)]}'
{"id":"openstack%2Fkeystone~929315","triplet_id":"openstack%2Fkeystone~master~I3c44ed0937679f01d554b2fa74c9916d54c68464","project":"openstack/keystone","branch":"master","topic":"claim-delimiter","attention_set":{},"removed_from_attention_set":{"5890":{"account":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"last_update":"2025-01-22 18:26:03.000000000","reason":"Change was submitted"},"7414":{"account":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"last_update":"2024-10-30 16:07:24.000000000","reason":"\u003cGERRIT_ACCOUNT_7414\u003e replied on the change","reason_account":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"}},"27900":{"account":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"last_update":"2024-10-21 15:53:47.000000000","reason":"\u003cGERRIT_ACCOUNT_27900\u003e replied on the change","reason_account":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}}},"hashtags":[],"change_id":"I3c44ed0937679f01d554b2fa74c9916d54c68464","subject":"extend docs explaining OIDC","status":"MERGED","created":"2024-09-13 17:56:17.000000000","updated":"2025-01-22 18:28:09.000000000","submitted":"2025-01-22 18:26:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"929315-claim-delimiter","meta_rev_id":"bb5ca23d5e11b93cd7e340268ee4fbc169f4ee34","_number":929315,"virtual_id_number":929315,"owner":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-01-22 18:26:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"all":[{"value":2,"date":"2024-10-30 16:07:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-01-22 15:31:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"all":[{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-01-22 15:31:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-13 18:12:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-30 22:18:34.000000000","updated_by":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"REVIEWER"},{"updated":"2024-10-04 14:02:13.000000000","updated_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"}],"messages":[{"id":"d754edabea3d1504e069df9c5cc7e727005de9c1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2024-09-13 17:56:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"45f543b26375b094f6aed19ac2ae04de3f7cc8a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-13 18:12:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b407cf0ed43443d8de7cb3dea0a6235\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff84da1c5eaa42aaaf9bf0404a90cc14 : SUCCESS in 5m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d892d7acb2a416585c2132e204a8c10 : SUCCESS in 11m 08s\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/f880ffd1c8234d378ec51846cce2163d : SUCCESS in 6m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"18d0f73766ac912b227e2d4dff35369ef724e5e9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2024-09-15 23:21:40.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"4b79f8885d80970ee7bd582ff4529409ad39301f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2024-09-15 23:22:24.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"0f22542bb20bc385af86335c51cb178e15ba459c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-15 23:31:40.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3659545a0d334bc6ab14dad55b90a94a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/123d69363cff44199bc176f9027cb9d0 : SUCCESS in 4m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0fd9156381964f1d952c004a5ac1dee4 : SUCCESS in 8m 29s\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/60f0d72d79f2457487f43840e44aa479 : SUCCESS in 3m 51s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"382775faea600d0ddfb24e65cff7a67f1b6f9481","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2024-09-30 22:18:34.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"9bf8f86eca8409aee0c295f84e63e98989d0d0ae","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2024-10-04 14:02:13.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"482932a8d67cbade47ed07a745c5e602a06a6485","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-10-10 13:20:41.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b73bb6ad2730fc965c6fda44c8f7869b5ebf9dec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2024-10-11 01:17:41.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"38b03e918b978fd8addb4f1cfe16a673ce0266fe","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2024-10-11 01:18:07.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"61b6d4d658b3c8ac8dff3dd43d03af3ae75367b8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-11 02:21:16.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb778741b1e947a081feb3069ee9eed9\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1e5789d3354b4c7c8ba8176c006b5194 : TIMED_OUT in 1h 01m 04s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4d9e53ad037e4d189bb130d8184edd54 : TIMED_OUT in 1h 01m 03s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"73e8c7848b717550cd86aaf445f9bc4f941abec2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-11 02:58:31.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/56324412ca7841a0b793873caa4e1621\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8083b25690fe42de94f89b8ac182cc3c : SUCCESS in 11m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/646e43bb496548069ab7a8b345a3710a : SUCCESS in 7m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0db555f2ecbf465ab9ea5b7307553655 : SUCCESS in 10m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a5bded16f74d4460912f827f69ccfb6d : SUCCESS in 10m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76999eb315e44061bdfc78dffe7fbbdc : SUCCESS in 13m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/3ac90bff2331456a9d8187790314da52 : SUCCESS in 42m 00s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ccbc3a5a154b449da25516409718084a : SUCCESS in 1h 40m 10s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/fb8eedeeda2a42da976c2c03f8a9f673 : SUCCESS in 31m 56s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/9feecf3203f147ce92fd11e3eae59403 : SUCCESS in 42m 07s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e98ced2abfeb49339aa9bf24a915a485 : FAILURE in 29m 14s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/1b2302fe8a0f4642a34db4acbe1dfc01 : SUCCESS in 28m 13s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/ebfca337f5b942d6993287723408cea2 : SUCCESS in 19m 08s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/5423cd9098c5413d9b943a8d54a9f210 : FAILURE in 10m 03s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/daa0520ad1154149b18fb3ed9b376048 : SUCCESS in 40m 23s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/a698e9427e37495e838e6254eb8a18dd : SUCCESS in 37m 18s (non-voting)\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/739c3927b72f441594c02915baca033d : SUCCESS in 5m 41s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"2b6eef005c21ed24a1d3c52aec32cb1472119603","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-10-21 15:53:47.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"6763647483865e7dc10d43b903228da9d3726578","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2024-10-30 16:07:24.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"75460bca5649b2c5a4027182771a96750a3cf44a","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2024-12-04 19:27:47.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"d13df17cd83e91f28f8af5ee9a820d6c99b7b828","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2025-01-22 15:31:40.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"6750a298833ea3136eec72fb1eb96cf50913fdaa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-22 15:32:17.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"318ab5bcc1370b72f88de7e51ded9952869e90cc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-22 18:26:02.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7969138955ca481694f3dd281e9f6aa7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ebfb7c64382b429b89c3252de9a42f69 : SUCCESS in 9m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/35991f0757414861af3adfac12af7fce : SUCCESS in 15m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1786c26e621f4faf8aed182005c0670e : SUCCESS in 13m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/342dbc06ad2d499fbaac55dfae40e2bd : SUCCESS in 12m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/55128ce5cfd54d81a0f5e405543ddc09 : SUCCESS in 38m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7c65cdbfa127473ba4df914338f290f7 : SUCCESS in 2h 37m 09s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/30973c56c57c4c9db465b64193f12dac : SUCCESS in 22m 31s\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/120ccf8f258642f5ae3ab1c4566a7297 : SUCCESS in 31m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/17a29745a6d44e229cda1550ef0be4f0 : SUCCESS in 49m 46s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/dfd48915076542e9a49be10ecf4f9d84 : FAILURE in 44m 36s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"b42705deb4d4d968481ca1a204edce2840e3ad01","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-22 18:26:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"bb5ca23d5e11b93cd7e340268ee4fbc169f4ee34","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-22 18:28:09.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c25544971dfa460b9959e7bc37f48ca9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d2908b067944a72a3a0d945ff06105c : SUCCESS in 1m 45s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"bc8bdce77ed3a93833feec8623159739f3fd6a6e","revisions":{"3bea34342bbc78eaf248e70a46a03bfe9738e761":{"kind":"REWORK","_number":1,"created":"2024-09-13 17:56:17.000000000","uploader":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"ref":"refs/changes/15/929315/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/15/929315/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/15/929315/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/15/929315/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/15/929315/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/15/929315/1"}}},"commit":{"parents":[{"commit":"97431ec99b255f3e04d4b21d38412e5d6b798618","subject":"Merge \"Remove logic to support pysaml2\u003c3.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/97431ec99b255f3e04d4b21d38412e5d6b798618"}]}],"author":{"name":"Doug Goldstein","email":"cardoe@cardoe.com","date":"2024-09-13 17:52:35.000000000","tz":-300},"committer":{"name":"Doug Goldstein","email":"cardoe@cardoe.com","date":"2024-09-13 17:56:14.000000000","tz":-300},"subject":"add note about delimiters for OIDC","message":"add note about delimiters for OIDC\n\nmod_auth_openidc takes lists in claims and when it converts them to a\nheader it uses a comma as its delimiter by default. But keystone uses ;\nas it\u0027s delimiter when using a field in a mapping as a list. So the user\nmust configure mod_auth_openidc to use a ; instead of a ,\n\nChange-Id: I3c44ed0937679f01d554b2fa74c9916d54c68464\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/3bea34342bbc78eaf248e70a46a03bfe9738e761"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/3bea34342bbc78eaf248e70a46a03bfe9738e761"}]},"branch":"refs/heads/master"},"a91a983c0b544dcd1bf1d196dc6416b9adfe3f18":{"kind":"REWORK","_number":2,"created":"2024-09-15 23:21:40.000000000","uploader":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"ref":"refs/changes/15/929315/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/15/929315/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/15/929315/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/15/929315/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/15/929315/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/15/929315/2"}}},"commit":{"parents":[{"commit":"97431ec99b255f3e04d4b21d38412e5d6b798618","subject":"Merge \"Remove logic to support pysaml2\u003c3.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/97431ec99b255f3e04d4b21d38412e5d6b798618"}]}],"author":{"name":"Doug Goldstein","email":"cardoe@cardoe.com","date":"2024-09-13 17:52:35.000000000","tz":-300},"committer":{"name":"Doug Goldstein","email":"cardoe@cardoe.com","date":"2024-09-15 23:16:52.000000000","tz":-300},"subject":"extend docs explaining OIDC","message":"extend docs explaining OIDC\n\nOpenID Connect has specific terms for different parts so add those to\nthe documentation or use them when the wrong term was used. Added some\nmore expanded details around the different settings and how they work.\nmod_auth_openidc takes lists in claims and when it converts them to a\nheader it uses a comma as its delimiter by default. But keystone uses ;\nas it\u0027s delimiter when using a field in a mapping as a list. So the user\nmust configure mod_auth_openidc to use a ; instead of a ,\n\nChange-Id: I3c44ed0937679f01d554b2fa74c9916d54c68464\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/a91a983c0b544dcd1bf1d196dc6416b9adfe3f18"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/a91a983c0b544dcd1bf1d196dc6416b9adfe3f18"}]},"branch":"refs/heads/master"},"5e7d3dc12b23b26102060146f599d101a7e3a933":{"kind":"REWORK","_number":3,"created":"2024-09-15 23:22:24.000000000","uploader":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"ref":"refs/changes/15/929315/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/15/929315/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/15/929315/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/15/929315/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/15/929315/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/15/929315/3"}}},"commit":{"parents":[{"commit":"97431ec99b255f3e04d4b21d38412e5d6b798618","subject":"Merge \"Remove logic to support pysaml2\u003c3.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/97431ec99b255f3e04d4b21d38412e5d6b798618"}]}],"author":{"name":"Doug Goldstein","email":"cardoe@cardoe.com","date":"2024-09-13 17:52:35.000000000","tz":-300},"committer":{"name":"Doug Goldstein","email":"cardoe@cardoe.com","date":"2024-09-15 23:22:19.000000000","tz":-300},"subject":"extend docs explaining OIDC","message":"extend docs explaining OIDC\n\nOpenID Connect has specific terms for different parts so add those to\nthe documentation or use them when the wrong term was used. Added some\nmore expanded details around the different settings and how they work.\nmod_auth_openidc takes lists in claims and when it converts them to a\nheader it uses a comma as its delimiter by default. But keystone uses ;\nas it\u0027s delimiter when using a field in a mapping as a list. So the user\nmust configure mod_auth_openidc to use a ; instead of a ,\n\nChange-Id: I3c44ed0937679f01d554b2fa74c9916d54c68464\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/5e7d3dc12b23b26102060146f599d101a7e3a933"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/5e7d3dc12b23b26102060146f599d101a7e3a933"}]},"branch":"refs/heads/master"},"bc8bdce77ed3a93833feec8623159739f3fd6a6e":{"kind":"REWORK","_number":4,"created":"2024-10-11 01:17:41.000000000","uploader":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"ref":"refs/changes/15/929315/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/15/929315/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/15/929315/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/15/929315/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/15/929315/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/15/929315/4"}}},"commit":{"parents":[{"commit":"97431ec99b255f3e04d4b21d38412e5d6b798618","subject":"Merge \"Remove logic to support pysaml2\u003c3.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/97431ec99b255f3e04d4b21d38412e5d6b798618"}]}],"author":{"name":"Doug Goldstein","email":"cardoe@cardoe.com","date":"2024-09-13 17:52:35.000000000","tz":-300},"committer":{"name":"Doug Goldstein","email":"cardoe@cardoe.com","date":"2024-10-11 01:17:20.000000000","tz":-300},"subject":"extend docs explaining OIDC","message":"extend docs explaining OIDC\n\nOpenID Connect has specific terms for different parts so add those to\nthe documentation or use them when the wrong term was used. Added some\nmore expanded details around the different settings and how they work.\nmod_auth_openidc takes lists in claims and when it converts them to a\nheader it uses a comma as its delimiter by default. But keystone uses ;\nas it\u0027s delimiter when using a field in a mapping as a list. So the user\nmust configure mod_auth_openidc to use a ; instead of a ,\n\nChange-Id: I3c44ed0937679f01d554b2fa74c9916d54c68464\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/bc8bdce77ed3a93833feec8623159739f3fd6a6e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/bc8bdce77ed3a93833feec8623159739f3fd6a6e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
