)]}'
{"id":"openstack%2Fkeystone~929736","triplet_id":"openstack%2Fkeystone~master~Ic44c945f3cb65b48ff72052fd2b4f6d45e118b2e","project":"openstack/keystone","branch":"master","topic":"idx_revocation_event_project_id_user_id","attention_set":{},"removed_from_attention_set":{"37313":{"account":{"_account_id":37313,"name":"Alma MAILLET-CONTOZ","display_name":"Alma MAILLET-CONTOZ","email":"alma.maillet-contoz@infomaniak.com","username":"nicolasmc","status":"Intern SRE at Infomaniak"},"last_update":"2025-12-05 18:11:18.000000000","reason":"Change was submitted"},"27442":{"account":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"last_update":"2025-12-05 18:11:18.000000000","reason":"Change was submitted"},"37315":{"account":{"_account_id":37315,"name":"Mathias Heyraud","display_name":"mathias.heyraud","email":"mathias.heyraud@infomaniak.com","username":"mathias.heyraud"},"last_update":"2025-12-05 18:11:18.000000000","reason":"Change was submitted"},"37203":{"account":{"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},"last_update":"2025-12-05 18:11:18.000000000","reason":"Change was submitted"},"29349":{"account":{"_account_id":29349,"name":"Simeon Gourlin","email":"simeon.gourlin@infomaniak.com","username":"6meon"},"last_update":"2025-12-05 18:11:18.000000000","reason":"Change was submitted"},"35895":{"account":{"_account_id":35895,"name":"Theo Gindre","display_name":"theogndr","email":"theo.gindre@gmail.com","username":"theogndr"},"last_update":"2025-12-05 18:11:18.000000000","reason":"Change was submitted"},"27900":{"account":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"last_update":"2025-12-05 18:11:18.000000000","reason":"Change was submitted"},"6476":{"account":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"last_update":"2025-12-05 18:11:18.000000000","reason":"Change was submitted"},"29422":{"account":{"_account_id":29422,"name":"Olivier Chaze","display_name":"Olivier Chaze","email":"olivier.chaze@infomaniak.com","username":"o.chaze"},"last_update":"2025-12-05 18:11:18.000000000","reason":"Change was submitted"},"11583":{"account":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"last_update":"2025-12-05 18:11:18.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic44c945f3cb65b48ff72052fd2b4f6d45e118b2e","subject":"Add a new index on revocation_event table","status":"MERGED","created":"2024-09-18 07:49:05.000000000","updated":"2025-12-05 18:11:18.000000000","submitted":"2025-12-05 18:11:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":12,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"929736-idx_revocation_event_project_id_user_id","meta_rev_id":"5156b522f7f77d133bf46b7912dd6c24a5beb5d3","_number":929736,"virtual_id_number":929736,"owner":{"_account_id":29422,"name":"Olivier Chaze","display_name":"Olivier Chaze","email":"olivier.chaze@infomaniak.com","username":"o.chaze"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":37313,"name":"Alma MAILLET-CONTOZ","display_name":"Alma MAILLET-CONTOZ","email":"alma.maillet-contoz@infomaniak.com","username":"nicolasmc","status":"Intern SRE at Infomaniak"},{"value":0,"_account_id":37315,"name":"Mathias Heyraud","display_name":"mathias.heyraud","email":"mathias.heyraud@infomaniak.com","username":"mathias.heyraud"},{"value":0,"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},{"value":0,"_account_id":29349,"name":"Simeon Gourlin","email":"simeon.gourlin@infomaniak.com","username":"6meon"},{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":35895,"name":"Theo Gindre","display_name":"theogndr","email":"theo.gindre@gmail.com","username":"theogndr"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-12-05 18:11:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"all":[{"value":0,"_account_id":37313,"name":"Alma MAILLET-CONTOZ","display_name":"Alma MAILLET-CONTOZ","email":"alma.maillet-contoz@infomaniak.com","username":"nicolasmc","status":"Intern SRE at Infomaniak"},{"value":0,"_account_id":37315,"name":"Mathias Heyraud","display_name":"mathias.heyraud","email":"mathias.heyraud@infomaniak.com","username":"mathias.heyraud"},{"value":0,"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},{"value":0,"_account_id":29349,"name":"Simeon Gourlin","email":"simeon.gourlin@infomaniak.com","username":"6meon"},{"value":2,"date":"2025-12-05 15:49:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":35895,"name":"Theo Gindre","display_name":"theogndr","email":"theo.gindre@gmail.com","username":"theogndr"},{"value":2,"date":"2025-12-05 09:48:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"all":[{"value":0,"_account_id":37313,"name":"Alma MAILLET-CONTOZ","display_name":"Alma MAILLET-CONTOZ","email":"alma.maillet-contoz@infomaniak.com","username":"nicolasmc","status":"Intern SRE at Infomaniak"},{"value":0,"_account_id":37315,"name":"Mathias Heyraud","display_name":"mathias.heyraud","email":"mathias.heyraud@infomaniak.com","username":"mathias.heyraud"},{"value":0,"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},{"value":0,"_account_id":29349,"name":"Simeon Gourlin","email":"simeon.gourlin@infomaniak.com","username":"6meon"},{"value":1,"date":"2025-12-05 15:49:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":35895,"name":"Theo Gindre","display_name":"theogndr","email":"theo.gindre@gmail.com","username":"theogndr"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"}],"reviewers":{"REVIEWER":[{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"_account_id":29349,"name":"Simeon Gourlin","email":"simeon.gourlin@infomaniak.com","username":"6meon"},{"_account_id":35895,"name":"Theo Gindre","display_name":"theogndr","email":"theo.gindre@gmail.com","username":"theogndr"},{"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},{"_account_id":37313,"name":"Alma MAILLET-CONTOZ","display_name":"Alma MAILLET-CONTOZ","email":"alma.maillet-contoz@infomaniak.com","username":"nicolasmc","status":"Intern SRE at Infomaniak"},{"_account_id":37315,"name":"Mathias Heyraud","display_name":"mathias.heyraud","email":"mathias.heyraud@infomaniak.com","username":"mathias.heyraud"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-18 08:00:01.000000000","updated_by":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"reviewer":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"state":"REVIEWER"},{"updated":"2024-09-18 08:00:15.000000000","updated_by":{"_account_id":35895,"name":"Theo Gindre","display_name":"theogndr","email":"theo.gindre@gmail.com","username":"theogndr"},"reviewer":{"_account_id":35895,"name":"Theo Gindre","display_name":"theogndr","email":"theo.gindre@gmail.com","username":"theogndr"},"state":"REVIEWER"},{"updated":"2024-09-18 08:03:47.000000000","updated_by":{"_account_id":29349,"name":"Simeon Gourlin","email":"simeon.gourlin@infomaniak.com","username":"6meon"},"reviewer":{"_account_id":29349,"name":"Simeon Gourlin","email":"simeon.gourlin@infomaniak.com","username":"6meon"},"state":"REVIEWER"},{"updated":"2024-09-18 08:05:46.000000000","updated_by":{"_account_id":37313,"name":"Alma MAILLET-CONTOZ","display_name":"Alma MAILLET-CONTOZ","email":"alma.maillet-contoz@infomaniak.com","username":"nicolasmc","status":"Intern SRE at Infomaniak"},"reviewer":{"_account_id":37313,"name":"Alma MAILLET-CONTOZ","display_name":"Alma MAILLET-CONTOZ","email":"alma.maillet-contoz@infomaniak.com","username":"nicolasmc","status":"Intern SRE at Infomaniak"},"state":"REVIEWER"},{"updated":"2024-09-18 08:41:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-09-18 08:48:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-18 09:42:19.000000000","updated_by":{"_account_id":37315,"name":"Mathias Heyraud","display_name":"mathias.heyraud","email":"mathias.heyraud@infomaniak.com","username":"mathias.heyraud"},"reviewer":{"_account_id":37315,"name":"Mathias Heyraud","display_name":"mathias.heyraud","email":"mathias.heyraud@infomaniak.com","username":"mathias.heyraud"},"state":"REVIEWER"},{"updated":"2024-09-18 09:44:33.000000000","updated_by":{"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},"reviewer":{"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},"state":"REVIEWER"},{"updated":"2025-11-24 08:25:32.000000000","updated_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"reviewer":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"state":"CC"},{"updated":"2025-11-24 08:26:54.000000000","updated_by":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"reviewer":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"state":"CC"},{"updated":"2025-11-24 09:35:34.000000000","updated_by":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"reviewer":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"state":"REVIEWER"},{"updated":"2025-11-24 13:52:13.000000000","updated_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"REVIEWER"},{"updated":"2025-11-24 13:53:16.000000000","updated_by":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"reviewer":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"state":"REVIEWER"},{"updated":"2025-12-05 09:48:35.000000000","updated_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"},{"updated":"2025-12-05 15:49:11.000000000","updated_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"}],"messages":[{"id":"807abf575c6c9321a4737917a5bc71260a1f77d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29422,"name":"Olivier Chaze","display_name":"Olivier Chaze","email":"olivier.chaze@infomaniak.com","username":"o.chaze"},"date":"2024-09-18 07:49:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7f13a8882028550a3163cabddd08a7f3532308e3","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2024-09-18 08:00:01.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"6a7b9ce37a419e43a67d8757911002408388d5cc","author":{"_account_id":35895,"name":"Theo Gindre","display_name":"theogndr","email":"theo.gindre@gmail.com","username":"theogndr"},"date":"2024-09-18 08:00:15.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"520618f8cdf3ab690132bc1cb2f3aceec985da9c","author":{"_account_id":29349,"name":"Simeon Gourlin","email":"simeon.gourlin@infomaniak.com","username":"6meon"},"date":"2024-09-18 08:03:47.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"3841130ad827edb98154ef5ea1e03f5aed05e6b6","author":{"_account_id":37313,"name":"Alma MAILLET-CONTOZ","display_name":"Alma MAILLET-CONTOZ","email":"alma.maillet-contoz@infomaniak.com","username":"nicolasmc","status":"Intern SRE at Infomaniak"},"date":"2024-09-18 08:05:46.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"6aa11ae1289c7423840b7703db69734e4f4a4029","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-18 08:41:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b8d689113781478b8dc66af1ddbd231c\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6a82f6c8fece40368b022ee833303159 : FAILURE in 46m 23s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a98d0c3752a2494aad1acd8f07bc430c : FAILURE in 51m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a1649b92fe2511a0dd69e4eb995bd52e50b3eecc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-18 08:48:34.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f7e13719ad0047d1b2e86f71f435a92a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/000ea398280b4671873495796393879d : FAILURE in 11m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd8cf34aa5f04ab6a008a1d5bb204b5b : FAILURE in 6m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c004c56130ab48478dd3e4444876f7fa : FAILURE in 10m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c618acfbec2d4096b9022f6d0284d5b5 : FAILURE in 11m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1459bfc1999d4b1a8f70066955d2fac0 : FAILURE in 11m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17bd40dc200d4b4492f18a96011991fb : SUCCESS in 13m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/b2a9b90a8f624dc5a8a295f16fac57cb : FAILURE in 52m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/363450b346a047bebf272399e3873227 : FAILURE in 12m 59s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/f145c609d4b04abea8b378035eb71216 : FAILURE in 25m 38s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/19e06627ce9544c2b5fdbf9238d889c3 : FAILURE in 18m 33s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fc70484fba3642388df46382f968ef7f : FAILURE in 13m 09s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/9c709e4e8e18459ca74c164aa583da90 : FAILURE in 12m 49s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/8630115d4c2a480fb3ef0b063c7acc65 : POST_FAILURE in 16m 05s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/5b9d09ea5ae342528df5df970d928213 : FAILURE in 12m 44s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/62b00a74bece48668573f0b9d70febb7 : FAILURE in 21m 38s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/f2a98a29558f4b3bafc93721f39c359c : FAILURE in 9m 13s (non-voting)\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/7dc845f9b1694b4fbe0effffc0dc8123 : SUCCESS in 6m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"40ff3b44979a8d6bf837261ab31677243a3a0892","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29422,"name":"Olivier Chaze","display_name":"Olivier Chaze","email":"olivier.chaze@infomaniak.com","username":"o.chaze"},"date":"2024-09-18 09:00:00.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"a7bb86ce16ed55bb09af1efd3c71a6bd656bf00d","author":{"_account_id":37315,"name":"Mathias Heyraud","display_name":"mathias.heyraud","email":"mathias.heyraud@infomaniak.com","username":"mathias.heyraud"},"date":"2024-09-18 09:42:19.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"4d5e2deee25ee17471e53e748cf20750ad2a522b","author":{"_account_id":37203,"name":"Bertrand Lanson","display_name":"Bertrand Lanson","email":"bertrand.lanson@infomaniak.com","username":"lanson","status":"Infomaniak Network SA"},"date":"2024-09-18 09:44:33.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"95e72f0bcd6ea0b66a9d4b448b14ff81da110231","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-18 09:51:45.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8b68f78ac1d241a3a4014f188c44cd19\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f1abb2e9d4b047f785f90a56a01ff1b6 : FAILURE in 16m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5392813a4564ec7be7e8eca0e0fcc9d : FAILURE in 5m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4044c8260d10493c96a54260403c7935 : FAILURE in 15m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d2debdf8bd9a4686bed0dafc09bb8e2f : FAILURE in 10m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0b6f1cc31c7745068a98d30ff80c0507 : FAILURE in 15m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb80abc5ce1d4c618f74bd965f44d2e1 : SUCCESS in 12m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/132ad188e5f04e6f945f019201807765 : FAILURE in 46m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7446c5047ede44068165e53e76f3a06c : FAILURE in 23m 35s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/4a7d57781dc740018a12ac4a9216cfe7 : FAILURE in 22m 30s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/d3a012d5bb28407ca7b7bf83a5f30ed2 : FAILURE in 19m 11s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f28f24af8fc045e0abf23c2373d719e2 : FAILURE in 13m 06s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/c48647e5fc62490d97c596699e8ab1ee : FAILURE in 13m 12s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/6623c04b045e4f418245c9ab260781d4 : POST_FAILURE in 16m 13s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/c9c7de1d2648456d809b072093856019 : FAILURE in 13m 15s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/29a67d3de5ed4fb589feaec2cec1a58c : FAILURE in 13m 07s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/f252b82569244cef8d2c5bf0aeff5cbc : FAILURE in 19m 30s (non-voting)\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/8bfb67974df04b44b034e3c959486d8f : SUCCESS in 4m 18s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7b746a9eadcab63b738a0effca7ad8b6bde5ea87","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-18 10:36:34.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bb412b9c451d41e1a32ddfc247444031\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/66e12f8feb6a4a9c98f70c9765695d77 : TIMED_OUT in 1h 01m 01s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/40889ca1eadf4a63886156bf74643889 : FAILURE in 50m 27s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"56f962ae58b3770043483c9802c0244dad171e95","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29422,"name":"Olivier Chaze","display_name":"Olivier Chaze","email":"olivier.chaze@infomaniak.com","username":"o.chaze"},"date":"2024-09-18 15:54:38.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"294098f99dbc473818db80a0f31edf4dcbf2da61","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-18 16:35:14.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6c8915c8ffaa48758a6287e2c6c6d0b6\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f35a7f3d74a54f38932ae626dfc70965 : FAILURE in 15m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ee1c71d370c4b37bd84aafe3eab0deb : FAILURE in 6m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/74a34a65744747b784d087ce893a952b : FAILURE in 12m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ac5e82a33ff64b1faff4281083c5dc1c : FAILURE in 10m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5fd2559851304e7db0598107ea5a385a : FAILURE in 14m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1f49208a5074e7583697913d168710d : SUCCESS in 13m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/f63e6b6dc32d442999b3e214b859fc1e : FAILURE in 39m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e1f62c8e8b4541f3b6f929c06865259a : FAILURE in 14m 00s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/e497ba5244b14fc182eb7992332d9e17 : FAILURE in 15m 03s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/806c2cea1770418ca60576a6117924ab : FAILURE in 21m 55s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b8a1639309594ba6a119e600bdae5d4f : FAILURE in 22m 48s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/8ccbf519b93047febb5de14b6bc69dbf : FAILURE in 12m 13s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/057179587ce349f884664f17277c76c3 : POST_FAILURE in 7m 33s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/2b5e54b00639400dbc9dd138d887bb58 : FAILURE in 20m 59s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7ae17ea3152a43479b5c536907a621ee : FAILURE in 20m 52s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/e1ac9d88a0824d029eb2a9e092d96a67 : FAILURE in 18m 11s (non-voting)\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/1bf2bd19b3354c29b39e5205955c2b9a : SUCCESS in 5m 59s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8d3d4a5488f71ebff2e6e4e80f64196eb4520c33","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-18 22:51:59.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/72297c1654504b5d8bea7bf2dbe60836\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8187a66688b040919287e020f825c5a4 : FAILURE in 58m 28s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/c2396226e05d4d6da76bea20564da1bb : TIMED_OUT in 1h 01m 06s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"d86e018a269e2c062348fa6175ddfa2b23c7bc76","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-11-24 08:32:27.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"9cd0cb65883c0d1c0a47e6ce68b7cc122b5b42e4","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-11-24 08:32:45.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c26c98fd56b4e32454a57a701fc9262457ce535c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"date":"2025-11-24 09:35:34.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"5cf1bbe5bc5ddc1b8711e6d17c0967cb49338f19","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-24 11:05:09.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b488bef3e9d342c38a9da5d2c01940fc\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6809534c507842df87c21ecf9982fa5a : SUCCESS in 31m 54s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/056b0ad10bc94d27920cddc6e002e5cc : SUCCESS in 15m 17s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/eb8066199a9f4b988a80bca61adbd3d8 : TIMED_OUT in 1h 01m 02s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"ca7ed07b888e7288d6dac16bbd5212a7d2f46383","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-24 11:27:52.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b8e9cfdc337b40aca65e55e4ed035df4\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cde958eb5cc948e8a07b35f7d3365a8b : SUCCESS in 15m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e9b128c1a1b40c29c17883cce5b1bcf : FAILURE in 9m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/463f489dfa7e45d1bfa68c2fbceed9da : SUCCESS in 13m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fdbc0259bc344fd7ba6aeac140bc10e4 : SUCCESS in 12m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ade1afb456d47a19342402b8080d7a5 : SUCCESS in 13m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/b87808a79a714358961ce51850675330 : SUCCESS in 45m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/20ea83b22f674bb48be45801f25a6aa1 : SUCCESS in 1h 46m 17s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/e41a6a2778e343418f37555592dfbec1 : SUCCESS in 7m 01s (non-voting)\n- keystone-tempest https://zuul.opendev.org/t/openstack/build/f4748e2752cf446e8fab06b0e57d4b53 : SUCCESS in 31m 58s\n- keystone-tempest-fips https://zuul.opendev.org/t/openstack/build/9a812e72455f4e429cf1e7744c4f9940 : FAILURE in 22m 09s (non-voting)\n- keystone-tempest-federation https://zuul.opendev.org/t/openstack/build/1d0cb3f544f040f1973b78923c8eb093 : FAILURE in 27m 44s (non-voting)\n- keystone-tempest-federation-k2k https://zuul.opendev.org/t/openstack/build/539d27cc1ff940dfa80fcbc53f61c23e : SUCCESS in 31m 02s\n- keystone-tempest-oidc-federation https://zuul.opendev.org/t/openstack/build/ace9ce04a86443078cf2854d3a9f19ee : SUCCESS in 12m 37s\n- keystone-tempest-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/18b117743fb74cc6a5556453c7df3bc9 : SUCCESS in 40m 27s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ba411dc71d3348efb1173eddb278da41 : SUCCESS in 1h 05m 31s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/a08653f7e18443eda2fb706849af117e : FAILURE in 17m 40s (non-voting)\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/77b3e297060a49158e7fcbaa26db2cd0 : SUCCESS in 8m 47s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"07944bc65983e685009a6a0178e4999e8a8c227e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2025-11-24 13:52:13.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"74d554b9cfca9eb673595e367cdcb7bc6ce9a7e9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-11-24 13:53:16.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"8bdebf0580739bfc09cb720186ecc1a98488caf9","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2025-11-24 13:55:09.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"ddcc4269f718bef6b0009a50e4e183599f87f6e5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-24 14:44:01.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/985577e48a974a5d828b9b3e15b4473b\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c334924e0f7040cca1844d6f6698fe70 : SUCCESS in 38m 56s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/92abf31fd950473d8839636353da668a : SUCCESS in 36m 32s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/70181b82ec7b47db940f060de52fb55a : SUCCESS in 47m 21s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"55fc281abef5a4ab6826b7a09ad6f7d800a46d18","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-24 15:40:33.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/000b7fc6a8d3479b94279d328a10c97a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e24d72150c294d818338ebfc4dea6417 : SUCCESS in 14m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00271855161e4194a0e9dd5bd296b337 : FAILURE in 6m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0e9559e4363c447aaf3102f32a0e2907 : SUCCESS in 12m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8d8c9717fc024197bcd467a7053aa8ae : SUCCESS in 13m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2384725a8254b90bdd9ab883989e2df : SUCCESS in 7m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/81ae4fb2feba41dd819f4a514a6c8fc5 : SUCCESS in 57m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0e10bfd55d5c491fa0c30b2a5e8a5261 : SUCCESS in 1h 44m 14s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/020822622f634242a33a61b500f99c08 : SUCCESS in 17m 34s (non-voting)\n- keystone-tempest https://zuul.opendev.org/t/openstack/build/730b28a2796b4418863fe38f86667b5f : SUCCESS in 28m 00s\n- keystone-tempest-fips https://zuul.opendev.org/t/openstack/build/61a654ad2d1847d2bfe4066c53ad11a0 : FAILURE in 19m 12s (non-voting)\n- keystone-tempest-federation https://zuul.opendev.org/t/openstack/build/d58a4aa342434fd18d08462516791ac5 : FAILURE in 27m 07s (non-voting)\n- keystone-tempest-federation-k2k https://zuul.opendev.org/t/openstack/build/776378b916754b20be5bef99c987db29 : SUCCESS in 29m 45s\n- keystone-tempest-oidc-federation https://zuul.opendev.org/t/openstack/build/24839efb9e164628b4b0d926713144b3 : SUCCESS in 32m 13s\n- keystone-tempest-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/79caf2175a3f4b6aa9dcba26390a08d7 : SUCCESS in 28m 11s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/679dd51e72e949a7a8d60fe71995ab7d : SUCCESS in 1h 17m 22s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/d02314335409473e9fdb7b6deace0e40 : FAILURE in 38m 43s (non-voting)\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/b81e67d1efe64a1baf78a2df31cd453d : SUCCESS in 5m 57s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"feaa0ad03d69f3af1c00a9b305af0795ed60a4ec","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-11-24 19:30:20.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"9df3435988bbbf05bab947050e3d017682a316c0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"date":"2025-11-25 08:02:27.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"082481b7bf7c73e6c22d35aefcaad21e03035e94","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-25 08:34:17.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/21afaee5356f40d9ad5ccc53938d20b6\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/0b4b278162a644a497c27b26145bad3e : SUCCESS in 22m 11s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0838d44b7ebb4aa3b9252ec1e000b12e : SUCCESS in 21m 47s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/6c25b5dc46da4545aafe49f91a94ccc6 : SUCCESS in 29m 28s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"ee357fe8900d56f6b1b222205f2158704d7e04b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-25 09:48:50.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/527e01439d1d4e618bdf8a5228b8ceec\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/202d18ea6edb4f62a15bd2a7a38d5d7f : SUCCESS in 11m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27f6be907f3c40c0aff71357a2f12411 : SUCCESS in 8m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/696443d5e6fa431180bc428ebf6bb255 : SUCCESS in 11m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/72029688620d4505b578c708c5463ff4 : SUCCESS in 11m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2baa112b39f341198d9e112a66ba3980 : SUCCESS in 11m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/7a5c08e4b9ff4774a289dee50b91c5f5 : SUCCESS in 1h 01m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/519bb1f9de9f4a06a6db8d7ab87c5539 : SUCCESS in 1h 40m 59s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/864bf05dd81c4d8cbb16e7b9e959cf1d : SUCCESS in 10m 24s (non-voting)\n- keystone-tempest https://zuul.opendev.org/t/openstack/build/3a12f98946a242c88468a42bdd313d25 : SUCCESS in 27m 30s\n- keystone-tempest-fips https://zuul.opendev.org/t/openstack/build/33d90e38797742dcbc0eddf508f85a06 : FAILURE in 9m 22s (non-voting)\n- keystone-tempest-federation https://zuul.opendev.org/t/openstack/build/99f4b04383f245608b0c5836b35c5ec5 : FAILURE in 17m 05s (non-voting)\n- keystone-tempest-federation-k2k https://zuul.opendev.org/t/openstack/build/4b6a832ad86543e394770f4a9fd7e8fe : SUCCESS in 21m 28s\n- keystone-tempest-oidc-federation https://zuul.opendev.org/t/openstack/build/422ed8dd2efe4947bb22f86078aa56ba : SUCCESS in 28m 55s\n- keystone-tempest-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/ed517307711245409c6be9f59e79d3ca : SUCCESS in 42m 29s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1adb843677114b79ac444f8eeecb7163 : SUCCESS in 46m 44s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/9c56664fc653472c8f021489e4661f4c : FAILURE in 40m 10s (non-voting)\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/ce22027af75c47cb9b947ac73c1d8ecc : SUCCESS in 6m 56s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"59bd7d5563da671865d1afce3abdcfb73fc73655","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2025-11-25 11:25:54.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"a2c0d1af814bee2110c2bfe70fbc0693b99b65ac","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-11-25 12:32:12.000000000","message":"Patch Set 7: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"450ddf76f03184dc557e0301d9efebc2adfd6881","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-11-25 12:34:20.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"6e1a1916dc087b07d713fd792931a15aaa718857","author":{"_account_id":29422,"name":"Olivier Chaze","display_name":"Olivier Chaze","email":"olivier.chaze@infomaniak.com","username":"o.chaze"},"date":"2025-11-25 13:19:38.000000000","message":"Patch Set 7:\n\n(3 comments)","accounts_in_message":[],"_revision_number":7},{"id":"283504bb7925c3379e896d6708ed9408239bcbb6","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-11-25 13:27:04.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"604518f200de4b46767720743065aaafeac62258","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2025-11-25 13:32:07.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"021be8da70cf3c68394260966859561e076cfa9d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-11-25 16:57:08.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"7511a18306dddb34a3e91c649a37d8d99c964903","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-25 17:24:20.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/049eeb21cf78421d9aa7eb8c7b7bd957\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/cb07bb4741b24e94b1b3150fd836d8e6 : FAILURE in 23m 37s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ad34796bf5624e45bc96628b9e3d537f : FAILURE in 22m 57s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a29fe8a2921d465fbe4cc36020043b5e : FAILURE in 25m 10s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"e17f6ae023bf37f1bf5740cc7eff517fd5b92c31","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-25 17:58:23.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/462060a6b29d49898a1fa1f98f7f0147\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f38a3025a2264128b25850d5e7b09338 : FAILURE in 11m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6487aea81c0475a9db241428a1f38fd : FAILURE in 8m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/93036199ee5448d5a161cc2d7d46a9e6 : FAILURE in 11m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6f996b0a32f54d039fbca15a8adf8909 : FAILURE in 16m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/462dd6d57cb043f8b7301f265866834e : SUCCESS in 14m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/a82e056cb66445b5ab929fa7dc79e466 : FAILURE in 54m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8e270a2aed9441e1b28372f27917eea0 : FAILURE in 17m 26s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/553b8c088b33448d83908a8009d68a5a : POST_FAILURE in 7m 32s (non-voting)\n- keystone-tempest https://zuul.opendev.org/t/openstack/build/6542258e54954ea6b1c6f3791b346b41 : FAILURE in 16m 57s\n- keystone-tempest-fips https://zuul.opendev.org/t/openstack/build/3e0b431235f046729d1e0f34733e163f : FAILURE in 11m 09s (non-voting)\n- keystone-tempest-federation https://zuul.opendev.org/t/openstack/build/c098ae6814444302a5a0708fe897485c : FAILURE in 17m 52s (non-voting)\n- keystone-tempest-federation-k2k https://zuul.opendev.org/t/openstack/build/98cbfba510ff40ab89263c4fc0c39423 : FAILURE in 18m 33s\n- keystone-tempest-oidc-federation https://zuul.opendev.org/t/openstack/build/fadd8b3ff2d24758b8f006e74f3fd18d : FAILURE in 19m 39s\n- keystone-tempest-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/4a3dc8e1604c422eb5130426be8a4ca9 : FAILURE in 16m 53s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/efa1d3fe480c47759c5f4a711b8d92d7 : FAILURE in 8m 40s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/1756c6259f574dfe8777cc1c2149b647 : FAILURE in 15m 08s (non-voting)\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/c47553e7312b4c64993ef58355aca172 : SUCCESS in 6m 53s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"e8778ac8cd1a523cae2132f63a8d7f80b250eed0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-11-25 19:00:10.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"b74f7cd2c72f2f2b58b89536f2690899f306a43d","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-11-25 19:26:59.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"b41b9fb924605a69f64aafbe6c60473317364b6d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-25 19:34:29.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3a68a860c94d42dc8fa8cf5d3bea3d7d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0845a384d0db4687a96ca354f8b2e5b5 : FAILURE in 14m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79d46d14e40c4c169471f99eec22265c : SUCCESS in 5m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/15c68b7999344f0495a86ae8e1f346ec : FAILURE in 12m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f019a31372cc4df9912fb03c75537125 : FAILURE in 12m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3488b17281d64e14bd0d2e86198d3e56 : SUCCESS in 11m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/7c28c0da9c6545aab3788e31bcb1419d : FAILURE in 30m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1fb0d296c2594580940eb94bd18fabfb : FAILURE in 19m 03s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/a7e07d77980d44baa9dcd84730460874 : POST_FAILURE in 15m 31s (non-voting)\n- keystone-tempest https://zuul.opendev.org/t/openstack/build/503a85bb97dd40e795a2d12c2f454aa6 : FAILURE in 17m 46s\n- keystone-tempest-fips https://zuul.opendev.org/t/openstack/build/baa85bda512b4dbf84a4a032c569b008 : FAILURE in 21m 49s (non-voting)\n- keystone-tempest-federation https://zuul.opendev.org/t/openstack/build/57f17cf1977f4556bc396eed49f8b0d8 : FAILURE in 18m 28s (non-voting)\n- keystone-tempest-federation-k2k https://zuul.opendev.org/t/openstack/build/08082b889f08411c9a5bb6ee2101d0c1 : FAILURE in 17m 09s\n- keystone-tempest-oidc-federation https://zuul.opendev.org/t/openstack/build/5d949e89ac6d4fef99a4cf4d499db53c : FAILURE in 18m 52s\n- keystone-tempest-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/a543116d6a934cad9c584470669663ef : FAILURE in 14m 09s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/49a10b96fd3d49299cc081571cce8cd2 : FAILURE in 10m 19s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/4b7930153c454deb9c1fa9a09acf3edf : FAILURE in 8m 27s (non-voting)\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/56ef9946e797466387b824fe7c046a47 : SUCCESS in 6m 32s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"c69a6033df0b0d1cdcec8b59be8baca52b8dc325","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-25 20:01:29.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/095ab0337a9041e6baaddc30318ec5c5\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6333b9bfb29a4dde806ed74601e5f672 : FAILURE in 30m 50s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5d2bb0f095c54a3a8b2a7d52950ec5b1 : FAILURE in 15m 03s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/72dc56f716224e16b45dfd773c9fb46f : FAILURE in 40m 05s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"f4047d73e44e964f6bc913bb52277b0f5bbe84cd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2025-12-04 22:20:48.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"00d63c4f30a0189131399ec292ddafc8aba7fc61","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-05 00:22:19.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/41469ad9c17449ee8e38744a870d7163\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a0ddd85287b74c55b4ab23a856d9f176 : SUCCESS in 16m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bac6069cf64d48aba2208ea0f9ab75bc : SUCCESS in 8m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7b8a37e81fa2487789143980d5fd9f86 : SUCCESS in 9m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/88619ffe5ef441f8b703b76422079628 : SUCCESS in 23m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1d82f10af994755aede86b8a2b6e646 : SUCCESS in 7m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/2588f1ef96694e77bbe2419208e0011b : SUCCESS in 31m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/fb76c145525e4ec68dc3c031956b9838 : SUCCESS in 1h 56m 12s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/9ff0a57b239049248745c49b35da4756 : SUCCESS in 16m 11s (non-voting)\n- keystone-tempest https://zuul.opendev.org/t/openstack/build/c7304a6bbae4489eadaf1c19b3b1f42d : SUCCESS in 12m 57s\n- keystone-tempest-fips https://zuul.opendev.org/t/openstack/build/1d1fd0514c3044b1acce23e2d677fcb2 : FAILURE in 9m 44s (non-voting)\n- keystone-tempest-federation https://zuul.opendev.org/t/openstack/build/227972bc8aa6426dbfc4ddbd8618a067 : FAILURE in 25m 04s (non-voting)\n- keystone-tempest-federation-k2k https://zuul.opendev.org/t/openstack/build/976bf5d924824280b7ef9e36bdafb3dd : SUCCESS in 20m 40s\n- keystone-tempest-oidc-federation https://zuul.opendev.org/t/openstack/build/d3770b9ae801421a876eabb24b575691 : SUCCESS in 14m 42s\n- keystone-tempest-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/b896ae7affb04a07bae3d0a48759f8ba : SUCCESS in 19m 25s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0a6790f2eb7a427aa2f20ad988f2b669 : SUCCESS in 56m 59s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/728ee69cc5cf4275aee610922c097742 : FAILURE in 41m 35s (non-voting)\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/e75226d19a474ff7942798065224dab7 : SUCCESS in 7m 07s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"33ed8fc5aa5326e7a08a1b8326fdb2810019b7e7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-05 01:40:58.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f6934d1f2345488c8d867f06105625e1\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/983a1a70d4fb4d9f9404b223d1e591f6 : SUCCESS in 17m 24s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/fec8bd0cf68746418ac9cbc23537d062 : SUCCESS in 18m 24s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/2f5d912c09354069a3e180732c3ebe45 : SUCCESS in 28m 13s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"b9f3cf260028475d352e7e67f1b2384cba935371","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2025-12-05 09:48:35.000000000","message":"Patch Set 10: Code-Review+2","accounts_in_message":[],"_revision_number":10},{"id":"80c0107cfe3bfe9dc38b456feaed622b5c463b59","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2025-12-05 15:49:11.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"8695f6cc12bfa7f6709688de0a4757e365a2d49c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-05 15:49:51.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"6a2d6964dc08bd72a6723a140fd4d37f8672b141","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-05 18:11:17.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0165607bf5a141e39129113d0b36172f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42c51027e23f401297974c42ad214282 : SUCCESS in 6m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ccb47d37155a45ffb11a281b6351eb64 : SUCCESS in 7m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9df56b2746784b1ba475bf48a0154248 : SUCCESS in 10m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59337d1474884b2e8600f515a15b756b : SUCCESS in 13m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/873ab5ad5fab48f09181a0486b4eb811 : SUCCESS in 49m 26s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cfe80ea564ca4aff9385d63706c4b7eb : SUCCESS in 1h 25m 21s\n- keystone-tempest https://zuul.opendev.org/t/openstack/build/4a72ce533dd1471d841c754bc731b958 : SUCCESS in 25m 14s\n- keystone-tempest-federation-k2k https://zuul.opendev.org/t/openstack/build/3ce7aec9c7764974814b07bb45eeb37d : SUCCESS in 22m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d6a7997326a249d49e583af2b7c141fe : SUCCESS in 52m 17s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/3cfe59c892f14c029b7a7e2a56b462c4 : FAILURE in 36m 37s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"5156b522f7f77d133bf46b7912dd6c24a5beb5d3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-05 18:11:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"2b7e19b1717a9a99738e2446719437082f705cb8","revisions":{"b3b0e94d066f7cd8a33122a8d46dfc37ef2f6daa":{"kind":"REWORK","_number":1,"created":"2024-09-18 07:49:05.000000000","uploader":{"_account_id":29422,"name":"Olivier Chaze","display_name":"Olivier Chaze","email":"olivier.chaze@infomaniak.com","username":"o.chaze"},"ref":"refs/changes/36/929736/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/36/929736/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/36/929736/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/36/929736/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/36/929736/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/36/929736/1"}}},"commit":{"parents":[{"commit":"97431ec99b255f3e04d4b21d38412e5d6b798618","subject":"Merge \"Remove logic to support pysaml2\u003c3.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/97431ec99b255f3e04d4b21d38412e5d6b798618"}]}],"author":{"name":"Olivier Chaze","email":"olivier.chaze@infomaniak.com","date":"2024-09-18 07:47:15.000000000","tz":120},"committer":{"name":"Olivier Chaze","email":"olivier.chaze@infomaniak.com","date":"2024-09-18 07:47:15.000000000","tz":120},"subject":"Index ON revocation_event table to optimize token checks.","message":"Index ON revocation_event table to optimize token checks.\n\nChange-Id: Ic44c945f3cb65b48ff72052fd2b4f6d45e118b2e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/b3b0e94d066f7cd8a33122a8d46dfc37ef2f6daa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/b3b0e94d066f7cd8a33122a8d46dfc37ef2f6daa"}]},"branch":"refs/heads/master"},"8016048b31a73598476722de28eb007d3b750215":{"kind":"REWORK","_number":2,"created":"2024-09-18 09:00:00.000000000","uploader":{"_account_id":29422,"name":"Olivier Chaze","display_name":"Olivier Chaze","email":"olivier.chaze@infomaniak.com","username":"o.chaze"},"ref":"refs/changes/36/929736/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/36/929736/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/36/929736/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/36/929736/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/36/929736/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/36/929736/2"}}},"commit":{"parents":[{"commit":"97431ec99b255f3e04d4b21d38412e5d6b798618","subject":"Merge \"Remove logic to support pysaml2\u003c3.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/97431ec99b255f3e04d4b21d38412e5d6b798618"}]}],"author":{"name":"Olivier Chaze","email":"olivier.chaze@infomaniak.com","date":"2024-09-18 07:47:15.000000000","tz":120},"committer":{"name":"Olivier Chaze","email":"olivier.chaze@infomaniak.com","date":"2024-09-18 08:59:45.000000000","tz":120},"subject":"Index ON revocation_event table to optimize token checks.","message":"Index ON revocation_event table to optimize token checks.\n\nChange-Id: Ic44c945f3cb65b48ff72052fd2b4f6d45e118b2e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/8016048b31a73598476722de28eb007d3b750215"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/8016048b31a73598476722de28eb007d3b750215"}]},"branch":"refs/heads/master"},"f5f54c143c61fb8753d96b9ff9c23741ba893285":{"kind":"REWORK","_number":3,"created":"2024-09-18 15:54:38.000000000","uploader":{"_account_id":29422,"name":"Olivier Chaze","display_name":"Olivier Chaze","email":"olivier.chaze@infomaniak.com","username":"o.chaze"},"ref":"refs/changes/36/929736/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/36/929736/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/36/929736/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/36/929736/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/36/929736/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/36/929736/3"}}},"commit":{"parents":[{"commit":"97431ec99b255f3e04d4b21d38412e5d6b798618","subject":"Merge \"Remove logic to support pysaml2\u003c3.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/97431ec99b255f3e04d4b21d38412e5d6b798618"}]}],"author":{"name":"Olivier Chaze","email":"olivier.chaze@infomaniak.com","date":"2024-09-18 07:47:15.000000000","tz":120},"committer":{"name":"Olivier Chaze","email":"olivier.chaze@infomaniak.com","date":"2024-09-18 15:54:25.000000000","tz":120},"subject":"Index ON revocation_event table to optimize token checks.","message":"Index ON revocation_event table to optimize token checks.\n\nChange-Id: Ic44c945f3cb65b48ff72052fd2b4f6d45e118b2e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/f5f54c143c61fb8753d96b9ff9c23741ba893285"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/f5f54c143c61fb8753d96b9ff9c23741ba893285"}]},"branch":"refs/heads/master"},"f06f93a402777a3b8072f69c4309fd15a3a8012d":{"kind":"REWORK","_number":4,"created":"2025-11-24 09:35:34.000000000","uploader":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"ref":"refs/changes/36/929736/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/36/929736/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/36/929736/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/36/929736/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/36/929736/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/36/929736/4"}}},"commit":{"parents":[{"commit":"19947349904082995b5477d4ec8272bf01f176b2","subject":"Merge \"Bytes-vs-str mismatch causes repeated ID mapping creates on list\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/19947349904082995b5477d4ec8272bf01f176b2"}]}],"author":{"name":"Olivier Chaze","email":"olivier.chaze@infomaniak.com","date":"2024-09-18 07:47:15.000000000","tz":120},"committer":{"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","date":"2025-11-24 09:35:11.000000000","tz":60},"subject":"Index ON revocation_event table to optimize token checks.","message":"Index ON revocation_event table to optimize token checks.\n\nCloses-Bug: #2081082\nChange-Id: Ic44c945f3cb65b48ff72052fd2b4f6d45e118b2e\nSigned-off-by: Quentin GROLLEAU \u003cquentin.grolleau@corp.ovh.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/f06f93a402777a3b8072f69c4309fd15a3a8012d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/f06f93a402777a3b8072f69c4309fd15a3a8012d"}]},"branch":"refs/heads/master"},"1818df7d25cc07321f297e32bdcc059b3e52df80":{"kind":"REWORK","_number":5,"created":"2025-11-24 13:52:13.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/36/929736/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/36/929736/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/36/929736/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/36/929736/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/36/929736/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/36/929736/5"}}},"commit":{"parents":[{"commit":"19947349904082995b5477d4ec8272bf01f176b2","subject":"Merge \"Bytes-vs-str mismatch causes repeated ID mapping creates on list\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/19947349904082995b5477d4ec8272bf01f176b2"}]}],"author":{"name":"Olivier Chaze","email":"olivier.chaze@infomaniak.com","date":"2024-09-18 07:47:15.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2025-11-24 13:51:57.000000000","tz":60},"subject":"Index ON revocation_event table to optimize token checks.","message":"Index ON revocation_event table to optimize token checks.\n\nCloses-Bug: #2081082\nChange-Id: Ic44c945f3cb65b48ff72052fd2b4f6d45e118b2e\nSigned-off-by: Quentin GROLLEAU \u003cquentin.grolleau@corp.ovh.com\u003e\nSigned-off-by: Artem Goncharov \u003cartem.goncharov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/1818df7d25cc07321f297e32bdcc059b3e52df80"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/1818df7d25cc07321f297e32bdcc059b3e52df80"}]},"branch":"refs/heads/master"},"17992c819178dfa272008a8587e92800e0b55820":{"kind":"REWORK","_number":6,"created":"2025-11-24 13:53:16.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/36/929736/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/36/929736/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/36/929736/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/36/929736/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/36/929736/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/36/929736/6"}}},"commit":{"parents":[{"commit":"19947349904082995b5477d4ec8272bf01f176b2","subject":"Merge \"Bytes-vs-str mismatch causes repeated ID mapping creates on list\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/19947349904082995b5477d4ec8272bf01f176b2"}]}],"author":{"name":"Olivier Chaze","email":"olivier.chaze@infomaniak.com","date":"2024-09-18 07:47:15.000000000","tz":120},"committer":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2025-11-24 13:50:48.000000000","tz":60},"subject":"Add a new index on revocation_event table","message":"Add a new index on revocation_event table\n\nWhen validating a token, under load, a full table scan could happen\nbecause of missing indexes.\nThis new index drastically help the database to get through the\nrevocation events.\n\nCloses-Bug: #2081082\nChange-Id: Ic44c945f3cb65b48ff72052fd2b4f6d45e118b2e\nSigned-off-by: Quentin GROLLEAU \u003cquentin.grolleau@corp.ovh.com\u003e\nSigned-off-by: Arnaud Morin \u003carnaud.morin@ovhcloud.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/17992c819178dfa272008a8587e92800e0b55820"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/17992c819178dfa272008a8587e92800e0b55820"}]},"branch":"refs/heads/master"},"b29af84e95f2671bb150542638e0eeabb56a5a60":{"kind":"REWORK","_number":7,"created":"2025-11-25 08:02:27.000000000","uploader":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"ref":"refs/changes/36/929736/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/36/929736/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/36/929736/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/36/929736/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/36/929736/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/36/929736/7"}}},"commit":{"parents":[{"commit":"19947349904082995b5477d4ec8272bf01f176b2","subject":"Merge \"Bytes-vs-str mismatch causes repeated ID mapping creates on list\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/19947349904082995b5477d4ec8272bf01f176b2"}]}],"author":{"name":"Olivier Chaze","email":"olivier.chaze@infomaniak.com","date":"2024-09-18 07:47:15.000000000","tz":120},"committer":{"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","date":"2025-11-25 08:01:37.000000000","tz":60},"subject":"Add a new index on revocation_event table","message":"Add a new index on revocation_event table\n\nWhen validating a token, under load, a full table scan could happen\nbecause of missing indexes.\nThis new index drastically help the database to get through the\nrevocation events.\n\nCloses-Bug: #2081082\nChange-Id: Ic44c945f3cb65b48ff72052fd2b4f6d45e118b2e\nSigned-off-by: Quentin GROLLEAU \u003cquentin.grolleau@corp.ovh.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/b29af84e95f2671bb150542638e0eeabb56a5a60"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/b29af84e95f2671bb150542638e0eeabb56a5a60"}]},"branch":"refs/heads/master"},"3fda04aae4a1bef14647d0ded1e89afe54ad383a":{"kind":"REWORK","_number":8,"created":"2025-11-25 16:57:08.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/36/929736/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/36/929736/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/36/929736/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/36/929736/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/36/929736/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/36/929736/8"}}},"commit":{"parents":[{"commit":"42daa9f271b883319341c0b8ae17b96466604287","subject":"Rename the migration folder","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/42daa9f271b883319341c0b8ae17b96466604287"}]}],"author":{"name":"Olivier Chaze","email":"olivier.chaze@infomaniak.com","date":"2024-09-18 07:47:15.000000000","tz":120},"committer":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2025-11-25 16:56:58.000000000","tz":60},"subject":"Add a new index on revocation_event table","message":"Add a new index on revocation_event table\n\nWhen validating a token, under load, a full table scan could happen\nbecause of missing indexes.\nThis new index drastically help the database to get through the\nrevocation events.\n\nCloses-Bug: #2081082\nChange-Id: Ic44c945f3cb65b48ff72052fd2b4f6d45e118b2e\nSigned-off-by: Quentin GROLLEAU \u003cquentin.grolleau@corp.ovh.com\u003e\nSigned-off-by: Arnaud Morin \u003carnaud.morin@ovhcloud.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/3fda04aae4a1bef14647d0ded1e89afe54ad383a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/3fda04aae4a1bef14647d0ded1e89afe54ad383a"}]},"branch":"refs/heads/master"},"92ad9b18952e573ad46a5112bae4a3c194ce76cf":{"kind":"REWORK","_number":9,"created":"2025-11-25 19:00:10.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/36/929736/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/36/929736/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/36/929736/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/36/929736/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/36/929736/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/36/929736/9"}}},"commit":{"parents":[{"commit":"11079bffc0ea4bb6b0ad3b8ac91893885bb8664a","subject":"Rename the migration folder","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/11079bffc0ea4bb6b0ad3b8ac91893885bb8664a"}]}],"author":{"name":"Olivier Chaze","email":"olivier.chaze@infomaniak.com","date":"2024-09-18 07:47:15.000000000","tz":120},"committer":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2025-11-25 18:59:59.000000000","tz":60},"subject":"Add a new index on revocation_event table","message":"Add a new index on revocation_event table\n\nWhen validating a token, under load, a full table scan could happen\nbecause of missing indexes.\nThis new index drastically help the database to get through the\nrevocation events.\n\nCloses-Bug: #2081082\nChange-Id: Ic44c945f3cb65b48ff72052fd2b4f6d45e118b2e\nSigned-off-by: Quentin GROLLEAU \u003cquentin.grolleau@corp.ovh.com\u003e\nSigned-off-by: Arnaud Morin \u003carnaud.morin@ovhcloud.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/92ad9b18952e573ad46a5112bae4a3c194ce76cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/92ad9b18952e573ad46a5112bae4a3c194ce76cf"}]},"branch":"refs/heads/master"},"2b7e19b1717a9a99738e2446719437082f705cb8":{"kind":"REWORK","_number":10,"created":"2025-12-04 22:20:48.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/36/929736/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/36/929736/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/36/929736/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/36/929736/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/36/929736/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/36/929736/10"}}},"commit":{"parents":[{"commit":"11079bffc0ea4bb6b0ad3b8ac91893885bb8664a","subject":"Rename the migration folder","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/11079bffc0ea4bb6b0ad3b8ac91893885bb8664a"}]}],"author":{"name":"Olivier Chaze","email":"olivier.chaze@infomaniak.com","date":"2024-09-18 07:47:15.000000000","tz":120},"committer":{"name":"Arnaud M.","email":"arnaud.morin@gmail.com","date":"2025-12-04 22:19:12.000000000","tz":60},"subject":"Add a new index on revocation_event table","message":"Add a new index on revocation_event table\n\nWhen validating a token, under load, a full table scan could happen\nbecause of missing indexes.\nThis new index drastically help the database to get through the\nrevocation events.\n\nCloses-Bug: #2081082\nChange-Id: Ic44c945f3cb65b48ff72052fd2b4f6d45e118b2e\nSigned-off-by: Quentin GROLLEAU \u003cquentin.grolleau@corp.ovh.com\u003e\nSigned-off-by: Arnaud Morin \u003carnaud.morin@ovhcloud.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/2b7e19b1717a9a99738e2446719437082f705cb8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/2b7e19b1717a9a99738e2446719437082f705cb8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
