)]}'
{"id":"openstack%2Fkeystone~930827","triplet_id":"openstack%2Fkeystone~master~I2e2e6fa12cc8280e1f6a5825764af0b2b48835db","project":"openstack/keystone","branch":"master","attention_set":{},"removed_from_attention_set":{"7973":{"account":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"last_update":"2024-11-15 15:50:59.000000000","reason":"Change was submitted"},"7414":{"account":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"last_update":"2024-11-15 15:10:25.000000000","reason":"\u003cGERRIT_ACCOUNT_7414\u003e replied on the change","reason_account":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-11-15 15:50:59.000000000","reason":"Change was submitted"},"14250":{"account":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"last_update":"2024-10-31 13:29:18.000000000","reason":"\u003cGERRIT_ACCOUNT_14250\u003e replied on the change","reason_account":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}}},"hashtags":[],"change_id":"I2e2e6fa12cc8280e1f6a5825764af0b2b48835db","subject":"Remove direct dependency on memcache","status":"MERGED","created":"2024-09-29 07:37:16.000000000","updated":"2024-11-15 15:52:53.000000000","submitted":"2024-11-15 15:50:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"930827","meta_rev_id":"afcdfb883fbf40d1867761d0f024e7b38028578c","_number":930827,"virtual_id_number":930827,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-15 15:50:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"all":[{"value":2,"date":"2024-11-15 15:10:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":2,"date":"2024-10-31 13:29:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"all":[{"value":1,"date":"2024-11-15 15:10:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"}],"reviewers":{"REVIEWER":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-29 18:35:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-09-29 19:18:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-31 03:16:32.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"},{"updated":"2024-10-31 03:16:32.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"state":"REVIEWER"},{"updated":"2024-10-31 03:16:32.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"}],"messages":[{"id":"a8ac22514b6805037556b7eb42c4e342126271ce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-09-29 07:37:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0a5646e16ba8d20962a0303318cd9e81b0f240af","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-29 18:35:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cdf5ee0e769b411b90a1c9648ee976c1\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b8543dad872242e8b9d87e7a694785b6 : TIMED_OUT in 1h 00m 50s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a881a5b0e881426cb5d650b2a9e9dcfc : TIMED_OUT in 1h 00m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0a9170fcf469d561a04ce84c8a9e3b1831af9aba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-29 19:18:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5af67706183348b89510fef6f8b9b1f4\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b160b70e69c5403d91f7f37dc51afbbc : SUCCESS in 1m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/29e6c06aa82c4cb69eb353c756ad4972 : SUCCESS in 11m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b33aad2d287e4d03a6f0f041ee6c6652 : SUCCESS in 8m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e301b3292e4d452ca0f3741d8b618f4b : SUCCESS in 16m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4b1972e60b1d4a359f5b939ecb813f21 : SUCCESS in 15m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12cc312360d344f791248c0bf2c80fd2 : SUCCESS in 12m 48s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/a14dfe4033fa404dbfd7a34257cde49b : SUCCESS in 20m 38s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/56cc9d43b6524a8187fd16381b8eef29 : SUCCESS in 44m 02s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5130a7bdec1c479cb38ddf36f6ad5c81 : FAILURE in 17m 31s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/f4ef9b1ade0b4781affb7149ccea65ea : SUCCESS in 30m 48s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/b8285ace25984af5b7118504ea8df26f : SUCCESS in 10m 35s (non-voting)\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/a734acc6424b4587a8a9448a50383871 : FAILURE in 42m 53s (non-voting)\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/e9ab634b27444693824f620fe8bf617e : SUCCESS in 6m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6f95fcbbc260c393caa3f2ce803a3cf89fd569d7","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2024-10-31 13:29:18.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"71bf1701ace1ae2a899ca34c0804447a6b2f83d3","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2024-11-15 15:10:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c4789e92231885037dc302b1a9a06b42d6a16fa9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 15:10:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"54ae12e987b087f3cb35324d6388e60ed1574e3d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 15:50:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9f6bc0d49af546b2b59b3e005320a687\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/5c1335662974435aaedfa97704a05ea4 : SUCCESS in 1m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57e2c3cafeef4851be98f7fca7db35d7 : SUCCESS in 4m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1a6460f2e0b14d6bb3351e6c2525afcd : SUCCESS in 11m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/640f746ff30143d8b5c48f1a123eb4c5 : SUCCESS in 9m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f4fd0bb31d543a9beaa3edf18f404f5 : SUCCESS in 12m 05s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/d4c5bcb4b56b4533b5a6362448a46dda : SUCCESS in 16m 38s\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/b16e9325c8164272a9ba497fb2c1bdd9 : SUCCESS in 29m 23s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/c6dca84bd13742f78b44162a9ff102e9 : SUCCESS in 31m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d19deb2d747810099bfd0496898ec34fcd6cd4d6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 15:50:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"afcdfb883fbf40d1867761d0f024e7b38028578c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 15:52:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ba924cc088146688418151b6a736cc2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/033dc09536364acd9b6557b207d1c67a : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a59ac9d823a93a945726ec75583222e18368091f","revisions":{"a59ac9d823a93a945726ec75583222e18368091f":{"kind":"REWORK","_number":1,"created":"2024-09-29 07:37:16.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/27/930827/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/27/930827/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/27/930827/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/27/930827/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/27/930827/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/27/930827/1"}}},"commit":{"parents":[{"commit":"97431ec99b255f3e04d4b21d38412e5d6b798618","subject":"Merge \"Remove logic to support pysaml2\u003c3.0.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/97431ec99b255f3e04d4b21d38412e5d6b798618"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-09-29 07:35:39.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-09-29 07:36:50.000000000","tz":540},"subject":"Remove direct dependency on memcache","message":"Remove direct dependency on memcache\n\npython-memcached was used by memcache token persistence which was\nalready removed, is no longer directly used by keystone and it\u0027s\nrequired via oslo.cache .\n\nChange-Id: I2e2e6fa12cc8280e1f6a5825764af0b2b48835db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/a59ac9d823a93a945726ec75583222e18368091f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/a59ac9d823a93a945726ec75583222e18368091f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
