)]}'
{"id":"openstack%2Fkeystone~932512","triplet_id":"openstack%2Fkeystone~master~Ie48bff1c7b9cf1d580f2dc08f35b46e4ce8e05c1","project":"openstack/keystone","branch":"master","hashtags":[],"change_id":"Ie48bff1c7b9cf1d580f2dc08f35b46e4ce8e05c1","subject":"Re-initiate OSA rolling upgrade tests","status":"NEW","created":"2024-10-16 14:11:10.000000000","updated":"2024-10-17 16:42:35.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":2,"unresolved_comment_count":1,"work_in_progress":true,"has_review_started":true,"meta_rev_id":"e64217b0696457d653f49c9d16c63dbe562134d6","_number":932512,"virtual_id_number":932512,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-10-16 17:19:16.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2024-10-16 15:14:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-10-16 17:19:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"300d8cd8e3739c41099d65f294b000ae6045a6a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-10-16 14:11:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c9c15451f5f54762d9f17898a00ae9f7a9e725d9","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-10-16 14:15:38.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"c54608f074e548cb336e96127029bec32bfbc194","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-16 15:14:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c395d53949954aa7b0cb3dc1642d3868\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/57969d11732948f6b318f531be8a26a6 : TIMED_OUT in 1h 01m 07s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5a48360f22f8493098eb83cecc9cdf58 : TIMED_OUT in 1h 01m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0a3260b18018eb791250b2754d1c28e967d7048e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-10-16 15:30:12.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"794ae3e16209b2d8df920932b8a3e6aac92bf824","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-16 16:38:08.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c7bd6fee85c34d59b6eebd203e805d20\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2f903ecbf94245d5a2ed809f57609ca1 : TIMED_OUT in 1h 01m 20s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/568c4499fd65491ba15d1fd2caece8e7 : TIMED_OUT in 1h 01m 01s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"144335fd2914d0964b301d1faf1c9cde11291aa5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-16 17:19:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fc044c5a879f4117a17dd844f3a4425a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d121981771b74bbcb6622d3b21c0db8a : SUCCESS in 12m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7924eafe65ae4daa915b817a4ca013e9 : SUCCESS in 7m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cce9757579014e8995bc2def71cea9dc : SUCCESS in 12m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5d06390bf15a4c1ebf27196179b24042 : SUCCESS in 12m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8cb49d78c0624d7b861dbb6ec2c96c75 : SUCCESS in 12m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/2c314238c3da4b0f9672722bb011ff82 : SUCCESS in 47m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7f4ebdcf7b2041858c6921a6cfc661e7 : SUCCESS in 1h 37m 59s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/749409b4be034235a6b744aee1e4b90d : SUCCESS in 20m 38s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/33201da1af3a4fdebe7808d624cc733d : SUCCESS in 31m 15s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4f3842df32124abfb469c4bf48d17a25 : FAILURE in 31m 59s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/f5f7df3c341449c09312841dc6ed3f50 : SUCCESS in 18m 01s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/9abdbef2c77c4cdca6c2ed27d7ea8d0d : SUCCESS in 24m 00s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/ca16fe8e2edb46a0a264f2319d6fbb09 : FAILURE in 23m 44s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cd0db1ee305d46a88cb3622594c647e2 : SUCCESS in 1h 02m 43s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/30819b4dbde14e57968e6156c966951d : SUCCESS in 38m 16s (non-voting)\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/96b6d94f69054c0c8688536b8dcc1cf8 : SUCCESS in 7m 31s (non-voting)\n- openstack-ansible-upgrade-infra_lxc-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5fd324db9add4e1b98542ee72b27ad31 : SUCCESS in 1h 44m 59s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"492012874b7b21a0ba4799ba08584539bd9110f1","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-10-17 08:20:45.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e64217b0696457d653f49c9d16c63dbe562134d6","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2024-10-17 16:42:35.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"382d65aac75a8410f892b706399b3a678d275c6c","revisions":{"0b2f25eb4620efe2a057e917597a8920d0ff7652":{"kind":"REWORK","_number":1,"created":"2024-10-16 14:11:10.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/12/932512/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/12/932512/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/12/932512/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/12/932512/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/12/932512/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/12/932512/1"}}},"commit":{"parents":[{"commit":"6433f1ae1c094e58216360d191a21d956a80f658","subject":"Merge \"Complete getting rid of passlib\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/6433f1ae1c094e58216360d191a21d956a80f658"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-10-16 14:11:00.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-10-16 14:11:00.000000000","tz":120},"subject":"Re-initiate OSA rolling upgrade tests","message":"Re-initiate OSA rolling upgrade tests\n\nChange-Id: Ie48bff1c7b9cf1d580f2dc08f35b46e4ce8e05c1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/0b2f25eb4620efe2a057e917597a8920d0ff7652"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/0b2f25eb4620efe2a057e917597a8920d0ff7652"}]},"branch":"refs/heads/master"},"382d65aac75a8410f892b706399b3a678d275c6c":{"kind":"REWORK","_number":2,"created":"2024-10-16 15:30:12.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/12/932512/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/12/932512/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/12/932512/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/12/932512/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/12/932512/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/12/932512/2"}}},"commit":{"parents":[{"commit":"6433f1ae1c094e58216360d191a21d956a80f658","subject":"Merge \"Complete getting rid of passlib\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/6433f1ae1c094e58216360d191a21d956a80f658"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-10-16 14:11:00.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-10-16 15:30:12.000000000","tz":0},"subject":"Re-initiate OSA rolling upgrade tests","message":"Re-initiate OSA rolling upgrade tests\n\nChange-Id: Ie48bff1c7b9cf1d580f2dc08f35b46e4ce8e05c1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/382d65aac75a8410f892b706399b3a678d275c6c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/382d65aac75a8410f892b706399b3a678d275c6c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
