)]}'
{"id":"openstack%2Fkeystone~933189","triplet_id":"openstack%2Fkeystone~master~Ieea676b76506880dcc6b7437b01f538c4dde2bcf","project":"openstack/keystone","branch":"master","topic":"pre-commit-skip-functional","attention_set":{},"removed_from_attention_set":{"7973":{"account":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"last_update":"2025-03-28 14:34:24.000000000","reason":"\u003cGERRIT_ACCOUNT_7973\u003e replied on the change","reason_account":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-03-28 14:34:51.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_37598\u003e by clicking the attention icon","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}},"hashtags":[],"change_id":"Ieea676b76506880dcc6b7437b01f538c4dde2bcf","subject":"Skip functional tests on pre-commit config update","status":"MERGED","created":"2024-10-24 01:20:29.000000000","updated":"2025-03-28 18:08:08.000000000","submitted":"2025-03-28 18:03:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"933189-pre-commit-skip-functional","meta_rev_id":"57395001985328d7f537898cb841b31515345035","_number":933189,"virtual_id_number":933189,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-28 18:03:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"all":[{"value":2,"date":"2025-03-28 14:34:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":2,"date":"2025-03-28 14:27:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-03-28 14:28:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"all":[{"value":1,"date":"2025-03-28 14:34:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-24 02:47:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-10-24 03:34:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-28 14:27:27.000000000","updated_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"},{"updated":"2025-03-28 14:28:18.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2025-03-28 14:28:18.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"state":"REVIEWER"}],"messages":[{"id":"eaffb6c635653cbbd1d90382af648b47097d14fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-24 01:20:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"edc10d36e639e3aec81fb28d86e95d41a920c32a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-24 01:25:27.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"6967189e559f08d12e7c55c070e1dda8dcc5b8e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-24 01:43:18.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"75e441089306fbb09ddb2cfb6bd5cdc4713e58d1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-24 02:47:20.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/81372e527b6e495caa543ab78585745e\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2b43f23261ff47ab827de21accedd28d : TIMED_OUT in 1h 01m 00s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/36f56bc6bffd40b39fb3dd650cb4381f : TIMED_OUT in 1h 01m 02s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"68ace05448f2b4f736b00a172f89d6a947910ccb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-24 03:34:05.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/18fc9eefe4534a6d9693d10ce24f9440\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/58e71f36af2c42c0908829d4d5928f59 : SUCCESS in 15m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91585cd1a727437d8f198f2ebe6c5a34 : SUCCESS in 5m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ad652405c5a344d4b0db4aab8e9b1b36 : SUCCESS in 11m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/261d697fbd574065adc31e78160a224f : SUCCESS in 11m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e759c0728da04685be66af2635ae4385 : SUCCESS in 12m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/4916b6b061a040048a7926da9d67d4eb : SUCCESS in 1h 06m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bbeee9fe8b724d10b6140f5f47239fa5 : SUCCESS in 1h 49m 46s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/a71832fb6b644f3d9c72159515a36399 : SUCCESS in 30m 31s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/39fe9b6f9d224b1db3059bc381ececa6 : SUCCESS in 41m 00s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f405079f105d4615b490c5f9d0475939 : FAILURE in 29m 32s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/4ecf1246cc0448b8bff0f146405a0f59 : SUCCESS in 33m 00s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/3bc87f9eee874514a2b0c13eb3c87446 : SUCCESS in 18m 40s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/9e31dba0faec482dac7a5a822db70fd7 : FAILURE in 15m 48s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c0c6734ddc69445291d0917ceb1a8757 : SUCCESS in 37m 48s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/7fabf40f7cbe446f9e07dd90f6c2bc28 : SUCCESS in 35m 36s (non-voting)\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/b643d2c5a7b74b4595dadf85f331cf2e : SUCCESS in 7m 24s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"15e736a51affdabe553f8d7e41b3cddd9881c6da","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2025-03-28 14:27:27.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"3118b2d58f0937cd5999ae880721a97895251468","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-03-28 14:28:18.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"16ff0e5c99ccd5d1074aa2cbfd7ab0628a320ef3","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2025-03-28 14:34:24.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"2ffa03dfb76699fc00a1416e3b641fa191bf30a8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-28 14:35:25.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"285d85b44af18a6378a1ef8aaa0b9f2eea01246c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-28 18:03:41.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/44cd946cb0c94a9eb316f32cfb38960a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c47e232f32449589a68458abdf7fe6e : SUCCESS in 9m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ca4fb20826594721a301c1823e5dc60c : SUCCESS in 12m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/443382c39ffb4defbf4dbbbea679f730 : SUCCESS in 7m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b261ac967a7543818e446d9d88cafad5 : SUCCESS in 13m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/07c794613f9e40a2992418a87cae3cb1 : SUCCESS in 26m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d939749733f34540a828d73867c11ea1 : SUCCESS in 1h 44m 13s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/1bc1c57b884f4aac909135704587e160 : SUCCESS in 25m 34s\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/b7cfb26078924763a556ab7f730b2b4e : SUCCESS in 18m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/356421bc02704e20b885ae33e12b1101 : SUCCESS in 45m 08s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/23a04f66baed4db6973b888db5b0fd1e : FAILURE in 39m 01s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"544214d089b80b9055c13e45a72f4fcf46e72260","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-28 18:03:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"57395001985328d7f537898cb841b31515345035","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-28 18:08:08.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/96c48369ff6c4103a4a415448627003b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/780b64fe175c4d88a53a504bd4f02ba5 : SUCCESS in 1m 14s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"11681e229778ec3e929458256eb4aa4eb7cdc5b9","revisions":{"444585467d99a509ac75f5af9e226459e57bc48d":{"kind":"REWORK","_number":1,"created":"2024-10-24 01:20:29.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/89/933189/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/89/933189/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/89/933189/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/89/933189/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/89/933189/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/89/933189/1"}}},"commit":{"parents":[{"commit":"e2b174a93337424b8475fa957c41954968919dce","subject":"Merge \"Switch to cryptography for pbkdf2_sha512 password hashing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/e2b174a93337424b8475fa957c41954968919dce"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-24 01:19:54.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-24 01:20:19.000000000","tz":540},"subject":"Skip functional tests on pre-commit-config.yaml update","message":"Skip functional tests on pre-commit-config.yaml update\n\nThis file is used only in pep8 .\n\nChange-Id: Ieea676b76506880dcc6b7437b01f538c4dde2bcf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/444585467d99a509ac75f5af9e226459e57bc48d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/444585467d99a509ac75f5af9e226459e57bc48d"}]},"branch":"refs/heads/master"},"8689d7c9fe746c40a1ebfdb2a56d4b85259d3919":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-10-24 01:25:27.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/89/933189/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/89/933189/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/89/933189/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/89/933189/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/89/933189/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/89/933189/2"}}},"commit":{"parents":[{"commit":"e2b174a93337424b8475fa957c41954968919dce","subject":"Merge \"Switch to cryptography for pbkdf2_sha512 password hashing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/e2b174a93337424b8475fa957c41954968919dce"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-24 01:19:54.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-24 01:25:27.000000000","tz":0},"subject":"Skip functional tests on pre-commit config update","message":"Skip functional tests on pre-commit config update\n\nThis file is used only in pep8 .\n\nChange-Id: Ieea676b76506880dcc6b7437b01f538c4dde2bcf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/8689d7c9fe746c40a1ebfdb2a56d4b85259d3919"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/8689d7c9fe746c40a1ebfdb2a56d4b85259d3919"}]},"branch":"refs/heads/master","description":"Edit commit message"},"11681e229778ec3e929458256eb4aa4eb7cdc5b9":{"kind":"REWORK","_number":3,"created":"2024-10-24 01:43:18.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/89/933189/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/89/933189/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/89/933189/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/89/933189/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/89/933189/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/89/933189/3"}}},"commit":{"parents":[{"commit":"e2b174a93337424b8475fa957c41954968919dce","subject":"Merge \"Switch to cryptography for pbkdf2_sha512 password hashing\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/e2b174a93337424b8475fa957c41954968919dce"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-24 01:19:54.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-24 01:42:56.000000000","tz":540},"subject":"Skip functional tests on pre-commit config update","message":"Skip functional tests on pre-commit config update\n\nThis file is used only in code style checks.\n\nChange-Id: Ieea676b76506880dcc6b7437b01f538c4dde2bcf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/11681e229778ec3e929458256eb4aa4eb7cdc5b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/11681e229778ec3e929458256eb4aa4eb7cdc5b9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
