)]}'
{"id":"openstack%2Fkeystone~933292","triplet_id":"openstack%2Fkeystone~master~Ie71dd98911c740b2b3d87157c326eedc40f9f17d","project":"openstack/keystone","branch":"master","topic":"remove-py38","attention_set":{},"removed_from_attention_set":{"7973":{"account":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"last_update":"2024-11-15 15:50:55.000000000","reason":"Change was submitted"},"7414":{"account":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"last_update":"2024-11-15 15:03:27.000000000","reason":"\u003cGERRIT_ACCOUNT_7414\u003e replied on the change","reason_account":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-11-15 15:50:55.000000000","reason":"Change was submitted"},"14250":{"account":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"last_update":"2024-10-31 13:30:07.000000000","reason":"\u003cGERRIT_ACCOUNT_14250\u003e replied on the change","reason_account":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}}},"hashtags":[],"change_id":"Ie71dd98911c740b2b3d87157c326eedc40f9f17d","subject":"Remove Python 3.8 support","status":"MERGED","created":"2024-10-24 12:30:14.000000000","updated":"2024-11-15 15:52:04.000000000","submitted":"2024-11-15 15:50:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"933292-remove-py38","meta_rev_id":"be8f3aa729fd9fd4a58310b311e310f2c4f3922b","_number":933292,"virtual_id_number":933292,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-15 15:50:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"all":[{"value":2,"date":"2024-11-15 15:03:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":2,"date":"2024-10-31 13:30:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"all":[{"value":1,"date":"2024-11-15 15:03:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"}],"reviewers":{"REVIEWER":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-24 13:07:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-31 03:17:16.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"},{"updated":"2024-10-31 03:17:16.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"state":"REVIEWER"},{"updated":"2024-10-31 03:17:16.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"}],"messages":[{"id":"0412091b88f11ca218c0a4a2121d41fff10768aa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-24 12:30:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0f707e0f1a6e355ba7fe001e922a6367d4b04fac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-24 13:07:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e550d1d6bfe403f89d44f3673a32d7e\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/231e7565727f4150bbee46c18380cd37 : SUCCESS in 2m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2defd402d3054c1292c81e2296a32136 : SUCCESS in 14m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6988fb94e7214d7b9f3e6662f545b854 : SUCCESS in 5m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4f45f2a66a81406281f1741e08583879 : SUCCESS in 11m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/aec4497af7754927a0cac60efb9fe4f0 : SUCCESS in 12m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68100849d4c64e48ad9164f70facf9a0 : SUCCESS in 12m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b546380c676040d984dd6da1882d5394 : SUCCESS in 13m 03s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/9feca97660b642a9859009f14cdfbaa0 : SUCCESS in 30m 21s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/da36e065bf6a4af8bed52d217ad4a714 : SUCCESS in 26m 59s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/15a9b71fda53448d9159552fb24a81d2 : FAILURE in 30m 33s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/660e9a02e34243f285b8a6563720efbb : SUCCESS in 30m 06s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/f6cd2630aad342d798327ce5f3b14363 : SUCCESS in 9m 39s (non-voting)\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/3d4527db70df4e02b28fc5ba1bc990cb : SUCCESS in 26m 17s (non-voting)\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/4fb221486a4c4eac9ebb0a37818cba6b : SUCCESS in 6m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e539c8720ca3a6c5b0508d5bf01034068eb52ae3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-24 14:03:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f749b3413edc4fddbbddd1d52ea52e42\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e7675181558f448b84c833ce6f0484c2 : TIMED_OUT in 1h 00m 52s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2c7f49402c88490b8d3de23d62013fb6 : TIMED_OUT in 1h 01m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b7a2889f7c8c6ae3f17dcc5efc2467ba6b5e8509","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2024-10-31 13:30:07.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d052c568dabaac95329c6575f93feff35475cbfc","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2024-11-15 15:03:27.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"622d5ce6e77877a51e4e32e00b697613b9450e8c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 15:04:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f13595d26feab4adddf41642dee1df586623a674","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 15:50:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a0cd54e940e4d2bba34ad21603a4b79\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/67578d35888e495d9925d21f62ddca9b : SUCCESS in 2m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/497608072e6e46cbb0f00890e48ccc23 : SUCCESS in 6m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9f53f4334e6d4a468fbc0692d02ddce9 : SUCCESS in 12m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/17e51b858c5746e1aae9136b2591f5b5 : SUCCESS in 13m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b96f55834e6741bd9f7afe59e9229bd6 : SUCCESS in 11m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9a0415071b3b4785b23a5f7800e5d989 : SUCCESS in 9m 15s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/dc5b2bab53184078aec346327e5566cc : SUCCESS in 33m 50s\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/e1897939f3b949f88a7f3597d249f88f : SUCCESS in 26m 57s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/824c82300fcc4c279b49bfc57f38dc4f : SUCCESS in 41m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"305cdb7d223695599eac7b9a5b79288dba175c80","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 15:50:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"be8f3aa729fd9fd4a58310b311e310f2c4f3922b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-15 15:52:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0843bbf5f23d47928f45a5668a8a05ac\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/afa7b911b3cf4109a9ce1293e89b9ae7 : SUCCESS in 49s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/825551c3cecc4757b30ce9935ece3821 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e1b4edb146aa8e0d41a2b620b8131a5352fda867","revisions":{"e1b4edb146aa8e0d41a2b620b8131a5352fda867":{"kind":"REWORK","_number":1,"created":"2024-10-24 12:30:14.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/92/933292/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/92/933292/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/92/933292/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/92/933292/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/92/933292/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/92/933292/1"}}},"commit":{"parents":[{"commit":"6433f1ae1c094e58216360d191a21d956a80f658","subject":"Merge \"Complete getting rid of passlib\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/6433f1ae1c094e58216360d191a21d956a80f658"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-24 12:30:05.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-24 12:30:05.000000000","tz":540},"subject":"Remove Python 3.8 support","message":"Remove Python 3.8 support\n\nPython 3.8 was removed from the tested runtimes for 2024.2[1] and has\nnot been tested since then.\n\nAlso add Python 3.12 which is part of the tested runtimes for 2025.1.\nNow unit tests job with Python 3.12 is voting.\n\n[1] https://governance.openstack.org/tc/reference/runtimes/2024.2.html\n\nChange-Id: Ie71dd98911c740b2b3d87157c326eedc40f9f17d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/e1b4edb146aa8e0d41a2b620b8131a5352fda867"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/e1b4edb146aa8e0d41a2b620b8131a5352fda867"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
