)]}'
{"id":"openstack%2Fkeystone~935685","triplet_id":"openstack%2Fkeystone~master~Ie5fc80d7e5506c8c426a289687acc6c7c30f5ebd","project":"openstack/keystone","branch":"master","topic":"diag","attention_set":{},"removed_from_attention_set":{"27900":{"account":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"last_update":"2024-11-20 22:02:03.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie5fc80d7e5506c8c426a289687acc6c7c30f5ebd","subject":"Replace sphinxcontrib-*diag","status":"MERGED","created":"2024-11-19 17:06:24.000000000","updated":"2024-11-20 22:03:23.000000000","submitted":"2024-11-20 22:02:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"935685-diag","meta_rev_id":"dac7bd0b0e7db0136352a7f1f9e6298fed296479","_number":935685,"virtual_id_number":935685,"owner":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-20 22:02:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"all":[{"value":2,"date":"2024-11-20 15:28:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":1,"date":"2024-11-20 04:40:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2024-11-20 20:51:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"all":[{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2024-11-20 20:51:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-19 19:26:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-20 04:40:36.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2024-11-20 15:28:41.000000000","updated_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"},{"updated":"2024-11-20 20:51:23.000000000","updated_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"}],"messages":[{"id":"aab5edcc1da8f033097f011834209038848cfc7d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-11-19 17:06:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"511a463407d7e79037e5f87a0d75f7e776b65e7d","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-11-19 17:07:38.000000000","message":"Topic diag removed","accounts_in_message":[],"_revision_number":1},{"id":"47d0602462408ce2885c3685c8c46468d4477421","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-11-19 17:07:41.000000000","message":"Topic set to remove-blockdiag","accounts_in_message":[],"_revision_number":1},{"id":"e3f1857156091141c6462bc399e6bb2df10bcfa2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 19:26:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/87af74d5e5eb4d8c943205578eb735cc\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/2acb29ba3b864652a18c33c3d191e2a0 : SUCCESS in 4m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2f8af39487c44351ae3b5b76b00a542d : SUCCESS in 13m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b7181bbbd0b4a67bd4b68edc10be2ac : SUCCESS in 5m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ac4c4c0d3640402ab35d1882ddc25bfa : SUCCESS in 12m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ffe84445879945e8a6e07f894641811d : SUCCESS in 10m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3534d041d09b40b8805a29b1e1a41147 : SUCCESS in 13m 17s\n- requirements-check https://zuul.opendev.org/t/openstack/build/3fb8682e4cd242bcb800082de770f086 : SUCCESS in 2m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/6eaf3bc6f3f74ac1a9602797f3ee1213 : SUCCESS in 1h 14m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8869da9aadf5424b91096d8f6bde60d5 : SUCCESS in 1h 54m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/31de72593ca546368527d77667f4cc49 : SUCCESS in 13m 24s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/4ce1acb6b56f40f7a903be553db79d7c : SUCCESS in 27m 48s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/dab9c0c2d5974ecf8da45bb4412cea29 : SUCCESS in 43m 33s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/de6a4c0c97dd49d792a66c52d6e98280 : FAILURE in 27m 58s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/12d8ac4474b747e3bdddde2d951a0b6c : SUCCESS in 26m 54s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/b1d07d6b81de47c1929a8481e8c2ba0b : SUCCESS in 12m 50s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/dcc32f764f914728bf4519759c6946ab : FAILURE in 15m 46s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ebc7aff866ab49e981c22894f6493aad : SUCCESS in 1h 06m 04s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/5f66ce3dc2d64497a38f81b071f98b38 : SUCCESS in 52m 16s (non-voting)\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/45140235b3e840f8beced33c4e6cb9d9 : SUCCESS in 6m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0b1be7f0e33e6a261e5fdcc9482b9aa733a4856e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-11-19 19:46:34.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"2b5761f6bbd1e52a129a2155571b344bf54724ac","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 21:29:53.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/07bc23a6eaea496992d703d5b6e510c6\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f0555bb32644417795a62206696fcf69 : TIMED_OUT in 1h 00m 56s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a58cd946898941ada28a1ec7c27285f9 : TIMED_OUT in 1h 00m 56s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a0baeb69324f843e2abf1d5dbbd31a0d1800c73c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-19 21:39:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/45d40756216e4e7caf4ee91759270bd7\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/3072501744f44a728b3e2eca72851758 : SUCCESS in 4m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0907da60aebd42128977aaa9b7886edd : SUCCESS in 13m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5bddc707d2d44deb31f42a1aaeb7dfe : SUCCESS in 3m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c713f271ede945bf9232b40005420cb9 : SUCCESS in 12m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f3cc29a3d965412eb7762d4e5446a15e : SUCCESS in 9m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea9090cc230f458f8b0659e3626c4a38 : SUCCESS in 10m 09s\n- requirements-check https://zuul.opendev.org/t/openstack/build/21f08967995c4eea9b02f2aebefab7b8 : SUCCESS in 2m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/ba779543e1324f16a72e16282a52d40a : SUCCESS in 1h 05m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/620f50161927465fa70dec01360fa46c : SUCCESS in 1h 45m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/381e646d7abf4638b6b36bc4b69d97e8 : SUCCESS in 12m 15s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/61beb5395e4e42c69e6bccd12b55d9cb : SUCCESS in 30m 58s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/8226948836fa48479f8e7efaa5035b77 : SUCCESS in 52m 39s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9cdec72c32e9476988798d8bdbbd002c : FAILURE in 37m 53s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/3cd1b5a44c464bc48d571f3a6b68bb6b : SUCCESS in 33m 38s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/02a455f988fa47f3b359b5b9f7c03780 : SUCCESS in 30m 53s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/ec475d17fa4b4629b0f5be821e0543cd : FAILURE in 22m 51s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0991966e0f83438b831e8b77d2dd2605 : SUCCESS in 1h 02m 59s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/f2652c91203542739107a78a56b6837f : SUCCESS in 44m 50s (non-voting)\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/cb62df1e47af40c28ec954fe929fdf86 : SUCCESS in 6m 50s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"bd8abba540b05ae372bb8f21d5b59ca329330ac4","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-20 04:40:36.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"0c4b74e54522b1a4722c5a437a2d5013173d2d21","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-20 04:42:53.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"968220699f2ce3ed74efe7a7cc61f7bbee34973c","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2024-11-20 15:28:41.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"db4418d9f61e7034cde13446ca414ffaffe464ff","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2024-11-20 20:51:23.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"74c5ba475bb42b5710ad893bb377402810c78a5f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 20:51:54.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"79fb69c61bafd6f17ac6d9aa1620c45fac053a97","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 22:02:03.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/767d12b21ecc4022ac7bf5c4dba8b503\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/bb26c750f6dc44cca9a38da7a929ae30 : SUCCESS in 3m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45d6a99039084c50b36a53d931020fed : SUCCESS in 4m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/acbaa8fa5f074b9987ccf033b8bae9ad : SUCCESS in 13m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/af3ce7a77a284da6b720a43545819fde : SUCCESS in 10m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d471f8a526f54646816da5d9cffcc2b3 : SUCCESS in 6m 38s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a39a0d55cfa24d95a6f4e7d01eea5dc7 : SUCCESS in 3m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/f0e52f8394c1447d9791424ba6758171 : SUCCESS in 1h 04m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/47dd8eec9d494fe39a281deabea38239 : SUCCESS in 56m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f657e4cbf6b04f27921ea6f0cd30f842 : SUCCESS in 10m 30s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/26f0db6a7bd546a5928681514c0ae182 : SUCCESS in 15m 21s\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/1ad73643b4df48ddab53f61d3aade377 : SUCCESS in 32m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/28e1974f594541d39aa30addb6a9ff51 : SUCCESS in 41m 04s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/f20182a3ee4f47b98cedf3eed3bc256c : SUCCESS in 41m 59s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"50127c89dfebdc78fb42d0a7b040e1f1602769af","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 22:02:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"dac7bd0b0e7db0136352a7f1f9e6298fed296479","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 22:03:23.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/90b703e136674802bcf79701ce59f0a9\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/80213356a03c4c4199addc1af964db56 : SUCCESS in 39s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f700d5f3b0f426e8986811296718872 : SUCCESS in 57s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cc062ab06cb9459f874f3c1b35ad3111 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"51045cc0dd1622fe03be78a0e2b273e4c975106e","revisions":{"54f08920f46f550054d54248bf4f016779fad588":{"kind":"REWORK","_number":1,"created":"2024-11-19 17:06:24.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/85/935685/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/85/935685/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/85/935685/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/85/935685/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/85/935685/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/85/935685/1"}}},"commit":{"parents":[{"commit":"15f34eff17f14074bdadc558e6fe80308e371476","subject":"Merge \"Ruff the code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/15f34eff17f14074bdadc558e6fe80308e371476"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-11-19 17:02:47.000000000","tz":60},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-11-19 17:02:47.000000000","tz":60},"subject":"Replace sphinxcontrib-*diag","message":"Replace sphinxcontrib-*diag\n\nAs expected (and as done in lot of other OpenStack services) *blockdiag\nand *seqdiag extensions stop working for us and we need to replace this\nusage. Other services go directly to graphviz. Sadly sequence diagrams\ndoesn\u0027t look good, but there is no other alternative so far. It would be\nnice if we would have mermaid allowed in global requirements, but that\ntends to be a long process while we need an urgent solution (since\notherwise doc build is broken).\n\nChange-Id: Ie5fc80d7e5506c8c426a289687acc6c7c30f5ebd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/54f08920f46f550054d54248bf4f016779fad588"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/54f08920f46f550054d54248bf4f016779fad588"}]},"branch":"refs/heads/master"},"51045cc0dd1622fe03be78a0e2b273e4c975106e":{"kind":"REWORK","_number":2,"created":"2024-11-19 19:46:34.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/85/935685/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/85/935685/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/85/935685/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/85/935685/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/85/935685/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/85/935685/2"}}},"commit":{"parents":[{"commit":"15f34eff17f14074bdadc558e6fe80308e371476","subject":"Merge \"Ruff the code\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/15f34eff17f14074bdadc558e6fe80308e371476"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-11-19 17:02:47.000000000","tz":60},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-11-19 19:45:53.000000000","tz":60},"subject":"Replace sphinxcontrib-*diag","message":"Replace sphinxcontrib-*diag\n\nAs expected (and as done in lot of other OpenStack services) *blockdiag\nand *seqdiag extensions stop working for us and we need to replace this\nusage. Other services go directly to graphviz. Sadly sequence diagrams\ndoesn\u0027t look good, but there is no other alternative so far. It would be\nnice if we would have mermaid allowed in global requirements, but that\ntends to be a long process while we need an urgent solution (since\notherwise doc build is broken).\n\nChange-Id: Ie5fc80d7e5506c8c426a289687acc6c7c30f5ebd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/51045cc0dd1622fe03be78a0e2b273e4c975106e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/51045cc0dd1622fe03be78a0e2b273e4c975106e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
