)]}'
{"id":"openstack%2Fkeystone~937191","triplet_id":"openstack%2Fkeystone~master~I428e0b65129eddf2d01bc29b702c4f1278d88bf4","project":"openstack/keystone","branch":"master","topic":"openapi","attention_set":{},"removed_from_attention_set":{"14250":{"account":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"last_update":"2024-12-06 14:30:11.000000000","reason":"\u003cGERRIT_ACCOUNT_14250\u003e replied on the change","reason_account":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}},"27900":{"account":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"last_update":"2024-12-10 09:29:58.000000000","reason":"\u003cGERRIT_ACCOUNT_27900\u003e replied on the change","reason_account":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}}},"hashtags":[],"change_id":"I428e0b65129eddf2d01bc29b702c4f1278d88bf4","subject":"Add jsonschema decorators for domain resource","status":"MERGED","created":"2024-12-05 16:21:23.000000000","updated":"2024-12-10 13:17:36.000000000","submitted":"2024-12-10 13:16:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"937191-openapi","meta_rev_id":"56f4bc1ea92bee4369d99ca7b5e7a8ab38be40f9","_number":937191,"virtual_id_number":937191,"owner":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-12-10 13:16:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"all":[{"value":2,"date":"2024-12-06 14:21:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":2,"date":"2024-12-06 14:30:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"all":[{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":1,"date":"2024-12-06 14:30:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-05 17:31:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-12-06 14:21:56.000000000","updated_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"},{"updated":"2024-12-06 14:26:34.000000000","updated_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"}],"messages":[{"id":"63cbf9c7976f8f66929a84b67520bab28a8b273a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-12-05 16:21:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e9ecd8fcf6eec6898f6a5a31f36619aae0fda8f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-05 17:31:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ffdd8452fbca48a2b88f6aa5b688b52b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9425a6f77aaf49abad8d7db82b3ac4dd : SUCCESS in 17m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a095a47902a4f599535bbe347691440 : SUCCESS in 6m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/28fdc38814b34a43b72d4c4eb6a774d4 : SUCCESS in 13m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/65e5ecdc4a3746f4b803ac42c08f9ce7 : SUCCESS in 12m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/157cfcb88b6d4c0a9494e5fb62e5a58f : SUCCESS in 11m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/a6f0617a927649f3887d14c6ae304c57 : SUCCESS in 1h 03m 57s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5cfbcda2ac43473788fcc90b03bee682 : SUCCESS in 1h 01m 27s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/9db9ed6195bc47e599c8e5427eb587ec : SUCCESS in 17m 56s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/ea4e9aabcb5e4133be702be68c4dad53 : FAILURE in 19m 17s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/48e6fab4a2f84673b3e480855b6d7e92 : FAILURE in 15m 31s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/289c18251ff04e12a0be3fd4c3732aec : SUCCESS in 15m 56s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/6f7e2672ff8f40ce85095340b626076c : SUCCESS in 12m 13s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/13216294de084b7f965112d54d822785 : FAILURE in 15m 07s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7c7feb5b38144e768cc7edac14991f3b : SUCCESS in 35m 43s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/acb1a8ce4c18420f813f2250cdb052bf : SUCCESS in 21m 22s (non-voting)\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/12bfb65b870447e4ba90ba4e41e63257 : FAILURE in 6m 05s (non-voting)\n- keystone-dsvm-functional-oidc-federation https://zuul.opendev.org/t/openstack/build/2183690c2e5f4364ad2c340b6a7a5f81 : SUCCESS in 24m 08s","accounts_in_message":[],"_revision_number":1},{"id":"fadb3c3d4c397759388cfb6c39f9c69f2eefdce1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-05 18:04:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/84acbe0d66b14723a7bec9eaca79e34c\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4f9623042e4d415d8410815592b42d5b : TIMED_OUT in 1h 01m 49s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c641f92f6bf94f7db97c67c514d6e36b : TIMED_OUT in 1h 01m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6026b96806ba7297e7ec5238ea45364488479b78","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-12-05 18:25:32.000000000","message":"Topic pagination removed","accounts_in_message":[],"_revision_number":1},{"id":"312a7acbd036283242e69c1b7ff504ab269d7790","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-12-05 18:25:40.000000000","message":"Topic set to openapi","accounts_in_message":[],"_revision_number":1},{"id":"2ee7656a471e01ce98479d5657113510c81cbe28","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-12-05 18:45:47.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"7b84dc66c2d080f8000c51c3ed641a3a564a6226","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-05 19:53:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1b2cb42090954215a0b7402759410da9\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d78f46b73a1b496c914204d569d6267d : SUCCESS in 17m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad57c0e530554241a957f9c868981998 : SUCCESS in 5m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dd33b47365e74d2b9d38b841c2a3c35e : SUCCESS in 9m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c3ca91876c74430588267d5442dc88e6 : SUCCESS in 14m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c378bfd4002437887da180438c10c69 : SUCCESS in 7m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/b9aaed3d15064cb58ed35bd73f55b86d : SUCCESS in 50m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a4554ca4b6214ba0a6455e42282e5c52 : SUCCESS in 1h 06m 44s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/27e962db360a49f787a53e63c8869253 : SUCCESS in 19m 20s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/5a9e75b1c7f7404abe4f78f6ae75ba0a : SUCCESS in 30m 25s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/67ab0c4b1a3546af8e8ed70aa507ef5f : FAILURE in 16m 39s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/f391d47513d0414da58a05f73e38e0eb : SUCCESS in 30m 55s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/ec2a42f38ecf4eccadabb07b56967c9f : SUCCESS in 9m 30s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/87b78e9e1b024e82afef2baaa10cf527 : FAILURE in 24m 47s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/abc6ea2acdf94bd180de0e549b3fd27e : SUCCESS in 37m 17s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/5d16d1603e2c438f994e82b672139197 : FAILURE in 35m 33s (non-voting)\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/5a338e4c53544592aa598642ddea98e5 : FAILURE in 5m 46s (non-voting)\n- keystone-dsvm-functional-oidc-federation https://zuul.opendev.org/t/openstack/build/6d8c76b3197b4e6fad8979843f406641 : SUCCESS in 37m 27s","accounts_in_message":[],"_revision_number":2},{"id":"878ec77ad8722db4ece6a85bd4b0adea8f26e2e4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-05 19:55:16.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d49279a20cfe4e2da4048d3dab21a707\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/64af8b81b9e746bda42ab1216d843143 : TIMED_OUT in 1h 01m 06s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a0122cbf579446fd90b896a930b0d078 : TIMED_OUT in 1h 02m 08s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"af5b7eacd3e3e2af78b29e828192561001e493d6","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-12-06 07:56:43.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0e1ffa4cacc6383f3ce7c4f41576327203416b42","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-06 08:30:58.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe51749009b7404ab8ee8361dcca7c18\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0ff95cdb1dfc4b82a7d681ef37b06cae : SUCCESS in 31m 18s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/206f5c5c8b75421887697944815038ad : SUCCESS in 32m 18s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"78767805a902cc6f6656a107caf960fbe25d3c65","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-06 09:09:25.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad88a6ffb62840c9b442f39a786c8abe\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2281c257458641a9a155a3bb014a6a8a : SUCCESS in 13m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95de7968cf8a45889c9d40bdc84c3f82 : SUCCESS in 5m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c0873198d5c141269eacf0f21a0644dc : SUCCESS in 12m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d0128daecb7848b595fcfb4fb9ee8bb1 : SUCCESS in 8m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ec6a8ffabbd405abcc1cc340c405832 : SUCCESS in 9m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/71e7302280664f81a19f83411d28ec0c : SUCCESS in 1h 06m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b071269594cc4c4397902bed347e11d1 : SUCCESS in 58m 03s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/e2c024a763ff4f3b9aa0e19578482c94 : SUCCESS in 22m 17s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/d379ecf6cc79440eb10847cdd44ae5b4 : SUCCESS in 43m 50s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ff568a69449e457a9290d18bad739689 : FAILURE in 30m 34s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/0465772ebb984f86bf4ec17bc4b226a6 : SUCCESS in 30m 03s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/b6d725412bf54dda85339b8474f69f82 : SUCCESS in 10m 58s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/caf19ea18b8c4fb5998fe1510a097f33 : FAILURE in 17m 51s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e91006d1330143fb86eaefe25be60041 : SUCCESS in 46m 44s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/e179454b01294be9a99e75e9d1a9aae9 : FAILURE in 37m 46s (non-voting)\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/22860d58100b45528c588abece557b48 : SUCCESS in 4m 54s (non-voting)\n- keystone-dsvm-functional-oidc-federation https://zuul.opendev.org/t/openstack/build/c063760de69247e5b68840d9ca6ea327 : SUCCESS in 18m 17s","accounts_in_message":[],"_revision_number":2},{"id":"82fad1c301a2d3dfdc00b5460badbb3a8a2062ab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-12-06 14:14:47.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"dffc867296da8db338c35457161e18263df4f290","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-12-06 14:20:31.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"15040873ba5133f03fc449b097f74b1a692804e1","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2024-12-06 14:21:56.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"de7e92860ade85e927cbcc502ab83fd81dd2a2a3","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2024-12-06 14:26:34.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"55e282686be165da4ad6764bcb5ebbcbddd2d93f","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2024-12-06 14:30:11.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"7698296d38f029a4f43ea2b6e5155ba61e624cf1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-06 15:16:45.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d4c40fefbf87458ca1a7f287b8b5e41a\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c06ee2e5a922425393f145e0c3e7f73e : SUCCESS in 51m 26s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/080f1be2a7c14b8d887b954b134b5913 : SUCCESS in 53m 32s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"9aeac6b1472336a0bdedd8c972ad933f524abf6d","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-12-10 09:29:58.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"af9ccb51e9731d602aed9d39ecc6ff95f446d8c7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-10 10:33:33.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1b0631e71d534c7298efc0d13de04c0b\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9f5f269b2c0e4e82b0765218ce6b44a0 : TIMED_OUT in 1h 01m 00s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ad665457d413480ba689e338692f8649 : TIMED_OUT in 1h 00m 58s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"086302e53a2d7716b613a3ce8bd3c529f77f2476","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-10 11:16:13.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f5c92ef43cbc48a4adadd21c217e2405\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a08d24d4cb074709bb856cfb7430ed35 : SUCCESS in 15m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4660d70676ee43cfb88fc2545aa00745 : SUCCESS in 6m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0a1a4f8b9cc546e0a64e1452fbc50e37 : SUCCESS in 10m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7e12f37fa7f349e59d53fdf41922ef07 : SUCCESS in 11m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5afd9b63aa543d19f3be123ab620fd8 : SUCCESS in 11m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/8f6d63ac6dd54523bad07ba8bb056bfb : SUCCESS in 1h 06m 43s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a26eceecf0f049e9bd0cfb1636ffd20d : SUCCESS in 1h 29m 41s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/91ddb47ec7f9477194e67a66c1636013 : SUCCESS in 39m 01s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/32fc929eaa204d4ca44867402e890488 : SUCCESS in 34m 33s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dd33fb495dcd4989b761bcd335f620dc : FAILURE in 19m 14s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/be4c9c73865c4d4eb06d51d6ff66ca11 : SUCCESS in 27m 13s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/4c674656e21b4997ba6cbd9586877898 : SUCCESS in 23m 39s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/48483c1ed7654027960d804d62ccc0dd : FAILURE in 29m 34s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7ff29b8792e14d96b13977204e514bd1 : SUCCESS in 47m 46s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/95c6ca80b25442059e40aeec1564e1b1 : FAILURE in 28m 37s (non-voting)\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/1bfc7fd791b64a65b1d5232c44bf644a : SUCCESS in 5m 52s (non-voting)\n- keystone-dsvm-functional-oidc-federation https://zuul.opendev.org/t/openstack/build/0c940c8049d443ff905b9e92a5dd0169 : SUCCESS in 30m 11s","accounts_in_message":[],"_revision_number":4},{"id":"db17489dbb7c4a66ddb00a8cd2f896c927682214","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-10 11:16:47.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"fe4f905ef0d16b2612829d6ee068f4699a5b09a3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-10 13:16:10.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2866f6a1fa2f4b8b8304962d693c50a9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/904196c1f5a74542b2d8101390c931f0 : SUCCESS in 4m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8d0c6e23be1b4aff8d1a8310879ea9d4 : SUCCESS in 10m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5e3b258b63184033af45dfd444598479 : SUCCESS in 9m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ce225fbb1b749b09e85bad6e3639df1 : SUCCESS in 6m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/994566b06d4c4889a7e8de3d09927c44 : SUCCESS in 36m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9051900bdad34ab2a94a114b8fe2852d : SUCCESS in 57m 35s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/51fc4e1de43a479eb0d3e68cb2314961 : SUCCESS in 31m 30s\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/2adcb2db3e534a538a92188b8d42b4ff : SUCCESS in 22m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1f6b9fe679a840d1a0a8ab4240173e9c : SUCCESS in 38m 36s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/39361fe69a8441b89709fff8bbbfdfbd : FAILURE in 32m 03s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"4b93d279e9483e3cbd11e5d2b7ea387f09666ca9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-10 13:16:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"56f4bc1ea92bee4369d99ca7b5e7a8ab38be40f9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-10 13:17:36.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/27b8a8ebb1a340a7a4e42ab1c3bc9255\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b57b10a0060041f6af21e71de3848734 : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"9f4ba188061db8aa5583d9779197aa3fe738b2de","revisions":{"3febd023634d12d807cf042e299e3f9cf0cc159a":{"kind":"REWORK","_number":1,"created":"2024-12-05 16:21:23.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/91/937191/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/91/937191/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/91/937191/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/91/937191/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/91/937191/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/91/937191/1"}}},"commit":{"parents":[{"commit":"37b02476e77750d349f9db29036d57fbd5a079d7","subject":"Enable projects and domains pagination","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/37b02476e77750d349f9db29036d57fbd5a079d7"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-12-05 16:20:51.000000000","tz":60},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-12-05 16:20:51.000000000","tz":60},"subject":"Add jsonschema decorators for domain resource","message":"Add jsonschema decorators for domain resource\n\nChange-Id: I428e0b65129eddf2d01bc29b702c4f1278d88bf4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/3febd023634d12d807cf042e299e3f9cf0cc159a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/3febd023634d12d807cf042e299e3f9cf0cc159a"}]},"branch":"refs/heads/master"},"6b8f931e98709ac7cac388eb2454c4014bd61593":{"kind":"REWORK","_number":2,"created":"2024-12-05 18:45:47.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/91/937191/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/91/937191/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/91/937191/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/91/937191/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/91/937191/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/91/937191/2"}}},"commit":{"parents":[{"commit":"37b02476e77750d349f9db29036d57fbd5a079d7","subject":"Enable projects and domains pagination","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/37b02476e77750d349f9db29036d57fbd5a079d7"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-12-05 16:20:51.000000000","tz":60},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-12-05 18:45:29.000000000","tz":60},"subject":"Add jsonschema decorators for domain resource","message":"Add jsonschema decorators for domain resource\n\nChange-Id: I428e0b65129eddf2d01bc29b702c4f1278d88bf4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/6b8f931e98709ac7cac388eb2454c4014bd61593"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/6b8f931e98709ac7cac388eb2454c4014bd61593"}]},"branch":"refs/heads/master"},"59ec5709e8ad36380c75246612c06527713f4e84":{"kind":"REWORK","_number":3,"created":"2024-12-06 14:14:47.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/91/937191/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/91/937191/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/91/937191/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/91/937191/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/91/937191/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/91/937191/3"}}},"commit":{"parents":[{"commit":"37b02476e77750d349f9db29036d57fbd5a079d7","subject":"Enable projects and domains pagination","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/37b02476e77750d349f9db29036d57fbd5a079d7"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-12-05 16:20:51.000000000","tz":60},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-12-06 14:14:38.000000000","tz":60},"subject":"Add jsonschema decorators for domain resource","message":"Add jsonschema decorators for domain resource\n\nChange-Id: I428e0b65129eddf2d01bc29b702c4f1278d88bf4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/59ec5709e8ad36380c75246612c06527713f4e84"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/59ec5709e8ad36380c75246612c06527713f4e84"}]},"branch":"refs/heads/master"},"9f4ba188061db8aa5583d9779197aa3fe738b2de":{"kind":"REWORK","_number":4,"created":"2024-12-06 14:20:31.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/91/937191/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/91/937191/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/91/937191/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/91/937191/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/91/937191/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/91/937191/4"}}},"commit":{"parents":[{"commit":"37b02476e77750d349f9db29036d57fbd5a079d7","subject":"Enable projects and domains pagination","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/37b02476e77750d349f9db29036d57fbd5a079d7"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-12-05 16:20:51.000000000","tz":60},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-12-06 14:20:24.000000000","tz":60},"subject":"Add jsonschema decorators for domain resource","message":"Add jsonschema decorators for domain resource\n\nChange-Id: I428e0b65129eddf2d01bc29b702c4f1278d88bf4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/9f4ba188061db8aa5583d9779197aa3fe738b2de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/9f4ba188061db8aa5583d9779197aa3fe738b2de"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
