)]}'
{"id":"openstack%2Fkeystone~937197","triplet_id":"openstack%2Fkeystone~master~I03ba754875050fdb93219e915fc099680679b6c4","project":"openstack/keystone","branch":"master","topic":"immutable","attention_set":{},"removed_from_attention_set":{"27900":{"account":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"last_update":"2025-01-10 18:20:40.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I03ba754875050fdb93219e915fc099680679b6c4","subject":"Fix domain `immutable` lockdown","status":"MERGED","created":"2024-12-05 17:41:38.000000000","updated":"2025-01-10 18:22:01.000000000","submitted":"2025-01-10 18:20:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"937197-immutable","meta_rev_id":"86194a9bac86e30d8a95d87771e03a1efa27b6b1","_number":937197,"virtual_id_number":937197,"owner":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-01-10 18:20:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"all":[{"value":2,"date":"2024-12-06 14:32:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":2,"date":"2024-12-13 14:16:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"all":[{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":1,"date":"2024-12-13 14:16:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-05 18:52:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-12-06 14:06:55.000000000","updated_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"},{"updated":"2024-12-13 14:16:21.000000000","updated_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"}],"messages":[{"id":"ba8d9186aa83ec846170406176306471873ad59a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-12-05 17:41:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fbcef8621ae0f7a6be20af5c7554fa55bb4472bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-05 18:52:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f19c184d4aa74646b7258b8d2fe93d45\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/813b7d597a0b46bb8085b7531e3fa562 : SUCCESS in 16m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c301cd971fe7453db433f55b98bab084 : SUCCESS in 6m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8ac4b3397abb4ea786c910a1fd49831a : SUCCESS in 13m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c7d585526c3242619bdfbf1d99802136 : SUCCESS in 8m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/045e576f7eda46f08586b932c7d0ecc9 : SUCCESS in 13m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/ba5bdd6d4ffc489ba8b782aa412a7570 : SUCCESS in 1h 04m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2bfaa6a202374710ae329b7416056db4 : SUCCESS in 57m 09s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/6451cf9bb6294c9cbbac939c07e64a36 : SUCCESS in 38m 50s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/35c46ca57c214a69ae4581eed037234c : SUCCESS in 21m 41s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3f85194962a54f009eaa80541edd4d91 : FAILURE in 21m 01s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/c7a334e410974b98a2d0bcdd9be69b64 : SUCCESS in 19m 27s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/4dd42b16e0f94ba59eebde16a61e4bfe : SUCCESS in 22m 02s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/7ff3780c81694eeaaf680ebc63c66867 : FAILURE in 12m 13s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/59a6c970dc39469ea470cdf6e05dbc3e : SUCCESS in 1h 03m 28s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/521a7051d563436d84570c9ae3ccd95b : SUCCESS in 32m 02s (non-voting)\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/a743453f68dd4bf689d18723562457d9 : FAILURE in 5m 22s (non-voting)\n- keystone-dsvm-functional-oidc-federation https://zuul.opendev.org/t/openstack/build/7c02f9e6973f4188b9bef8d1295c3b4b : SUCCESS in 31m 13s","accounts_in_message":[],"_revision_number":1},{"id":"3dc467b120f532895b53c40e473f622d7070387b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-05 19:07:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5861370ecd5a4eedba4f91564d2b24d0\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/82ef008b083342bcaab73d45d4d04a34 : TIMED_OUT in 1h 00m 59s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4466f4909e884304b21c55171cde8c36 : TIMED_OUT in 1h 02m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0b62218bd0ac3ed80e13e4345ff3f867a660898e","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2024-12-06 14:06:55.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"251af92913b7f4ae3522cf1db395f20ec2d75c79","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-12-06 14:32:11.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"6ee6fad107afddb5593ee1fce044bdda4a0437bd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-06 15:34:51.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/42ceecae79a745eeab5dc3905b9eee0b\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b25c5051fdca4c04bd849c2c31e8e7ca : TIMED_OUT in 1h 00m 58s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1351f2cbb60c408687924689455d6e88 : SUCCESS in 47m 34s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"265d06e1f4201fefccf40a08af050759e3b0bc19","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2024-12-13 14:16:21.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"da10f3b5c1e70d678a446101e91e4df84240a3ab","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2024-12-13 14:16:36.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b75beaea15bea2b7228d7843c42305a46ee50863","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2025-01-10 14:14:19.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"118d2d62159be8c971c832e604f2cd50fe4f8294","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 16:24:57.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/56907cff998743c1a78516a94e4d5fb3\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/eac1da3f42aa41b6bd3f0a14270e8b68 : TIMED_OUT in 1h 01m 15s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/598743b55c924df99e9e890b06ba330b : TIMED_OUT in 1h 01m 27s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"20c618a4211b5e825e08f1627fa1929b7ea0bcf7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 16:25:39.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a47de612034f446589a76a30a66e486e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4714c517e6cf401fa1c453d8f8cc4644 : SUCCESS in 19m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad8ad7b22ee34477ae6b30128aa5d5eb : SUCCESS in 6m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c60365e6bd774b3eb78faa49e987bbc7 : SUCCESS in 12m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1d05feb7ef484dbc8f13d0d2825faa4f : SUCCESS in 11m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64cd188f252a49d589058128709dba58 : SUCCESS in 11m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/028c8d3852bd4a289a51abbc587db533 : SUCCESS in 1h 02m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a89eac46d9344e5fb1752f529bdb5283 : SUCCESS in 1h 34m 08s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/3826a061373e41feb18da0283d25f623 : SUCCESS in 30m 42s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/c8caef894be34f02868bb694bc3f27d2 : SUCCESS in 48m 06s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c416f54d37ff443d95f75b614b643a29 : FAILURE in 26m 41s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/3547f23c55e04adbb0daacccca2a3dab : SUCCESS in 28m 11s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/d259bf228d0145cc9548eb467ad1c201 : SUCCESS in 20m 42s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/2ff1f885d8454252b2d5b75573449c54 : FAILURE in 25m 34s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bf941211821742adbd9f1f2ddce56212 : SUCCESS in 55m 35s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/488925e99f2744e7a2228d18c636ea30 : FAILURE in 46m 23s (non-voting)\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/bb808ac9951740adbe945322c239ad10 : SUCCESS in 6m 00s (non-voting)\n- keystone-dsvm-functional-oidc-federation https://zuul.opendev.org/t/openstack/build/1f1cbd9a3ea54b5dac5e2231142f1b8a : SUCCESS in 31m 39s","accounts_in_message":[],"_revision_number":2},{"id":"cdc218323d5d2ce65a88ff5c2c4444e375b6974e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 16:26:15.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"3c8b4e9fa8c0462d79864161e3fd0db1a39bcf36","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 18:20:40.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d00a1d147dd941a98d28c24efe8b69a5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9b6e34dfa794c3291eb5c4d6e2da9fd : SUCCESS in 6m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1d0c418c2c09416f80f0d0ba13c2a442 : SUCCESS in 14m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a0d9df9ed3ff4df9a7e125df935f951d : SUCCESS in 9m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2e6a605739849318fcf4b727bc96a83 : SUCCESS in 12m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/fc0d549148a543158e058fb4449be9f3 : SUCCESS in 1h 02m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7768d19752824963b776cdcbc07a30d0 : SUCCESS in 1h 42m 46s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/6e0c158128a54d29b6869d2b53a24ae3 : SUCCESS in 25m 10s\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/b12db56380124389985a9d3b830de4d7 : SUCCESS in 31m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c8599ba4e22a4fedba76d1d4e9f68e44 : SUCCESS in 51m 33s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/73c0b0a3f9dc4ca09cf79a5808869767 : FAILURE in 36m 42s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"09dc5abf60d05bd52d26ff9e951306fd39595668","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 18:20:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"86194a9bac86e30d8a95d87771e03a1efa27b6b1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-10 18:22:01.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/56cc0bad72df4948945a65f0c64e41d3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80f37e01d5a547f3ab19253d11681f93 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"628db853fb902282ea260ea0c5f9ba213e6ea2b0","revisions":{"c8fd201560b4482905f3f60e1789cd1c25a995a0":{"kind":"REWORK","_number":1,"created":"2024-12-05 17:41:38.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/97/937197/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/97/937197/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/97/937197/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/97/937197/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/97/937197/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/97/937197/1"}}},"commit":{"parents":[{"commit":"3febd023634d12d807cf042e299e3f9cf0cc159a","subject":"Add jsonschema decorators for domain resource","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/3febd023634d12d807cf042e299e3f9cf0cc159a"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-12-05 17:15:29.000000000","tz":60},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-12-05 17:34:19.000000000","tz":60},"subject":"Fix domain `immutable` lockdown","message":"Fix domain `immutable` lockdown\n\nA bug in the test design have hidden the issue with options of the\ndomain. It is possible to set immutable to true during creation or\nupdate of the domain, but it is impossible to change it back through\nAPI. Reason is the check that refuses update operation when there are\nmore then 1 updated props with immutable switched on. For projects it is\nno problem to send 1 update but for domain we pull additional base props\n(is_domain\u003dtrue, domain_id\u003dnone, parent_id\u003dnone) in the\nget_project_from_domain method. And since the test for unsetting\nimmutable started with immutable already unset (in difference to similar\ntest for the project) it was never identified.\n- discard default props of the domain to keep the length check happy\n- ensure test for unsetting immutable starts with immutable domain\n\nChange-Id: I03ba754875050fdb93219e915fc099680679b6c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/c8fd201560b4482905f3f60e1789cd1c25a995a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/c8fd201560b4482905f3f60e1789cd1c25a995a0"}]},"branch":"refs/heads/master"},"628db853fb902282ea260ea0c5f9ba213e6ea2b0":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-12-06 14:32:11.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/97/937197/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/97/937197/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/97/937197/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/97/937197/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/97/937197/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/97/937197/2"}}},"commit":{"parents":[{"commit":"8fe72cfe5dba5acd10e1b40fb34366a347d9cbed","subject":"Fix wsgi service name for the federation tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/8fe72cfe5dba5acd10e1b40fb34366a347d9cbed"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-12-05 17:15:29.000000000","tz":60},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-12-06 14:32:11.000000000","tz":0},"subject":"Fix domain `immutable` lockdown","message":"Fix domain `immutable` lockdown\n\nA bug in the test design have hidden the issue with options of the\ndomain. It is possible to set immutable to true during creation or\nupdate of the domain, but it is impossible to change it back through\nAPI. Reason is the check that refuses update operation when there are\nmore then 1 updated props with immutable switched on. For projects it is\nno problem to send 1 update but for domain we pull additional base props\n(is_domain\u003dtrue, domain_id\u003dnone, parent_id\u003dnone) in the\nget_project_from_domain method. And since the test for unsetting\nimmutable started with immutable already unset (in difference to similar\ntest for the project) it was never identified.\n- discard default props of the domain to keep the length check happy\n- ensure test for unsetting immutable starts with immutable domain\n\nChange-Id: I03ba754875050fdb93219e915fc099680679b6c4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/628db853fb902282ea260ea0c5f9ba213e6ea2b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/628db853fb902282ea260ea0c5f9ba213e6ea2b0"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
