)]}'
{"id":"openstack%2Fkeystone~937483","triplet_id":"openstack%2Fkeystone~master~Ie52dcda09f76bfd86b8e1a8b6d71dfd7078ecf22","project":"openstack/keystone","branch":"master","topic":"openapi/roles","attention_set":{},"removed_from_attention_set":{"36482":{"account":{"_account_id":36482,"name":"Oria Weng","display_name":"0weng","email":"oweng@osuosl.org","username":"0weng"},"last_update":"2024-12-13 16:38:20.000000000","reason":"Change was submitted"},"27900":{"account":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"last_update":"2024-12-13 14:21:54.000000000","reason":"\u003cGERRIT_ACCOUNT_27900\u003e replied on the change","reason_account":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}}},"hashtags":[],"change_id":"Ie52dcda09f76bfd86b8e1a8b6d71dfd7078ecf22","subject":"Add JSON schema and validation for `role` resource","status":"MERGED","created":"2024-12-10 21:09:09.000000000","updated":"2024-12-13 16:38:20.000000000","submitted":"2024-12-13 16:38:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"937483-openapi/roles","meta_rev_id":"8a21d298cc24aa615621b99dd4e89b650c3e71bf","_number":937483,"virtual_id_number":937483,"owner":{"_account_id":36482,"name":"Oria Weng","display_name":"0weng","email":"oweng@osuosl.org","username":"0weng"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-12-13 16:38:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"all":[{"value":2,"date":"2024-12-13 14:46:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":2,"date":"2024-12-13 14:23:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":1,"date":"2024-12-13 14:21:54.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"all":[{"value":1,"date":"2024-12-13 14:46:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-10 22:06:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-12-10 22:21:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-12-11 00:34:50.000000000","updated_by":{"_account_id":36482,"name":"Oria Weng","display_name":"0weng","email":"oweng@osuosl.org","username":"0weng"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"REVIEWER"},{"updated":"2024-12-13 14:23:42.000000000","updated_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"},{"updated":"2024-12-13 14:46:34.000000000","updated_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"}],"messages":[{"id":"126ab53c82d3333af106344000e3632f542f7a9c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36482,"name":"Oria Weng","display_name":"0weng","email":"oweng@osuosl.org","username":"0weng"},"date":"2024-12-10 21:09:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"405c844302349f6d72744b1306f04ba4a1105d61","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36482,"name":"Oria Weng","display_name":"0weng","email":"oweng@osuosl.org","username":"0weng"},"date":"2024-12-10 21:09:56.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"519b7b1add6d4e23988804aa391fbc6035916841","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":36482,"name":"Oria Weng","display_name":"0weng","email":"oweng@osuosl.org","username":"0weng"},"date":"2024-12-10 21:23:16.000000000","message":"Topic openapi/roles removed","accounts_in_message":[],"_revision_number":2},{"id":"3b247888f7e872aa47340a0275d06d072488ed32","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":36482,"name":"Oria Weng","display_name":"0weng","email":"oweng@osuosl.org","username":"0weng"},"date":"2024-12-10 21:23:21.000000000","message":"Topic set to openapi","accounts_in_message":[],"_revision_number":2},{"id":"4d74465824e03eec4a2518712bb3f6a308e27a91","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-10 22:06:19.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0cd4545cb60c4507873ee1e84a397e29\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/daf4a2f898da4226abfcb9983ec7ba7e : FAILURE in 49m 58s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/fad50a068cb6491c95c65973cd74e7f7 : FAILURE in 54m 05s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3dd0bc1b90ab50ee9faa4b655a0cd676f6c9073d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-10 22:21:49.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/dd942d77690c4dd7b29576c0910b4035\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8700a2ab94f941268c6a8c572065b8d0 : FAILURE in 12m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a254e2c15d140878714669f418ecedf : SUCCESS in 9m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/316154e74fc84c92bc1bd9d2a7ef1040 : FAILURE in 11m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ef7b69f453db4627b1e260f779c3ace4 : FAILURE in 8m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91897de216754722a2f751ee740e8832 : SUCCESS in 11m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/bc487dba262f406fb0ae49df5df3aabe : FAILURE in 1h 06m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/47231885f0c4485d9ed6b04838c36534 : FAILURE in 24m 33s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/3c8b33f29f414b4a80cc41670d9eca9b : FAILURE in 22m 21s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/f2e6060a15ca430ea0e2d117a0953783 : FAILURE in 40m 45s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3a406da9145a4836878a2e905f14ec12 : FAILURE in 21m 56s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/fdac5d2e649d48f5a6697955e0015554 : FAILURE in 19m 59s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/bb8cd67461964f77a170fa701d48a598 : FAILURE in 23m 30s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/1b8032c53bd4458dae6f1d63104cb43c : FAILURE in 17m 03s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3b2a79d9e0054c35a530ac20498b3911 : FAILURE in 22m 46s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/7fb2025a677e480bbf4e880221247f8d : FAILURE in 34m 21s (non-voting)\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/709eb6e123724ebe8e6f50a4e0dfdb48 : SUCCESS in 7m 04s (non-voting)\n- keystone-dsvm-functional-oidc-federation https://zuul.opendev.org/t/openstack/build/17e066ad2bca47e5adbe92bca4560807 : FAILURE in 18m 26s","accounts_in_message":[],"_revision_number":2},{"id":"46db44c8aa49d101db0c9ad049082a1fbfe5bab2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36482,"name":"Oria Weng","display_name":"0weng","email":"oweng@osuosl.org","username":"0weng"},"date":"2024-12-10 22:33:01.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"364488ee1ad8ed89029c38460002f8d45db6e136","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-10 23:35:53.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ce8f39a1d014b01aa83bb8160f86dd7\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8225dae3cbb842289640ddb2de3b6fec : TIMED_OUT in 1h 01m 08s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8d4f1bb9c03a452db400b978e370fade : SUCCESS in 50m 40s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"cd3aa77567474cdb9b0651743b0a61389308a569","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-10 23:43:12.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/536c7252d49d4f38b027a41fd01119fe\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4ebf089d019a4909be130e67eac8fe07 : SUCCESS in 14m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b5055de07dd4c2aa7b95a5eb4cc7a26 : SUCCESS in 6m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3826abbf440446dca86034015922c465 : SUCCESS in 10m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/567b5257cf8b4125930a6de5d5c1d72d : SUCCESS in 8m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0222b3936d9d43549da42c7b669b92b3 : SUCCESS in 6m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/ba0ff3bcd68043faa75e0c8dbc0cc855 : SUCCESS in 35m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/96db5aa5cb8a486cae1cbbb794882719 : SUCCESS in 54m 31s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/20011cb4fbed46b39cb5843df3161ee9 : SUCCESS in 36m 44s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/731f5d4f3fe640809874a75a16cea253 : SUCCESS in 44m 15s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b73c4b5efeed4a648f6d459a06f3b997 : FAILURE in 23m 25s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/29ff75e1596042698852718300af5a58 : SUCCESS in 31m 05s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/41f073ed74114ee495f201f3a50ee4de : SUCCESS in 20m 35s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/a14c413ecf6e443a9aa7b18bb786ba5a : FAILURE in 12m 59s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/74febc1911d64b74af26b46feca7a3ee : SUCCESS in 1h 04m 43s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/17f7911635274796a65fe2ace96638fa : FAILURE in 37m 59s (non-voting)\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/0fa4240ccb4a416b9b74beed08cb9852 : SUCCESS in 6m 32s (non-voting)\n- keystone-dsvm-functional-oidc-federation https://zuul.opendev.org/t/openstack/build/0d88c0029c5d49e2a223d16d662d15e5 : SUCCESS in 35m 01s","accounts_in_message":[],"_revision_number":3},{"id":"7660722014e27cd7b316416cff7e79899733b072","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-12-13 14:21:54.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"aa3677b3bab41564c4bae1acd6ce1e8033d4f27b","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2024-12-13 14:23:42.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"435fb4d0931a9e9c78c785ae224d3d361a7cce62","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2024-12-13 14:46:34.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"9b3e8e6eda361dbbcd746551e104086b66c71d6a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-13 14:47:08.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"3b460933d56fda1207ac8cbf87e1ea18b38a89fc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-13 16:38:20.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2aa174fef91341128608931aac374d46\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d80747468d543c4843e7fce2ff4e472 : SUCCESS in 6m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/25bbcd295a1e44fe96e4cbd2163ce390 : SUCCESS in 12m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e4e4cf0509964e819c245158ea622c29 : SUCCESS in 13m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0a337bfe09a479ea87b978f3e475eb6 : SUCCESS in 12m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/d4c277ded4fa406c93399eff92cfbb1d : SUCCESS in 51m 47s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/96d7ef89c9164a90a6a116b13970c821 : SUCCESS in 1h 03m 41s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/1ca8a9d805954cb2adce3e3d7bad3674 : SUCCESS in 32m 20s\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/dc3e91fc3ff6404d9cc06e87edc03412 : SUCCESS in 33m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c7d521d183fe4a9cb0204011b66f7377 : SUCCESS in 36m 32s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/d603561967a84a518c4c764cd9b93ad4 : FAILURE in 38m 46s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8a21d298cc24aa615621b99dd4e89b650c3e71bf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-13 16:38:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"0a690d47b2498e94a3aed437c92e86f627d5e247","revisions":{"7dcefc1b34fbd8c1602a452d88c3472e74ece154":{"kind":"REWORK","_number":1,"created":"2024-12-10 21:09:09.000000000","uploader":{"_account_id":36482,"name":"Oria Weng","display_name":"0weng","email":"oweng@osuosl.org","username":"0weng"},"ref":"refs/changes/83/937483/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/83/937483/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/83/937483/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/83/937483/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/83/937483/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/83/937483/1"}}},"commit":{"parents":[{"commit":"8fe72cfe5dba5acd10e1b40fb34366a347d9cbed","subject":"Fix wsgi service name for the federation tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/8fe72cfe5dba5acd10e1b40fb34366a347d9cbed"}]}],"author":{"name":"0weng","email":"oweng@osuosl.org","date":"2024-12-10 21:06:37.000000000","tz":-480},"committer":{"name":"0weng","email":"oweng@osuosl.org","date":"2024-12-10 21:06:37.000000000","tz":-480},"subject":"Add JSON schema and validation for `role` resource","message":"Add JSON schema and validation for `role` resource\n\nChange-Id: Ie52dcda09f76bfd86b8e1a8b6d71dfd7078ecf22\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/7dcefc1b34fbd8c1602a452d88c3472e74ece154"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/7dcefc1b34fbd8c1602a452d88c3472e74ece154"}]},"branch":"refs/heads/master"},"03bebb2f307da7a383db73c93e5f5d6017794ef2":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-12-10 21:09:56.000000000","uploader":{"_account_id":36482,"name":"Oria Weng","display_name":"0weng","email":"oweng@osuosl.org","username":"0weng"},"ref":"refs/changes/83/937483/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/83/937483/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/83/937483/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/83/937483/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/83/937483/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/83/937483/2"}}},"commit":{"parents":[{"commit":"7dc31993637b5df41748c4135ba6740011aea9eb","subject":"Merge \"Add jsonschema decorators for domain resource\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/7dc31993637b5df41748c4135ba6740011aea9eb"}]}],"author":{"name":"0weng","email":"oweng@osuosl.org","date":"2024-12-10 21:06:37.000000000","tz":-480},"committer":{"name":"Oria Weng","email":"oweng@osuosl.org","date":"2024-12-10 21:09:56.000000000","tz":0},"subject":"Add JSON schema and validation for `role` resource","message":"Add JSON schema and validation for `role` resource\n\nChange-Id: Ie52dcda09f76bfd86b8e1a8b6d71dfd7078ecf22\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/03bebb2f307da7a383db73c93e5f5d6017794ef2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/03bebb2f307da7a383db73c93e5f5d6017794ef2"}]},"branch":"refs/heads/master","description":"Rebase"},"0a690d47b2498e94a3aed437c92e86f627d5e247":{"kind":"REWORK","_number":3,"created":"2024-12-10 22:33:01.000000000","uploader":{"_account_id":36482,"name":"Oria Weng","display_name":"0weng","email":"oweng@osuosl.org","username":"0weng"},"ref":"refs/changes/83/937483/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/83/937483/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/83/937483/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/83/937483/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/83/937483/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/83/937483/3"}}},"commit":{"parents":[{"commit":"7dc31993637b5df41748c4135ba6740011aea9eb","subject":"Merge \"Add jsonschema decorators for domain resource\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/7dc31993637b5df41748c4135ba6740011aea9eb"}]}],"author":{"name":"0weng","email":"oweng@osuosl.org","date":"2024-12-10 21:06:37.000000000","tz":-480},"committer":{"name":"0weng","email":"oweng@osuosl.org","date":"2024-12-10 22:32:49.000000000","tz":-480},"subject":"Add JSON schema and validation for `role` resource","message":"Add JSON schema and validation for `role` resource\n\nChange-Id: Ie52dcda09f76bfd86b8e1a8b6d71dfd7078ecf22\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/0a690d47b2498e94a3aed437c92e86f627d5e247"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/0a690d47b2498e94a3aed437c92e86f627d5e247"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
