)]}'
{"id":"openstack%2Fkeystone~937731","triplet_id":"openstack%2Fkeystone~master~I6ea33652a8b3fca32e74e00062b59c1e0513c633","project":"openstack/keystone","branch":"master","hashtags":[],"change_id":"I6ea33652a8b3fca32e74e00062b59c1e0513c633","subject":"Revert \"Update notification tests to work with o-m 9.0.0\"","status":"NEW","created":"2024-12-13 18:54:01.000000000","updated":"2024-12-13 19:57:36.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f964d38c7f1afaa9327b12b787bb044354cfce94","_number":937731,"virtual_id_number":937731,"owner":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-12-13 19:40:29.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-12-13 19:40:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0c010c891f303b4d1c8dea60d06879ae54386772","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"date":"2024-12-13 18:54:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"68974164b90e5bb836c833b9942b04de1f785dd7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-13 19:40:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/89edb98579f44b7db24bb0f0df4c6a67\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/93c1504883e44ea8ba1e30dc697b81ea : SUCCESS in 17m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/001b9b00bca743efa23f478ada812a29 : SUCCESS in 5m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eb8539c5294f491687da4ce24086259c : SUCCESS in 11m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f58dc856eeb34caa8aa5be22ab13883c : SUCCESS in 14m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3cd4400c85684d4fb8027d297cf9ca74 : SUCCESS in 12m 30s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/69c99505fd2d4c8c958c075c76bd68b9 : FAILURE in 39m 04s (non-voting)\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/dc0f46e318584d85b84ce680c6dfa91a : SUCCESS in 6m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f964d38c7f1afaa9327b12b787bb044354cfce94","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-13 19:57:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc3d33fc77ce4c58aa9610865a353cb6\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0bb560a0140c4213a84abee7d440de65 : SUCCESS in 51m 46s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/01d637d077024373930fe644d333324e : SUCCESS in 1h 01m 29s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b25dae1b589c4b360c650b0b8877a5e7783c2279","revisions":{"b25dae1b589c4b360c650b0b8877a5e7783c2279":{"kind":"REWORK","_number":1,"created":"2024-12-13 18:54:01.000000000","uploader":{"_account_id":11583,"name":"Arnaud Morin","email":"arnaud.morin@gmail.com","username":"arnaudmorin"},"ref":"refs/changes/31/937731/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/31/937731/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/31/937731/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/31/937731/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/31/937731/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/31/937731/1"}}},"commit":{"parents":[{"commit":"6bcc050ddd62da5860c1969d6d17e0e8aed76bd8","subject":"Merge \"Add JSON schema and validation for `implied role`\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/6bcc050ddd62da5860c1969d6d17e0e8aed76bd8"}]}],"author":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2024-12-13 18:48:08.000000000","tz":60},"committer":{"name":"Arnaud Morin","email":"arnaud.morin@ovhcloud.com","date":"2024-12-13 18:53:23.000000000","tz":60},"subject":"Revert \"Update notification tests to work with o-m 9.0.0\"","message":"Revert \"Update notification tests to work with o-m 9.0.0\"\n\nThis reverts commit f26efd719df0a74ccb4f6649fc4c465ef1ccab86.\n\nWe can safely revert this commit now because oslo.messaging 9.0.0 is not\nin the testing scope anymore.\n\nAlso, this commit introduce a slight issue that could appear randomly:\nno such group [oslo_messaging_notifications]\n\nIt\u0027s appearing everytime if you run this single test:\ntox -e py3 keystone.tests.unit.common.test_notifications.NotificationsTestCase.test_opt_out_authenticate_event\n\nReverting it fixes the issue.\n\nChange-Id: I6ea33652a8b3fca32e74e00062b59c1e0513c633\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/b25dae1b589c4b360c650b0b8877a5e7783c2279"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/b25dae1b589c4b360c650b0b8877a5e7783c2279"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
