)]}'
{"id":"openstack%2Fkeystone~938788","triplet_id":"openstack%2Fkeystone~master~I4e72b24ecae19606abb7ba076ce8d8d87c21c085","project":"openstack/keystone","branch":"master","topic":"openapi","attention_set":{},"removed_from_attention_set":{"27900":{"account":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"last_update":"2025-01-17 14:36:06.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4e72b24ecae19606abb7ba076ce8d8d87c21c085","subject":"Add jsonschemas to the User resource","status":"MERGED","created":"2025-01-09 11:21:11.000000000","updated":"2025-01-17 14:37:22.000000000","submitted":"2025-01-17 14:36:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"938788-openapi","meta_rev_id":"98d878f47059a82cb71f908effb978c5d3f4f78c","_number":938788,"virtual_id_number":938788,"owner":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-01-17 14:36:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"all":[{"value":2,"date":"2025-01-10 14:11:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":2,"date":"2025-01-17 12:36:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"all":[{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":1,"date":"2025-01-17 12:36:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-09 13:00:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-01-09 13:01:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-10 14:11:31.000000000","updated_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"},{"updated":"2025-01-17 12:36:58.000000000","updated_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"}],"messages":[{"id":"32a0a0a674f3f05973895060ca75c59468d06dbc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2025-01-09 11:21:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fdcbd7361417e1c79825e525da0f2c937e0bcf18","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-09 13:00:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6d3b14183660454ea354d9a08634d29b\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e3cc73db028e4f9d96835040051b5deb : TIMED_OUT in 1h 01m 04s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e3dcc227613b4842bd430dd2d91d9028 : TIMED_OUT in 1h 00m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a021fe3a43dc2e2833d1c805f3749e30718c6881","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-09 13:01:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a63561920204090b1ead72b96bc34d5\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/541ed95f91e54575bb8d4ad82bbf4c82 : SUCCESS in 16m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37fda182747f4a46b49066bfc905de7a : SUCCESS in 6m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/984a55b79c5c48b38585c59f12f9e75a : SUCCESS in 11m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8a896039db184a19b4ca38e66be8ef04 : SUCCESS in 13m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b7d8c77d84124786b954a43be1aa71c3 : SUCCESS in 10m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/0fb14d26edf14ad19cb731151eafa4b3 : SUCCESS in 1h 04m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ab9e6076ff1d49b4bbb93e0db067cc8c : SUCCESS in 1h 35m 22s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/6fa150a094024d4087f44f2a238bcfb1 : SUCCESS in 26m 06s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/26fd38b727c547ed8809aa7d02cc0137 : SUCCESS in 28m 55s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/954862af37e94d14ad07622795f7a07f : FAILURE in 28m 50s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/08e344cf9b7f47a79130892bae5b4367 : SUCCESS in 30m 54s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/79b84374da084dfd8be8f264c3b2e6a5 : SUCCESS in 11m 00s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/d91a64bdcc2842618a393bcb93ba2893 : FAILURE in 16m 44s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f1eeb7a6ddd14c14a873171abb2d6f31 : SUCCESS in 1h 00m 21s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/eebe71338fb34d6cb04a5dfa10a2bbcc : FAILURE in 38m 23s (non-voting)\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/025f029bb64647858b64b96d6b083abc : POST_FAILURE in 4m 06s (non-voting)\n- keystone-dsvm-functional-oidc-federation https://zuul.opendev.org/t/openstack/build/fa4eecf927804b788ffe260231659862 : SUCCESS in 24m 03s","accounts_in_message":[],"_revision_number":1},{"id":"21f274ba80584daf0a6411738bf911591517fa18","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2025-01-09 13:42:17.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3afcfda7a8d9861af331cb0caa0144e9d2f90a85","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-09 14:54:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d03f33e7d59948598aef76009ab825b3\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/959ab1b411244a0bad53263a31bd59e0 : TIMED_OUT in 1h 00m 50s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/fa955b7e2cea4274910c0791a459bf7b : TIMED_OUT in 1h 01m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bbf996722a1ba0e0a10b9d43cb4ea30e53ab76d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-09 15:39:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c257c10853f498d9297f08dd5d0bba3\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/85d03630902742c797e5b96ffe6fc234 : SUCCESS in 22m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f8b2b03c54048459bd67a0d6ebb3446 : SUCCESS in 4m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f1dda8df4411468e989684a5dcae6811 : SUCCESS in 14m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5b564ea119014beb92d3121886427822 : SUCCESS in 11m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26ea7c267a66473695b5dcac699a8e1b : SUCCESS in 14m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/94be0d5b84b644bab3f7225c883ecfc3 : SUCCESS in 58m 13s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c5b48f0c0dc2491099553b60b9484012 : SUCCESS in 1h 33m 56s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/980afd1d197c4d7cba6be475ef5b5a3d : SUCCESS in 35m 25s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/e1a5b8fbc90248e098c54ee4390e7aff : SUCCESS in 42m 44s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c14f923eedb14090a8d9f5f263bd5597 : FAILURE in 27m 57s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/0e4f00dcc1464808b0c1ba8200df83b6 : SUCCESS in 27m 43s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/967be17640e345d293c99c2cee78a110 : SUCCESS in 22m 03s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/92cee2ec59d249fa86f2be4a2396c5d0 : FAILURE in 17m 31s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/28a5d8b1a4c84b23971c7ec94fa3a700 : SUCCESS in 58m 28s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/dedff66ce9ab457dae2c598454762f7d : FAILURE in 21m 28s (non-voting)\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/a405c6d4850b49d0abc315637807dc6b : SUCCESS in 6m 08s (non-voting)\n- keystone-dsvm-functional-oidc-federation https://zuul.opendev.org/t/openstack/build/c2f7957f15054a1c93b80e2b74493888 : SUCCESS in 18m 49s","accounts_in_message":[],"_revision_number":1},{"id":"82b4bf16019fdfb4551e042421f325f325dbfb0f","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2025-01-10 14:11:31.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a0a2be8de8763cee99534a5ebafd2e166e469f0c","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2025-01-17 12:36:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"99a0912771ee1a67dc2cbc03140201597ee405ea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-17 12:37:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5fed7dffcc9c140d691ad2236cefbfd8777a27b3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-17 14:36:06.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/99e70d504f804ed9b27360d09e6a9eff\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2469eb14441e4d6285df7eecee75dc91 : SUCCESS in 7m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2aa92c2655364dd89e0052f5d3a0768d : SUCCESS in 12m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1a59e3e196bb4feaa67ade066303b5f2 : SUCCESS in 11m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/559cd3e2c5554610b69fba9961407538 : SUCCESS in 10m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/830fb72acc2e486d9a29a0ca33127d0f : SUCCESS in 50m 39s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cff9bef5621447bd984e5264eba06655 : SUCCESS in 1h 52m 55s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/6a4a3083b7984a4caed11f15bf208fef : SUCCESS in 36m 10s\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/56f8920a0e7c4b4da914fbe69cdeb8f1 : SUCCESS in 32m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/812ddcaba66d49a1a86be74be92f770e : SUCCESS in 1h 06m 28s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/1f5ace274f7d4623b27162c220d6a5a8 : FAILURE in 42m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a7c3aa200789ec01147fcbfae4757c6f4002d36a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-17 14:36:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"98d878f47059a82cb71f908effb978c5d3f4f78c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-17 14:37:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eef78d1d17d442fc83020e5e6f26a801\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab550ddc0aec42f68eb73e2e49dea7e2 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fcf8b4c6cef8f7421865602853a1340e03b33ce9","revisions":{"fcf8b4c6cef8f7421865602853a1340e03b33ce9":{"kind":"REWORK","_number":1,"created":"2025-01-09 11:21:11.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/88/938788/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/88/938788/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/88/938788/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/88/938788/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/88/938788/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/88/938788/1"}}},"commit":{"parents":[{"commit":"6bcc050ddd62da5860c1969d6d17e0e8aed76bd8","subject":"Merge \"Add JSON schema and validation for `implied role`\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/6bcc050ddd62da5860c1969d6d17e0e8aed76bd8"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2025-01-09 09:30:54.000000000","tz":60},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2025-01-09 11:12:06.000000000","tz":60},"subject":"Add jsonschemas to the User resource","message":"Add jsonschemas to the User resource\n\nChange-Id: I4e72b24ecae19606abb7ba076ce8d8d87c21c085\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/fcf8b4c6cef8f7421865602853a1340e03b33ce9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/fcf8b4c6cef8f7421865602853a1340e03b33ce9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
