)]}'
{"id":"openstack%2Fkeystone~938814","triplet_id":"openstack%2Fkeystone~master~Ieb6db300e5728d65451211dc73a690af4a384508","project":"openstack/keystone","branch":"master","topic":"openapi","attention_set":{},"removed_from_attention_set":{"27900":{"account":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"last_update":"2025-01-17 14:36:12.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ieb6db300e5728d65451211dc73a690af4a384508","subject":"Implement pagination for users and groups","status":"MERGED","created":"2025-01-09 15:22:36.000000000","updated":"2025-01-17 14:38:36.000000000","submitted":"2025-01-17 14:36:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"938814-openapi","meta_rev_id":"ccbbf3ef108c7f378c3e07f9d2c3271dbd9d93a9","_number":938814,"virtual_id_number":938814,"owner":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-01-17 14:36:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"all":[{"value":2,"date":"2025-01-10 14:22:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":2,"date":"2025-01-17 12:39:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"all":[{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":1,"date":"2025-01-17 12:39:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"}],"REVIEWER":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-09 16:26:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-01-09 16:38:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-10 14:22:27.000000000","updated_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"},{"updated":"2025-01-17 12:39:37.000000000","updated_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"CC"},{"updated":"2025-01-17 12:39:37.000000000","updated_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"reviewer":{"_account_id":8648,"name":"Radomir Dopieralski","email":"openstack@dopieralski.pl","username":"thesheep"},"state":"CC"},{"updated":"2025-01-17 12:39:42.000000000","updated_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"}],"messages":[{"id":"e73da779b4041173d6404566b8b8fae9b3c57ace","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2025-01-09 15:22:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3d13836df1012ec36eaa109f07afa917ae9cf311","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-09 16:26:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bed60e2d9071443fb02da9c01b5db211\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d32aa925a1e445eaa3a3aa47b5a42b65 : TIMED_OUT in 1h 01m 03s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/895bd7c6138d427fa766aa33eb125895 : TIMED_OUT in 1h 01m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"74cd6d356d54722377199e30979d2d84ab140463","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-09 16:38:18.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/35aeed41b51b423fa7ebab992098f68b\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/d4c654ecfc704d74b45512da22c4676d : SUCCESS in 5m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d190ae4826594300b5efccf1281f40cf : FAILURE in 16m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55c61c92fd5845ea8c382cba485d0fd3 : SUCCESS in 7m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/07bcb13389c84b2ba96622501e5fdf05 : FAILURE in 12m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/09649cb94e9e463dacec8a9028ecb04a : FAILURE in 13m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/855441429b1f433593ccb4d6d2d3f580 : SUCCESS in 12m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/b77b5bd4d65b4fb99b350a53466352a4 : SUCCESS in 57m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/51d0d98500b34399beba3b11acad31a0 : SUCCESS in 1h 02m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/98ecbfe7688a4b85a6e535b010138191 : SUCCESS in 11m 19s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/70cb269fef46432d83f4e8eae58dff3b : SUCCESS in 19m 19s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/540474580bca4c49af5593a197a76651 : SUCCESS in 41m 46s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1ad5f2caf4744240838a8884b25c2870 : FAILURE in 16m 56s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/53c310716d96454481d4346fda1a4e00 : SUCCESS in 33m 11s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/df85453d9ba44c9b95e0df4f7cc8748d : SUCCESS in 21m 41s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/28a5f4e10d6b439fa82685a35d10145a : FAILURE in 23m 51s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9fae6affbfbb4e019bc1d986d1325d58 : SUCCESS in 1h 09m 02s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/54ca06c74f214b31a4ed72d8b9c5405e : FAILURE in 51m 06s (non-voting)\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/63a66b72df2c4e5abc24b114fe82fa2e : SUCCESS in 6m 03s (non-voting)\n- keystone-dsvm-functional-oidc-federation https://zuul.opendev.org/t/openstack/build/3f70136be4354648acfdc8c34772a512 : SUCCESS in 36m 28s","accounts_in_message":[],"_revision_number":1},{"id":"8f8f07d60d7d19c6733fe8d039f5b96ce197c8c5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2025-01-09 17:52:41.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"dcdff2f12c63732b99ae0876f98aa153aa285cc0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-09 18:56:53.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dc750ca58d4c418b8ece158a7d4a6e6a\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6b809eca85324067878c75aa7a4dcff4 : TIMED_OUT in 1h 00m 54s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d445cd33b51e47089acd897e3cfe94bc : TIMED_OUT in 1h 01m 19s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"64a52e45db7544203331a02563ccf1a53c01f7c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-09 19:29:30.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f19b29780af1471797ab7a4d6e6824ca\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/f6bba41987cb49b793e1a839051281b9 : SUCCESS in 4m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d9c1d8bc043c49988128bba70415f563 : SUCCESS in 16m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71ac746cfc2f4a0c913a00de67a5a851 : SUCCESS in 6m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e7b2030534fa4c4fa110134f799eeb89 : SUCCESS in 12m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a8ef8728378046db979127ce4ad29963 : SUCCESS in 11m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f238059ccb5c4ee68a78b7f97dff2c29 : SUCCESS in 13m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/f0b5731d5b2a4bc69050f9281cfc5883 : SUCCESS in 1h 08m 55s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/80631764ad2a41899ec572d8dadca21e : FAILURE in 1h 22m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a81f14705af4448b8bf96637764ea604 : SUCCESS in 11m 32s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/131eb5a1dcf74acdaaaf3222f4a163ef : SUCCESS in 39m 03s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/c6f58bea95d04e3ea1dbe8bb63a397a5 : SUCCESS in 29m 04s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ca9b3ce03f284a4caa434d56abeff0c5 : FAILURE in 32m 24s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/1d5a200789a242f4853c9fd2698ec048 : SUCCESS in 22m 20s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/d41a1fd1792c4f7eb58cca00bd312895 : SUCCESS in 10m 29s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/f1066d9ab48942d69908ef14d8de6256 : FAILURE in 29m 14s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/764ec8f5690e4c1ca5ebb53cdc0526ab : SUCCESS in 48m 01s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/afea1a8f5b31416daeb3d0457667acad : FAILURE in 31m 18s (non-voting)\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/7d285385de0c43f6a50fa2a4f631285c : SUCCESS in 6m 58s (non-voting)\n- keystone-dsvm-functional-oidc-federation https://zuul.opendev.org/t/openstack/build/50b5fbd183d541db880910073209fdc5 : SUCCESS in 27m 24s","accounts_in_message":[],"_revision_number":2},{"id":"0f025d0a95da68a1ee650f68110a7f9ef4228775","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2025-01-09 19:42:28.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0e0412504faeb2040ba5b7a1c705ca1aa9dfbd2f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-09 20:39:25.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0cd813da502846729e58506693b47f44\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/95fc4f88e5df4899a1374fc51bcb8103 : SUCCESS in 53m 54s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f7a237cd0bb94a69984683ef9b819070 : SUCCESS in 54m 49s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"999db0a850a69b43a4e2bc86e4265d737ffb331f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-09 21:15:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b8ba689121c43beb9721848db21b3ef\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/beb2483dc1ac4bf7bbd2f48bd0229ae9 : SUCCESS in 5m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8b891b1632b94ff0934de23fa6944820 : SUCCESS in 14m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b916032800f4451f8d54de02297f29f7 : SUCCESS in 4m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7821b7d12e0e451ea29bbac44bf8dbc7 : SUCCESS in 11m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d71717b0000648b982e1777239f8cf5b : SUCCESS in 11m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b325485488834a3c957abadfd648b808 : SUCCESS in 6m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/4f16dda32e4b461481e9b2ad51b8650a : SUCCESS in 39m 17s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a2bbf6a752154933ab04ddfc72df4932 : SUCCESS in 1h 29m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/12f024740d34488cb8c25d2f59736875 : SUCCESS in 7m 49s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/99074c40d49f48db9327f19ddbd4dae5 : SUCCESS in 35m 35s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/c62fe155f6c647f0b92dc29b5d03c9c8 : SUCCESS in 43m 37s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d67ba3d05263471d9ae80aa978175ea8 : FAILURE in 17m 23s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/688ed538a3c140198a256d61e7018ad0 : SUCCESS in 16m 28s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/304bebef08364ac7b5d9128c287e0d84 : SUCCESS in 9m 22s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/65bc1b1604bf45bb889e438664434070 : FAILURE in 12m 32s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6535c94f43aa4d20bb2b769f31ec90af : SUCCESS in 56m 06s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/fdd5ce29a3df46c5af8ac36ba87c59ce : FAILURE in 37m 28s (non-voting)\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/69a1af562bda4b02ab34760e5c2a5ff9 : SUCCESS in 5m 46s (non-voting)\n- keystone-dsvm-functional-oidc-federation https://zuul.opendev.org/t/openstack/build/4f37a726cc024e089e16c976ff71d3a0 : SUCCESS in 34m 19s","accounts_in_message":[],"_revision_number":2},{"id":"c6a0189968b2a87290274f7c126927cdba69a1d4","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2025-01-10 14:22:27.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"a91fe3c3b27a2b54aa15064d8050a3edc2d8272e","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2025-01-17 12:39:42.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"592c7d4bc638c2b98a3af080a234550f2a069828","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-17 12:40:23.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1e82c55e47d9eab04ea9b293ee59d616cc6c86df","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-17 14:36:12.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e54b4601ae554164be69f3266329a4df\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/4f35a42fdbf1430fbd9abc6608baffd6 : SUCCESS in 3m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c17b91b579a424fa11c77cdf6d9d4a5 : SUCCESS in 7m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/32c66b9e62424d51a8ecc555fc8e95cb : SUCCESS in 12m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c026da15b91c4757a9c7b57eeb94cbe6 : SUCCESS in 14m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2dfec822474f442482af216acf2d2e0d : SUCCESS in 12m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/c4836cf4084c4e53a2cab8f984f460e0 : SUCCESS in 48m 23s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f5055f306c3e44f98023be4380ef26c6 : SUCCESS in 1h 03m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/411668367bda4d81978e6d46f6c814da : SUCCESS in 11m 27s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/f934cc007a584a46b44ce14583e5b48e : SUCCESS in 33m 07s\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/102574633e2644fcacd24e899b747d59 : SUCCESS in 24m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2e4118b834094775a9af6efff7c13634 : SUCCESS in 46m 13s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/4b7055bb2bfe405296bcddb982ec1681 : FAILURE in 33m 39s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"86a28bb0a3a389391c7b8a5cbbf67ca569364363","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-17 14:36:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"ccbbf3ef108c7f378c3e07f9d2c3271dbd9d93a9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-17 14:38:36.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/12ba6084f37b443989f30b2b62018d26\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/58b8a4f1dafe4bf7af71f5a005cc6c31 : SUCCESS in 44s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/570cacd9e1a64206b44e3e76972c0f5c : SUCCESS in 1m 06s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c52caf6e3b634e8caa076112d25e9668 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ba8bf97faca4f95a0278081c3794f83e2c6cd48f","revisions":{"4512f8bc1843f986498c02bac700edc1d55c27ff":{"kind":"REWORK","_number":1,"created":"2025-01-09 15:22:36.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/14/938814/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/14/938814/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/14/938814/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/14/938814/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/14/938814/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/14/938814/1"}}},"commit":{"parents":[{"commit":"4f4a3e0972c68fd646c98d0d0b40cfb9652ae499","subject":"Add jsonschema for Group resource","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/4f4a3e0972c68fd646c98d0d0b40cfb9652ae499"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2025-01-09 15:21:46.000000000","tz":60},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2025-01-09 15:21:46.000000000","tz":60},"subject":"Implement pagination for users and groups","message":"Implement pagination for users and groups\n\nChange-Id: Ieb6db300e5728d65451211dc73a690af4a384508\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/4512f8bc1843f986498c02bac700edc1d55c27ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/4512f8bc1843f986498c02bac700edc1d55c27ff"}]},"branch":"refs/heads/master"},"ba8bf97faca4f95a0278081c3794f83e2c6cd48f":{"kind":"REWORK","_number":2,"created":"2025-01-09 17:52:41.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/14/938814/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/14/938814/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/14/938814/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/14/938814/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/14/938814/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/14/938814/2"}}},"commit":{"parents":[{"commit":"4f4a3e0972c68fd646c98d0d0b40cfb9652ae499","subject":"Add jsonschema for Group resource","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/4f4a3e0972c68fd646c98d0d0b40cfb9652ae499"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2025-01-09 15:21:46.000000000","tz":60},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2025-01-09 17:51:21.000000000","tz":60},"subject":"Implement pagination for users and groups","message":"Implement pagination for users and groups\n\nChange-Id: Ieb6db300e5728d65451211dc73a690af4a384508\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/ba8bf97faca4f95a0278081c3794f83e2c6cd48f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/ba8bf97faca4f95a0278081c3794f83e2c6cd48f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
