)]}'
{"id":"openstack%2Fkeystone~939172","triplet_id":"openstack%2Fkeystone~master~Ieea94c4b409a9fdd2af494e43bc6c891c81504f8","project":"openstack/keystone","branch":"master","topic":"fix-ldap-backend","attention_set":{},"removed_from_attention_set":{"7973":{"account":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"last_update":"2025-01-16 15:59:36.000000000","reason":"\u003cGERRIT_ACCOUNT_7973\u003e replied on the change","reason_account":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"}},"37374":{"account":{"_account_id":37374,"name":"Stanislav Zaprudskiy","email":"s.zaprudskiy@sap.com","username":"stanislav-z"},"last_update":"2025-01-16 15:59:40.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ieea94c4b409a9fdd2af494e43bc6c891c81504f8","subject":"Implement reset_last_active for LDAP backend","status":"ABANDONED","created":"2025-01-13 21:20:12.000000000","updated":"2025-01-16 15:59:40.000000000","total_comment_count":4,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"8bc8e8f604af79e2894044b81a512d3315ef179e","_number":939172,"virtual_id_number":939172,"owner":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-01-13 22:44:08.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37374,"name":"Stanislav Zaprudskiy","email":"s.zaprudskiy@sap.com","username":"stanislav-z"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"disliked":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"all":[{"value":1,"date":"2025-01-13 21:23:53.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"value":-1,"date":"2025-01-15 17:05:28.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2025-01-15 13:24:42.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":37374,"name":"Stanislav Zaprudskiy","email":"s.zaprudskiy@sap.com","username":"stanislav-z"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37374,"name":"Stanislav Zaprudskiy","email":"s.zaprudskiy@sap.com","username":"stanislav-z"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"_account_id":37374,"name":"Stanislav Zaprudskiy","email":"s.zaprudskiy@sap.com","username":"stanislav-z"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-13 21:23:53.000000000","updated_by":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"reviewer":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"state":"REVIEWER"},{"updated":"2025-01-13 22:44:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-15 13:24:42.000000000","updated_by":{"_account_id":37374,"name":"Stanislav Zaprudskiy","email":"s.zaprudskiy@sap.com","username":"stanislav-z"},"reviewer":{"_account_id":37374,"name":"Stanislav Zaprudskiy","email":"s.zaprudskiy@sap.com","username":"stanislav-z"},"state":"REVIEWER"},{"updated":"2025-01-15 17:05:28.000000000","updated_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"REVIEWER"}],"messages":[{"id":"83215f290c637a812f7883c49dfb7278a5fdfcb9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2025-01-13 21:20:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7810fa2631df90a7da7a7d1c55dee4dbd8886316","author":{"_account_id":9914,"name":"Ade Lee","email":"alee@redhat.com","username":"alee"},"date":"2025-01-13 21:23:53.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"e39126bdfc6e2b3e3750f7c7b8231fb37e4b6c04","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-13 22:44:08.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a4940ca2e14d4c6babcfc6357cb5e959\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/07b1aacdaa66417ba8c7f27efd4c17a2 : SUCCESS in 15m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/baa3c37c68424fa6b7726e97e7bdc35d : FAILURE in 7m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4b26d5afe4c2413f9c2e486a886d574f : SUCCESS in 11m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1ed8abeb2da948be92ce32864a589473 : SUCCESS in 11m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4f882c1173649d78b75ae3bed5b0a13 : SUCCESS in 11m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/6164e96e9d1745c3a5f1a0fb1e49e545 : SUCCESS in 1h 00m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e419a8cc442e4820b2144dae7b845e20 : SUCCESS in 1h 22m 04s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/06aba875679e40d491eba0d101fc2682 : SUCCESS in 34m 41s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/221a92b8533a4e69ac561897b1d542c1 : SUCCESS in 46m 34s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/40205b9cf53641d48a17542f3f5fbf25 : FAILURE in 17m 20s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/86d1c11254b04c2f9721dea43662308f : SUCCESS in 31m 51s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/4ada73794f3e4a33bef9608d35025228 : SUCCESS in 23m 50s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/2f21433ec8dd4591b63d1ce189572bd2 : FAILURE in 23m 34s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8b6484c9e17547ea992f80e06bde2db0 : SUCCESS in 55m 28s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/ceaa392cd04e4ec4a8534fa147e2a740 : FAILURE in 36m 48s (non-voting)\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/b4b9ddf16e6e49a29906fc88d3f20e59 : SUCCESS in 7m 01s (non-voting)\n- keystone-dsvm-functional-oidc-federation https://zuul.opendev.org/t/openstack/build/ca0365e20a0f4a748f3c0b8a96310f82 : SUCCESS in 30m 33s","accounts_in_message":[],"_revision_number":1},{"id":"fa1d8f985dd761b7eee7ea76beedac17f6792f9f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-13 23:21:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e6097af9def47e5b2e394635f8978ae\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4da8e9b3290845b7bdfd01e4bf07479b : TIMED_OUT in 1h 00m 52s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f8f59f8ed5ca4357b81f6856df67eec1 : SUCCESS in 47m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6de82d61c5bea8ea879ef3e08ff1cd552c69cdfc","author":{"_account_id":37374,"name":"Stanislav Zaprudskiy","email":"s.zaprudskiy@sap.com","username":"stanislav-z"},"date":"2025-01-15 13:24:42.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"e3d971670802aa8eca3768944fb2361ff56c79bd","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2025-01-15 17:05:28.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2aefb232446b914ef98766c9efea9296f27df6ad","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2025-01-16 15:59:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8bc8e8f604af79e2894044b81a512d3315ef179e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"date":"2025-01-16 15:59:40.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1f69166426a70eb2fac805511f21b9007c01f992","revisions":{"1f69166426a70eb2fac805511f21b9007c01f992":{"kind":"REWORK","_number":1,"created":"2025-01-13 21:20:12.000000000","uploader":{"_account_id":7973,"name":"Douglas Mendizábal","email":"dmendiza@redhat.com","username":"dougmendizabal"},"ref":"refs/changes/72/939172/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/72/939172/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/72/939172/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/72/939172/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/72/939172/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/72/939172/1"}}},"commit":{"parents":[{"commit":"86eeb2d2066f4504c47c7b790c8c68fc4aa1fcbe","subject":"Merge \"Fix domain `immutable` lockdown\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/86eeb2d2066f4504c47c7b790c8c68fc4aa1fcbe"}]}],"author":{"name":"Douglas Mendizabal","email":"dmendiza@redhat.com","date":"2025-01-13 20:45:43.000000000","tz":-300},"committer":{"name":"Douglas Mendizabal","email":"dmendiza@redhat.com","date":"2025-01-13 21:18:38.000000000","tz":-300},"subject":"Implement reset_last_active for LDAP backend","message":"Implement reset_last_active for LDAP backend\n\nA new abstract method `reset_last_active` for the abstract base class\n`keystone.identity.backends.base.IdentityDriverBase` was introduced in\ncommit e9513f8e4f25e1f20bc6fcab71d9177120000abf\n\nThis method is used by the SQL backend to reset NULL values in the\n`last_active_at` column in the User table.  However, the abstract method\nwas not implemented for the LDAP backend, which results in a runtime\nerror when attempting to run Keystone using LDAP.\n\nThis patch implements the method for the LDAP backend by raising an\nexception if the method is called.  This will result in an error when a\nuser attempts to run `barbican-manage reset_last_active` when using the\nLDAP backend.\n\nChange-Id: Ieea94c4b409a9fdd2af494e43bc6c891c81504f8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/1f69166426a70eb2fac805511f21b9007c01f992"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/1f69166426a70eb2fac805511f21b9007c01f992"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
