)]}'
{"id":"openstack%2Fkeystone~939778","triplet_id":"openstack%2Fkeystone~master~Ic4f60bc39b1143f436100a169f1afbf08e1b0fea","project":"openstack/keystone","branch":"master","topic":"drop-sha512-crypt","attention_set":{},"removed_from_attention_set":{"935":{"account":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"last_update":"2025-02-07 18:31:23.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ic4f60bc39b1143f436100a169f1afbf08e1b0fea","subject":"Drop the sha512_crypt module","status":"MERGED","created":"2025-01-22 10:16:34.000000000","updated":"2025-02-07 18:33:15.000000000","submitted":"2025-02-07 18:31:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"939778-drop-sha512-crypt","meta_rev_id":"65a8607a8749f40cb32a44485249e5a5b8daecd8","_number":939778,"virtual_id_number":939778,"owner":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-07 18:31:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37374,"name":"Stanislav Zaprudskiy","email":"s.zaprudskiy@sap.com","username":"stanislav-z"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"all":[{"value":2,"date":"2025-01-31 14:09:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":2,"date":"2025-01-27 14:01:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-01-22 10:21:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37374,"name":"Stanislav Zaprudskiy","email":"s.zaprudskiy@sap.com","username":"stanislav-z"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"all":[{"value":1,"date":"2025-01-31 14:09:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37374,"name":"Stanislav Zaprudskiy","email":"s.zaprudskiy@sap.com","username":"stanislav-z"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"_account_id":37374,"name":"Stanislav Zaprudskiy","email":"s.zaprudskiy@sap.com","username":"stanislav-z"}],"CC":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-22 10:21:55.000000000","updated_by":{"_account_id":37374,"name":"Stanislav Zaprudskiy","email":"s.zaprudskiy@sap.com","username":"stanislav-z"},"reviewer":{"_account_id":37374,"name":"Stanislav Zaprudskiy","email":"s.zaprudskiy@sap.com","username":"stanislav-z"},"state":"REVIEWER"},{"updated":"2025-01-22 11:27:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-01-22 12:03:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-27 14:01:13.000000000","updated_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"REVIEWER"},{"updated":"2025-01-31 14:09:55.000000000","updated_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"},{"updated":"2025-02-07 14:19:58.000000000","updated_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"CC"}],"messages":[{"id":"9a504a933e61787e1ccaddf6110c992e6c7b7095","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"date":"2025-01-22 10:16:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d7ee51ae4aa1105488b8f218b7ac31e5676b1c0c","author":{"_account_id":37374,"name":"Stanislav Zaprudskiy","email":"s.zaprudskiy@sap.com","username":"stanislav-z"},"date":"2025-01-22 10:21:55.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8cd667e7f0a73f087a54714add703c14b0afaecf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-22 11:27:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0bcf78257222436eacf244e9d080dfb9\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/206067c4c25242df9cf63c027b427a06 : TIMED_OUT in 1h 00m 53s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9aa6e206ce7c4f13829604ec256ff425 : TIMED_OUT in 1h 01m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3fd7c2e73ad9ff1cc0816ea7f6b38361656c4786","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-22 12:03:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/77dda8d7171846b78e579be57d7e2b19\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0deca82ee2954455807e04310ba833bb : SUCCESS in 15m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51363bf8b0034acfa0edda810af2a974 : SUCCESS in 4m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ab01b0abfa7a4cce8e28414a413d1f1a : SUCCESS in 21m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/607ad65da590421abc6fe64668aa15e0 : SUCCESS in 15m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ab446167f8d48b6bf25a67950e3a49a : SUCCESS in 11m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/b62c9ef3e4fa4486854df564edb2a21b : SUCCESS in 1h 04m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/bee7cfb9440343cb9cfcc624989da016 : SUCCESS in 1h 40m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/acbcfb75b61c4327b4dbcd7f5cbb70e3 : SUCCESS in 12m 55s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/274182dd691242238a4ca7f584d20c0d : SUCCESS in 30m 50s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/e390473c2ad94af994fed4937ce6812d : SUCCESS in 36m 06s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/673b61709eee4872a5c408ac662ba150 : FAILURE in 24m 36s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/ad3e4b23804b408287c308d7b19f4b1f : SUCCESS in 18m 58s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/5929b984e6e84ea1ba266d146e51e53c : SUCCESS in 12m 23s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/f371fc06ee614facac35ff4efd3605b8 : FAILURE in 15m 17s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/08ec1fb4836b4e21903776ac566b71c1 : SUCCESS in 46m 52s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/c80b6e3e86744412bd886e688d9c778a : FAILURE in 24m 18s (non-voting)\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/2f743b50694b4437978c20a82c2c8537 : POST_FAILURE in 6m 45s (non-voting)\n- keystone-dsvm-functional-oidc-federation https://zuul.opendev.org/t/openstack/build/b76272d3c57b4861a0c326ccbecea362 : SUCCESS in 35m 32s","accounts_in_message":[],"_revision_number":1},{"id":"8067b38ea69d2c999714216d229a00649bd43933","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2025-01-27 14:01:13.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c03ee77fcffeb9b75dc1e3ad1b35c73a24266797","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2025-01-31 14:09:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8cc4c145356a27b59598fb1a301a9fab3638bbaf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-31 14:10:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"16d11e9fb11cc58fb01ab9ce1dc478d3c5758a12","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-31 15:23:30.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6cf155e00d4d4b2f960c80c2b4d6b7e9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8ebb2d5facd4a2cae7419eb45604aeb : SUCCESS in 9m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/70884a4252d04fbca68fbcc732c34b9e : SUCCESS in 11m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d72ed744a90a4bf1860ddaf62b4e46d5 : SUCCESS in 10m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5b6e1f379da43529248ff3f825c824c : SUCCESS in 13m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/9d186f7ba1254f48a71c37e6eae4eda1 : SUCCESS in 1h 04m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8aa5819530cb4d52a14db147d35ffc71 : FAILURE in 6m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bbd59d50c190477aa3d020b4ad7d090f : SUCCESS in 12m 45s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/498bf5a403a04385b4abbf69a8b207d6 : FAILURE in 9m 34s\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/2d5034f9ae6340aa9c07c1e8eb019323 : SUCCESS in 27m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c6bdfd13dee94b619e746873ee5147bf : FAILURE in 6m 35s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/7738208f50d443549b0fb718ebad4801 : FAILURE in 5m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"389f94f4816d82c3eac166a9f0f8a1bee576ecc4","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2025-02-07 14:19:58.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"03ce6354e7c1d470cd338cdc79dfd341ce4af88e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-07 15:23:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6da46060895848c6a0679beb8a9d90fc\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/24ee2b63200b47799a66f0c2589cff4a : TIMED_OUT in 1h 01m 16s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/60fb6670e0404c77ab8296afe4c87b05 : TIMED_OUT in 1h 01m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"98c736944ae3e455de1a8f0f35f8cbf82e538d0e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-07 16:11:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/88344e9153e9465090d6b5c99497b32d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a91b794e0b9248a78f29105f64a107e2 : SUCCESS in 29m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1a74c3314b74825b6283b772ed41584 : SUCCESS in 8m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/90109242a9574c149ec2dfdae539a3e7 : SUCCESS in 12m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a8f6d9ca259446aca06bc396d10d8254 : SUCCESS in 13m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64a65fcb7e3b4965ac54522075b9ff45 : SUCCESS in 11m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/82d6118c3efb4225bf6452e360940f3b : SUCCESS in 1h 00m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5dead72a391c4feea42aeb7197c0e737 : SUCCESS in 1h 49m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2a06f80c82544f8480dd9f7809e84199 : SUCCESS in 7m 46s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/a628d6de71384f3f9a3c9726d68fb655 : SUCCESS in 37m 19s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/9d1015995601473d9ba3d900cc5e6cdf : SUCCESS in 40m 47s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3342b54e82e943e4ba60f8bb97e63a91 : FAILURE in 17m 15s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/c0f7388bfb9e446aaab514e8f353f644 : SUCCESS in 25m 48s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/224a4e347fe8407c911d0d1428e543d8 : SUCCESS in 21m 35s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/9266814321f34dfaa41c588784d3c455 : FAILURE in 26m 54s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b54c463c10f64bde8d0943350e14968e : SUCCESS in 44m 06s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/ed4ff7c0f8ec49d9966253572b12d1ea : FAILURE in 37m 50s (non-voting)\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/9c99a216c28641bbbe7176eee1f83af2 : SUCCESS in 7m 51s (non-voting)\n- keystone-dsvm-functional-oidc-federation https://zuul.opendev.org/t/openstack/build/1fca5ca9ec11429ba2a3ca3477f87b03 : SUCCESS in 42m 34s","accounts_in_message":[],"_revision_number":1},{"id":"46335ae2b7c708cc73ddd1443caf42db86b31f2b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-07 16:11:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ad876cda566ff59e6dde25c1965b9a7fcdae135a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-07 18:31:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0fcce1d46612409ebe97441fe80e1ea7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b0bb70ad7944a6eaccb8000c5f1d5e2 : SUCCESS in 9m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b218f3ccd89544b2adfd8536bb0480e2 : SUCCESS in 13m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2965e6d90ea54161983b54b7fbc823f5 : SUCCESS in 13m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d78e74c8500b489987776252ac00b11e : SUCCESS in 31m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/09cce21cf3c442cd806b1e91407ffe26 : SUCCESS in 1h 23m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cf65c041e155426bbee54d57b942ac7d : SUCCESS in 2h 05m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6b9bd5d3d72b4f6a8c909c8b99bb294b : SUCCESS in 20m 59s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/d5ec88a0455248268a9c44a67a99573a : SUCCESS in 32m 20s\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/4b1a45bc28214f159fe7b585662e8f1f : SUCCESS in 30m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a5d9fb83dc4d4049bf3b6260ddc17267 : SUCCESS in 1h 01m 40s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/34df4ea9101b448f898da27ad23711ec : FAILURE in 42m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fac0f05ee5d5bb8c6765d50c9b5906871984ca42","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-07 18:31:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"65a8607a8749f40cb32a44485249e5a5b8daecd8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-07 18:33:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ba744c845b145a3befc780972ad6d7d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9912697401e24a5ca310ca31445256ea : SUCCESS in 1m 21s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a16bf57834f5465f862ebf7ace8b699e : SUCCESS in 1m 31s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4a4137d5f77ced3f4f011846fb9c287444047649","revisions":{"4a4137d5f77ced3f4f011846fb9c287444047649":{"kind":"REWORK","_number":1,"created":"2025-01-22 10:16:34.000000000","uploader":{"_account_id":935,"name":"James Page","email":"ringo.page@gmail.com","username":"james-page"},"ref":"refs/changes/78/939778/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/78/939778/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/78/939778/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/78/939778/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/78/939778/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/78/939778/1"}}},"commit":{"parents":[{"commit":"9eaf07c97d3126f5d9006f4d2586ba76d15171b2","subject":"Merge \"Add JSON schema to `trust` and validation decorators to trust resource.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/9eaf07c97d3126f5d9006f4d2586ba76d15171b2"}]}],"author":{"name":"James Page","email":"james.page@canonical.com","date":"2025-01-22 10:14:00.000000000","tz":0},"committer":{"name":"James Page","email":"james.page@canonical.com","date":"2025-01-22 10:16:06.000000000","tz":0},"subject":"Drop the sha512_crypt module","message":"Drop the sha512_crypt module\n\nThis feature was deprecated for removal in the Epoxy cycle.\n\nComplete the deprecation process and enable support for\nPython \u003e\u003d 3.13 which drops the crypt module this feature\nuses.\n\nChange-Id: Ic4f60bc39b1143f436100a169f1afbf08e1b0fea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/4a4137d5f77ced3f4f011846fb9c287444047649"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/4a4137d5f77ced3f4f011846fb9c287444047649"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
