)]}'
{"id":"openstack%2Fkeystone~940344","triplet_id":"openstack%2Fkeystone~master~I972efbd39bd47772dabcafee136b060fdd7a5437","project":"openstack/keystone","branch":"master","topic":"openapi/role_assignments","attention_set":{},"removed_from_attention_set":{"36482":{"account":{"_account_id":36482,"name":"Oria Weng","display_name":"0weng","email":"oweng@osuosl.org","username":"0weng"},"last_update":"2025-02-14 16:32:58.000000000","reason":"Change was submitted"},"27900":{"account":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"last_update":"2025-02-14 16:32:58.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I972efbd39bd47772dabcafee136b060fdd7a5437","subject":"Add JSON schema \u0026 validation for `role assignment`","status":"MERGED","created":"2025-01-28 22:01:50.000000000","updated":"2025-02-14 16:34:19.000000000","submitted":"2025-02-14 16:32:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"940344-openapi/role_assignments","meta_rev_id":"7f234c2249ea99e7b7af84241ecf1ce26c3d038b","_number":940344,"virtual_id_number":940344,"owner":{"_account_id":36482,"name":"Oria Weng","display_name":"0weng","email":"oweng@osuosl.org","username":"0weng"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-14 16:32:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"all":[{"value":2,"date":"2025-02-07 14:23:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":2,"date":"2025-02-07 16:40:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"all":[{"value":1,"date":"2025-02-14 14:28:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-28 22:52:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-01-29 01:01:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-29 21:40:38.000000000","updated_by":{"_account_id":36482,"name":"Oria Weng","display_name":"0weng","email":"oweng@osuosl.org","username":"0weng"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"REVIEWER"},{"updated":"2025-02-07 14:23:25.000000000","updated_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"}],"messages":[{"id":"fd290bd522f07397a90aed0fe87eda6bbc265c1c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36482,"name":"Oria Weng","display_name":"0weng","email":"oweng@osuosl.org","username":"0weng"},"date":"2025-01-28 22:01:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b06d562d7d6dc024de6107fe0aac9a7aa7397d65","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-28 22:52:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fce673c1a2134e41951c3f0e36289bec\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/80e496ab23814c66a995d1da133f2225 : FAILURE in 48m 37s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a380d65179dd4caa85281020c1c5455b : FAILURE in 48m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9afc031507573ef23e080cd2ccc0f02405d54fb2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36482,"name":"Oria Weng","display_name":"0weng","email":"oweng@osuosl.org","username":"0weng"},"date":"2025-01-28 23:13:55.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0f84def675b0eaf4b6f6f2ba7c55674a4936d6e2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-29 00:17:11.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e17b3f810cb94201ad4b4a0bc9f5565a\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/52ee022cd5a34e239b2b62ce62ee7e3c : TIMED_OUT in 1h 01m 02s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/18717a0986114f369cf4ca9d44f0df90 : TIMED_OUT in 1h 01m 05s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5ae60d3bf4d5b5ea4fe10d1d69bafdf38571d57a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-29 01:01:10.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1aef1b441e9b473081bf742e5011c6f6\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/10e2376e8d224938bfef8ce72d217bff : SUCCESS in 16m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af38c6da81514fbeaf07f1421ddafc7e : SUCCESS in 6m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ed5cced858d441709845c4bee01b3f28 : SUCCESS in 12m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0654eda236d747389f5f62fa6a745c80 : SUCCESS in 12m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c585974453b4b3aba25770cdea94103 : SUCCESS in 12m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/2c9619846342421d97a14009164e4d64 : SUCCESS in 1h 04m 15s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0b725db2ccd34a98b966fefc3896db2d : SUCCESS in 1h 41m 03s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/f74c9b9ff7644e0ca68c52ca896c5b6f : SUCCESS in 33m 23s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/d63eb5e604f0461c9bcda457ff0b9870 : SUCCESS in 46m 27s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d254bb7232d24d56831d993e569832b0 : FAILURE in 32m 54s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/ab7869fa440749d28f48d62e6bb5e514 : SUCCESS in 17m 34s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/cc3ce63830cf407a9f0b20a9bd860871 : SUCCESS in 11m 25s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/95d361f67e1c4086b6672f06ce392d02 : FAILURE in 14m 18s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5f62f10187eb4c5588c99933324f6bb7 : SUCCESS in 38m 04s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/e46b223f76ea4a4ba10a3c4a847355ae : FAILURE in 23m 09s (non-voting)\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/69b02a7874a94b8eac68468479309cd0 : POST_FAILURE in 5m 31s (non-voting)\n- keystone-dsvm-functional-oidc-federation https://zuul.opendev.org/t/openstack/build/deacb6606dce4432866935c6c06e977e : SUCCESS in 19m 17s","accounts_in_message":[],"_revision_number":2},{"id":"1d2960b26d94c8b03fcacf6cb0b88a20c84ebf4c","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2025-02-07 14:23:25.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"674bd0209e88d019b9a6bf33543fb207cadd0262","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2025-02-07 16:40:12.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"56341271eb7efc2fda7dd8f063b5a1470f117ed6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-07 18:23:46.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/74f84771d8c14b7fbb34f7ee5ed49612\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fbb4a2098a44442c85974c22d04a19da : SUCCESS in 24m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9c6ebeec68e49f089d334d0f6465c00 : SUCCESS in 7m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b1aecc46db614f3f86027c3c4b117a61 : SUCCESS in 11m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/85e0b4be476d43ed8ee5fec93a3ed227 : SUCCESS in 15m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63eb035bbf2243f4b1b428d74a198b16 : SUCCESS in 12m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/99b2b04b9c964b6abd633189a870cc33 : SUCCESS in 1h 00m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/87ba4f96aaa64a48b2745ef42a20cf8a : SUCCESS in 1h 40m 59s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/ba346b9e22ea4698a84c64df35d1a22d : SUCCESS in 24m 40s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/6c8e4bd378e142fcb0c11a63d8c131b1 : SUCCESS in 44m 53s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/aa37ec58fc3c48a98975bb7aa7c1fefb : FAILURE in 19m 34s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/a74686239df946eda605190d18f42944 : SUCCESS in 19m 14s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/e287fa50cbc44a588d0201d4ac9ee00e : SUCCESS in 10m 28s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/d6a3699137974959bf464b6570a30d80 : FAILURE in 12m 49s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d121595a9d334def8c24913dc781a7e5 : SUCCESS in 47m 51s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/a88f608628bf48ac882bbc185ba237bb : FAILURE in 31m 02s (non-voting)\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/b98962f47a574255bdba95f8193dc147 : FAILURE in 6m 53s (non-voting)\n- keystone-dsvm-functional-oidc-federation https://zuul.opendev.org/t/openstack/build/e8bc7833965f484eb8558c204afbc804 : SUCCESS in 27m 56s","accounts_in_message":[],"_revision_number":2},{"id":"56918db6e0425a327470de3d35014c9e790690ff","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-07 18:29:11.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d8f8dcdf26ff4a638d12962e650ee7ba\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b51dcebf0c4849da8c6f6b8593843f73 : TIMED_OUT in 1h 00m 58s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f46cd4c18bac4b3784f041085e6377e8 : TIMED_OUT in 1h 01m 01s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0b6164b7bbf2a785192e970815251b1dd80e7d97","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2025-02-14 14:28:03.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b7895175cee1c0202e48e0139b98ceb9a32a2b43","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-14 14:28:49.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"511c2fd6d07990d436171c487de0314fa5462d2c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-14 16:32:57.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/370c4331e34542e49d030c758cf0838c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4177c5572fa4523bfe144ee8ce6152e : SUCCESS in 7m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/965ff8c9d4414649b36248874971ca9a : SUCCESS in 11m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/871b25e0cdf845eca88157ac84cb87d5 : SUCCESS in 15m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4cd6c35f07cf421bbd835764b88dc7dc : SUCCESS in 12m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/aa33e4ed1f1c40bea4793e8d8b37c9d0 : SUCCESS in 49m 35s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/02278ca166e140938cfdb0360f78db26 : SUCCESS in 1h 58m 23s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/21d2924b3c824b7cb8934bbfa68183a2 : SUCCESS in 42m 36s\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/ed8dd52dbf0042399d02fc90a90b85d0 : SUCCESS in 30m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d2105c2bef824e49bbe5b9db2954d82c : SUCCESS in 1h 08m 14s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/cb6165d0edf54871a849427dc984e0e7 : FAILURE in 42m 52s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"de73a4b0f6042ee447e898ddc4107ca8a7cfd9c8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-14 16:32:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"7f234c2249ea99e7b7af84241ecf1ce26c3d038b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-14 16:34:19.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7407186f6ffc436c912157586af3aebe\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/412d981fe5b14d1393398d9e96e410c1 : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f3dfcad86e540c2a37b681ca5539e0b94de5238f","revisions":{"16492f0c0365adee89dba1f2fd9a044032605c00":{"kind":"REWORK","_number":1,"created":"2025-01-28 22:01:50.000000000","uploader":{"_account_id":36482,"name":"Oria Weng","display_name":"0weng","email":"oweng@osuosl.org","username":"0weng"},"ref":"refs/changes/44/940344/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/44/940344/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/44/940344/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/44/940344/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/44/940344/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/44/940344/1"}}},"commit":{"parents":[{"commit":"6965c2a24bdf8bc2f5adb7ff559148cffe5b2694","subject":"Merge \"extend docs explaining OIDC\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/6965c2a24bdf8bc2f5adb7ff559148cffe5b2694"}]}],"author":{"name":"0weng","email":"oweng@osuosl.org","date":"2025-01-28 22:00:14.000000000","tz":-480},"committer":{"name":"0weng","email":"oweng@osuosl.org","date":"2025-01-28 22:00:14.000000000","tz":-480},"subject":"Add JSON schema \u0026 validation for `role assignment`","message":"Add JSON schema \u0026 validation for `role assignment`\n\nChange-Id: I972efbd39bd47772dabcafee136b060fdd7a5437\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/16492f0c0365adee89dba1f2fd9a044032605c00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/16492f0c0365adee89dba1f2fd9a044032605c00"}]},"branch":"refs/heads/master"},"f3dfcad86e540c2a37b681ca5539e0b94de5238f":{"kind":"REWORK","_number":2,"created":"2025-01-28 23:13:55.000000000","uploader":{"_account_id":36482,"name":"Oria Weng","display_name":"0weng","email":"oweng@osuosl.org","username":"0weng"},"ref":"refs/changes/44/940344/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/44/940344/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/44/940344/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/44/940344/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/44/940344/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/44/940344/2"}}},"commit":{"parents":[{"commit":"6965c2a24bdf8bc2f5adb7ff559148cffe5b2694","subject":"Merge \"extend docs explaining OIDC\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/6965c2a24bdf8bc2f5adb7ff559148cffe5b2694"}]}],"author":{"name":"0weng","email":"oweng@osuosl.org","date":"2025-01-28 22:00:14.000000000","tz":-480},"committer":{"name":"0weng","email":"oweng@osuosl.org","date":"2025-01-28 23:05:16.000000000","tz":-480},"subject":"Add JSON schema \u0026 validation for `role assignment`","message":"Add JSON schema \u0026 validation for `role assignment`\n\nChange-Id: I972efbd39bd47772dabcafee136b060fdd7a5437\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/f3dfcad86e540c2a37b681ca5539e0b94de5238f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/f3dfcad86e540c2a37b681ca5539e0b94de5238f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
