)]}'
{"id":"openstack%2Fkeystone~943035","triplet_id":"openstack%2Fkeystone~master~I0c4fd741069f162706b5ba06fd2335efc6dbade4","project":"openstack/keystone","branch":"master","topic":"openapi/federation","attention_set":{},"removed_from_attention_set":{"36482":{"account":{"_account_id":36482,"name":"Oria Weng","display_name":"0weng","email":"oweng@osuosl.org","username":"0weng"},"last_update":"2026-03-06 18:07:29.000000000","reason":"Change was submitted"},"27900":{"account":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"last_update":"2026-02-17 09:56:56.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I0c4fd741069f162706b5ba06fd2335efc6dbade4","subject":"Add JSON schema and validation for `federation protocol`","status":"MERGED","created":"2025-02-28 17:17:51.000000000","updated":"2026-03-06 18:08:37.000000000","submitted":"2026-03-06 18:07:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"943035-openapi/federation","meta_rev_id":"dfcd9a34ed01b787b6bda3d6b7efdc42f3538091","_number":943035,"virtual_id_number":943035,"owner":{"_account_id":36482,"name":"Oria Weng","display_name":"0weng","email":"oweng@osuosl.org","username":"0weng"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-06 18:07:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"all":[{"value":2,"date":"2026-03-06 15:51:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":2,"date":"2026-02-17 09:56:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"all":[{"value":1,"date":"2026-03-06 15:51:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-28 18:57:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-02-28 19:32:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-13 21:59:08.000000000","updated_by":{"_account_id":36482,"name":"Oria Weng","display_name":"0weng","email":"oweng@osuosl.org","username":"0weng"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"REVIEWER"},{"updated":"2026-03-06 15:51:36.000000000","updated_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"}],"messages":[{"id":"ad1c2014a0b00c01d2b141f79af14a853962f0f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36482,"name":"Oria Weng","display_name":"0weng","email":"oweng@osuosl.org","username":"0weng"},"date":"2025-02-28 17:17:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f508375eb160ecdb54b4cce700f91cc039f1d441","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36482,"name":"Oria Weng","display_name":"0weng","email":"oweng@osuosl.org","username":"0weng"},"date":"2025-02-28 17:21:34.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c610e9b8359346533292defa42a52629e01347fe","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-28 18:57:12.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5f38a649c3174674bd9b1a22d9cfef0a\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ad841ba0e1ad4a80bea2d8aaeee20e59 : SUCCESS in 51m 06s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ec0fa115adad410bbbde450832fe20de : SUCCESS in 51m 01s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"924472dab60b3ab9d4df134c183cef9c94da3f05","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-28 19:32:31.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/66aed369cb0449eaa23102cbcdcd3d16\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/66fcf98f1bc74018b398e5bc42812732 : SUCCESS in 17m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ee883abbc0247c5a6c6ee390f25e166 : SUCCESS in 9m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/633deed5e130484cb86b63434b8aef2d : SUCCESS in 11m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b3d98421f4b242939a3bfc5c9bf4e023 : SUCCESS in 13m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/adfb266893424795a6fdacc2979fa478 : SUCCESS in 14m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/6b10e03b51314087a54064756ee20325 : SUCCESS in 56m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/191c3dc579fe4b748e1d6082e6328f0b : SUCCESS in 1h 55m 01s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/bdde204efe2f43e8bfab94b6a03a51e5 : SUCCESS in 27m 31s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/dd502cc474804229b4d88c90906f46f5 : SUCCESS in 37m 07s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/16a361a2e4db4a7095226b0c67e61f78 : FAILURE in 22m 55s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/2ddbac63c3e94298a799d1e2b573eb1d : FAILURE in 31m 47s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/e9581243464f453aae29acd6ac83bf4c : SUCCESS in 10m 51s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/85db0c7722c74932a973a0ced273d1a7 : FAILURE in 13m 49s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/247972de5a5246d3bfb24e7469d7a1d8 : SUCCESS in 1h 10m 17s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/f52aad0073c741ad8721ce449873d1e2 : FAILURE in 59m 16s (non-voting)\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/96fbda69d1d240e2af75c55961fde219 : POST_FAILURE in 6m 43s (non-voting)\n- keystone-dsvm-functional-oidc-federation https://zuul.opendev.org/t/openstack/build/f6726d0408284d148aaacb57a641a327 : FAILURE in 42m 10s","accounts_in_message":[],"_revision_number":2},{"id":"245fbef48fb14ec3985a275f75f8f7276d460430","author":{"_account_id":36482,"name":"Oria Weng","display_name":"0weng","email":"oweng@osuosl.org","username":"0weng"},"date":"2025-03-13 20:14:19.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"fef339ec21ef43410dbb14ac4b3873e0410670df","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36482,"name":"Oria Weng","display_name":"0weng","email":"oweng@osuosl.org","username":"0weng"},"date":"2025-03-13 20:14:27.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"88681995f8c1afa693a7376792e3706222cbe1e2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-13 20:33:13.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c3890abbab184975836d059aa4de0c31\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8a0de1de4dfd42a89aac80804ab9509d : SUCCESS in 17m 08s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5add905aaddf413eb03a964add4c48bf : SUCCESS in 16m 51s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5eac9f60369ade5d29251caa5cf724ff35853441","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-13 21:58:30.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c849809bbf0e4178be353fac17a90bed\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d3d3bd6a87d143a0b1214e1be5e5151f : SUCCESS in 15m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fda0f886c783481293006b48b5b4d2a8 : SUCCESS in 8m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/162f4113031a41a1a848d24a77b165a4 : SUCCESS in 9m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/833e30c02e204b8c8a37fe5bb4e6da31 : SUCCESS in 7m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41bcac0d6b5a4a8a938a6d7b513d5aea : SUCCESS in 11m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/ba371f84f7964ca8be300a21aa61ef6a : SUCCESS in 38m 40s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d7ccae6c75804bb9921c430a71919381 : SUCCESS in 1h 43m 21s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/849d7fbdc7aa4643b27dfacc27888421 : SUCCESS in 13m 44s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/530729233d6c4ceda8e284a9edb19f96 : SUCCESS in 46m 50s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/69370dfe46a14489a757440022921569 : FAILURE in 18m 20s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/b7f691d65c644caf87ef80d52bd7a16a : SUCCESS in 30m 40s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/0dc2c825ed3e446d8cb7decb6e1b64dd : SUCCESS in 8m 14s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/9bb82527b27648fd9c75d96e71b06fc8 : FAILURE in 13m 18s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/97e2fd5a643f4683bd43a112526caa15 : SUCCESS in 1h 10m 16s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/a86feda9a9e54caba41dcc9c4f26058e : FAILURE in 34m 22s (non-voting)\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/264fc33800154c0a9f51aa98e24dd819 : POST_FAILURE in 5m 33s (non-voting)\n- keystone-dsvm-functional-oidc-federation https://zuul.opendev.org/t/openstack/build/9c126dae11244d81aaaa5f6158010249 : SUCCESS in 19m 15s","accounts_in_message":[],"_revision_number":3},{"id":"786ea3adaf28a591bf95ee3f372da881b21d772b","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2025-03-20 11:17:03.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6282844d53e4aba6f6fa013fbf89bd5a4b68dfe0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-20 11:37:22.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/515c9c3b24064d0483ec3c6109d5737e\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/00163a0c515c4305a791c994e4020c63 : SUCCESS in 17m 42s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/480d164d60434480914b1dd7b24efcfd : SUCCESS in 17m 51s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"3318adad1f0c9ae01bbd1387d5eac7d1acc9d033","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-20 12:10:01.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/287fd774ad9646199728027391cb1261\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/29d274c79b9b43bd945169887d85c6f6 : SUCCESS in 17m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aaaa11f245014e198c7471b7c8865b5a : SUCCESS in 6m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eefe70e1b4124a7abc72ed37188b11bb : SUCCESS in 11m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/aed18e070cee4df18d622a9278c37c0a : SUCCESS in 11m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92b13a9f21314c84be81a82fe52708ca : SUCCESS in 11m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/b9501ae3cd1d4f3bbc888bdbc31e7227 : SUCCESS in 31m 45s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6fcbf560ec8c4c67b110b545282428d6 : SUCCESS in 51m 51s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/e411b33b35954c08b80d28133e9d7d42 : SUCCESS in 21m 30s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/29470a9245ca4fefa7a12448df54bb97 : SUCCESS in 33m 50s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/36a035f2c5584bb49e2bf87153fe8f83 : FAILURE in 17m 12s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/2dc81d7ff0a24eae9b575b964d3ce582 : SUCCESS in 10m 06s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/53959f72a11640f68b55e410f182c090 : SUCCESS in 7m 31s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/9b3206612d574b45b3e4083ed937bfed : FAILURE in 10m 09s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0b5c5934e9254873b637e6f2875eab29 : SUCCESS in 50m 56s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/f117e4f82f5c4d70add0e34bde8375be : FAILURE in 27m 29s (non-voting)\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/ec1e2099a709452a90b38424353b5714 : SUCCESS in 5m 59s (non-voting)\n- keystone-dsvm-functional-oidc-federation https://zuul.opendev.org/t/openstack/build/0bca9f9722d243f692eb98c036dbf0a6 : SUCCESS in 31m 29s","accounts_in_message":[],"_revision_number":3},{"id":"952eee157a3ecedf9c405884dadf9d19b5427dff","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2025-04-25 13:53:56.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"86707e06db6e856902be7b5156ac028763442398","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-25 14:52:06.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/312002a76f8a4d82affadbfc64d4f789\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/cc296b6c84a143809e953801d5778aaa : SUCCESS in 30m 40s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b1f2dc6f05804bcca67276a6cbf69218 : SUCCESS in 36m 05s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/529d2ee3008046caad4ee62f09fc7929 : SUCCESS in 30m 23s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5df20b3de1bd312ed11eb79002412cc826c85669","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-25 15:34:23.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/349541bcba0d43078d4426c6f839bcad\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fa74ed6021a74a8e93a862d27a5d1fdc : SUCCESS in 10m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/814210205be44cb3ae3ae3a3c4ce3dcb : SUCCESS in 6m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0ba299790ca446ca82472ead4c43595c : SUCCESS in 10m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3affdc334d5e4689880e6b875aa8dc46 : SUCCESS in 11m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/970d8926db3b4b3aa9cf302943741aa6 : SUCCESS in 13m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3dbe6b8206a9415bba6d68945f5811d5 : SUCCESS in 12m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/bd19e1c645f04c4d905c05ad096a567e : SUCCESS in 51m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ab8d6cb0cc644383ad0ba18a7b6c9667 : SUCCESS in 1h 35m 48s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/0b6dd1bacc4141b6b5bbb342e8486cec : SUCCESS in 24m 27s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/d5d6b978b2424a58a5fa16383fd39ec6 : SUCCESS in 35m 06s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6bf08c91723f4937b0da61e3b0838e3c : FAILURE in 26m 33s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/c890bacabb9b4f1c8d6fdb1aa4cce48b : SUCCESS in 24m 40s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/50a09eac9bb24954a53e5f31b1159fc8 : SUCCESS in 13m 27s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/c8de812ed21d45228ba4f878a4645adb : FAILURE in 8m 16s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2f3d6f5fad284839ba739927a0a3369c : SUCCESS in 55m 11s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/620a35fb39864b1aae0be7c2a8d5d971 : FAILURE in 15m 45s (non-voting)\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/5a1a5b698b754082a5fd12c5c10fd79e : SUCCESS in 6m 05s (non-voting)\n- keystone-dsvm-functional-oidc-federation https://zuul.opendev.org/t/openstack/build/0db411aeb92440989c4147db31c735fa : SUCCESS in 28m 55s","accounts_in_message":[],"_revision_number":3},{"id":"e7eedb670cdddb76b22249d7bec30f56fee63f7e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36482,"name":"Oria Weng","display_name":"0weng","email":"oweng@osuosl.org","username":"0weng"},"date":"2025-11-12 18:36:26.000000000","message":"Uploaded patch set 4: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"f2b485bb35e2c6604f5a2e4ff2eaae21c80b86b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-12 19:37:11.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/93b55c2913394dc681fbcdd915c29b91\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ae661d4c7cd4415e8c0d2c51d37da7e4 : SUCCESS in 14m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5bb362405a304b5da633c9d7df7cf51a : SUCCESS in 6m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bdeb38841a17403f9d1e8b7a35733211 : SUCCESS in 11m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8b0bac83684d4e89ae215ec771135ec0 : SUCCESS in 11m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f81a2bfe8a464c599a30badfe3cf67fe : SUCCESS in 14m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/2c4209cdf1504050935419c0ecc27a38 : SUCCESS in 51m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f37928f679aa4223969ab9c8b83a39ba : SUCCESS in 58m 35s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/bb933ca8f8e0436a9833b3f148b9e73b : SUCCESS in 11m 11s (non-voting)\n- keystone-tempest https://zuul.opendev.org/t/openstack/build/bd517232bea04f338c018bb81200fbfd : SUCCESS in 19m 28s\n- keystone-tempest-fips https://zuul.opendev.org/t/openstack/build/c3e18cd6da1f42a7b3ef852a66e0cd62 : FAILURE in 19m 31s (non-voting)\n- keystone-tempest-federation https://zuul.opendev.org/t/openstack/build/60381735ad464aaa8cf033611f02f5ff : FAILURE in 18m 04s (non-voting)\n- keystone-tempest-federation-k2k https://zuul.opendev.org/t/openstack/build/bb31a83bc3df47cd9a3f856d6f48bcc6 : SUCCESS in 29m 59s\n- keystone-tempest-oidc-federation https://zuul.opendev.org/t/openstack/build/7e584414bd174c34b7c85965c7f45efb : SUCCESS in 30m 42s\n- keystone-tempest-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/7e904a95d2504088a81b9a8d9aabef3f : SUCCESS in 37m 50s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/46887e6bbdc446009f1f1126a5dd05c2 : SUCCESS in 55m 32s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/637738a1e4f6478ab30d32c350371abb : FAILURE in 38m 22s (non-voting)\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/75f8e1b13e454fae94b2e983c410fc50 : SUCCESS in 7m 11s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"9ad552b19186950ae2ec4555953192b858d09495","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-12 22:09:43.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ef44945b5e88475597f7a6de16ccd944\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/ec4b83632aa54081a33aea4e827bc5eb : SUCCESS in 28m 36s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/95123f78b23a416abf5942a385439ac4 : SUCCESS in 18m 35s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b752691292e341b2918e747d6c71dea8 : SUCCESS in 20m 19s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"e11d1fc52c653844978988cb61fc7933f3838aa8","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2026-02-17 09:56:56.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"8f9619d1df02508e9698a84b66158aa98255726b","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2026-03-06 15:51:36.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"f2143e24ed1ef07f9fc448ec64a5c1f07fc25481","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-06 15:51:58.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"dab13557147eddee64549c90494e933be85f08bd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-06 18:07:29.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cde0d0e80a5649e89f1ad2f67a0133ef\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f23e029f9414a7dabbb906e95ebafce : SUCCESS in 4m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4e60b918128f4df497e46bdabefe368f : SUCCESS in 12m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/592060e6218d46a1b446827f71bb15a3 : SUCCESS in 12m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/626e61ae37e3410baba135e722f9073e : SUCCESS in 12m 13s\n- grenade https://zuul.opendev.org/t/openstack/build/6ee8858d5d0d4f008bd184edf6a30b7c : SUCCESS in 30m 36s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1e7cc6d9f29e4245bc601d5ab29139c2 : SUCCESS in 2h 10m 19s\n- keystone-tempest https://zuul.opendev.org/t/openstack/build/f7796cbd2a52413db0397519a2fd49d7 : SUCCESS in 33m 14s\n- keystone-tempest-federation-k2k https://zuul.opendev.org/t/openstack/build/fc62acfd78ed4c7798149216cd9cb1f4 : SUCCESS in 16m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c49a37d32db7443b98efdfb356dbaf49 : SUCCESS in 1h 01m 30s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/a29f6ed7342b4713a232e8e96a98b0a1 : FAILURE in 35m 49s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"db7ba2b169caa4ce1d8f4000ba0aacc5a20e5698","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-06 18:07:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"dfcd9a34ed01b787b6bda3d6b7efdc42f3538091","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-06 18:08:37.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/03b25a907b4846d181189705b5690d6f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/afe0212b2cb64097bf095c8f8b9ca370 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"057a3c3c092502a3d0e331b100819319c195fe8e","revisions":{"aa265950a958ea50e69e16dadbf94067eaf9b4fa":{"kind":"REWORK","_number":1,"created":"2025-02-28 17:17:51.000000000","uploader":{"_account_id":36482,"name":"Oria Weng","display_name":"0weng","email":"oweng@osuosl.org","username":"0weng"},"ref":"refs/changes/35/943035/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/35/943035/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/35/943035/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/35/943035/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/35/943035/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/35/943035/1"}}},"commit":{"parents":[{"commit":"463b6a9b3fdb014d2778259c201a00040308c01b","subject":"Merge \"Add JSON schema to `identity provider` and validation decorators to identity provider resource.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/463b6a9b3fdb014d2778259c201a00040308c01b"}]}],"author":{"name":"0weng","email":"oweng@osuosl.org","date":"2025-02-28 17:17:01.000000000","tz":-480},"committer":{"name":"0weng","email":"oweng@osuosl.org","date":"2025-02-28 17:17:01.000000000","tz":-480},"subject":"Add JSON schema and validation for `federation protocol`","message":"Add JSON schema and validation for `federation protocol`\n\nChange-Id: I0c4fd741069f162706b5ba06fd2335efc6dbade4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/aa265950a958ea50e69e16dadbf94067eaf9b4fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/aa265950a958ea50e69e16dadbf94067eaf9b4fa"}]},"branch":"refs/heads/master"},"009d444838a4d315fd4afe3db8b820f9b9d87dd5":{"kind":"REWORK","_number":2,"created":"2025-02-28 17:21:34.000000000","uploader":{"_account_id":36482,"name":"Oria Weng","display_name":"0weng","email":"oweng@osuosl.org","username":"0weng"},"ref":"refs/changes/35/943035/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/35/943035/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/35/943035/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/35/943035/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/35/943035/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/35/943035/2"}}},"commit":{"parents":[{"commit":"463b6a9b3fdb014d2778259c201a00040308c01b","subject":"Merge \"Add JSON schema to `identity provider` and validation decorators to identity provider resource.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/463b6a9b3fdb014d2778259c201a00040308c01b"}]}],"author":{"name":"0weng","email":"oweng@osuosl.org","date":"2025-02-28 17:17:01.000000000","tz":-480},"committer":{"name":"0weng","email":"oweng@osuosl.org","date":"2025-02-28 17:21:15.000000000","tz":-480},"subject":"Add JSON schema and validation for `federation protocol`","message":"Add JSON schema and validation for `federation protocol`\n\nChange-Id: I0c4fd741069f162706b5ba06fd2335efc6dbade4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/009d444838a4d315fd4afe3db8b820f9b9d87dd5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/009d444838a4d315fd4afe3db8b820f9b9d87dd5"}]},"branch":"refs/heads/master"},"c052e3bb265ff53c023966a62a8fb3efa73f04bd":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2025-03-13 20:14:27.000000000","uploader":{"_account_id":36482,"name":"Oria Weng","display_name":"0weng","email":"oweng@osuosl.org","username":"0weng"},"ref":"refs/changes/35/943035/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/35/943035/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/35/943035/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/35/943035/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/35/943035/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/35/943035/3"}}},"commit":{"parents":[{"commit":"bd2b97a0412b8ea01ef47fd8c0c4908ca9fcee8a","subject":"Merge \"Replace tabs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/bd2b97a0412b8ea01ef47fd8c0c4908ca9fcee8a"}]}],"author":{"name":"0weng","email":"oweng@osuosl.org","date":"2025-02-28 17:17:01.000000000","tz":-480},"committer":{"name":"Oria Weng","email":"oweng@osuosl.org","date":"2025-03-13 20:14:27.000000000","tz":0},"subject":"Add JSON schema and validation for `federation protocol`","message":"Add JSON schema and validation for `federation protocol`\n\nChange-Id: I0c4fd741069f162706b5ba06fd2335efc6dbade4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/c052e3bb265ff53c023966a62a8fb3efa73f04bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/c052e3bb265ff53c023966a62a8fb3efa73f04bd"}]},"branch":"refs/heads/master","description":"Rebase"},"057a3c3c092502a3d0e331b100819319c195fe8e":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2025-11-12 18:36:26.000000000","uploader":{"_account_id":36482,"name":"Oria Weng","display_name":"0weng","email":"oweng@osuosl.org","username":"0weng"},"ref":"refs/changes/35/943035/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/35/943035/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/35/943035/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/35/943035/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/35/943035/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/35/943035/4"}}},"commit":{"parents":[{"commit":"bd2b97a0412b8ea01ef47fd8c0c4908ca9fcee8a","subject":"Merge \"Replace tabs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/bd2b97a0412b8ea01ef47fd8c0c4908ca9fcee8a"}]}],"author":{"name":"0weng","email":"oweng@osuosl.org","date":"2025-02-28 17:17:01.000000000","tz":-480},"committer":{"name":"0weng","email":"oweng@osuosl.org","date":"2025-11-12 18:36:18.000000000","tz":-480},"subject":"Add JSON schema and validation for `federation protocol`","message":"Add JSON schema and validation for `federation protocol`\n\nChange-Id: I0c4fd741069f162706b5ba06fd2335efc6dbade4\nSigned-off-by: 0weng \u003coweng@osuosl.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/057a3c3c092502a3d0e331b100819319c195fe8e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/057a3c3c092502a3d0e331b100819319c195fe8e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
