)]}'
{"id":"openstack%2Fkeystone~943101","triplet_id":"openstack%2Fkeystone~master~Ic81e4761f6a18f0b6a7770ad0e662510beb2c531","project":"openstack/keystone","branch":"master","attention_set":{},"removed_from_attention_set":{"7414":{"account":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"last_update":"2025-03-07 17:36:32.000000000","reason":"Change was submitted"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-03-07 17:36:32.000000000","reason":"Change was submitted"},"14250":{"account":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"last_update":"2025-03-07 14:32:26.000000000","reason":"\u003cGERRIT_ACCOUNT_14250\u003e replied on the change","reason_account":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}}},"hashtags":[],"change_id":"Ic81e4761f6a18f0b6a7770ad0e662510beb2c531","subject":"Replace tabs","status":"MERGED","created":"2025-03-01 16:02:15.000000000","updated":"2025-03-07 17:37:50.000000000","submitted":"2025-03-07 17:36:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"943101","meta_rev_id":"4f32e16cb82c6e32d06fda50dacd1098e2d51677","_number":943101,"virtual_id_number":943101,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-07 17:36:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"all":[{"value":2,"date":"2025-03-07 14:32:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":2,"date":"2025-03-01 16:19:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-03-01 21:27:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"all":[{"value":1,"date":"2025-03-07 14:32:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"}],"reviewers":{"REVIEWER":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-01 16:19:05.000000000","updated_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"REVIEWER"},{"updated":"2025-03-01 17:06:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-03-01 17:48:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-01 21:27:08.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2025-03-01 21:27:13.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"},{"updated":"2025-03-03 16:57:39.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"}],"messages":[{"id":"211d2fa17a33844c85f06521a79e25d403453d21","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-03-01 16:02:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c3d9fa9e1894891be437a21404923e56dab65978","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2025-03-01 16:19:05.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5a36bfbeb104993a77ffe7e2d5b9d4e3c2f974b8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-01 17:06:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/202c6b9a02c74b1a9e4fad2cd383c2bf\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0ed194e2c8734e0b96b6871c0d6b07a6 : TIMED_OUT in 1h 00m 56s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0c708c885eeb4b5393d62156413b5e45 : TIMED_OUT in 1h 01m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5dce6e256e2570bcc0b61aae84e972a87881dc41","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-01 17:48:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a6a9fb4ebf264b30a003780645a4b7f5\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e87a015f626946aeb48407d32feda8c7 : SUCCESS in 17m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99c14da74c2143948923bf46f71867cf : SUCCESS in 6m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/350ba10884bd4a57b84f777730388806 : SUCCESS in 12m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8923be440a5c40c286de82471071dc9c : SUCCESS in 15m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13905dbedd6a417a9e88e8a1e6c6924d : SUCCESS in 10m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/e44d9fae84d3436287778d7c98c893fd : SUCCESS in 46m 42s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/73d1d04d946f4a1a8ab3b8d796a025a8 : SUCCESS in 1h 40m 31s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/3c8c80866d9d4ba19d34d50a555f1419 : SUCCESS in 24m 26s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/b86f174e85954c4aadad541cd5875cb6 : SUCCESS in 56m 03s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b104ac3ade914fbb865d56635f0656c6 : FAILURE in 20m 50s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/68bda50971bd46319b63f810999c18a3 : SUCCESS in 31m 26s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/b0134c4a231e4856be9b4ea6950219f7 : SUCCESS in 10m 13s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/437efe67aec740e1b2a574ddd5b5ca7c : FAILURE in 28m 54s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ed6bda8428b246cab7a568a3ba0cabeb : SUCCESS in 59m 08s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/6bcdc1f407b942548a8887c505366190 : FAILURE in 32m 11s (non-voting)\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/f82ed891133943bab2b4858554a9e7ee : SUCCESS in 6m 21s (non-voting)\n- keystone-dsvm-functional-oidc-federation https://zuul.opendev.org/t/openstack/build/1f474c37762443a2a238a3bd426fbd3b : SUCCESS in 40m 38s","accounts_in_message":[],"_revision_number":1},{"id":"ca9a475a1c4082799265d76641011c65b870c770","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-03-01 21:27:08.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"d05198e5a67e3026e9ada5c6e108c0612e08dae6","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2025-03-07 14:32:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"54784b4de1e65248e6fe47cf3cdb46d149fe5a01","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-07 14:34:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ceb445ccefffeb9abb7c272ba1d95b6b505489bd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-07 17:36:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7fa6c25e9e8b4bce8df13d8da495469c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf161c180ada408aba519526c8468673 : SUCCESS in 9m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b019d259ee5747bca662194db15d93b5 : SUCCESS in 13m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7ff02616636b4f5899e51231755e5ec9 : SUCCESS in 11m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45b571a139c248c982a96a9eeaaf73bb : SUCCESS in 10m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/1328bba09e45493db10feb271aeb3351 : SUCCESS in 50m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6de2cc883901410ea907c9ba065105d9 : SUCCESS in 1h 50m 07s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/bbe0672d8b594b2fa4e62b8c6c835f91 : SUCCESS in 17m 16s\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/11fbcabf45de457aa02d9dbbdd79c61a : SUCCESS in 34m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/392fc8f1613e4f62b661e2440c024b58 : SUCCESS in 53m 37s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/bbeb5c0a6d8840d8893712f9efbc1a14 : FAILURE in 8m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c895b15062f160499f63c583df2c54fde3e4a855","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-07 17:36:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"4f32e16cb82c6e32d06fda50dacd1098e2d51677","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-07 17:37:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bb9dc2815b154689b7aad7b1e71da02a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac4faac3ea9c4448bbb0ad7ab2484204 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"94535666d35a4b95967201066d24fc7ab38aefd5","revisions":{"94535666d35a4b95967201066d24fc7ab38aefd5":{"kind":"REWORK","_number":1,"created":"2025-03-01 16:02:15.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/01/943101/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/01/943101/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/01/943101/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/01/943101/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/01/943101/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/01/943101/1"}}},"commit":{"parents":[{"commit":"ad93445fdc94aedc659ba4d629b86524f97d134d","subject":"Use default member role in federation tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/ad93445fdc94aedc659ba4d629b86524f97d134d"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-03-01 16:01:34.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-03-01 16:01:34.000000000","tz":540},"subject":"Replace tabs","message":"Replace tabs\n\nIn the devstack plugin files \"4 spaces\" is commonly used instead.\n\nChange-Id: Ic81e4761f6a18f0b6a7770ad0e662510beb2c531\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/94535666d35a4b95967201066d24fc7ab38aefd5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/94535666d35a4b95967201066d24fc7ab38aefd5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
