)]}'
{"id":"openstack%2Fkeystone~944230","triplet_id":"openstack%2Fkeystone~stable%2F2025.1~I23bf9b0d14b1f7b3ab5982a2f213844e066dcf34","project":"openstack/keystone","branch":"stable/2025.1","topic":"create-2025.1","hashtags":[],"change_id":"I23bf9b0d14b1f7b3ab5982a2f213844e066dcf34","subject":"Update .gitreview for stable/2025.1","status":"MERGED","created":"2025-03-13 13:34:05.000000000","updated":"2025-03-21 11:42:22.000000000","submitted":"2025-03-21 11:41:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"944230-create-2025.1","meta_rev_id":"228d3ba36ddb52763520344dc9d76d43e103d31f","_number":944230,"virtual_id_number":944230,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-21 11:41:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2025-03-21 09:31:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2025-03-14 14:16:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2025-03-21 09:31:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-13 13:57:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-14 14:16:09.000000000","updated_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"},{"updated":"2025-03-21 09:31:48.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"03c9ed3b933372245c4524a8479c21ad66e47dc8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2025-03-13 13:34:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d6abc57ed2b7a900519bf625096861fbbce72ba3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-13 13:57:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6afb779fe3a24eddb327de6c82f91a9e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1e3b99526f8e43c98ee0ddb3a41527bb : SUCCESS in 15m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b047425677844d478bba32ffa8c6134e : SUCCESS in 6m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0549a24e0be1487880dcdfd8be20f72a : SUCCESS in 10m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/da52a1348e27492e887b7668e6eb2651 : SUCCESS in 12m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4fb4b4ccd5654335b35961594b07830b : SUCCESS in 6m 39s\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/b2d9297765034d289f53e421aeb4f482 : SUCCESS in 6m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"334101900b5674dc6316c96e8eda4dee4bef192a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-13 14:04:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1080be2d9e154fc1b9d4232d95d4585f\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fc7bc583a07c4bfc97ad5b44005e4e7f : SUCCESS in 26m 24s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c9a16417ad5e4577bd7179c6872b9c8a : SUCCESS in 22m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"13448a52ddff1c301784b5496dcc8493ea5963bd","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2025-03-14 14:16:09.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c06c8e455c0b2b7c6797e6414870f7d1fe7c98d5","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-03-21 09:31:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e98a5e61f46098f59e9b3b6b2f864bc47e970926","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-21 09:32:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"94e7f39c72f53e0700596e276f1b1e595ea63fda","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-21 11:41:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/636030993ecc4e15bce8b4981f9dcb1a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db96ff86612741dfa8eaf446c2b2640e : SUCCESS in 3m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/320284eb222b4c249e4fa941039c2965 : SUCCESS in 12m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/acbe7da6d8ca420a948f1ca511b52593 : SUCCESS in 14m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9eb7245b9d24829b81d269b5ba2f803 : SUCCESS in 23m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/f42d045fc4df4a079cdd42a1e1900723 : SUCCESS in 29m 38s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/33aa0db6048249738e76b901200211ec : SUCCESS in 1h 50m 46s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/6275c31240b3477ab5af85f27cc72265 : SUCCESS in 30m 08s\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/7ca17eb79aa54782875660ad6bf65df5 : SUCCESS in 20m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6bfef5fab3cb4fcf99deeb4dd9565163 : SUCCESS in 1h 03m 55s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/9cb05e9348394d23b922c4ce7a364367 : FAILURE in 35m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6c758ab33e9b4229301241da0700ed931f698714","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-21 11:41:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"228d3ba36ddb52763520344dc9d76d43e103d31f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-21 11:42:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0790e6306a9c4234aec8506003d92b7d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb95050d4a2f4005b8e0262477fd7475 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ba29ce20ff60765aa83f462af8f7424828b5e4a1","revisions":{"ba29ce20ff60765aa83f462af8f7424828b5e4a1":{"kind":"REWORK","_number":1,"created":"2025-03-13 13:34:05.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/30/944230/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/30/944230/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/30/944230/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/30/944230/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/30/944230/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/30/944230/1"}}},"commit":{"parents":[{"commit":"bd2b97a0412b8ea01ef47fd8c0c4908ca9fcee8a","subject":"Merge \"Replace tabs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/bd2b97a0412b8ea01ef47fd8c0c4908ca9fcee8a"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2025-03-13 13:34:03.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2025-03-13 13:34:03.000000000","tz":0},"subject":"Update .gitreview for stable/2025.1","message":"Update .gitreview for stable/2025.1\n\nChange-Id: I23bf9b0d14b1f7b3ab5982a2f213844e066dcf34\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/ba29ce20ff60765aa83f462af8f7424828b5e4a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/ba29ce20ff60765aa83f462af8f7424828b5e4a1"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
