)]}'
{"id":"openstack%2Fkeystone~945505","triplet_id":"openstack%2Fkeystone~master~I17c4e5116ded437a8561b5f721899cdc653c352e","project":"openstack/keystone","branch":"master","topic":"bug/2104185","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-03-28 16:07:57.000000000","reason":"Change was submitted"},"27900":{"account":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"last_update":"2025-03-26 18:43:27.000000000","reason":"\u003cGERRIT_ACCOUNT_27900\u003e replied on the change","reason_account":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}}},"hashtags":[],"change_id":"I17c4e5116ded437a8561b5f721899cdc653c352e","subject":"api: Correct query string schema for access rules API","status":"MERGED","created":"2025-03-25 18:11:26.000000000","updated":"2025-03-28 16:07:57.000000000","submitted":"2025-03-28 16:07:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"945505-bug/2104185","meta_rev_id":"5f6da5163290ca9401396837a41620ce41a10de9","_number":945505,"virtual_id_number":945505,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-28 16:07:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"all":[{"value":2,"date":"2025-03-28 14:17:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":2,"date":"2025-03-26 18:43:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"all":[{"value":1,"date":"2025-03-28 14:17:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}],"CC":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-25 18:27:49.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"REVIEWER"},{"updated":"2025-03-25 19:12:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-03-25 19:50:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-25 20:39:34.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2025-03-28 14:17:06.000000000","updated_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"}],"messages":[{"id":"dbb7ad32c0289bc94b8d1e98e7c191a0f4250345","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-03-25 18:11:26.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dadb2ab4a0c6a107f83886e71d644fff1552c153","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-03-25 18:25:10.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"73254cc7b7e59303f7c0e86924534b475f46f4b4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-25 19:12:15.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f595f39ca154f2ab08b03641e133e78\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0bfb6d8f05a94da18b49a59d1337ad46 : SUCCESS in 20m 21s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0f36083b32f749369f037be664245e88 : SUCCESS in 29m 13s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1b2cf2c5660c0d79132858bf8172919b47d2852b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-25 19:50:48.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/46bf8a7268b447f7a058e51cdecc522e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/04e4efad34214598ad8a7fb9b48d5942 : SUCCESS in 10m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0599f7eb60e044c6928945c6daa69cdc : SUCCESS in 8m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/660fbb603f2f4313b928b08096184e38 : SUCCESS in 11m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/906e0ec84d9a49efb6f033aa3abe53ee : SUCCESS in 13m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b02c3d859f6e4084ac2447a0b353b8e2 : TIMED_OUT in 30m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/e66f9f0138d6488eb77d14c64a12e876 : SUCCESS in 54m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/86daeb316093463eb3f04fa512875e0a : SUCCESS in 58m 01s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/09429729cc9949a1bccfd3d1f88a52e5 : SUCCESS in 27m 18s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/b8c11c0780464675812bb28cb046f098 : SUCCESS in 34m 50s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e7d934ce6d8147c18eae960986a31a89 : FAILURE in 22m 38s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/9b7384dd46dd48a4bd67676d27b1c022 : SUCCESS in 22m 44s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/3342f8ffd6934e67bb62df301a9c1837 : SUCCESS in 11m 04s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/5e5abc583bc64e9685ddd84fbaf7c601 : FAILURE in 16m 59s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bc06c7873c8a41949a84a492fbfecb68 : SUCCESS in 57m 06s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/128f2a63ac5148d88c1c9577b79b433b : FAILURE in 38m 33s (non-voting)\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/4f3ed37e4b334911a2bf6c20db2141b9 : SUCCESS in 6m 41s (non-voting)\n- keystone-dsvm-functional-oidc-federation https://zuul.opendev.org/t/openstack/build/735d8f0f49bb42ee8e420e330bed18e8 : SUCCESS in 14m 04s","accounts_in_message":[],"_revision_number":2},{"id":"41283570b78b01358dba5bf4601642fcc22fa42b","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-03-25 20:39:34.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"db062b0b887db1d8c5b2e68c266e128249249e56","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-25 21:06:52.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f1d3629617964bd6b3eebb1c8c915982\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fe19b2dea80b421ca3aa8d4ebd0cda1f : SUCCESS in 17m 54s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f7ab6491455d4c7f905daf07326a018d : SUCCESS in 25m 37s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"abbeb33ebe23bce26afd6b5c4bd4d5bfd607b87a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-25 22:07:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab83618434f74e18bd609984941a0a98\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3782b5249d9c4d85967096b1db2df312 : SUCCESS in 16m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/acb3a7b1a9994656bef02071370b2f8e : SUCCESS in 6m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/408859e97b3d4180bfa2eb98fdda36d7 : SUCCESS in 11m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/04e5e5cfe3a94447af6d65f020620e60 : SUCCESS in 12m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f01639603bba407eb5cb63712fb31b02 : SUCCESS in 11m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/1f7eb43a016d4cfeb26a13945ace8272 : SUCCESS in 56m 09s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3fbcea2ecd2a4a799d6f6f62180cfbc3 : SUCCESS in 1h 25m 52s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/593ad8113fff4df4bf2d43ef478e659a : SUCCESS in 27m 44s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/7d703f39521740f894de72f2b2d2bf82 : SUCCESS in 15m 33s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7a0a309314304aabae8036d8cd0decb7 : FAILURE in 12m 23s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/c242838f6f814f4facf2165c91633e2c : SUCCESS in 22m 06s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/7782687ea06e499ba40cb65dec557df4 : SUCCESS in 13m 14s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/ac97ee90170349108c9e5d6da9639872 : FAILURE in 14m 43s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c3c9afcb314a48d4b6877717e8378a53 : SUCCESS in 52m 33s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/1ae132f4d4514d628a159bd709d72ca0 : FAILURE in 30m 26s (non-voting)\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/4a2759901ec347bda5d0909ee663a632 : SUCCESS in 7m 23s (non-voting)\n- keystone-dsvm-functional-oidc-federation https://zuul.opendev.org/t/openstack/build/26dc466a0ba84eb280617cd57456a493 : SUCCESS in 21m 21s","accounts_in_message":[],"_revision_number":2},{"id":"e2672b3f73f9234cc54de7fbab3291cd22cf4115","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2025-03-26 18:43:27.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"1a06b546ec0b25b4e244779366fd32eaefae811d","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2025-03-28 14:17:06.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"439718d1b7205c337e436e442ec7a40d9a097c15","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-28 14:17:48.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"577533de837a24e11111de7478d8de316182f034","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-28 16:07:57.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0843e285e8c3428c82eb8ad8a8b781ab\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6226d7a8948e489b9677c00cd36d0750 : SUCCESS in 6m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e52b5af48a9745c6a5535c578a06dcd7 : SUCCESS in 12m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/698f59428cea4656b8c57169dbfec25f : SUCCESS in 11m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40a6e62c3bdf42dcb6bf23daabdbda1a : SUCCESS in 12m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/795f4ef3009c4cc599ab62559fe73466 : SUCCESS in 54m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/68c18e85513749018802d0bd6f3b4091 : SUCCESS in 54m 16s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/5c215a3ff60f49f29c890c6235953d2c : SUCCESS in 24m 57s\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/ce1c9138c68f458583dc04d3902b3206 : SUCCESS in 10m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4da921a6b6ed440a9c17395828a66c8a : SUCCESS in 51m 22s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/c7e16d5fa1d04085853959c3c42c0d33 : FAILURE in 16m 25s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5f6da5163290ca9401396837a41620ce41a10de9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-28 16:07:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"187c1af5068997ab1ebd29cb7deaa8138b496af6","revisions":{"18acd3202a84f873e40c3d100e86742c8c5e72f5":{"kind":"REWORK","_number":1,"created":"2025-03-25 18:11:26.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/05/945505/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/05/945505/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/05/945505/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/05/945505/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/05/945505/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/05/945505/1"}}},"commit":{"parents":[{"commit":"1f877ff424406aa7add2972a7c81894ad5a4f344","subject":"api: Don\u0027t restrict unknown querystring parameters yet","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/1f877ff424406aa7add2972a7c81894ad5a4f344"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-03-25 17:55:46.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-03-25 18:11:20.000000000","tz":0},"subject":"api: Correct query string schema for access rules API","message":"api: Correct query string schema for access rules API\n\nThe presence of a call to \u0027build_driver_hints\u0027 in the method indicates\nthat filters are permitted for this API, even if they\u0027re not currently\ndocumented.\n\nWe also rename the variables and fix a docstring to avoid confusion\nabout what the schemas are intended for.\n\nChange-Id: I17c4e5116ded437a8561b5f721899cdc653c352e\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/18acd3202a84f873e40c3d100e86742c8c5e72f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/18acd3202a84f873e40c3d100e86742c8c5e72f5"}]},"branch":"refs/heads/master"},"187c1af5068997ab1ebd29cb7deaa8138b496af6":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-03-25 18:25:10.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/05/945505/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/05/945505/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/05/945505/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/05/945505/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/05/945505/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/05/945505/2"}}},"commit":{"parents":[{"commit":"05cc3d1903c1bfe669a1032746aa9e7b197238e9","subject":"api: Don\u0027t restrict unknown querystring parameters yet","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/05cc3d1903c1bfe669a1032746aa9e7b197238e9"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-03-25 17:55:46.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-03-25 18:24:28.000000000","tz":0},"subject":"api: Correct query string schema for access rules API","message":"api: Correct query string schema for access rules API\n\nThe presence of a call to \u0027build_driver_hints\u0027 in the method indicates\nthat filters are permitted for this API, even if they\u0027re not currently\ndocumented.\n\nWe also rename the variables and fix a docstring to avoid confusion\nabout what the schemas are intended for.\n\nChange-Id: I17c4e5116ded437a8561b5f721899cdc653c352e\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/187c1af5068997ab1ebd29cb7deaa8138b496af6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/187c1af5068997ab1ebd29cb7deaa8138b496af6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
