)]}'
{"id":"openstack%2Fkeystone~945509","triplet_id":"openstack%2Fkeystone~stable%2F2025.1~I1e862cb1c5e9c218cea59800ff759a1b094b5906","project":"openstack/keystone","branch":"stable/2025.1","topic":"bug/2104185","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-03-28 16:02:57.000000000","reason":"Change was submitted"},"27900":{"account":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"last_update":"2025-03-26 18:42:36.000000000","reason":"\u003cGERRIT_ACCOUNT_27900\u003e replied on the change","reason_account":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}}},"hashtags":[],"change_id":"I1e862cb1c5e9c218cea59800ff759a1b094b5906","subject":"api: Don\u0027t restrict unknown querystring parameters yet","status":"MERGED","created":"2025-03-25 18:25:41.000000000","updated":"2025-03-28 16:04:03.000000000","submitted":"2025-03-28 16:02:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"945509-bug/2104185","meta_rev_id":"9ce2d83fc45dcc95b8f1bca31eea0beacaca273e","_number":945509,"virtual_id_number":945509,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-28 16:02:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"all":[{"value":2,"date":"2025-03-28 14:05:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":2,"date":"2025-03-26 18:42:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"all":[{"value":1,"date":"2025-03-28 14:09:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-25 18:27:02.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"REVIEWER"},{"updated":"2025-03-25 19:24:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-03-25 19:31:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-28 14:05:52.000000000","updated_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"}],"messages":[{"id":"35476b69c26162d6fba766e22a3562f1200d25d0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-03-25 18:25:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7855e2374e3a0deadee24b398509338d7cebcaff","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-25 19:24:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/59bc0cc5d36b40c9b66b513f5799c66c\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6ce58105fb174748b4c6c87202e38843 : SUCCESS in 21m 27s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d5837ed1468648feaa030bd5e8966a95 : SUCCESS in 28m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"374745cf9b21f8fb9a4cd31e249ab6a3021358fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-25 19:31:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b9b4d53be054869892d098968ba7f8d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c5d4631f9fdf4ca5ba3ad72f093e07ff : SUCCESS in 16m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa5b1c2854b34ea7af645829a016d609 : SUCCESS in 6m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1061c44cfe22474bbca55aa4245d8a3d : SUCCESS in 11m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/aaf8f8499f6240fab5d96f219bf17f75 : SUCCESS in 11m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/483303e8f94447c0a5b84455394ff858 : SUCCESS in 12m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/1aac0b5187234d20a49ece04ca6b9f0b : SUCCESS in 52m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/136eb892e5c945d1b7072d0c01beae18 : SUCCESS in 55m 43s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/e2f0349f1ace47babb758b9a108ae38b : SUCCESS in 23m 52s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/73b33859c82d476782c70ba7330b93b5 : SUCCESS in 16m 20s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7acd85b95c7e45f59eada3d79253dc2d : FAILURE in 27m 27s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/addfe503fe254acfa09081b9ab6aaed8 : SUCCESS in 23m 06s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/d395e8b072604c30b2a499fd31b27529 : SUCCESS in 11m 30s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/8b684c011f924260b8335010ad274427 : FAILURE in 15m 13s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/91bc7acdc6494601b04f35e1dbcf32b5 : SUCCESS in 46m 06s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/a898436cfd5540d19695c71142c22a2b : FAILURE in 38m 24s (non-voting)\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/4764bd8e9867437eb2f6d05aa9a5c0a1 : SUCCESS in 5m 41s (non-voting)\n- keystone-dsvm-functional-oidc-federation https://zuul.opendev.org/t/openstack/build/9a50d4758d624b3aa37989567229cab5 : SUCCESS in 26m 01s","accounts_in_message":[],"_revision_number":1},{"id":"ace1b711bb670af99ecc3aad4b80b13977c17045","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2025-03-26 18:42:36.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d7f9f57ed8d493d3444f2bfd73bc857b1becea0a","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2025-03-28 14:05:52.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9399ea6741b098416a641db36f2288b39c97f85e","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2025-03-28 14:09:39.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"dcc2e743beb17068ba281f29bf519b911a68b4c7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-28 14:11:49.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fda75f83f52ba27b24862c2200db7424e4057a47","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-28 16:02:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/14a0c3d101cb47fa9189fb2ff0aacb65\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ad4cd79f77648b79e8e8b1ca0a1c01e : SUCCESS in 8m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e4656016cce449eba2ab22f51df9a32f : SUCCESS in 14m 12s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b4edb415ba4f43e5926bd86078731015 : SUCCESS in 10m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/933ddee576dc4baa85d94a6675d05a46 : SUCCESS in 13m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/a240e02b36ba4e6a8ac265fa4c312213 : SUCCESS in 50m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f4a86a7df85244f1ba12755fab2ce627 : SUCCESS in 1h 45m 38s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/e9fe870613144f5ea0232c0b8b2cb65f : SUCCESS in 25m 29s\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/56062d8ff4ea41e4b1c9612c4a363e56 : SUCCESS in 22m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/47afc856306d4e71b3b53f5a3cf73995 : SUCCESS in 50m 01s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/291a42e7632f4cf990d4187b5a9cb052 : FAILURE in 36m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c108289e50958905437cb1148c7f051035fd3341","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-28 16:02:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"9ce2d83fc45dcc95b8f1bca31eea0beacaca273e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-28 16:04:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8e65102d40e44e87bd86f201a2b5ae01\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a645a79a9d7e4582a7b7739e4f9800b8 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"79c90794adc8fd0f81caf27d2e9a5d48cde0f20d","revisions":{"79c90794adc8fd0f81caf27d2e9a5d48cde0f20d":{"kind":"REWORK","_number":1,"created":"2025-03-25 18:25:41.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/09/945509/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/09/945509/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/09/945509/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/09/945509/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/09/945509/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/09/945509/1"}}},"commit":{"parents":[{"commit":"bb6dd30158a34f4332100b05456b3a0ae62b0aba","subject":"Update TOX_CONSTRAINTS_FILE for stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/bb6dd30158a34f4332100b05456b3a0ae62b0aba"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-03-25 17:57:26.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-03-25 18:25:31.000000000","tz":0},"subject":"api: Don\u0027t restrict unknown querystring parameters yet","message":"api: Don\u0027t restrict unknown querystring parameters yet\n\nWe will address this in a new API version. For now, such a change would\nbe a breaking one.\n\nNo release note is included since these changes haven\u0027t been released\nyet.\n\nChange-Id: I1e862cb1c5e9c218cea59800ff759a1b094b5906\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nCloses-Bug: #2104185\n(cherry picked from commit 05cc3d1903c1bfe669a1032746aa9e7b197238e9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/79c90794adc8fd0f81caf27d2e9a5d48cde0f20d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/79c90794adc8fd0f81caf27d2e9a5d48cde0f20d"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
