)]}'
{"id":"openstack%2Fkeystone~948185","triplet_id":"openstack%2Fkeystone~master~Ib100f1943e223f30fd5ff53042b3f5f5bae2dd25","project":"openstack/keystone","branch":"master","attention_set":{},"removed_from_attention_set":{"27900":{"account":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"last_update":"2025-05-16 15:37:14.000000000","reason":"\u003cGERRIT_ACCOUNT_27900\u003e replied on the change","reason_account":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}}},"hashtags":[],"change_id":"Ib100f1943e223f30fd5ff53042b3f5f5bae2dd25","subject":"Start building openapi doc","status":"MERGED","created":"2025-04-25 10:19:54.000000000","updated":"2025-05-16 18:21:17.000000000","submitted":"2025-05-16 18:20:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"948185","meta_rev_id":"8e47df3c31148a1c4cc4ff4fe55e2becdfc6ccf3","_number":948185,"virtual_id_number":948185,"owner":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-16 18:20:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"all":[{"value":2,"date":"2025-05-07 15:14:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":2,"date":"2025-05-16 09:22:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"all":[{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":1,"date":"2025-05-16 09:22:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-25 10:55:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-04-25 12:20:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-07 15:14:59.000000000","updated_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"},{"updated":"2025-05-16 09:22:36.000000000","updated_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"}],"messages":[{"id":"c2a1c0ee7c40bfe6feb845d377d0d917e37d835d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2025-04-25 10:19:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"018103f7d989109c00732e2a99d826061944cb35","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-25 10:55:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6ed482741c554828bd238010525b9b5c\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f50538aeb0854a88bdd59b377617eb2e : SUCCESS in 29m 09s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/50e76e4ce2684e55913cad9c4b196102 : SUCCESS in 27m 41s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f55e8d980f234f8bb6ca971674dfeb79 : SUCCESS in 33m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a5efce99b9e2c7903d0967ca9dc1327e45fc3667","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-25 12:20:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f46e69ef510b471fa6f5961c86b047f3\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/91eb52de59be44749d312127e830739b : SUCCESS in 6m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b9d473ce20094264b8e8502841bbdade : SUCCESS in 17m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/726a2ab174144c03aae8202f9c636603 : FAILURE in 7m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d114f02c6750463ab7579af52ca6e1f2 : SUCCESS in 10m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0aa1a9dd56ac416b9af24fd282e1b4e6 : SUCCESS in 10m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4e266ae20363430bbdedd2df874fbdc7 : SUCCESS in 15m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c993aabcfa094d5abb5206243fb653f0 : SUCCESS in 13m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/ce60fbcaf58e48618728d634ba3d0c03 : SUCCESS in 50m 51s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/cc7ec9b75df34e06b392e0c3bb6962c8 : SUCCESS in 1h 55m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/32974c849c354740a2c7ee3f68ed1e8b : SUCCESS in 7m 40s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/1696ccd9e3294a38988a7a32e17e74ff : SUCCESS in 27m 31s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/ee8f1bf0ef414461846a50a97e640298 : SUCCESS in 38m 27s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/20398a6ad47243d9a9ad7e6560d35d64 : FAILURE in 15m 10s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/c3655ff74c5b4ed68cf3ff19fb43d766 : SUCCESS in 17m 21s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/61d00291b85943c087cb81ba07c8cba3 : SUCCESS in 6m 35s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/4716575b4f9042b0970b0a8b1500be37 : FAILURE in 12m 04s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c2a7b6c63ee74b799c9c75d6cebafdf7 : SUCCESS in 52m 27s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/3b66d2c7c9ff4ae2a7caf6e2e905e009 : FAILURE in 31m 12s (non-voting)\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/0ac163cf91824f5fbbded9c71f6096e9 : SUCCESS in 6m 42s (non-voting)\n- keystone-dsvm-functional-oidc-federation https://zuul.opendev.org/t/openstack/build/23316f8d2eb14e078b33717728f1b23e : SUCCESS in 15m 07s","accounts_in_message":[],"_revision_number":1},{"id":"5480826b10b94c96813ebf709b0c384276259294","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2025-05-05 12:06:54.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"1b1c266aecf2b31077a87627dd8516554f45271b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-05 12:57:57.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a3d45e11510442c9ad637ee92fb02d57\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/d350f2ba2d474815978595abe4dd4d64 : SUCCESS in 44m 53s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b26c531b5f7e4d4fbb3f3043da28a709 : SUCCESS in 47m 48s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/806016adb27b45b9941b4f8aa4f8c65a : SUCCESS in 49m 08s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b5460b7f2ed5c6ca8ba72ff88e8f70088b06142b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-05 13:48:16.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/34c0fc020d2948ceb53c746545a30194\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/4a76db5b3c964211a2bc6eac4c563e24 : SUCCESS in 5m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b91b69b8beca4c90b05f667b5b4bee62 : SUCCESS in 18m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/343c01b13036492b9faae0ab524df29c : FAILURE in 5m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/008f07a9bd68401aa63b79fb117798b1 : SUCCESS in 14m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4f3028fc6e5c4b819ec09fc9671cdb3d : SUCCESS in 11m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/45dcec80deac4b56b35d7e24c88c8357 : SUCCESS in 15m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/728be92a53cf47bcb642348fceb69ad8 : SUCCESS in 10m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/6ce15d15446646cb96ec07cdd7a56f76 : SUCCESS in 57m 44s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f45d1fd1e90549ada5b0aa49afbf12fd : SUCCESS in 1h 36m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/36467d9fe7dd4fed996824a8191e357f : SUCCESS in 7m 40s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/10259464b4d14222a27ece1484b0b70f : SUCCESS in 19m 10s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/26fdf78005d54239949cf6e77aa8b822 : FAILURE in 8m 39s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/322b851acd8d463aab91f9eb0956f8f4 : FAILURE in 23m 26s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/019d5be2ed3a4d88a24d52f399602a7c : SUCCESS in 25m 21s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/b6e329c1b45848c4a0f7b4e781d80e4b : SUCCESS in 8m 38s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/8a8dcae8a1e54dd9ae3df10aa3cfd471 : FAILURE in 8m 03s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9b61050a7f974d39b90a1c9baa9bef36 : SUCCESS in 49m 45s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/fe012df59d1b4518aef4b53fe466f4b6 : FAILURE in 15m 52s (non-voting)\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/1fbb76e4b62349839d7906ad0c5f970b : SUCCESS in 5m 34s (non-voting)\n- keystone-dsvm-functional-oidc-federation https://zuul.opendev.org/t/openstack/build/793aca07112c441f94f01b6914f7627e : SUCCESS in 11m 56s","accounts_in_message":[],"_revision_number":2},{"id":"7466a8916c914be0c00854a58ffd0816de614638","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2025-05-05 13:55:16.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"52ffe8a8c68b9b3a5ab3f7d9816f9dd4de813004","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-05 14:47:55.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b94144ca46a45169ede41ef724cce03\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9a774141d61841918cd2bd2ee30e793e : SUCCESS in 42m 57s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8265c5c550a64867888306d4c8cb88dd : SUCCESS in 48m 59s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/83b59b4d2adb4feb8308c4716900c9e9 : SUCCESS in 35m 56s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e8d865b90034f140800618af9f3712b6bfbb0f4b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-05 15:23:02.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0feff7a2bb2840bfaad8fa7f93eb194b\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/79941be2abe749c08d43f89e6f794445 : SUCCESS in 6m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0b9af63d6d4342cdb88af0b7c021d87b : SUCCESS in 36m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4738135f10b747f5a28964cc8651d6fb : SUCCESS in 5m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/304db3234e644d53ad4cc3e6a40a4094 : SUCCESS in 13m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3c293fba27404062a92dfc0c4b66ceed : SUCCESS in 11m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/200ec474529a415ea635e063ebd144dc : SUCCESS in 15m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e05ffe7a9cd740fca9cac398601e21f3 : SUCCESS in 12m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/6e2a1a86c35d4e0b9394c6bc6c4982f5 : SUCCESS in 1h 06m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/14eff61bb9624a8d98ce8f86bbc3550a : SUCCESS in 1h 26m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8f1ee1ba2aa34112aeff53f8ce2c460c : SUCCESS in 9m 31s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/ab89001db0324acaa141268744bee329 : SUCCESS in 33m 01s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/95a86507c0ac42cd9584834aa9687f3e : FAILURE in 11m 48s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c774a6eb40c24ecd9001ce2336191c07 : FAILURE in 24m 21s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/bed4b069f7d541eab811ba5562c4fbed : SUCCESS in 26m 07s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/35e459166e774e18a983a647a997aa5d : SUCCESS in 15m 58s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/ecccd47b5c3e4c2badf2f54d5ae61162 : FAILURE in 12m 29s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d06673d7fcb24b91b375e61698d858ed : SUCCESS in 50m 06s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/193b3095642441fbb12fdbd11d4e9b28 : FAILURE in 44m 19s (non-voting)\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/377726851e194c958e54bfc5277d6499 : SUCCESS in 7m 45s (non-voting)\n- keystone-dsvm-functional-oidc-federation https://zuul.opendev.org/t/openstack/build/b9a2961b8d7a474e8981275869208bd2 : SUCCESS in 28m 40s","accounts_in_message":[],"_revision_number":3},{"id":"c2e322e5916764b0f6beab86dc064a78f6fe75b2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2025-05-07 09:39:08.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"d060c489eb7c1c32058be6d1dc2132acda6b3fc6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-07 10:13:01.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0f2cd9825a1c4473a18a4f791e8d1898\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/20151956c7f6442c8a3775ce7a324e90 : SUCCESS in 31m 35s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a2d758d8f43d4bef995365a296e5f0f1 : SUCCESS in 27m 41s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/17def646025845e0bd89b980a386be29 : SUCCESS in 31m 52s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"ac8f83d343c24f83742c2a4da4ee14e7d4601b23","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-07 11:19:26.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ec674a98b2be4de28d6ed616bfa0d7c5\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/0a552ed34d254c838d993260ad0ef564 : SUCCESS in 4m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/09e58b2bc03f4a51b7a3dbd50a3ca251 : SUCCESS in 15m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3d9338b6007410cadc0b825c824f428 : FAILURE in 6m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/792f140d597f4926bb4d5c5df25f4667 : SUCCESS in 11m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a17f51054a2b46ac95fb25e872ce2f93 : SUCCESS in 10m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bf6a4dedcba341c3a7cd4e48aec346fc : SUCCESS in 10m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5cab2ea57c2a42dc8bec6cc2183e7918 : SUCCESS in 14m 00s\n- requirements-check https://zuul.opendev.org/t/openstack/build/4a290b0025bd4d219a269a5fb91242d4 : SUCCESS in 3m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/3ba65401476448b4ab38d226b743dcc8 : SUCCESS in 57m 12s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b2de4952e0d046a3a9a7cc23cdcfa2ee : SUCCESS in 1h 35m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2a6d130f78854d0793efcfc2b1f98858 : SUCCESS in 10m 07s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/3344879f1ce4489d9c56043d9977f2f5 : SUCCESS in 19m 35s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/f6ad8ecad72048e3bca88f11dc66e5fa : FAILURE in 18m 15s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1753fbb61184425593641da7796b4ad1 : FAILURE in 24m 59s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/0ccaf837b9704439af70d6461781f77c : SUCCESS in 10m 56s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/b570ed38750b45f9b7798574ed68d46a : SUCCESS in 6m 17s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/f219144ba37f4ff1a21d6945603f20ef : FAILURE in 7m 48s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/781db40f5e3c4dfa9e55130f28b77371 : SUCCESS in 59m 04s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/f2614b017bd5460d86314911cb1be7ed : FAILURE in 15m 50s (non-voting)\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/e3c448862b3d490ca0471298983de1c1 : SUCCESS in 6m 44s (non-voting)\n- keystone-dsvm-functional-oidc-federation https://zuul.opendev.org/t/openstack/build/87c40ef8e17c400198b95a6895b9e052 : SUCCESS in 28m 04s","accounts_in_message":[],"_revision_number":4},{"id":"81bda42e1c7480e4799500e601682f4261158114","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2025-05-07 11:30:16.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"46f9d11a27390f902aa6dbc4b3ed91e1f08f1931","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-07 12:06:44.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/359e014f9a044f61ba5f8c8f78ec9eb9\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/dd33737fc23d4ebd8cdda043fc2e1621 : SUCCESS in 23m 40s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9a351ce69b244288ba9e6cf43f27ce7b : SUCCESS in 28m 05s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ec48cc04dc8c431b91fcbd418c9d4cd9 : SUCCESS in 34m 52s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"68c4d9bf7cdf7e54428f0e463ea75d991ca88fa2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-07 13:17:01.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ebf96cd0e3c24e9083f1f5c4c7518f05\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/45dde9189ff445b39a989eb92a53e561 : SUCCESS in 7m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ac594e1abf4a4f48901c6e11371e21e1 : SUCCESS in 15m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e802c3eea7f945498e70e367b13f922f : SUCCESS in 6m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/eb1afd0db1b94c56a8f97af35529d92a : SUCCESS in 13m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6760d0dd87f94ca58fe7c6d62b315f3c : SUCCESS in 10m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2ca296b0c3e8435195a137e1cb28472d : SUCCESS in 15m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a290842a5d534819add44dc313bba781 : SUCCESS in 13m 43s\n- requirements-check https://zuul.opendev.org/t/openstack/build/2c1e210361b44ec99491ee52b5d30cd0 : SUCCESS in 3m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/31b06a3cdcc84a13b3bf991e6db05c38 : SUCCESS in 50m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b08aa99bbf714b99a548104704d0f0d4 : SUCCESS in 1h 42m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/080d9518197841128ba3a9da4beeca42 : SUCCESS in 9m 27s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/2fbad6c725b5456d8102a51aed02a911 : SUCCESS in 26m 49s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/bea4f842a2614dceba58f6ddb1cdbd64 : FAILURE in 18m 20s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f2ca318bcee740ebae988d4d3eba4010 : FAILURE in 17m 32s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/ded71df7d4f54686b8f622c8d01e2e36 : SUCCESS in 23m 37s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/186446183a644745b89e3c295a32e080 : SUCCESS in 7m 42s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/9d5fd87b891b4396ad1ae1b99219b271 : FAILURE in 14m 12s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/617b56f0eb7746f49571b17db50fcce6 : SUCCESS in 50m 34s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/eb6010bdb909407e8104b3fa5053dad2 : FAILURE in 35m 45s (non-voting)\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/d3aafcecc87d4b91b7f3661773d69c0a : SUCCESS in 7m 14s (non-voting)\n- keystone-dsvm-functional-oidc-federation https://zuul.opendev.org/t/openstack/build/5132ef3431b94c45ace6b351c0309181 : SUCCESS in 31m 20s","accounts_in_message":[],"_revision_number":5},{"id":"50952f1a76f705c24f613e84088cb982c476e012","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2025-05-07 15:14:59.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"3b10a9e957d85057bdd166a1eaf8b2aea54b8054","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2025-05-16 09:22:36.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"c19444eed36757f6610772f921ef9ecc8c714b93","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-16 09:22:57.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"f7cfb0b19b24bde3ee4f167d46af06cf5ceca827","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-16 10:51:58.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5ae6d42e3e63461999d05803131480c4\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/c827a7c7235a47ebb40a7e1e828ecbc0 : SUCCESS in 5m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6455f73229824e2ca29a302e0053fbc3 : FAILURE in 7m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/644f1187b11346ecbc525321afb7dbd1 : SUCCESS in 9m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ad7a799e9d1943d88772f0a5439c7b87 : SUCCESS in 10m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e42f78e5e0f14927b5e58308a7d689ba : SUCCESS in 12m 57s\n- requirements-check https://zuul.opendev.org/t/openstack/build/3bc30ec45b924916819f1b91fa1e4e56 : SUCCESS in 3m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/abe7c77a334f4d2fbfcc18086d4f5f2c : SUCCESS in 46m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4ffd7bd79c4c4eabae1501c51b52da5a : SUCCESS in 1h 22m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e945e78e63f041e2a4d91b4db7f680c8 : SUCCESS in 11m 01s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/3eccbab68dd44c0e9b667ef0d7df8de0 : SUCCESS in 25m 34s\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/77007bfa47d94e65abad85c1f4d05b94 : SUCCESS in 26m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e04db323764b45a89904dce12018c539 : SUCCESS in 1h 01m 21s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/3a0ebfd56f4444eba936bdfe2f1e529d : FAILURE in 42m 55s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"440dc10cbb63aebac9ee27b7f38b50a7121fc03c","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2025-05-16 15:37:14.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"7933e27aeff75dbbdc9bfecacf647c5c32ec8e4f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-16 16:31:07.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/23158d200b8348f1aff75e87e768ecec\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/feeaf19ea98e4492ba75d9039d820900 : SUCCESS in 5m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2b6b8dd461b44952ae0dd093b5f65019 : SUCCESS in 21m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b367f1735e4146468767b42a9fccf2ce : SUCCESS in 7m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/137087e9bf2642a2a53be8c90f9c4df6 : SUCCESS in 10m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/68db9300239843a8bd9c980526eb51d4 : SUCCESS in 14m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/43d097b5a61344c199d577606071a32c : SUCCESS in 16m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a8872800fc04644af30e70027bd0557 : SUCCESS in 15m 06s\n- requirements-check https://zuul.opendev.org/t/openstack/build/50a46e41a3344393be36b9b06471e3a5 : SUCCESS in 3m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/90b26ba90b0a41ca9fb94856a6105caa : SUCCESS in 48m 49s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/e4ae6b08841d468dae7586eef4cb6401 : SUCCESS in 53m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/247b37264a0043a6ad21ffbb559222e7 : SUCCESS in 9m 12s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/48fedf08b763446892c33de78638d7f8 : SUCCESS in 25m 43s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/e2859f3f7c004dec9d92a80d9e1dcf05 : FAILURE in 24m 39s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/17912f9a068245a39265ce1b7dbbfaba : FAILURE in 13m 41s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/77c031561db542f3be949a233ace4229 : SUCCESS in 24m 51s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/37556f0523d247f0bd03f7a209ac2113 : SUCCESS in 12m 42s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/04ecfce02958406781bf3af1a6b425c1 : FAILURE in 15m 56s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/12f36ba80f574c5889b732f108a1b56a : SUCCESS in 28m 39s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/aa809af435524aa797e4a4370e712e6e : FAILURE in 32m 19s (non-voting)\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/fb293315faf040da914dcfc7351d91ed : SUCCESS in 7m 22s (non-voting)\n- keystone-dsvm-functional-oidc-federation https://zuul.opendev.org/t/openstack/build/8bc31d359d26416c81261bf3d4206674 : SUCCESS in 15m 45s","accounts_in_message":[],"_revision_number":5},{"id":"ca85ad1df34deb048940cfa1b6fd2f7e71cb7a87","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-16 16:31:29.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"b42ca56d75dbe332b444873337b61b5e17414e2f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-16 16:57:41.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e7f31731471e4e2cbaea854e4e9b8f5a\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/926886dce6b7474e8f28667439b9a21e : SUCCESS in 28m 03s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7096aa2d7c1a4f09b749fdea59d1c616 : SUCCESS in 46m 57s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f9b03bed297a437ab692a30a8e49aeeb : SUCCESS in 34m 57s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"0ed23a9fa33838d717de1f6945e8dade03b2705c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-16 18:20:07.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee243c85d34c426f8f2adcc5c5ab33ec\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/8e3774164aaf4780ae8b60918967bdbc : SUCCESS in 7m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5e0a89a2a374b9a94b770db996cd479 : SUCCESS in 6m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ec9b8b83e5d64ec4bca13f30b6d5bd8d : SUCCESS in 10m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1d8805975238420299a08f7660e07011 : SUCCESS in 11m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d21ee10215854b33a1b770581a28896b : SUCCESS in 11m 28s\n- requirements-check https://zuul.opendev.org/t/openstack/build/54be8d1b9c5140cb820ad40cea7ef06b : SUCCESS in 3m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/e528f980c7f849788c3d7abb73bb64c8 : SUCCESS in 47m 25s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c50002022d7b4bb8832266c7b5dfaca2 : SUCCESS in 1h 43m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5b47d1d25adf4dd78c3b39b101310762 : SUCCESS in 7m 53s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/f17783229e794061902e134d40980a15 : SUCCESS in 27m 10s\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/40715b71dde943db898d61d6320159d4 : SUCCESS in 26m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/02305561a0a343ec8d4c34588722834c : SUCCESS in 1h 00m 31s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/e3fe76ddc94c485cbcb648e0ade7d910 : FAILURE in 38m 26s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"767129c45e571ba06921863ab299ff506a17c939","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-16 18:20:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"8e47df3c31148a1c4cc4ff4fe55e2becdfc6ccf3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-16 18:21:17.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/65064118b71141c7a4ef1dd76ed238df\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/cb8b21b016e247c7b43fb33e2249a64c : SUCCESS in 38s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2339e7b933bb461d8f89787bdadbe315 : SUCCESS in 1m 00s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4f41e3318b3243d9a9327afeac09bcd6 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"d688c3c9592cbce5fb85eeb02271859fa92f811f","revisions":{"3193e2124f0f886798f28eac5c08dafcb81b5915":{"kind":"REWORK","_number":1,"created":"2025-04-25 10:19:54.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/85/948185/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/85/948185/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/85/948185/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/85/948185/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/85/948185/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/85/948185/1"}}},"commit":{"parents":[{"commit":"46ba4f455dd14fe509e2e83dc39a3fc6894f7dbb","subject":"api: Don\u0027t restrict unknown querystring parameters yet (redux)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/46ba4f455dd14fe509e2e83dc39a3fc6894f7dbb"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2025-04-25 10:16:46.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2025-04-25 10:16:46.000000000","tz":120},"subject":"[WIP]: Start building openapi doc","message":"[WIP]: Start building openapi doc\n\n- Build openapi spec as part of the api-ref job\n- render openapi as invisible doc (can be accessed knowing the path -\n  /openapi.html)\n- forcibly update openstackdocstheme and os-api-ref beyond\n  upper-constraints to make openapi look like it is supposed to be\n- since os-openapi and os_api_ref conflict in certain css things openapi\n  must be build using dedicated conf.py\n\nChange-Id: Ib100f1943e223f30fd5ff53042b3f5f5bae2dd25\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/3193e2124f0f886798f28eac5c08dafcb81b5915"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/3193e2124f0f886798f28eac5c08dafcb81b5915"}]},"branch":"refs/heads/master"},"458da5b7d66f8b9c14abcc96944425c0e22c7a29":{"kind":"REWORK","_number":2,"created":"2025-05-05 12:06:54.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/85/948185/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/85/948185/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/85/948185/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/85/948185/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/85/948185/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/85/948185/2"}}},"commit":{"parents":[{"commit":"46ba4f455dd14fe509e2e83dc39a3fc6894f7dbb","subject":"api: Don\u0027t restrict unknown querystring parameters yet (redux)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/46ba4f455dd14fe509e2e83dc39a3fc6894f7dbb"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2025-04-25 10:16:46.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2025-05-05 12:06:18.000000000","tz":120},"subject":"[WIP]: Start building openapi doc","message":"[WIP]: Start building openapi doc\n\n- Build openapi spec as part of the api-ref job\n- render openapi as separate doc\n\nChange-Id: Ib100f1943e223f30fd5ff53042b3f5f5bae2dd25\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/458da5b7d66f8b9c14abcc96944425c0e22c7a29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/458da5b7d66f8b9c14abcc96944425c0e22c7a29"}]},"branch":"refs/heads/master"},"52d6a1ecb30d513d84e3eb91cc6aefaa65f43bd1":{"kind":"REWORK","_number":3,"created":"2025-05-05 13:55:16.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/85/948185/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/85/948185/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/85/948185/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/85/948185/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/85/948185/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/85/948185/3"}}},"commit":{"parents":[{"commit":"46ba4f455dd14fe509e2e83dc39a3fc6894f7dbb","subject":"api: Don\u0027t restrict unknown querystring parameters yet (redux)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/46ba4f455dd14fe509e2e83dc39a3fc6894f7dbb"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2025-04-25 10:16:46.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2025-05-05 13:54:37.000000000","tz":120},"subject":"[WIP]: Start building openapi doc","message":"[WIP]: Start building openapi doc\n\n- Build openapi spec as part of the api-ref job\n- render openapi as separate doc\n\nChange-Id: Ib100f1943e223f30fd5ff53042b3f5f5bae2dd25\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/52d6a1ecb30d513d84e3eb91cc6aefaa65f43bd1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/52d6a1ecb30d513d84e3eb91cc6aefaa65f43bd1"}]},"branch":"refs/heads/master"},"329ef7887c62e8c250190e5c981baa9b39e83b4f":{"kind":"REWORK","_number":4,"created":"2025-05-07 09:39:08.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/85/948185/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/85/948185/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/85/948185/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/85/948185/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/85/948185/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/85/948185/4"}}},"commit":{"parents":[{"commit":"46ba4f455dd14fe509e2e83dc39a3fc6894f7dbb","subject":"api: Don\u0027t restrict unknown querystring parameters yet (redux)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/46ba4f455dd14fe509e2e83dc39a3fc6894f7dbb"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2025-04-25 10:16:46.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2025-05-07 09:38:50.000000000","tz":120},"subject":"Start building openapi doc","message":"Start building openapi doc\n\n- Build openapi spec as part of the api-ref job\n- render openapi as separate doc (preview)\n\nChange-Id: Ib100f1943e223f30fd5ff53042b3f5f5bae2dd25\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/329ef7887c62e8c250190e5c981baa9b39e83b4f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/329ef7887c62e8c250190e5c981baa9b39e83b4f"}]},"branch":"refs/heads/master"},"d688c3c9592cbce5fb85eeb02271859fa92f811f":{"kind":"REWORK","_number":5,"created":"2025-05-07 11:30:16.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/85/948185/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/85/948185/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/85/948185/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/85/948185/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/85/948185/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/85/948185/5"}}},"commit":{"parents":[{"commit":"46ba4f455dd14fe509e2e83dc39a3fc6894f7dbb","subject":"api: Don\u0027t restrict unknown querystring parameters yet (redux)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/46ba4f455dd14fe509e2e83dc39a3fc6894f7dbb"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2025-04-25 10:16:46.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2025-05-07 11:30:10.000000000","tz":120},"subject":"Start building openapi doc","message":"Start building openapi doc\n\n- Build openapi spec as part of the api-ref job\n- render openapi as separate doc (preview)\n\nChange-Id: Ib100f1943e223f30fd5ff53042b3f5f5bae2dd25\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/d688c3c9592cbce5fb85eeb02271859fa92f811f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/d688c3c9592cbce5fb85eeb02271859fa92f811f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
