)]}'
{"id":"openstack%2Fkeystone~948630","triplet_id":"openstack%2Fkeystone~stable%2F2024.1~I854d37e3b4a34a7880f157564466bde61a3f886a","project":"openstack/keystone","branch":"stable/2024.1","attention_set":{},"removed_from_attention_set":{"7414":{"account":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"last_update":"2025-05-02 14:45:11.000000000","reason":"\u003cGERRIT_ACCOUNT_7414\u003e replied on the change","reason_account":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"}},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2025-05-02 16:10:43.000000000","reason":"Change was submitted"},"27900":{"account":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"last_update":"2025-05-02 14:43:53.000000000","reason":"\u003cGERRIT_ACCOUNT_27900\u003e replied on the change","reason_account":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}}},"hashtags":[],"change_id":"I854d37e3b4a34a7880f157564466bde61a3f886a","subject":"Fix DB migrations after alembic integration","status":"MERGED","created":"2025-05-01 10:39:55.000000000","updated":"2025-05-02 16:11:52.000000000","submitted":"2025-05-02 16:10:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"948630","cherry_pick_of_change":930589,"cherry_pick_of_patch_set":3,"meta_rev_id":"82db9735c06006f0949535a57af84f118a3d4364","_number":948630,"virtual_id_number":948630,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-02 16:10:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"all":[{"value":2,"date":"2025-05-02 14:45:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":2,"date":"2025-05-02 14:43:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-05-01 11:51:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"all":[{"value":1,"date":"2025-05-02 14:45:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-01 11:06:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-05-01 11:51:49.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2025-05-01 11:51:49.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"REVIEWER"},{"updated":"2025-05-01 11:51:49.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"},{"updated":"2025-05-01 12:17:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"44549ef32a7d876685d1703cbeb9b9ab8c8530b4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-05-01 10:39:55.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"066ae3ffe9a43ae2d5da2d90d34a27b7c9b02718","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-01 11:06:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/278c1bbdecd641b4b81a3b105746668a\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c4faf4cbb1be424aa4b44d5144971a00 : SUCCESS in 24m 26s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a63b1e90d0434348ac7c158ce10fc59a : SUCCESS in 19m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"530ebc23d0f7b6838f3bb66ecd14b427bd1b7dfd","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-05-01 11:51:49.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"b4bb16155021564b6e54138423d5d15bac8ee82d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-01 12:17:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/940cd3805eba4240a8ea1aaab2a1b6f5\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6c53abc2f27a49bdb873b7844b55cb22 : SUCCESS in 13m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c81e1dfe1f714847a65235b7645fb417 : SUCCESS in 4m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0f294026ce57448087db64ea52c03791 : SUCCESS in 10m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b6a2833251a64e2da7c622194c6e15b3 : SUCCESS in 12m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e6db5323a8c4d0e8610052daa050f6d : SUCCESS in 10m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/b41fc6c86743493f9a9c4b829ed3ff68 : SUCCESS in 52m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6ba733f3d84e4e7e81086cdcdfcc5bbf : SUCCESS in 1h 31m 29s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/a8efd6cb51904f71912e51e8378b55c1 : SUCCESS in 25m 52s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/f8ad1bb037e7414c97f8542653d6c215 : FAILURE in 35m 21s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a54d8eb411c84ef48ab0d48959543518 : FAILURE in 25m 29s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/f7cda9cc55ca478f83a70ac548fe5bb1 : SUCCESS in 25m 11s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/58e4638c7e9943ed99af77d0a4ec8c3a : SUCCESS in 14m 37s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/2d243f4575714c039575ed445377e294 : FAILURE in 15m 42s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e2e750734fa7416881151e4862fa7483 : SUCCESS in 55m 59s\n- keystone-dsvm-functional-oidc-federation https://zuul.opendev.org/t/openstack/build/9e76d11f38b949da91a0a8a991c5b3c7 : SUCCESS in 31m 54s","accounts_in_message":[],"_revision_number":1},{"id":"b851e570503d80e684d6069feb8962fb1f3feb43","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2025-05-02 14:43:53.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"31dfb234993c0dec4558639d56ece9aeb131bee7","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2025-05-02 14:45:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"43ca51b35f1400fb6d1f0aeb4590e588b0648e78","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-02 14:45:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"98f16aeac1e0fc35ba933ec6529f9caf2fb134ed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-02 16:10:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6717db546b964c13a4d5a666965b16bc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/879120cc00a64b13a1fd3b3d3f0fe34e : SUCCESS in 3m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/76328d8206ca47bfa2a50a269731a842 : SUCCESS in 10m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e4867f1ff023474aac98bc3b6a462778 : SUCCESS in 8m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce2cdd352d8841d6b212596e1cb573d0 : SUCCESS in 9m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/975471e54b5b4143b45e53934b98f080 : SUCCESS in 48m 34s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/44f92580e2a64319b8019e15d4ee383b : SUCCESS in 1h 24m 43s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/e0945a9fd8e34b7588107ce99d0021eb : SUCCESS in 17m 53s\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/ca85ad313f7447ea83f86691965d2a7f : SUCCESS in 18m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/80684f06f5c54bd583ae3eba772ad9be : SUCCESS in 52m 29s","accounts_in_message":[],"_revision_number":1},{"id":"7672dbe7e7170cadb8041c174647670c0f3441e0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-02 16:10:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"82db9735c06006f0949535a57af84f118a3d4364","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-02 16:11:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/40e5722861d84b77be7c7f30663dcbf7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/269d5eb747584dcb9c71166e371b694c : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4e2e36ad7767a459277987d21f748cec54ac09e0","revisions":{"4e2e36ad7767a459277987d21f748cec54ac09e0":{"kind":"REWORK","_number":1,"created":"2025-05-01 10:39:55.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/30/948630/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/30/948630/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/30/948630/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/30/948630/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/30/948630/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/30/948630/1"}}},"commit":{"parents":[{"commit":"5a27264673e42ff0a91b9ed64d54d9fc41ee76af","subject":"Merge \"Move OIDC job to check\" into stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/5a27264673e42ff0a91b9ed64d54d9fc41ee76af"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2024-09-26 14:36:06.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2025-05-01 10:39:55.000000000","tz":0},"subject":"Fix DB migrations after alembic integration","message":"Fix DB migrations after alembic integration\n\nWith a change to swap sqlalchemy-migrate with alembic [1]\na `db_sync --check` was broken.\n\nThis is due to both `upgrades.get_db_version` and\n`upgrades.get_current_heads` are actually checking \"current\" state of\nthe dabase by calling _get_current_heads[2][3], while obvious intention\nwas to compare intended state with current state.\n\nWith that we\u0027re introducing upgrade.get_head_revisions which will fetch\nrevisions not from the database, but from the environment [4]\n\nAs a result `db_sync --check` does compare desired state of the DB with\nactual state and exists with corresponsive status again.\n\n[1] https://opendev.org/openstack/keystone/commit/f174b4fa7c4fb010bbacc8c5a5f3625a8fcb41f3\n[2] https://opendev.org/openstack/keystone/src/branch/master/keystone/common/sql/upgrades.py#L147\n[3] https://opendev.org/openstack/keystone/src/branch/master/keystone/common/sql/upgrades.py#L191\n[4] https://alembic.sqlalchemy.org/en/latest/api/runtime.html#alembic.runtime.environment.EnvironmentContext.get_head_revisions\n\nCloses-Bug: #2080542\nChange-Id: I854d37e3b4a34a7880f157564466bde61a3f886a\n(cherry picked from commit 5125d9feed8dc8f9f0ad01f4c042594475f5fb84)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/4e2e36ad7767a459277987d21f748cec54ac09e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/4e2e36ad7767a459277987d21f748cec54ac09e0"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
