)]}'
{"id":"openstack%2Fkeystone~949329","triplet_id":"openstack%2Fkeystone~master~I9fa7cd60bf27768db06f69681675fce2e55712af","project":"openstack/keystone","branch":"master","topic":"pip-23.1-support","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-05-23 08:28:15.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_37598\u003e by clicking the attention icon","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}},"hashtags":[],"change_id":"I9fa7cd60bf27768db06f69681675fce2e55712af","subject":"setup: Remove pbr\u0027s wsgi_scripts","status":"MERGED","created":"2025-05-09 11:46:36.000000000","updated":"2025-05-23 09:51:33.000000000","submitted":"2025-05-23 09:50:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"949329-pip-23.1-support","meta_rev_id":"957f9ff0a3bee302c2a75f249a9df752e520ea91","_number":949329,"virtual_id_number":949329,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-23 09:50:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"all":[{"value":2,"date":"2025-05-16 14:42:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":2,"date":"2025-05-09 14:06:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-05-23 07:46:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"all":[{"value":1,"date":"2025-05-16 14:42:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-09 12:16:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-05-09 12:31:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-09 14:06:10.000000000","updated_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"REVIEWER"},{"updated":"2025-05-16 14:42:12.000000000","updated_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"},{"updated":"2025-05-23 07:46:02.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"},{"updated":"2025-05-23 07:46:06.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"}],"messages":[{"id":"3f41d05b1093b45f9a893141684b78f6fc16fa6f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-05-09 11:46:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7635d0aa094da3d84fecd4ae44cdcf85b03cd0ef","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-09 12:16:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/56776961cbd9403da7b27c43843c51af\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/fb80e257cc0e4276b345ee5e10f4540b : SUCCESS in 22m 48s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/00a582dc669543a089c61d2483c3e1a5 : SUCCESS in 27m 57s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/f134376df4d0431fbd2aa3b773237fff : SUCCESS in 22m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"589b1e5280cc4ad029e25a976f5de90010c6b4d1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-09 12:31:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/498d1bff51bf4989aaaac3289e2db46c\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/669c1d8c5e794a788e703e3032c4aec2 : SUCCESS in 2m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/992ff627263e43d2824b920895f656f6 : SUCCESS in 16m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81a4d064232b42e78e3fad68ea794cfe : SUCCESS in 6m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/39157dcc61154b4ea0da5b2dae9b93d7 : SUCCESS in 11m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1bb93bf603c84b22b392b06e5887835c : SUCCESS in 11m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bbffd1df6aae4f5baa7f9de3b7a6013f : SUCCESS in 16m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7016381c862c4dac8cdff44db4169734 : SUCCESS in 11m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5fd37fe93746488dbdec3a1ccfebc27e : SUCCESS in 9m 39s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/4830ffeae0d94010a6fb894e23122eee : SUCCESS in 25m 28s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/58f5c5a10f60452b91b3cb6127a950ae : FAILURE in 16m 39s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/52a8b61c7827421e9aee522170c07d9e : FAILURE in 17m 17s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/1645353f49e34d0c8e3e55f65c945c85 : SUCCESS in 18m 58s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/689a9997c9aa4bb1a6d52269d0f6ac35 : FAILURE in 10m 06s (non-voting)\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/dbce17d82324460b986040a44a71dd0e : FAILURE in 30m 45s (non-voting)\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/a14337bb8f9646b4ac275e6d4c3483b8 : SUCCESS in 5m 52s (non-voting)\n- keystone-dsvm-functional-oidc-federation https://zuul.opendev.org/t/openstack/build/006f61375bdc4eaf9994b79c7894e58b : SUCCESS in 22m 35s","accounts_in_message":[],"_revision_number":1},{"id":"2eba3fa65112ae75b37156225fe22f2c9a8df79b","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2025-05-09 14:06:10.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7f344ee31da478948903dd312c1775e41d920baa","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2025-05-16 14:42:12.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"74e0d7c37c5ce6c1350efc31307077f062ec3672","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-16 14:42:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"36fdced514dc7a8ac513ce20095bef8bbd5a9b3b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-16 15:43:30.000000000","message":"Patch Set 1: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/949328 is needed.","accounts_in_message":[],"_revision_number":1},{"id":"8105f69d26dbb4c74f58eb0ca629e0d6f890bf2c","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-05-23 07:46:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"da4f74ea02ab668f3f669946cc1ac0ce053975d0","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-05-23 07:46:06.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"c1c97c4db289c0b67d87998a46d9b9774a3dabcb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-23 08:25:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aece0dfbcc0446d598e7b6c333f00b1c\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a762dd2e26204fc4995666cffb6e16df : SUCCESS in 28m 58s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f0a37509144a4b6db25c110b1a3ff4f5 : SUCCESS in 25m 57s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e6b88f31d0af4bc2b05f63fc03dc4412 : SUCCESS in 34m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6c6246ebba970a142d806e18c9aa774b56750687","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-23 08:25:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c1f3effbbf34473aaa6399a66d8ff6b8\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/25f6b4dcc9ac42a49daba10f663f07b7 : SUCCESS in 3m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1fd77e533a81419784b78a38eaf0fde8 : SUCCESS in 15m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ec91e3e8bb84464b2da91bd507788d4 : SUCCESS in 6m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c4a1e634737a4cf59197227e363fd39e : SUCCESS in 12m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/981cefbbab3b44a8bb34923cdc632b2b : SUCCESS in 12m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7e2d23cc16524a868755d2ac9cee1579 : SUCCESS in 15m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f281381c9a34492bb72a58220575f57 : SUCCESS in 12m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/76d677b2ecfb46a3ac5ad9441223eaaf : SUCCESS in 11m 30s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/d2064f0645c242ab81c31540ba9da85a : SUCCESS in 30m 16s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/330945e5cb454ab9888c12d5acb47bae : FAILURE in 24m 44s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/57cf9a42d0754c82adcb9825a31f05a3 : FAILURE in 28m 28s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/494266074a1f430f9949bdf5cdaccd13 : SUCCESS in 24m 10s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/6f15a1c8db8d441988da01464bbbf461 : SUCCESS in 15m 02s (non-voting)\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/e39b2df8f1824b59999e0f0068f3e2e2 : FAILURE in 35m 14s (non-voting)\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/891e503310344a839d568a7ca92d2dd2 : SUCCESS in 7m 17s (non-voting)\n- keystone-dsvm-functional-oidc-federation https://zuul.opendev.org/t/openstack/build/1c981dd584cf42a0aecbe3bea694371f : SUCCESS in 29m 29s","accounts_in_message":[],"_revision_number":1},{"id":"5f33a8bbfb6b6af46b243c051be9870969e00bef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-23 08:26:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9b9d6fcca9e47f6835b7ab0531551e825cf877e5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-23 09:50:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f6c9a28fdeb347f2acbcf353451a121e\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/9ef0c46f2f594cc0802c1c58322a4819 : SUCCESS in 2m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ce0f3b09c72445280389807926ee098 : SUCCESS in 8m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6119600ca2c041829cb1ef83c3f59601 : SUCCESS in 12m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2edec70a5b2b4266a5e5f31b68719826 : SUCCESS in 12m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3eab4fce3cf4347a898dbc06c17ac67 : SUCCESS in 13m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b2418f990b3543c2bf9fc80e72388181 : SUCCESS in 4m 54s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/66e170216e834736aee67a779442e02a : SUCCESS in 11m 25s\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/8c296c5338014526acb253a9c460081f : SUCCESS in 21m 08s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/fedb23de179246ec8fcfbe35c9c2e763 : FAILURE in 37m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cd7e9255c677eb3845055bc98ddbae6b83a90779","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-23 09:50:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"957f9ff0a3bee302c2a75f249a9df752e520ea91","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-23 09:51:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3e6e95e6f5904ee6996c9fcdb233b149\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44f66c7ad61940cf8a07582d8eae260b : SUCCESS in 57s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/940be9be7e3943d2819a6193a04d7075 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0d2cc1a3af4dbd2825cef5992056bffe935eaadd","revisions":{"0d2cc1a3af4dbd2825cef5992056bffe935eaadd":{"kind":"REWORK","_number":1,"created":"2025-05-09 11:46:36.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/29/949329/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/29/949329/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/29/949329/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/29/949329/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/29/949329/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/29/949329/1"}}},"commit":{"parents":[{"commit":"037bf7c4d5fc671b0b822d611e850abc702281d0","subject":"wsgi: Don\u0027t create, use lock in same line","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/037bf7c4d5fc671b0b822d611e850abc702281d0"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-05-09 09:43:32.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-05-09 11:31:41.000000000","tz":60},"subject":"setup: Remove pbr\u0027s wsgi_scripts","message":"setup: Remove pbr\u0027s wsgi_scripts\n\nThis is technical dead end and not something we\u0027re going to be able to\nsupport long-term in pbr. We need to push users away from this. Doing so\nhighlights quite a few place where our docs need some work.\n\nChange-Id: I9fa7cd60bf27768db06f69681675fce2e55712af\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/0d2cc1a3af4dbd2825cef5992056bffe935eaadd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/0d2cc1a3af4dbd2825cef5992056bffe935eaadd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
