)]}'
{"id":"openstack%2Fkeystone~951088","triplet_id":"openstack%2Fkeystone~master~Ifbe487ff911ebc006f9639c4d8894a16864e3f2f","project":"openstack/keystone","branch":"master","hashtags":[],"change_id":"Ifbe487ff911ebc006f9639c4d8894a16864e3f2f","subject":"Fixed the lists numbering","status":"NEW","created":"2025-05-28 07:33:40.000000000","updated":"2025-05-28 07:47:30.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"7a259fc58dbdf3bbe6c345c47d7882fb32eca157","_number":951088,"virtual_id_number":951088,"owner":{"_account_id":38057,"name":"Emmanuel Lécharny","display_name":"Emmanuel Lecharny","email":"emmlec@worteks.com","username":"elecharny","status":"Working at worteks.com, also Apache Software member, working on directory and MINA projects."},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-05-28 07:47:30.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-28 07:47:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"008105d20cf34168bcb493d2b40649f9166a1d90","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38057,"name":"Emmanuel Lécharny","display_name":"Emmanuel Lecharny","email":"emmlec@worteks.com","username":"elecharny","status":"Working at worteks.com, also Apache Software member, working on directory and MINA projects."},"date":"2025-05-28 07:33:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7a259fc58dbdf3bbe6c345c47d7882fb32eca157","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-28 07:47:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd28fd56e5494e88ac876b2fcb1be5c8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b100987c4b34de2be9e645eaae9df7c : SUCCESS in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d3e3a6e21494d9c89b44e90eefa1a86 : SUCCESS in 13m 31s\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/47bd4439cee84846be052a8d376d5fcc : SUCCESS in 8m 25s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"93dcfd4ca414deb6661d8cd413b84134388d4f98","revisions":{"93dcfd4ca414deb6661d8cd413b84134388d4f98":{"kind":"REWORK","_number":1,"created":"2025-05-28 07:33:40.000000000","uploader":{"_account_id":38057,"name":"Emmanuel Lécharny","display_name":"Emmanuel Lecharny","email":"emmlec@worteks.com","username":"elecharny","status":"Working at worteks.com, also Apache Software member, working on directory and MINA projects."},"ref":"refs/changes/88/951088/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/88/951088/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/88/951088/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/88/951088/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/88/951088/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/88/951088/1"}}},"commit":{"parents":[{"commit":"9f47af17e18db29d39027f077bfb710cca51ed22","subject":"Merge \"Remove openSUSE/SLES from install guide\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/9f47af17e18db29d39027f077bfb710cca51ed22"}]}],"author":{"name":"Emmanuel Lécharny","email":"emmlec@worteks.com","date":"2025-05-28 07:33:14.000000000","tz":120},"committer":{"name":"Emmanuel Lécharny","email":"emmlec@worteks.com","date":"2025-05-28 07:33:14.000000000","tz":120},"subject":"Fixed the lists numbering","message":"Fixed the lists numbering\n\nChange-Id: Ifbe487ff911ebc006f9639c4d8894a16864e3f2f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/93dcfd4ca414deb6661d8cd413b84134388d4f98"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/93dcfd4ca414deb6661d8cd413b84134388d4f98"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
