)]}'
{"id":"openstack%2Fkeystone~951327","triplet_id":"openstack%2Fkeystone~master~I7c347af983cb8af9be9d19a010fa0f5bf20ab804","project":"openstack/keystone","branch":"master","topic":"logging","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-05-30 16:08:10.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7c347af983cb8af9be9d19a010fa0f5bf20ab804","subject":"api: Add log when creating unscoped token","status":"MERGED","created":"2025-05-29 17:15:21.000000000","updated":"2025-05-30 16:09:14.000000000","submitted":"2025-05-30 16:08:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"951327-logging","meta_rev_id":"1daba99cc839e93d5fe89d909cf69711c00cd770","_number":951327,"virtual_id_number":951327,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-30 16:08:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"all":[{"value":2,"date":"2025-05-30 14:11:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":2,"date":"2025-05-30 10:52:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"all":[{"value":1,"date":"2025-05-30 14:11:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-29 17:44:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-05-29 18:49:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-30 10:52:27.000000000","updated_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"REVIEWER"},{"updated":"2025-05-30 14:11:11.000000000","updated_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"}],"messages":[{"id":"95eed4e6b37927a06588700ee5086a9bade102c3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-05-29 17:15:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"631fdc4b79824633774cf0b93ee3930849abdec5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-29 17:44:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/412fc92f17ca4a7a8f87cd86bc2ccda7\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2c54cf155714495ea4cc65745c653fa6 : SUCCESS in 17m 36s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8ec706cacaf345e6bc2da474284cb4a5 : SUCCESS in 26m 57s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/e6c1a1e46e53498cbfe961d9e9751ff6 : SUCCESS in 21m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8b330433c374f1948741ad246bfab27b8d52479e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-29 18:49:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bce04130a2b34befa4f3c8e5136319f2\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fe321421bb334fa2b18c8484b1a76585 : SUCCESS in 16m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/640fea44531d4a5dbd93893a914aa6a7 : SUCCESS in 5m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bb860a1ababb40c7a6a0e05c01e24396 : SUCCESS in 10m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3b4244694f804ed0aec08d81f628c205 : SUCCESS in 11m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4ca5686827f24491869e90389baa4765 : SUCCESS in 14m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/861b7360128a4eadacb133f3ca93a611 : SUCCESS in 13m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/c5bc5d1ce75c4d5e949d2ebb566bebe1 : SUCCESS in 1h 03m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7cea43556a9f4e059389ec87230a2f91 : SUCCESS in 1h 20m 36s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/8c3df431782c4d148b8bbc9c2892ccd4 : SUCCESS in 27m 33s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/b9025a7724a743a9bfa83e5a73f45cec : FAILURE in 21m 28s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cc251df65bc14af69af22f93c879c10b : FAILURE in 17m 25s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/db08adec4cbb46e49cd60107b2303ad6 : SUCCESS in 24m 21s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/f58932e69c634d8c8d2fbfc644ad8c92 : SUCCESS in 9m 45s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/6fe0b4112c9843109da0f8580c3fb7a8 : FAILURE in 11m 13s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cd407b0424f04493b1743cceaf16ae8a : SUCCESS in 55m 44s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/d9df54f170844dc685285fb58558548a : FAILURE in 36m 23s (non-voting)\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/364da52e569f40c1834bb4e84c539c84 : SUCCESS in 6m 49s (non-voting)\n- keystone-dsvm-functional-oidc-federation https://zuul.opendev.org/t/openstack/build/f6895d4e588d433d9fa46adce20ca1e8 : SUCCESS in 29m 31s","accounts_in_message":[],"_revision_number":1},{"id":"0d2bd29c0bee3dbd195264ff612a46481d009037","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2025-05-30 10:52:27.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"529896dfd195c38af96f6938e1613b00b5dabfbf","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2025-05-30 14:11:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2c45593f7c489b6cfd5b90118f709597912debe0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-30 14:11:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"383d2a8afe1c014c3cb53b6f9c2b9661bfb7f5c6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-30 16:08:10.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5961c4fdc17c4ece9471509e2f723847\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/926a97fe90574e25b0c2c26feb43a5dc : SUCCESS in 9m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/72ae1a323fd2490994e9eaa663aaafd1 : SUCCESS in 11m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c58a335bc51e4b1799de90d37f45fe4f : SUCCESS in 10m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4591e46a9f1a4a6c9ac124cbd55ad7d6 : SUCCESS in 14m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/e87d17bde1dc4f0c9b3bba8be4694a61 : SUCCESS in 53m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/037acdbdb15b4786b42613304e6a9f76 : SUCCESS in 1h 49m 48s\n- keystone-tempest https://zuul.opendev.org/t/openstack/build/a16bca25390141fb8080cbd6816008c1 : SUCCESS in 24m 00s\n- keystone-tempest-federation-k2k https://zuul.opendev.org/t/openstack/build/ba8ee79a5d064937b6fc315658493d03 : SUCCESS in 18m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f9d142d1c7f04192a34aaae73ad185a4 : SUCCESS in 1h 01m 46s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/5134807a4a654d4bba15d9a3e5126a18 : FAILURE in 39m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c4f31e06380658e20c86a2c8032088f120e34d0b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-30 16:08:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"1daba99cc839e93d5fe89d909cf69711c00cd770","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-30 16:09:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/375ee7d0352845149ed5f64ef4bd5c0b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e8099a51c1f4b998cecd66e8ed1d33c : SUCCESS in 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b6f955b8e40ef3f0bc7388f9c6de406069c6870b","revisions":{"b6f955b8e40ef3f0bc7388f9c6de406069c6870b":{"kind":"REWORK","_number":1,"created":"2025-05-29 17:15:21.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/27/951327/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/27/951327/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/27/951327/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/27/951327/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/27/951327/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/27/951327/1"}}},"commit":{"parents":[{"commit":"069fa4566b9c4835554d993780f1010362f4420b","subject":"Merge \"Allow additional properties in endpoint schema\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/069fa4566b9c4835554d993780f1010362f4420b"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-05-29 17:04:51.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-05-29 17:09:37.000000000","tz":60},"subject":"api: Add log when creating unscoped token","message":"api: Add log when creating unscoped token\n\nOtherwise you get different behavior depending on the user used, if\nsome have default projects set and others do not.\n\nChange-Id: I7c347af983cb8af9be9d19a010fa0f5bf20ab804\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/b6f955b8e40ef3f0bc7388f9c6de406069c6870b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/b6f955b8e40ef3f0bc7388f9c6de406069c6870b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
