)]}'
{"id":"openstack%2Fkeystone~951331","triplet_id":"openstack%2Fkeystone~master~I4ceb0dfabb90f6176688e8633ae411a7e65d4c8b","project":"openstack/keystone","branch":"master","topic":"bug/2111836","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-05-30 11:15:33.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I4ceb0dfabb90f6176688e8633ae411a7e65d4c8b","subject":"RFC: Do not consider other method types if using token auth","status":"ABANDONED","created":"2025-05-29 18:02:36.000000000","updated":"2025-05-30 11:15:33.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6cdddfa5c8cc1a7aaa514b9757a7ec289b4df63a","_number":951331,"virtual_id_number":951331,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-05-29 19:33:33.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"all":[{"value":-2,"date":"2025-05-30 09:39:38.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-29 18:42:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-05-29 19:33:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-30 09:39:38.000000000","updated_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"REVIEWER"}],"messages":[{"id":"3fce9f883d6af9000c9a3f1a0fb6af5dccc63101","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-05-29 18:02:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"928ac97fe7b890c5cf3b46592555a528e4825060","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-29 18:42:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6623bd6a580e4297bdde2153f2c5fb5c\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2f231faa556c48a6a3106bddaf86b948 : FAILURE in 27m 38s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7deb0e7e69fa4b11aba7db74ef35250f : FAILURE in 33m 46s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/950bb93cdc8647a3ac6b6dc28c66d03b : FAILURE in 26m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c68057d93ddce7121d3aec8ef78c01f50bf2f628","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-29 19:33:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b32dc7bf0b5141848f15b3b42e6173d3\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3af9daedb495431c9a1a1bacc548fb58 : FAILURE in 17m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d4e805875574a438b0f54b8e788ec28 : SUCCESS in 6m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4854c822a63f4c56bbf0ce024c07161b : FAILURE in 14m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5efae41bcba540e89a2ad6c7a60105c3 : FAILURE in 13m 05s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2a7749477d5f4bdbac6bb7b16154ca49 : FAILURE in 14m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17e9fcffa290464f8dc2804be20bed6f : SUCCESS in 12m 41s\n- grenade https://zuul.opendev.org/t/openstack/build/4acc31da0316434fbda2f42bf14b08bd : SUCCESS in 52m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/adf6bf77df56496090662c5ae6fbed59 : FAILURE in 1h 20m 08s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/2c4d989ba3684aa3b4a743bd7d21c77d : SUCCESS in 25m 47s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/e1c1e4551fca4566b0c48d83ad6864b8 : FAILURE in 20m 09s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a001b2f7bdbe4dcaade8a1bf20981e18 : FAILURE in 26m 30s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/b704ee13ee4e4d88a8ddac985e910f98 : SUCCESS in 28m 07s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/1adcf8a4f15b422a80688c3403781613 : SUCCESS in 9m 25s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/7e759b91735c4b1ba6bc38c015f51174 : FAILURE in 13m 03s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/db341fb2501748e792595bfa23fb600b : SUCCESS in 58m 53s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/fd79362e9e664214b91438b03c0555b1 : POST_FAILURE in 34m 46s (non-voting)\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/37f94ea8e1134148a58798d496f9e96f : SUCCESS in 7m 39s (non-voting)\n- keystone-dsvm-functional-oidc-federation https://zuul.opendev.org/t/openstack/build/12ab7e95ed954fd49b1aae3952794d14 : SUCCESS in 28m 51s","accounts_in_message":[],"_revision_number":1},{"id":"5c5b3e35e7957a8d69ce06f526768044399c1735","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2025-05-30 09:39:38.000000000","message":"Patch Set 1: Code-Review-2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6cdddfa5c8cc1a7aaa514b9757a7ec289b4df63a","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-05-30 11:15:33.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2e2de6c7184d1532446d9c2fe40fe792df1445d8","revisions":{"2e2de6c7184d1532446d9c2fe40fe792df1445d8":{"kind":"REWORK","_number":1,"created":"2025-05-29 18:02:36.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/31/951331/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/31/951331/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/31/951331/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/31/951331/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/31/951331/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/31/951331/1"}}},"commit":{"parents":[{"commit":"069fa4566b9c4835554d993780f1010362f4420b","subject":"Merge \"Allow additional properties in endpoint schema\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/069fa4566b9c4835554d993780f1010362f4420b"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-05-29 17:50:58.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-05-29 18:01:57.000000000","tz":60},"subject":"RFC: Do not consider other method types if using token auth","message":"RFC: Do not consider other method types if using token auth\n\nRFC because I do not know this code well and therefore have no idea if\nthis is a sensible approach or not.\n\nChange-Id: I4ceb0dfabb90f6176688e8633ae411a7e65d4c8b\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nCloses-bug: #2111836\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/2e2de6c7184d1532446d9c2fe40fe792df1445d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/2e2de6c7184d1532446d9c2fe40fe792df1445d8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Code-Review\u003dMIN"],"failing_atoms":["label:Code-Review\u003dMAX"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
