)]}'
{"id":"openstack%2Fkeystone~951381","triplet_id":"openstack%2Fkeystone~master~I44d2df472b4f977e1ff803ad70a9f565efacfc69","project":"openstack/keystone","branch":"master","topic":"ci-fixes","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-05-30 15:58:26.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I44d2df472b4f977e1ff803ad70a9f565efacfc69","subject":"zuul: Remove keystone-tox-patch_cover job","status":"MERGED","created":"2025-05-30 08:40:12.000000000","updated":"2025-05-30 16:01:39.000000000","submitted":"2025-05-30 15:58:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"951381-ci-fixes","meta_rev_id":"667441871c5a8b7a264d46ab50f1e69546a04d0d","_number":951381,"virtual_id_number":951381,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-30 15:58:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"all":[{"value":2,"date":"2025-05-30 14:09:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":2,"date":"2025-05-30 12:39:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"all":[{"value":1,"date":"2025-05-30 14:09:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-30 09:33:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-05-30 10:35:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-30 12:39:06.000000000","updated_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"REVIEWER"},{"updated":"2025-05-30 14:09:34.000000000","updated_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"reviewer":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"state":"REVIEWER"}],"messages":[{"id":"207c09dde5ad5d430c127ef970c03ba98641659f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-05-30 08:40:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9d93cfb8cbd646f7e0211fc553607ed05ba243f6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-05-30 08:46:01.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"cbed7f88aa77006cdcd5ba21dde3be495b65a8aa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-05-30 08:55:29.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"d15fb0256a08b9e36645683c58da7c2ca2599f09","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-30 09:33:52.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/34446808b3ad4af08aee4fe9ff8d85fa\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/62c711ad755c4502b50acdb6bbac8c8d : SUCCESS in 21m 05s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4bec9e2484b24bbbbec5a3a06d38c1ad : SUCCESS in 26m 22s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/1164ecc315624b159470fd17e20e534d : SUCCESS in 34m 31s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"9d49d420f090f162ce1014c41c49e14af8265a41","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-30 10:35:40.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eae695b414234b188545c789c4da6cff\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ab0938249ad04ce7af0f27bfc9174232 : SUCCESS in 17m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ada914ab6125404bb3b13fb24c1fdaff : SUCCESS in 5m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a89e2720e64d463e97c63127ef6f672d : SUCCESS in 9m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a508b281cefc45c286ed95e23fae9106 : SUCCESS in 12m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/27d3eaa8ce924c9bac18e4fa3ba8ab40 : SUCCESS in 15m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2701fb1c444443f8dd7434359cfd097 : SUCCESS in 11m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/dbe00bce12b74e5592950fd86f5da419 : SUCCESS in 1h 01m 06s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/955efe4a35eb4e3ea484bf43148028c1 : SUCCESS in 1h 33m 56s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/bb1fab171da741ab9faa230752bd058e : SUCCESS in 14m 22s (non-voting)\n- keystone-tempest https://zuul.opendev.org/t/openstack/build/101312454ccc43ac8a7117cfd63f1111 : SUCCESS in 21m 38s\n- keystone-tempest-fips https://zuul.opendev.org/t/openstack/build/45db16f532ee418f920c642711b434f7 : FAILURE in 16m 19s (non-voting)\n- keystone-tempest-federation https://zuul.opendev.org/t/openstack/build/d72dc9bcfe454d169ad3b8ccb31d3240 : FAILURE in 25m 54s (non-voting)\n- keystone-tempest-federation-k2k https://zuul.opendev.org/t/openstack/build/43d3cc50732b40c1895ac4c706772a9d : SUCCESS in 19m 41s\n- keystone-tempest-oidc-federation https://zuul.opendev.org/t/openstack/build/d09cb3efef674b748bd6970296359ab6 : SUCCESS in 32m 01s\n- keystone-tempest-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/6cec6a746ee34ef38c0ea2642cfc1529 : FAILURE in 11m 54s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bcbc281a780041a7b781bbb71c4d7bc7 : SUCCESS in 58m 00s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/3c2894ed69e5438fa2e9ccd608c77784 : FAILURE in 26m 52s (non-voting)\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/f62bc3acc61d4c00b07d5c36509e04dc : SUCCESS in 7m 54s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f24f55aa337bc094ada98675fa1682060e4e0458","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2025-05-30 12:39:06.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"c1db2e83d1b46a37f1dfdd99a3b98e2d1955d4cb","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2025-05-30 14:09:34.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"84d87b7981cd33532178ca983a26ab7c4d038867","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-30 14:11:02.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"21a8d25763bb69bfc286f719def12780b2b7780d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-30 15:58:26.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3eb32bdaeb934fbe874aec4e33eacc40\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f69a470011d4f1a9a38a2e7a1e832a9 : SUCCESS in 7m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/95b92f5095174928b142780d1c7da5a8 : SUCCESS in 11m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7a58f9b2b4484b9e9f26cfa03badda0b : SUCCESS in 11m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e52155da1424ba4b679fb476d282651 : SUCCESS in 12m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/0a80f1d60f66470bb2156798cc72d404 : SUCCESS in 1h 02m 28s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ab63d30783c64142a46a7e2d128bbb0b : SUCCESS in 1h 41m 21s\n- keystone-tempest https://zuul.opendev.org/t/openstack/build/34a8b7e9c04f4493b5a7406099bf344a : SUCCESS in 26m 15s\n- keystone-tempest-federation-k2k https://zuul.opendev.org/t/openstack/build/7f5b4e13ed524b0ea85541c82c415765 : SUCCESS in 24m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6e5d6668462b412ebdaebc3df9103cc6 : SUCCESS in 1h 00m 47s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/9e5e41a6cf8d43cd8ea11e9315d731a9 : FAILURE in 41m 06s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"bcc2bc8243b61c493169f325df46e6f991a3f1b5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-30 15:58:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"667441871c5a8b7a264d46ab50f1e69546a04d0d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-30 16:01:39.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/588f8ef7382640cba8990208b59efc00\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69efb5b2c33c49d887ea0db6716414c7 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"4033945248829c94840801996bd7086c8ba93934","revisions":{"f67d5ec7b4b2bf6c9bb5b7e37f5406a09f2f3246":{"kind":"REWORK","_number":1,"created":"2025-05-30 08:40:12.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/81/951381/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/81/951381/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/81/951381/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/81/951381/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/81/951381/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/81/951381/1"}}},"commit":{"parents":[{"commit":"bef20fd42f172e3ae7d265de4a729a3f4d44340d","subject":"zuul: Rename, reorder tempest jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/bef20fd42f172e3ae7d265de4a729a3f4d44340d"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-05-30 08:36:37.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-05-30 08:40:05.000000000","tz":60},"subject":"zuul: Remove keystone-tox-patch_cover job","message":"zuul: Remove keystone-tox-patch_cover job\n\nThis has been experimental since it was added. It seems clear that it\nwill not graduate to voting at this point.\n\nChange-Id: I44d2df472b4f977e1ff803ad70a9f565efacfc69\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/f67d5ec7b4b2bf6c9bb5b7e37f5406a09f2f3246"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/f67d5ec7b4b2bf6c9bb5b7e37f5406a09f2f3246"}]},"branch":"refs/heads/master"},"b71f6248ea1bd6bdc4d32de794b10050f4bc405f":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-05-30 08:46:01.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/81/951381/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/81/951381/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/81/951381/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/81/951381/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/81/951381/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/81/951381/2"}}},"commit":{"parents":[{"commit":"0734c283d3d26de93e86bb7fcb549032a08a184c","subject":"zuul: Rename, reorder tempest jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/0734c283d3d26de93e86bb7fcb549032a08a184c"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-05-30 08:36:37.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-05-30 08:44:55.000000000","tz":60},"subject":"zuul: Remove keystone-tox-patch_cover job","message":"zuul: Remove keystone-tox-patch_cover job\n\nThis has been experimental since it was added. It seems clear that it\nwill not graduate to voting at this point.\n\nChange-Id: I44d2df472b4f977e1ff803ad70a9f565efacfc69\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/b71f6248ea1bd6bdc4d32de794b10050f4bc405f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/b71f6248ea1bd6bdc4d32de794b10050f4bc405f"}]},"branch":"refs/heads/master"},"4033945248829c94840801996bd7086c8ba93934":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2025-05-30 08:55:29.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/81/951381/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/81/951381/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/81/951381/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/81/951381/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/81/951381/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/81/951381/3"}}},"commit":{"parents":[{"commit":"9bde660aea8114dd7a825966cdc9d5a885e7176c","subject":"zuul: Rename, reorder tempest jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/9bde660aea8114dd7a825966cdc9d5a885e7176c"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-05-30 08:36:37.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-05-30 08:54:17.000000000","tz":60},"subject":"zuul: Remove keystone-tox-patch_cover job","message":"zuul: Remove keystone-tox-patch_cover job\n\nThis has been experimental since it was added. It seems clear that it\nwill not graduate to voting at this point.\n\nChange-Id: I44d2df472b4f977e1ff803ad70a9f565efacfc69\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/4033945248829c94840801996bd7086c8ba93934"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/4033945248829c94840801996bd7086c8ba93934"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
