)]}'
{"id":"openstack%2Fkeystone~953104","triplet_id":"openstack%2Fkeystone~master~Ie3b44527c956fb6d3a2114b14a49e95abd408ccd","project":"openstack/keystone","branch":"master","hashtags":[],"change_id":"Ie3b44527c956fb6d3a2114b14a49e95abd408ccd","subject":"Fix json schema of user groups","status":"MERGED","created":"2025-06-23 13:50:43.000000000","updated":"2025-06-23 16:56:36.000000000","submitted":"2025-06-23 16:55:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"953104","meta_rev_id":"e4f89f7ca8c61d7fe48a9c16079fc27c83f146ae","_number":953104,"virtual_id_number":953104,"owner":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-06-23 16:55:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"all":[{"value":2,"date":"2025-06-23 15:23:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"all":[{"value":1,"date":"2025-06-23 15:23:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-23 15:22:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5b2502ede6ffeda58fdabfaf6dba8aefb788fd54","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2025-06-23 13:50:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e614f852fddab003fbd1d1996898b547bdcc5301","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-23 15:22:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/175c25bdf32341c2a84a5fb1f305f13a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6e95ff7787c44703858147f1525009c9 : SUCCESS in 16m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7914d75abf72428397bf8eaedaf10f3e : SUCCESS in 9m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3a751824e4b3484faff94542d034c3a6 : SUCCESS in 11m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9e73ecd50fb14a3f81247a66e250909e : SUCCESS in 13m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a5c79b978a844fb0b1d047826ad892c3 : SUCCESS in 15m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7b260d9d35547a89643b0f939128452 : SUCCESS in 13m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/c4029c4ed00f45f3a9e3a35513e47fdc : SUCCESS in 58m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c84c3ac185d74c909413907f6debc722 : SUCCESS in 1h 29m 21s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/23bddf2f0b164e0384520ae8cfc0f6e8 : FAILURE in 11m 06s (non-voting)\n- keystone-tempest https://zuul.opendev.org/t/openstack/build/50e6a18ff6284472b00c0b69edcdc8c5 : SUCCESS in 25m 23s\n- keystone-tempest-fips https://zuul.opendev.org/t/openstack/build/988e9044853e4619b18cefe08f6e7879 : FAILURE in 12m 36s (non-voting)\n- keystone-tempest-federation https://zuul.opendev.org/t/openstack/build/cc25131f40274d61883718fdf323f16c : FAILURE in 26m 20s (non-voting)\n- keystone-tempest-federation-k2k https://zuul.opendev.org/t/openstack/build/a9a06c4e3903406eb83c5dd9ab471db3 : SUCCESS in 26m 51s\n- keystone-tempest-oidc-federation https://zuul.opendev.org/t/openstack/build/8d3a0525fcac4ab5840adb6ac3af5a01 : SUCCESS in 27m 53s\n- keystone-tempest-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/28892225c7654006a1f2ddcd4e17b661 : FAILURE in 17m 44s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f4296ea599dc44818a306a2916cea720 : SUCCESS in 52m 32s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/d684802d19c64f439aaf6fb6813e960e : FAILURE in 40m 07s (non-voting)\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/806201a7f8fb48e7b5d9a602e8dc7405 : SUCCESS in 8m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"19d4fdc4ec9491054e869c9cf0a9d5778a4fd813","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2025-06-23 15:23:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"80252696659addd6aea9eadef82aba007a67c86c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-23 15:23:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6e843a7124842d778ce9f337290b507430e0c050","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-23 15:29:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9bdc21e523c04030ba5e9e360c99eaab\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/cbe31960c8af4cee94ac5905a50d3aae : SUCCESS in 40m 38s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d17fad3f017c4a899fe9c11a78c2170e : SUCCESS in 38m 54s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4a1a44d7971e4cb4ab0f41b6531686b8 : SUCCESS in 48m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9f7aba6c0647757e0169b27dc9f905a78d43dd15","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-23 16:55:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b775644b3e1f470daa8d4baa954522b5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e26d71b8ff74490b9437081d3c90992b : SUCCESS in 6m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2012ce688fa344daa875e9e4c3ed85ef : SUCCESS in 10m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0daebfeabd9e4e2caef96fc80a2188b6 : SUCCESS in 11m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83bfc969cbbf40e7b68b2b223d686439 : SUCCESS in 12m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/f1ab2e1fe24d40caa91cb19b892c9939 : SUCCESS in 47m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a953353472794fa9af06a92d25027557 : SUCCESS in 1h 30m 54s\n- keystone-tempest https://zuul.opendev.org/t/openstack/build/0f160cc8766f457e83ef7aee68e6231d : SUCCESS in 29m 16s\n- keystone-tempest-federation-k2k https://zuul.opendev.org/t/openstack/build/d47f806007ff448e9d2c5b997ccbd2fc : SUCCESS in 26m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ea8980c22f2f41a4b3c02caa29bd5c37 : SUCCESS in 49m 10s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/85c6a62bd2544bb8acfe2df58ac4e488 : FAILURE in 40m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5dd5dc8fd9e89e2a41c4b33ba238a41d7d47b529","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-23 16:55:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e4f89f7ca8c61d7fe48a9c16079fc27c83f146ae","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-23 16:56:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1fb0aad052594d9792c1ef16062cb5f4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e03130d4ce6346ae930faa3e41bf513b : SUCCESS in 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"398e05c213261acf53d56176d7322e706fc991aa","revisions":{"398e05c213261acf53d56176d7322e706fc991aa":{"kind":"REWORK","_number":1,"created":"2025-06-23 13:50:43.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/04/953104/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/04/953104/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/04/953104/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/04/953104/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/04/953104/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/04/953104/1"}}},"commit":{"parents":[{"commit":"72230d60fdb9f3d1086aa7cab7ac6f818abeca04","subject":"Merge \"trust schema: don\u0027t require project_id to be in uuid format\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/72230d60fdb9f3d1086aa7cab7ac6f818abeca04"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2025-06-23 13:48:01.000000000","tz":120},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2025-06-23 13:50:36.000000000","tz":120},"subject":"Fix json schema of user groups","message":"Fix json schema of user groups\n\n2 properties are placed on the wrong level.\n\nChange-Id: Ie3b44527c956fb6d3a2114b14a49e95abd408ccd\nSigned-off-by: Artem Goncharov \u003cartem.goncharov@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/398e05c213261acf53d56176d7322e706fc991aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/398e05c213261acf53d56176d7322e706fc991aa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
