)]}'
{"id":"openstack%2Fkeystone~956176","triplet_id":"openstack%2Fkeystone~master~I82f2f35630ac315e8f57b2faa873b763679354ac","project":"openstack/keystone","branch":"master","attention_set":{},"removed_from_attention_set":{"13478":{"account":{"_account_id":13478,"name":"Boris Bobrov","email":"b.bobrov@sap.com","username":"bbobrov"},"last_update":"2026-02-11 09:37:54.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I82f2f35630ac315e8f57b2faa873b763679354ac","subject":"Invalidate cache if expiring groups change","status":"ABANDONED","created":"2025-07-30 15:01:23.000000000","updated":"2026-02-11 09:37:54.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"84ec639779c0cd248337f4e2e3bbfa3445de5f81","_number":956176,"virtual_id_number":956176,"owner":{"_account_id":13478,"name":"Boris Bobrov","email":"b.bobrov@sap.com","username":"bbobrov"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-07-30 16:35:00.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-30 16:35:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"978e362e6ba7f308b3bc83d330ad9606dd13ca53","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13478,"name":"Boris Bobrov","email":"b.bobrov@sap.com","username":"bbobrov"},"date":"2025-07-30 15:01:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a225d7f48fb0e516fc3b5464550f7ce89bced62a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-30 16:35:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/84a3e9d5d9084331afd29ce1cb0023a0\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/505f6a2048ee48f9b4ee3b959433fcb8 : FAILURE in 9m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b0cca3724c84ab08c9949223f3e8ae9 : FAILURE in 3m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/628d1ab4fee0408c9561aed6d9ac6916 : FAILURE in 7m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/afcfe707e363450db00a0a55b44e8cfb : FAILURE in 7m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/630e11a5747749ef88d5c8416896ab6e : FAILURE in 9m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/440ef17af7784e8cb785d3a36d96b228 : SUCCESS in 6m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/0d420477915d402593c80175c6ef245f : SUCCESS in 29m 19s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/35afebe0004f4054891a716b84a1a33d : SUCCESS in 1h 27m 03s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/c719c4ffbe7040f9a06cece4f454dfad : SUCCESS in 14m 13s (non-voting)\n- keystone-tempest https://zuul.opendev.org/t/openstack/build/9ba2213a03ce42b19ca86637df3a0d52 : SUCCESS in 21m 56s\n- keystone-tempest-fips https://zuul.opendev.org/t/openstack/build/487673447054499a9f98504d9ad73ef2 : FAILURE in 20m 35s (non-voting)\n- keystone-tempest-federation https://zuul.opendev.org/t/openstack/build/dca74a717c6a4159b24af51926808a46 : FAILURE in 24m 37s (non-voting)\n- keystone-tempest-federation-k2k https://zuul.opendev.org/t/openstack/build/d02855f2df3c42edb74fcf5b8cea4d0d : FAILURE in 11m 21s\n- keystone-tempest-oidc-federation https://zuul.opendev.org/t/openstack/build/66bef36deec5458c97742fd07c4e648d : FAILURE in 29m 54s\n- keystone-tempest-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/fa0ed56f41c5432391fb836b5d7221d4 : FAILURE in 18m 10s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b9d0dd3277744109a777e75cd7e6b738 : SUCCESS in 45m 53s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/f11807e5cbad452c9371281e642b4a61 : FAILURE in 39m 29s (non-voting)\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/8bb9b5094c774a5a93629ab545f8933e : SUCCESS in 7m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"51f2e811c1becfd9f785f7943d148fb9fe206a9e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-30 17:27:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d37812995d8841ba8e4d7deb401054b5\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/8189d757f9d041b58bcf4c6aca826c6f : TIMED_OUT in 1h 01m 03s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8bf40d4c445047cf9b854f2e8d19a839 : FAILURE in 52m 26s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/6eeadc7ea3a94bab9b993a416079f5ef : TIMED_OUT in 1h 01m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"84ec639779c0cd248337f4e2e3bbfa3445de5f81","tag":"autogenerated:gerrit:abandon","author":{"_account_id":13478,"name":"Boris Bobrov","email":"b.bobrov@sap.com","username":"bbobrov"},"date":"2026-02-11 09:37:54.000000000","message":"Abandoned\n\ndone differently","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cab1ba6b7dd63f213b0bca2f90dd70b6651c97d4","revisions":{"cab1ba6b7dd63f213b0bca2f90dd70b6651c97d4":{"kind":"REWORK","_number":1,"created":"2025-07-30 15:01:23.000000000","uploader":{"_account_id":13478,"name":"Boris Bobrov","email":"b.bobrov@sap.com","username":"bbobrov"},"ref":"refs/changes/76/956176/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/76/956176/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/76/956176/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/76/956176/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/76/956176/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/76/956176/1"}}},"commit":{"parents":[{"commit":"e066e18abf5dc9a061fe2fc479a851aa0b65c794","subject":"Merge \"Remove the password element from the generated dict in update_user\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/e066e18abf5dc9a061fe2fc479a851aa0b65c794"}]}],"author":{"name":"Boris Bobrov","email":"b.bobrov@sap.com","date":"2025-07-30 14:58:09.000000000","tz":120},"committer":{"name":"Boris Bobrov","email":"b.bobrov@sap.com","date":"2025-07-30 15:01:09.000000000","tz":120},"subject":"Invalidate cache if expiring groups change","message":"Invalidate cache if expiring groups change\n\nInvalidate assignment cache when adding users to expiring groups. This\nhappens only after the set of groups changes.\n\nChange-Id: I82f2f35630ac315e8f57b2faa873b763679354ac\nCloses-Bug: 2119031\nSigned-off-by: Boris Bobrov \u003cb.bobrov@sap.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/cab1ba6b7dd63f213b0bca2f90dd70b6651c97d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/cab1ba6b7dd63f213b0bca2f90dd70b6651c97d4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
