)]}'
{"id":"openstack%2Fkeystone~959279","triplet_id":"openstack%2Fkeystone~stable%2F2024.1~I4a16401b914c92fd7db9d626cb1642570b36d600","project":"openstack/keystone","branch":"stable/2024.1","topic":"migrate-password-hashes","attention_set":{},"removed_from_attention_set":{"35986":{"account":{"_account_id":35986,"name":"Adrian Jarvis","email":"adrian.r.jarvis@gmail.com","username":"adrian_jarvis"},"last_update":"2025-09-15 21:53:36.000000000","reason":"\u003cGERRIT_ACCOUNT_35986\u003e replied on the change","reason_account":{"_account_id":35986,"name":"Adrian Jarvis","email":"adrian.r.jarvis@gmail.com","username":"adrian_jarvis"}},"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2025-10-31 13:45:34.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I4a16401b914c92fd7db9d626cb1642570b36d600","subject":"Migrates deprecated password hashes.","status":"ABANDONED","created":"2025-09-03 05:45:19.000000000","updated":"2025-10-31 13:45:34.000000000","total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1a2356e9e0e3a4ec7214cd38ca3aeb7c050f5222","_number":959279,"virtual_id_number":959279,"owner":{"_account_id":35986,"name":"Adrian Jarvis","email":"adrian.r.jarvis@gmail.com","username":"adrian_jarvis"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-09-04 03:07:35.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-03 06:23:38.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2025-09-03 06:28:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-09-03 07:18:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-03 19:23:26.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2025-09-03 19:48:26.000000000","updated_by":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"reviewer":{"_account_id":23084,"name":"Bartosz Bezak","email":"bartosz@stackhpc.com","username":"b.bezak"},"state":"CC"},{"updated":"2025-10-31 13:45:34.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"}],"messages":[{"id":"6746eb2a20176d72112aa961f790911068df9a91","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35986,"name":"Adrian Jarvis","email":"adrian.r.jarvis@gmail.com","username":"adrian_jarvis"},"date":"2025-09-03 05:45:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3fc00886454dbf5d1b35ceb532d1732b9ead8c65","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35986,"name":"Adrian Jarvis","email":"adrian.r.jarvis@gmail.com","username":"adrian_jarvis"},"date":"2025-09-03 05:55:26.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8530a30352403f955aa4f5f5c58ff8c584692efc","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":35986,"name":"Adrian Jarvis","email":"adrian.r.jarvis@gmail.com","username":"adrian_jarvis"},"date":"2025-09-03 05:58:33.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":2},{"id":"2544aa6a494bc0590aa19bc169921ffa4a9537f3","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-09-03 06:23:38.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"189e3f66474c1e6f9c16741a96e8d1772b5381d9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-03 06:28:33.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/18bad81ba93745ec840e16fef6a76970\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/15c8b01dffca415c9bc0a18a89fd4602 : FAILURE in 31m 17s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/41ec93c204ef4cd9acf4c558b1eec8a3 : FAILURE in 24m 48s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3e662b5f198c3992b26ed1b7152447aeb5251f3a","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-09-03 06:32:55.000000000","message":"Patch Set 2: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b9b258d5dc8ae3ea88436ebef2193798f97bbe54","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-03 07:18:45.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/63354356a6b14c7c8a05e83583dedb65\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e540b78c22184e66af05e8300133a487 : FAILURE in 13m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7cebe945924d451dba6a47f1e82f27d8 : FAILURE in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b4abd8022e614cb78754fc8775168773 : FAILURE in 7m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a1175655d6ea46a296a8cc263f04365d : FAILURE in 13m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8e663f7269f4ae4af67b64dd168bb1c : SUCCESS in 13m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/2b0bca0ab63a450981f5046df8122907 : SUCCESS in 52m 41s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/3b371123c62045259575331151ab175c : SUCCESS in 1h 21m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ecd984d4f0254bc78af0194f56fab1d5 : SUCCESS in 9m 42s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/fb0de27dc1ee4dc7b75cdde08d2d1d38 : SUCCESS in 22m 17s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/99f03b4767524e0e8e0c74b2f311a856 : FAILURE in 39m 27s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/59cd67a784824e5390b20bb0fe77b251 : FAILURE in 16m 13s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/39e790bfb31d412eb4eaec013d1f0684 : SUCCESS in 12m 51s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/d5b5c440cf8f49a4ab1d2750bfca6a56 : SUCCESS in 7m 24s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/0c8f0d2401e7481fad7505432add57e2 : FAILURE in 11m 48s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f8979b7a927f436d8287a8e93e3ce9d8 : SUCCESS in 27m 01s\n- keystone-dsvm-functional-oidc-federation https://zuul.opendev.org/t/openstack/build/52c2f02544494acba4bdd97195a5dc4d : SUCCESS in 21m 16s","accounts_in_message":[],"_revision_number":2},{"id":"8735f5335a668e3370773c523722e04780119a0f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":35986,"name":"Adrian Jarvis","email":"adrian.r.jarvis@gmail.com","username":"adrian_jarvis"},"date":"2025-09-04 01:14:35.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"8a9e969e22cb77e7227cb4b6a3889e66f49e9b27","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-04 01:39:47.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/51648ccba7944ec2b731123cf8702cbc\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/359776b72daf45f7a2cc0b615faacf6b : SUCCESS in 23m 15s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ac7d06f13f2b483590f791356521bab0 : SUCCESS in 20m 37s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"5abc3178b034360d165742ff67f6860219c210b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-04 03:07:35.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/64c703868fad47efb6718d69be0627f5\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e9f7cfa9dc3743779f418454b3005e9a : SUCCESS in 10m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92200c46d37b4aee89ae0caf47e5402d : FAILURE in 4m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/944cfb67df2249e5b2c1c945f08d3b7d : SUCCESS in 10m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8e879dd64dd248fc9dfb65c740b79885 : SUCCESS in 12m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc590a958e214aeb93b0eca0190221e8 : SUCCESS in 5m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/c84d9cc311bb453a9f49e60fc3cf99be : SUCCESS in 1h 05m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/70e105cc1cec49079da839fef3d898fd : SUCCESS in 1h 48m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3c4c6fcec4324573af704e414152f1e9 : SUCCESS in 13m 04s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/bd8454645a0d4959bd493d4192c5515d : SUCCESS in 13m 19s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/5a5576689bc1429cb6a9b68d351ff0f6 : FAILURE in 16m 53s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2fa4b89517524ad49e048c0fa7007522 : FAILURE in 25m 51s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/d2a3eed64e1b488397fe8af1570c7bc4 : SUCCESS in 26m 54s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/0791a5393c0946adba3505206a029a77 : SUCCESS in 15m 00s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/c96fe7e35b134dcbab0549cb0e411029 : FAILURE in 18m 01s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c70713b9064b40ce950ab7fa169438f9 : SUCCESS in 56m 15s\n- keystone-dsvm-functional-oidc-federation https://zuul.opendev.org/t/openstack/build/da759e73ec6e4527a7d0dbe0851f611a : SUCCESS in 31m 01s","accounts_in_message":[],"_revision_number":3},{"id":"c49a13faeb9853f16ff7e14bdbbe23cecb946b5e","author":{"_account_id":35986,"name":"Adrian Jarvis","email":"adrian.r.jarvis@gmail.com","username":"adrian_jarvis"},"date":"2025-09-15 21:53:36.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"56d1ee274d8fa7148315e9d6b9912ba58aa4b183","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":35986,"name":"Adrian Jarvis","email":"adrian.r.jarvis@gmail.com","username":"adrian_jarvis"},"date":"2025-09-15 21:53:46.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":3},{"id":"335a29d6ae59eff6202462b4b26cfb303e80f4fc","author":{"_account_id":35986,"name":"Adrian Jarvis","email":"adrian.r.jarvis@gmail.com","username":"adrian_jarvis"},"date":"2025-09-18 05:14:28.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1a2356e9e0e3a4ec7214cd38ca3aeb7c050f5222","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-10-31 13:45:34.000000000","message":"Abandoned\n\nstable/2024.1 branch of openstack/keystone is about to be deleted. To be able to do that, all open patches need to be abandoned. Please cherry pick the patch to unmaintained/2024.1 if you want to further work on this patch.","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"ad7231fff7beea1a2b9b164706fe522311b71e0b","revisions":{"0ef35001a4d687ac7a97a1311e2175d9d5f423a8":{"kind":"REWORK","_number":1,"created":"2025-09-03 05:45:19.000000000","uploader":{"_account_id":35986,"name":"Adrian Jarvis","email":"adrian.r.jarvis@gmail.com","username":"adrian_jarvis"},"ref":"refs/changes/79/959279/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/79/959279/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/79/959279/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/79/959279/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/79/959279/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/79/959279/1"}}},"commit":{"parents":[{"commit":"4e2e36ad7767a459277987d21f748cec54ac09e0","subject":"Fix DB migrations after alembic integration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/4e2e36ad7767a459277987d21f748cec54ac09e0"}]}],"author":{"name":"Adrian Jarvis","email":"adrian.jarvis@catalystcloud.nz","date":"2025-09-01 05:15:24.000000000","tz":720},"committer":{"name":"Adrian Jarvis","email":"adrian.jarvis@catalystcloud.nz","date":"2025-09-03 05:45:15.000000000","tz":720},"subject":"Migrates deprecated password hashes.","message":"Migrates deprecated password hashes.\n\nWith Epoxy release the support for the sha512_crypt hash is dropped.\nThis change adds a check of the password hash when the user authenticates.\nIf the hash of the users password is deprecated then the password will be\nre-hashed using the default hasher and updated in the database.\n\nChange-Id: I4a16401b914c92fd7db9d626cb1642570b36d600\nSigned-off-by: Adrian Jarvis \u003cadrian.jarvis@catalystcloud.nz\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/0ef35001a4d687ac7a97a1311e2175d9d5f423a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/0ef35001a4d687ac7a97a1311e2175d9d5f423a8"}]},"branch":"refs/heads/stable/2024.1"},"46ca6953831d667c7e58d563ef122699384cae43":{"kind":"REWORK","_number":2,"created":"2025-09-03 05:55:26.000000000","uploader":{"_account_id":35986,"name":"Adrian Jarvis","email":"adrian.r.jarvis@gmail.com","username":"adrian_jarvis"},"ref":"refs/changes/79/959279/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/79/959279/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/79/959279/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/79/959279/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/79/959279/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/79/959279/2"}}},"commit":{"parents":[{"commit":"4e2e36ad7767a459277987d21f748cec54ac09e0","subject":"Fix DB migrations after alembic integration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/4e2e36ad7767a459277987d21f748cec54ac09e0"}]}],"author":{"name":"Adrian Jarvis","email":"adrian.jarvis@catalystcloud.nz","date":"2025-09-01 05:15:24.000000000","tz":720},"committer":{"name":"Adrian Jarvis","email":"adrian.jarvis@catalystcloud.nz","date":"2025-09-03 05:55:15.000000000","tz":720},"subject":"Migrates deprecated password hashes.","message":"Migrates deprecated password hashes.\n\nWith Epoxy release the support for the sha512_crypt hash is dropped.\nThis change adds a check of the password hash when the user authenticates.\nIf the hash of the users password is deprecated then the password will be\nre-hashed using the default hasher and updated in the database.\n\nChange-Id: I4a16401b914c92fd7db9d626cb1642570b36d600\nSigned-off-by: Adrian Jarvis \u003cadrian.jarvis@catalystcloud.nz\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/46ca6953831d667c7e58d563ef122699384cae43"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/46ca6953831d667c7e58d563ef122699384cae43"}]},"branch":"refs/heads/stable/2024.1"},"ad7231fff7beea1a2b9b164706fe522311b71e0b":{"kind":"REWORK","_number":3,"created":"2025-09-04 01:14:35.000000000","uploader":{"_account_id":35986,"name":"Adrian Jarvis","email":"adrian.r.jarvis@gmail.com","username":"adrian_jarvis"},"ref":"refs/changes/79/959279/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/79/959279/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/79/959279/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/79/959279/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/79/959279/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/79/959279/3"}}},"commit":{"parents":[{"commit":"4e2e36ad7767a459277987d21f748cec54ac09e0","subject":"Fix DB migrations after alembic integration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/4e2e36ad7767a459277987d21f748cec54ac09e0"}]}],"author":{"name":"Adrian Jarvis","email":"adrian.jarvis@catalystcloud.nz","date":"2025-09-01 05:15:24.000000000","tz":720},"committer":{"name":"Adrian Jarvis","email":"adrian.jarvis@catalystcloud.nz","date":"2025-09-04 01:14:17.000000000","tz":720},"subject":"Migrates deprecated password hashes.","message":"Migrates deprecated password hashes.\n\nWith Epoxy release the support for the sha512_crypt hash is dropped.\nThis change adds a check of the password hash when the user authenticates.\nIf the hash of the users password is deprecated then the password will be\nre-hashed using the default hasher and updated in the database.\n\nChange-Id: I4a16401b914c92fd7db9d626cb1642570b36d600\nSigned-off-by: Adrian Jarvis \u003cadrian.jarvis@catalystcloud.nz\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/ad7231fff7beea1a2b9b164706fe522311b71e0b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/ad7231fff7beea1a2b9b164706fe522311b71e0b"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
