)]}'
{"id":"openstack%2Fkeystone~960526","triplet_id":"openstack%2Fkeystone~stable%2F2025.2~I2a72ee5e990a6e26320fbd0caf10d93e03cc8624","project":"openstack/keystone","branch":"stable/2025.2","topic":"create-2025.2","hashtags":[],"change_id":"I2a72ee5e990a6e26320fbd0caf10d93e03cc8624","subject":"Update .gitreview for stable/2025.2","status":"MERGED","created":"2025-09-11 12:36:46.000000000","updated":"2025-09-12 08:42:44.000000000","submitted":"2025-09-12 08:41:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"960526-create-2025.2","meta_rev_id":"c2dfc7651ee0758f50332eafd6de77afc6ab6c41","_number":960526,"virtual_id_number":960526,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-12 08:41:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2025-09-12 08:24:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2025-09-12 08:24:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-11 13:02:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-12 08:24:29.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"d56a2ee514bea0ee25f7feb4c232c19143e4dcdf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2025-09-11 12:36:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a0d856801328cac735b1abf49dd136bc813a6ced","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-11 13:02:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/90c6676d854444f3bac87099088a5fd2\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aa96b5ec8aad4655993f0ed7ee529d73 : SUCCESS in 16m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95924e3deddc48dcb084bfd28faaae46 : SUCCESS in 6m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b5c53e38faac4a7c822cff0f0ef0649d : SUCCESS in 7m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1414fe3d825d4d169306df5a3c728083 : SUCCESS in 9m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/766cfef4860448948916bdace9ebb81a : SUCCESS in 14m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd86a4292c454e619e0a6bd17d4284ba : SUCCESS in 14m 08s\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/369baaab05934089bf61128eb34352f2 : SUCCESS in 6m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9adc7f0d67529775070a76b81d285c781339f7d3","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-09-12 08:24:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ee51acfbaa5fb77e2e41607fcdd23721905b0a90","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-12 08:24:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ab1e509c4cb046aad5a675a8b3d9284be932d050","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-12 08:41:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c83f83b208fc46aeae2a0dac664d3089\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae7df4b607df4ac1bc84c7564a919b22 : SUCCESS in 7m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/557125cddab74d528e8a8b205e62e0d3 : SUCCESS in 11m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c265c56d791a45a6b5396bb9a4c1fe34 : SUCCESS in 11m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e128a513a87a4680b08f7ea71eaebe7f : SUCCESS in 12m 40s","accounts_in_message":[],"_revision_number":1},{"id":"ef97e368053581dcc587d9c4ae9af868003f46bd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-12 08:41:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c2dfc7651ee0758f50332eafd6de77afc6ab6c41","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-12 08:42:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4edf8f0d9ed247dbb5da68eeec557036\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8e704bc136147989fc0120725e7c49c : SUCCESS in 55s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"38f294559261c3f4c5296fb7f004555141d715ad","revisions":{"38f294559261c3f4c5296fb7f004555141d715ad":{"kind":"REWORK","_number":1,"created":"2025-09-11 12:36:46.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/26/960526/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/26/960526/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/26/960526/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/26/960526/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/26/960526/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/26/960526/1"}}},"commit":{"parents":[{"commit":"a568938e0c967a56ade269c93f42718008487b14","subject":"Merge \"Fix AD nested groups issues\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/a568938e0c967a56ade269c93f42718008487b14"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2025-09-11 12:36:44.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2025-09-11 12:36:44.000000000","tz":0},"subject":"Update .gitreview for stable/2025.2","message":"Update .gitreview for stable/2025.2\n\nChange-Id: I2a72ee5e990a6e26320fbd0caf10d93e03cc8624\nSigned-off-by: OpenStack Release Bot \u003cinfra-root@openstack.org\u003e\nGenerated-By: openstack/project-config:roles/copy-release-tools-scripts/files/release-tools/functions\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/38f294559261c3f4c5296fb7f004555141d715ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/38f294559261c3f4c5296fb7f004555141d715ad"}]},"branch":"refs/heads/stable/2025.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
