)]}'
{"id":"openstack%2Fkeystone~965135","triplet_id":"openstack%2Fkeystone~stable%2F2024.2~I36bb9ab43c3263b5064f35f4fc550722a0e83279","project":"openstack/keystone","branch":"stable/2024.2","topic":"remove-wsgi_scripts","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-10-30 20:45:09.000000000","reason":"Change was submitted"},"27900":{"account":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"last_update":"2025-10-30 13:14:05.000000000","reason":"\u003cGERRIT_ACCOUNT_27900\u003e replied on the change","reason_account":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}}},"hashtags":[],"change_id":"I36bb9ab43c3263b5064f35f4fc550722a0e83279","subject":"Add new keystone.wsgi module","status":"MERGED","created":"2025-10-28 15:20:18.000000000","updated":"2025-10-30 20:46:11.000000000","submitted":"2025-10-30 20:45:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"965135-remove-wsgi_scripts","meta_rev_id":"21c24155cd356a42bdeeccbf0787f0a9c699a00d","_number":965135,"virtual_id_number":965135,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-30 20:45:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"all":[{"value":2,"date":"2025-10-30 15:44:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"all":[{"value":1,"date":"2025-10-30 15:44:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-28 15:20:18.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"REVIEWER"},{"updated":"2025-10-28 16:22:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"afcf107694f6d150f2db42277e212feea45fb238","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-10-28 15:20:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"59e39e8bfd3d8cba9f6306487590526cbcb9068c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-28 16:22:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7c07ddbc1b6744fd9678637c0e453cc4\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ca79256ed1234b3e8d3f6ba8ac2f9011 : SUCCESS in 17m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87d68c94b87d442fb61f57be3691e236 : SUCCESS in 7m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e4102afebcf543d19d7c5d350dedd487 : SUCCESS in 15m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e1ae772dea854994b6a6645fcab95993 : SUCCESS in 10m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/85760222afaa457e961a0683a938c8d8 : SUCCESS in 11m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01f0ae6202f74c44886d9e92e74cf4ed : SUCCESS in 13m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/9df6532d67184e269aa11e07e4aa61dc : SUCCESS in 56m 37s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/475a742f082c4bbaacdf0649244a479e : FAILURE in 17m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a57b3f97b9764501affd0e54ed6c75b8 : SUCCESS in 12m 48s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/417bbad36e064517a8a8003b9ead46c9 : FAILURE in 18m 37s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/b61334a0dac94a488a29b9a558b7d252 : FAILURE in 40m 12s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/4a3d922bda674c479662d8186970f1e9 : FAILURE in 17m 13s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/0295e80ef4594cdebdfb1406ccdbd87b : FAILURE in 17m 44s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/ad800399725c4541acfad8ad52dc803f : POST_FAILURE in 12m 42s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/037642a0cacf4e44969a34b161db46b4 : FAILURE in 18m 27s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e4e7a063f69f4da984ce59007f4e15e0 : FAILURE in 8m 52s\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/a262e83c2fc64d8c8ed3019b98b88b62 : FAILURE in 5m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"dcc7ef59d8168e9ac39c19c3bde491eafb19f3c9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-28 17:01:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/783a04268ff04ae89fc1b4908b2afbdf\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c65921662c1d4eafa034bc6e7ddaa5be : SUCCESS in 31m 08s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e959392592c44896978ba25e441ab941 : SUCCESS in 17m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2b66c7d6ba96b71986e7fad28b8d007b7b58c14b","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2025-10-30 13:14:05.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"02b1a7ddf9e40ebf2e917dd3c543d7dae4fb9a35","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-30 14:00:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c92ebaccbe34edca341b96e012c70ae\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7af34c644ba54c45981e2c4de3dcb985 : SUCCESS in 26m 23s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/db9f255b3c3d43da826ab261d23e4ed3 : SUCCESS in 14m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ce86e5192b629fcde1738573c40f59ae1d6265cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-30 14:56:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f48a80d32a62476b90ada8f67b701027\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f5a19d2258044fd9945524c256ba36e6 : SUCCESS in 14m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e9b631c1bcc48c881e4dcb94ab6238e : SUCCESS in 6m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cca2d3bc35ae4a779332fa34ebfdb219 : SUCCESS in 14m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/96f811b8a0d94cbeb1565414fb26360c : SUCCESS in 13m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ee3d4d80117b419e82696a2169ade91d : SUCCESS in 10m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26071e9cac044f0cb6041b29178e80ac : SUCCESS in 10m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/4b50fc3c043b4492bd12d94f983f78a7 : SUCCESS in 33m 10s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/1688586595094dd787f00e8cd75d737d : SUCCESS in 1h 37m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5c33a07e5f354764b47788934911f3a6 : SUCCESS in 11m 41s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/dbbddbd9413b4fd28e7278b31adf617f : SUCCESS in 14m 59s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/e938b79fb0994455a7b8a84c94c61b5d : FAILURE in 15m 51s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f1a0cd10c98c4c249f07c345707a59ba : FAILURE in 14m 44s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/02d1c676a48548c4a610e335a6dc9448 : SUCCESS in 27m 30s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/05da22c8f6e84231a783b0cd10d3b530 : SUCCESS in 14m 02s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/720fb816efcb4308930854bfcb1b625a : FAILURE in 10m 06s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/21b33b21f4ce4b1299f1659c4fe1b6f3 : SUCCESS in 46m 16s\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/12b381708f2f486a9583ad5c7469c670 : FAILURE in 6m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"daf6db8a7de90ed116e470bfa539df533680a80a","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2025-10-30 15:44:27.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"610baaa5e9c26de0543caa80a9d99a515d35386c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-30 15:45:43.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5fca146eb94297d966b2697ade6d4cb9bea95f46","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-30 16:29:01.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/12f933503e394a559f0c8e3bb41c7ba7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70de9158d5a44ece86f39d317bfcbce0 : SUCCESS in 3m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5a1229dc651a4fa781d0ebe645dc5414 : SUCCESS in 9m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/22db9abb1ac143dc90b365b9c89c7421 : SUCCESS in 7m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9eb4a015b4d048dfa30b842f46772adb : SUCCESS in 6m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/d3af65e6d3b44513a927f05a936fbbaa : SUCCESS in 29m 48s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/c31aa4aa69ad4186ac1cd211c50ed222 : FAILURE in 42m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5e76c7a5b000426e81afabea7d3ac414 : SUCCESS in 6m 34s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/1eb3124cfa064bafa798e9aff5a22417 : SUCCESS in 13m 54s\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/8dd75289c42b4a098518dfc4d59c38dc : SUCCESS in 13m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/af351a296be847e093bf1c4feffc6233 : SUCCESS in 30m 30s","accounts_in_message":[],"_revision_number":1},{"id":"c9722fd2bda3efbdf2b0cde768cfc20e1d5962e9","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2025-10-30 16:36:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f378e83bdf264283edfb57b317e9f4b4d9b96398","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-30 17:53:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2dfa4bbcd78c4391aa83335864768f96\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/650e17e2365944d9bbd9f342c12ed794 : SUCCESS in 26m 21s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3724fb8dbdc6414faf05f30a579cd344 : SUCCESS in 14m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e03ae2377b4d5e2c9c4be0dcbf9945d0cbbb1af6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-30 18:11:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1ed3c6e3fb274a8cac3002c703dce3b7\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/07a272d67b4041f38763cbeca7732ac4 : SUCCESS in 14m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1959b1e9f7b437bb4d0de1eda296fb7 : SUCCESS in 6m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2dd6a6e520a742b8a3a19b309d1ff819 : SUCCESS in 14m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f1cea24d7d8c4cbd9917e815f60ac1b4 : SUCCESS in 15m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cbc89ab3a117407d9eda9be18fa3dd9e : SUCCESS in 11m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5fd105ceea1c4916ad45941ec6f62634 : SUCCESS in 6m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/e51167c609a74852b9c03dc48847a143 : SUCCESS in 1h 06m 16s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/50ae4aa2067548eead1d7852990a8b6d : SUCCESS in 1h 29m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b81ac6d04b3b4ac9821dfeb73f3481fe : SUCCESS in 13m 08s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/3eb2116b185a45c49a5682f052305112 : SUCCESS in 25m 24s\n- keystone-dsvm-py3-functional-fips https://zuul.opendev.org/t/openstack/build/1d281015ba11412cba9d680fc438446e : FAILURE in 40m 17s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cd179181ff64456eb892e1e772d94ec1 : FAILURE in 25m 14s (non-voting)\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/94b45fadd9f94e6db1f3f97ca795b7e7 : SUCCESS in 28m 38s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/dba0d837aed14315b393f00ec22f2cfc : SUCCESS in 14m 01s (non-voting)\n- keystone-dsvm-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/08899f7f9d244c909bb98c36e9bf9cb0 : FAILURE in 16m 52s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c57f943b22e54c55abf8ded8fbd3e8bd : SUCCESS in 51m 55s\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/32ea4dde8e00499180081bc7fba0be71 : FAILURE in 5m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"beb020cb189d516d5b3f20bb3ba265c330f50995","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-30 18:11:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a360188fe1ce749b34454375fa8272352de10e15","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-30 20:45:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/731e34e5ecde4044863d7abebf39af48\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5ad2161f10d470a97125b24f9dc66c5 : SUCCESS in 6m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3def67027f074cc6b8f87bb71a55edde : SUCCESS in 9m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3c8f8394220749aa838750e8f0867987 : SUCCESS in 7m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/faecb2c3354049bd8b94fb27b281370a : SUCCESS in 8m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/f8867605c0184ca58fc30ccf6efa9275 : SUCCESS in 1h 03m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8d855853ee564543bb31b0f70b30c4aa : SUCCESS in 1h 26m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9cacc7017ddb4e7c96c9b927cba43af5 : SUCCESS in 8m 43s\n- keystone-dsvm-py3-functional https://zuul.opendev.org/t/openstack/build/a984541dc72a4bbe8fa1371ee5f77715 : SUCCESS in 26m 48s\n- keystone-dsvm-py3-functional-federation-ubuntu-jammy-k2k https://zuul.opendev.org/t/openstack/build/aaaed2726e8f44539c185c1b022b38f8 : SUCCESS in 28m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6dd9bb6de2eb447d81f8a790133b894d : SUCCESS in 53m 06s","accounts_in_message":[],"_revision_number":1},{"id":"28c1f15b3b0cf1f4879d56ecb075aefb05a62bdf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-30 20:45:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"21c24155cd356a42bdeeccbf0787f0a9c699a00d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-30 20:46:11.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/206622e98dbe4f57a4699a8c761ec31b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be460f2166c34080ba8f51ee0e8019b0 : SUCCESS in 49s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b27b00e7e7414158864a3ad3a08d146f : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"eaf73231a46b7a1bd6b594e8a8c86f997f5bed65","revisions":{"eaf73231a46b7a1bd6b594e8a8c86f997f5bed65":{"kind":"REWORK","_number":1,"created":"2025-10-28 15:20:18.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/35/965135/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/35/965135/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/35/965135/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/35/965135/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/35/965135/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/35/965135/1"}}},"commit":{"parents":[{"commit":"daaf698ba6d3987290176c77cd622101cdae2547","subject":"Ignore typing on the single import","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/daaf698ba6d3987290176c77cd622101cdae2547"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-10-10 12:02:32.000000000","tz":120},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-10-28 15:20:05.000000000","tz":0},"subject":"Add new keystone.wsgi module","message":"Add new keystone.wsgi module\n\nThis allows deployment tooling to easily switch from passing a binary\npath to passing a Python module path. We\u0027ll use it shortly.\n\nChange-Id: I36bb9ab43c3263b5064f35f4fc550722a0e83279\nSigned-off-by: Artem Goncharov \u003cartem.goncharov@gmail.com\u003e\n(cherry picked from commit 57c6b3187fc47e7d5cf6f1b5cde10fd867edb49f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/eaf73231a46b7a1bd6b594e8a8c86f997f5bed65"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/eaf73231a46b7a1bd6b594e8a8c86f997f5bed65"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
