)]}'
{"id":"openstack%2Fkeystone~971194","triplet_id":"openstack%2Fkeystone~master~I6bd50b75e907586cc5db3ba14c8e21a43a0ec712","project":"openstack/keystone","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-02-06 18:12:10.000000000","reason":"Change was submitted"},"27900":{"account":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"last_update":"2026-02-06 16:14:05.000000000","reason":"\u003cGERRIT_ACCOUNT_27900\u003e replied on the change","reason_account":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}}},"hashtags":[],"change_id":"I6bd50b75e907586cc5db3ba14c8e21a43a0ec712","subject":"Replace deprecated warn method","status":"MERGED","created":"2025-12-16 15:02:11.000000000","updated":"2026-02-06 18:13:40.000000000","submitted":"2026-02-06 18:12:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"971194","meta_rev_id":"1a42625185825481e2ab2fa4a56c9e3a75f2bbe5","_number":971194,"virtual_id_number":971194,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13478,"name":"Boris Bobrov","email":"b.bobrov@sap.com","username":"bbobrov"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-06 18:12:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"recommended":{"_account_id":13478,"name":"Boris Bobrov","email":"b.bobrov@sap.com","username":"bbobrov"},"all":[{"value":1,"date":"2026-02-05 13:31:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13478,"name":"Boris Bobrov","email":"b.bobrov@sap.com","username":"bbobrov"},{"value":2,"date":"2025-12-17 14:25:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":2,"date":"2026-02-06 16:14:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-12-23 22:39:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"all":[{"value":0,"_account_id":13478,"name":"Boris Bobrov","email":"b.bobrov@sap.com","username":"bbobrov"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":1,"date":"2026-02-06 16:14:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13478,"name":"Boris Bobrov","email":"b.bobrov@sap.com","username":"bbobrov"},{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-16 15:36:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-12-16 16:58:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-12-17 14:25:38.000000000","updated_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"},{"updated":"2025-12-23 22:39:01.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2025-12-23 22:39:01.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"REVIEWER"},{"updated":"2026-02-05 13:31:34.000000000","updated_by":{"_account_id":13478,"name":"Boris Bobrov","email":"b.bobrov@sap.com","username":"bbobrov"},"reviewer":{"_account_id":13478,"name":"Boris Bobrov","email":"b.bobrov@sap.com","username":"bbobrov"},"state":"REVIEWER"}],"messages":[{"id":"74f364177d70d809ce81d0107452c00d6bc88d15","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-12-16 15:02:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eceeb728cbbee6010ffe30fdbcc53ff7f22de54e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-16 15:36:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cf60df64a18d4651aadc3647aa23adba\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f4336f07fb6040e0a50d6a18e71340bc : SUCCESS in 23m 16s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b99f4413b12548f8b59cd47f8723c9ce : SUCCESS in 17m 32s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/14f8b4118a5c4cfaa66703627dd6df1a : SUCCESS in 32m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b874d375b749b62cb7f705510d34df8c567c44c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-16 16:58:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c77ede29a6834f0dbdcd244153133b51\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/37882b2422f645e8a19c81ad3ca6b2b3 : SUCCESS in 11m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28301b7980f6485cab0a3eedb39f5552 : FAILURE in 4m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4f64552b4968488c8631e8726f2335a9 : SUCCESS in 8m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/391edcfda5fe483ba29dda3d122b1edc : SUCCESS in 10m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09718ea9c61c4bada6213db3165523d9 : SUCCESS in 14m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/0b763fc29ea94eb0812ab04b04119b60 : SUCCESS in 57m 02s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4c1d1371405a4aefa9e619c17607c514 : SUCCESS in 1h 50m 18s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/9836a47308ab448695b913ad463188e5 : SUCCESS in 17m 59s (non-voting)\n- keystone-tempest https://zuul.opendev.org/t/openstack/build/f1282fb4b7cc42bc80e668ceeb81169c : SUCCESS in 31m 11s\n- keystone-tempest-fips https://zuul.opendev.org/t/openstack/build/a60877d8e91e493f959b4a6c03b8fa06 : FAILURE in 14m 09s (non-voting)\n- keystone-tempest-federation https://zuul.opendev.org/t/openstack/build/b5b6370376714fda9aa8143ee8f975d7 : FAILURE in 22m 59s (non-voting)\n- keystone-tempest-federation-k2k https://zuul.opendev.org/t/openstack/build/939b3dff0425446fbcdcb1f94b378fad : SUCCESS in 14m 21s\n- keystone-tempest-oidc-federation https://zuul.opendev.org/t/openstack/build/7108458e746c47d0a83004cf6c270dd3 : SUCCESS in 22m 13s\n- keystone-tempest-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/40e93b552b0f43a5a8d324a2ae82f368 : SUCCESS in 35m 38s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d78b115defe044fbb7c2cc3d461c6061 : SUCCESS in 44m 21s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/be1681eb02714b0cbe15e940785f6368 : FAILURE in 26m 33s (non-voting)\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/f55132c27d1e4a49ae1480458a5f293b : SUCCESS in 3m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1747af69862f0dffa96354f86cd2fd7afddf35a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-12-16 22:46:30.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"17b7ac5cb5f675d08f6aa2c8df4966a682eb1d92","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-16 23:20:42.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/69d20d802789447fbaf7e912dc1204e3\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/fe3e5b1318fc47658918b8376ac1d27a : SUCCESS in 19m 15s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0be73cc94b13439f82fd68debbc4f9a8 : SUCCESS in 25m 47s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/cc0c341eb375406da9eff9cefd570589 : SUCCESS in 32m 12s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c70972cdf65a2ec79a7f6daf7c06e3b1b559932a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-17 00:37:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5a44d789d6e14a50b2da173262c859f5\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f4c7dc9df1cf4cf397bfa79aa4ba3502 : SUCCESS in 10m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a2f3c57d24441a88b9c85d13102c789 : SUCCESS in 3m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b5b101b9175e463e9c5336c4db2c5124 : SUCCESS in 10m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7ece39e7d42b44b6adee2ca551577061 : SUCCESS in 10m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4553ed3a5db4c8caa97ddeae406e5a0 : SUCCESS in 13m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/42dbb006ef6f4442a21496dc66930e9b : SUCCESS in 29m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/812caf1b057f41d3b96455779bcd70c2 : SUCCESS in 1h 41m 48s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/e103eef6218b48778c8cc9e0713ae8b4 : SUCCESS in 8m 32s (non-voting)\n- keystone-tempest https://zuul.opendev.org/t/openstack/build/a4163da3651f48008167cf0a90a814ee : SUCCESS in 18m 53s\n- keystone-tempest-fips https://zuul.opendev.org/t/openstack/build/70f59fafaadd4d63b1876d57df4cb6a6 : FAILURE in 21m 24s (non-voting)\n- keystone-tempest-federation https://zuul.opendev.org/t/openstack/build/3f9f1f23d2794bc2ab31603592a2f7ef : FAILURE in 24m 57s (non-voting)\n- keystone-tempest-federation-k2k https://zuul.opendev.org/t/openstack/build/49e92a3f838d4d23bb25b200162124db : SUCCESS in 29m 19s\n- keystone-tempest-oidc-federation https://zuul.opendev.org/t/openstack/build/9c2ae9efb8854cad969f14aef71432fb : SUCCESS in 30m 53s\n- keystone-tempest-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/a61fa370e07a41049af528dfcae5ab0d : SUCCESS in 31m 25s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ec4bd15696f34aae869e6736d0d53164 : SUCCESS in 56m 30s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/7eb44571312f426da1d0c39b24aff9b2 : FAILURE in 19m 50s (non-voting)\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/2e36e00cd0304b528d20acd667ef4db8 : SUCCESS in 6m 44s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"89bb26c6e16a735b0e19bd66de1c6b7cc99779fd","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2025-12-17 14:25:38.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"b689acc9e5c3f40811188704ffc78b26ec15c6f9","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-23 22:39:01.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"f83183c7d1c7ab3d693cef56cbc6b27baa73f2c3","author":{"_account_id":13478,"name":"Boris Bobrov","email":"b.bobrov@sap.com","username":"bbobrov"},"date":"2026-02-05 13:31:34.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"87425b307401339659025ae1ddc2b28658ae2f73","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2026-02-06 16:14:05.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"32352ff5a681e66c08b492dcb6fa5d06e8403fbc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-06 16:14:41.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4936bf30fa8c707048b085f2a44bf02aed452a17","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-06 18:12:09.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2446be6a3ce9422387a6164847c98749\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0c51cbfb17f42c397589da0ef23dfba : SUCCESS in 2m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1d4455bfc0de49d4bfa004fcbe1ca2f3 : SUCCESS in 7m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e9608877e72c40dc94e68046571ec3c3 : SUCCESS in 11m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6622adc9b31400598124246064e76b0 : SUCCESS in 6m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/f0aa946292044fdc8276997bd3be13f0 : SUCCESS in 52m 38s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/475e673131224fe6b80c53e867f689e6 : SUCCESS in 1h 52m 35s\n- keystone-tempest https://zuul.opendev.org/t/openstack/build/79ab06e3a8894ce08ee6c3e4b26f7a28 : SUCCESS in 19m 53s\n- keystone-tempest-federation-k2k https://zuul.opendev.org/t/openstack/build/9c6017642c2842f8ad4224edca77f857 : SUCCESS in 17m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8598e62fef934dbc8cc01476bf77e700 : SUCCESS in 53m 57s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/2d0f4e8f36e14af6a0e66e237db0a9cd : FAILURE in 28m 17s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7867dc94871839c478d93e7608b2709ede7d37f6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-06 18:12:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"1a42625185825481e2ab2fa4a56c9e3a75f2bbe5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-06 18:13:40.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9174b0454fc04e33bb64e106a26a9786\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c412a19f29a495398e170e6edd77a3e : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d9d20173bb1a10700ab457c5a7c61e634dd5dffb","revisions":{"dc280c02a67e7b3efdf229ea27a1c21d6ebe3bdf":{"kind":"REWORK","_number":1,"created":"2025-12-16 15:02:11.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/94/971194/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/94/971194/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/94/971194/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/94/971194/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/94/971194/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/94/971194/1"}}},"commit":{"parents":[{"commit":"19e1b3fc91d960de232dbc4d356e681ad12317d1","subject":"Merge \"Drop unused tempest from test requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/19e1b3fc91d960de232dbc4d356e681ad12317d1"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-12-16 15:01:36.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-12-16 15:01:56.000000000","tz":540},"subject":"Replace deprecated warn method","message":"Replace deprecated warn method\n\nThe warn method of logger instances has been deprecated. Use warning\ninstead.\n\nAlso delay string interpolation in logs for efficiency.\n\nChange-Id: I6bd50b75e907586cc5db3ba14c8e21a43a0ec712\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/dc280c02a67e7b3efdf229ea27a1c21d6ebe3bdf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/dc280c02a67e7b3efdf229ea27a1c21d6ebe3bdf"}]},"branch":"refs/heads/master"},"d9d20173bb1a10700ab457c5a7c61e634dd5dffb":{"kind":"REWORK","_number":2,"created":"2025-12-16 22:46:30.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/94/971194/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/94/971194/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/94/971194/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/94/971194/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/94/971194/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/94/971194/2"}}},"commit":{"parents":[{"commit":"19e1b3fc91d960de232dbc4d356e681ad12317d1","subject":"Merge \"Drop unused tempest from test requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/19e1b3fc91d960de232dbc4d356e681ad12317d1"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-12-16 15:01:36.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-12-16 22:46:20.000000000","tz":540},"subject":"Replace deprecated warn method","message":"Replace deprecated warn method\n\nThe warn method of logger instances has been deprecated. Use warning\ninstead.\n\nAlso delay string interpolation in logs for efficiency.\n\nChange-Id: I6bd50b75e907586cc5db3ba14c8e21a43a0ec712\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/d9d20173bb1a10700ab457c5a7c61e634dd5dffb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/d9d20173bb1a10700ab457c5a7c61e634dd5dffb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
