)]}'
{"id":"openstack%2Fkeystone~977303","triplet_id":"openstack%2Fkeystone~master~Ie22885d19dbc0acb60f335454b988a1c29ac9157","project":"openstack/keystone","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-02-19 23:13:19.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"Ie22885d19dbc0acb60f335454b988a1c29ac9157","subject":"Move keystone-tempest-federation to noble","status":"MERGED","created":"2026-02-19 09:56:47.000000000","updated":"2026-02-20 03:07:53.000000000","submitted":"2026-02-20 03:05:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"977303","meta_rev_id":"a6b4565636463166ae8b4a3800896ae74d9b6de0","_number":977303,"virtual_id_number":977303,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-20 03:05:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"recommended":{"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"},"all":[{"value":1,"date":"2026-02-19 14:06:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"},{"value":2,"date":"2026-02-19 11:02:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":2,"date":"2026-02-19 11:10:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"all":[{"value":0,"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"},{"value":0,"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"value":1,"date":"2026-02-19 11:10:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-19 10:24:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-02-19 11:02:11.000000000","updated_by":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"reviewer":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"state":"REVIEWER"},{"updated":"2026-02-19 11:10:28.000000000","updated_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"reviewer":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"state":"REVIEWER"},{"updated":"2026-02-19 11:36:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-19 14:06:36.000000000","updated_by":{"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"},"reviewer":{"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"},"state":"REVIEWER"}],"messages":[{"id":"3553b1fde8c4891c5740aa9e263aba48e2a2e6f3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-02-19 09:56:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"37dd86a19d9380c1a7259fd41c11707c58bd8427","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-19 10:24:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/183b3305291e476eb8d6838c7aabdb4e\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3f199a6f5ab146b4ac19bcbd41be27af : POST_FAILURE in 25m 22s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/abb71030e5264f38968cc73482b16aa4 : SUCCESS in 24m 59s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c8c95e8f017844459f9bfde91d66fb33 : RETRY_LIMIT in 5m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fb6dcd3b39438f86e55a0fb25acdaca99c880cc0","author":{"_account_id":14250,"name":"Grzegorz Grasza","email":"xek@redhat.com","username":"xek"},"date":"2026-02-19 11:02:11.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a58bdbb69e177de42d738dff1f8ba657180e4c56","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2026-02-19 11:10:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5dade5ce0fea6b7709f871f781a05f8cdcab0f20","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-19 11:36:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d13e9b3d91584ff5a219efbc2536608a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8682d71e821e4ed196ac348a4d5951ce : SUCCESS in 11m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6e4af23c3024914b751824ea3f3d982 : SUCCESS in 2m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ca81c2c6b3cd49889057b893eda76216 : POST_FAILURE in 11m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e3199b55673e4b5a861b0bd4b9e25fcd : SUCCESS in 6m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ae2f160128741139233a080f46730e4 : SUCCESS in 7m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/425de64c32ad48b2b752ee4891aa3c02 : POST_FAILURE in 1h 04m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/25b73792dba9480ba72311da758cd7b2 : SUCCESS in 1h 36m 47s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/4feb185007954f0b8f62cc5228383374 : SUCCESS in 14m 15s (non-voting)\n- keystone-tempest https://zuul.opendev.org/t/openstack/build/93bb834302734f0ca216f0c0be450109 : SUCCESS in 31m 17s\n- keystone-tempest-fips https://zuul.opendev.org/t/openstack/build/049dc5f7e34649e08178a5174f595622 : FAILURE in 19m 37s (non-voting)\n- keystone-tempest-federation https://zuul.opendev.org/t/openstack/build/bcfb471300cd4bb9ac10827b789c556f : FAILURE in 18m 50s (non-voting)\n- keystone-tempest-federation-k2k https://zuul.opendev.org/t/openstack/build/938bf2cc397649ab8f45fbc99cd239d8 : SUCCESS in 29m 29s\n- keystone-tempest-oidc-federation https://zuul.opendev.org/t/openstack/build/bbcb370405d2438fa1015e2313ebf863 : SUCCESS in 30m 12s\n- keystone-tempest-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/8d38099c9c59488185ddd62dd4a23009 : SUCCESS in 23m 58s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8376e16717e74f20b1b4f939ffc01254 : SUCCESS in 1h 00m 39s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/922233fc05dd497780a0585f6a1ed127 : FAILURE in 17m 13s (non-voting)\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/8cfa435b0ad64be282d8abfc2513bbe0 : SUCCESS in 5m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7ce078e26b8de8d5c9ef4e5ccaf75e9b57271630","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-02-19 11:38:59.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"11f036a0917cdb4030c3194b8811443a87a863f5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-19 12:18:30.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a1540662a214bfd91955e597e683ebe\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/fa3149bf922343d3a81d15f862353167 : POST_FAILURE in 26m 25s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c79ed6d5da8f4855836ae8dfba7037b7 : SUCCESS in 24m 41s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c088cf8d865c4981af9e2712230796cf : RETRY_LIMIT in 4m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"86c411df98a26fd4bc7c07aec9cee2deb2f4251a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-19 13:26:06.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/37c6832affc344a88a82117685867e0f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5fc8b1f6f85449e7a0e3f84fcefb6a97 : SUCCESS in 17m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01c538fb4db74a99b4691deb2a7fc59b : SUCCESS in 2m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/065d919edea94d42b5f42d5c49e5bbf7 : POST_FAILURE in 11m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bb8a37ac956b46f0bc5c05d34c2e8bed : SUCCESS in 9m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/343f180759a94cf69b52f7483d154224 : SUCCESS in 13m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/f1d11d3d5770440b87ebdbbb10d1a44a : POST_FAILURE in 48m 58s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f310b4992db8432482403cb0537b08bb : SUCCESS in 1h 44m 56s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/2b54214e1b8b44ae80c923e7e39db4f2 : SUCCESS in 14m 26s (non-voting)\n- keystone-tempest https://zuul.opendev.org/t/openstack/build/c3894985feb0404dbc47af466a0b34f7 : SUCCESS in 23m 58s\n- keystone-tempest-fips https://zuul.opendev.org/t/openstack/build/259157e4f58b45b9af7770ac5a182bf6 : FAILURE in 16m 17s (non-voting)\n- keystone-tempest-federation https://zuul.opendev.org/t/openstack/build/10e324df72d3428ebe1d92f1216cc93d : FAILURE in 16m 45s (non-voting)\n- keystone-tempest-federation-k2k https://zuul.opendev.org/t/openstack/build/88013f6ca4964245aa457ca612b6d1dc : SUCCESS in 17m 40s\n- keystone-tempest-oidc-federation https://zuul.opendev.org/t/openstack/build/b7199af83e7e4c03b84e94293dc4fb7b : SUCCESS in 31m 40s\n- keystone-tempest-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/acc082fda32843aaa4164378bac9c6d8 : SUCCESS in 38m 45s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e5777c6ba4bb412f89d378c7b9946d5c : SUCCESS in 1h 03m 47s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/0dad81eae04e451e8b120d1037afc657 : FAILURE in 48m 24s (non-voting)\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/30015deebcd4407189d3808651b2297d : SUCCESS in 8m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7cbab8fd9a269c1199ee6c5145a731d2d77b20fa","author":{"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"},"date":"2026-02-19 14:06:36.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"7ac898707767b42d254a819ab181fb1adb39d1b3","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-02-19 23:13:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4bbbd100c51d3c3e6a0c1d057ad1c69968aa9e5b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-19 23:37:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8ab78a6d84454ec09746cb9f1afbe906\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/3667c6bbf6354c668f7279c8fc03710d : POST_FAILURE in 16m 16s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8ff5e7df877e4ff2a9f47c70fd48f4bc : SUCCESS in 21m 39s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c775afcb5c3f403b9014e2b3dfd30580 : RETRY_LIMIT in 3m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2af2f227254f4b0f133344c810243887d147ed04","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-20 01:03:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e366a648edec44568e26568fe2d23fff\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/17be759fa9bb4e1ba571ce77c3870567 : SUCCESS in 18m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7cf502d7ffdb42cda940667e74c65f66 : SUCCESS in 6m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/24cf14f6070f4ee48febe164c390da88 : SUCCESS in 7m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c3f94a6658d344de93d8a48e2b4f0af0 : SUCCESS in 7m 57s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/81b53e7ab16d4f43b97c21dbed787d77 : SUCCESS in 12m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca0d99d9ed2347c0b1f0fa329e90f454 : SUCCESS in 12m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/2a6934c6578d49069b51564122c6a341 : SUCCESS in 37m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f378716e51964cbb807cc678caaea550 : SUCCESS in 1h 44m 45s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/8954ec9ba5f74398b13183ce7b2c3854 : SUCCESS in 9m 28s (non-voting)\n- keystone-tempest https://zuul.opendev.org/t/openstack/build/d9fabfd87e56409588e5d9597ad82c06 : SUCCESS in 32m 40s\n- keystone-tempest-fips https://zuul.opendev.org/t/openstack/build/0557e4f594d347939e7ddab63f431fb1 : FAILURE in 9m 13s (non-voting)\n- keystone-tempest-federation https://zuul.opendev.org/t/openstack/build/7566d73db53448e5be9e97e4d2e6038e : FAILURE in 12m 31s (non-voting)\n- keystone-tempest-federation-k2k https://zuul.opendev.org/t/openstack/build/94988879f5e14a788808dd5d8159f82b : SUCCESS in 23m 03s\n- keystone-tempest-oidc-federation https://zuul.opendev.org/t/openstack/build/3ee5e7548bdb491db7e7928cefb3658f : SUCCESS in 33m 25s\n- keystone-tempest-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/38e34b925f2b4f7682e9cee86c1e3ee5 : SUCCESS in 22m 52s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1281e6aae8364d3093474192e4a35e25 : SUCCESS in 1h 01m 47s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/85a0a1a915264060a28818df05d8cef2 : FAILURE in 17m 14s (non-voting)\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/4a6fea979a2e4ba6ba81165c0e302acb : SUCCESS in 3m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ae27a689e7fdf00eb5bac9e5f888182139354aaf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-20 01:05:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0a4792502eb72ab3771f682ad75bcf20cbf8a9b1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-20 03:05:10.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d39491b1bf9e44a9a78f9c33fc465600\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25feaf9784364f99a8b0291211139eb1 : SUCCESS in 3m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/704796a5cc7b42328910876903f56283 : SUCCESS in 11m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b60c3978636f49ef97968ca0a4b72057 : SUCCESS in 11m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07f81f74061f4a1aadb6424f53ae2d9e : SUCCESS in 7m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/c710f685ad0544cfa0dcf9225bb59f64 : SUCCESS in 42m 11s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8c6d2bd678d747a2a6ddd209ce940ee8 : SUCCESS in 1h 54m 01s\n- keystone-tempest https://zuul.opendev.org/t/openstack/build/47d18037cf3142729c3d15bed80b9fc7 : SUCCESS in 36m 43s\n- keystone-tempest-federation-k2k https://zuul.opendev.org/t/openstack/build/9f0668fc3e5c4ed8ba85fb3fa64aab1c : SUCCESS in 32m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/61ad1273be8947a7b87bd32caa4997cb : SUCCESS in 1h 02m 40s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/fb35ab2d2b0244da9a3850659576ec40 : FAILURE in 48m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9364b0b3be04e670625e9a0a1fbf496853d7c0a9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-20 03:05:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a6b4565636463166ae8b4a3800896ae74d9b6de0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-20 03:07:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe0be31e44524218bc26b6d77c7f5cd3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ac5825109ba48c08be969cef78204d7 : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"456c24bbd334026392401fd2771a0e86b4ec04c9","revisions":{"456c24bbd334026392401fd2771a0e86b4ec04c9":{"kind":"REWORK","_number":1,"created":"2026-02-19 09:56:47.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/03/977303/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/03/977303/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/03/977303/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/03/977303/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/03/977303/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/03/977303/1"}}},"commit":{"parents":[{"commit":"a903b505429410435aab4c96db6e89413929fcf3","subject":"Merge \"Improve federated mapping documentation clarity\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/a903b505429410435aab4c96db6e89413929fcf3"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-02-19 09:55:46.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-02-19 09:56:38.000000000","tz":540},"subject":"Move keystone-tempest-federation to noble","message":"Move keystone-tempest-federation to noble\n\nSupport for Ubuntu Jammy was removed from devstack by [1].\n\n[1] ea0fdd2c806cf93e48a698e68165306da45314a2\n\nChange-Id: Ie22885d19dbc0acb60f335454b988a1c29ac9157\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/456c24bbd334026392401fd2771a0e86b4ec04c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/456c24bbd334026392401fd2771a0e86b4ec04c9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":35125,"name":"Mauricio Harley","email":"mharley@redhat.com","username":"mharley-rh"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
