)]}'
{"id":"openstack%2Fkeystone~985931","triplet_id":"openstack%2Fkeystone~master~I2d9d04113c9c8dfca37c64f0e35798e9f1d8e6e1","project":"openstack/keystone","branch":"master","topic":"allow-insecure-fernet-key-repository","attention_set":{"13478":{"account":{"_account_id":13478,"name":"Boris Bobrov","email":"b.bobrov@sap.com","username":"bbobrov"},"last_update":"2026-05-01 08:42:58.000000000","reason":"\u003cGERRIT_ACCOUNT_39086\u003e replied on the change","reason_account":{"_account_id":39086,"name":"Matthias Haag","display_name":"Matthias Haag","email":"matthias.haag@uhurutec.com","username":"mauhau"}}},"removed_from_attention_set":{"39086":{"account":{"_account_id":39086,"name":"Matthias Haag","display_name":"Matthias Haag","email":"matthias.haag@uhurutec.com","username":"mauhau"},"last_update":"2026-05-01 08:42:58.000000000","reason":"\u003cGERRIT_ACCOUNT_39086\u003e replied on the change","reason_account":{"_account_id":39086,"name":"Matthias Haag","display_name":"Matthias Haag","email":"matthias.haag@uhurutec.com","username":"mauhau"}}},"hashtags":[],"change_id":"I2d9d04113c9c8dfca37c64f0e35798e9f1d8e6e1","subject":"Allow insecure fernet key repository","status":"NEW","created":"2026-04-23 08:58:19.000000000","updated":"2026-05-01 08:42:58.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"08100868c060764d8d7820eb77e562cbfa4af27a","_number":985931,"virtual_id_number":985931,"owner":{"_account_id":39086,"name":"Matthias Haag","display_name":"Matthias Haag","email":"matthias.haag@uhurutec.com","username":"mauhau"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":13478,"name":"Boris Bobrov","email":"b.bobrov@sap.com","username":"bbobrov"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-04-23 10:57:33.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":13478,"name":"Boris Bobrov","email":"b.bobrov@sap.com","username":"bbobrov"},"all":[{"value":1,"date":"2026-04-29 08:53:38.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":13478,"name":"Boris Bobrov","email":"b.bobrov@sap.com","username":"bbobrov"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":13478,"name":"Boris Bobrov","email":"b.bobrov@sap.com","username":"bbobrov"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13478,"name":"Boris Bobrov","email":"b.bobrov@sap.com","username":"bbobrov"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-04-23 09:45:56.000000000","updated_by":{"_account_id":13478,"name":"Boris Bobrov","email":"b.bobrov@sap.com","username":"bbobrov"},"reviewer":{"_account_id":13478,"name":"Boris Bobrov","email":"b.bobrov@sap.com","username":"bbobrov"},"state":"CC"},{"updated":"2026-04-23 09:52:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2026-04-23 10:57:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-29 08:53:38.000000000","updated_by":{"_account_id":13478,"name":"Boris Bobrov","email":"b.bobrov@sap.com","username":"bbobrov"},"reviewer":{"_account_id":13478,"name":"Boris Bobrov","email":"b.bobrov@sap.com","username":"bbobrov"},"state":"REVIEWER"}],"messages":[{"id":"d598c932ee1a7620de95b81a94d945951e143b32","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":39086,"name":"Matthias Haag","display_name":"Matthias Haag","email":"matthias.haag@uhurutec.com","username":"mauhau"},"date":"2026-04-23 08:58:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e3938d7a1853d9bfff95f18f93a25012c513d09d","author":{"_account_id":13478,"name":"Boris Bobrov","email":"b.bobrov@sap.com","username":"bbobrov"},"date":"2026-04-23 09:45:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f17dfcf19bda23e00ab48d7b738b14a2fa73023b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-23 09:52:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7510f06b8da047ba82b10d49a0678aef\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/dd97c657865c43fa83eb691a08bd3305 : SUCCESS in 18m 15s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/9cc5c19d02704f028884d42d88f8b408 : SUCCESS in 14m 58s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/ce85af8b2f1b4286a0026f7f2556e7e3 : SUCCESS in 24m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"507a94dcd5e8099da9929db2e145fc6a275f8627","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-23 10:57:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5f197e2c554b49dfa5971d52eb1eecfc\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cfb9719cafc648b9abead4af61950e39 : SUCCESS in 18m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51fdfa14dd9f434e9668c1f3c29fbd0e : SUCCESS in 5m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e3e6bf072f48466189b1423143d2ac42 : SUCCESS in 11m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fff5c125335449bb9b1cd1c5094c60a0 : SUCCESS in 10m 43s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/face1b89473e457b901398f0e557cf55 : SUCCESS in 15m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8bd8834959434cfeacda630c6c0a5ff8 : SUCCESS in 13m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/6ea9705fef194dab809f3f8dcb1908c2 : SUCCESS in 1h 02m 50s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/69a006ef21b04347b04a325bf3e077a8 : SUCCESS in 1h 53m 46s\n- keystoneclient-devstack-functional https://zuul.opendev.org/t/openstack/build/5a03a474112f4cb8aa8c37d4924b5092 : SUCCESS in 17m 46s (non-voting)\n- keystone-tempest https://zuul.opendev.org/t/openstack/build/9de3b27f2a6346f5879a83470268af6c : SUCCESS in 32m 11s\n- keystone-tempest-fips https://zuul.opendev.org/t/openstack/build/b7d7f636835840f3b3552f634950a2f5 : FAILURE in 11m 45s (non-voting)\n- keystone-tempest-federation https://zuul.opendev.org/t/openstack/build/2a40ab87c59e41618a4cdaf3009086a5 : FAILURE in 25m 49s (non-voting)\n- keystone-tempest-federation-k2k https://zuul.opendev.org/t/openstack/build/4839e4e6f9af4102aa71fa99bd6082a8 : SUCCESS in 31m 34s\n- keystone-tempest-oidc-federation https://zuul.opendev.org/t/openstack/build/fbfae536f82248c7bbfb1c7319a7a274 : SUCCESS in 33m 16s\n- keystone-tempest-ldap-domain-specific-driver https://zuul.opendev.org/t/openstack/build/6b0d2ad522934f179bc2ec4f965659d3 : SUCCESS in 37m 22s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/78fcae82507641dc9b3882fa3969367c : SUCCESS in 57m 06s\n- keystone-protection-functional https://zuul.opendev.org/t/openstack/build/f5546b68651c49b4842e0379b5230bd4 : FAILURE in 32m 40s (non-voting)\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/1eb9f93cef154f0b860a0baee7b5f4f5 : SUCCESS in 7m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"099f8a37993584dcf2f4ef0a354067384e68bdaf","author":{"_account_id":39086,"name":"Matthias Haag","display_name":"Matthias Haag","email":"matthias.haag@uhurutec.com","username":"mauhau"},"date":"2026-04-23 12:26:49.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"342b3ee69b612791d1d8160ae5c9107f8eacf09f","author":{"_account_id":13478,"name":"Boris Bobrov","email":"b.bobrov@sap.com","username":"bbobrov"},"date":"2026-04-29 08:53:38.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"08100868c060764d8d7820eb77e562cbfa4af27a","author":{"_account_id":39086,"name":"Matthias Haag","display_name":"Matthias Haag","email":"matthias.haag@uhurutec.com","username":"mauhau"},"date":"2026-05-01 08:42:58.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9ac57015cc48eb556c725014409efd3d66c704c3","revisions":{"9ac57015cc48eb556c725014409efd3d66c704c3":{"kind":"REWORK","_number":1,"created":"2026-04-23 08:58:19.000000000","uploader":{"_account_id":39086,"name":"Matthias Haag","display_name":"Matthias Haag","email":"matthias.haag@uhurutec.com","username":"mauhau"},"ref":"refs/changes/31/985931/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/keystone","ref":"refs/changes/31/985931/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/keystone refs/changes/31/985931/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/keystone refs/changes/31/985931/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/keystone refs/changes/31/985931/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/keystone refs/changes/31/985931/1"}}},"commit":{"parents":[{"commit":"b6fd80996b882890a51f3e2aab41d952d7ff68ae","subject":"Enforce app cred project boundary on EC2 credential paths","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/b6fd80996b882890a51f3e2aab41d952d7ff68ae"}]}],"author":{"name":"Matthias Haag","email":"matthias.haag@uhurutec.com","date":"2026-04-23 08:55:06.000000000","tz":120},"committer":{"name":"Matthias Haag","email":"matthias.haag@uhurutec.com","date":"2026-04-23 08:55:06.000000000","tz":120},"subject":"Allow insecure fernet key repository","message":"Allow insecure fernet key repository\n\nWhen Keystone gets deployed in Kubernetes and fernet keys are\nmounted as secret the directory is always world-readable as\nKubernetes is designed to be a non-multi-user system in context\nof a pod. This secure but world-readable directory leads to useless\nwarnings on every Keystone authentication request.\n\nA configurable option to allow insecure key repository is added to\navoid these warnings.\n\nChange-Id: I2d9d04113c9c8dfca37c64f0e35798e9f1d8e6e1\nSigned-off-by: Matthias Haag \u003cmatthias.haag@uhurutec.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/9ac57015cc48eb556c725014409efd3d66c704c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/keystone/commit/9ac57015cc48eb556c725014409efd3d66c704c3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13478,"name":"Boris Bobrov","email":"b.bobrov@sap.com","username":"bbobrov"}},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
