)]}'
{"id":"openstack%2Fkolla-ansible~1001621","triplet_id":"openstack%2Fkolla-ansible~master~Ia2d04e3ecaf861dd28cc15fcdf9ea0d0c930c1ff","project":"openstack/kolla-ansible","branch":"master","topic":"bug/2161170-keystone-oidc-cert-path","hashtags":[],"change_id":"Ia2d04e3ecaf861dd28cc15fcdf9ea0d0c930c1ff","subject":"Fix Keystone OIDC certificate paths","status":"NEW","created":"2026-08-20 07:20:01.000000000","updated":"2026-08-20 10:26:25.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f6ee1466a861e4513ea5a9a937b099794bddc432","_number":1001621,"virtual_id_number":1001621,"owner":{"_account_id":38898,"name":"Kim soo","display_name":"500k","email":"popododo0720@naver.com","username":"popododo0720"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-08-20 10:26:25.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-20 10:26:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a4f63636dfa10e21681a146dfd5a608830c62790","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38898,"name":"Kim soo","display_name":"500k","email":"popododo0720@naver.com","username":"popododo0720"},"date":"2026-08-20 07:20:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f6ee1466a861e4513ea5a9a937b099794bddc432","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-20 10:26:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa87c1a25b394b45803833488f5194b7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/483111e5b4da40e39d118f287a27edf6 : SUCCESS in 5m 44s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/673a2c97115b4827beaccfe177ea0514 : SUCCESS in 4m 56s\n- kolla-ansible-debian-trixie-multi-compute-ironic https://zuul.opendev.org/t/openstack/build/9235c4af5b064432b7cefa3a02098830 : SUCCESS in 1h 09m 55s (non-voting)\n- kolla-ansible-rocky-10-multi-compute-ironic https://zuul.opendev.org/t/openstack/build/04a59c74faca45a082d8d8b9e824f3d0 : FAILURE in 57m 51s (non-voting)\n- kolla-ansible-ubuntu-noble-multi-compute-ironic https://zuul.opendev.org/t/openstack/build/3913ccd90b644a06b5bbde66b5a18980 : SUCCESS in 1h 14m 07s (non-voting)\n- kolla-ansible-debian-trixie https://zuul.opendev.org/t/openstack/build/9d2125e3dda04609a5b968c01a160520 : SUCCESS in 55m 15s (non-voting)\n- kolla-ansible-debian-trixie-podman https://zuul.opendev.org/t/openstack/build/ca0466dc41b24b69ad3aac01aa715e51 : SUCCESS in 53m 21s (non-voting)\n- kolla-ansible-debian-trixie-upgrade https://zuul.opendev.org/t/openstack/build/a75dd657e40d4eed999156e28727992a : SUCCESS in 1h 44m 03s (non-voting)\n- kolla-ansible-rocky-10 https://zuul.opendev.org/t/openstack/build/527d8c0fe71d48a09ac52638a98c9824 : SUCCESS in 56m 05s\n- kolla-ansible-rocky-10-upgrade https://zuul.opendev.org/t/openstack/build/e752bc5bfc774ea2a6df03bc57cb1468 : SUCCESS in 1h 21m 22s\n- kolla-ansible-rocky-10-podman https://zuul.opendev.org/t/openstack/build/0f8a95a22a774284b156d01131262232 : SUCCESS in 57m 18s\n- kolla-ansible-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ed720fe8b11449f28112ebb68d2ffe42 : SUCCESS in 1h 12m 02s\n- kolla-ansible-ubuntu-noble-podman https://zuul.opendev.org/t/openstack/build/9ab46b2350e8408a8d90adba998be4e6 : SUCCESS in 54m 58s\n- kolla-ansible-ubuntu-noble-upgrade https://zuul.opendev.org/t/openstack/build/b7f06fcb58a146a9a75b8b6c9cb293a2 : SUCCESS in 1h 39m 05s\n- kolla-ansible-debian-trixie-bifrost https://zuul.opendev.org/t/openstack/build/28e37fff3d554170b53c1580c3697b5e : SUCCESS in 15m 13s (non-voting)\n- kolla-ansible-rocky-10-bifrost https://zuul.opendev.org/t/openstack/build/62a8fb1b591c46cc9c592e56612cf5f0 : SUCCESS in 13m 42s (non-voting)\n- kolla-ansible-ubuntu-noble-bifrost https://zuul.opendev.org/t/openstack/build/e5773554e0504f8190fd66536a0a94ae : SUCCESS in 15m 35s (non-voting)\n- kolla-ansible-debian-trixie-cells https://zuul.opendev.org/t/openstack/build/7af6f03091c74343a7b14ecff9ffe51c : SUCCESS in 1h 15m 08s (non-voting)\n- kolla-ansible-rocky-10-cells https://zuul.opendev.org/t/openstack/build/a125a5ecc5e645fca9f3c6bf1c742df5 : FAILURE in 1h 02m 29s (non-voting)\n- kolla-ansible-ubuntu-noble-cells https://zuul.opendev.org/t/openstack/build/59e63efb554b46f09f04c6231b9e1c6b : SUCCESS in 1h 20m 40s (non-voting)\n- kolla-ansible-debian-trixie-cephadm https://zuul.opendev.org/t/openstack/build/e8c7015b97424a59946c45591a64dd5c : SUCCESS in 1h 35m 52s (non-voting)\n- kolla-ansible-debian-trixie-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/1eb2d82873c9461ba7a1e6454daa5a90 : SUCCESS in 2h 19m 42s (non-voting)\n- kolla-ansible-rocky-10-cephadm https://zuul.opendev.org/t/openstack/build/e11b36d801e64490ad9fdca03db1946a : TIMED_OUT in 3h 03m 22s (non-voting)\n- kolla-ansible-rocky-10-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/9dc213d2b2134084ad9fd6b1711ed5c4 : FAILURE in 1h 04m 46s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm https://zuul.opendev.org/t/openstack/build/c185cbb9e51a4492a7c02ab817e7ca39 : FAILURE in 47m 20s (non-voting)\n- kolla-ansible-ubuntu-noble-cephadm-upgrade https://zuul.opendev.org/t/openstack/build/c77177a4c2e8435bbb2c368175ae1cf5 : FAILURE in 46m 03s (non-voting)\n- kolla-ansible-debian-trixie-cloudkitty https://zuul.opendev.org/t/openstack/build/e45dfd61460e41028bf02bf36d4bc1e3 : SUCCESS in 43m 12s (non-voting)\n- kolla-ansible-rocky-10-cloudkitty https://zuul.opendev.org/t/openstack/build/5cc431d61d224e49bef081a90c5a372e : SUCCESS in 30m 35s (non-voting)\n- kolla-ansible-ubuntu-noble-cloudkitty https://zuul.opendev.org/t/openstack/build/0b1094aed7294708b0ba5dd68d428311 : SUCCESS in 1h 06m 43s (non-voting)\n- kolla-ansible-debian-trixie-federation https://zuul.opendev.org/t/openstack/build/2776279219e749fc8b940c691148ad2e : SUCCESS in 55m 56s (non-voting)\n- kolla-ansible-rocky-10-federation https://zuul.opendev.org/t/openstack/build/8382e615f7b540e393003ce3e5f492f2 : SUCCESS in 1h 25m 44s (non-voting)\n- kolla-ansible-ubuntu-noble-federation https://zuul.opendev.org/t/openstack/build/1ef07ea4b4254f209432fa5c51813263 : SUCCESS in 1h 18m 34s (non-voting)\n- kolla-ansible-debian-trixie-haproxy-fqdn https://zuul.opendev.org/t/openstack/build/02ada5d7607a4fd3a4175f236f9e1a0f : SUCCESS in 1h 02m 49s (non-voting)\n- kolla-ansible-rocky-10-haproxy-fqdn https://zuul.opendev.org/t/openstack/build/9c0c99c36ec94c6691718ed54367f02d : SUCCESS in 45m 57s (non-voting)\n- kolla-ansible-ubuntu-noble-haproxy-fqdn https://zuul.opendev.org/t/openstack/build/ea6e8d71fca24f28ad4401816d7a33ab : SUCCESS in 53m 36s (non-voting)\n- kayobe-overcloud-rocky10 https://zuul.opendev.org/t/openstack/build/b20b555cb60a49d1bd95fc508c0997ca : SUCCESS in 1h 10m 20s (non-voting)\n- kayobe-overcloud-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c5d949cad87141e59f737c9b847f4baa : FAILURE in 1h 32m 59s (non-voting)\n- kolla-ansible-debian-trixie-openbao https://zuul.opendev.org/t/openstack/build/c204c38e99a64282bf64c3408a2fac6e : SUCCESS in 3m 14s (non-voting)\n- kolla-ansible-rocky-10-openbao https://zuul.opendev.org/t/openstack/build/ec1fd42553ce40ecb1607fae036cef60 : SUCCESS in 4m 39s (non-voting)\n- kolla-ansible-ubuntu-noble-openbao https://zuul.opendev.org/t/openstack/build/bfb830d04d374563bf8252b46529f53e : SUCCESS in 3m 13s (non-voting)\n- kolla-ansible-debian-trixie-kvm https://zuul.opendev.org/t/openstack/build/70b7cbd0ed004f07bb0e26834128f430 : SUCCESS in 50m 47s (non-voting)\n- kolla-ansible-rocky-10-kvm https://zuul.opendev.org/t/openstack/build/3ad77d19008d479b852d8a1fb8887ef5 : FAILURE in 35m 31s (non-voting)\n- kolla-ansible-ubuntu-noble-kvm https://zuul.opendev.org/t/openstack/build/0246403ad6c442c682448082051a0a6a : SUCCESS in 55m 12s (non-voting)\n- kolla-ansible-debian-trixie-lets-encrypt https://zuul.opendev.org/t/openstack/build/295775f2033d42f9955b7a90b5d7dc85 : SUCCESS in 1h 06m 06s (non-voting)\n- kolla-ansible-ubuntu-noble-lets-encrypt https://zuul.opendev.org/t/openstack/build/a0219e66ea0e43249ddd3e3ecfd1df2f : FAILURE in 40m 37s (non-voting)\n- kolla-ansible-debian-trixie-magnum https://zuul.opendev.org/t/openstack/build/464bf4e6b06440588cd87a553572ea3e : FAILURE in 25m 25s (non-voting)\n- kolla-ansible-rocky-10-magnum https://zuul.opendev.org/t/openstack/build/9e5928e7e912474da2cadc517cb563db : FAILURE in 46m 13s (non-voting)\n- kolla-ansible-ubuntu-noble-magnum https://zuul.opendev.org/t/openstack/build/27c13a11f5494f178b3703cdbd5793e3 : FAILURE in 27m 45s (non-voting)\n- kolla-ansible-debian-trixie-mariadb https://zuul.opendev.org/t/openstack/build/0302633c4e9e4544bf8bf1ca0fc3d0bb : SUCCESS in 18m 27s\n- kolla-ansible-debian-trixie-mariadb-upgrade https://zuul.opendev.org/t/openstack/build/d0fcea42d2594351bfff521633e2f4e8 : SUCCESS in 27m 01s\n- kolla-ansible-rocky-10-mariadb https://zuul.opendev.org/t/openstack/build/bb8be6ac18664b0db39f43622ca4699a : SUCCESS in 19m 00s\n- kolla-ansible-rocky-10-mariadb-upgrade https://zuul.opendev.org/t/openstack/build/7c08d1d7b967433dbf696b7a5ec9baf3 : SUCCESS in 26m 33s\n- kolla-ansible-ubuntu-noble-mariadb https://zuul.opendev.org/t/openstack/build/bfe30dd106d748409713ee9dd5a66ca3 : SUCCESS in 25m 36s\n- kolla-ansible-ubuntu-noble-mariadb-upgrade https://zuul.opendev.org/t/openstack/build/c223bab826b54dea8d275b7e14029916 : SUCCESS in 19m 45s\n- kolla-ansible-debian-trixie-masakari https://zuul.opendev.org/t/openstack/build/bafeffa9d55d426190cc53fd9f15c2d9 : FAILURE in 39m 42s (non-voting)\n- kolla-ansible-rocky-10-masakari https://zuul.opendev.org/t/openstack/build/79358878d753459394d958278e9b2385 : SUCCESS in 54m 15s (non-voting)\n- kolla-ansible-ubuntu-noble-masakari https://zuul.opendev.org/t/openstack/build/3344c8ef037c4b8fb8dd95bbc28bb6ef : SUCCESS in 1h 10m 35s (non-voting)\n- kolla-ansible-debian-trixie-nfv https://zuul.opendev.org/t/openstack/build/92af58e30a8e4a4b93699ef285a2f1f6 : POST_FAILURE Host unreachable in 1h 39m 06s (non-voting)\n- kolla-ansible-ubuntu-noble-nfv https://zuul.opendev.org/t/openstack/build/5fbc973b08544c4a9a901dde34b8dcd4 : FAILURE in 1h 38m 31s (non-voting)\n- kolla-ansible-debian-trixie-ironic https://zuul.opendev.org/t/openstack/build/c07371cb95be48c79115d9229342af9a : SUCCESS in 1h 11m 50s (non-voting)\n- kolla-ansible-debian-trixie-ironic-upgrade https://zuul.opendev.org/t/openstack/build/0b39df8727c74ab6ada2b7f443d421d9 : SUCCESS in 1h 41m 42s (non-voting)\n- kolla-ansible-rocky-10-ironic https://zuul.opendev.org/t/openstack/build/71f4e914341740589b59bb8cd8a3baaf : SUCCESS in 1h 08m 00s (non-voting)\n- kolla-ansible-rocky-10-ironic-upgrade https://zuul.opendev.org/t/openstack/build/61564148b1aa43ee94f18084ea04d03a : FAILURE in 51m 14s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic https://zuul.opendev.org/t/openstack/build/340e0d70316a47828bb4699af141f687 : SUCCESS in 1h 14m 44s (non-voting)\n- kolla-ansible-ubuntu-noble-ironic-upgrade https://zuul.opendev.org/t/openstack/build/6c612f5c926c41cf9318b822cc33262b : SUCCESS in 1h 44m 47s (non-voting)\n- kolla-ansible-debian-trixie-octavia https://zuul.opendev.org/t/openstack/build/83b99efcfa654f2bb75b3ff0fae5c58f : SUCCESS in 54m 51s (non-voting)\n- kolla-ansible-ubuntu-noble-octavia https://zuul.opendev.org/t/openstack/build/28e76b85376c458f92d5ca030e4e8b40 : SUCCESS in 59m 14s (non-voting)\n- kolla-ansible-debian-trixie-ovn https://zuul.opendev.org/t/openstack/build/4d7b53d01a2c41e2b7c05374567e0aa9 : SUCCESS in 1h 30m 06s (non-voting)\n- kolla-ansible-debian-trixie-ovn-upgrade https://zuul.opendev.org/t/openstack/build/7d1afff363704ad5aea3f9fe27b64602 : FAILURE in 59m 37s (non-voting)\n- kolla-ansible-rocky-10-ovn https://zuul.opendev.org/t/openstack/build/872ce87a58ae44d887ad3e1238a97b14 : SUCCESS in 1h 35m 20s (non-voting)\n- kolla-ansible-rocky-10-ovn-upgrade https://zuul.opendev.org/t/openstack/build/ead146a46bf849539bfc87a4777e3a5c : SUCCESS in 2h 03m 29s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn https://zuul.opendev.org/t/openstack/build/f24013d01cda4b08bc54ea9e92131438 : SUCCESS in 1h 35m 06s (non-voting)\n- kolla-ansible-ubuntu-noble-ovn-upgrade https://zuul.opendev.org/t/openstack/build/405aa21b0cf94c82a6da2b455532bf8e : SUCCESS in 2h 21m 53s (non-voting)\n- kolla-ansible-debian-trixie-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/4ee1109a27b74ea9820e5294d78cdb8e : SUCCESS in 37m 52s\n- kolla-ansible-debian-trixie-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/d2cfe251ca34443fa1b0c454b8102beb : SUCCESS in 48m 17s\n- kolla-ansible-rocky-10-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/1691c17c969442ddbffa631361bfb3ae : SUCCESS in 29m 08s\n- kolla-ansible-rocky-10-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/3f797b79db2d43b5bc50a63fb15412a1 : SUCCESS in 59m 32s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch https://zuul.opendev.org/t/openstack/build/b750c43dcd6f47b08a812432dfe57fa4 : SUCCESS in 39m 26s\n- kolla-ansible-ubuntu-noble-prometheus-opensearch-upgrade https://zuul.opendev.org/t/openstack/build/171f91cf71d04270aa8b0408514f1e4b : SUCCESS in 40m 57s\n- kolla-ansible-debian-trixie-telemetry https://zuul.opendev.org/t/openstack/build/744768fa74004ac29d06de897280705f : FAILURE in 1h 06m 32s (non-voting)\n- kolla-ansible-rocky-10-telemetry https://zuul.opendev.org/t/openstack/build/1f779707bae3402db2926348f199dde6 : FAILURE in 1h 07m 19s (non-voting)\n- kolla-ansible-ubuntu-noble-telemetry https://zuul.opendev.org/t/openstack/build/95a65b0fe9ab4fc68c147b1e9b63c35c : TIMED_OUT in 2h 32m 46s (non-voting)\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5ddb7eca33c54c6495fe21f7786dba78 : SUCCESS in 3m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd8af32913ad4d2998158934fd077d3d : SUCCESS in 4m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4890d0e221834cbbb76af5ffb0c16d06 : SUCCESS in 3m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/82193fd2de394f659ff4b72b5b9e403b : SUCCESS in 3m 24s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9a68f7394ff74c68b9dc2c72807cc575 : SUCCESS in 5m 43s (non-voting)\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dd0136a36dc64c228238c39affec59a3 : SUCCESS in 18m 13s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"37a6781e852dd307bcc3d5b97df978b443db289a","revisions":{"37a6781e852dd307bcc3d5b97df978b443db289a":{"kind":"REWORK","_number":1,"created":"2026-08-20 07:20:01.000000000","uploader":{"_account_id":38898,"name":"Kim soo","display_name":"500k","email":"popododo0720@naver.com","username":"popododo0720"},"ref":"refs/changes/21/1001621/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/1001621/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/1001621/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/1001621/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/1001621/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/1001621/1"}}},"commit":{"parents":[{"commit":"1a681814eb19a4840ca307a25f21b2907b432272","subject":"Merge \"Extend the nova_scheduler healthcheck timeout.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/1a681814eb19a4840ca307a25f21b2907b432272"}]}],"author":{"name":"popododo0720","email":"popododo0720@naver.com","date":"2026-08-20 07:19:49.000000000","tz":540},"committer":{"name":"popododo0720","email":"popododo0720@naver.com","date":"2026-08-20 07:19:49.000000000","tz":540},"subject":"Fix Keystone OIDC certificate paths","message":"Fix Keystone OIDC certificate paths\n\nUse the correct folded-scalar escaping for OIDC certificate key IDs.\n\nCover certificate files in the federation scenario.\n\nCloses-Bug: #2161170\n\nAssisted-By: OpenAI Codex\nChange-Id: Ia2d04e3ecaf861dd28cc15fcdf9ea0d0c930c1ff\nSigned-off-by: popododo0720 \u003cpopododo0720@naver.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/37a6781e852dd307bcc3d5b97df978b443db289a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/37a6781e852dd307bcc3d5b97df978b443db289a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"UNSATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1"],"atom_explanations":{"label:Backport-Candidate\u003d1":"","label:Backport-Candidate\u003d-1":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
