)]}'
{"id":"openstack%2Fkolla-ansible~659821","triplet_id":"openstack%2Fkolla-ansible~stable%2Frocky~Ia6d3935263a5909c71750b34eb69e72e6e558b7a","project":"openstack/kolla-ansible","branch":"stable/rocky","topic":"bug/1794249-stable/rocky","hashtags":[],"change_id":"Ia6d3935263a5909c71750b34eb69e72e6e558b7a","subject":"Constrain the size of Docker logs","status":"MERGED","created":"2019-05-17 15:18:49.000000000","updated":"2019-05-31 12:15:31.000000000","submitted":"2019-05-31 12:15:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"659821-1559304931755-efcb9f90","meta_rev_id":"49a3900b28ba644a0ab39486105d84505e37d359","_number":659821,"virtual_id_number":659821,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23942,"name":"Matt Kucia","email":"maciej@kucia.net","username":"maciejkucia"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2019-05-31 12:15:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"recommended":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":1,"date":"2019-05-22 09:42:45.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":2,"date":"2019-05-31 10:20:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2019-05-30 19:54:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23942,"name":"Matt Kucia","email":"maciej@kucia.net","username":"maciejkucia"},{"value":2,"date":"2019-05-31 11:09:56.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23942,"name":"Matt Kucia","email":"maciej@kucia.net","username":"maciejkucia"},{"value":1,"date":"2019-05-31 11:09:56.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23942,"name":"Matt Kucia","email":"maciej@kucia.net","username":"maciejkucia"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":23942,"name":"Matt Kucia","email":"maciej@kucia.net","username":"maciejkucia"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2019-05-29 10:56:52.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":23942,"name":"Matt Kucia","email":"maciej@kucia.net","username":"maciejkucia"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-22 09:42:45.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2019-05-30 19:54:56.000000000","updated_by":{"_account_id":23942,"name":"Matt Kucia","email":"maciej@kucia.net","username":"maciejkucia"},"reviewer":{"_account_id":23942,"name":"Matt Kucia","email":"maciej@kucia.net","username":"maciejkucia"},"state":"REVIEWER"},{"updated":"2019-05-31 10:20:21.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2019-05-31 11:09:56.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-05-31 12:15:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ed82cf3185d2b8e0dfcc4df1d8c2c2efd84d0ef2","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-05-17 15:18:49.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"d87c6ac4186be60b452561c515cb6c1ef6faca67","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-17 16:48:24.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/21/659821/1/check/openstack-tox-pep8/2c57d86/ : SUCCESS in 3m 48s\n- openstack-tox-py27 http://logs.openstack.org/21/659821/1/check/openstack-tox-py27/1640ee3/ : SUCCESS in 3m 32s\n- openstack-tox-py35 http://logs.openstack.org/21/659821/1/check/openstack-tox-py35/8defcb6/ : SUCCESS in 3m 11s\n- openstack-tox-py36 http://logs.openstack.org/21/659821/1/check/openstack-tox-py36/5596b38/ : SUCCESS in 2m 45s\n- openstack-tox-docs http://logs.openstack.org/21/659821/1/check/openstack-tox-docs/7dfec25/html/ : SUCCESS in 3m 20s\n- build-openstack-releasenotes http://logs.openstack.org/21/659821/1/check/build-openstack-releasenotes/aec977a/html/ : SUCCESS in 3m 57s\n- kolla-ansible-centos-source http://logs.openstack.org/21/659821/1/check/kolla-ansible-centos-source/41b2f97/ : SUCCESS in 54m 36s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/21/659821/1/check/kolla-ansible-ubuntu-source/12c27ca/ : FAILURE in 13m 42s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/21/659821/1/check/kolla-ansible-oraclelinux-source/0d1c59d/ : SUCCESS in 49m 36s (non-voting)\n- kolla-ansible-ubuntu-source-ceph http://logs.openstack.org/21/659821/1/check/kolla-ansible-ubuntu-source-ceph/d15bc63/ : SUCCESS in 1h 05m 14s (non-voting)\n- kolla-ansible-centos-source-ceph http://logs.openstack.org/21/659821/1/check/kolla-ansible-centos-source-ceph/45c7928/ : SUCCESS in 1h 21m 03s (non-voting)\n- kolla-ansible-oraclelinux-source-ceph http://logs.openstack.org/21/659821/1/check/kolla-ansible-oraclelinux-source-ceph/08be725/ : SUCCESS in 1h 18m 57s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/21/659821/1/check/openstack-tox-lower-constraints/3ab6692/ : SUCCESS in 4m 01s","accounts_in_message":[],"_revision_number":1},{"id":"955f49752cd81b18b805713a03614bf1c35f4d96","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-05-20 09:09:55.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b6b53d93483374b5df340c824b0d327a5bebf045","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 09:25:16.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-pep8 http://logs.openstack.org/21/659821/1/check/openstack-tox-pep8/6027e65/ : SUCCESS in 3m 36s\n- openstack-tox-py27 http://logs.openstack.org/21/659821/1/check/openstack-tox-py27/1dc3d6a/ : SUCCESS in 3m 26s\n- openstack-tox-py35 http://logs.openstack.org/21/659821/1/check/openstack-tox-py35/fe250e7/ : SUCCESS in 3m 37s\n- openstack-tox-py36 http://logs.openstack.org/21/659821/1/check/openstack-tox-py36/ef538cc/ : SUCCESS in 2m 25s\n- openstack-tox-docs http://logs.openstack.org/21/659821/1/check/openstack-tox-docs/01104a5/html/ : SUCCESS in 2m 56s\n- build-openstack-releasenotes http://logs.openstack.org/21/659821/1/check/build-openstack-releasenotes/98aee5b/html/ : SUCCESS in 3m 25s\n- kolla-ansible-centos-source http://logs.openstack.org/21/659821/1/check/kolla-ansible-centos-source/47507fb/ : POST_FAILURE in 3m 13s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/21/659821/1/check/kolla-ansible-ubuntu-source/5a75395/ : POST_FAILURE in 3m 21s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/21/659821/1/check/kolla-ansible-oraclelinux-source/6d4b17a/ : POST_FAILURE in 3m 37s (non-voting)\n- kolla-ansible-ubuntu-source-ceph http://logs.openstack.org/21/659821/1/check/kolla-ansible-ubuntu-source-ceph/552fcdd/ : POST_FAILURE in 5m 13s (non-voting)\n- kolla-ansible-centos-source-ceph http://logs.openstack.org/21/659821/1/check/kolla-ansible-centos-source-ceph/7063a29/ : POST_FAILURE in 4m 40s (non-voting)\n- kolla-ansible-oraclelinux-source-ceph http://logs.openstack.org/21/659821/1/check/kolla-ansible-oraclelinux-source-ceph/e480d04/ : POST_FAILURE in 5m 41s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/21/659821/1/check/openstack-tox-lower-constraints/3e8861e/ : SUCCESS in 3m 36s","accounts_in_message":[],"_revision_number":1},{"id":"dd1cf103df3b69cda271a146c206934676900a1c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-05-20 11:34:17.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c705c74bc5104ffb73f663d09ccc796d926c0c3c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 12:47:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/21/659821/1/check/openstack-tox-pep8/794a55f/ : SUCCESS in 3m 35s\n- openstack-tox-py27 http://logs.openstack.org/21/659821/1/check/openstack-tox-py27/c10a46a/ : SUCCESS in 3m 13s\n- openstack-tox-py35 http://logs.openstack.org/21/659821/1/check/openstack-tox-py35/227035b/ : SUCCESS in 3m 01s\n- openstack-tox-py36 http://logs.openstack.org/21/659821/1/check/openstack-tox-py36/9cd8195/ : SUCCESS in 3m 33s\n- openstack-tox-docs http://logs.openstack.org/21/659821/1/check/openstack-tox-docs/a7f75f4/html/ : SUCCESS in 3m 26s\n- build-openstack-releasenotes http://logs.openstack.org/21/659821/1/check/build-openstack-releasenotes/51989a6/html/ : SUCCESS in 3m 35s\n- kolla-ansible-centos-source http://logs.openstack.org/21/659821/1/check/kolla-ansible-centos-source/4bdfc75/ : SUCCESS in 44m 15s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/21/659821/1/check/kolla-ansible-ubuntu-source/9fcfdeb/ : SUCCESS in 46m 09s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/21/659821/1/check/kolla-ansible-oraclelinux-source/17505f4/ : SUCCESS in 50m 52s (non-voting)\n- kolla-ansible-ubuntu-source-ceph http://logs.openstack.org/21/659821/1/check/kolla-ansible-ubuntu-source-ceph/253aabf/ : FAILURE in 20m 47s (non-voting)\n- kolla-ansible-centos-source-ceph http://logs.openstack.org/21/659821/1/check/kolla-ansible-centos-source-ceph/b32b7f2/ : FAILURE in 1h 05m 03s (non-voting)\n- kolla-ansible-oraclelinux-source-ceph http://logs.openstack.org/21/659821/1/check/kolla-ansible-oraclelinux-source-ceph/7106d89/ : FAILURE in 44m 56s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/21/659821/1/check/openstack-tox-lower-constraints/ca43916/ : SUCCESS in 3m 17s","accounts_in_message":[],"_revision_number":1},{"id":"1f7b38b9d69bca17ddeef7af1a0222e5a7c038cd","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-05-20 13:00:56.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"efbe6400b9a8447dfb03f7891b0fb43349ff6559","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 14:34:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/21/659821/1/check/openstack-tox-pep8/17ea6e8/ : SUCCESS in 3m 23s\n- openstack-tox-py27 http://logs.openstack.org/21/659821/1/check/openstack-tox-py27/377bac4/ : SUCCESS in 3m 38s\n- openstack-tox-py35 http://logs.openstack.org/21/659821/1/check/openstack-tox-py35/b81f742/ : SUCCESS in 3m 24s\n- openstack-tox-py36 http://logs.openstack.org/21/659821/1/check/openstack-tox-py36/d09f910/ : SUCCESS in 3m 12s\n- openstack-tox-docs http://logs.openstack.org/21/659821/1/check/openstack-tox-docs/cbf2431/html/ : SUCCESS in 2m 47s\n- build-openstack-releasenotes http://logs.openstack.org/21/659821/1/check/build-openstack-releasenotes/da6cae0/html/ : SUCCESS in 2m 57s\n- kolla-ansible-centos-source http://logs.openstack.org/21/659821/1/check/kolla-ansible-centos-source/0c63310/ : SUCCESS in 56m 46s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/21/659821/1/check/kolla-ansible-ubuntu-source/22175ae/ : SUCCESS in 36m 41s\n- kolla-ansible-oraclelinux-source http://logs.openstack.org/21/659821/1/check/kolla-ansible-oraclelinux-source/8f44c0c/ : SUCCESS in 53m 49s (non-voting)\n- kolla-ansible-ubuntu-source-ceph http://logs.openstack.org/21/659821/1/check/kolla-ansible-ubuntu-source-ceph/c5c6bee/ : SUCCESS in 1h 09m 13s (non-voting)\n- kolla-ansible-centos-source-ceph http://logs.openstack.org/21/659821/1/check/kolla-ansible-centos-source-ceph/7faca1e/ : SUCCESS in 1h 17m 41s (non-voting)\n- kolla-ansible-oraclelinux-source-ceph http://logs.openstack.org/21/659821/1/check/kolla-ansible-oraclelinux-source-ceph/07ad80e/ : SUCCESS in 1h 25m 36s (non-voting)\n- openstack-tox-lower-constraints http://logs.openstack.org/21/659821/1/check/openstack-tox-lower-constraints/27dab7c/ : SUCCESS in 3m 18s","accounts_in_message":[],"_revision_number":1},{"id":"f5c8bc96d2c200187aebb8b9abd46ecbc3c700c5","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2019-05-22 09:42:45.000000000","message":"Patch Set 1: Code-Review+1\n\nThanks for back-porting.","accounts_in_message":[],"_revision_number":1},{"id":"ca2a066a93ab43729399684cf405a7ad4067e126","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-05-29 10:56:52.000000000","message":"Patch Set 1: Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"4fd0c943125dcc46ad0bb4d8ad6285436f92a174","author":{"_account_id":23942,"name":"Matt Kucia","email":"maciej@kucia.net","username":"maciejkucia"},"date":"2019-05-30 19:54:56.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"7b1214994f5fe83fa34b257234aec7aeafc04a16","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2019-05-31 10:20:21.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ced65d93fbeb5b84b0feed01d00a9073831dadca","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-05-31 11:09:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6115289dc54ff33680bfae8b73a90b2c9bd2e366","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-31 11:10:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ec99722f58e3abef675b91e5cb910698715b6352","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-31 12:15:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 http://logs.openstack.org/21/659821/1/gate/openstack-tox-pep8/1334555/ : SUCCESS in 3m 54s\n- openstack-tox-py27 http://logs.openstack.org/21/659821/1/gate/openstack-tox-py27/183995b/ : SUCCESS in 4m 04s\n- openstack-tox-py35 http://logs.openstack.org/21/659821/1/gate/openstack-tox-py35/3157ff5/ : SUCCESS in 2m 48s\n- openstack-tox-py36 http://logs.openstack.org/21/659821/1/gate/openstack-tox-py36/3f47e60/ : SUCCESS in 3m 08s\n- openstack-tox-docs http://logs.openstack.org/21/659821/1/gate/openstack-tox-docs/55d4077/html/ : SUCCESS in 2m 53s\n- build-openstack-releasenotes http://logs.openstack.org/21/659821/1/gate/build-openstack-releasenotes/3e8eba5/html/ : SUCCESS in 3m 24s\n- kolla-ansible-centos-source http://logs.openstack.org/21/659821/1/gate/kolla-ansible-centos-source/5c788bf/ : SUCCESS in 46m 57s\n- kolla-ansible-ubuntu-source http://logs.openstack.org/21/659821/1/gate/kolla-ansible-ubuntu-source/a8ff9ef/ : SUCCESS in 39m 17s\n- openstack-tox-lower-constraints http://logs.openstack.org/21/659821/1/gate/openstack-tox-lower-constraints/2ded3b2/ : SUCCESS in 3m 13s","accounts_in_message":[],"_revision_number":1},{"id":"a7759513faa632c8c1ef615079a048ca237611d0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-31 12:15:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"354aa1d19b2b8fbeaeda35bed15626bab4acfa8b","revisions":{"354aa1d19b2b8fbeaeda35bed15626bab4acfa8b":{"kind":"REWORK","_number":1,"created":"2019-05-17 15:18:49.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/21/659821/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/659821/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/659821/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/659821/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/659821/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/659821/1"}}},"commit":{"parents":[{"commit":"ce533592e41950bf6207e8b854e7032983d6f3f8","subject":"Fix kolla-docker possible undefined variable","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/ce533592e41950bf6207e8b854e7032983d6f3f8"}]}],"author":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2018-10-11 09:12:41.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-05-17 15:18:49.000000000","tz":0},"subject":"Constrain the size of Docker logs","message":"Constrain the size of Docker logs\n\nEven though Kolla services are configured to log output to file rather than\nstdout, some stdout still occurs when for example the container re(starts).\nSince the Docker logs are not constrained in size, they can fill up the\ndocker volumes drive and bring down the host. One example of when this is\nparticularly problematic is when Fluentd cannot parse a log message. The\nwarning output is written to the Docker log and in production we have seen\nit eat 100GB of disk space in less than a day. We could configure Fluentd\nnot to do this, but the problem may still occur via another mechanism.\n\nChange-Id: Ia6d3935263a5909c71750b34eb69e72e6e558b7a\nCloses-Bug: #1794249\n(cherry picked from commit bd54b99132d019b747dcd98af9371628625c8a1b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/354aa1d19b2b8fbeaeda35bed15626bab4acfa8b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/354aa1d19b2b8fbeaeda35bed15626bab4acfa8b"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
