)]}'
{"id":"openstack%2Fkolla-ansible~675421","triplet_id":"openstack%2Fkolla-ansible~master~Icbebc415d643a237fa128840d5f5a9c91d22c12d","project":"openstack/kolla-ansible","branch":"master","topic":"telegraf-docker","hashtags":[],"change_id":"Icbebc415d643a237fa128840d5f5a9c91d22c12d","subject":"Configure Telegraf to monitor Docker containers","status":"MERGED","created":"2019-08-08 17:21:23.000000000","updated":"2019-08-14 14:00:17.000000000","submitted":"2019-08-14 14:00:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"675421-1565791217935-8c476a95","meta_rev_id":"47bf3cabf5238b5cf79b8b2188031c108277b20b","_number":675421,"virtual_id_number":675421,"owner":{"_account_id":30272,"name":"Keith Plant","email":"kplantjr@gmail.com","username":"kplant"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30272,"name":"Keith Plant","email":"kplantjr@gmail.com","username":"kplant"},{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":0,"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":2,"date":"2019-08-14 14:00:17.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},"all":[{"value":0,"_account_id":30272,"name":"Keith Plant","email":"kplantjr@gmail.com","username":"kplant"},{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":0,"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":2,"date":"2019-08-14 11:36:14.000000000","_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"value":1,"date":"2019-08-14 12:34:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2019-08-14 12:47:34.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-08-14 11:41:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":30272,"name":"Keith Plant","email":"kplantjr@gmail.com","username":"kplant"},{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":0,"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2019-08-14 12:47:34.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":30272,"name":"Keith Plant","email":"kplantjr@gmail.com","username":"kplant"},{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":0,"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":30272,"name":"Keith Plant","email":"kplantjr@gmail.com","username":"kplant"},{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":0,"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"value":0,"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},{"_account_id":30272,"name":"Keith Plant","email":"kplantjr@gmail.com","username":"kplant"},{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-08-08 17:21:35.000000000","updated_by":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"reviewer":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"state":"REVIEWER"},{"updated":"2019-08-11 13:52:52.000000000","updated_by":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"reviewer":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"state":"REVIEWER"},{"updated":"2019-08-12 07:51:58.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2019-08-14 11:36:14.000000000","updated_by":{"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},"reviewer":{"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},"state":"REVIEWER"},{"updated":"2019-08-14 11:41:44.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2019-08-14 12:34:23.000000000","updated_by":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"reviewer":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"state":"REVIEWER"},{"updated":"2019-08-14 12:47:34.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-08-14 14:00:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"07b707383ae4b75152056e45fd9ec9f5531f2b31","author":{"_account_id":30272,"name":"Keith Plant","email":"kplantjr@gmail.com","username":"kplant"},"date":"2019-08-08 17:21:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f93e5de4e2843d2dbd53fc76a82259d588303ef4","author":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"date":"2019-08-08 17:21:35.000000000","message":"Patch Set 1:\n\nThank you for your first contribution to OpenStack.\n\nYour patch will now be tested automatically by OpenStack testing frameworks\nand once the automatic tests pass, it will be reviewed by other friendly\ndevelopers. They will give you feedback and may require you to refine it.\n\nPeople seldom get their patch approved on the first try, so don\u0027t be\nconcerned if requested to make corrections. Feel free to modify your patch\nand resubmit a new change-set.\n\nPatches usually take 3 to 7 days to be reviewed so be patient and be\navailable on IRC to ask and answer questions about your work. Also it\ntakes generally at least a couple of weeks for cores to get around to\nreviewing code. The more you participate in the community the more\nrewarding it is for you. You may also notice that the more you get to know\npeople and get to be known, the faster your patches will be reviewed and\neventually approved. Get to know others and become known by doing code\nreviews: anybody can do it, and it\u0027s a great way to learn the code base.\n\nThanks again for supporting OpenStack, we look forward to working with you.\n\nIRC: https://wiki.openstack.org/wiki/IRC\nWorkflow: https://docs.openstack.org/infra/manual/developers.html\nCommit Messages: https://wiki.openstack.org/wiki/GitCommitMessages","accounts_in_message":[],"_revision_number":1},{"id":"1f8eab7c4bf432a6e88ba2e99c7175ce69af429e","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-08-08 18:44:13.000000000","message":"Patch Set 1: Code-Review-1\n\nWhile probably good practice, I don\u0027t think we should force it on people?","accounts_in_message":[],"_revision_number":1},{"id":"1cbb6c843c41a5082ce55286d0fd459dc34ec612","author":{"_account_id":30272,"name":"Keith Plant","email":"kplantjr@gmail.com","username":"kplant"},"date":"2019-08-08 18:52:53.000000000","message":"Patch Set 1:\n\nWould this be better as a toggleable flag in globals.yml? Maybe wrap {% if enable_telegraf_docker_input | bool } around it?","accounts_in_message":[],"_revision_number":1},{"id":"bce6ffabe66db6ccd7a286d55aae362336d8c39b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-08 19:01:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://logs.opendev.org/21/675421/1/check/openstack-tox-cover/2d1ad18/cover/ : SUCCESS in 4m 00s\n- openstack-tox-lower-constraints https://logs.opendev.org/21/675421/1/check/openstack-tox-lower-constraints/41434fd/ : SUCCESS in 4m 20s\n- openstack-tox-pep8 https://logs.opendev.org/21/675421/1/check/openstack-tox-pep8/c82b9e3/ : SUCCESS in 3m 51s\n- openstack-tox-py27 https://logs.opendev.org/21/675421/1/check/openstack-tox-py27/c8f684d/ : SUCCESS in 3m 44s\n- openstack-tox-py36 https://logs.opendev.org/21/675421/1/check/openstack-tox-py36/45ed502/ : SUCCESS in 5m 10s\n- openstack-tox-py37 https://logs.opendev.org/21/675421/1/check/openstack-tox-py37/e76b8a5/ : SUCCESS in 5m 22s\n- openstack-tox-docs https://logs.opendev.org/21/675421/1/check/openstack-tox-docs/a9a8f99/html/ : SUCCESS in 3m 00s\n- kolla-ansible-centos-source https://logs.opendev.org/21/675421/1/check/kolla-ansible-centos-source/20029d3/ : SUCCESS in 53m 14s\n- kolla-ansible-ubuntu-source https://logs.opendev.org/21/675421/1/check/kolla-ansible-ubuntu-source/ca5249c/ : SUCCESS in 39m 03s\n- kolla-ansible-ubuntu-source-ceph https://logs.opendev.org/21/675421/1/check/kolla-ansible-ubuntu-source-ceph/90be6f3/ : SUCCESS in 59m 48s (non-voting)\n- kolla-ansible-centos-source-ceph https://logs.opendev.org/21/675421/1/check/kolla-ansible-centos-source-ceph/3f677fd/ : SUCCESS in 1h 10m 09s (non-voting)\n- kolla-ansible-centos-source-upgrade https://logs.opendev.org/21/675421/1/check/kolla-ansible-centos-source-upgrade/a67c97d/ : SUCCESS in 1h 00m 58s\n- kolla-ansible-ubuntu-source-upgrade https://logs.opendev.org/21/675421/1/check/kolla-ansible-ubuntu-source-upgrade/e49dc03/ : SUCCESS in 44m 38s\n- kolla-ansible-centos-source-upgrade-ceph https://logs.opendev.org/21/675421/1/check/kolla-ansible-centos-source-upgrade-ceph/8fced67/ : SUCCESS in 1h 36m 17s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://logs.opendev.org/21/675421/1/check/kolla-ansible-ubuntu-source-upgrade-ceph/1b75147/ : SUCCESS in 1h 11m 50s (non-voting)\n- kolla-ansible-centos-binary https://logs.opendev.org/21/675421/1/check/kolla-ansible-centos-binary/2754e5c/ : SUCCESS in 1h 00m 55s (non-voting)\n- kolla-ansible-ubuntu-binary https://logs.opendev.org/21/675421/1/check/kolla-ansible-ubuntu-binary/664bcc7/ : SUCCESS in 47m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d60f02dee4595c3faf2f6f0b34a12b7578d0786b","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-08-09 16:43:04.000000000","message":"Patch Set 1:\n\n\u003e Would this be better as a toggleable flag in globals.yml? Maybe\n \u003e wrap {% if enable_telegraf_docker_input | bool } around it?\n\ntelegraf_enable_docker_input looks better as it is telegraf-related, goes to role defaults and globals for telegraf","accounts_in_message":[],"_revision_number":1},{"id":"168a23c5e9904b9cf57d10b7c32b38944b48aca3","author":{"_account_id":30272,"name":"Keith Plant","email":"kplantjr@gmail.com","username":"kplant"},"date":"2019-08-09 18:31:37.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8cd92514108a5330b9867440bf4f06ce1b6b0df7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-09 22:31:39.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://logs.opendev.org/21/675421/2/check/openstack-tox-cover/4bdfc35/cover/ : SUCCESS in 4m 37s\n- openstack-tox-lower-constraints https://logs.opendev.org/21/675421/2/check/openstack-tox-lower-constraints/f383ec1/ : SUCCESS in 4m 00s\n- openstack-tox-pep8 https://logs.opendev.org/21/675421/2/check/openstack-tox-pep8/efc6716/ : SUCCESS in 4m 10s\n- openstack-tox-py27 https://logs.opendev.org/21/675421/2/check/openstack-tox-py27/94e851a/ : SUCCESS in 3m 20s\n- openstack-tox-py36 https://logs.opendev.org/21/675421/2/check/openstack-tox-py36/49cf781/ : SUCCESS in 3m 31s\n- openstack-tox-py37 https://logs.opendev.org/21/675421/2/check/openstack-tox-py37/e94be9d/ : SUCCESS in 3m 06s\n- openstack-tox-docs https://logs.opendev.org/21/675421/2/check/openstack-tox-docs/5d938ad/html/ : SUCCESS in 3m 27s\n- kolla-ansible-centos-source https://logs.opendev.org/21/675421/2/check/kolla-ansible-centos-source/be9d950/ : SUCCESS in 48m 09s\n- kolla-ansible-ubuntu-source https://logs.opendev.org/21/675421/2/check/kolla-ansible-ubuntu-source/5dd1b0c/ : SUCCESS in 40m 48s\n- kolla-ansible-ubuntu-source-ceph https://logs.opendev.org/21/675421/2/check/kolla-ansible-ubuntu-source-ceph/74aea9f/ : SUCCESS in 59m 11s (non-voting)\n- kolla-ansible-centos-source-ceph https://logs.opendev.org/21/675421/2/check/kolla-ansible-centos-source-ceph/6bd6de9/ : SUCCESS in 1h 13m 55s (non-voting)\n- kolla-ansible-centos-source-upgrade https://logs.opendev.org/21/675421/2/check/kolla-ansible-centos-source-upgrade/9389afd/ : SUCCESS in 1h 02m 38s\n- kolla-ansible-ubuntu-source-upgrade https://logs.opendev.org/21/675421/2/check/kolla-ansible-ubuntu-source-upgrade/29f1e58/ : SUCCESS in 53m 53s\n- kolla-ansible-centos-source-upgrade-ceph https://logs.opendev.org/21/675421/2/check/kolla-ansible-centos-source-upgrade-ceph/f4aaf7c/ : SUCCESS in 1h 30m 21s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://logs.opendev.org/21/675421/2/check/kolla-ansible-ubuntu-source-upgrade-ceph/876f846/ : SUCCESS in 1h 13m 43s (non-voting)\n- kolla-ansible-centos-binary https://logs.opendev.org/21/675421/2/check/kolla-ansible-centos-binary/f86df62/ : SUCCESS in 59m 45s (non-voting)\n- kolla-ansible-ubuntu-binary https://logs.opendev.org/21/675421/2/check/kolla-ansible-ubuntu-binary/54950b7/ : SUCCESS in 43m 42s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"552d51869469a256ffd8dcff41c427c6f5784bf5","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-08-11 07:22:52.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"1ce3075fc775ed5db4176526aa05eb7afd30d2f0","author":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"date":"2019-08-11 13:52:52.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"ab13999d074023dad9328b6ddad173bbd1720a6d","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2019-08-12 07:51:58.000000000","message":"Patch Set 2:\n\n(venv3) 06:24 (220s) hrw@j13-qrep-04:kolla$ ls -l /var/run/docker*\n-rw-r--r-- 1 root root     4 Aug  8 17:07 /var/run/docker.pid\nsrw-rw---- 1 root docker   0 Aug  8 17:07 /var/run/docker.sock\n\nDoes default user (used by telegraf) in docker group to be able to access docker.sock?","accounts_in_message":[],"_revision_number":2},{"id":"3e36ced565700374b62adb4f6e340c8447109567","author":{"_account_id":30272,"name":"Keith Plant","email":"kplantjr@gmail.com","username":"kplant"},"date":"2019-08-12 11:25:20.000000000","message":"Patch Set 2:\n\n\u003e (venv3) 06:24 (220s) hrw@j13-qrep-04:kolla$ ls -l /var/run/docker*\n \u003e -rw-r--r-- 1 root root     4 Aug  8 17:07 /var/run/docker.pid\n \u003e srw-rw---- 1 root docker   0 Aug  8 17:07 /var/run/docker.sock\n \u003e \n \u003e Does default user (used by telegraf) in docker group to be able to\n \u003e access docker.sock?\n\nHey Marcin, this is what we were discussing in IRC last week. I think I\u0027m going to need to submit a separate patch to Kolla to create the docker group with a matching GID and add the telegraf user to it.\n\nI wanted to submit this patch ahead of time but leave it defaulted off just to get the framework in place in kolla-ansible.","accounts_in_message":[],"_revision_number":2},{"id":"19bfe7a078ad607c2ba4bd51eb8880bbac40f651","author":{"_account_id":30272,"name":"Keith Plant","email":"kplantjr@gmail.com","username":"kplant"},"date":"2019-08-13 12:17:45.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"418fb5283c7563d150ad73158fba7e5237db18e1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-13 14:04:46.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2706c1e2ec7941e2a3935c33f4ae699d : SUCCESS in 3m 46s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/72e3e4ad11684f24b9df0fc36af8b780 : SUCCESS in 3m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/675a9f1f1c1b45adac8888a5c3d8beca : SUCCESS in 3m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/dc7043a0425545d89e5ae013b497db82 : SUCCESS in 4m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/720ef3edbd924ceba8b348fd102cbb0f : SUCCESS in 3m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/89ebeb02c401452b88fde6c25931b0b5 : SUCCESS in 3m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41701b18f2dd4cd69c7b6a968ddcf93d : SUCCESS in 2m 53s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/485f85718adf4f588ed90ec00d5e045f : SUCCESS in 48m 12s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/522cf627b14d41e3a6d2c22dfb9b67ee : SUCCESS in 39m 42s\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/8b10a815bd8f466faea7ad2b969a6cb8 : SUCCESS in 57m 31s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/34923c160d4a4ae8a2ebc51ec8fb6a94 : SUCCESS in 1h 12m 52s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/8afd37404cf8485d96c82430f58b89ea : SUCCESS in 1h 00m 31s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/8ede462e44d9453f812823f471908ae3 : SUCCESS in 46m 55s\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/a79dca4513a143e584eda7df3ba6736e : SUCCESS in 1h 31m 49s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/f2a66de882cf4c1394ca03f73860de90 : SUCCESS in 1h 13m 52s (non-voting)\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/9e565be1c1874232870cd9815d710be1 : SUCCESS in 58m 02s (non-voting)\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/37f8379b98b0480bbed4c1455b7ba758 : SUCCESS in 38m 26s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"40b06859b35e2682de9baccddbb525462a6bf3e0","author":{"_account_id":23717,"name":"Surya Prakash Singh","email":"singh.surya64mnnit@gmail.com","username":"SuryaPrakash"},"date":"2019-08-14 11:36:14.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"e7e96ef0399713deb15b8370960cd458d30d8300","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2019-08-14 11:41:44.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"5d25ae063f31bec0200fddd290d918d2639ff3c0","author":{"_account_id":28935,"name":"98k","email":"18552437190@163.com","username":"98k"},"date":"2019-08-14 12:34:23.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"11f12caf4bb31aa7ad41db62c8f31a90ac6afb74","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-08-14 12:47:34.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"d4557b874abd8c92d5ca0d941ec48568f36ded82","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-14 12:47:47.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"6c208db899e24f71bfb4b7a40b8e6928bc64cd82","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-14 14:00:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"2a8182f6d9e3a215f6ca076fd63bbe8cd3664d2b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-08-14 14:00:17.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4c31362a1bc64302b97e3365045565a4 : SUCCESS in 3m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0eea9c4cf7443ab927e1e83d36976ef : SUCCESS in 3m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3d2e4a1393814724b52bc7ea48daead2 : SUCCESS in 2m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e8ddccdd28ef4811a9add3d81109b917 : SUCCESS in 3m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e8e4e8ec43634f64b3e535c10cd9d9d2 : SUCCESS in 3m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/268573d609bf4a418a0dc20bc27833e2 : SUCCESS in 2m 51s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/241cd8d198824d6fb833a8f8335ba8e1 : SUCCESS in 47m 27s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/282597bd03324a1faba89661dde0ba00 : SUCCESS in 39m 13s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/98ffc9cf1879415ea1c6863e5062cedf : SUCCESS in 1h 02m 09s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/033d2fae7bfe4d4f9dae31b22b29f54b : SUCCESS in 44m 50s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"b95ff2d1dba3f857381675f38be297250e6433f7","revisions":{"6edd6a645f9b85f76a6a66aa1cee909ec24db1ac":{"kind":"REWORK","_number":1,"created":"2019-08-08 17:21:23.000000000","uploader":{"_account_id":30272,"name":"Keith Plant","email":"kplantjr@gmail.com","username":"kplant"},"ref":"refs/changes/21/675421/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/675421/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/675421/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/675421/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/675421/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/675421/1"}}},"commit":{"parents":[{"commit":"ee5e99fcf5110c6bf96aa9939732ff0d9cbcb337","subject":"Merge \"Stop using MountFlags\u003dshared in Docker configuration\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/ee5e99fcf5110c6bf96aa9939732ff0d9cbcb337"}]}],"author":{"name":"Keith Plant","email":"kplantjr@gmail.com","date":"2019-08-08 17:20:20.000000000","tz":-240},"committer":{"name":"Keith Plant","email":"kplantjr@gmail.com","date":"2019-08-08 17:20:20.000000000","tz":-240},"subject":"Configure Telegraf to monitor Docker containers","message":"Configure Telegraf to monitor Docker containers\n\nAdded configuration to ansible/roles/telegraf/templates/telegraf.conf.j2 to\nallow telegraf to grab telemetry data from docker directly.\n\nChange-Id: Icbebc415d643a237fa128840d5f5a9c91d22c12d\nSigned-off-by: Keith Plant \u003ckplantjr@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/6edd6a645f9b85f76a6a66aa1cee909ec24db1ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/6edd6a645f9b85f76a6a66aa1cee909ec24db1ac"}]},"branch":"refs/heads/master"},"208fbea83144a19315a6a69350a62703181758ff":{"kind":"REWORK","_number":2,"created":"2019-08-09 18:31:37.000000000","uploader":{"_account_id":30272,"name":"Keith Plant","email":"kplantjr@gmail.com","username":"kplant"},"ref":"refs/changes/21/675421/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/675421/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/675421/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/675421/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/675421/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/675421/2"}}},"commit":{"parents":[{"commit":"4468250b951ed4a95942ec58be361fd74c809a99","subject":"Merge \"Remove support for Docker legacy packages\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/4468250b951ed4a95942ec58be361fd74c809a99"}]}],"author":{"name":"Keith Plant","email":"kplantjr@gmail.com","date":"2019-08-08 17:20:20.000000000","tz":-240},"committer":{"name":"Keith Plant","email":"kplantjr@gmail.com","date":"2019-08-09 18:30:27.000000000","tz":-240},"subject":"Configure Telegraf to monitor Docker containers","message":"Configure Telegraf to monitor Docker containers\n\nAdded configuration to ansible/roles/telegraf/templates/telegraf.conf.j2 to\nallow telegraf to grab telemetry data from docker directly.\n\nAdded option to etc/kolla/globals.yml to switch on/off the configuration to\ningest data from the docker daemon into telegraf.\n\nChange-Id: Icbebc415d643a237fa128840d5f5a9c91d22c12d\nSigned-off-by: Keith Plant \u003ckplantjr@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/208fbea83144a19315a6a69350a62703181758ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/208fbea83144a19315a6a69350a62703181758ff"}]},"branch":"refs/heads/master"},"b95ff2d1dba3f857381675f38be297250e6433f7":{"kind":"REWORK","_number":3,"created":"2019-08-13 12:17:45.000000000","uploader":{"_account_id":30272,"name":"Keith Plant","email":"kplantjr@gmail.com","username":"kplant"},"ref":"refs/changes/21/675421/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/675421/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/675421/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/675421/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/675421/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/675421/3"}}},"commit":{"parents":[{"commit":"5c70e0a61521c714120d06b4ee36086e572fec29","subject":"Merge \"Set \u0027distro_python_version\u0027 variable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/5c70e0a61521c714120d06b4ee36086e572fec29"}]}],"author":{"name":"Keith Plant","email":"kplantjr@gmail.com","date":"2019-08-08 17:20:20.000000000","tz":-240},"committer":{"name":"Keith Plant","email":"kplantjr@gmail.com","date":"2019-08-13 12:17:00.000000000","tz":-240},"subject":"Configure Telegraf to monitor Docker containers","message":"Configure Telegraf to monitor Docker containers\n\nAdded configuration to ansible/roles/telegraf/templates/telegraf.conf.j2 to\nallow telegraf to grab telemetry data from docker directly.\n\nAdded option to etc/kolla/globals.yml to switch on/off the configuration to\ningest data from the docker daemon into telegraf.\n\nChange-Id: Icbebc415d643a237fa128840d5f5a9c91d22c12d\nSigned-off-by: Keith Plant \u003ckplantjr@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/b95ff2d1dba3f857381675f38be297250e6433f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/b95ff2d1dba3f857381675f38be297250e6433f7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
