)]}'
{"id":"openstack%2Fkolla-ansible~685721","triplet_id":"openstack%2Fkolla-ansible~master~I4d96f921e025bdea783c8e88682683b25d844ce1","project":"openstack/kolla-ansible","branch":"master","hashtags":[],"change_id":"I4d96f921e025bdea783c8e88682683b25d844ce1","subject":"Use virtualenv for vagrant bootstrap.sh script","status":"NEW","created":"2019-09-30 14:20:40.000000000","updated":"2020-09-21 13:36:33.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":11,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6c5a670f1f5512b8678a47c85fb512f048b768e8","_number":685721,"virtual_id_number":685721,"owner":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"date":"2019-10-02 08:51:52.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:check","value":1,"date":"2019-10-01 19:18:18.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2020-09-21 13:36:33.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":-1,"date":"2019-10-01 16:54:34.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-01 19:18:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-02 08:51:52.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-09-21 13:36:33.000000000","updated_by":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"reviewer":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"state":"REVIEWER"}],"messages":[{"id":"cb268d3ddfbb6fae7009d6344a5e15d9c7fd2128","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2019-09-30 14:20:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4f5b08d53e37578f11e3dd725fe242abac8f28b4","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2019-09-30 14:27:59.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"668b22b8c767dd562b3b0daf0beab8c1dd941e8b","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2019-09-30 14:35:13.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"274e17f1557575c244d6fa475a8b09d7845dab70","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2019-09-30 15:39:59.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a76047971e9eac0b7112ee4e0c94ba1e3562ae68","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-30 17:44:44.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cadb1253ae5d4cb093632013de532ee7 : SUCCESS in 3m 10s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e23af7d114bb45d182fb5efd34556c2d : SUCCESS in 3m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fcf78cba81e44cd791afe8576c05d202 : SUCCESS in 4m 11s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5ded51546b6f4024bc904579c98c794a : SUCCESS in 2m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/66e6bf6cf5cb4fec8d7a480970eeaf26 : SUCCESS in 3m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ddbff7500f3d4c038be15124d102cef8 : SUCCESS in 3m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d3b527acd99443fb3d78fe9658c3072 : SUCCESS in 6m 13s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/130a663b342146f195046ff075c4ecf3 : SUCCESS in 56m 22s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/564347f1857f4dfc96fc1bc34170c743 : SUCCESS in 44m 07s\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/756ad11e049548f885c4da53fc9e29b8 : SUCCESS in 1h 09m 39s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/0f6b34152e034dec84e23eb722ad68e5 : FAILURE in 1h 00m 54s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/f95b404ed8204b08a0568ab7913b8a78 : SUCCESS in 1h 08m 18s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/360e47ff37884faa9f5f8ba5b0f3a1e9 : SUCCESS in 53m 09s\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/47a0bab7b2ee4408988562d54999425e : SUCCESS in 2h 01m 26s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/9bcfb51269c84fbaac5b378b3a9b886f : FAILURE in 1h 19m 31s (non-voting)\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/6cdd9a14f8fe48cc891a1fb83debf8c9 : SUCCESS in 54m 24s (non-voting)\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/efa33b3bc3b240dcb0b5721b2ca9f475 : SUCCESS in 39m 42s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b93154122477e7b030efe6450385ba5465b80d49","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-01 08:31:46.000000000","message":"Patch Set 3:\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"9c7761a240aa3a21254957aebf85f2c9384363e3","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2019-10-01 09:08:55.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"fd9667dc50ace62393d6a45a3e33498ce01ff732","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-01 09:09:57.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"d2b523e3f5c8577b769c7160a50d2bf462c80887","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2019-10-01 10:12:10.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"e8bcbe6b3b99ced9760823ab0da29ecc400aac2f","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2019-10-01 10:13:13.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"dbfe9cb1f56f1eb926ddd4e370263748dba0f2d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 12:13:02.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ad90aa84639247ca84f1d1fcb34738d7 : SUCCESS in 4m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/59483f1f169e41da964bf94183ce4db6 : SUCCESS in 3m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2fc05f3e2ab7479ea05bb575c2ae0fba : SUCCESS in 4m 01s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c389bbd3c1cd418ba5ee07f1dff1798a : SUCCESS in 3m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/499b185a332c451daf15c36fe9fbdac4 : SUCCESS in 3m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/bf74bd679a86421688f30462dfb548e4 : SUCCESS in 3m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6b3443c9f234ad7beeecc52052230c3 : SUCCESS in 6m 36s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/95639d0f8d9e42f7a972eecf94e6ba11 : SUCCESS in 1h 03m 24s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/25aa9754774c4101b71d957533baade8 : SUCCESS in 46m 39s\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/5c689502825d49a8852e9c8ba1ab218c : SUCCESS in 1h 26m 00s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/73b5832a8d1a42648ad9438e6b228419 : SUCCESS in 1h 35m 54s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/d4c5ccac9ac34921842165127c8735bb : SUCCESS in 1h 07m 20s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/7951142920b74fbba887a4294f03fe55 : SUCCESS in 43m 48s\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/63af5982c1454dadb2300044faf039e3 : SUCCESS in 1h 50m 59s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/aa2a0ee4494345b1ac0c5c2f84006dfc : FAILURE in 1h 23m 19s (non-voting)\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/2be3db78c83a44bda743a63ed1188922 : SUCCESS in 1h 08m 11s (non-voting)\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c4bb347b7d3644a9a5d7261833d3863b : SUCCESS in 41m 45s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"2ac7a45c5879742265f1fbda3c52a6f9540ae545","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-01 12:21:50.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"41199e1146a789b1d849b8e5c9a0fd5e2f3db0c3","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2019-10-01 16:08:30.000000000","message":"Patch Set 4: Workflow-1\n\nThere are some issues with stuff not being found, e.g:\n\n  TASK [prechecks : Checking docker SDK version] ***************************************\n  fatal: [localhost]: FAILED! \u003d\u003e {\"changed\": false, \"cmd\": [\"/usr/bin/python\", \"-c\", \"import docker; print(docker.__version__)\"], \"delta\": \"0:00:00.047217\", \"end\": \"2019-10-01 16:04:14.706860\", \"failed_when_result\": true, \"msg\": \"non-zero return code\", \"rc\": 1, \"start\": \"2019-10-01 16:04:14.659643\", \"stderr\": \"Traceback (most recent call last):\\n  File \\\"\u003cstring\u003e\\\", line 1, in \u003cmodule\u003e\\nImportError: No module named docker\", \"stderr_lines\": [\"Traceback (most recent call last):\", \"  File \\\"\u003cstring\u003e\\\", line 1, in \u003cmodule\u003e\", \"ImportError: No module named docker\"], \"stdout\": \"\", \"stdout_lines\": []}","accounts_in_message":[],"_revision_number":4},{"id":"c3b90f830e66583a3f6bf81100818a8e34c573fc","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-01 16:20:23.000000000","message":"Patch Set 4: -Code-Review","accounts_in_message":[],"_revision_number":4},{"id":"52981759e024b5e2e8af4144069568b7982b1520","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-01 16:21:09.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"b556f4567c58b4bbe558acb98ff33025e0a7e0f0","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2019-10-01 16:54:15.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"d31dc90d17be21e7c93e36f6946855e86e02b375","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2019-10-01 16:54:34.000000000","message":"Patch Set 5: Workflow-1","accounts_in_message":[],"_revision_number":5},{"id":"d1208adf77606757e73f8f6617077b6851877bc1","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2019-10-01 19:04:08.000000000","message":"Patch Set 5:\n\ndarn!\n\n    operator: Installing collected packages: chardet, six, websocket-client, certifi, ipaddress, urllib3, backports.ssl-match-hostname, idna, requests, docker\n    operator:   Found existing installation: chardet 2.2.1\n    operator:     Uninstalling chardet-2.2.1:\n    operator:       Successfully uninstalled chardet-2.2.1\n    operator:   Found existing installation: ipaddress 1.0.16\n    operator: ERROR: Cannot uninstall \u0027ipaddress\u0027. It is a distutils installed project and thus we cannot accurately determine which files belong to it which would lead to only a partial uninstall.\n\nGoing to fix the venv install instead","accounts_in_message":[],"_revision_number":5},{"id":"28bfb84dd94f1531b7461b3f01d87fc592a0ef2f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-01 19:18:18.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2a337e9dc6464fadb847bf11339ac623 : SUCCESS in 3m 34s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/511cadaef2a64e2d9eead628406d953a : SUCCESS in 3m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1d577ed1d18459e87d3584110550c6f : SUCCESS in 4m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/09435dcfb2254849bad278e5c60befaa : SUCCESS in 5m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7918b1e1f1ab4c2da27609698f09a946 : SUCCESS in 4m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d87b6cfd103f4538bcb2375c22badc62 : SUCCESS in 5m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5862d43fa93a43f7929fdbacd4744fe2 : SUCCESS in 5m 45s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/05708b83b6fc49eca9d9c9b16d357d11 : SUCCESS in 1h 04m 22s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c97cc4fa4115454ead9906b641b1ae1d : SUCCESS in 43m 43s\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/4d8d3984d3294de6b9ec9121382f1dfe : SUCCESS in 1h 23m 52s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/23eb9ed117214d4496382bdd7f03dc1f : SUCCESS in 1h 41m 40s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/e26af5e98a9e46f7a6840229e4c66924 : SUCCESS in 1h 00m 12s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/10f4f06542f84b2ca6d350b6414d607c : SUCCESS in 50m 02s\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/a01890a107574f01b064e5a0164e4157 : SUCCESS in 1h 49m 44s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/ba254f961bdc4355b7b0ab6d1141731c : SUCCESS in 1h 29m 27s (non-voting)\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/0b437ec4afe14d9aaa49d25160ff8d7a : SUCCESS in 1h 08m 13s (non-voting)\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/78fc3991a4aa46948ad43410f0033505 : SUCCESS in 43m 46s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"c3f180084ae8ba9b0152ef573217290886443200","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-02 08:51:52.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"6c5a670f1f5512b8678a47c85fb512f048b768e8","author":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"},"date":"2020-09-21 13:36:33.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"c38154fcc66197cacd753898d26d560089e73193","revisions":{"2bebfad09a4a950bbc6c1b4119ec877a2944a0ee":{"kind":"REWORK","_number":1,"created":"2019-09-30 14:20:40.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/21/685721/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/685721/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/685721/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/685721/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/685721/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/685721/1"}}},"commit":{"parents":[{"commit":"78caa69440438bf0175be2856a04e358f133c0cd","subject":"Merge \"Remove deprecated option \u0027ovsdb_interface\u0027\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/78caa69440438bf0175be2856a04e358f133c0cd"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2019-09-30 14:16:21.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2019-09-30 14:16:21.000000000","tz":60},"subject":"Use virtualenv for vagrant bootstrap.sh script","message":"Use virtualenv for vagrant bootstrap.sh script\n\nmore-itertools \u003e 5.0.0 is no longer compatible with python2.\nOne way to fix this is to upgrade pip, see:\n\nhttps://review.opendev.org/#/c/682119/\n\nbut it seems dirty to overwrite stuff installed by the system\npackage manager, so this change uses a virtualenv and upgrades\npip in there.\n\nChange-Id: I4d96f921e025bdea783c8e88682683b25d844ce1\nCloses-Bug: #1845963\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/2bebfad09a4a950bbc6c1b4119ec877a2944a0ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/2bebfad09a4a950bbc6c1b4119ec877a2944a0ee"}]},"branch":"refs/heads/master"},"11964a4d4cae28655332962ee0aece88c971a11e":{"kind":"REWORK","_number":2,"created":"2019-09-30 14:27:59.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/21/685721/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/685721/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/685721/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/685721/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/685721/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/685721/2"}}},"commit":{"parents":[{"commit":"78caa69440438bf0175be2856a04e358f133c0cd","subject":"Merge \"Remove deprecated option \u0027ovsdb_interface\u0027\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/78caa69440438bf0175be2856a04e358f133c0cd"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2019-09-30 14:16:21.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2019-09-30 14:28:22.000000000","tz":60},"subject":"Use virtualenv for vagrant bootstrap.sh script","message":"Use virtualenv for vagrant bootstrap.sh script\n\nmore-itertools \u003e 5.0.0 is no longer compatible with python2.\nOne way to fix this is to upgrade pip, see:\n\nhttps://review.opendev.org/#/c/682119/\n\nbut it seems dirty to overwrite stuff installed by the system\npackage manager, so this change uses a virtualenv and upgrades\npip in there.\n\nChange-Id: I4d96f921e025bdea783c8e88682683b25d844ce1\nCloses-Bug: #1845963\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/11964a4d4cae28655332962ee0aece88c971a11e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/11964a4d4cae28655332962ee0aece88c971a11e"}]},"branch":"refs/heads/master"},"6a00267fe13401cca585d3b83d2e2937fb6a7ca5":{"kind":"REWORK","_number":3,"created":"2019-09-30 15:39:59.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/21/685721/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/685721/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/685721/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/685721/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/685721/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/685721/3"}}},"commit":{"parents":[{"commit":"78caa69440438bf0175be2856a04e358f133c0cd","subject":"Merge \"Remove deprecated option \u0027ovsdb_interface\u0027\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/78caa69440438bf0175be2856a04e358f133c0cd"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2019-09-30 14:16:21.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2019-09-30 15:40:21.000000000","tz":60},"subject":"Use virtualenv for vagrant bootstrap.sh script","message":"Use virtualenv for vagrant bootstrap.sh script\n\nmore-itertools \u003e 5.0.0 is no longer compatible with python2.\nOne way to fix this is to upgrade pip, see:\n\nhttps://review.opendev.org/#/c/682119/\n\nbut it seems dirty to overwrite stuff installed by the system\npackage manager, so this change uses a virtualenv and upgrades\npip in there.\n\nChange-Id: I4d96f921e025bdea783c8e88682683b25d844ce1\nCloses-Bug: #1845963\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/6a00267fe13401cca585d3b83d2e2937fb6a7ca5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/6a00267fe13401cca585d3b83d2e2937fb6a7ca5"}]},"branch":"refs/heads/master"},"18580c1d3330eee9201196e106b47169c715652a":{"kind":"REWORK","_number":4,"created":"2019-10-01 10:12:10.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/21/685721/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/685721/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/685721/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/685721/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/685721/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/685721/4"}}},"commit":{"parents":[{"commit":"78caa69440438bf0175be2856a04e358f133c0cd","subject":"Merge \"Remove deprecated option \u0027ovsdb_interface\u0027\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/78caa69440438bf0175be2856a04e358f133c0cd"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2019-09-30 14:16:21.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2019-10-01 10:12:28.000000000","tz":60},"subject":"Use virtualenv for vagrant bootstrap.sh script","message":"Use virtualenv for vagrant bootstrap.sh script\n\nmore-itertools \u003e 5.0.0 is no longer compatible with python2.\nOne way to fix this is to upgrade pip, see:\n\nhttps://review.opendev.org/#/c/682119/\n\nbut it seems dirty to overwrite stuff installed by the system\npackage manager, so this change uses a virtualenv and upgrades\npip in there.\n\nChange-Id: I4d96f921e025bdea783c8e88682683b25d844ce1\nCloses-Bug: #1845963\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/18580c1d3330eee9201196e106b47169c715652a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/18580c1d3330eee9201196e106b47169c715652a"}]},"branch":"refs/heads/master"},"c38154fcc66197cacd753898d26d560089e73193":{"kind":"REWORK","_number":5,"created":"2019-10-01 16:54:15.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/21/685721/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/685721/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/685721/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/685721/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/685721/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/685721/5"}}},"commit":{"parents":[{"commit":"78caa69440438bf0175be2856a04e358f133c0cd","subject":"Merge \"Remove deprecated option \u0027ovsdb_interface\u0027\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/78caa69440438bf0175be2856a04e358f133c0cd"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2019-09-30 14:16:21.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2019-10-01 16:53:51.000000000","tz":60},"subject":"Use virtualenv for vagrant bootstrap.sh script","message":"Use virtualenv for vagrant bootstrap.sh script\n\nmore-itertools \u003e 5.0.0 is no longer compatible with python2.\nOne way to fix this is to upgrade pip, see:\n\nhttps://review.opendev.org/#/c/682119/\n\nbut it seems dirty to overwrite stuff installed by the system\npackage manager, so this change uses a virtualenv and upgrades\npip in there.\n\nChange-Id: I4d96f921e025bdea783c8e88682683b25d844ce1\nCloses-Bug: #1845963\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/c38154fcc66197cacd753898d26d560089e73193"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/c38154fcc66197cacd753898d26d560089e73193"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":27380,"name":"Boyron Sébastien","email":"sboyron@redhat.com","username":"sboyron"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"NonZeroBackportCandidate","description":"A non-zero vote on Backport-Candidate is required before submitting changes","status":"UNSATISFIED","is_legacy":false,"applicability_expression_result":{"fulfilled":true,"status":"PASS"},"submittability_expression_result":{"expression":"label:Backport-Candidate\u003d1 OR label:Backport-Candidate\u003d-1","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Backport-Candidate\u003d1","label:Backport-Candidate\u003d-1"],"atom_explanations":{"label:Backport-Candidate\u003d1":"","label:Backport-Candidate\u003d-1":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
