)]}'
{"id":"openstack%2Fkolla-ansible~686921","triplet_id":"openstack%2Fkolla-ansible~master~I77a6534dd2ddd188f795e17d17a44be249d01f31","project":"openstack/kolla-ansible","branch":"master","hashtags":[],"change_id":"I77a6534dd2ddd188f795e17d17a44be249d01f31","subject":"CI: Use any_errors_fatal in pre.yml and run.yml","status":"MERGED","created":"2019-10-06 07:05:06.000000000","updated":"2019-10-11 16:09:39.000000000","submitted":"2019-10-11 16:07:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"686921-1570810058359-9e252c24","meta_rev_id":"b3e37bf335f96bc2c5f04a3a4b4375f68dba81ae","_number":686921,"virtual_id_number":686921,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-11 16:07:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"all":[{"value":2,"date":"2019-10-11 15:08:14.000000000","_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":1,"date":"2019-10-09 06:31:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":1,"date":"2019-10-10 02:27:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":2,"date":"2019-10-07 17:15:39.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"all":[{"value":1,"date":"2019-10-11 15:08:14.000000000","_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"value":0,"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-07 17:15:39.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2019-10-09 06:31:44.000000000","updated_by":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"reviewer":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"state":"REVIEWER"},{"updated":"2019-10-10 02:27:57.000000000","updated_by":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"reviewer":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"state":"REVIEWER"},{"updated":"2019-10-11 15:08:14.000000000","updated_by":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"reviewer":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"state":"REVIEWER"},{"updated":"2019-10-11 16:07:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"176b21add13c05ea7c94b9bbed794129746f3289","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-06 07:05:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1796c66f2c854a4b0533f8d2762090d8da3b199d","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-06 07:47:22.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b037c1c09552fa8f741a77c7025ecec63685fa8a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-06 09:07:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4f5625e8d87647d5932fc862bfc093f8 : SUCCESS in 3m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/91a677f323a4443898d6860ce226517a : SUCCESS in 3m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24fa2850bb334709a39cc66a767fa2d7 : SUCCESS in 3m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/09affb00b0ef4ea0a782afc991789954 : SUCCESS in 3m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e5e7fa272dcb424fb3bc06aa6ca0d913 : SUCCESS in 3m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1ee45f9c7d444d8983abc15f4f127ca8 : SUCCESS in 3m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e7b75ad096b4abea9c277f7a4d5d8d0 : SUCCESS in 5m 14s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/b616b9ebfcf147a18633f92bd50fe350 : SUCCESS in 58m 27s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/61174fef512a47059950b94849cfdd3f : SUCCESS in 50m 12s (non-voting)\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7de2a48fae844fecb1a31a99157beeb9 : SUCCESS in 47m 26s\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/12d3ee0485c34a9aa2b63e14484596b0 : SUCCESS in 1h 17m 33s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/c5a953e33fbb4ca5a3c469fa686ef0af : SUCCESS in 1h 37m 47s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/d783d9db956948b9965495846a4de0f5 : SUCCESS in 1h 01m 49s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/275c3f6fa26345469c094548c9531b36 : SUCCESS in 47m 42s\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/75ccb51532cf4e6d9aebf8df4d338e4b : SUCCESS in 1h 43m 58s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/b34ea61cd2a94b3fbb6f688eebb8d084 : SUCCESS in 1h 20m 50s (non-voting)\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/1a49082d0d1a4ae0add1d6d722cef2ba : SUCCESS in 59m 09s (non-voting)\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/dc09db719a2f43b18ab68172ddd193d0 : SUCCESS in 41m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"08684842fb3737f992c374fb3ee5ac8c1344bce0","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2019-10-07 09:33:45.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"48f83e5bddeeaf7be74b97d6e625b93ad715ee6a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-07 11:25:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/720d51eee50b4a969e07f8db69d34419 : SUCCESS in 3m 31s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7c8bf8704f4742a3a7a82a3be31f6870 : SUCCESS in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1df4b9d5156458b90889e0363856912 : SUCCESS in 3m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bafd0c21d22b420eb68aa5773e570858 : SUCCESS in 3m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b35e64487f334d19bc3d0a11bb42ea62 : SUCCESS in 3m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1895f9e687bb42c8a28424dd2803c27f : SUCCESS in 3m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ceb9c9763e684651aa7dafe26bfab9cb : SUCCESS in 5m 27s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/fc4547c657394174ab314ecf0bde5538 : SUCCESS in 57m 22s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/9ae5805ea8c041e282a08cacfc1429cc : SUCCESS in 46m 01s (non-voting)\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/f7160610a4d94dceb8e665e89fac449a : SUCCESS in 46m 56s\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/31dc62766f204df2ab9cfdc21d430e4c : SUCCESS in 1h 16m 01s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/ba81287e56df45bd9270a121f0005cd6 : SUCCESS in 1h 39m 08s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/865c9b4dd767497ea40f5b25e9d1315f : SUCCESS in 1h 05m 38s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/e9c72b12cc3e476f888fdc54f83d4323 : SUCCESS in 48m 03s\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/9320aaeff679417badd667488839c0d2 : SUCCESS in 1h 46m 37s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/60b7140c37b24cba91f1fe13ccc7c61e : FAILURE in 55m 49s (non-voting)\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/0fbf16b5379548e8bbd449a7fce0c3c5 : SUCCESS in 57m 52s (non-voting)\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2dd15d365b964dd0810dd1159c16ce3c : SUCCESS in 39m 14s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0f36c0f5369654edf969f6d75618850e5917fe94","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2019-10-07 17:15:39.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"11ee0a7e13fedd025599ff8204d1c2d34ae0fff0","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2019-10-09 06:31:44.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"d8e916e8bb14a0ac6173dce8c80d0ce66a5f7670","author":{"_account_id":17130,"name":"melissaml","email":"ma.lei@99cloud.net","username":"malei"},"date":"2019-10-10 02:27:57.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"2a99148d0db6934adfa820e59c8996700c007c8f","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2019-10-11 15:08:14.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ee90bbe097af8a7838f37b34b3894b02d2387018","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-11 15:08:28.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"94d6b3b90cffff3f99632e35247df854773a1830","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-11 16:07:38.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/26b6068978d34ae19a3102608d015f58 : SUCCESS in 5m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae7f8fa77864495bbff2809980b40e31 : SUCCESS in 3m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/cb4cd0d013be4464936f406a686bc6fd : SUCCESS in 3m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/515c68916e2a4a3ab1ca6c793d9c1d55 : SUCCESS in 3m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6607f2d78df641b996b6c51af6957039 : SUCCESS in 3m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f966ce0c04d48f285f47d9982e53f51 : SUCCESS in 5m 14s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/4faeb04c765d4938bf1eb17c5c3ed62b : SUCCESS in 53m 25s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/50489f42f60d44d498c3990bf521f037 : SUCCESS in 45m 15s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/18cd40fce45444fdaa7484afd8be57f5 : SUCCESS in 57m 41s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/8df40d2b60aa423596ff2c8ddd2b8156 : SUCCESS in 50m 33s","accounts_in_message":[],"_revision_number":2},{"id":"3eba1fd820313ad0c4e79d279346394b2f1a26a7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-11 16:07:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"b3e37bf335f96bc2c5f04a3a4b4375f68dba81ae","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-11 16:09:39.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b307b4429a944c2284d142c0c6c86018 : SUCCESS in 1m 29s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"fac1670433dcca250cc49086d4353c36fe17bb8e","revisions":{"140bd78bd25c82891c9cf0784b33420c003a6414":{"kind":"REWORK","_number":1,"created":"2019-10-06 07:05:06.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/21/686921/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/686921/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/686921/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/686921/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/686921/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/686921/1"}}},"commit":{"parents":[{"commit":"42bfc533756d961d7cc0935706b301f00640774c","subject":"Merge \"Update mention of Kayobe in documentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/42bfc533756d961d7cc0935706b301f00640774c"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-10-06 07:02:42.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-10-06 07:04:49.000000000","tz":60},"subject":"CI: Use any_errors_fatal in pre.yml and post.yml","message":"CI: Use any_errors_fatal in pre.yml and post.yml\n\nThis ensures that failure of a single host fails the whole play at that\ntask. This can avoid confusing errors such as when the task\n\"Assert that the nodepool private IPv4 address is assigned\" fails on one\nhost, causing subsequent errors on other hosts.\n\nNote that this only affects the Zuul playbooks, not Kolla Ansible\u0027s\nplaybooks.\n\nChange-Id: I77a6534dd2ddd188f795e17d17a44be249d01f31\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/140bd78bd25c82891c9cf0784b33420c003a6414"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/140bd78bd25c82891c9cf0784b33420c003a6414"}]},"branch":"refs/heads/master"},"fac1670433dcca250cc49086d4353c36fe17bb8e":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-10-07 09:33:45.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/21/686921/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/686921/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/686921/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/686921/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/686921/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/686921/2"}}},"commit":{"parents":[{"commit":"42bfc533756d961d7cc0935706b301f00640774c","subject":"Merge \"Update mention of Kayobe in documentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/42bfc533756d961d7cc0935706b301f00640774c"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-10-06 07:02:42.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-10-07 09:33:41.000000000","tz":0},"subject":"CI: Use any_errors_fatal in pre.yml and run.yml","message":"CI: Use any_errors_fatal in pre.yml and run.yml\n\nThis ensures that failure of a single host fails the whole play at that\ntask. This can avoid confusing errors such as when the task\n\"Assert that the nodepool private IPv4 address is assigned\" fails on one\nhost, causing subsequent errors on other hosts.\n\nNote that this only affects the Zuul playbooks, not Kolla Ansible\u0027s\nplaybooks.\n\nChange-Id: I77a6534dd2ddd188f795e17d17a44be249d01f31\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/fac1670433dcca250cc49086d4353c36fe17bb8e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/fac1670433dcca250cc49086d4353c36fe17bb8e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
