)]}'
{"id":"openstack%2Fkolla-ansible~723021","triplet_id":"openstack%2Fkolla-ansible~master~Ib8f2e6b6672e7c06aa94bc226c4d72640d25d8c2","project":"openstack/kolla-ansible","branch":"master","topic":"bug/1856346","hashtags":[],"change_id":"Ib8f2e6b6672e7c06aa94bc226c4d72640d25d8c2","subject":"Check that used Ansible can see Kolla Ansible","status":"MERGED","created":"2020-04-24 19:42:38.000000000","updated":"2020-05-25 09:14:19.000000000","submitted":"2020-05-24 13:39:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"723021-1590327598147-7d9b3224","meta_rev_id":"cfe746d5f1246bbde4a1b406b9272afa199516d4","_number":723021,"virtual_id_number":723021,"owner":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"date":"2020-05-23 15:17:33.000000000","_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":27336,"name":"XiaojueGuan","email":"xiaojueguan@gmail.com","username":"XiaojueGuan"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"date":"2020-05-23 15:32:31.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-24 13:39:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":2,"date":"2020-05-17 11:05:12.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":1,"date":"2020-05-18 00:10:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27336,"name":"XiaojueGuan","email":"xiaojueguan@gmail.com","username":"XiaojueGuan"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":2,"date":"2020-05-24 12:37:38.000000000","_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":27336,"name":"XiaojueGuan","email":"xiaojueguan@gmail.com","username":"XiaojueGuan"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":1,"date":"2020-05-24 12:37:38.000000000","_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":27336,"name":"XiaojueGuan","email":"xiaojueguan@gmail.com","username":"XiaojueGuan"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":27336,"name":"XiaojueGuan","email":"xiaojueguan@gmail.com","username":"XiaojueGuan"},{"value":1,"date":"2020-04-29 15:51:24.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"_account_id":27336,"name":"XiaojueGuan","email":"xiaojueguan@gmail.com","username":"XiaojueGuan"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-25 14:35:42.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2020-04-29 15:51:24.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-05-17 11:05:12.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2020-05-18 00:10:39.000000000","updated_by":{"_account_id":27336,"name":"XiaojueGuan","email":"xiaojueguan@gmail.com","username":"XiaojueGuan"},"reviewer":{"_account_id":27336,"name":"XiaojueGuan","email":"xiaojueguan@gmail.com","username":"XiaojueGuan"},"state":"REVIEWER"},{"updated":"2020-05-23 15:17:33.000000000","updated_by":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"reviewer":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"state":"REVIEWER"},{"updated":"2020-05-24 12:37:38.000000000","updated_by":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"reviewer":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"state":"REVIEWER"},{"updated":"2020-05-24 13:39:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"01a42e1424bb07e9b1d9558d1806012d663ccbda","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-24 19:42:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d0a711deab63cf64acf159fa77eb042fc5e646aa","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-24 19:42:51.000000000","message":"Patch Set 1:\n\nbp to Train","accounts_in_message":[],"_revision_number":1},{"id":"1a186f8abd7537c11a3b13b8cbaea39b2f91f655","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 19:43:49.000000000","message":"Patch Set 1: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"4989cbd0552ee14824ed9c364320225fcfeeb2e6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 19:43:50.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"e7c7489a2bb530b2d655345bce1ac10c9e2c15d4","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-24 19:46:04.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"55e2539b319ac6ba0b55719d2ea1c7a2753552a6","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-24 19:46:43.000000000","message":"Patch Set 2:\n\nweird that it could not merge by itself, looks like bug in gerrit","accounts_in_message":[],"_revision_number":2},{"id":"2f7942d74cea63cfa0d318c90d2b9b53c667a0ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 19:47:15.000000000","message":"Patch Set 2: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":2},{"id":"230c57749a039d26de193f0c32772b19ab94ce0e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 19:47:16.000000000","message":"Patch Set 2:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":2},{"id":"ecffc3dc2520092d7fc3f33b98fa518f9bec6b4b","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-24 19:47:52.000000000","message":"Patch Set 2:\n\n??","accounts_in_message":[],"_revision_number":2},{"id":"52e6e59a7419768778b326d55d89c3011e38168f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-24 19:53:13.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"dc6a76b676b432a0b65cf56dc7522aef849fd243","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 19:54:24.000000000","message":"Patch Set 3: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":3},{"id":"a5e746cac06b33ee552339a85cc0561847fec896","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 19:54:28.000000000","message":"Patch Set 3:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":3},{"id":"b8f32ed1699e92b307724321252bd2af53503e5c","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-24 19:59:42.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"72c28c0b348665f3a8bf1f6e206c9f1e4467ac8a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 20:00:53.000000000","message":"Patch Set 4: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":4},{"id":"70ee135b4747cb58cd151f26e5b43878bb6487aa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 20:00:54.000000000","message":"Patch Set 4:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":4},{"id":"6172c8e6b83e59959b34c1721227669e196a4027","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-24 20:05:33.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"27317fbe0f365d06574977dec67676898b86528f","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-25 04:56:33.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"c23658d6c028d285980ceb223672059280a80294","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-25 05:01:12.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/30d0daaaa6214e30a292ecd5bc87a691 : FAILURE in 4m 07s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"afd1cc2db8ca2a9783cefab016b968486cf31211","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-25 07:12:34.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/008318f1dede449b82e7c6f9c13bbec5 : SUCCESS in 3m 38s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a5255dd79e824e4bafff63dc2f2f01ad : SUCCESS in 3m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e7e6f5d2b9d74362b999c61203c3877e : SUCCESS in 3m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ed8cde7313c24d5b939c47e1cb9c6bbd : SUCCESS in 3m 49s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/367fb18a039a48f790a8faf08f1783ea : SUCCESS in 3m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/92cecf1bfa894e1598f68cc5a13c6e9d : SUCCESS in 3m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4baa9a9938c41ab8a5e5c54208fbbc2 : SUCCESS in 5m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8ac633001b764c90a80ae7a08169ac50 : SUCCESS in 4m 56s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/5f0c4e6c2be347c4b23521b7f03a7994 : SUCCESS in 1h 04m 56s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/20685bf0cde94c658e8f620262d47f8b : SUCCESS in 50m 38s (non-voting)\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/f15d39e5c38f4f34920e31ba2731cf31 : SUCCESS in 50m 28s\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/75b5f85e5e194a14bce6c781db241d11 : SUCCESS in 1h 13m 33s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/dbaf792c460743398a9e76d74f32bf1b : SUCCESS in 1h 05m 43s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/7ab4f77e62be4f708370f364396c6e83 : SUCCESS in 54m 29s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d09de8c627bc4d2eb7588f3e8a75ea26 : SUCCESS in 57m 46s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/f585a96679f040aea5990cd91d40c670 : SUCCESS in 58m 15s (non-voting)\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ba7ace08dc414d5b9842b9bf35f640ca : SUCCESS in 49m 01s (non-voting)\n- kolla-ansible-centos8-source-cells https://zuul.opendev.org/t/openstack/build/d49d377c7023434892e33a3a12e8cc4a : SUCCESS in 1h 18m 06s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/5d7de8edfa9c447e83f52239d373d773 : SUCCESS in 1h 41m 21s (non-voting)\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/e1384e94b3c243df8f0f0dec3fedbe25 : SUCCESS in 1h 22m 41s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/da54e4f7198948719ed859958c3b2079 : SUCCESS in 1h 50m 21s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/b2f3dc4fc89f4627a49cda2651a1324a : SUCCESS in 1h 36m 06s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"99bb01c40528955c79f0d062613f822cbe52c362","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-25 09:12:14.000000000","message":"Patch Set 4:\n\nthanks, Andreas","accounts_in_message":[],"_revision_number":4},{"id":"9daa53ba773a56ff10736a3fc9d43d4f5bfd2734","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-25 09:13:05.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"8f67c13d81dd0161f84a28545ae4c4313d825a6f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-25 14:33:44.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"e99c396ea1327f58fa42d217eee2b46062b88c0c","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-25 14:35:28.000000000","message":"Patch Set 5:\n\nMark, I tried via Ansible but it turned out to make basic Kolla-Ansible calls sluggish... We might want to replace this bash script with Python at some point so it is not that bad to rely on Python. That said, this change keeps the logic that Ansible drives the Python choice and not the other way around.","accounts_in_message":[],"_revision_number":5},{"id":"74b5a54598684728b9c1cb0eb28203fdb692da88","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-25 14:38:07.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/0a69a5cdfe884e8584ef23461eaaf93d : FAILURE in 3m 55s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"6193fda8b7066bedd0280edc8cbed8e07d322d67","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-25 15:53:26.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/07b57b96dc584d13ba174344879a54d2 : SUCCESS in 4m 11s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/85d5818cf2524a109ea4d764aa80bc24 : SUCCESS in 3m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ec8d5b863914664ab92798fac83b2ef : SUCCESS in 4m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8631bbe7962d475aadd5d3e088e0d90d : SUCCESS in 3m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/233e1fb59bb14a4fac5fff603b019840 : SUCCESS in 4m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8e21d77efa9944ffb2a607e8bf0315d9 : SUCCESS in 3m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/607a1f36a0914ec58bc668ebed049b20 : SUCCESS in 6m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0b794850b5de454fac3b8296377d4cdf : SUCCESS in 4m 25s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/80969e9c4a694d6ea4463bae6cd49560 : EXCEPTION\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/baacbc96cac04c8fa83074db1014a082 : SUCCESS in 42m 13s (non-voting)\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/5d6f591ea1d942909c719aecaf27270e : SUCCESS in 58m 56s\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/bf1fd879c15a4e079c484f1ba3498068 : EXCEPTION (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/aad427e66c3e4cc989671aee5e088072 : EXCEPTION\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/adf60b7645dd4cc98155f116da4d4dca : SUCCESS in 49m 36s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/1b49b4bd5ee649eaa20fdc26f3ec83b5 : EXCEPTION\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/6bfd41d2bfaf491a905d23ef2807ab1d : EXCEPTION (non-voting)\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3ef1f69052e941ae955713d8675cc6a1 : SUCCESS in 55m 37s (non-voting)\n- kolla-ansible-centos8-source-cells https://zuul.opendev.org/t/openstack/build/59d6349e5d5b44e9b60420694cf0b719 : EXCEPTION (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/78fa949445f0493bb7fcd0da118b9b24 : EXCEPTION (non-voting)\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/5400170456ce4a6a8676afd40c2007d4 : EXCEPTION (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/09bbfadc30b04d198612e062cbe8dd11 : EXCEPTION (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/39126fedc8d0460b8b23b4ca9efa40fc : EXCEPTION (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"fee1360d2c1835baa764e50a122525af41415235","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-25 19:39:04.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0b9c284d52c84c7f86f1d7a9201dfc0b : SUCCESS in 6m 23s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3cf28523dbd24e9896e02659db391fb8 : SUCCESS in 6m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba6944bd950144fdbdd374336a2b6821 : SUCCESS in 5m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d63c7ea2ea564c50b43e383857ed88b5 : SUCCESS in 5m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5513bf6ec1e9480782e6cc4ed28b78ee : SUCCESS in 5m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/89fcdc687d964544b70e2ea223ce9c77 : SUCCESS in 5m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/260b94afd0d84d828cf1ea989681bb77 : SUCCESS in 8m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ce4eb531f6f542f48aac69b9b71ad92d : SUCCESS in 6m 20s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/ccda07ba65d04c96bb884d455ec69463 : SUCCESS in 1h 05m 09s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/c45f6f5dffb74338b48e077d642f1891 : SUCCESS in 51m 31s (non-voting)\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/1e7e8cc4f0e14ec5a60af7aed542edc8 : SUCCESS in 53m 12s\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/0ef78e708f114f278b535717f5046f4c : SUCCESS in 1h 24m 19s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/8f6d748328d447fdbfda27d65fa39fbf : SUCCESS in 1h 12m 24s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/cdfe5bca2ab2413e8daf3ee670987ddf : SUCCESS in 1h 03m 10s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/8dec2b0438ab4e229dda98f6c619147a : SUCCESS in 1h 18m 50s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/9a73d7f67faf4f6d97f44a917fab1689 : SUCCESS in 1h 10m 13s (non-voting)\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/5d0c312c6e3143bdaf522f3a62283ed7 : SUCCESS in 1h 10m 49s (non-voting)\n- kolla-ansible-centos8-source-cells https://zuul.opendev.org/t/openstack/build/30df7e55ccd74c99a521093b69727897 : SUCCESS in 1h 19m 26s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/ca21fb51f2fa42d8b2347810dad185a3 : SUCCESS in 1h 36m 54s (non-voting)\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/635ea41d1ef24ac9a9cf78d8ef092549 : SUCCESS in 1h 27m 07s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/77a5e2e6304e43708431f26d191f10b6 : SUCCESS in 2h 00m 51s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/f67f3de3998049a5b1b49ad6699beb76 : SUCCESS in 1h 50m 10s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"749a3878a76f625aeb1521a4da32c294c2d2baf2","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-27 14:34:24.000000000","message":"Patch Set 5:\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"2bdbc618fc0a88e6345122eaaa743c09be150c67","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-27 14:45:23.000000000","message":"Patch Set 5:\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"b901babba8952860419c81e87cc0abb12dfd8266","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-27 15:18:41.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"2c180ca6ae7f64f278c0bb5d28f8c4e893758c4f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-04-27 15:18:56.000000000","message":"Patch Set 6:\n\nAll done, please bless.","accounts_in_message":[],"_revision_number":6},{"id":"daa18a39b2b66b593cbc934191092109316c025d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 15:34:19.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/8454475b2e764808bb324507ae47c9f6 : FAILURE in 5m 33s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/38b2611a855543a2960e772aa82094ef : FAILURE in 4m 32s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"966e1b4585bf34e832f8811489f805f054a6c79b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 17:14:56.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/38bbd77fd0d44a98a0ae4713d70247ab : SUCCESS in 4m 41s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/710928608b414e7cb4f3fee86a727895 : SUCCESS in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2654a860dfd94c8a9d8799289f423f5e : SUCCESS in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/67f48623e5164790aae2a3376f891026 : SUCCESS in 4m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9fa50314e0154b3b8c8b1f76062e6e27 : SUCCESS in 6m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4b1100f204524a80b976bc57ab43fcf9 : SUCCESS in 5m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35013e7efe5e416980d0289e858eb762 : SUCCESS in 7m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a5b4ef5f982d4c96ae00d07c58ee3eb0 : SUCCESS in 5m 29s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/a27cffbb5be442bebfba01945ab87289 : SUCCESS in 1h 12m 30s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/010a0cc954b4410885c7f927deb9bbf5 : SUCCESS in 48m 45s (non-voting)\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/6302f6106f08441c96b72bcc13984a25 : SUCCESS in 1h 04m 06s\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/19255c6be40d47e487166ac71e8785c6 : SUCCESS in 1h 07m 52s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/faee3f3984ce4811890c7b64bdc22204 : SUCCESS in 1h 10m 27s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/172ca6ef5e4f4be09512d422d3729512 : SUCCESS in 57m 39s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/e034e4073ba44e54b99b435ba3b5b7dd : SUCCESS in 1h 00m 11s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/fee58a261b794a058edb61b116bd4c79 : SUCCESS in 1h 00m 37s (non-voting)\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/522867dff1574e81b594cb658dcfe6a0 : SUCCESS in 56m 46s (non-voting)\n- kolla-ansible-centos8-source-cells https://zuul.opendev.org/t/openstack/build/00cedfa49a8f47be899d96249512a62f : SUCCESS in 1h 29m 43s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/42b3de36d0dd4a0e981cc237453b86ce : FAILURE in 59m 13s (non-voting)\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/7ef680ed5191489aae3feb8d366d8c6c : SUCCESS in 1h 33m 00s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/fc104e5ea3c649468a343b1210f7cab8 : SUCCESS in 1h 50m 03s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/4c4b46c7a83b4702af699ed22196247e : SUCCESS in 1h 28m 11s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"ca250cf02930bbaa08093f7fb8175430351f3b26","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-29 15:51:24.000000000","message":"Patch Set 6: Review-Priority+1","accounts_in_message":[],"_revision_number":6},{"id":"d176a1bb6fafcf06be578871b994c418d5f6e848","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-05-17 11:05:12.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"4c5de540cf96461e3a0ae422ffd798ac62fb2d4a","author":{"_account_id":27336,"name":"XiaojueGuan","email":"xiaojueguan@gmail.com","username":"XiaojueGuan"},"date":"2020-05-18 00:10:39.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"c0cd3f8c95715fd5ac1b59f6202b662a8f208f93","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2020-05-23 15:17:33.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"a25ddca735ace6abc676fb36e83f1e43f60869d4","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-05-23 15:32:31.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"3585704b4ae89a221a4bc0564e1c8520dd7b32ac","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2020-05-24 12:37:38.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"03ef6a6963252f34748c24ccf161176e21660ab3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-24 12:37:50.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"ac8e8586e42c1c761366a136900c0cb0ae4654ee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-24 13:39:57.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e7e77c498dce4e9d894a3b8e3148456c : SUCCESS in 4m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7bf1ad71e0f434293bffaf5ee035c00 : SUCCESS in 4m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a61a48325aaf46f5828a5c9a5f3ddcd5 : SUCCESS in 4m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0998ce457a854739bd75d24255a67efe : SUCCESS in 4m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c7a6e3abcaf4654944f6454e4bf4fa9 : SUCCESS in 6m 05s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/de4b4b09cb0247588713f33c86ff4fc9 : SUCCESS in 9m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c83383e2621a431aba2173647ed4b4b8 : SUCCESS in 4m 12s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/e57066cb21404fad8059738484a3683a : SUCCESS in 57m 53s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/1421ce0d445844e3a3d47cbc594b460d : SUCCESS in 1h 01m 19s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/1273972d254c41c68f0672ad0327f46d : SUCCESS in 48m 45s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/4b793a62988841238e4238cbb5bd1782 : SUCCESS in 58m 22s","accounts_in_message":[],"_revision_number":6},{"id":"b17caeefbdef9132ffc28ebb404d32df4048d625","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-24 13:39:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"ad0c2e37057f5252fc62da9a87a44de04ca2519a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-24 13:41:11.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b29ce34297704cb3ad7f0044d79e3a33 : SUCCESS in 56s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c121c72b4364479cab0e9ea6a93bbad9 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":6},{"id":"03f920a6f7e2465f48762c20dfa2b108a6de4e09","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-05-25 09:14:19.000000000","message":"Patch Set 6: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit 1f16c04a50e0c6e9614ede128736c2c97d4efe1d","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"32fc2599a665e01b47c74588cdc2b547dc208874","revisions":{"45454de12219e94c7bb2205927c16a68e74478e3":{"kind":"REWORK","_number":1,"created":"2020-04-24 19:42:38.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/21/723021/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/723021/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/723021/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/723021/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/723021/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/723021/1"}}},"commit":{"parents":[{"commit":"9251315b36e3aad4aa9f0bcee3d239105eb7abd3","subject":"Merge \"CI: Test ironic on nova-cell change\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/9251315b36e3aad4aa9f0bcee3d239105eb7abd3"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-04-24 19:38:15.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-04-24 19:38:15.000000000","tz":120},"subject":"Check that used Ansible can see Kolla Ansible","message":"Check that used Ansible can see Kolla Ansible\n\nFix-feature following up on the original check [1] to make it\ntest the correct interpreter.\n\nAdditionally, this change removes last, unneeded call to\nrandom python - getting script directory is perfectly\ndoable in bash.\n\n[1] Icf0399d21b3fde8d530d73e6e7ee4a57665da276\n\nChange-Id: Ib8f2e6b6672e7c06aa94bc226c4d72640d25d8c2\nCloses-bug: #1856346\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/45454de12219e94c7bb2205927c16a68e74478e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/45454de12219e94c7bb2205927c16a68e74478e3"}]},"branch":"refs/heads/master"},"bbe35fb26aecefba6d780e45fd849630284f6897":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-04-24 19:46:04.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/21/723021/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/723021/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/723021/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/723021/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/723021/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/723021/2"}}},"commit":{"parents":[{"commit":"3036c7cfda0a6ef01f7eb3ac25c70ba2cddb2f23","subject":"CI: Make bashate happy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/3036c7cfda0a6ef01f7eb3ac25c70ba2cddb2f23"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-04-24 19:38:15.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-04-24 19:45:53.000000000","tz":120},"subject":"Check that used Ansible can see Kolla Ansible","message":"Check that used Ansible can see Kolla Ansible\n\nFix-feature following up on the original check [1] to make it\ntest the correct interpreter.\n\nAdditionally, this change removes last, unneeded call to\nrandom python - getting script directory is perfectly\ndoable in bash.\n\n[1] Icf0399d21b3fde8d530d73e6e7ee4a57665da276\n\nChange-Id: Ib8f2e6b6672e7c06aa94bc226c4d72640d25d8c2\nCloses-bug: #1856346\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/bbe35fb26aecefba6d780e45fd849630284f6897"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/bbe35fb26aecefba6d780e45fd849630284f6897"}]},"branch":"refs/heads/master"},"aa142f73131cfeabacb2cfee0d53d15498cc9d49":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-04-24 19:53:13.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/21/723021/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/723021/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/723021/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/723021/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/723021/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/723021/3"}}},"commit":{"parents":[{"commit":"3036c7cfda0a6ef01f7eb3ac25c70ba2cddb2f23","subject":"CI: Make bashate happy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/3036c7cfda0a6ef01f7eb3ac25c70ba2cddb2f23"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-04-24 19:38:15.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-04-24 19:53:10.000000000","tz":0},"subject":"Check that used Ansible can see Kolla Ansible","message":"Check that used Ansible can see Kolla Ansible\n\nFix-feature following up on the original check [1] to make it\ntest the correct interpreter.\n\nAdditionally, this change removes last, unneeded call to\nrandom python - getting script directory is perfectly\ndoable in bash.\n\n[1] LOREM IPSUM TODO FIXME\n\nChange-Id: Ib8f2e6b6672e7c06aa94bc226c4d72640d25d8c2\nCloses-bug: #1856346\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/aa142f73131cfeabacb2cfee0d53d15498cc9d49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/aa142f73131cfeabacb2cfee0d53d15498cc9d49"}]},"branch":"refs/heads/master"},"6b8c4feda8f5a5951ecebf06b49337d8fd80ab15":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2020-04-24 19:59:42.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/21/723021/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/723021/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/723021/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/723021/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/723021/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/723021/4"}}},"commit":{"parents":[{"commit":"3036c7cfda0a6ef01f7eb3ac25c70ba2cddb2f23","subject":"CI: Make bashate happy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/3036c7cfda0a6ef01f7eb3ac25c70ba2cddb2f23"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-04-24 19:38:15.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-04-24 19:59:40.000000000","tz":0},"subject":"Check that used Ansible can see Kolla Ansible","message":"Check that used Ansible can see Kolla Ansible\n\nFix-feature following up on the original check [1] to make it\ntest the correct interpreter.\n\nAdditionally, this change removes last, unneeded call to\nrandom python - getting script directory is perfectly\ndoable in bash.\n\n[1] Icf0399d21b3fde8d530d73e6e7ee4a57665da276\n\nChange-Id: Ib8f2e6b6672e7c06aa94bc226c4d72640d25d8c2\nCloses-bug: #1856346\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/6b8c4feda8f5a5951ecebf06b49337d8fd80ab15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/6b8c4feda8f5a5951ecebf06b49337d8fd80ab15"}]},"branch":"refs/heads/master"},"223c1f89df70a0418c9d1f0e92952005cd02c16c":{"kind":"REWORK","_number":5,"created":"2020-04-25 14:33:44.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/21/723021/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/723021/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/723021/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/723021/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/723021/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/723021/5"}}},"commit":{"parents":[{"commit":"3036c7cfda0a6ef01f7eb3ac25c70ba2cddb2f23","subject":"CI: Make bashate happy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/3036c7cfda0a6ef01f7eb3ac25c70ba2cddb2f23"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-04-24 19:38:15.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-04-25 14:28:58.000000000","tz":120},"subject":"Check that used Ansible can see Kolla Ansible","message":"Check that used Ansible can see Kolla Ansible\n\nFix-feature following up on the original check [1] to make it\ntest the correct interpreter.\n\nAdditionally, this change removes last, unneeded call to\nrandom python - getting script directory is perfectly\ndoable in bash.\n\nAll checks are done from Python, not Ansible, due to its\nperformance. Python version feels snappy (0.2 s to check),\ncompared to sluggish Ansible (2.0 s to check).\nWhat is more, relying on Ansible would require hacky solutions\nto e.g. prevent custom config from interfering with it.\nWe might be willing to steer Ansible from Python in the future\nanyhow.\n\n[1] Icf0399d21b3fde8d530d73e6e7ee4a57665da276\n\nChange-Id: Ib8f2e6b6672e7c06aa94bc226c4d72640d25d8c2\nCloses-Bug: #1856346\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/223c1f89df70a0418c9d1f0e92952005cd02c16c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/223c1f89df70a0418c9d1f0e92952005cd02c16c"}]},"branch":"refs/heads/master"},"32fc2599a665e01b47c74588cdc2b547dc208874":{"kind":"REWORK","_number":6,"created":"2020-04-27 15:18:41.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/21/723021/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/723021/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/723021/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/723021/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/723021/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/723021/6"}}},"commit":{"parents":[{"commit":"3036c7cfda0a6ef01f7eb3ac25c70ba2cddb2f23","subject":"CI: Make bashate happy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/3036c7cfda0a6ef01f7eb3ac25c70ba2cddb2f23"}]}],"author":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-04-24 19:38:15.000000000","tz":120},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2020-04-27 15:18:31.000000000","tz":120},"subject":"Check that used Ansible can see Kolla Ansible","message":"Check that used Ansible can see Kolla Ansible\n\nFix-feature following up on the original check [1] to make it\ntest the correct interpreter.\n\nAdditionally, this change removes last, unneeded call to\nrandom python - getting script directory is perfectly\ndoable in bash.\n\nAll checks are done from Python, not Ansible, due to its\nperformance. Python version feels snappy (0.2 s to check),\ncompared to sluggish Ansible (2.0 s to check).\nWhat is more, relying on Ansible would require hacky solutions\nto e.g. prevent custom config from interfering with it.\nWe might be willing to steer Ansible from Python in the future\nanyhow.\n\n[1] Icf0399d21b3fde8d530d73e6e7ee4a57665da276\n\nChange-Id: Ib8f2e6b6672e7c06aa94bc226c4d72640d25d8c2\nCloses-Bug: #1856346\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/32fc2599a665e01b47c74588cdc2b547dc208874"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/32fc2599a665e01b47c74588cdc2b547dc208874"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
