)]}'
{"id":"openstack%2Fkolla-ansible~723402","triplet_id":"openstack%2Fkolla-ansible~stable%2Fstein~Ida61cef4b9c9622a5e925bac4583fba281469a39","project":"openstack/kolla-ansible","branch":"stable/stein","topic":"bug/1867926-stable/train-stable/stein","hashtags":[],"change_id":"Ida61cef4b9c9622a5e925bac4583fba281469a39","subject":"[octavia] Adds region_name if enable_barbican","status":"MERGED","created":"2020-04-27 12:26:20.000000000","updated":"2020-04-27 17:31:29.000000000","submitted":"2020-04-27 17:29:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"723402-1588008564767-bde3a662","meta_rev_id":"f55e38be08ad78e94ceeb4e7324357687b84536b","_number":723402,"virtual_id_number":723402,"owner":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-27 17:29:24.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":2,"date":"2020-04-27 15:44:58.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":2,"date":"2020-04-27 13:31:04.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"all":[{"value":1,"date":"2020-04-27 15:44:58.000000000","_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-27 13:31:04.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-04-27 15:44:58.000000000","updated_by":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"reviewer":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"state":"REVIEWER"},{"updated":"2020-04-27 17:29:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8bce453fcb59cfd12936e8083db4efd141b81b4a","author":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"date":"2020-04-27 12:26:20.000000000","message":"Patch Set 1: Cherry Picked from branch stable/train.","accounts_in_message":[],"_revision_number":1},{"id":"84f623344c7a093495f69d184498f17c3e928e43","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-27 13:31:04.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6314554cdb84c3704651fafbb1ca8fd1fbc3f075","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 13:47:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0de6c6a95618439782b9ddad359d674c : SUCCESS in 5m 09s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/56dce93f58554e5d96b0642c1195eb8d : SUCCESS in 5m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6a0bb7b70ff46aaa24e4912e228df5a : SUCCESS in 4m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/33939e18ac254d2db1a6afadddeb0e01 : SUCCESS in 5m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b4b5f8471d94419eab9a1bd518dacb68 : SUCCESS in 5m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/19bb24baaf21468a9425edf6654ad360 : SUCCESS in 6m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/698eff130aba40b2b99e2097522c101e : SUCCESS in 3m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8536dc78ddcc48428f17f5c27d4ff165 : SUCCESS in 5m 28s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/fc6bbcedf72d47bfba7e097264e9ea73 : SUCCESS in 1h 04m 53s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/cacdb19e03f24c9c936ad61d9d8127e4 : SUCCESS in 42m 39s\n- kolla-ansible-ubuntu-source-ceph https://zuul.opendev.org/t/openstack/build/11dde334f2bc42339f9147d8a36b0f84 : SUCCESS in 1h 06m 32s (non-voting)\n- kolla-ansible-centos-source-ceph https://zuul.opendev.org/t/openstack/build/498f2f6a5a9a43fc88328d160fd6cab4 : FAILURE in 1h 17m 53s (non-voting)\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/e15f2d01449948dab3319019924d48f4 : SUCCESS in 51m 48s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f80345c6698e467c9c73d2773b0ac145 : SUCCESS in 42m 36s\n- kolla-ansible-centos-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/dbceb4c98852455bbcfcf3f21646c9f8 : FAILURE in 31m 11s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph https://zuul.opendev.org/t/openstack/build/50a9c1fef9c34b52b772fc916eee67fb : SUCCESS in 1h 10m 37s (non-voting)\n- kolla-ansible-centos-binary https://zuul.opendev.org/t/openstack/build/bd3c442be4ff4d27b02235f1e182c115 : SUCCESS in 47m 14s (non-voting)\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a4cbfa401bb241658aba751bcb73b274 : FAILURE in 23m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bfe292bb28aaab95e35dd24e91d404501f845c81","author":{"_account_id":24072,"name":"Marcin Juszkiewicz","email":"mjuszkiewicz@redhat.com","username":"hrw"},"date":"2020-04-27 15:44:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d0efdd255cc8761aa1d34724038154db877da8a7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 15:45:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6709a65e8bd9361170523b719f03a01d6db3c1ff","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 17:29:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"5259bf01013918e6f1976b50d372417fdaca5f24","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 17:29:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d444c6bbd6a0431c830fffed88eb0131 : SUCCESS in 6m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/823d13d8fd3f4b6383d79c81b6438a1b : SUCCESS in 7m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ab1ce07acd1442dba6d5317cd5f37a83 : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/339cdb48f0c94c9e960b278959a71b96 : SUCCESS in 5m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/984967964a214d97baf5bfcc5b66609a : SUCCESS in 5m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1326c6b6b74a422c856d14a82d6c0d55 : SUCCESS in 6m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d2e11e60a50045c6a42e7eac9cc82871 : SUCCESS in 8m 08s\n- kolla-ansible-centos-source https://zuul.opendev.org/t/openstack/build/2495eaf9b2444c199c935a8afd06fbc9 : SUCCESS in 1h 08m 39s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/bc00038da03a4bf0b0aecb91e7bfb8e5 : SUCCESS in 47m 57s\n- kolla-ansible-centos-source-upgrade https://zuul.opendev.org/t/openstack/build/a6305b6bd50a484082340b5c565a251a : SUCCESS in 53m 40s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/9b0c1e1e9bc946108531330893f3084f : SUCCESS in 47m 24s","accounts_in_message":[],"_revision_number":1},{"id":"a1f6dd884c71b6372e638fd218409f8f5da6c37e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 17:31:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/afbb7304d314428f9babc1c1706c7a44 : SUCCESS in 1m 37s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b55f02c7fddc42088f44406a9e17f8fa : SUCCESS in 1m 34s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"967b3be7182cb9501eb3478de0f758529d22f991","revisions":{"967b3be7182cb9501eb3478de0f758529d22f991":{"kind":"REWORK","_number":1,"created":"2020-04-27 12:26:20.000000000","uploader":{"_account_id":30523,"name":"Dincer Celik","email":"hello@dincercelik.com","username":"osmanlicilegi"},"ref":"refs/changes/02/723402/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/02/723402/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/02/723402/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/02/723402/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/02/723402/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/02/723402/1"}}},"commit":{"parents":[{"commit":"fadaca8c85ff4fa6ffa76ff561940b6afa2e6c9f","subject":"Merge \"[skydive] fix: Use Keystone backend to authenticate API users\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/fadaca8c85ff4fa6ffa76ff561940b6afa2e6c9f"}]}],"author":{"name":"ramboman","email":"lijie@unitedstack.com","date":"2020-03-18 12:47:39.000000000","tz":480},"committer":{"name":"Dincer Celik","email":"hello@dincercelik.com","date":"2020-04-27 12:26:20.000000000","tz":0},"subject":"[octavia] Adds region_name if enable_barbican","message":"[octavia] Adds region_name if enable_barbican\n\nAdds necessary \"region_name\" to octavia.conf when\n\"enable_barbican\" is set to \"true\".\n\nCloses-Bug: #1867926\n\nChange-Id: Ida61cef4b9c9622a5e925bac4583fba281469a39\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/967b3be7182cb9501eb3478de0f758529d22f991"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/967b3be7182cb9501eb3478de0f758529d22f991"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
