)]}'
{"id":"openstack%2Fkolla-ansible~737821","triplet_id":"openstack%2Fkolla-ansible~stable%2Fussuri~I7fa62726d0d5b0aeb3bd5fa06dc0e59667f94fa0","project":"openstack/kolla-ansible","branch":"stable/ussuri","hashtags":[],"change_id":"I7fa62726d0d5b0aeb3bd5fa06dc0e59667f94fa0","subject":"nova-cell role clone failed","status":"MERGED","created":"2020-06-24 14:39:22.000000000","updated":"2020-06-24 18:11:21.000000000","submitted":"2020-06-24 18:09:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"737821-1593022184577-4f349c0a","meta_rev_id":"aa6905b060aebf3addfb9e9ae6145cfb1a84d372","_number":737821,"virtual_id_number":737821,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-24 18:09:44.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2020-06-24 14:39:47.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-06-24 14:57:39.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2020-06-24 14:57:39.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-06-24 14:39:22.000000000","_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-24 14:39:22.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2020-06-24 14:57:39.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-06-24 18:09:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9ce676f8985ead6fece68263dfb48e9b8071764f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-24 14:39:22.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"0f271be087b0b119903e0de349f3e53083a8a1c2","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-24 14:39:47.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"034b3728398a4dca4f9c6ceb5d78c7c6b351b6b3","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-06-24 14:57:39.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"31b147456cbe4ee971fc0005eabc0909445a465a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 15:57:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/2f3a7fba16f14dadb1c1c1aae05786af : SUCCESS in 1h 16m 55s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/de466fb917af45748b20f94fbfcc3cd9 : FAILURE in 41m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"73856eb9a7d70252e628566886e7d3cce02a8e8a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 16:45:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f6e26ee380f24a49965bf6857eb237b8 : SUCCESS in 5m 44s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b8b5401ad10c4ee2aeeadf0300f656fb : SUCCESS in 5m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ee22f6153b8b48779dd767b0ef3ef40a : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c4867639ed144c318dfee374098e1f02 : SUCCESS in 5m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4edd84768c9c42d58e2c3393f8056bdc : SUCCESS in 6m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca5efc281d60420ca62366f4360b94ce : SUCCESS in 10m 21s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/780cf1127a8d42f3a34e98c78329fb51 : SUCCESS in 13m 21s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/216715937a274e6b896dd031de85c433 : SUCCESS in 1h 21m 56s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/d03fb74755ef4702bb58d0b706a761e4 : SUCCESS in 53m 28s (non-voting)\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/fc03e51a203d4d9e9133c3b341bd896e : SUCCESS in 52m 10s\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/37632bb91f3c4f348c585ddaf8f9da29 : SUCCESS in 1h 21m 42s (non-voting)\n- kolla-ansible-centos8-source-ironic https://zuul.opendev.org/t/openstack/build/929950cd39714314bf4ce1cff98832f3 : SUCCESS in 1h 16m 38s (non-voting)\n- kolla-ansible-centos8-binary-ironic https://zuul.opendev.org/t/openstack/build/3a6b9b59ca2a4ac4bbe37cdf6b07d9d1 : SUCCESS in 1h 14m 28s (non-voting)\n- kolla-ansible-ubuntu-source-ironic https://zuul.opendev.org/t/openstack/build/efa4712354b843cf83a0829de8cd198d : SUCCESS in 59m 16s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/e0f9ff17c32041049617a63bc2f1ce11 : SUCCESS in 1h 13m 49s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/f78ddb7814e2413a8505660f0139af72 : SUCCESS in 1h 14m 16s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/6b33efcbf2f44206ae7c606d6a074f52 : SUCCESS in 1h 17m 09s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/f85eade67f064650b1c13fc9d7ff2cf6 : SUCCESS in 1h 00m 32s (non-voting)\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/91259323220e4ff7a03a7e8f7f21c9a8 : SUCCESS in 53m 59s (non-voting)\n- kolla-ansible-centos8-source-cells https://zuul.opendev.org/t/openstack/build/e0e2eba54302468d91503550a2b1024d : SUCCESS in 1h 17m 58s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/362d02d5b3bb454d9f5c1288e1a4991f : SUCCESS in 1h 38m 01s (non-voting)\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/81060738254a4df3b987b2120ef7f748 : SUCCESS in 1h 25m 12s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/f85bd77577e4448781ba1390d6971283 : POST_FAILURE in 1h 41m 52s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/eeef9e8b23054b6e8d873fefd5a26934 : SUCCESS in 1h 29m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1110527330c1fb02f1953fcdc59b53a8cb002c98","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 16:45:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"69af09bfe951ac9ec8e98f0f03c4a0608c8768d7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 18:09:44.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"44fc38722e87108dc23cd8dd37944d9300a8231e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 18:09:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6caa57813d2940a4b23b67df79754b3d : SUCCESS in 4m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5807e8c98694ff7afa9560c63748e45 : SUCCESS in 7m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/efc762023eab4280a886865d700ef701 : SUCCESS in 6m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/167e0269db0747d0a253f16a551de5af : SUCCESS in 9m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db4286dfbf204ed4b4b6301fedf41c59 : SUCCESS in 11m 50s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/91f24cbeff004151a85bd3bd6dca70d3 : SUCCESS in 16m 03s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/30302be0889e45a7a86d3badfecd3e3e : SUCCESS in 1h 09m 46s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/5cdf911f37f947e9ba6d83a1290aee33 : SUCCESS in 1h 18m 25s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/76ca5fe5e3414ff8a481717292ac3036 : SUCCESS in 55m 14s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/91b51ea9ae7d4aefa1ebd12b0f1250c0 : SUCCESS in 1h 07m 34s","accounts_in_message":[],"_revision_number":1},{"id":"aa6905b060aebf3addfb9e9ae6145cfb1a84d372","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 18:11:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8762060057fe42dfa3935ffac54ed8e0 : SUCCESS in 1m 14s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2ca01f25f0d0bc11db841807655da9a436f7f47b","revisions":{"2ca01f25f0d0bc11db841807655da9a436f7f47b":{"kind":"REWORK","_number":1,"created":"2020-06-24 14:39:22.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/21/737821/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/737821/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/737821/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/737821/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/737821/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/737821/1"}}},"commit":{"parents":[{"commit":"55eac5947d091e24d59fa434cf54a89b232633d9","subject":"Merge \"Victoria: Add releasenote for remove congress role\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/55eac5947d091e24d59fa434cf54a89b232633d9"}]}],"author":{"name":"wu.chunyang","email":"wuchunyang@yovole.com","date":"2020-06-21 14:40:56.000000000","tz":480},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-06-24 14:39:22.000000000","tz":0},"subject":"nova-cell role clone failed","message":"nova-cell role clone failed\n\nwhen enable kolla_dev_mod, nova-cell role clones code failed,\nbecause we use nova-cell repository which is not exists.\nin fact, nova-cell role should use nova repository too\n\nChange-Id: I7fa62726d0d5b0aeb3bd5fa06dc0e59667f94fa0\n(cherry picked from commit a9c94aee394a4e520ab47619d9ad8bf3898de0e2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/2ca01f25f0d0bc11db841807655da9a436f7f47b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/2ca01f25f0d0bc11db841807655da9a436f7f47b"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
