)]}'
{"id":"openstack%2Fkolla-ansible~737921","triplet_id":"openstack%2Fkolla-ansible~master~Id22eb3e74867230da02543abd93234a5fb12b31d","project":"openstack/kolla-ansible","branch":"master","hashtags":[],"change_id":"Id22eb3e74867230da02543abd93234a5fb12b31d","subject":"openvswitch: Use ansible_hostname for system-id","status":"MERGED","created":"2020-06-25 08:30:41.000000000","updated":"2020-06-26 08:56:26.000000000","submitted":"2020-06-26 08:46:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"737921-1593161174877-f0b752bd","meta_rev_id":"c5f37e130f74f2604b108c9b8392d9ce9092d9b5","_number":737921,"virtual_id_number":737921,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":26689,"name":"Justinas Balciunas","email":"justinas.balciunas@gmail.com","username":"stacklablt"},{"value":0,"date":"2020-06-25 12:20:39.000000000","_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-26 08:46:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"recommended":{"_account_id":26689,"name":"Justinas Balciunas","email":"justinas.balciunas@gmail.com","username":"stacklablt"},"all":[{"value":1,"date":"2020-06-25 13:33:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26689,"name":"Justinas Balciunas","email":"justinas.balciunas@gmail.com","username":"stacklablt"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2020-06-25 13:31:36.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2020-06-26 07:35:34.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":26689,"name":"Justinas Balciunas","email":"justinas.balciunas@gmail.com","username":"stacklablt"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2020-06-26 07:35:34.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":26689,"name":"Justinas Balciunas","email":"justinas.balciunas@gmail.com","username":"stacklablt"},"all":[{"value":1,"date":"2020-06-25 13:33:40.000000000","_account_id":26689,"name":"Justinas Balciunas","email":"justinas.balciunas@gmail.com","username":"stacklablt"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2020-06-26 07:35:37.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":26689,"name":"Justinas Balciunas","email":"justinas.balciunas@gmail.com","username":"stacklablt"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":26689,"name":"Justinas Balciunas","email":"justinas.balciunas@gmail.com","username":"stacklablt"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-25 13:31:36.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-06-25 13:33:40.000000000","updated_by":{"_account_id":26689,"name":"Justinas Balciunas","email":"justinas.balciunas@gmail.com","username":"stacklablt"},"reviewer":{"_account_id":26689,"name":"Justinas Balciunas","email":"justinas.balciunas@gmail.com","username":"stacklablt"},"state":"REVIEWER"},{"updated":"2020-06-26 07:35:34.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-06-26 08:46:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4b30bef750e99744843f4cb2e291d97417a198f4","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-06-25 08:30:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a3d603811ed3abd6ae55b73ce78ca95a71343eee","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-06-25 08:55:44.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"728d6248be50c8a34b4bf76f67048f2a98eadcb2","author":{"_account_id":26689,"name":"Justinas Balciunas","email":"justinas.balciunas@gmail.com","username":"stacklablt"},"date":"2020-06-25 09:15:23.000000000","message":"Patch Set 2: Backport-Candidate+1 Code-Review+1\n\nWorking as expected now with IP addresses in the inventory.\n\nTested on CentOS 8 source multinode deployment.","accounts_in_message":[],"_revision_number":2},{"id":"00f967420d9336d7aa90f23c3c51765ca0c63a96","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 10:46:25.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/cd23a3ddb5ea4e30907262ef6e04f2bb : SUCCESS in 1h 36m 58s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/771b9f3155ec4f5698c00830c53e19ce : SUCCESS in 1h 42m 32s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d98f7d41723d0c7d66fc6a0fe6b6b487423cae3c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 11:00:05.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c46a34f8a7234d01b274b3225e525591 : SUCCESS in 6m 15s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9629424b99ca458eaeb8bdc3a3145ab4 : SUCCESS in 6m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b20cd4c745684bc2ac92866a5de4ee14 : SUCCESS in 5m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/df7a161e983f415485bfdeef25574c7b : SUCCESS in 5m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5e68162bc2bc4ed3aecab2d772190674 : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13d67a88450b480985b413fed76e6b8f : SUCCESS in 7m 58s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/a76c0a007b904b3fb811f6a4a7e6a984 : SUCCESS in 12m 39s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/bc01f2981d684bce979b96ac02c662e1 : SUCCESS in 1h 07m 18s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/971c5746b377411692e9e686b9f16ccf : SUCCESS in 45m 38s (non-voting)\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/190dde733c14490886acbd0cb39f09ab : SUCCESS in 59m 34s\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/56cc1fbc9269426e98697cb19842e8a5 : SUCCESS in 1h 11m 38s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/17ba5ca1a80d452e87b2bcf94d116d67 : SUCCESS in 1h 11m 44s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/32c6bdd8152e4206960fb778b1763a00 : SUCCESS in 58m 30s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/11387b6d81324e6ea3d8b849d10070bb : SUCCESS in 1h 06m 21s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/4cf300bd1390483b84b67c7a968e342c : SUCCESS in 1h 01m 32s (non-voting)\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/35c9723a13704359b149e3b5f2baf111 : SUCCESS in 55m 13s (non-voting)\n- kolla-ansible-centos8-source-cells https://zuul.opendev.org/t/openstack/build/1d2db94650d44287916aa282b24539f2 : SUCCESS in 1h 30m 33s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/e65034738a964efab90e99c2c7a8c30e : FAILURE in 1h 33m 53s (non-voting)\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/e0ccfc9f4728471daf78f0371391f713 : SUCCESS in 1h 31m 58s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/3a47b27dc979404cabceb9227714ebf8 : SUCCESS in 1h 58m 18s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/8a1b8a28d6aa4bdc8e0d5a02c7ceee82 : SUCCESS in 1h 45m 48s (non-voting)\n- kolla-ansible-centos8-source-ovn https://zuul.opendev.org/t/openstack/build/a1f0a33d3cb64335972987c33283b997 : SUCCESS in 1h 21m 45s (non-voting)\n- kolla-ansible-ubuntu-source-ovn https://zuul.opendev.org/t/openstack/build/579dda439fe0416abcac5c0b5a20f0f7 : SUCCESS in 1h 24m 27s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"120d035c623118e711e4f8799145d6e609d5badb","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-25 11:10:19.000000000","message":"Patch Set 2:\n\nRelease note?","accounts_in_message":[],"_revision_number":2},{"id":"c881b07510a1a8baaa70a13c36b6e3fb6c14f13e","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-06-25 12:20:20.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"672d2a495d73e77b49eb2bd80f8cc026163f292d","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-06-25 12:20:39.000000000","message":"Patch Set 3:\n\n\u003e Release note?\n\nHere it is. I keep forgetting about them...","accounts_in_message":[],"_revision_number":3},{"id":"14154559b9e12ad77ab3106c44f86baa617880e8","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-25 13:31:36.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"d23661abb83dfd0bea3541c16e0cc4d10d2d74e0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 13:32:31.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/4f21b6aeafc64af898d8f8bb4f6a12a9 : SUCCESS in 1h 02m 58s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/56656402ea4b4289a8ea3169ac98cbbf : SUCCESS in 1h 07m 20s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e2e4aedaeed872b41e32dd9d31e3eec408891681","author":{"_account_id":26689,"name":"Justinas Balciunas","email":"justinas.balciunas@gmail.com","username":"stacklablt"},"date":"2020-06-25 13:33:40.000000000","message":"Patch Set 3: Backport-Candidate+1 Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"fbd3b2ef1cce1e35fa6d7498b0c5f0464ff0205c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 14:59:05.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e701629e34374c778bdd36b480732d52 : SUCCESS in 5m 01s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a78e321e738a483693a63c603b8b7fa5 : SUCCESS in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/abc215bda7b44a588ed15a10be9f92b4 : SUCCESS in 5m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5d2ea26abae4479a87591ae69efbfe50 : SUCCESS in 5m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3dafbf3598694db2a968171d99a2384c : SUCCESS in 5m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15e46dabaf2843e98821b2d7a6379170 : SUCCESS in 6m 16s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/dce54b31470748c2a066d866045e9ca9 : SUCCESS in 12m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/12abfd0639464a89a946443576c5d597 : SUCCESS in 4m 12s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/d85e30ca4c894c74a848aab48213e5f5 : SUCCESS in 1h 17m 17s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/8f8a21536579458b8e8bdcbf791b38ea : SUCCESS in 43m 00s (non-voting)\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d8809509b97346d980d733414d9ae1bd : SUCCESS in 57m 32s\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/d95e8268adda490eb8085dda00465e20 : SUCCESS in 1h 14m 59s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/856d2eca22d3461da7ca0eaff47c4f1f : SUCCESS in 1h 04m 13s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/23bedffa850a462e9babfa2409598fa7 : SUCCESS in 58m 55s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/800130774558450ca6dfa6779b2f1a35 : SUCCESS in 55m 38s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/7d33316f962044b7b1c1358a4d4b0894 : SUCCESS in 1h 03m 13s (non-voting)\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/75ca2e5a01bb4d67b647781a85fd1ee0 : SUCCESS in 54m 53s (non-voting)\n- kolla-ansible-centos8-source-cells https://zuul.opendev.org/t/openstack/build/364ba19c1fd54a9491e8637daa8c57bf : SUCCESS in 1h 20m 35s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/d3b36ba73f614a6eb0606434ebb6724c : SUCCESS in 1h 22m 10s (non-voting)\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/659878ab533c4361842f1f49d71f0a96 : SUCCESS in 1h 34m 51s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/b2743d89546f42c7a750ff466ab8a8ea : SUCCESS in 1h 54m 52s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/65eab6046b114916bef9734195621a93 : SUCCESS in 1h 32m 25s (non-voting)\n- kolla-ansible-centos8-source-ovn https://zuul.opendev.org/t/openstack/build/cf442a424c0d41939ecf8d5a3c1af54c : SUCCESS in 1h 26m 21s (non-voting)\n- kolla-ansible-ubuntu-source-ovn https://zuul.opendev.org/t/openstack/build/2b73d275f4cf40b3a196a65357fd9481 : SUCCESS in 1h 09m 58s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a82709d5fa283e578fad7fd5a4090ad6764a6090","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-06-26 07:35:34.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"948999b9da9d63a83fd497f15ba44c01335a7108","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-06-26 07:35:37.000000000","message":"Patch Set 3: Backport-Candidate+2","accounts_in_message":[],"_revision_number":3},{"id":"43656942a2b0958e8f4b76c683a000af58f8d6e3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 07:35:53.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"b645fae156afb08cb7ac045fba0361b1c67b912f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 08:46:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"c4e5aec210f892ce85e16737e321e76686902bf4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 08:46:14.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e141da994d4d4d0d9bd05e088d7f355d : SUCCESS in 5m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4434320b744476884fdf83817a676c1 : SUCCESS in 5m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9117ddb0e42e4f25939a3a75707328c9 : SUCCESS in 5m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f3358193590440da876a5ef8704a7adc : SUCCESS in 5m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7a6e457be9d459f967a22237284a5f1 : SUCCESS in 7m 14s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/4a13471f57be4617acc0725014c08023 : SUCCESS in 11m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b42e9adcad914770a5a5963d1a6bcaee : SUCCESS in 4m 06s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/f3a7df2e1be84cfa8b13e82b06b6fcc9 : SUCCESS in 1h 04m 24s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/6222ee4663d94bbe9123c20631446a32 : SUCCESS in 1h 05m 28s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/22a5cd6d66534eb0aa4115ab2b8921ff : SUCCESS in 52m 28s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/297851fb071240c0a5b6723ce94974e7 : SUCCESS in 59m 22s","accounts_in_message":[],"_revision_number":3},{"id":"8efcc734ec7932a29cd87074636990856978817b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 08:49:32.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8fa52f49632d44eab53ccc2ba765193b : SUCCESS in 1m 33s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0d61eca3812341c0a7ae765d70671b84 : SUCCESS in 1m 24s","accounts_in_message":[],"_revision_number":3},{"id":"c5f37e130f74f2604b108c9b8392d9ce9092d9b5","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-06-26 08:56:26.000000000","message":"Patch Set 3: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit c5ddd19eacb6d6ea1fead663c98d965ee1575af8","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"cecdb6a175eb8fe1bfff1d9321d06b7ba59990b1","revisions":{"1cde20c3bf97e83d819357fb0c9eb399717a4df8":{"kind":"REWORK","_number":1,"created":"2020-06-25 08:30:41.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/21/737921/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/737921/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/737921/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/737921/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/737921/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/737921/1"}}},"commit":{"parents":[{"commit":"475901d9f60cd4d061102286d1970bf9b8dc25cc","subject":"Merge \"Enable ZooKeeper when Storm is enabled\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/475901d9f60cd4d061102286d1970bf9b8dc25cc"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-06-25 08:24:22.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-06-25 08:30:38.000000000","tz":120},"subject":"openvswitch: Use ansible_hostname for system-id","message":"openvswitch: Use ansible_hostname for system-id\n\nCurrently openvswitch sets system-id based on inventory_hostname, but when\nAnsible inventory contains ip addresses - then it will only take first ip\noctet - resulting in multiple OVN chassis being named i.e. \"10\".\nThen Neutron and OVN have problems functioning, because a chassis named \"10\"\nwill be created and deleted multiple times per second - this ends up in\novsdb and neutron-server processes using up to 100% CPU.\n\nChange-Id: Id22eb3e74867230da02543abd93234a5fb12b31d\nCloses-Bug: #1884734\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/1cde20c3bf97e83d819357fb0c9eb399717a4df8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/1cde20c3bf97e83d819357fb0c9eb399717a4df8"}]},"branch":"refs/heads/master"},"ae9cdf6a8b5a35c041d504a9415feb55089e9048":{"kind":"REWORK","_number":2,"created":"2020-06-25 08:55:44.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/21/737921/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/737921/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/737921/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/737921/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/737921/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/737921/2"}}},"commit":{"parents":[{"commit":"475901d9f60cd4d061102286d1970bf9b8dc25cc","subject":"Merge \"Enable ZooKeeper when Storm is enabled\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/475901d9f60cd4d061102286d1970bf9b8dc25cc"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-06-25 08:24:22.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-06-25 08:55:21.000000000","tz":120},"subject":"openvswitch: Use ansible_hostname for system-id","message":"openvswitch: Use ansible_hostname for system-id\n\nCurrently openvswitch sets system-id based on inventory_hostname, but when\nAnsible inventory contains ip addresses - then it will only take first ip\noctet - resulting in multiple OVN chassis being named i.e. \"10\".\nThen Neutron and OVN have problems functioning, because a chassis named \"10\"\nwill be created and deleted multiple times per second - this ends up in\novsdb and neutron-server processes using up to 100% CPU.\n\nAdding openvswitch role to ovn CI job triggers.\n\nChange-Id: Id22eb3e74867230da02543abd93234a5fb12b31d\nCloses-Bug: #1884734\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/ae9cdf6a8b5a35c041d504a9415feb55089e9048"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/ae9cdf6a8b5a35c041d504a9415feb55089e9048"}]},"branch":"refs/heads/master"},"cecdb6a175eb8fe1bfff1d9321d06b7ba59990b1":{"kind":"REWORK","_number":3,"created":"2020-06-25 12:20:20.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/21/737921/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/737921/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/737921/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/737921/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/737921/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/737921/3"}}},"commit":{"parents":[{"commit":"475901d9f60cd4d061102286d1970bf9b8dc25cc","subject":"Merge \"Enable ZooKeeper when Storm is enabled\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/475901d9f60cd4d061102286d1970bf9b8dc25cc"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-06-25 08:24:22.000000000","tz":120},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-06-25 12:20:08.000000000","tz":120},"subject":"openvswitch: Use ansible_hostname for system-id","message":"openvswitch: Use ansible_hostname for system-id\n\nCurrently openvswitch sets system-id based on inventory_hostname, but when\nAnsible inventory contains ip addresses - then it will only take first ip\noctet - resulting in multiple OVN chassis being named i.e. \"10\".\nThen Neutron and OVN have problems functioning, because a chassis named \"10\"\nwill be created and deleted multiple times per second - this ends up in\novsdb and neutron-server processes using up to 100% CPU.\n\nAdding openvswitch role to ovn CI job triggers.\n\nChange-Id: Id22eb3e74867230da02543abd93234a5fb12b31d\nCloses-Bug: #1884734\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/cecdb6a175eb8fe1bfff1d9321d06b7ba59990b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/cecdb6a175eb8fe1bfff1d9321d06b7ba59990b1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
