)]}'
{"id":"openstack%2Fkolla-ansible~742721","triplet_id":"openstack%2Fkolla-ansible~stable%2Fussuri~I3afd117633a84d342effb6baadf16fa42c16776c","project":"openstack/kolla-ansible","branch":"stable/ussuri","topic":"bug/1888501-stable/ussuri","hashtags":[],"change_id":"I3afd117633a84d342effb6baadf16fa42c16776c","subject":"Update glance configuration file for backend","status":"MERGED","created":"2020-07-23 18:19:17.000000000","updated":"2020-07-24 10:38:13.000000000","submitted":"2020-07-24 10:36:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"742721-1595587001946-78954b85","meta_rev_id":"50b466f6b1313b31ec1e00565556d80987abaa38","_number":742721,"virtual_id_number":742721,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-24 10:36:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2020-07-24 05:48:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2020-07-24 09:18:26.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2020-07-24 09:18:26.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-07-23 18:19:17.000000000","_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-23 18:19:17.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"},{"updated":"2020-07-24 05:48:17.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2020-07-24 09:18:26.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2020-07-24 10:36:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"66f28d1de38fa3fb78b32b56d00b284bc0dadbbc","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-07-23 18:19:17.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"7acb296a84dd3361efeb2f3473fb2abcf6eb73a8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-23 19:40:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/fca04e716bf4401289111d1a392afbc9 : SUCCESS in 1h 15m 12s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/4dc59c9678474909b36f114dae49042a : FAILURE in 39m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0d05df87d9f4f14248faa5e5ad4cd5c7a604744f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-23 20:18:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3556029dfa8d413095a87e3245d2cbfd : SUCCESS in 5m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d0e1e3e910c045e5b6259742020e6e8e : SUCCESS in 8m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ed42643bd1945ff81d573be5aa0b0a3 : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1cf45233e4494816902c9b7c6bcd15a2 : SUCCESS in 8m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/15d378547ef546409e65577b6fa4ded1 : SUCCESS in 5m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8bed88432234a66b60143fd58b24f79 : SUCCESS in 9m 41s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/c2d7e7f9dde34d0eadbeb2d1b1d3d1a6 : SUCCESS in 13m 10s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/0da67b3e2717410b9eb8697800c8217f : SUCCESS in 1h 08m 25s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/722390477f0048909793e4110ad4813b : SUCCESS in 49m 33s (non-voting)\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/db2b6de2c67e4c2393bcc31744dbbf20 : SUCCESS in 59m 35s\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/c7b192385da743c4979b8fff5c706fb8 : SUCCESS in 1h 07m 55s (non-voting)\n- kolla-ansible-centos8-source-swift https://zuul.opendev.org/t/openstack/build/0b7bbef2a2df477bb9d88bda5010d7ac : SUCCESS in 1h 53m 26s (non-voting)\n- kolla-ansible-ubuntu-source-swift https://zuul.opendev.org/t/openstack/build/77f266bb9af34ca382ec43f36310d6c9 : SUCCESS in 1h 10m 30s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/cf1457a4c9b54a94b21f1cc28de30285 : SUCCESS in 1h 15m 02s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/d16ef6cce0a5403caaaa2725e5c03cda : SUCCESS in 1h 05m 03s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/9f4ae2deb2df4ceabc1b9bc977e79887 : SUCCESS in 1h 03m 46s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/f75076ccf7fa485a966859d3b9c0c759 : SUCCESS in 1h 03m 29s (non-voting)\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8f8c24a37efe4f5f9c8418831f36d58c : SUCCESS in 53m 46s (non-voting)\n- kolla-ansible-centos8-source-cells https://zuul.opendev.org/t/openstack/build/d711833cba884f5d911e79c99fdcaa1f : SUCCESS in 1h 29m 00s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/5ace9b794c03433ba2e61cf9365cd84b : SUCCESS in 1h 45m 58s (non-voting)\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/8d6e2d38957747f594cadf86e604bf29 : SUCCESS in 1h 24m 59s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/970d1d9edd1f4c9f931db540ae1ce4c2 : SUCCESS in 1h 48m 42s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/14400f12aacc4eafad84473e7f2cdbdc : SUCCESS in 1h 41m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"23d7736e1e64e71e13f8b650c10e14d233dbc5eb","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-07-24 05:48:17.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b586980608dbae2e6962f52877a4f11792f768ba","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2020-07-24 09:18:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b16b539a5df59002929611fc0b2a3d8bb56551aa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 09:18:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"01b8bca82bff04cc8b1ac01ecd0ad2ba7db186dd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 10:36:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"4c01ee1e230b26f2a491c83ad47dc0570c50ba2f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 10:36:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5bb59155b59f47c2bdd8ea4115d16e4d : SUCCESS in 6m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d320aa0221f74b4d8eacdaed749db215 : SUCCESS in 5m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/48b053dc5e3647f7948182073902f95f : SUCCESS in 6m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/29bd26eb50a44ffaa93b216215251628 : SUCCESS in 6m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d207440ef20409bb998a682f0bb3279 : SUCCESS in 7m 01s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/a799ba702a76428c9ca8b9e3655f52b9 : SUCCESS in 14m 11s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/6a443999bdd6487a87c64c37eec2d0cb : SUCCESS in 1h 10m 36s\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/a88f1b7093de41408dc3f0f2a75a59ce : SUCCESS in 1h 08m 33s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/28cfef6af9084e758a283cc943fc0001 : SUCCESS in 42m 00s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/dab249bbce0444cfaa67302a655efc0d : SUCCESS in 1h 10m 23s","accounts_in_message":[],"_revision_number":1},{"id":"d13d0b6d91eb299d1453aeabb05d811edc478efd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-24 10:38:13.000000000","message":"Patch Set 1:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/053213ce8f6d4b138f8e2f59f558b905 : FAILURE in 1m 02s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3c33cd6a3faaeca2d1ab44992d5c3056c9d4d0fe","revisions":{"3c33cd6a3faaeca2d1ab44992d5c3056c9d4d0fe":{"kind":"REWORK","_number":1,"created":"2020-07-23 18:19:17.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/21/742721/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/742721/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/742721/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/742721/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/742721/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/742721/1"}}},"commit":{"parents":[{"commit":"a0cace81bcd24e9a2d1159f27a74201c1dff2839","subject":"Merge \"fix deploy freezer failed when kolla_dev_mod enabled\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/a0cace81bcd24e9a2d1159f27a74201c1dff2839"}]}],"author":{"name":"likui","email":"likui@yovole.com","date":"2020-07-22 11:08:24.000000000","tz":480},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-07-23 18:19:17.000000000","tz":0},"subject":"Update glance configuration file for backend","message":"Update glance configuration file for backend\n\nglobal file glance_backend_file parameters not take effect\n\nCloses-Bug: #1888501\n\nChange-Id: I3afd117633a84d342effb6baadf16fa42c16776c\n(cherry picked from commit 3d0419969263c3ed18bf9cbc4f2e32451c80fcbe)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/3c33cd6a3faaeca2d1ab44992d5c3056c9d4d0fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/3c33cd6a3faaeca2d1ab44992d5c3056c9d4d0fe"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
