)]}'
{"id":"openstack%2Fkolla-ansible~770621","triplet_id":"openstack%2Fkolla-ansible~master~If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b","project":"openstack/kolla-ansible","branch":"master","topic":"proxysql","attention_set":{},"removed_from_attention_set":{"27339":{"account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"last_update":"2022-08-09 15:42:46.000000000","reason":"\u003cGERRIT_ACCOUNT_27339\u003e replied on the change","reason_account":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}},"30491":{"account":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"last_update":"2022-04-21 15:13:45.000000000","reason":"Radosław Piliszek replied on the change"}},"hashtags":["proxysql"],"change_id":"If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b","subject":"[CI] Test ProxySQL with shards in the nova cells scenario","status":"MERGED","created":"2021-01-13 13:56:00.000000000","updated":"2022-08-10 12:24:55.000000000","submitted":"2022-08-10 12:22:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":15,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"770621-proxysql","meta_rev_id":"ff403f4ad9b9028fe32a612ec8e6f11a312ca541","_number":770621,"virtual_id_number":770621,"owner":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":33493,"name":"chenwei","display_name":"chenwei","email":"jzpcoming@163.com","username":"chenwei"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-10 12:22:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":29759,"name":"zhangmeng","email":"zhangmeng9502@gmail.com","username":"zhangmeng9502"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"recommended":{"_account_id":33493,"name":"chenwei","display_name":"chenwei","email":"jzpcoming@163.com","username":"chenwei"},"all":[{"value":1,"date":"2022-08-09 13:14:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":1,"date":"2022-08-09 10:17:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":33493,"name":"chenwei","display_name":"chenwei","email":"jzpcoming@163.com","username":"chenwei"},{"value":2,"date":"2022-08-09 13:17:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2022-08-09 10:17:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":1,"date":"2022-08-09 10:17:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":2,"date":"2022-08-09 10:17:34.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-08-09 10:17:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":1,"date":"2022-08-09 10:17:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29759,"name":"zhangmeng","email":"zhangmeng9502@gmail.com","username":"zhangmeng9502"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":33493,"name":"chenwei","display_name":"chenwei","email":"jzpcoming@163.com","username":"chenwei"},{"value":1,"date":"2022-08-09 13:17:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":29759,"name":"zhangmeng","email":"zhangmeng9502@gmail.com","username":"zhangmeng9502"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":33493,"name":"chenwei","display_name":"chenwei","email":"jzpcoming@163.com","username":"chenwei"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":29759,"name":"zhangmeng","email":"zhangmeng9502@gmail.com","username":"zhangmeng9502"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"all":[{"value":0,"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"value":0,"_account_id":33493,"name":"chenwei","display_name":"chenwei","email":"jzpcoming@163.com","username":"chenwei"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":1,"date":"2022-08-09 13:12:07.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":29759,"name":"zhangmeng","email":"zhangmeng9502@gmail.com","username":"zhangmeng9502"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22072,"name":"Viktor Křivák","email":"viktor.krivak@ultimum.io","username":"viktor-krivak"}],"REVIEWER":[{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},{"_account_id":29759,"name":"zhangmeng","email":"zhangmeng9502@gmail.com","username":"zhangmeng9502"},{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"_account_id":33493,"name":"chenwei","display_name":"chenwei","email":"jzpcoming@163.com","username":"chenwei"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-13 14:43:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-01-13 20:07:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-01-26 12:06:23.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"CC"},{"updated":"2021-01-26 12:36:28.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"CC"},{"updated":"2021-01-27 11:16:52.000000000","updated_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"reviewer":{"_account_id":22072,"name":"Viktor Křivák","email":"viktor.krivak@ultimum.io","username":"viktor-krivak"},"state":"CC"},{"updated":"2021-02-02 19:00:45.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"},{"updated":"2022-05-13 17:49:22.000000000","updated_by":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"reviewer":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"state":"REVIEWER"},{"updated":"2022-07-13 06:45:42.000000000","updated_by":{"_account_id":29759,"name":"zhangmeng","email":"zhangmeng9502@gmail.com","username":"zhangmeng9502"},"reviewer":{"_account_id":29759,"name":"zhangmeng","email":"zhangmeng9502@gmail.com","username":"zhangmeng9502"},"state":"REVIEWER"},{"updated":"2022-07-13 09:34:33.000000000","updated_by":{"_account_id":33493,"name":"chenwei","display_name":"chenwei","email":"jzpcoming@163.com","username":"chenwei"},"reviewer":{"_account_id":33493,"name":"chenwei","display_name":"chenwei","email":"jzpcoming@163.com","username":"chenwei"},"state":"REVIEWER"},{"updated":"2022-07-24 14:55:37.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"},{"updated":"2022-07-27 12:45:02.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2022-07-29 07:09:37.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"},{"updated":"2022-08-09 13:14:38.000000000","updated_by":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"reviewer":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"state":"REVIEWER"}],"messages":[{"id":"631a92c958c59f3b839e4c8286fea2abb5cdab25","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-01-13 13:56:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d787f516334953f2800a8b2951b716dab1dd66f5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 14:43:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/d9c8a34bd5424e9c99c857e472f2e88a : RETRY_LIMIT in 3m 01s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/ed102b0692314bb1861a99586af84433 : RETRY_LIMIT in 4m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0af50ebf1d98fcbacd940b7ce72d1befc94b10f4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-01-13 15:01:34.000000000","message":"Uploaded patch set 2: New patch set was added with same tree, parent, and commit message as Patch Set 1.","accounts_in_message":[],"_revision_number":2},{"id":"103f38718b581f38611912740120af475ad13f9d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-01-13 15:12:01.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"72aa038ce3e534f218703da715e865a484dba611","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 16:16:32.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/5c441cc5cb614a99ae307840d35b9a43 : RETRY_LIMIT in 2m 31s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/e3462528ae5947f681ac39a37eebb3c9 : RETRY_LIMIT in 2m 25s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"0d9ebf0a76d1b1b2e7e99b049c1ca919d8d957c5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-01-13 18:17:22.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"f63e79e1638dfe103e2d3712927a7a04bc4f4e8e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 20:01:56.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/4f5033b1cf6d4263943ef89a9a283f52 : FAILURE in 1h 06m 22s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/542bb6c5c8a54f88bee0065571bc83af : FAILURE in 1h 38m 32s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"fd1bc13c9347b87543049f8a7754eb046d7d3c23","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-13 20:07:02.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/28219a78c11d4c7f98ca2a73158f1fcb : SUCCESS in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/907eacdf01f34ae09e22c378988ffd8a : SUCCESS in 3m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0e67408f89b94b83849d8ce79b182e08 : SUCCESS in 4m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8fd41dbdd419430386df22e831782dd7 : SUCCESS in 4m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/69bc48e0fad746df97f63d325a97d6e7 : SUCCESS in 7m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c5eb2af178c41f2a75dea136984d58c : SUCCESS in 7m 34s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b29054b9bf2a4dedb738913465f56739 : SUCCESS in 8m 12s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/1586634e5e3d46bba10d135ccca6b091 : FAILURE in 46m 55s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/917e8d5c5ac345aebb23e78e6e44c40a : FAILURE in 25m 46s (non-voting)\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4e21b133f31545bc8f89c09d1b5adf2b : FAILURE in 39m 29s\n- kolla-ansible-centos8-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/3157e690f3c448f49b5c29fc8369923a : FAILURE in 23m 19s (non-voting)\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/8e6d0696e5894440b65b2642cb1f38e4 : FAILURE in 1h 28m 23s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/0231b28f086a4ccda9294f6b1dfbad9f : FAILURE in 1h 01m 21s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/4e228792f1464e46a86c6d66b55049b8 : FAILURE in 1h 23m 19s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/92821c486bb14f5b9f45d4629c51f5b4 : FAILURE in 1h 15m 07s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/3c81488b56cc42b2bd83f41a63d3f876 : FAILURE in 26m 01s (non-voting)\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/41d36603007c44a494544536ee80910e : FAILURE in 27m 42s (non-voting)\n- kolla-ansible-centos8-source-cells https://zuul.opendev.org/t/openstack/build/5ecb22e404ec46da84da18d5c7d3d787 : FAILURE in 28m 34s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/65d8014404aa4a76af8c2a5d2690aae1 : FAILURE in 34m 52s (non-voting)\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/66f0f2f43f5841aa93b0de32c196503c : FAILURE in 35m 50s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/317d20f9a1fe460c89f4f5adacd1b8fa : FAILURE in 1h 19m 52s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/bf3c93ade2114b47a13061f62f285b12 : FAILURE in 1h 09m 19s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"2a019f79c233cfb57d167a74cb2734a24846ad67","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-01-13 22:55:08.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"4ba9fa53181eec8755ee5acfc13e8bce00db8277","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 00:24:59.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a153fec39f7f42d8bde774818969b8ae : FAILURE in 44m 33s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/c02d3bf504094f3bb925745303bc8e4c : FAILURE in 1h 21m 08s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"05b621c8e293fcf0ce65819585913d5c947beea6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-01-14 00:30:33.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"63cf108d09118924916c1e9a29f7d20f9c36ae56","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 02:02:11.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/1d56db61888f4929a49aebbf898a4fb0 : FAILURE in 45m 38s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/f5cbba1fe6b84d7bac8d40c2038aa1b7 : FAILURE in 1h 23m 43s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"338b5705b5502b74445326043eaf81dbdc9ebc17","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 02:17:55.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9b9f76ab26054e4ba0efb1980452f7fe : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13919fa5c9a14649ba97b9b3aafb8ac4 : SUCCESS in 3m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d4e37a92215e484f90d978dce0523560 : SUCCESS in 4m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ebabce9923b64cf0b3484645d9c5ed02 : SUCCESS in 4m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6db6cfe5a7c74404a97812f8a24acae7 : SUCCESS in 7m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56a4f0c4233f440ca4a609c89f189b9c : SUCCESS in 8m 00s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/db6efa8793584ebdb27d122b2cb59b79 : SUCCESS in 7m 02s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/af2321a996dd468186c5d68954a516be : FAILURE in 33m 15s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/c49021d3b6174f69836cc99229c62f1e : FAILURE in 46m 22s (non-voting)\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4d28ff8f10e14f9ab05169c18444d47a : FAILURE in 26m 59s\n- kolla-ansible-centos8-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/8847982a13704445bb2cd8f4ceba3fb3 : FAILURE in 23m 38s (non-voting)\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/997e9ba7e1f5400aa98096c92d8a83ec : FAILURE in 21m 46s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/d565c1e01ac34cc7bfc8c6c5884319e3 : FAILURE in 1h 02m 39s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/e438062ca68c4e32a02d1e2154f0695f : FAILURE in 57m 38s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/693752ea51d24f8b93b681f43e30fd81 : FAILURE in 55m 21s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/f0bb8df105be4b4982a0de154ce88d31 : FAILURE in 34m 38s (non-voting)\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/821e689d0cc240e6830fa1cda6a79cbb : FAILURE in 44m 44s (non-voting)\n- kolla-ansible-centos8-source-cells https://zuul.opendev.org/t/openstack/build/a2edf23d009543f9b138d5ccecdf6063 : FAILURE in 58m 09s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/d4bbaec1e1be4bc9a677496b36ed5a0a : FAILURE in 41m 39s (non-voting)\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/bcce257ef36042feb47ea9cbab51ff36 : FAILURE in 49m 35s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/61abf373e79a44fca96f4875e7a6a145 : FAILURE in 1h 33m 46s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/cc56ae857e0044b5875e995ea248f630 : FAILURE in 1h 21m 23s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"9512590c9478ebae98b5ae1cc4bcbc0448081989","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-01-14 08:30:01.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"e5efa322d930cf0cf380f8faffc14ca91879dfc5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 08:31:55.000000000","message":"Patch Set 6:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":6},{"id":"1e0ef2aa5bbcb19782076cc12e1d7453aad0d897","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 08:31:59.000000000","message":"Patch Set 6:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":6},{"id":"598b285f5cc28ce4047fce704fa905e77cd5a375","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-01-14 08:37:11.000000000","message":"Hashtag added: proxysql","accounts_in_message":[],"_revision_number":6},{"id":"212327fc56361f89c146c544252c2c30fb242f96","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-01-14 09:27:33.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"6a5abf20eecdd7a480183e2cc2ce8b1f0ccebed9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-01-14 10:43:39.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"c373d1a5ce8c8dd444c9aa7414b74dcd82745bbf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-01-14 10:46:10.000000000","message":"Uploaded patch set 8: New patch set was added with same tree, parent, and commit message as Patch Set 7.","accounts_in_message":[],"_revision_number":8},{"id":"79871be06af1a9ff610fb1f38e9b759ce1e56568","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 17:32:48.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/80939375ea3849b48dc66a4d784aeff6 : FAILURE in 1h 09m 23s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"55259e5ee79c0742c4e4c096ca0e2492f9bfd8dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 18:04:49.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/40feacd0264e47fe91e8503d16af753a : SUCCESS in 12m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f504b6eb278d441ead31e46a257e6b22 : SUCCESS in 5m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/514219ef504441d5b4369e75a46955b2 : SUCCESS in 5m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0de6e81372284252a46ac994ed6dca66 : SUCCESS in 4m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ff579bd674dd4db3b4b3fd793b7fc519 : SUCCESS in 6m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/269ec350f6894fc489afd8a11d26d93b : SUCCESS in 6m 01s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/0a24f314aeb142c48c7ea137c5e1e9e7 : SUCCESS in 6m 04s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/4e9f0522bf664171a3d595c934ec8469 : SUCCESS in 1h 13m 38s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/7f4bed89f00740139eec06f6d5b42bfb : SUCCESS in 54m 21s (non-voting)\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/03d955dd63ff42739b5cd145848a9d38 : SUCCESS in 51m 50s\n- kolla-ansible-centos8-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/63823903901d4866bbc58b11fe347d16 : FAILURE in 25m 52s (non-voting)\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/9d3cfe7aaae8498395f67c77c01b4d8b : FAILURE in 20m 44s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/06c2d19e512a43e9b62c26f7b4cdaac4 : SUCCESS in 2h 22m 28s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/92c488e55e0543f38c4f25f24fe7b9f6 : FAILURE in 1h 56m 18s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/952f0bc8dc5c4701a50fdcd31a144775 : FAILURE in 30m 00s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/c351fef4fd514c2f8c8526724001a9ad : SUCCESS in 1h 02m 34s (non-voting)\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/462f7581d32048dfb3b341bf1b9f6c4e : SUCCESS in 57m 30s (non-voting)\n- kolla-ansible-centos8-source-cells https://zuul.opendev.org/t/openstack/build/f58d1c4a324048548fd2ed3c9ca4a15f : FAILURE in 45m 43s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/ecb089fae9414142be59227ec1747a9a : SUCCESS in 1h 34m 18s (non-voting)\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/92fd861ee60649af8749a723a62b50bb : DISK_FULL in 1h 26m 28s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/b2ebb801260d4bcfb9e4e8d45a0709c6 : TIMED_OUT in 2h 35m 56s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/9731006a3f4444a6802286906476da09 : DISK_FULL in 1h 30m 54s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"edf58fe9c42f7f4c1af04c6810bc41111d6ee92d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-01-14 23:20:11.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.","accounts_in_message":[],"_revision_number":9},{"id":"ad160dcea5578ad40d0a3cb447e5bd3df8bdf65d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 23:21:18.000000000","message":"Patch Set 9: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":9},{"id":"90be361aaece04b42c652ad6546f8d4801b39a48","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-14 23:21:43.000000000","message":"Patch Set 9:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":9},{"id":"6f94c552f05cdf105d7b109ed7bac0f408db95ec","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-01-15 01:00:03.000000000","message":"Patch Set 9:\n\nrecheck","accounts_in_message":[],"_revision_number":9},{"id":"67adf728404e119bd1e7268c087538979aa98b8e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 03:00:15.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/1f6baa1afb7f4b8c867716a796ce9986 : FAILURE in 1h 17m 59s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/dcab786286ab41349fb669807b9e1350 : FAILURE in 1h 54m 32s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"8d7a6f75e8e243d51caa929469aa971738e9d9ab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-15 04:13:53.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2833a9bd4ac64dd08fdd421d45277472 : SUCCESS in 4m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9529efef856b4d3f832d1fec9546d4fe : SUCCESS in 3m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1f8ab9b662d848d189ac916241bc5f79 : SUCCESS in 4m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d5d5ec5b2b934d438156031bb98f9529 : SUCCESS in 4m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/54f8e63d2cba4841a9b95d6bed22c6f5 : SUCCESS in 4m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2bc8c51f52434e348ebc204d1eacec90 : SUCCESS in 10m 21s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/32e3f94410be40f1a6299745259d8734 : SUCCESS in 5m 02s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/edce37a39d854f69bf3d8a4f09d3e28b : SUCCESS in 1h 08m 57s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/f3368546279e4f9989446bab4166236a : SUCCESS in 1h 01m 33s (non-voting)\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/1d9926c5458b48b2b9a4ffb388cb564e : SUCCESS in 57m 39s\n- kolla-ansible-centos8-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/68b6257e8fed4b6d9f1be1f67bfb9898 : FAILURE in 24m 48s (non-voting)\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/42c364bf103c4b6e92b7a8e627894ad6 : FAILURE in 21m 53s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/bb5dcd6736f749b6b266a2760e8ea988 : SUCCESS in 1h 31m 47s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/e2646676cab74d619cad3ef239ddaa4a : SUCCESS in 1h 06m 59s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/bc4622119f544048b651bba7358987a3 : SUCCESS in 1h 12m 17s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/efceaa75303848b296900e57342c6e94 : SUCCESS in 1h 50m 15s (non-voting)\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/ee8bdfd3b6f94acc950e55a573f328e2 : SUCCESS in 53m 59s (non-voting)\n- kolla-ansible-centos8-source-cells https://zuul.opendev.org/t/openstack/build/15daec5bfa9a4d28b1a3faa427b6a284 : FAILURE in 52m 57s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/e1278cfe8c3945eda7d147c7fcf00186 : SUCCESS in 1h 49m 18s (non-voting)\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/b475083f0ebc40a590853d6a2ee58b2d : SUCCESS in 1h 28m 34s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/87a281be80974b6f866260d22e7ed224 : SUCCESS in 2h 10m 50s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/9e991bc6c0294803b42c68892441aba2 : DISK_FULL in 1h 30m 17s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"da6eef298138bfb84019664271a98e3f314bfe59","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-01-18 11:53:40.000000000","message":"Uploaded patch set 10: Patch Set 9 was rebased.","accounts_in_message":[],"_revision_number":10},{"id":"4b37664e6cf8e59e8d5249a54e996d95ea0a9f2e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-18 12:47:50.000000000","message":"Patch Set 10:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":10},{"id":"34af247f26943c1bd2244b4050486fcde368bb31","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-18 12:48:28.000000000","message":"Patch Set 10: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":10},{"id":"cd4e198306cac49c0d33b560d40929712a148d8c","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-01-18 12:55:21.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"7fdbbd0894a4fba8bbad89342aa529d03ecb4386","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-18 14:21:58.000000000","message":"Patch Set 10:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":10},{"id":"1583c829e0a0299a62a5f10d7de71907af1e976b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-18 14:22:03.000000000","message":"Patch Set 10:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":10},{"id":"19ecb1a59289824e1959827f2df092935de78970","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-01-18 14:48:09.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"d6a2ba7f259dc11896d8d36985e290e173e3a8e1","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-01-18 15:51:46.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"ebd1366507f88a073b1240f281f95af63c94a2e2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-18 15:52:45.000000000","message":"Patch Set 10:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":10},{"id":"fea66d779f2aa4bb56b2af1e1aaca9a6c1a292e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-18 15:54:32.000000000","message":"Patch Set 10:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":10},{"id":"1dba6fd70064302b18987348bbb8f3fd9a7bd32c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-18 16:26:06.000000000","message":"Patch Set 10:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":10},{"id":"44b44b589532d6d9e79e6ad9f653c3cf17f57201","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-01-18 16:27:51.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"af0426b849770ba296bcfba008c56ec3436a74cb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-18 19:08:04.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/ccafbb1e38254dd38cda4c28a6ca30f7 : FAILURE in 1h 38m 50s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/f5a16f76a12644f1aa180cfec390af6d : FAILURE in 1h 43m 37s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"500dd76b33b5858c7d3e8786e85996c3c00a2181","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-01-18 21:31:28.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"33225a8d172b748c198c2ad4a6c15895cca4b098","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-18 22:34:08.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/fd82b723308f487fa5abd797ee17576e : FAILURE in 55m 25s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/f0a7d9495d494191bb3942fd3aeeb747 : FAILURE in 58m 07s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"d3f64de4327e379c828234b3cc6efa27c3c7f121","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-19 01:04:46.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3797fb02b980404994bca406f51e1409 : SUCCESS in 4m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab1668dd3aeb4c56ad00d4fbe4208ace : SUCCESS in 5m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e58f3a288e1f4c0aa80c2cde178fe865 : SUCCESS in 4m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aeaa163ff36a446695d2d3fe3656556f : SUCCESS in 4m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1d30631febc241bfa27e02f559fd63ff : SUCCESS in 6m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1bfc1433ff24fbfa6179b1094e3de5a : SUCCESS in 7m 45s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/31acb5f38d8e470e83e4e7d4e33839fb : SUCCESS in 6m 16s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/612742611f0440e791f51870b0dde253 : SUCCESS in 1h 10m 59s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/16f48123671b403692d97fa9397bb0ff : SUCCESS in 59m 24s (non-voting)\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/6f6da93d5a5341029850fee0a934c548 : SUCCESS in 58m 33s\n- kolla-ansible-centos8-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/6a7ab948a5a247a5899a74cd4d9fb7d1 : FAILURE in 27m 21s (non-voting)\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/414e0381d34c4f83ace988734f61612c : FAILURE in 20m 05s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/c0cc1bf3a9df4b6db010284319b1bacf : FAILURE in 1h 09m 33s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/7207c23744f04479a78d7848eff55f14 : SUCCESS in 55m 23s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/0efb7ed3f4d64f4ab7c7d4a60af13016 : SUCCESS in 1h 17m 03s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/de061a357d5a48e89816bba97d9f8c6d : SUCCESS in 1h 11m 13s (non-voting)\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/da1ada0a6719464494e493e408a1493b : SUCCESS in 52m 29s (non-voting)\n- kolla-ansible-centos8-source-cells https://zuul.opendev.org/t/openstack/build/fd01cd74a40c4204a374de3fcf8d427d : FAILURE in 44m 19s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/db1b9cfd2ca94319aba083d2e367b46b : SUCCESS in 1h 39m 55s (non-voting)\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/399b1f063ae14f2e86661090788f7bda : DISK_FULL in 1h 12m 18s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/d7124ffd210f4fd8a33679318e4d597a : FAILURE in 1h 51m 52s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/952a55ee0f61468eb261182eeb5507c4 : FAILURE in 2h 08m 13s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"9e347e180e95c37a2ed8440ac1153a185642abee","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-01-19 07:56:31.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"7c503fadca56cd211f2aebf3f6634088fd081ae0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-19 08:44:38.000000000","message":"Patch Set 10:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":10},{"id":"8de3e1ae34890933378ead942bddb2fe557a35a0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-19 08:44:45.000000000","message":"Patch Set 10:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":10},{"id":"ea9f3ba213e23ff28cbd0c3dfe7a3ca162bea687","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-01-19 08:45:15.000000000","message":"Patch Set 10:\n\nrecheck","accounts_in_message":[],"_revision_number":10},{"id":"6693aa3e1cfef88685864f11f9c288264be9c643","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-19 10:00:01.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/1f67f906d59b44e19171301fd0b3803b : FAILURE in 1h 07m 29s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/c0a75b09f8e74a1596dd536a71fc9173 : FAILURE in 1h 08m 33s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"7b5360a076c786ee5d9a9adfbcf334b2b0c69691","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-19 11:24:13.000000000","message":"Patch Set 10:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/15076e337acf4a2cb3e7e1761c1e2cc0 : SUCCESS in 5m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88cb7028b7414ac4bed4a06fec53b060 : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/81221948d70249febd0876b79b395696 : SUCCESS in 4m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ce77e8d29f9f4c5db316b4fb926daa99 : SUCCESS in 4m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/14b98df9ae9849fdbf42a100baea0a40 : SUCCESS in 5m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f2bf066033d45e6874436387233e632 : SUCCESS in 5m 52s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f103299cce4b402e955a23f0ce3ed616 : SUCCESS in 5m 19s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/12db6dda57dc4013b6076ee57e37d30b : SUCCESS in 1h 11m 38s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/86b30ff6b1cc4a0ba8bf10714cde7e56 : SUCCESS in 55m 45s (non-voting)\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/9c9f135b420e4e5c88e4f2d74ef7ad54 : SUCCESS in 55m 57s\n- kolla-ansible-centos8-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/2da6c36cf3404fff83370399e5713126 : FAILURE in 24m 08s (non-voting)\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/d652d03d2f944d20b4c2dbadacf131a7 : FAILURE in 20m 19s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/e66d95963bce4f7ba0e89972e496763f : SUCCESS in 1h 44m 52s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/9146d3c447d2447896c307da6fb94bcc : SUCCESS in 1h 35m 25s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/402f6c092c834d20ae790c33f126e554 : FAILURE in 1h 08m 47s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/b2ffd89475b646a488085289a0f0c736 : SUCCESS in 1h 15m 49s (non-voting)\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/3e6530b83ec2406d8ff6e374f64f54b6 : SUCCESS in 1h 01m 23s (non-voting)\n- kolla-ansible-centos8-source-cells https://zuul.opendev.org/t/openstack/build/369206c846604484b6876efd3567336d : FAILURE in 49m 37s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/933b91227b274f279717738aefeffc88 : SUCCESS in 2h 32m 54s (non-voting)\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/2453ed7150504659ac48f5431533c25e : SUCCESS in 1h 21m 07s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/475985b79b094f59a1639a7bb19021cb : FAILURE in 1h 51m 40s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/00b30a898f074580a36a2def6579b9df : POST_FAILURE in 1h 32m 30s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"203257839e1f53587fc90e52f45e7b1a2a2fbbc6","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-01-19 13:47:31.000000000","message":"Patch Set 10:\n\nIt seems this will not work until all patches in chain will be merged.","accounts_in_message":[],"_revision_number":10},{"id":"5a2f475de66c4f873658c4378bce1b222946d401","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-01-20 00:33:51.000000000","message":"Uploaded patch set 11: Patch Set 10 was rebased.","accounts_in_message":[],"_revision_number":11},{"id":"6ae62e178ebf4737b6714388920ee93e7a48e8fa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-20 02:03:14.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/adf021bac3d745c89ff8fdd5ea904077 : FAILURE in 1h 17m 46s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/57da57d49b8d4b478c2693c02b931b13 : FAILURE in 1h 19m 19s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"7dcc1b02dae358cb072de1c3d2a78fc91ea8271f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-20 03:42:06.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/89706adb936f42de8fea963b66d7e373 : SUCCESS in 5m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d6cf5185899d4fb2b23a0c381d94dbe1 : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bb9999a2af204436b5d0efb93dc480f3 : SUCCESS in 5m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b1e47bb846344411bf7c56d6d8df7d9a : SUCCESS in 4m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dcde0ecda89e4a5d8544736658115412 : SUCCESS in 5m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de7856f458d94660823fe8b11e292ea7 : SUCCESS in 5m 49s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/39fccbbf9fc946889fe438a70093bb84 : SUCCESS in 4m 59s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/d1b244a0053d4f1aae12f00bdd50c95e : SUCCESS in 1h 11m 25s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/d219b318575d4c8cb90a0b15fb5f7d29 : SUCCESS in 51m 05s (non-voting)\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/85c00f43070c4e17ba8e070ff68311c7 : SUCCESS in 1h 31m 21s\n- kolla-ansible-centos8-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/b987642f2283438fb4679ecc690a77aa : FAILURE in 26m 31s (non-voting)\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/2e42a384796947fab1d4991c491d6279 : FAILURE in 21m 09s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/e25f2b8e55464c828fcbfc49e5b79d54 : SUCCESS in 1h 06m 58s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/a470ea5e13564b79af157d2f2cde24c3 : SUCCESS in 1h 11m 25s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d78cae19f5224abcaf0fc70f1b2a2f74 : SUCCESS in 1h 10m 25s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/67ffd05ac72042db8c42d95a69e738a4 : SUCCESS in 1h 08m 21s (non-voting)\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a62fa510d53a459cbf523987c1522d19 : SUCCESS in 59m 54s (non-voting)\n- kolla-ansible-centos8-source-cells https://zuul.opendev.org/t/openstack/build/9323c780323a46018912edd6e8d858f8 : FAILURE in 46m 13s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/1aa9c0de641641db9c17722d7b4a4ed5 : SUCCESS in 1h 42m 59s (non-voting)\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/2a856564816349df8e99a7040560842a : DISK_FULL in 1h 33m 55s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/cf66f99808aa41f7992b96be297e7651 : FAILURE in 40m 21s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/6e0c0f353d2041c994780e1cc3377783 : DISK_FULL in 1h 33m 02s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"2683d95f922493b30d3aaf702f617e70101a0c9b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-01-20 09:09:43.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.","accounts_in_message":[],"_revision_number":12},{"id":"c1168646c7c7f44d418918bcc99e77bed8804802","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-20 10:41:50.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/9c874f00b7b54bea98d721cb56db8e92 : FAILURE in 1h 22m 46s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/cd5a16d3c1dd4cd9a4d22c0c63ccb5b6 : FAILURE in 1h 25m 05s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"adddd42d851de020145f1f264520da503b3452c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-20 11:29:43.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6216320a618842e0bcb5b969bd9115ad : SUCCESS in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/772d914423f5461a9ff9d0ae4a521473 : SUCCESS in 3m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cf1f3655461741d89bc816570cbd9f2b : SUCCESS in 5m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ef67677277714d938e94fbf28c5baa34 : SUCCESS in 4m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8826318bef24441683eee5aeea196442 : SUCCESS in 5m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d88903350c343aebed531b933dfa171 : SUCCESS in 7m 46s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/ac5df2315ceb4774ab61410e8d16d49c : SUCCESS in 6m 35s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/25592b91f5c14648bbfcfea6f9f5b53b : SUCCESS in 1h 10m 51s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/2fd6cb02186e4cce9e73156f044ba48c : SUCCESS in 1h 10m 17s (non-voting)\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/baa9208c2c2c44629d8e871e731f74c0 : SUCCESS in 1h 03m 18s\n- kolla-ansible-centos8-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/ab5c0c95db1646799819489634429820 : FAILURE in 24m 49s (non-voting)\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/bf0acb56cff04a51ad80a9094583d519 : FAILURE in 22m 00s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/618cf48d52a94f13b04a5bfb8ba8a37f : SUCCESS in 1h 10m 54s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/cf361482c1744140bac5397b324c7b95 : SUCCESS in 1h 03m 01s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/aed4a2460e5b484f8650acd8cb8b3a99 : SUCCESS in 1h 20m 27s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/28d81773240849998232038e3a0b6c48 : SUCCESS in 1h 14m 07s (non-voting)\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/05d4970ef0b243049935f85912840446 : SUCCESS in 55m 33s (non-voting)\n- kolla-ansible-centos8-source-cells https://zuul.opendev.org/t/openstack/build/06c663360e344eee9e31e89597ac19af : FAILURE in 47m 46s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/db35dd748b4f4b69813356e4ef3bac67 : SUCCESS in 1h 31m 21s (non-voting)\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/7bde14dde6be4a81bbc455e584e49e93 : DISK_FULL in 1h 09m 54s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/8ea9d80f495c498b9fe49421b546297f : POST_FAILURE in 53m 14s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/7e46d81e5e8241399ed30706f0b97dab : POST_FAILURE in 1h 33m 39s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"998ec0e94c5a9f80f8389113b93b2a787acc1d9f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-01-26 12:06:23.000000000","message":"Patch Set 12:\n\n\u003e Patch Set 10:\n\u003e \n\u003e It seems this will not work until all patches in chain will be merged.\n\nIt should do if you set up the dependencies.","accounts_in_message":[],"_revision_number":12},{"id":"baeb2f5403ea49437d06809c443137af9016fc8e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-01-26 12:06:53.000000000","message":"Patch Set 12:\n\nI would suggest trying to use multiple shards in the cells job.","accounts_in_message":[],"_revision_number":12},{"id":"5debb0fb5db560650c6aab49c7d2d8181de87589","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-01-26 12:36:28.000000000","message":"Patch Set 12:\n\n\u003e Patch Set 12:\n\u003e \n\u003e I would suggest trying to use multiple shards in the cells job.\n\n++, I can help with that. 😊","accounts_in_message":[],"_revision_number":12},{"id":"e7c55f4faa7a95c020e76b9bb06566562b63fcea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-01-26 20:07:45.000000000","message":"Patch Set 13: Patch Set 12 was rebased","accounts_in_message":[],"_revision_number":13},{"id":"89b3bcf2a247e3edc5ebb0e59cf7c81a2ccde24d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 21:15:50.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/2cd7acf77ee64c868cfea6b6c1836c85 : FAILURE in 46m 48s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/2d4163e4eb4d4bec86f1090998c10870 : FAILURE in 1h 02m 26s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"21a42f323e0ec9b735a97619c09151408d9f7bf9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 23:22:54.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0640fda44c1b46178e718deef9f6948b : SUCCESS in 5m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50d3d7d014634b9584568d70ce5e9040 : SUCCESS in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9ca7c333f6424cbdb520c65af7a301ab : SUCCESS in 5m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6b8293fa763b48e5b7dbb7004621b291 : SUCCESS in 5m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2f689b2339914b4196fe772cd53253d9 : SUCCESS in 5m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b83e665ba2041daba057c77ef40dd29 : SUCCESS in 7m 09s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/5138602a675a4e1a8e2e7048c27251cd : SUCCESS in 4m 24s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/93341c94a1c94740a28abc6cf674d048 : FAILURE in 29m 59s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/f2066c3aaa7e439b80dd64c45f93eba2 : FAILURE in 40m 35s (non-voting)\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c8274a497a134203a48c8ad1eeed9d48 : FAILURE in 24m 24s\n- kolla-ansible-centos8-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/b86a26206f37424b854649ee83d37283 : FAILURE in 37m 27s (non-voting)\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/d63717bc08f543f0af983d0a5f71e034 : FAILURE in 21m 13s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/95c09b8eecdf4ef39fe65d40fa63b6a7 : FAILURE in 59m 07s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/79ecc772806d4f7b829d78ddef975fc0 : FAILURE in 51m 52s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d51c9c2c12b14222bbcc0d64619a8590 : TIMED_OUT in 2h 34m 09s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/caf4380a9ae84feb9145d733c0f7a509 : TIMED_OUT in 2h 02m 22s (non-voting)\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/bdb10ee26ed14c0e908a54f35ccec9f9 : FAILURE in 24m 21s (non-voting)\n- kolla-ansible-centos8-source-cells https://zuul.opendev.org/t/openstack/build/50cec07688d24885a759e7eaae4d53f4 : FAILURE in 47m 46s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/e7eee20d1c4e49ff85d82fa9daf71d62 : FAILURE in 38m 51s (non-voting)\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/ab26cd429492403c9d0d2ff868e98d6c : FAILURE in 36m 09s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/9b3ab4acd6db46dca19e2a3d59647808 : FAILURE in 1h 28m 41s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/84db9e67ae2e4218a94b83da2373cfec : DISK_FULL in 2h 36m 48s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"630e66e2f08ad685b9ee28c2448e496c1acbc999","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-01-27 07:25:35.000000000","message":"Patch Set 14: Patch Set 13 was rebased","accounts_in_message":[],"_revision_number":14},{"id":"e4c89386898183bb5d6298d010c9ae13899bf2c2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-27 08:48:45.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/6ea82ee364ba4d219b7abdc3dad6da7b : FAILURE in 50m 14s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/1444910d917d425ba4db1c21d426ce52 : FAILURE in 1h 13m 04s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"b9190bc53dc729fe7f76825e383ef0b04d42678a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-27 09:00:44.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a23f0afefa504446a4a0f384dcd6c330 : SUCCESS in 5m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f1c75499ade4ed38ac53d1a58a49378 : SUCCESS in 3m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6968780f53c946a4ba04538da5d3848b : SUCCESS in 4m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c3999d7eedfa47098eefcb13799a4d70 : SUCCESS in 4m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/61b9c0e6f4834e4eb5ef02d9cf602ea4 : SUCCESS in 4m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/663fbde6c69e4572bfe590c898984022 : SUCCESS in 5m 39s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/8c3304e352974968801cb38f0303127f : SUCCESS in 5m 18s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/484bc40ddd2a4f908e103d6f324ab509 : FAILURE in 27m 27s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/82ed782287e74d2dbd351b6b9f699ca3 : FAILURE in 26m 50s (non-voting)\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/0c1d925d15a2437f89a99bc338a68092 : FAILURE in 23m 15s\n- kolla-ansible-centos8-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/b04208f25a7f43488b58dcba1d260d65 : FAILURE in 23m 58s (non-voting)\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/064b1dea4f864acd9da3fb819370089a : FAILURE in 20m 45s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/914a92e22ee44c8e9044f77be2db339c : FAILURE in 57m 11s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/3bca259a5ff74862bf3b3d6ef4323e88 : FAILURE in 47m 03s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/7d68d5ace5b74f20971a58c3f2371269 : FAILURE in 44m 21s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/691ffe43b60345258229c169b8e95829 : FAILURE in 33m 54s (non-voting)\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2dc8cf068e0d47e3bf0bfad2764f637e : FAILURE in 24m 24s (non-voting)\n- kolla-ansible-centos8-source-cells https://zuul.opendev.org/t/openstack/build/b83e7670af8344e89334d710e3f197db : FAILURE in 32m 38s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/f79cb1c495904b8985059424fe38a777 : FAILURE in 41m 20s (non-voting)\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/76714989e9e447c2abf4b0f69ae6fc4d : FAILURE in 36m 22s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/2a102a6441fc4548886e35bd25393c1e : FAILURE in 1h 30m 40s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/d6becb29f1744ef099fbc3ebafa6a84f : DISK_FULL in 1h 19m 45s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"39c60eb14f22cbb4bb6da6831f7dde46751f549c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-01-27 11:36:48.000000000","message":"Uploaded patch set 15: Patch Set 14 was rebased.","accounts_in_message":[],"_revision_number":15},{"id":"2116169ac19a4ef1f386c4e454bda629d8678604","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-27 13:07:03.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/6104cd2c2dd34b29b9f4f146fb985e1b : SUCCESS in 1h 29m 15s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/d00d0507ec984ef4add18ac7d8201bfa : FAILURE in 56m 46s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"d24c84632b37e2ef72e0deb9a9834ef078a16cf3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-27 14:44:17.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aa98a6544b1b43fc85a2aa61eb770bbf : SUCCESS in 3m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a78b1c8785e42819632a55fb98a68f1 : SUCCESS in 5m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/df6394e8cc6840a6bf1f6ed84b48a84a : SUCCESS in 5m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e79c7711d83d43698d5bd762465f4e03 : SUCCESS in 4m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ed97509422d84a59a1e75cec6e29b671 : SUCCESS in 4m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2055cf71f27147c099a59628027bfab8 : SUCCESS in 6m 22s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b64aa66cf49f49f08e31a88b967a03ee : SUCCESS in 4m 49s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/3c536d377bb14a07ad608893ca976a11 : SUCCESS in 59m 37s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/6adeee032cf94b098012290600b03325 : SUCCESS in 51m 41s (non-voting)\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/9534f2b28883458796ec64ca99677b87 : SUCCESS in 50m 04s\n- kolla-ansible-centos8-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/40200d95b8544333b356d4f4e5a869c2 : FAILURE in 25m 21s (non-voting)\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/8d75375582b24b5ea3bd5ca5b9bfb1a2 : FAILURE in 19m 25s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/460822fbdfa14f77893af8d27ace3a4e : SUCCESS in 1h 09m 23s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/c59201a4ebb3470abb0b49cbe150ebea : SUCCESS in 1h 09m 45s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f5564c95f3f944969ac88eea6d469227 : SUCCESS in 1h 24m 00s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/a7e2b9db39ed48b784f62bc31587c942 : SUCCESS in 1h 14m 19s (non-voting)\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c97418e7a93c4879a0f2e00679db5815 : SUCCESS in 1h 02m 18s (non-voting)\n- kolla-ansible-centos8-source-cells https://zuul.opendev.org/t/openstack/build/c0c48f119b7d4670b4a4642419164ea4 : SUCCESS in 1h 29m 23s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/b5ba0ae3c4e9453cb4f0476b3edefb12 : SUCCESS in 1h 25m 21s (non-voting)\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/54a2979a161b442aa96a3f5c3dfe3e88 : SUCCESS in 1h 37m 42s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/9e01612634444f9485405d7d19a8c263 : FAILURE in 1h 58m 23s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/0d7679816a4e4cb8b98c2e2cb64794db : DISK_FULL in 1h 40m 36s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"3217c4934bd0096c9a39fcebfb0cb176c366ba07","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-02-01 12:13:25.000000000","message":"Uploaded patch set 16: Patch Set 15 was rebased.","accounts_in_message":[],"_revision_number":16},{"id":"6dcb858c63b1d654188078bb9b76cc7680caebc0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-01 13:55:51.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/77762a47c7444aef956cd8c4b15b803d : FAILURE in 1h 07m 10s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/e8f9010eaab643a39f33e98d2c611488 : FAILURE in 1h 28m 04s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"5ebf17b0cbee991e7949c140bc9acd1fa2470d25","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-01 15:49:50.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/acfdb17fb27340c4a3b0c1b72914dbd7 : SUCCESS in 4m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cfc91bed6b2944bf858203c6d75084d2 : SUCCESS in 4m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c4acc4d15dd749c7bc2610c26b3c720d : SUCCESS in 3m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a18ff5e8e77143a89eed969fe95bdf02 : SUCCESS in 3m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ca1399f4840c43239aba2f8ee0de6d26 : SUCCESS in 4m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba0ef028241542588266f3d6941cd107 : SUCCESS in 7m 42s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/dddcf71193d64ac0a9a00887e8f7e470 : SUCCESS in 6m 02s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/aadac81e6ea4450d9d60b04912d692e7 : FAILURE in 24m 31s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/d305a783348045d7aeeb19a8ec12717e : FAILURE in 23m 32s (non-voting)\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/92068c0be67c47de957af16dc299abed : FAILURE in 20m 33s\n- kolla-ansible-centos8-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/d1f8106304844bf589c7f7cc4403edd7 : FAILURE in 25m 19s (non-voting)\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/7d249041e8ef4d4c801890070400c1dc : FAILURE in 25m 22s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/df4d737bee5342ffa40823cd1a1610d2 : FAILURE in 56m 08s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/f8fe584053524ea2b5cea56a7768d47d : FAILURE in 1h 00m 16s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/804d094afc674094859ed4e125ff5453 : FAILURE in 54m 06s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/831f52ad66f0416898aa7798ecefbf81 : FAILURE in 30m 12s (non-voting)\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/945082024d95470cadb120e2c66d3130 : FAILURE in 22m 00s (non-voting)\n- kolla-ansible-centos8-source-cells https://zuul.opendev.org/t/openstack/build/536254ab53644898b05e3d484a88fc06 : FAILURE in 27m 02s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/8e8e31fd86924ac394c815b4c2a015b1 : FAILURE in 43m 17s (non-voting)\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/c047325afdb34597b2a00859414b4135 : FAILURE in 37m 05s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/a64c0748001446af88512a2238179f8b : FAILURE in 1h 25m 09s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/7686a657a4934a1e82d1be8faf9f47aa : FAILURE in 1h 15m 34s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"11dd5c08019a726df1082c8274dec33e3a5f9d6b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-02-01 17:19:28.000000000","message":"Uploaded patch set 17: Patch Set 16 was rebased.","accounts_in_message":[],"_revision_number":17},{"id":"2851a440e49876ebbacfb8efe78eccfe72453764","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-02-01 17:34:43.000000000","message":"Uploaded patch set 18: Patch Set 17 was rebased.","accounts_in_message":[],"_revision_number":18},{"id":"a48b705766dd960e20958ccd2d10482aa2d7a76f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-01 18:50:20.000000000","message":"Patch Set 18:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/9c0eb02288d44ec0adfd8581fccc21fc : FAILURE in 1h 05m 12s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/dfa08d04d01544048e39eccee374a013 : FAILURE in 1h 07m 34s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"3a78e39ce5875377e69d2f25eeeee418167861f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-01 21:46:24.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dfff7c3bbef541f6a94a58de6cf9aafe : SUCCESS in 4m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e6cea71446747dcb18a8da64f6a676c : SUCCESS in 4m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bed149665d494e7b80a0711eb8237d8d : SUCCESS in 5m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9ccb9b51ce4d4d258024f4856d8104b1 : SUCCESS in 4m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/93c43c1ac3e042ce8b8081bf97067bd7 : SUCCESS in 4m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a27aceaeb189472da325a3e17b706e62 : SUCCESS in 6m 19s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/619af25902c34b04873c29015f55c051 : SUCCESS in 5m 44s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/8e4a13819fe54daf8d223e2c9dda1f58 : SUCCESS in 1h 03m 50s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/b98a6055c9da408f919995bf3f28f07f : FAILURE in 34m 35s (non-voting)\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4a832b929f3044bbb7b3a51f6cc46df3 : SUCCESS in 55m 54s\n- kolla-ansible-centos8-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/60823376800f44f682f901004c25b030 : FAILURE in 24m 23s (non-voting)\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/7222fb818fdd4212abcbd8ffc4ebda81 : FAILURE in 20m 33s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/998ed69e7d87431a9e404811f78b6772 : SUCCESS in 1h 05m 33s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/12dbeba4493640b3b354d7c96b3a92af : FAILURE in 58m 31s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/1cd201ae976442ab946b59023b86d70c : SUCCESS in 1h 00m 32s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/91ed318c9f654e44be63957437ae812c : SUCCESS in 1h 17m 47s (non-voting)\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c75bd8fd659c48d98a366573c43f59fc : SUCCESS in 56m 20s (non-voting)\n- kolla-ansible-centos8-source-cells https://zuul.opendev.org/t/openstack/build/df01d581860f4ec3b6d00561c68056df : SUCCESS in 1h 16m 46s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/5059355ec19042119e88fe02f119fda9 : SUCCESS in 1h 31m 20s (non-voting)\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/eccfa461814541d9b8e60e0bea78c23b : SUCCESS in 1h 22m 43s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/f3acd7727672412fa4e8a17ca2756f1a : FAILURE in 1h 55m 18s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/9fd4d09529ec45d693fbea7a0b592238 : DISK_FULL in 2h 04m 40s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"b14446d54d7c91506a82507d0d3b05d55a15bc67","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-02-02 19:00:45.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"a1d6cf62dc5e43b749291ee726aab2189dbe2c58","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-02 20:29:03.000000000","message":"Patch Set 19:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/388d0bcbea7044008f555a0c189ec240 : FAILURE in 1h 12m 24s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/e395f0b46d74453eada5b175047e5d15 : FAILURE in 1h 14m 37s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"ce36840fdfc6e289f20dc456f52ae1b45f9b0b27","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-03 00:25:11.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c607e203f2e14c8ebdd6f1703e505e1e : SUCCESS in 5m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94280dcbe1124b1c929f95c09186eb9f : SUCCESS in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/80b864365e00415c9cce48fc0b6f387f : SUCCESS in 5m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d33842e079a8456397032e874bca0d07 : SUCCESS in 5m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8411288692764c14bcfe9756ab01ff1c : SUCCESS in 5m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3170dc9a50db4e8a80159eed59b7abdf : SUCCESS in 6m 22s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/c0ef81f360724b319d2142ca62b4a991 : SUCCESS in 5m 34s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/33128ba0a5c5416d8314a18a8b843023 : SUCCESS in 1h 11m 07s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/7e439fb8375147f0a93dd44806a87810 : FAILURE in 35m 10s (non-voting)\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/792a81bf801f4465a1745c7278fca7c5 : SUCCESS in 51m 44s\n- kolla-ansible-centos8-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/5a6dd9b4596c4ab09488957a159d0454 : FAILURE in 31m 35s (non-voting)\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/d2e42142964044d282cdf5efb91b1481 : FAILURE in 19m 27s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/e7fe8fba3ce541fd91ec9f4ed9de87a8 : SUCCESS in 1h 15m 44s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/21cb30e8e0d647b2a22303b0397d4a5f : FAILURE in 1h 28m 30s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/9d2ce300b6c64e9fb3fc77a4e5e91a41 : SUCCESS in 1h 04m 50s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/afa4b8849b464319bf4e22893485623c : SUCCESS in 1h 03m 11s (non-voting)\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/90108b29df124d7f84228ca2b6e1c5e3 : SUCCESS in 44m 54s (non-voting)\n- kolla-ansible-centos8-source-cells https://zuul.opendev.org/t/openstack/build/d933f6d9541044d4b31a4846f09b822f : SUCCESS in 1h 11m 19s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/eedc6a65684644eeb25fef50fccb9384 : SUCCESS in 1h 20m 43s (non-voting)\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/b8dd30fdc4da443eb3cc043f1cee9dbe : FAILURE in 49m 53s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/95c17e2178684403bbb8549e9da09ad2 : SUCCESS in 2h 11m 04s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/107aa9522eb54c50866ba7b215a14212 : DISK_FULL in 1h 25m 59s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"ae1eb29aa42f02a17eec34b5cacdf66e02bcf9e1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-02-09 11:19:31.000000000","message":"Uploaded patch set 20: Patch Set 19 was rebased.","accounts_in_message":[],"_revision_number":20},{"id":"107d676a639a56ee6d5905b99fd1bd2ceb49d1d9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-02-09 12:00:28.000000000","message":"Uploaded patch set 21: Patch Set 20 was rebased.","accounts_in_message":[],"_revision_number":21},{"id":"b94eb32b48ed0fb5c4f439a87410b8b752a60d31","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-09 13:39:48.000000000","message":"Patch Set 21:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/0da01bd0db1845c3b5aea083a0960188 : FAILURE in 6m 17s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/ad7ffb0a369f4defac3c258491b0fb17 : FAILURE in 1h 27m 41s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"d896b3d3c4def0fe1272aa9b3ec68ad591aeb530","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-09 13:43:34.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7cb0720ec44f4323b7158fecba7efef1 : SUCCESS in 5m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c80eaee47b2d464994fca367ca482fc5 : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ba84d7c4ea084348af928be3ec57a808 : SUCCESS in 4m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1f484b85b71e49ed903139515654ec4c : SUCCESS in 4m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fa8e5ffc727344b2a8c2685ddaf0dbb2 : SUCCESS in 4m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd4a0e9ad2e342d38578283b88d16633 : SUCCESS in 6m 17s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/9df022459dc14c27847e485748c1594a : SUCCESS in 5m 24s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/9a6791e4eb534c38a2d67a922d3710ad : FAILURE in 4m 45s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/cd206cecd639444cb482f31a892929a8 : FAILURE in 5m 43s (non-voting)\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a2fb43f3cf554f9f80c02de10fe0c500 : SUCCESS in 58m 35s\n- kolla-ansible-centos8-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/7653e6245a0b43b49e70cebbbdf51448 : FAILURE in 4m 52s (non-voting)\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/81c8b6fc04c44ce0b7219be471142169 : FAILURE in 23m 08s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/e5f32486656c49e6847a6cc53515507b : FAILURE in 52m 49s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/fad5e7ced69d43fa98b51f8b32a78132 : FAILURE in 46m 56s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5c952b94e88845d3a065a8deda7212b8 : FAILURE in 19m 32s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/ef65b3b06ab3407cbd7e69fd6e5d32b4 : FAILURE in 4m 33s (non-voting)\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1e82e1a7672149e2a96941e90a7c4c9f : SUCCESS in 56m 09s (non-voting)\n- kolla-ansible-centos8-source-cells https://zuul.opendev.org/t/openstack/build/2c546f7c5205446f9e3c54d6f1347a23 : FAILURE in 5m 56s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/2080ab23b28c47feae84bc80deb3172a : FAILURE in 7m 09s (non-voting)\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/0eaa0dc7eb3042e291e5b746d23384f4 : SUCCESS in 1h 18m 27s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/82fd4f758b39476487fb65213a905c4e : FAILURE in 20m 04s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/e3f8903c75b94bf5a29afc192835af04 : POST_FAILURE in 1h 21m 21s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"7883389bc15e55a6a961ccccf393951f0ed6077f","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-02-10 10:22:57.000000000","message":"Patch Set 21:\n\nrecheck","accounts_in_message":[],"_revision_number":21},{"id":"05964ca1a58c173d20cf7705b88c724dd442f758","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-10 11:34:49.000000000","message":"Patch Set 21:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/fede22ddb26c4008b4969ecaecf1cf3b : FAILURE in 7m 04s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/7910f994011f4fd7a4e2ea471a517459 : FAILURE in 59m 04s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"4799bd06944b54d3f42ed01fe3d86b23b4635e16","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-02-10 11:49:01.000000000","message":"Uploaded patch set 22: Patch Set 21 was rebased.","accounts_in_message":[],"_revision_number":22},{"id":"3d9a4fdba2bbedeffdf13fce689f08272cbdf7ce","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-10 12:08:16.000000000","message":"Patch Set 22:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/56c2297fd9ab4518b4384114b9f7081c : FAILURE in 6m 59s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/827722b9185e49abbfecf4e5b437b1fa : FAILURE in 5m 33s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"0dde31bd8c0546bd636500303d9a9f9d9543e7b8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-02-10 12:32:00.000000000","message":"Uploaded patch set 23: Patch Set 22 was rebased.","accounts_in_message":[],"_revision_number":23},{"id":"8c17e22a0e9a6847ec00e723773548919b9d4a48","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-10 12:44:21.000000000","message":"Patch Set 23: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/81ae76ff659246cc97608e17a0ad2439 : SUCCESS in 4m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93773458fd564683b1d4ed5dcdf409f4 : SUCCESS in 4m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d2ceb4543efc423aae977392efd79773 : SUCCESS in 5m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/08d8e630c2fc4ad99dddf7b16c0abec4 : SUCCESS in 4m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/82c3acce50a74877819b7630f2382578 : SUCCESS in 5m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/453eb175abee4dd98876152290fe9db4 : SUCCESS in 6m 32s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/2fc09f7b3ba14339a5eb62c496b322b4 : SUCCESS in 6m 42s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/2900990ea4ab4a74845d1fa044629147 : FAILURE in 5m 24s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/7e1b394cc27a492192f236107d80cbfa : FAILURE in 4m 37s (non-voting)\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/71894ca1bd7749068bcea7f1959a08ba : FAILURE in 3m 15s","accounts_in_message":[],"_revision_number":23},{"id":"6fb239af3645d605a61c1056b9930dda22226181","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-02-11 13:41:59.000000000","message":"Uploaded patch set 24: Patch Set 23 was rebased.","accounts_in_message":[],"_revision_number":24},{"id":"a4c31a94279c2a15cf7f2c3ddcf7e7ce61337bc7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-11 15:28:49.000000000","message":"Patch Set 24:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/5d1448998cce410fb5ca5632e1fec441 : SUCCESS in 1h 31m 59s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/db7059b1a7464132b7f35154ec80f892 : POST_FAILURE in 1h 22m 58s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"ba56436754392aa746107af6a218670431951439","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-11 17:13:56.000000000","message":"Patch Set 24: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/23dfe076fe254d6794b2cd413401bb23 : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d09c29006e64427ebb1ea08d1261916e : SUCCESS in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/89017206b9b3443d8ca0e17f5f7626a2 : SUCCESS in 4m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/db37ce757846412c94eb0d349dd110c8 : SUCCESS in 4m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/88c337609422468bac7176c13f96bc9b : SUCCESS in 4m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f400e7ac32c4feaa44742fe91770eec : SUCCESS in 6m 36s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/36ca60d132934c70af0507282f47e5ad : SUCCESS in 5m 50s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/dc54075b6e3749d7a327ebd14febf48c : SUCCESS in 59m 39s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/246b1da364544b80a1db779b4d411da8 : FAILURE in 34m 24s (non-voting)\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ef323508635a48d78d272c37ec2ff53f : SUCCESS in 48m 08s\n- kolla-ansible-centos8-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/ecf01a8bf2e44e2ebae1d958b734cce6 : FAILURE in 27m 58s (non-voting)\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/ee637c426241403faf84999e5ebdad1c : FAILURE in 20m 37s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/85e1a5498d8148559bf8ceee01ef66b2 : SUCCESS in 1h 41m 06s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/1a0ab07a708a4c23be82d2faec077cea : FAILURE in 1h 01m 49s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/af7c40c7fadf4e06bbd50da0f1c46578 : SUCCESS in 1h 08m 15s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/eb28a8e674774c958a1f3c4914882c15 : SUCCESS in 1h 00m 39s (non-voting)\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/33c9200915d5475ca98af28eabbbff84 : SUCCESS in 53m 30s (non-voting)\n- kolla-ansible-centos8-source-cells https://zuul.opendev.org/t/openstack/build/1a792845f4cc46179a54756476f77e07 : POST_FAILURE in 22m 02s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/ec355f7c4f5a4196a60fb9cf1695305a : FAILURE in 28m 45s (non-voting)\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/c97b2f8464be4616831c5d76e912ab54 : SUCCESS in 1h 31m 00s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/bb34614ed97c481db0538fb6002a3556 : FAILURE in 33m 13s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/50a61da9f1454540aae8f39f57b6c481 : FAILURE in 7m 55s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"8c657c83093e25531b07f4853e2bc295462220af","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-02-15 15:35:52.000000000","message":"Patch Set 24:\n\nrecheck","accounts_in_message":[],"_revision_number":24},{"id":"276ace9efcb1f03900ce9320ea48486742ec8687","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-15 15:36:11.000000000","message":"Patch Set 24:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":24},{"id":"95698012a59489904fa1184980eb6e89b7873057","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-15 15:36:17.000000000","message":"Patch Set 24: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":24},{"id":"4a275b3d1fe8c7815e88b58f2fc1157d8bff2360","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-02-15 16:32:02.000000000","message":"Patch Set 24:\n\nrecheck","accounts_in_message":[],"_revision_number":24},{"id":"898ccfc7ea6b9496a2d4df055abc6069adf514ac","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-02-15 16:48:26.000000000","message":"Patch Set 24:\n\nrecheck","accounts_in_message":[],"_revision_number":24},{"id":"aacd499e39ad618d1f39c9c16d58cb30df361c0c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-15 16:51:06.000000000","message":"Patch Set 24:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":24},{"id":"e68dbe1715ba0c085e95a3e31fd84100cd85298c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-15 16:51:42.000000000","message":"Patch Set 24:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":24},{"id":"b82b346bf7864b7db8d44e99b32afb5ff6859826","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-02-15 18:16:50.000000000","message":"Uploaded patch set 25: Patch Set 24 was rebased.","accounts_in_message":[],"_revision_number":25},{"id":"8a28290bec8bea3b92c16460fe9fb302edf9d842","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-15 21:33:54.000000000","message":"Patch Set 25:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/7e48981b4b144d1f84928914375eb6dc : TIMED_OUT in 3h 03m 55s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/312bc8f72fe9462ab41463f56fb44eb5 : TIMED_OUT in 2h 32m 54s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"c7255143d808e79fd3ce0bbf1ae9d087aeda6d44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-16 04:39:18.000000000","message":"Patch Set 25: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a4c6da1930ee4bc188bda94bbe5607f8 : SUCCESS in 6m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f6abc0c50c04f589ff8996c9d529ba1 : SUCCESS in 4m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1657328060ed4d5c88b3d3d653f485c9 : SUCCESS in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/285a1e27d6fa40bd963cf315437af3ad : SUCCESS in 4m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e6fff8a11a894e61990e76c5f8408a6b : SUCCESS in 4m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b682dbb819143f4a9580951484e3987 : SUCCESS in 7m 35s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/aedeeb6f77e647d98d3dbc84e65508b7 : SUCCESS in 6m 54s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/f3c971c7b4614a48af7fde718c5bf1b9 : SUCCESS in 1h 10m 42s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/212d13ad7113435c8636666cc12bd539 : SUCCESS in 1h 03m 49s (non-voting)\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/0333ecfb00124c9e87f7f6a4f4171783 : SUCCESS in 1h 10m 32s\n- kolla-ansible-centos8-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/24e0c89840984c42b06d85c4e2ac0101 : FAILURE in 35m 38s (non-voting)\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/a0d0421a304e45ad9ee93e5e15782936 : FAILURE in 39m 30s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/6a93ef0e98a949bca74c9fff483e308f : SUCCESS in 1h 28m 31s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/67a166f0b4ea4ce6871d0362805acc9d : SUCCESS in 1h 21m 53s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/1e0dbacd9c4847d5a7e970ff566c94af : SUCCESS in 1h 01m 41s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/4bcb820789134feb9079c21db4e4865a : SUCCESS in 1h 07m 10s (non-voting)\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4b5fb6d128224b9f87a411c71b2a810a : SUCCESS in 50m 10s (non-voting)\n- kolla-ansible-centos8-source-cells https://zuul.opendev.org/t/openstack/build/e64de46ed63e4fee92a248250c6d3ddd : FAILURE in 1h 04m 05s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/b243189a9fa64077bbadf5479a15a19d : SUCCESS in 1h 25m 27s (non-voting)\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/720b0ffa2c1547ae84bccd221f781ced : SUCCESS in 1h 29m 55s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/f07585669c514bc4a8a8694b90203bf4 : SUCCESS in 2h 07m 49s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/102d2809cfdf4282b188f3e099b1f97b : POST_FAILURE in 1h 45m 54s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"5ed445b5e0948831345f93ee04b06244c4a3919f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-02-26 11:25:29.000000000","message":"Uploaded patch set 26: Patch Set 25 was rebased.","accounts_in_message":[],"_revision_number":26},{"id":"db941333eb8fced0ced773d3bf686021bf993e23","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 12:12:20.000000000","message":"Patch Set 26:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/51b03e4bf4eb44079cc893eca50f99c6 : FAILURE in 30m 47s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/6d6e9affd2254e0cad558bd53e96c161 : FAILURE in 31m 36s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"da77aced046e6d1a7d5c8fc610023cf0fb4fa809","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 12:27:39.000000000","message":"Patch Set 26: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9f1bf93ed95c48678b202d24ca057ae3 : SUCCESS in 7m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b86252ce09e4fd695425d02cb59a076 : SUCCESS in 4m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ead82feaa019426b90d4395789ad6dc6 : SUCCESS in 4m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a4627b294a4a48e99a2aea0d2fa3eed1 : SUCCESS in 4m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3ff6c5c90b04492fbbf8354c37e9e0ac : SUCCESS in 6m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8fd2ceccaf1f400baf2c5cfcd2009b59 : SUCCESS in 6m 22s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d36ed4a9df4342939c96f178d444e33a : SUCCESS in 7m 50s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/808f5033d06249b4a8845cf68089a7ee : FAILURE in 20m 07s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/9aa39fbb73a44e57bc2a0fc99dfbead6 : FAILURE in 28m 49s (non-voting)\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b00b6cf8d1704e0b9500991c4444be08 : FAILURE in 15m 52s\n- kolla-ansible-centos8-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/6a5c26f528514a658e4271566c2ca017 : FAILURE in 32m 06s (non-voting)\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/7757d26a13bb45a0906462c54d4640c2 : FAILURE in 17m 59s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/a7a06ec6399a4a62bbc8e83d444b7848 : FAILURE in 29m 37s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/6d5fdd563e38444d9305a26873159dcf : FAILURE in 15m 52s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/93b38211ac1943deb6a42ef87fed2a8c : FAILURE in 25m 19s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/a8fa87ef7bcd4a04bfeb6651baf03dd0 : SUCCESS in 52m 34s (non-voting)\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b59bb10695ac4a1b8045c36197569602 : SUCCESS in 58m 05s (non-voting)\n- kolla-ansible-centos8-source-cells https://zuul.opendev.org/t/openstack/build/e06ed0456fd04af2b9727d0b1f1d701e : FAILURE in 22m 37s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/9907a2c00fc54417843e602e0f43b479 : FAILURE in 31m 46s (non-voting)\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/c3f0adec5c5149828fccf26a9656e57b : FAILURE in 17m 21s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/e4ef8f53ca50479ca8e94ad3357d7532 : FAILURE in 31m 06s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/dcc19e79f0354520b490a0ff40e3c1a9 : FAILURE in 18m 17s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"bd8f45b5ad749a8964173ca106714baa4a3ef7b2","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-02-26 14:02:02.000000000","message":"Patch Set 26:\n\nrecheck","accounts_in_message":[],"_revision_number":26},{"id":"af7da646ca4856a70658747b15121cf9fa5443a3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 14:29:47.000000000","message":"Patch Set 26:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/dde9a41d7d0840f39003a484fa07ed6c : FAILURE in 24m 58s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/5ed964e5af5844e0b800f0ec999b3712 : FAILURE in 23m 06s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"c707ea36124f531036da82a85da25eb5d85f658e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-26 15:11:39.000000000","message":"Patch Set 26:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a7beca25344b4a9ab47131cc35819852 : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a85f87fe3ae2436b96b990ce579a3d01 : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c62badd00ee34dbfa12e431a740ccaf1 : SUCCESS in 5m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ee5e1fca9bf441e4951d34249003dbbd : SUCCESS in 4m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3b685685d00a4f8fb7385c11a0bd9d8e : SUCCESS in 4m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04771fc1d6c2455f8bf40148db1036ae : SUCCESS in 6m 40s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/2287761a6f314dfa8a55b100ec87d7b3 : SUCCESS in 8m 05s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/1eef056e792e408fbbd2d10b8280f894 : FAILURE in 22m 28s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/a7a6fe21522a45ea82d5c7a9252c0c80 : FAILURE in 15m 45s (non-voting)\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/5a8eabc94fc24a26b4ebf4dcd1f28366 : FAILURE in 18m 53s\n- kolla-ansible-centos8-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/fe2afcbbec19402581a7f1c056576f60 : FAILURE in 21m 50s (non-voting)\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/91f8f58bc8294a37a052e69e8864067c : FAILURE in 18m 57s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/441fde5f379c444bb0de6def65de9550 : FAILURE in 22m 08s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/14c852939d794de09bbc447c304da003 : FAILURE in 14m 23s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/8ae8c0fcf6b64534983da4d0e2336205 : FAILURE in 18m 11s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/4d633a4777714020868af180cf656768 : SUCCESS in 1h 04m 08s (non-voting)\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f7d8d06f081f46d39a54c967ecc18400 : SUCCESS in 59m 21s (non-voting)\n- kolla-ansible-centos8-source-cells https://zuul.opendev.org/t/openstack/build/6759a29c6ad24b0c8417ce0c4f9bb381 : FAILURE in 23m 27s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/2386bbb85bea42618a820b2534a5dead : FAILURE in 26m 28s (non-voting)\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/4755cfc2ef62409a8de8bd7fa92afb29 : FAILURE in 22m 10s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/33984105f3444f6287ab00233e2b5302 : FAILURE in 11m 58s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/77f637b2377b4095bab1d160c85d79f1 : FAILURE in 21m 53s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"a1587f5d7f12f769432bab8de5f603807778f87a","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-03-03 17:10:57.000000000","message":"Patch Set 26:\n\nrecheck","accounts_in_message":[],"_revision_number":26},{"id":"0e7783713824481beef5d57fccdebb7c01492514","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 18:34:17.000000000","message":"Patch Set 26:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b8b325be1286491c83722c20dc191299 : FAILURE in 54m 36s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/e7244a2254424928b710862d6f8ca598 : FAILURE in 1h 15m 47s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"a27aa82fd7ad6b2acdd73d4dc347358f11ec2d43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-03 19:23:08.000000000","message":"Patch Set 26: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/71720a25f8184d3f9fed0e9abe85bfb7 : SUCCESS in 6m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/433bbf3bbead498fbc2f6d9a3b5e85c9 : SUCCESS in 6m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3be88319d6bd46e9ae36d0e239f17cba : SUCCESS in 5m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4efbe9db4c9e4415b12d5384e49877e4 : SUCCESS in 6m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/88fb77d8d0154d2abb03ac515f3071a7 : SUCCESS in 6m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/512220213cda44858889e9a1604d223a : SUCCESS in 7m 51s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/ad899a051058481b937356c560064faa : SUCCESS in 7m 08s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/8c106886d2c94933a45763d8b1c05218 : SUCCESS in 1h 15m 19s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/b1d14d75f0484a028634ab9cbaa8e561 : SUCCESS in 1h 03m 09s (non-voting)\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/3c04f96b5e7b4d73a8061e85269cbd6e : SUCCESS in 59m 44s\n- kolla-ansible-centos8-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/4fefbc21e7ac46e6a2602572c73bd3dc : FAILURE in 31m 27s (non-voting)\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/41f083a430564a4c80199ba9753c4404 : FAILURE in 28m 03s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/0b2e63e398c64d18bd58ebb4e54c52e8 : SUCCESS in 1h 18m 46s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/bdf60342ed2e42c3b7698ca0d86fc82f : SUCCESS in 1h 14m 42s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/0dc53cf0334b401eba9da09838be80c3 : SUCCESS in 1h 10m 16s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/cbb62061e062411a8a8535c8d7d7ce1e : SUCCESS in 1h 07m 27s (non-voting)\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/822fae5d498147609d5bc0046e0bd829 : SUCCESS in 56m 41s (non-voting)\n- kolla-ansible-centos8-source-cells https://zuul.opendev.org/t/openstack/build/9e0e27579607420ca0e6dd3cec3f99ec : SUCCESS in 1h 27m 05s (non-voting)\n- kolla-ansible-centos8-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/1a343bd3d28f44cc95334c1e06737460 : SUCCESS in 1h 43m 01s (non-voting)\n- kolla-ansible-ubuntu-source-ceph-ansible https://zuul.opendev.org/t/openstack/build/efe846129bbc482f8d7a037f07861e3c : DISK_FULL in 1h 32m 23s (non-voting)\n- kolla-ansible-centos8-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/5812b638351a4600bdf6ce106b90d3c1 : SUCCESS in 1h 51m 19s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-ceph-ansible https://zuul.opendev.org/t/openstack/build/1dc15a07dbb2446fa53f8abcd8d573cd : DISK_FULL in 1h 27m 02s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"5fade05b41631fcc5722c972bdff57d64513c355","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-03-16 12:39:58.000000000","message":"Uploaded patch set 27: Patch Set 26 was rebased.","accounts_in_message":[],"_revision_number":27},{"id":"b4c81e473accb61cc92aa85cbbcd395b5471812e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-16 12:41:45.000000000","message":"Patch Set 27: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":27},{"id":"090ae4630c3772aa6f567f176c1c457e3fdbf134","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-16 12:42:47.000000000","message":"Patch Set 27:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":27},{"id":"c325402b3059777711528def9204535c6ae27ac2","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-03-16 13:54:16.000000000","message":"Patch Set 27:\n\nrecheck","accounts_in_message":[],"_revision_number":27},{"id":"be06302f535592f4026e1267a343d7db3a4a0bfa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-16 15:18:29.000000000","message":"Patch Set 27:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/811a4800515b4580962e5b66331ef4c5 : FAILURE in 1h 05m 49s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/8bfb9885dc01489c845c2731e1b73392 : FAILURE in 1h 17m 00s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"d254c1feba4eec2cebcc13c104adc67c4d4dc8f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-16 16:30:01.000000000","message":"Patch Set 27: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/986d0e6c787c45cc8aad267126489675 : SUCCESS in 6m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0109e59908384f2f933976d7c9710dc3 : SUCCESS in 6m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7060eb56c2634ab39ce27cc07a6367e1 : SUCCESS in 6m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/60520851c5464b84983a219812e4a931 : SUCCESS in 6m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d644c1233a4944c98c4ab24143cef3aa : SUCCESS in 6m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9151a178ceb7404c884731784d78cc31 : SUCCESS in 8m 09s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/b3e61a3f87644072bbe46580c496b41a : SUCCESS in 8m 46s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/959ef3644bc0438ab20d598c6a033e30 : SUCCESS in 1h 16m 19s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/8fab6f4564e149be959c016cf49ffd89 : FAILURE in 42m 56s (non-voting)\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/9a88787e9c954b0ca727463654a74f9a : SUCCESS in 53m 24s\n- kolla-ansible-centos8-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/8ed4571eeb8148d7a6dc6f89121dd424 : FAILURE in 1h 51m 24s (non-voting)\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/e0f68b24af364261950d2c30f64fecf3 : FAILURE in 26m 57s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/4b85683ceb0e4a2ba9006aef8920becd : SUCCESS in 1h 20m 20s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/bcc441813de54050a5ab77580bbd9c1c : FAILURE in 1h 01m 06s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/0145e95ef1544f34b61619d28a6410cb : SUCCESS in 1h 06m 56s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/88037a4b82374b29a918116e76361107 : FAILURE in 24m 45s (non-voting)\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c4b37ac33f114237886b155166c74402 : SUCCESS in 57m 10s (non-voting)\n- kolla-ansible-centos8-source-cells https://zuul.opendev.org/t/openstack/build/1ea8f57a53774426b1073293defdddaa : SUCCESS in 1h 22m 28s (non-voting)\n- kolla-ansible-centos8-source-cephadm https://zuul.opendev.org/t/openstack/build/0dc262ea034346bea9d2156709b180b5 : FAILURE in 1h 24m 32s (non-voting)\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/87fceeaa11864c4cbdeae6a87192190b : SUCCESS in 2h 23m 21s (non-voting)\n- kolla-ansible-centos8-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/ac7faea9b11149739cfec600c7e1e1e8 : FAILURE in 1h 37m 09s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/2fb608c77a724401b6250195322c95d1 : SUCCESS in 1h 29m 39s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"99c453aa8ed5050303305f314463837c376a3ba2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-03-17 23:57:21.000000000","message":"Uploaded patch set 28: Patch Set 27 was rebased.","accounts_in_message":[],"_revision_number":28},{"id":"ef47d5e7aaccfd3b86346d5867d7620e0f06eaf0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 00:08:09.000000000","message":"Patch Set 28:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":28},{"id":"2fd1915260461ebd0d7071b9c51f1bcf2a498e4e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 02:33:57.000000000","message":"Patch Set 28: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1d44fd8436ed426787279a4d5cc6567f : SUCCESS in 7m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/522f9aa537024eada9f4fe1fd2f4418a : SUCCESS in 7m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f12a07bdc2544f5b9d1038b337d95c05 : SUCCESS in 8m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a1e9b6b5993e4654bd8026312fb7f06f : SUCCESS in 7m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d7a9283dfb4649dfae0d3ab1b5c547d3 : SUCCESS in 7m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4f4f7863df64329a04f1a656e2e93da : SUCCESS in 8m 41s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/cabd0ccb4fb34b8dafa6df7486f7cd4d : FAILURE in 8m 39s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/bd8d02ea170441cc9c02509592eaa3c1 : SUCCESS in 1h 27m 06s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/7e3dc8d94cf74493b3ecfb6fd5a59e1e : FAILURE in 50m 23s (non-voting)\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/cc49eac5176c4e448f0b6bb3be478a09 : SUCCESS in 1h 02m 41s\n- kolla-ansible-centos8-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/9c2ae518c7d5448080682c484a8a97f3 : SUCCESS in 1h 21m 17s (non-voting)\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/48ec7d03ec9b49d99d30508012457a80 : SUCCESS in 1h 15m 49s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/c304971009d74dd6af20d1692fe1dc78 : SUCCESS in 1h 25m 26s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/2bfdea3bfd1b4319a430da18581aba85 : FAILURE in 1h 24m 57s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5d16ee9681a8489d893b25d630c55684 : SUCCESS in 1h 16m 21s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/6bfd41bc5a8b4823948f44e9455e616a : FAILURE in 32m 00s (non-voting)\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f9c8fc78aa4e4fd7b9db165ac3bfdd2b : SUCCESS in 55m 33s (non-voting)\n- kolla-ansible-centos8-source-cells https://zuul.opendev.org/t/openstack/build/6a284ae081f4465e9adb7a8bf2decd8c : FAILURE in 41m 46s (non-voting)\n- kolla-ansible-centos8-source-cephadm https://zuul.opendev.org/t/openstack/build/21c2518579cc4b289e2eb17dd9962671 : SUCCESS in 1h 24m 35s (non-voting)\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/152cde6cd8d64520a474f86612bada01 : SUCCESS in 1h 14m 29s (non-voting)\n- kolla-ansible-centos8-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/dfa1184ee44147ea82db42b553da4b60 : FAILURE in 1h 46m 08s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/a03c0dc9d97f4695afe47e8279b53084 : SUCCESS in 1h 21m 14s (non-voting)","accounts_in_message":[],"_revision_number":28},{"id":"38763e14f67808700515c55a3ef7ff3f38bc64f3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-03-18 14:58:34.000000000","message":"Uploaded patch set 29: Patch Set 28 was rebased.","accounts_in_message":[],"_revision_number":29},{"id":"4dbf33cc3d8b93b5c94485226cee883d23dc9a0b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 16:11:35.000000000","message":"Patch Set 29:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/d023f94b9d454ff2a969b76602269b80 : FAILURE in 46m 20s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":29},{"id":"6fc3dbd4e463e30723d1782b6e5781ff57eef863","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 16:52:36.000000000","message":"Patch Set 29: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a81ca50dd87d426398be077793d140a7 : SUCCESS in 6m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a50249d2314d4b74b626c1e42cbb39fb : SUCCESS in 8m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/191bbf160b604000b63615cb3634bf2e : SUCCESS in 7m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/98c5cdeab5fc43b7b4435bb3815edf73 : SUCCESS in 6m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/294918a73cd64dff835c8553dd115b5e : SUCCESS in 6m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1cb5ebc9124c49c0a71454d3ed5358c0 : SUCCESS in 6m 26s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/6288f14840c2459592071463bdfca2d3 : FAILURE in 10m 06s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/b5cfebadd8b7493fa509154ec1b3dc9f : SUCCESS in 58m 48s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/a016147c4905400bba8237b251543fc7 : FAILURE in 45m 32s (non-voting)\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c418cecf99904f3a96e90beec77d3e40 : SUCCESS in 58m 14s\n- kolla-ansible-centos8-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/c1ff0d3b0e8846e283fa7e2205cb73f4 : SUCCESS in 1h 31m 24s (non-voting)\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/cfb2abfe7265497586f05d0d4fdef890 : SUCCESS in 1h 04m 31s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/fc7c2cfbfb894db59edd5b4b9c577e05 : SUCCESS in 1h 14m 52s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/986a2437be684407a8583edd483fc1b3 : FAILURE in 52m 57s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/99736c3ea25c42d193073b2b13a4e3f0 : SUCCESS in 1h 05m 47s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/ba4c133e63184a948310652e2d8009c5 : FAILURE in 26m 12s (non-voting)\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a452319ba768476b87e796f890918554 : SUCCESS in 50m 23s (non-voting)\n- kolla-ansible-centos8-source-cells https://zuul.opendev.org/t/openstack/build/9e5aeb42e5744f10ad1b6e0893b391d3 : FAILURE in 37m 47s (non-voting)\n- kolla-ansible-centos8-source-cephadm https://zuul.opendev.org/t/openstack/build/89569a05978a43149ab12ab2b08cbd48 : FAILURE in 57m 30s (non-voting)\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/c20969755997410594e9ce7b00e2269b : SUCCESS in 1h 04m 15s (non-voting)\n- kolla-ansible-centos8-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/7fb3fce5903245ac98282ca691484dfc : POST_FAILURE in 39m 31s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/d61ada3cbe354f0baa838d9b3e2b8317 : SUCCESS in 1h 13m 15s (non-voting)","accounts_in_message":[],"_revision_number":29},{"id":"11aaefffc83611930ba311422df20544600dc2d1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-03-18 19:33:54.000000000","message":"Uploaded patch set 30: Patch Set 29 was rebased.","accounts_in_message":[],"_revision_number":30},{"id":"7ab35b300ff90581c047c22e6c6258984befa0b5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-03-18 19:46:00.000000000","message":"Uploaded patch set 31: Patch Set 30 was rebased.","accounts_in_message":[],"_revision_number":31},{"id":"7915c896d0ee728614366a1cdfb5af36cb269735","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 21:20:40.000000000","message":"Patch Set 31:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/41f8c2c9b93141499e8f1a167af2db1b : FAILURE in 59m 35s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/c3ca9846d60f4827b458eaa4f6327172 : FAILURE in 1h 17m 16s (non-voting)","accounts_in_message":[],"_revision_number":31},{"id":"8bafa24abd8df4f6a978c58cf40ce02ecaba634b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 21:55:39.000000000","message":"Patch Set 31: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9ee7b647a20846749c4af728176baa3a : SUCCESS in 6m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/825b4f3da1324ac3b9039aadff0d97af : SUCCESS in 7m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d599441205ac41f28cdea5a432d0a359 : SUCCESS in 6m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0aeb8d7f2c2845b98e88ebb461184f0e : SUCCESS in 8m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2160f39beaf44f60ab12c73c322baebd : SUCCESS in 6m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d14c8ad8cc54a35b9d588831bced048 : SUCCESS in 9m 45s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/aaca62e498554384a061874c82bbb402 : FAILURE in 8m 15s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/4237e4f227a24e33b3f7ca609735f1c0 : SUCCESS in 1h 07m 20s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/3e77c4d569e94ca8bd14b3c10477f556 : FAILURE in 47m 05s (non-voting)\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/0a8254559f73458897702f8e983c29e3 : SUCCESS in 1h 00m 37s\n- kolla-ansible-centos8-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/2d5567f9c5504c59a8592eaa39f35850 : SUCCESS in 1h 24m 35s (non-voting)\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/221ecd0caf8d4c45af965d9eb531f417 : SUCCESS in 1h 07m 39s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/897cb82187c5465dab9a54bb4107bf1e : SUCCESS in 1h 21m 36s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/ef778ad9192b4edbbc44a62b7899b6fa : FAILURE in 1h 09m 51s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/453167478e3e43f39b4328b9e42b50e5 : SUCCESS in 1h 07m 45s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/457f0998b2184feaa5cbcf5c189e961f : FAILURE in 25m 04s (non-voting)\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/99d95408130143f696e2753aa699b446 : SUCCESS in 56m 53s (non-voting)\n- kolla-ansible-centos8-source-cells https://zuul.opendev.org/t/openstack/build/9c3d0329504440b3b0b96e22f478f4fc : FAILURE in 38m 30s (non-voting)\n- kolla-ansible-centos8-source-cephadm https://zuul.opendev.org/t/openstack/build/15a06078f3ec4e24843c3b498a558484 : FAILURE in 40m 54s (non-voting)\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/823e56ac0ea4418d84b98372bdcc8e0c : SUCCESS in 1h 08m 16s (non-voting)\n- kolla-ansible-centos8-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/8d6901febbd444238d2d7d412f21d79b : SUCCESS in 1h 29m 46s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/1b17c87e5f3d4f6aa5160405b2439c5c : SUCCESS in 1h 59m 11s (non-voting)","accounts_in_message":[],"_revision_number":31},{"id":"8f4ec1234781dfa467502b8769439bc2dfce2f8a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-03-19 14:37:45.000000000","message":"Uploaded patch set 32: Patch Set 31 was rebased.","accounts_in_message":[],"_revision_number":32},{"id":"e72354096fa5c92bdf5a530305e4ab7603fcbcd9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-19 15:32:05.000000000","message":"Patch Set 32:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/98f9dac071f14a83baaec036c6069103 : FAILURE in 41m 16s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":32},{"id":"337a8a39ddc04159fa229dba3e01f815ea308a90","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-03-19 16:50:59.000000000","message":"Uploaded patch set 33: New patch set was added with same tree, parent, and commit message as Patch Set 32.","accounts_in_message":[],"_revision_number":33},{"id":"4fb9553387f8b903829ed2aa0c93abc526c1012c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-19 18:03:38.000000000","message":"Patch Set 33:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/435f79718fd04c9eb5dde35e34c097d7 : FAILURE in 1h 02m 00s (non-voting)","accounts_in_message":[],"_revision_number":33},{"id":"f7fbd3ac3c99228d562ad6cf2ce6b8fbe92392f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-19 19:52:35.000000000","message":"Patch Set 33: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9538271828a742f582beb36d19b25d58 : SUCCESS in 6m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0282562eb2a49e1aea4700803f3d4f8 : SUCCESS in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9d85cc6a233e461783577506ed49bd1f : SUCCESS in 6m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/304efd3da4da442bb12860953bd5549b : SUCCESS in 6m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e24884d7d7154c06bfa2f1faf2cabb8d : SUCCESS in 8m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bf31b2717374dfd947c266cbf61b702 : SUCCESS in 8m 10s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/2740da028f274f2b9ddc1f3b01d6e11d : SUCCESS in 8m 27s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/e0cb5d02371042738d15afbf99a86734 : SUCCESS in 1h 18m 00s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/ed2a7dffb6754dd4b3df151d950419a8 : FAILURE in 40m 37s (non-voting)\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/8ca39cbc5dcb43c7a44e62e3ed426546 : SUCCESS in 48m 03s\n- kolla-ansible-centos8-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/81ac692e90ec4b5689755ab9f922b019 : TIMED_OUT in 2h 02m 28s (non-voting)\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/e783bad5e2e740e6b5c8327ded24d241 : FAILURE in 57m 23s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/9330839b8c514a659851cdadf85d33e7 : SUCCESS in 1h 13m 22s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/c6a33f43eab34b85b2ea4617c5b72bd2 : FAILURE in 54m 54s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/57c9c0bb699b4c3e949ca728d124c0ce : SUCCESS in 59m 05s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/ef0a537de6504eaeae83d2424e5bf333 : FAILURE in 31m 10s (non-voting)\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7d5cb2a77ad645618cbb7e3356c51f67 : SUCCESS in 58m 20s (non-voting)\n- kolla-ansible-centos8-source-cells https://zuul.opendev.org/t/openstack/build/a95039da562c4679b8b812badb7e9dd7 : SUCCESS in 1h 19m 44s (non-voting)\n- kolla-ansible-centos8-source-cephadm https://zuul.opendev.org/t/openstack/build/5903cd6a4aff45c6a9bbefaf06357bf7 : SUCCESS in 1h 28m 09s (non-voting)\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/c2e3851d1224454cb2fe4f6de20c6444 : SUCCESS in 1h 27m 40s (non-voting)\n- kolla-ansible-centos8-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/0be35d98646342ba9268dc169bc20641 : TIMED_OUT in 2h 32m 27s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/d6f3f888212a48f986b7893dcb03d743 : SUCCESS in 1h 19m 42s (non-voting)","accounts_in_message":[],"_revision_number":33},{"id":"467e8a47843bc0f9aa7b3402dcf55c8f4d4e0106","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-03-20 07:44:31.000000000","message":"Uploaded patch set 34: New patch set was added with same tree, parent, and commit message as Patch Set 33.","accounts_in_message":[],"_revision_number":34},{"id":"45e2acb92d8a37ffbb6d5ef7366dbf8c38d94c99","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-20 08:52:09.000000000","message":"Patch Set 34:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/80fa184914b04c198c098b7c7052cd73 : FAILURE in 41m 24s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/bc9cb7b13e394a7fb5077b3194c6d995 : FAILURE in 58m 20s (non-voting)","accounts_in_message":[],"_revision_number":34},{"id":"056908db2c80b74ca8c3e9b720e1db3977d0f671","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-20 10:15:41.000000000","message":"Patch Set 34:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f96dad12119f44f8a3e8220061efd9da : SUCCESS in 6m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6287ce450dd7438aaa9dcc3edc21e1cd : SUCCESS in 6m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/af972dbb066c4940a542361bab207cf8 : SUCCESS in 7m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ff8b5e71cf97497e8817aac6abafc888 : SUCCESS in 6m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0b9fb1eddf13473186b763abf1b63e53 : SUCCESS in 6m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5334db93201a45728eac15e9152719ef : SUCCESS in 6m 45s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/db1beb196656419a888a34f2781504a1 : SUCCESS in 7m 17s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/cd9aea0641204357a6ceb83ccdb6f7db : SUCCESS in 1h 02m 00s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/4175e92053c34529af8ca4b906eb022a : FAILURE in 31m 44s (non-voting)\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/0e550ac6a5a9440e81de05f0096d782c : SUCCESS in 50m 20s\n- kolla-ansible-centos8-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/a0b9721e9e7e4d359a11b5cc24ec6876 : SUCCESS in 1h 45m 33s (non-voting)\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/5c0bd30101ee4b40a6ac38c398a77dbd : SUCCESS in 59m 28s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/c91ad6dbbf3f4da3958908a5671e6ccd : SUCCESS in 1h 16m 34s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/7cfa5f7efef6446c8de1f38d5300cc9b : FAILURE in 1h 00m 11s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f43816724e0a4865a0daf1299a138cb3 : SUCCESS in 1h 04m 36s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/50e18443115c45328c24516e25ac6773 : FAILURE in 26m 53s (non-voting)\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c025e1da26cc485f9d83cec38112160f : SUCCESS in 49m 40s (non-voting)\n- kolla-ansible-centos8-source-cells https://zuul.opendev.org/t/openstack/build/df06b0ab299e4ddbb476de5ce75803fe : SUCCESS in 1h 03m 50s (non-voting)\n- kolla-ansible-centos8-source-cephadm https://zuul.opendev.org/t/openstack/build/833d115486044ece86773bc492f99ad0 : SUCCESS in 1h 15m 11s (non-voting)\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/901dc64f17eb491786b4b4ceed77c0e2 : SUCCESS in 1h 02m 48s (non-voting)\n- kolla-ansible-centos8-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/ee6b2503cea74e64a973622425332655 : FAILURE in 2h 22m 33s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/5b52b4a0f57c4580a0afe556ccbedf5c : SUCCESS in 1h 41m 46s (non-voting)","accounts_in_message":[],"_revision_number":34},{"id":"d9fb1c7b69e23ed9a303916fe83b0a19ac66982a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-03-23 15:23:00.000000000","message":"Uploaded patch set 35: Patch Set 34 was rebased.","accounts_in_message":[],"_revision_number":35},{"id":"442264697c07d3a74e8eec285facbafe66e81c7d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-23 16:39:25.000000000","message":"Patch Set 35:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":35},{"id":"624c3230fb5dcb04a453bbe8a5c1bc3f057b4b9b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-23 17:30:36.000000000","message":"Patch Set 35: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ec0f2d93dd52464f85273c6c797ff9ab : SUCCESS in 7m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6326f14e4224ca8976e7eb0c2b599e5 : SUCCESS in 6m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2ef23e63ae5f4c298be237dd13b2594c : SUCCESS in 6m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/94bbb5cc71554919b8727ca7f76edaac : SUCCESS in 6m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c9972a70b2834508a8d96736550140c7 : SUCCESS in 7m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa430e200bb947c3aea7cbfa5444d0dd : SUCCESS in 7m 33s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/fce2f4422bd2443896b9287e85c085b4 : SUCCESS in 9m 15s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/1e82d1e18ddf4d8fb8651ce47888b68c : SUCCESS in 1h 08m 51s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/da6f7523e25f41398c72388961cb6270 : FAILURE in 42m 30s (non-voting)\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a0ef20f0e88f44ae98208a390567dab3 : SUCCESS in 50m 29s\n- kolla-ansible-centos8-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/bf784a39671f4a0b830bdc380da728c3 : SUCCESS in 1h 14m 38s (non-voting)\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/c97498491a25465b8073b56d51eff475 : SUCCESS in 1h 39m 51s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/5da6fdc9cbda410a83a41083814a80c7 : SUCCESS in 1h 13m 08s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/aeb63bfa97384dec955750bb2722a3bb : FAILURE in 1h 03m 06s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/7413a51acc724fbfab2d2645f179114e : SUCCESS in 1h 53m 16s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/9585597f33cf44488badd3a54603ef37 : SUCCESS in 1h 00m 47s (non-voting)\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d43bebdf41204cde8f5aefbc48ddc57e : SUCCESS in 1h 20m 27s (non-voting)\n- kolla-ansible-centos8-source-cells https://zuul.opendev.org/t/openstack/build/77864b14647441f1bdbe07b7d909a033 : FAILURE in 35m 33s (non-voting)\n- kolla-ansible-centos8-source-cephadm https://zuul.opendev.org/t/openstack/build/6f9744ea1168472688a64e665cacb0b6 : FAILURE in 1h 26m 55s (non-voting)\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/9978b63998de4702a41bf15a65176237 : SUCCESS in 1h 57m 39s (non-voting)\n- kolla-ansible-centos8-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/2ec86f568ab243fa83a6a40b8b833928 : FAILURE in 35m 19s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/cfc757b76a974a08907707ee5ba51131 : FAILURE in 12m 15s (non-voting)","accounts_in_message":[],"_revision_number":35},{"id":"ce1cc2c3b9f263f5959fc68b74a0ce12bd762f23","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-03-24 19:43:44.000000000","message":"Uploaded patch set 36: Patch Set 35 was rebased.","accounts_in_message":[],"_revision_number":36},{"id":"9eb6cf2dfab64e9408160c054065b8cef05296c5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-24 20:33:14.000000000","message":"Patch Set 36:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/9dea6ac0f6f44143ac6dcb8c6fa823c3 : FAILURE in 28m 19s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":36},{"id":"5ed1941447793789d5814fe29911b26a0c1327f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-24 21:09:24.000000000","message":"Patch Set 36: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b62ef8fb264e416880757292a809f81f : SUCCESS in 6m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ee22c5903a94e73ae4df9c7e4a3ae5b : SUCCESS in 5m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/61bffbbad50e4e17ad4949e404a9672d : SUCCESS in 7m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b9ad67d7ef464e0092031310ad7bfd6c : SUCCESS in 7m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/91fe12df03694c6ebd559fb42c58f1db : SUCCESS in 8m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac8f56d353304afca79914387868d50e : SUCCESS in 6m 53s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/45121b03cdaf4188ba802059acb2a283 : SUCCESS in 8m 05s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/5a2ed884b4154ca986bd5ad45097cbd5 : SUCCESS in 55m 51s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/b9e8331fe5a8468880d1e1fa1fbedaa2 : FAILURE in 46m 22s (non-voting)\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/df101f49816c458ca8c45d73aa06eb74 : SUCCESS in 47m 29s\n- kolla-ansible-centos8-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/105587cd9a2543b9bffca4dbade30e9c : SUCCESS in 1h 17m 48s (non-voting)\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/0ba6d927a72a41d0aedb8a30e79d0370 : SUCCESS in 1h 07m 41s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/03a002ca96224f39829d0f0af5e3dd19 : SUCCESS in 1h 19m 36s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/f09edc300b1540a68ac0407b8a833374 : FAILURE in 1h 08m 14s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/00fe5a6be61d4641bc3d355f403d05a1 : SUCCESS in 1h 03m 47s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/a80c407f4ccb42768530aecc66d774a5 : SUCCESS in 58m 01s (non-voting)\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f9bf74c9eca0431e8c209433ebc67617 : SUCCESS in 46m 26s (non-voting)\n- kolla-ansible-centos8-source-cells https://zuul.opendev.org/t/openstack/build/9c27985750b843c8a47f73fb887f2a14 : FAILURE in 37m 11s (non-voting)\n- kolla-ansible-centos8-source-cephadm https://zuul.opendev.org/t/openstack/build/f0aea6ccebf140dfb71583d50322be57 : SUCCESS in 1h 12m 31s (non-voting)\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/bdcd4174b6de437d9143bfa1d1863d09 : SUCCESS in 1h 09m 28s (non-voting)\n- kolla-ansible-centos8-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/ee870c25389d4173882037cdc84dc2c9 : SUCCESS in 1h 18m 59s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/859105b6b6574692b298fbd9483d1a8f : FAILURE in 21m 39s (non-voting)","accounts_in_message":[],"_revision_number":36},{"id":"067d554eadce9263517cfbb04a2b9b80e9ee7d18","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-03-27 18:19:07.000000000","message":"Uploaded patch set 37: Patch Set 36 was rebased.","accounts_in_message":[],"_revision_number":37},{"id":"06e4601d5b11b79ab6e52cfd0e198d65dec99f7f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-27 20:12:54.000000000","message":"Patch Set 37: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ca2f201883ac4d868dfb45e71859869a : SUCCESS in 7m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0910a3e6623541f7a6f065f4b4045225 : SUCCESS in 6m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c85b3ad142da4c0884491de281826d3c : SUCCESS in 7m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b72d1d7d793a466cb594b3a140194c16 : SUCCESS in 7m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1d8de4a337584eb1bf343fe1e56c6770 : SUCCESS in 6m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f64ad7662e434b6491fcc63c4414b38f : SUCCESS in 9m 47s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/0a84bc5622004bb0bc88346aea78cdf0 : SUCCESS in 7m 02s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/22c0e7f428454fcd948d1dfa4aeaeea1 : SUCCESS in 1h 30m 51s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/8763eb90358740189fce8305c1435aa3 : SUCCESS in 58m 41s (non-voting)\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/23a47d0c6a804f32a258ab72a44f8bf6 : SUCCESS in 42m 37s\n- kolla-ansible-centos8-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/9ca5eb06b8a2433cb93126415e9b4235 : SUCCESS in 1h 00m 01s (non-voting)\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/2331b79b7ebf462fae844475b20990a7 : SUCCESS in 50m 56s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/4621d59e3d644d8d857ed092ab880972 : SUCCESS in 59m 25s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/a1c25e331ea6418a8cec48772b42eff0 : SUCCESS in 1h 10m 47s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ccb1ee89460c4719b8a2bad7f78583eb : SUCCESS in 53m 58s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/c1a07f9b30714aebb111b50709987b8c : SUCCESS in 54m 16s (non-voting)\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7f7bf606bbca461a9b6392c3f9157628 : SUCCESS in 44m 33s (non-voting)\n- kolla-ansible-centos8-source-cells https://zuul.opendev.org/t/openstack/build/f9cd1a2173cf487daeea0e20342e6432 : SUCCESS in 1h 43m 00s (non-voting)\n- kolla-ansible-centos8-source-cephadm https://zuul.opendev.org/t/openstack/build/9321e643a8a74202b43bdaf1f30e446a : SUCCESS in 1h 05m 45s (non-voting)\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/3eea7cf66a4b42b4a29708d62e02c9a1 : SUCCESS in 1h 32m 18s (non-voting)\n- kolla-ansible-centos8-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/cc2759cea29c4a0084937efe3dfacd7b : SUCCESS in 1h 09m 08s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/8cdcbd0ca35f4dabb03383832f317cc3 : SUCCESS in 1h 45m 05s (non-voting)","accounts_in_message":[],"_revision_number":37},{"id":"0d0796f265419ea250142485254bbaa47b708e81","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-03-28 22:22:44.000000000","message":"Uploaded patch set 38: Patch Set 37 was rebased.","accounts_in_message":[],"_revision_number":38},{"id":"ce7a83b104fedce2e5295ab7746e4d5e31b7e42c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-28 23:54:40.000000000","message":"Patch Set 38: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/471643d4b5404913b30a06ad1c5b9701 : SUCCESS in 5m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd9c4320c05641d3a79970d824abbb09 : SUCCESS in 5m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b76955a2da36461d8e4dc68c08df347d : SUCCESS in 5m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/05a5e156ef8d43609170e97b3c599152 : SUCCESS in 5m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e56329a76005481dbabfdd72c3c1fa74 : SUCCESS in 6m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47e57c04865b4fcb88bbd4af1e27ba4b : SUCCESS in 6m 58s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/6a38cc1f9b8f42fdae81e4a1f310825a : SUCCESS in 7m 01s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/7137b38e9e22495fa01755acf2bc2afb : SUCCESS in 56m 52s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/37fa9dd12dc948a39a8b9fcd0e995ed2 : SUCCESS in 46m 30s (non-voting)\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c87bfdc5023b4731ab5fa0db03a7e5a3 : SUCCESS in 54m 33s\n- kolla-ansible-centos8-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/597c742326ec437b97bbd24ada03a42e : SUCCESS in 1h 07m 32s (non-voting)\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/a7475fe6bfc440cdb850ec5671b64ce4 : SUCCESS in 1h 05m 51s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/717dde57041d4d16ac8a3d0e2cffddbe : SUCCESS in 58m 13s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/e589584c86ea417984d8be75b10c2e75 : SUCCESS in 1h 20m 21s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/e4a3060f999241a487a2a54cae968cfc : SUCCESS in 1h 04m 02s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/d7cd007a2f834d969d131c26db200b85 : SUCCESS in 44m 54s (non-voting)\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c99ac60648b24aefac7620e41e975d56 : SUCCESS in 47m 49s (non-voting)\n- kolla-ansible-centos8-source-cells https://zuul.opendev.org/t/openstack/build/bd2b72c6e5d84b049b123f4b227addf7 : FAILURE in 1h 26m 11s (non-voting)\n- kolla-ansible-centos8-source-cephadm https://zuul.opendev.org/t/openstack/build/b3ae17333b8a40e3b28ea1eb303fb609 : SUCCESS in 1h 05m 43s (non-voting)\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/05d8f2abe8674137bbb6f081e50a0c32 : SUCCESS in 1h 00m 03s (non-voting)\n- kolla-ansible-centos8-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/0cef4c112bb349e3bf325e153f40dbab : SUCCESS in 1h 29m 16s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/b910b920b5ec40b6a1bea06082bc917e : SUCCESS in 1h 09m 55s (non-voting)","accounts_in_message":[],"_revision_number":38},{"id":"d7927c9ab1a50417ec71b46181a5ab05a8e6ed5e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-29 00:07:54.000000000","message":"Patch Set 38:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/dcc85469a4cd40e29e7ec06432a2ccfe : SUCCESS in 1h 35m 04s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/1d71281350db4b0fb3eed28def7ffbfc : FAILURE in 1h 14m 15s (non-voting)","accounts_in_message":[],"_revision_number":38},{"id":"a910d99a447515e546aafbbe523ff75c476e696e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-03-29 10:23:41.000000000","message":"Uploaded patch set 39: Patch Set 38 was rebased.","accounts_in_message":[],"_revision_number":39},{"id":"c709053cc52055870941c911a2ee16cc876e862b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-03-29 11:00:25.000000000","message":"Uploaded patch set 40: Patch Set 39 was rebased.","accounts_in_message":[],"_revision_number":40},{"id":"c2a0f97852e4d5744ceda7ec28e0246e1da0c236","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-29 12:31:10.000000000","message":"Patch Set 40: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ef58c4e493e44c2db99018eeca750221 : SUCCESS in 8m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3fa696b8aac8459e88453980ec45ccba : SUCCESS in 6m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/497e5850245b46bd8880cd5239406e80 : SUCCESS in 5m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/08d5f9f9cc4d49af91b8fab85cb0e88a : SUCCESS in 6m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/10d7533f331e4e3fbdefb436212fbf90 : SUCCESS in 5m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fecc9a0efafd42c9895b006742c37309 : SUCCESS in 7m 29s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/64ac9019343347a4b7f005cbee84a147 : SUCCESS in 8m 12s\n- kolla-ansible-centos8-source https://zuul.opendev.org/t/openstack/build/5d5403b5911d44849f40813a28aba6d2 : SUCCESS in 1h 00m 09s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/fc85cc0822a942adbb5c211dd8434510 : SUCCESS in 1h 04m 27s (non-voting)\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/53048769fcb24620a2be383c982d2ecd : SUCCESS in 49m 44s\n- kolla-ansible-centos8-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/228a5ed312684bad93ad27df01295633 : SUCCESS in 1h 00m 37s (non-voting)\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/154a1c4ee63f4d3881e9f10d2123b1b4 : SUCCESS in 1h 14m 52s (non-voting)\n- kolla-ansible-centos8-source-upgrade https://zuul.opendev.org/t/openstack/build/26558b813495495689df410b82dad45a : SUCCESS in 1h 21m 36s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/a03c2f3b57fb4d6d9041aec6db65412c : SUCCESS in 1h 05m 32s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/1dee5ba46745432da06c35efa58ab279 : SUCCESS in 1h 03m 14s\n- kolla-ansible-centos8-binary https://zuul.opendev.org/t/openstack/build/6f193ca564b3429b95fd1269a736a517 : SUCCESS in 1h 07m 58s (non-voting)\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c519ce291b8647538ea19499d608481e : SUCCESS in 53m 11s (non-voting)\n- kolla-ansible-centos8-source-cells https://zuul.opendev.org/t/openstack/build/e5c9615bc4804cb3b4361501b89fe18c : SUCCESS in 1h 06m 54s (non-voting)\n- kolla-ansible-centos8-source-cephadm https://zuul.opendev.org/t/openstack/build/ea67252c292b44d98c7bf9c90120fe06 : SUCCESS in 1h 24m 05s (non-voting)\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/2480805bbb6d44a890c2eb9e9c9476fb : SUCCESS in 56m 12s (non-voting)\n- kolla-ansible-centos8-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/036214fc6ed74a6f82b48b44c4389770 : SUCCESS in 1h 14m 54s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/fe1370f7dfb84aea8ae0417bc6ef601e : SUCCESS in 1h 03m 31s (non-voting)","accounts_in_message":[],"_revision_number":40},{"id":"e81a6f768a214dd6b4aa3b254eb1ad040eb6a706","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-04-07 12:22:10.000000000","message":"Uploaded patch set 41: Patch Set 40 was rebased.","accounts_in_message":[],"_revision_number":41},{"id":"ab8b86daef9e3ab683b33b9238c10e9ba67e5d19","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 12:25:58.000000000","message":"Patch Set 41: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":41},{"id":"e6a44f152fe7af3f552c85f644aca6293b1efa49","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 12:27:39.000000000","message":"Patch Set 41:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":41},{"id":"ddfcbaf1cebe81e39e9a342574be0a0d49eb9607","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-04-07 12:42:33.000000000","message":"Uploaded patch set 42: Patch Set 41 was rebased.","accounts_in_message":[],"_revision_number":42},{"id":"ee38410c46fe006b3245fb99d78ed5a1f2975de9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 12:44:36.000000000","message":"Patch Set 42: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":42},{"id":"96382d7c74175dc172aa1643ede2df6741e8ae6b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 12:46:46.000000000","message":"Patch Set 42:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":42},{"id":"6109bb03b4a8eba8063ab0a95c84dbf3b280c84a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-04-07 21:27:30.000000000","message":"Uploaded patch set 43: Patch Set 42 was rebased.","accounts_in_message":[],"_revision_number":43},{"id":"96ca8dd7c36eaaddab9ce6128ac40a0b975b4fe1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-07 23:35:19.000000000","message":"Patch Set 43: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7e12676b8d8242bc99dc72daad730540 : SUCCESS in 6m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d3f6c69230149589cda827d47dbba56 : SUCCESS in 6m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b0e4f4e4b02248109f5905bd7289b4bc : SUCCESS in 7m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/035aa5fa174840549d7087d933470718 : SUCCESS in 7m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0ba86c2d70b94a44b6337a9419b71e06 : SUCCESS in 7m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/017f1cff1f8346b3b608163bd6745ef0 : SUCCESS in 6m 44s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/832f081a89554eafb1dab3f95b611f63 : SUCCESS in 8m 18s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/7fb6fc8f2836421ea1361058a8711d01 : SUCCESS in 1h 19m 54s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/8e01bf1615a440d582b0264df0aa001a : FAILURE in 22m 19s (non-voting)\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/fe64a95f798c4e4cacba1cb3913c0a2b : SUCCESS in 52m 11s\n- kolla-ansible-centos8s-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/9f3c9834672140889b73c9a1c6147e74 : SUCCESS in 1h 30m 29s (non-voting)\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/e514c684681147bab88e7a1d00d340d3 : SUCCESS in 1h 15m 16s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/1e7ea092c3264dd799a133b9046f21e6 : SUCCESS in 1h 03m 24s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/46716784126a4fa0882c8227fe690a5d : FAILURE in 19m 38s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ac5989c244a24494a29e8d7f1178b40b : SUCCESS in 52m 50s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/3baa91b24bf74343875312c5cffc6bc5 : SUCCESS in 48m 39s (non-voting)\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/801b190d538843139e694c70ec143fb6 : SUCCESS in 55m 38s (non-voting)\n- kolla-ansible-centos8s-source-cells https://zuul.opendev.org/t/openstack/build/02a55da59af6453f9f7def1dcd5d8e10 : FAILURE in 1h 11m 49s (non-voting)\n- kolla-ansible-centos8s-source-cephadm https://zuul.opendev.org/t/openstack/build/3de8615118da474bb5d6775bfcdc1d69 : FAILURE in 36m 41s (non-voting)\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/aa99849c2d8445dfba90f7ed7f581ba1 : SUCCESS in 1h 13m 47s (non-voting)\n- kolla-ansible-centos8s-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/4f6167a8a2854c0cbb2eae79f37ddbf0 : FAILURE in 40m 51s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/58477942bb65462e870c9d5fbd47db7c : FAILURE in 1h 40m 50s (non-voting)","accounts_in_message":[],"_revision_number":43},{"id":"b9e4704e955aee8012a423114c8dc770c774491f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-04-10 07:27:27.000000000","message":"Uploaded patch set 44: Patch Set 43 was rebased.","accounts_in_message":[],"_revision_number":44},{"id":"4b1618ae9e904084c35667e32b89f01ccb9019d7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-04-10 07:30:38.000000000","message":"Uploaded patch set 45: Patch Set 44 was rebased.","accounts_in_message":[],"_revision_number":45},{"id":"31ed5f20e35bd6764ae38e514cfe16cf36d200cd","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-04-10 07:39:15.000000000","message":"Topic change-780970-23 removed","accounts_in_message":[],"_revision_number":45},{"id":"c71f3b9399e72d54c578af1ed7f418a63120702d","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-04-10 07:39:35.000000000","message":"Topic set to mariadb-changes","accounts_in_message":[],"_revision_number":45},{"id":"144baa328204c984283b071768d87f9db90cbfff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-10 09:03:33.000000000","message":"Patch Set 45: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f99206d5e3aa4b8da50d835b53e993be : SUCCESS in 6m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/46fb3fec42774b71a295f668c44cbc22 : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6cc26272006b4d8ea7e16ab12f33d0b8 : SUCCESS in 5m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/649cb2ade0a843a88a3a7a8eaf3c0532 : SUCCESS in 5m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dfe0ea110f334b99a8b3bcdee2cb0d2a : SUCCESS in 5m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/172e91b3280d460aa8a1d243a7a1d742 : SUCCESS in 8m 29s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/a56a543a4f614993aaa4fc423da0877f : SUCCESS in 6m 20s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/9bf250fe33f4450799ef6fa4e499285f : SUCCESS in 1h 09m 48s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/15ca9fd204ef4410abf72f25628e3376 : SUCCESS in 41m 38s (non-voting)\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/84232a795dd24ef6a0599a71c58c9236 : SUCCESS in 41m 06s\n- kolla-ansible-centos8s-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/7ff603eeead044108b9760ae1e56ef9b : SUCCESS in 1h 09m 34s (non-voting)\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/eef8d2362f904107a7e499ecbe0c2bb8 : SUCCESS in 58m 53s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/48af07486aa54bcd8988a11892eec895 : FAILURE in 47m 34s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/2b03e6776344407dafbbe89555007a3e : SUCCESS in 1h 02m 25s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f0dd8e2e4d48481babf6939e99a39944 : SUCCESS in 1h 00m 15s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/3b710dfb872d4e7e923c7c42a30793be : SUCCESS in 1h 06m 00s (non-voting)\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/de8ea11eefa24667b55f35e7390425d3 : SUCCESS in 51m 26s (non-voting)\n- kolla-ansible-centos8s-source-cells https://zuul.opendev.org/t/openstack/build/780475d9bc644853b67b990421409c56 : SUCCESS in 1h 17m 00s (non-voting)\n- kolla-ansible-centos8s-source-cephadm https://zuul.opendev.org/t/openstack/build/99a11440a8f046eabfd3d6ac44a111fa : SUCCESS in 1h 14m 23s (non-voting)\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/aa9a489baef84ebdb21535fe6e82c03a : SUCCESS in 1h 25m 48s (non-voting)\n- kolla-ansible-centos8s-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/14a36b8b8ca34fcca9a4937e562bd545 : SUCCESS in 1h 26m 33s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/f2e559cbcc844da892da6f8dd8eac384 : SUCCESS in 1h 01m 06s (non-voting)","accounts_in_message":[],"_revision_number":45},{"id":"da58dd104dc204b9b4fecc8f71a4c11ea740538d","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-04-10 09:31:59.000000000","message":"Patch Set 45:\n\nrecheck","accounts_in_message":[],"_revision_number":45},{"id":"d129fa2a3def5be8da8e819657b75a334b233756","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-10 11:04:49.000000000","message":"Patch Set 45: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/042c1cf0aa074eb39deb220937a65244 : SUCCESS in 6m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/968dc9414b0a4b2bbc9fcd7fa76a018c : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/236e1b587fb94b4783a28845b7720318 : SUCCESS in 6m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/00e8aa30e0b944189705255832646497 : SUCCESS in 6m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b1556704404545278680891f1a59d8d7 : SUCCESS in 6m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6f8a8372c7e45979ff902837b5957be : SUCCESS in 7m 44s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/07f40b5bf3534efaa60b576a0f4c5f17 : SUCCESS in 8m 44s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/acd21d3f46fc4387bccf87f0654e4eef : SUCCESS in 1h 13m 12s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/53dbe61c10a24621b6e44e22fd5cdac8 : SUCCESS in 54m 56s (non-voting)\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/00cbf6fdee1f49d597c2aeb7b991637b : SUCCESS in 54m 26s\n- kolla-ansible-centos8s-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/ad5b9ed3e622481289520ede64b29e41 : SUCCESS in 1h 15m 18s (non-voting)\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/b0c366f5df024853b78a77c7b5acb54b : SUCCESS in 1h 02m 45s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/24220195cc524cba8887c2fbfe7a8e10 : SUCCESS in 1h 13m 31s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/6ad4b80acb3d47ab871f437b25691a45 : SUCCESS in 1h 03m 55s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/a161d6a0e2a34b97857bbef1f48f10fe : SUCCESS in 1h 04m 47s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/9c7e2fdab71d446c932695682124113b : FAILURE in 49m 42s (non-voting)\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8938b91ead5848178bdb886ade18838e : SUCCESS in 1h 00m 34s (non-voting)\n- kolla-ansible-centos8s-source-cells https://zuul.opendev.org/t/openstack/build/4d351fb3448240d09d7e38bba7a676ca : SUCCESS in 1h 15m 39s (non-voting)\n- kolla-ansible-centos8s-source-cephadm https://zuul.opendev.org/t/openstack/build/06601e2ca3f847f3ba491c4a7eb4ed41 : SUCCESS in 1h 17m 26s (non-voting)\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/db01ae77fc2b47c5a8dfb48d5adf009c : SUCCESS in 1h 13m 12s (non-voting)\n- kolla-ansible-centos8s-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/2bf811c444e34b979f2a55888b56108d : SUCCESS in 1h 30m 22s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/0916354965b44067bd8a6c6dfa6a1ca6 : SUCCESS in 1h 18m 31s (non-voting)","accounts_in_message":[],"_revision_number":45},{"id":"a37bfcd33c1ebad98451e6d0f1daf1d9521faeb7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-12 03:53:46.000000000","message":"Patch Set 45:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/None : NODE_FAILURE in 0s (non-voting)","accounts_in_message":[],"_revision_number":45},{"id":"84c29546340e4f6321815b5f4446562d52c327b7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-04-14 14:54:25.000000000","message":"Uploaded patch set 46: Patch Set 45 was rebased.","accounts_in_message":[],"_revision_number":46},{"id":"c2bcceff7d63429e690fbc44873f6d581bef8bdd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 16:37:45.000000000","message":"Patch Set 46: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/02e39dea643f467dbcd6344f434e1c46 : SUCCESS in 7m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1462ad6179e2411da6266cb2972e4324 : SUCCESS in 7m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1f54f4b416f34b128c97d32e7420dfb2 : SUCCESS in 6m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0210cd5735c74e20ad428f7d4a3b5957 : SUCCESS in 7m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/adea793b21074702a5fc3203dd18b16e : SUCCESS in 8m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f85c8aa2b6f46639cb5bb945fd4d68f : SUCCESS in 7m 22s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/5de63d991aa64301be7e660cfa15a923 : SUCCESS in 9m 46s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/db72c97c8b1e483cb62c02bc8f6e430b : SUCCESS in 1h 14m 49s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/3a5cf624977643dba0cc35eb885e54d9 : SUCCESS in 1h 01m 59s (non-voting)\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4f1c8910e19e4851ba285fdab809e85c : SUCCESS in 1h 00m 15s\n- kolla-ansible-centos8s-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/217929f003b34b2aba2769574fbbf2bd : SUCCESS in 1h 21m 01s (non-voting)\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/89e6ade78f5d4644825e4a30154fdcf5 : FAILURE in 1h 12m 10s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/514fd521c39b428ebca7822b60f6440b : SUCCESS in 1h 23m 25s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/bcb4d471c1d24699812b3a0153e22ecf : SUCCESS in 1h 10m 09s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5cae81cd3b124772a6ea36cd91974b92 : SUCCESS in 1h 05m 11s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/53621e99b6984000bfb0f82c7adcb9f6 : SUCCESS in 1h 14m 42s (non-voting)\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8fa8366dee6b452a962f0e3f12eda5eb : SUCCESS in 54m 17s (non-voting)\n- kolla-ansible-centos8s-source-cells https://zuul.opendev.org/t/openstack/build/e908bb973204461ba7e601e281554223 : FAILURE in 54m 29s (non-voting)\n- kolla-ansible-centos8s-source-cephadm https://zuul.opendev.org/t/openstack/build/a1aa857d8bb94e5782ea7d964dcfc69a : SUCCESS in 1h 29m 21s (non-voting)\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/8fd53fe778a7410bb24ff1544127dbdd : SUCCESS in 1h 18m 52s (non-voting)\n- kolla-ansible-centos8s-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/eddcb51eb49d49d4ae86bcbd2f141081 : POST_FAILURE in 44m 24s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/2e373488025f448980db7915043766ec : SUCCESS in 1h 29m 42s (non-voting)","accounts_in_message":[],"_revision_number":46},{"id":"f251e9288be5c78ba142d3987e41a43a45c4c39d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-14 16:55:46.000000000","message":"Patch Set 46:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/b82668153d2a48eeb625f1d0b5e82283 : SUCCESS in 1h 41m 10s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/7c970fb23a9340648d17b4d2ada24987 : FAILURE in 51m 37s (non-voting)","accounts_in_message":[],"_revision_number":46},{"id":"9f17205faa68e187bb72540d79a3b5b4c29d2f83","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-04-19 16:34:03.000000000","message":"Uploaded patch set 47: Patch Set 46 was rebased.","accounts_in_message":[],"_revision_number":47},{"id":"7046925d9d35c5a24c1e37fc650cd86077568526","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-19 18:27:04.000000000","message":"Patch Set 47: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e779b357b2394c358d031cab18cf623d : SUCCESS in 7m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18a16cbfe993418780d59f07f0462ef8 : SUCCESS in 6m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/591f35ccc3ea4dc293fc4789c7dd985d : SUCCESS in 7m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/97eeabb9bc394d90afcf05082a4f136e : SUCCESS in 6m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6c122a22d6964af5addca8e008ad38f5 : SUCCESS in 6m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/826a335bde1a488290abd66940189f11 : SUCCESS in 7m 05s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/c0347bd282224ba1abd838bbbf60363a : SUCCESS in 7m 59s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/3649dd8e27b84a4683ccdd1fc761ad5d : FAILURE in 38m 59s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/6dcd5a2a60154350af096fc763aee61a : FAILURE in 36m 43s (non-voting)\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4aa2e493443949ffaf1baf4862d6e659 : FAILURE in 32m 41s\n- kolla-ansible-centos8s-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/1a6c136faf7842788bf6e2dade0400a9 : FAILURE in 54m 49s (non-voting)\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/4e8084c3b9624f71b788ccc41e59afbc : FAILURE in 49m 31s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/5282084bdaa04378a5468d77e9ed4049 : FAILURE in 1h 14m 35s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/7ac806f14b284b3a9815b0886a17d069 : FAILURE in 1h 09m 07s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/e6d9b6e9f0ea43609cfe29374c9d9620 : FAILURE in 1h 00m 46s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/329d4d06d1a74a8fb8c980e88c521208 : FAILURE in 39m 58s (non-voting)\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1c728282ab2e43678648c8d0f19e0ec3 : FAILURE in 33m 33s (non-voting)\n- kolla-ansible-centos8s-source-cells https://zuul.opendev.org/t/openstack/build/c1f9c4f3dcc8488599771f331b5cda0f : SUCCESS in 1h 25m 17s (non-voting)\n- kolla-ansible-centos8s-source-cephadm https://zuul.opendev.org/t/openstack/build/3f1212f821e942d19ca17d00f9eb2a45 : FAILURE in 1h 10m 11s (non-voting)\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/388947130df74e36b46dc910f8e1e73d : FAILURE in 34m 55s (non-voting)\n- kolla-ansible-centos8s-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/556e4df9a3b04c58b34a95ac29a5c6ec : FAILURE in 29m 06s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/ec17e692abba4a39a495b7dce0700469 : FAILURE in 1h 32m 40s (non-voting)","accounts_in_message":[],"_revision_number":47},{"id":"662135d2a7a63f3e00cda835bfe9e03de7c6b642","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-19 19:15:26.000000000","message":"Patch Set 47:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/cca411776a2a4991b821b032900c0c03 : FAILURE in 1h 12m 20s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/89526aae7bc3430991d9fd27a51279cc : TIMED_OUT in 2h 33m 24s (non-voting)","accounts_in_message":[],"_revision_number":47},{"id":"4b49e751394a1678c7b6546ece140004c2030413","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-06-24 10:48:52.000000000","message":"Uploaded patch set 48: Patch Set 47 was rebased.","accounts_in_message":[],"_revision_number":48},{"id":"9aa0c83d94127f5d7a670406d7c558c0610cc039","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-24 11:22:16.000000000","message":"Patch Set 48:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/16c5daec6c70469db5d8a058eb78bc43 : FAILURE in 19m 24s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/a53689ae038742beae1d604ba90c0913 : FAILURE in 17m 19s (non-voting)","accounts_in_message":[],"_revision_number":48},{"id":"193ce56585255887730c803310707ca094491924","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-06-24 13:30:51.000000000","message":"Uploaded patch set 49: Patch Set 48 was rebased.","accounts_in_message":[],"_revision_number":49},{"id":"f5893543abc2a9034636d95c131f76195d9f635d","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-06-24 18:05:43.000000000","message":"Patch Set 49:\n\nrecheck","accounts_in_message":[],"_revision_number":49},{"id":"fa08c18c29cfc7e6c6edfe9615ed4040141454b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-24 18:30:28.000000000","message":"Patch Set 49: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d94a18006267474a9fd5ef9d3b1d59ae : SUCCESS in 9m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67961486504340dc9a9c677e2200096a : SUCCESS in 9m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4d8f6a2f948d46879d24438bdd2eb150 : SUCCESS in 8m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/28f99dbdae984415ad24cadc8f451de2 : SUCCESS in 8m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/56943205c24f4c7889882f094fac4fed : SUCCESS in 9m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba0352fc36bb4860ad96b396a07d0dbb : SUCCESS in 8m 24s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/62fba51d045e49b2a3ca8f038e857baf : SUCCESS in 10m 44s","accounts_in_message":[],"_revision_number":49},{"id":"1a8b93297fa974ff8c2f417059a4e0416e30c352","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-07-09 10:38:43.000000000","message":"Uploaded patch set 50: Patch Set 49 was rebased.","accounts_in_message":[],"_revision_number":50},{"id":"0d77c032ba574bca9fe7a48978f7bccd1877735b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-09 11:11:28.000000000","message":"Patch Set 50:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":50},{"id":"17f4c328539774ccb1030b8f6bffbdeccc8b1475","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-09 11:12:05.000000000","message":"Patch Set 50: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":50},{"id":"4e170ed6958a267087009c18753719feca199501","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-07-09 11:18:36.000000000","message":"Patch Set 50:\n\nrecheck","accounts_in_message":[],"_revision_number":50},{"id":"7bc897717d1d3d0096c5d890363b40ecb198f6e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-09 14:27:41.000000000","message":"Patch Set 50:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/458d322abfb042aaa7a06c7bbcbd541b : SUCCESS in 6m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb787aa789294eb7998599fb43119844 : SUCCESS in 7m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1ebbc9d3d7b8413fa00250e16dd0d949 : SUCCESS in 6m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dbfd990fa348407b9ddc4348368d460b : SUCCESS in 6m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c440ebcba0f4402abd4f38701bfa25de : SUCCESS in 6m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/709c48375fcc4bbf85d1783dd6c112ae : SUCCESS in 7m 46s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/89317a4377cd4a3396421cab145af16f : SUCCESS in 8m 04s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/c6538394aee1429a8eeec8003600471e : FAILURE in 30m 21s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/5fcb8461558a41a084a2f4ca0c3f5c31 : FAILURE in 53m 28s (non-voting)\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/181c3ebfa1024e69885d16c45095f89c : FAILURE in 22m 19s\n- kolla-ansible-centos8s-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/4f6ce720027647848b213da857695a81 : FAILURE in 27m 57s (non-voting)\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/e6df63ae15184089adb8d04e6b67e393 : FAILURE in 22m 49s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/75554bfbdf9c40ae865078111a936f50 : SUCCESS in 1h 25m 06s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/51646273b6924018ba25251189d3e0eb : FAILURE in 56m 47s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/8ba9959c78a24586bd22f48b1981e933 : SUCCESS in 1h 05m 16s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/d7345d134b93475e9d3c9fdef8aa78b2 : FAILURE in 1h 07m 50s (non-voting)\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/a1264fe8eaa54153ba654fe78a80f138 : FAILURE in 22m 20s (non-voting)\n- kolla-ansible-centos8s-source-cephadm https://zuul.opendev.org/t/openstack/build/193787ce1b63409290fde995f1e9a423 : FAILURE in 1h 14m 05s (non-voting)\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/a774cabbabb74718b2b24ca037898c46 : FAILURE in 24m 21s (non-voting)\n- kolla-ansible-centos8s-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/e47ec29023144022a707121ca26147da : SUCCESS in 1h 40m 09s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/1d5e1605f9f546de89a0f93bb6a65bf5 : SUCCESS in 1h 34m 33s (non-voting)\n- kolla-ansible-centos8s-hashi-vault https://zuul.opendev.org/t/openstack/build/ea8172b121f34e348fbcf8533bae637d : SUCCESS in 4m 09s (non-voting)","accounts_in_message":[],"_revision_number":50},{"id":"cbbd2c89ce907bde3ca68f9eb590cf12a8d6b7d4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-09 15:23:23.000000000","message":"Patch Set 50:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/cb2bee95573d47c7a059db6ff5c5bcdf : FAILURE in 1h 30m 27s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/6795d85d381d45ddb920511c31db9884 : FAILURE in 1h 59m 38s (non-voting)","accounts_in_message":[],"_revision_number":50},{"id":"4d4d8ca920330836d87213ef5a7f6849e2ea9cf0","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-07-13 09:11:03.000000000","message":"Patch Set 50:\n\nrecheck","accounts_in_message":[],"_revision_number":50},{"id":"7b521353a302eb5273c776da6fecb360f2f24234","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-13 10:06:10.000000000","message":"Patch Set 50:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/4196b8ef607941e784976426c720a4ab : FAILURE in 18m 56s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/e963f523e68049629b07f5f2ef6d7ba0 : FAILURE in 53m 06s (non-voting)","accounts_in_message":[],"_revision_number":50},{"id":"6d030ce20a2c4b8e7f2f3335aff284e27213ce69","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-13 10:35:31.000000000","message":"Patch Set 50:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/df2d66eb2fa547e7a420898d6105ad29 : SUCCESS in 5m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea2bbff775cf47d99ad1a208afa2b204 : SUCCESS in 7m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b74d61eb72cd41549be979793d382768 : SUCCESS in 7m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/06685ab8489744dab6055f1b660c456c : SUCCESS in 6m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3c19e37316b54859a84c6cedd71d71f0 : SUCCESS in 7m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df98fc258bfd4ffaa4321d0d59787ecb : SUCCESS in 9m 14s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d444919bf55e4528a6368aca2c5e0e3e : SUCCESS in 6m 57s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/248fa838383d4073b52df13bdb8994df : FAILURE in 12m 55s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/94fc4557e01b43cbbaf09f8e3c23a9b9 : FAILURE in 15m 27s (non-voting)\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/499986088391477ebdd372181fdeed57 : FAILURE in 11m 11s\n- kolla-ansible-centos8s-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/cf4348db099848bbb55818741cb33781 : FAILURE in 14m 52s (non-voting)\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/89d183a3ed374030851226896958185b : FAILURE in 14m 53s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/ce98c4b7f15d461487409b2d718593d0 : SUCCESS in 1h 14m 51s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/beccdbe508ca4b3caf184d977cd14e59 : FAILURE in 43m 13s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/9364f9476df745eb8b029ca3e5737955 : SUCCESS in 54m 18s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/db101170ce694aeb83eff56173c7f44f : FAILURE in 12m 20s (non-voting)\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7066fe0745eb4baa84ba23496e479f37 : FAILURE in 11m 34s (non-voting)\n- kolla-ansible-centos8s-source-cephadm https://zuul.opendev.org/t/openstack/build/d930054911da4dbf8fbfe36d6152c973 : FAILURE in 18m 42s (non-voting)\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/a994ca2717e94407aa4a3d5366e7afb4 : FAILURE in 18m 31s (non-voting)\n- kolla-ansible-centos8s-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/eda94c97ae6c4b58a57f19e1b0c5eb10 : FAILURE in 1h 21m 49s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/e5ab1507ee054c6aa2a8cde2c6d0a08e : SUCCESS in 1h 05m 09s (non-voting)\n- kolla-ansible-centos8s-hashi-vault https://zuul.opendev.org/t/openstack/build/664ef88f9dbe445b9ad69b5c12355201 : SUCCESS in 4m 26s (non-voting)","accounts_in_message":[],"_revision_number":50},{"id":"db9d967dfd59f5ea972685f1907bbb5e3bbc3969","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-08-02 09:04:24.000000000","message":"Uploaded patch set 51: Patch Set 50 was rebased.","accounts_in_message":[],"_revision_number":51},{"id":"5a914d0faaa37cb2443b6cfe121779b9a784d2e4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-08-02 10:41:32.000000000","message":"Uploaded patch set 52: Patch Set 51 was rebased.","accounts_in_message":[],"_revision_number":52},{"id":"948c3a8fa94aa23e8abc48f29c0ac38f8ac9d19d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-02 12:31:11.000000000","message":"Patch Set 52: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8e0f543a77ff47c381f0c7ef7df07177 : SUCCESS in 8m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c8e7af0936743c3b5b6e74f9b06aa28 : SUCCESS in 6m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/07247799d82e40899a59f7cf33307fd5 : SUCCESS in 6m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8c65ace533ed4a519a991c5cb870d914 : SUCCESS in 6m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3104d0606e8b410b90b14c2f1545851a : SUCCESS in 8m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/364c98c96a9f4803a1c0f1fe9d83f717 : SUCCESS in 7m 20s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/87c8c435399c48b3977cd447935d9466 : SUCCESS in 8m 46s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/184f1c6d767c4dabafe4a6f2a2158c0a : SUCCESS in 1h 10m 56s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/fb5de75a53ea48709f817bd4731d271e : SUCCESS in 52m 36s (non-voting)\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/1214eb8c37b041c28514cf3b311bba22 : SUCCESS in 1h 02m 53s\n- kolla-ansible-centos8s-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/c087e216014148d381b5c065192113ca : SUCCESS in 1h 16m 17s (non-voting)\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/b47d4413b4fb49029d54370e145d33d1 : SUCCESS in 1h 13m 59s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/5ec1b8d5d8e24484825b67feb449fde1 : SUCCESS in 1h 12m 26s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/cf6a385ff3764f2b90005f8cf704d50a : FAILURE in 52m 43s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/21a7158c120f4810a3f003ae5303a72e : SUCCESS in 1h 09m 54s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/b85a87873b37410a8f8be25349ac79ae : SUCCESS in 1h 07m 58s (non-voting)\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1ef551200c17408bbcc76fa1ae5d8f34 : SUCCESS in 51m 27s (non-voting)\n- kolla-ansible-centos8s-source-cephadm https://zuul.opendev.org/t/openstack/build/4c876db35fe14a3ba53b8d537a9e7ca2 : SUCCESS in 1h 32m 15s (non-voting)\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/bdfece616588475aab1106d7f3585c21 : SUCCESS in 1h 18m 31s (non-voting)\n- kolla-ansible-centos8s-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/709421e5d8a04ce08f847a20e8a0f793 : FAILURE in 35m 47s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/cb13a354599449ffb1dcb1c114c80498 : SUCCESS in 1h 34m 11s (non-voting)\n- kolla-ansible-centos8s-hashi-vault https://zuul.opendev.org/t/openstack/build/5074aa5274a74859b071039480f439bb : SUCCESS in 4m 37s (non-voting)","accounts_in_message":[],"_revision_number":52},{"id":"138a014038892f527441aed44de205a56f3fd0ed","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-02 12:59:45.000000000","message":"Patch Set 52:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/52b609af9872451aae1578f1e481cd7c : SUCCESS in 2h 16m 11s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/18ecd5a5e0744272b8885a87c7fb5744 : FAILURE in 1h 25m 36s (non-voting)","accounts_in_message":[],"_revision_number":52},{"id":"e246c8322887655e670b73f3c76223828918c183","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-08-17 15:20:48.000000000","message":"Uploaded patch set 53: Patch Set 52 was rebased.","accounts_in_message":[],"_revision_number":53},{"id":"9d368d3693acf26fc584b944c94db9815a0a6778","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-17 16:56:47.000000000","message":"Patch Set 53: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c05972c51cda4d7ebbfeb3581eb29f03 : SUCCESS in 7m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7cee698011834bd3bb245a4d58cfa3cd : SUCCESS in 6m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/12aa1778279f4e599d50eb08802ffa5a : SUCCESS in 6m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/052451de6eaa4113b36c6f5e4ad237bf : SUCCESS in 6m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/01d27ddece744b1aa9bdf344c89867c9 : SUCCESS in 6m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc05c6afa8674f87ba72d080c76156ef : SUCCESS in 7m 46s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/0acd0445ddd14c2999420395838a87b3 : SUCCESS in 7m 52s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/85655e4531f04a99bec030041ed9bbe4 : SUCCESS in 1h 04m 18s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/eeb681b134d54b68986df4473ac61305 : SUCCESS in 53m 08s (non-voting)\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/74d7187020cc46b6bde43b927e3677c4 : SUCCESS in 53m 30s\n- kolla-ansible-centos8s-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/895c52895aa54430a3bc413d38bcb53d : SUCCESS in 1h 16m 44s (non-voting)\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/cd491eba7314475db600214fb75d29f9 : SUCCESS in 1h 03m 19s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/6c38cdf19511426491f455d3e4619d62 : SUCCESS in 1h 13m 15s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/bde3f1f99ba6476fb1d0645d223d637a : FAILURE in 48m 54s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/c00a865ae9c74cc4bcfe55d9d6546a05 : SUCCESS in 55m 39s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/8ba212bb4c4149d4a1fbb7b29e79f35f : SUCCESS in 1h 07m 45s (non-voting)\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/693e1a951cc74b68b78860682dac06ff : SUCCESS in 47m 49s (non-voting)\n- kolla-ansible-centos8s-source-cephadm https://zuul.opendev.org/t/openstack/build/d3677bf6cb3a4f92ae51518cf1862560 : FAILURE in 43m 40s (non-voting)\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/48b43b0b96664e6a9c3735aecd9f12dc : SUCCESS in 1h 13m 34s (non-voting)\n- kolla-ansible-centos8s-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/adaac69b0677403ca6c7e3523ec87094 : POST_FAILURE in 34m 47s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/6bc8f4cd0e324316b7447ff11c8dc550 : FAILURE in 1h 33m 20s (non-voting)\n- kolla-ansible-centos8s-hashi-vault https://zuul.opendev.org/t/openstack/build/ad33592f323d4b979742e453e4bfc1a2 : SUCCESS in 4m 23s (non-voting)","accounts_in_message":[],"_revision_number":53},{"id":"fbbf68ae668a8b31979ed9432929d7c2b2ef84b6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-17 17:02:19.000000000","message":"Patch Set 53:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/2844d77074174521a0937f17ee4da975 : SUCCESS in 1h 33m 10s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/946bdbb49658400a9d324f36d40b5236 : FAILURE in 1h 24m 17s (non-voting)","accounts_in_message":[],"_revision_number":53},{"id":"a5e26235f7d9fadffb2ad1680736577665885cc0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-08-17 21:22:57.000000000","message":"Uploaded patch set 54: Patch Set 53 was rebased.","accounts_in_message":[],"_revision_number":54},{"id":"ce4535e2d745d539e9e96d3bddba9008be1509d2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-18 00:15:49.000000000","message":"Patch Set 54:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/674ae777204947f7bcb26e9026f3fbf7 : SUCCESS in 1h 57m 20s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/7b395e78efca4ed2897acce4f9db0635 : FAILURE in 1h 38m 58s (non-voting)","accounts_in_message":[],"_revision_number":54},{"id":"73d5f15905d0129b947201d723b2740dc6e5cd18","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-18 00:38:06.000000000","message":"Patch Set 54: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eb79a357384d4949847e9f5f6821dc19 : SUCCESS in 6m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8b3a8c6d3ca471481e33a61378d4b93 : SUCCESS in 5m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cfc86c8297ea40b9b5569db5fd5da773 : SUCCESS in 5m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9c1f67762c3d43a9abe9522d68785d95 : SUCCESS in 5m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/01164426e34c4c8da753496e836a99e6 : SUCCESS in 6m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/958e4cb4ff2740769090869fe60e07a3 : SUCCESS in 7m 24s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f57ea7bd6b714bb693f733d8a3d63180 : SUCCESS in 6m 57s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/4a100d4fc88746d69018da1742d61ce8 : SUCCESS in 1h 01m 51s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/ed02a8e0fa184653aba758c10ffdadfb : SUCCESS in 51m 52s (non-voting)\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/151bfe2ec3104f8cb00ff5e3dc429ff0 : SUCCESS in 52m 16s\n- kolla-ansible-centos8s-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/afcfccf250814ef782e5fe37652594ff : SUCCESS in 1h 29m 02s (non-voting)\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/ae11efe32411454f85c7bec3fdec88cf : SUCCESS in 1h 03m 09s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/b8f23f667f964372bbbd2bc90882e654 : SUCCESS in 1h 19m 39s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/2e7ce166c7964e61924cbe48673601f6 : FAILURE in 42m 41s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/360d9d4b3c194b4db332180877ed7930 : SUCCESS in 51m 55s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/ebdee1efb9b440dd8aeae697ff376b6b : SUCCESS in 1h 13m 32s (non-voting)\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/2f1327a6fdad45988a65c5e4450262cd : SUCCESS in 41m 52s (non-voting)\n- kolla-ansible-centos8s-source-cephadm https://zuul.opendev.org/t/openstack/build/041a71f352f64a9196e3645daf6cd6c0 : SUCCESS in 1h 34m 36s (non-voting)\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/2ff3e0ef12994a52843194bf068b6855 : SUCCESS in 1h 11m 27s (non-voting)\n- kolla-ansible-centos8s-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/cf78617c73584b02a846e543ebe31d40 : SUCCESS in 1h 40m 02s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/930e29fd89ca446f95ad9ce982b18abe : SUCCESS in 1h 25m 50s (non-voting)\n- kolla-ansible-centos8s-hashi-vault https://zuul.opendev.org/t/openstack/build/a59e262fa079426fbdef54f719c4b7e3 : SUCCESS in 4m 22s (non-voting)","accounts_in_message":[],"_revision_number":54},{"id":"8e3e49a5864fb191e8ae8d58665c3d5c05f8806d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-08-19 10:07:22.000000000","message":"Uploaded patch set 55: Patch Set 54 was rebased.","accounts_in_message":[],"_revision_number":55},{"id":"d6039743b14f5499c43bbadefc14d7d6e9949633","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 10:07:52.000000000","message":"Patch Set 55: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":55},{"id":"50756610a258eb199ffa0a2c47dc2a1af491d62e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 10:07:54.000000000","message":"Patch Set 55:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":55},{"id":"23bffc56755797b849bfdace6e7fbdff119835b1","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-08-19 13:24:52.000000000","message":"Patch Set 55:\n\nrecheck","accounts_in_message":[],"_revision_number":55},{"id":"823557966d06062aa300f34ce2bd6a3822db4e50","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 15:22:44.000000000","message":"Patch Set 55:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/84d7db8ef50c42a1b839a4973887941f : SUCCESS in 1h 55m 46s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/c7b1fabf3d8b4598944dfb74699dfc4c : FAILURE in 1h 48m 56s (non-voting)","accounts_in_message":[],"_revision_number":55},{"id":"be6771ed7b9a4497ad9abbee35a3b93349f25798","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 15:25:49.000000000","message":"Patch Set 55: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c18bdce775e243ab9a4d0f94afd953ca : SUCCESS in 7m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/817acebfb0c6425e8767fd61d563a366 : SUCCESS in 6m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/883106a1f49d447ab6028e5eea2ae895 : SUCCESS in 8m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ee927893ffa1483d82a50830902d9c70 : SUCCESS in 7m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fc3ac16006c64b66b41eadfa6ec214ce : SUCCESS in 7m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd97b3c6e8e24330a8bb22904951dfda : SUCCESS in 7m 48s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d83a1a766fc54d6eb55a7d6c070fe352 : SUCCESS in 8m 57s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/d477f279a0f549ad96f16941766bed35 : SUCCESS in 1h 07m 29s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/012686073df543919bca99ec06b89c16 : SUCCESS in 57m 58s (non-voting)\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/674212ed6f2c49c79c52ca06dc7ea580 : SUCCESS in 1h 03m 03s\n- kolla-ansible-centos8s-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/5d55f64deb794e598eeb7f793e7fde0a : SUCCESS in 1h 16m 20s (non-voting)\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/e0032051c0c8405998d911c403da20d8 : SUCCESS in 1h 03m 45s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/43c0953128904da7b11d78a4843f1ab8 : SUCCESS in 1h 18m 55s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/3e4bd9ce07ab4bdda8f910ce10369ce4 : FAILURE in 45m 28s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ddf9294b248d46849289e7215de8be3c : SUCCESS in 59m 06s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/3656fb65d0334fe8a104900c1ab29361 : SUCCESS in 1h 05m 01s (non-voting)\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/0d8cef1a7b0a47ccb0b5f169983efd2a : SUCCESS in 50m 23s (non-voting)\n- kolla-ansible-centos8s-source-cephadm https://zuul.opendev.org/t/openstack/build/f7f64dbc9e3e432bbd7017e1b05b80be : SUCCESS in 1h 30m 11s (non-voting)\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/23cfc0c7e4fc4dcd937f729d21104ef7 : SUCCESS in 1h 12m 08s (non-voting)\n- kolla-ansible-centos8s-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/08c1d3376b564d38b64aeb04c8bc918b : SUCCESS in 1h 58m 09s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/f793b91041c64e6a9b0a61d90b9c5c82 : SUCCESS in 1h 38m 23s (non-voting)\n- kolla-ansible-centos8s-hashi-vault https://zuul.opendev.org/t/openstack/build/7124b53eb5cb43d7ae48bcbe76844a94 : SUCCESS in 4m 26s (non-voting)","accounts_in_message":[],"_revision_number":55},{"id":"a77df444f165674841cc207749617d9e5df5a3d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-08-19 19:28:46.000000000","message":"Uploaded patch set 56: Patch Set 55 was rebased.","accounts_in_message":[],"_revision_number":56},{"id":"2411dd0a91065db6aa0aab4b5626877e2e619285","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 21:26:13.000000000","message":"Patch Set 56: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/019f82205ff045e08dec25f32dd5f207 : SUCCESS in 7m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8b8d7f98b3c415aa587d04b8d4c617b : SUCCESS in 6m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cc2bb58f2a6c400d8ac213c305d7a036 : SUCCESS in 7m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/68f3f90955d344d183ae12726a625814 : SUCCESS in 6m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/47ea9d9c66fe4d1d9790097abf415da9 : SUCCESS in 6m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42b1743d44d54672b6db4ccc29d81014 : SUCCESS in 9m 32s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f7932f6588934db08e5ea2c36fae49e7 : SUCCESS in 9m 39s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/e2f87d187ebc40408cc08471b0a01555 : SUCCESS in 1h 19m 45s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/44227792e81e4d6599bf537cb14b6fcd : SUCCESS in 55m 51s (non-voting)\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/36fdce92d9aa4daa8bd1e7faebfa84c7 : SUCCESS in 1h 03m 14s\n- kolla-ansible-centos8s-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/6e922394301f47b29214377079383efb : SUCCESS in 1h 16m 25s (non-voting)\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/a30241a69a654293b1e3c6b748fecfdd : SUCCESS in 1h 11m 36s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/44f3878c46f349dd83dfc7c6560fa9ca : SUCCESS in 1h 32m 35s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/fa38c63e6792432ab110e4ed5b316425 : FAILURE in 57m 55s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/2cc843b0bdf645be82aeb25fe4cc88b3 : SUCCESS in 1h 14m 30s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/f5c3dc0efc3643d5b0aad3888d4b988d : SUCCESS in 1h 11m 55s (non-voting)\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/b9f9be8c01b14798863b8d4871013394 : SUCCESS in 56m 24s (non-voting)\n- kolla-ansible-centos8s-source-cephadm https://zuul.opendev.org/t/openstack/build/e7ab4898523c40b19ac618399e73cd0c : SUCCESS in 1h 39m 21s (non-voting)\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/b333f8036ebe481a81142b85057e401c : SUCCESS in 1h 24m 23s (non-voting)\n- kolla-ansible-centos8s-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/00c3885ed2f5404288e3bab375624feb : SUCCESS in 1h 53m 28s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/16c6a8c1a9b74232ab9aa73e6babbb1d : SUCCESS in 1h 37m 30s (non-voting)\n- kolla-ansible-centos8s-hashi-vault https://zuul.opendev.org/t/openstack/build/c42c0d25c9fa45fa83950b184c822823 : SUCCESS in 4m 37s (non-voting)","accounts_in_message":[],"_revision_number":56},{"id":"ab7548f6b4c7b44e31acdc53f5514ed6e7138741","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-19 21:30:40.000000000","message":"Patch Set 56:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a1f38f45a42442a0bd978f1cd6ac3b33 : SUCCESS in 1h 59m 33s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/d28b30640f4c44328b707d562980e35b : FAILURE in 1h 15m 05s (non-voting)","accounts_in_message":[],"_revision_number":56},{"id":"941ae8bfffea5e19687a2bb5000a5819462600d1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-09-15 19:57:14.000000000","message":"Uploaded patch set 57: Patch Set 56 was rebased.","accounts_in_message":[],"_revision_number":57},{"id":"a672fd2c7ed413fef0f069f4d9bdf8fc87833d66","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-15 21:36:04.000000000","message":"Patch Set 57: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e047cce85e34450592dea22f95be2aba : SUCCESS in 6m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c73d4a2ecba346dfb81caf70db88a431 : SUCCESS in 6m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/441251c9e0be4115bd36ecde63e86355 : SUCCESS in 7m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bb64082f94b040f9bf6007d189667ebb : SUCCESS in 6m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d4081b04e1104cdc9357bfb768b63652 : SUCCESS in 7m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52076ff0efb74de280599df14743b7d8 : SUCCESS in 6m 59s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/e6c8920a80c34eeebb23ee5609c9c365 : SUCCESS in 8m 56s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/9cedefc6372042b49c73f21a3f7c6208 : SUCCESS in 1h 11m 39s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/c3f1d7a2e7ea45b98c0b75544c78dc8f : SUCCESS in 1h 00m 51s (non-voting)\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/df40171a28da41838c4e54afadf5f3e9 : SUCCESS in 49m 51s\n- kolla-ansible-centos8s-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/eab797ffcc3a4c158f07704607715eb2 : SUCCESS in 1h 22m 00s (non-voting)\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/1e4443bfc8bf413382e0973b5f024c74 : SUCCESS in 1h 09m 36s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/fd5d5b84167d4bf0a5b24923bc3958f2 : SUCCESS in 1h 11m 35s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/3bf6d644a6204c4ba918ab0790789694 : SUCCESS in 1h 07m 01s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/36c4327487e74542b38ae61d2db06ca5 : SUCCESS in 1h 04m 28s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/00982c78fe17437ba08aedef90641c28 : SUCCESS in 1h 25m 53s (non-voting)\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/99fb5207fd5242d79e3163c2d8cfdef2 : SUCCESS in 54m 30s (non-voting)\n- kolla-ansible-centos8s-source-cephadm https://zuul.opendev.org/t/openstack/build/1ae6d5b2e8ec45cab3e5ddb6cb906ead : SUCCESS in 1h 29m 26s (non-voting)\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/88470b32a9084a4a9d101533059b7a2d : SUCCESS in 1h 19m 03s (non-voting)\n- kolla-ansible-centos8s-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/60860203bd8748559402e50e383eb971 : FAILURE in 1h 33m 22s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/d4999ba151fb454da11ac8e62c80f8e9 : SUCCESS in 1h 18m 03s (non-voting)\n- kolla-ansible-centos8s-hashi-vault https://zuul.opendev.org/t/openstack/build/06faabcae3bf4bb28504777844c6d099 : SUCCESS in 4m 36s (non-voting)","accounts_in_message":[],"_revision_number":57},{"id":"8e1c6c12c999ff744591b08b7a5e67e93a1f190b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-15 22:01:08.000000000","message":"Patch Set 57:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/1da5c62ea1a54a25b1cd8d940e4ea63f : SUCCESS in 1h 37m 16s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/c0521ad307bc4085ae2164254f39f660 : SUCCESS in 2h 01m 31s (non-voting)","accounts_in_message":[],"_revision_number":57},{"id":"864abe6708b3247048cda85dc31c2aa47f7cf9da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-09-21 15:30:40.000000000","message":"Uploaded patch set 58: Patch Set 57 was rebased.","accounts_in_message":[],"_revision_number":58},{"id":"b2be06ca1d47a1302c2fc57a35b426a5f29191e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-21 17:10:11.000000000","message":"Patch Set 58: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d0835d7a2d9d4be2a0fbcb3c16c51f87 : SUCCESS in 7m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07c1838069fe47d9be0c17b52023c1c2 : SUCCESS in 9m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a75ff706372b428595e4bdf3afd55e33 : SUCCESS in 7m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/81f353b12e684c22bb2f39ad216d0c02 : SUCCESS in 8m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/acfc0c481bd940379fa7ee9e6ac8c0f5 : SUCCESS in 8m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00997ea108bb4a15977079291d5b7dc8 : SUCCESS in 7m 44s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/99f3fa7f1aee4929a0a469b9de21ebdc : SUCCESS in 8m 18s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/1c5a2a38f5bf4343b767268088f69883 : FAILURE in 29m 55s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/af2347f35cf946209b713ce411f48987 : SUCCESS in 1h 01m 18s (non-voting)\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7b865acf4d5e422e9d5f21a68a172611 : FAILURE in 20m 35s\n- kolla-ansible-centos8s-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/ff9b5f74a48340e0b685722070ef116e : FAILURE in 30m 33s (non-voting)\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/7bc827459ba34506813823c128cb0ab1 : SUCCESS in 56m 08s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/a2d37c08d51144269f1984fe78857731 : SUCCESS in 1h 15m 53s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/ff223695dc5d4c499a992b99429b74cd : SUCCESS in 1h 04m 55s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/bc7b45ce9ca847aeae69f47520fe4038 : FAILURE in 20m 49s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/29266f6a09bc4d07bf03b2b08fa5d9ac : SUCCESS in 1h 09m 39s (non-voting)\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/1813d91fa20c406c8210b9e3c2ee9e68 : SUCCESS in 55m 03s (non-voting)\n- kolla-ansible-centos8s-source-cephadm https://zuul.opendev.org/t/openstack/build/ce8187a6f4924db3ab5a51d6f30d9fd4 : SUCCESS in 1h 30m 06s (non-voting)\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/5a085ee4285c46918c3369f59e966d19 : SUCCESS in 1h 12m 30s (non-voting)\n- kolla-ansible-centos8s-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/4cec913b5de84f53955fac436297c853 : FAILURE in 30m 03s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/ffc7df4c681f4f9a8cf1ab10bd120f7d : SUCCESS in 1h 27m 19s (non-voting)\n- kolla-ansible-centos8s-hashi-vault https://zuul.opendev.org/t/openstack/build/bee1d8bfebe54c49838c09dd4b1a5369 : SUCCESS in 5m 02s (non-voting)","accounts_in_message":[],"_revision_number":58},{"id":"48bcbc78e749f01cf97e6c82f8c07a87d080a6a2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-21 18:01:20.000000000","message":"Patch Set 58:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/3fc2ad3c648a42668ae98fedcbb35568 : SUCCESS in 2h 09m 44s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/f24af21ac42c4f9db49f64a1d458c006 : FAILURE in 2h 25m 28s (non-voting)","accounts_in_message":[],"_revision_number":58},{"id":"da6cd42cd75267c568462bb995601cac6c9b1fb2","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-09-22 06:57:38.000000000","message":"Patch Set 58:\n\nrecheck","accounts_in_message":[],"_revision_number":58},{"id":"21dfadd244788d28d2ca4b9846c7d59eabe80df3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-22 09:13:28.000000000","message":"Patch Set 58:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/2c1a8dab781d4314baf591550f639e74 : SUCCESS in 1h 49m 55s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/82ed3aa0f84247188bea3d90dc8bc669 : SUCCESS in 1h 54m 09s (non-voting)","accounts_in_message":[],"_revision_number":58},{"id":"4f6b518c04ebbc5bb67646bd6a5b1b5e39c092ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-22 09:49:50.000000000","message":"Patch Set 58:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3d6156320e984227b16a92180027da29 : FAILURE in 5m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/61f06f7f370e409b95af7553e0aec4e5 : SUCCESS in 5m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2ed216436b6f440e830db0ae910bded9 : SUCCESS in 8m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f58b975e30184748ae5cd0a89d969056 : SUCCESS in 6m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fd16f78ca8b1424b9014554bc442dfe9 : SUCCESS in 6m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7b46366b47745a2ac6984c83ffeeca9 : SUCCESS in 7m 00s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/5a551d8de3264b648ff97179329dccae : SUCCESS in 7m 55s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/28129cebd1f443c0897de3681b92fc09 : SUCCESS in 1h 20m 51s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/4189fc36c5d04cc2ab02bfc5b8920af1 : SUCCESS in 1h 00m 59s (non-voting)\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e2cf0e1b44974c7aafdca81fde636632 : SUCCESS in 1h 00m 47s\n- kolla-ansible-centos8s-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/ab2173903953485db2d4bc54cad2737d : SUCCESS in 1h 25m 13s (non-voting)\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/8b4f03f3eb6a4c479dc85c3498065676 : SUCCESS in 1h 16m 17s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/56cd6593c9c6473e9e554775dd7b873c : SUCCESS in 1h 11m 05s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/3796b796bd654ada98a94df5369cd02c : SUCCESS in 1h 06m 05s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/873101ad85ab42eebc2b0650d4347f39 : SUCCESS in 1h 13m 34s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/33d827cf9a8542b886ab84b78b51e82f : SUCCESS in 1h 09m 18s (non-voting)\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/cd623d6828664b4e8204c927fb16aa7f : SUCCESS in 57m 07s (non-voting)\n- kolla-ansible-centos8s-source-cephadm https://zuul.opendev.org/t/openstack/build/5c45f27cd9dd4af982aedc2c1696e9c0 : SUCCESS in 1h 20m 54s (non-voting)\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/586f84c7638246f68419f828ef4f115c : SUCCESS in 1h 13m 38s (non-voting)\n- kolla-ansible-centos8s-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/d66f806d2ce04ca39a1bc0240d5364de : POST_FAILURE in 38m 28s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/aca7b872226d40c4b9ed35ef072bb50d : SUCCESS in 1h 27m 35s (non-voting)\n- kolla-ansible-centos8s-hashi-vault https://zuul.opendev.org/t/openstack/build/3500ff9b114f499ba0986d911489707b : SUCCESS in 5m 05s (non-voting)","accounts_in_message":[],"_revision_number":58},{"id":"def01b1369466d19d4f64d1246c0611810faecc2","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-09-22 10:21:15.000000000","message":"Patch Set 58:\n\nrecheck","accounts_in_message":[],"_revision_number":58},{"id":"10506dc6ba8e6d6241f0fb04068d43237ba5c2c4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-22 12:01:39.000000000","message":"Patch Set 58:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/7adf88748b9b4621a99e112a0b29faea : SUCCESS in 1h 16m 58s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/407f267006c546bcbb04b0eaa8b45c43 : SUCCESS in 1h 31m 44s (non-voting)","accounts_in_message":[],"_revision_number":58},{"id":"72e0174673c94093070590414f9c27165161c81c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-09-22 12:06:22.000000000","message":"Patch Set 58: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0e3dc4ea1c6440899ac999d181c0c145 : SUCCESS in 8m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7a9ad6233e245d58837d9806faeaf71 : SUCCESS in 6m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c6f8ab2cf8b043a5b13377e3b55db809 : SUCCESS in 6m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b2d3361977a14fd79d93c659f9e3901f : SUCCESS in 6m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6aff042b5c054a7a8546d6d44081aee0 : SUCCESS in 6m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8adad659da53449093951253df991894 : SUCCESS in 7m 20s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/1c959cb2bf644a6aa588371ef8fbba4e : SUCCESS in 8m 37s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/59f7aeaed3354b61961316d23f758395 : SUCCESS in 1h 16m 49s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/e37d0aca3d0c4baab39c1c1b23306410 : SUCCESS in 1h 06m 36s (non-voting)\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/128d37b6fdce434ea5f071a0968ac995 : SUCCESS in 55m 39s\n- kolla-ansible-centos8s-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/a230672194054fa69255cf24cb8d585d : SUCCESS in 1h 31m 20s (non-voting)\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/f4ea538965674580b29d470230e5f3b4 : SUCCESS in 1h 00m 26s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/2cc106b13ebd4fa69b4007650ce906da : SUCCESS in 1h 13m 46s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/2832dec91909453b848df0fc1c0e5942 : SUCCESS in 1h 06m 01s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/5d80d34b1e8948bd8f25203091265ff0 : SUCCESS in 1h 09m 44s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/dec12c7a53a34154a04eaae2344add76 : SUCCESS in 1h 12m 43s (non-voting)\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/4c7ff27a5df14254ac293e0d5f15dc1c : SUCCESS in 57m 07s (non-voting)\n- kolla-ansible-centos8s-source-cephadm https://zuul.opendev.org/t/openstack/build/ce23ed194bb84573bf07526d28c6c435 : SUCCESS in 1h 32m 31s (non-voting)\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/188f7201369841dab76b2207694b4513 : SUCCESS in 1h 08m 18s (non-voting)\n- kolla-ansible-centos8s-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/b4f6c4ffab844ba1abb71787233b7332 : POST_FAILURE in 38m 02s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/feaa5b0ec7ec4170a5c86b23e23d3a3c : SUCCESS in 1h 30m 35s (non-voting)\n- kolla-ansible-centos8s-hashi-vault https://zuul.opendev.org/t/openstack/build/c72e1749f67a442cb06e35f6eb11bf2b : SUCCESS in 4m 33s (non-voting)","accounts_in_message":[],"_revision_number":58},{"id":"21756aaff57bb4f22b5fbaaa7528be8e121ec983","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-10-01 14:36:36.000000000","message":"Uploaded patch set 59: Patch Set 58 was rebased.","accounts_in_message":[],"_revision_number":59},{"id":"0f0a7a40d40e579d27e4d0d2d9f91f66861198f1","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-10-01 14:38:00.000000000","message":"Topic change-780970-41 removed","accounts_in_message":[],"_revision_number":59},{"id":"2016e8e54ee4b9144f7008ef13bf77480eaa25c4","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-10-01 14:38:04.000000000","message":"Topic set to proxysql","accounts_in_message":[],"_revision_number":59},{"id":"2f6a495d09f8d061a727e3fc4c199e4da6d6ec43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-01 16:27:50.000000000","message":"Patch Set 59: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8e3c58af115f42598a1b137bbb0921aa : SUCCESS in 8m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/256e72e54f1a4afb9cd546d329ac2f80 : SUCCESS in 8m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0c1ceef67afd4d74bb42f7be81c3ef8d : SUCCESS in 8m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d07a66e2009c4083a9763d95efef1d6e : SUCCESS in 9m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/897279c44df54422bdb4b43890efa067 : SUCCESS in 8m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50befee506e3452290dae97aa1a7dab2 : SUCCESS in 6m 55s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/c60624a17c464dacaafcf41b1decc6a9 : SUCCESS in 10m 41s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/594b76ffbec44c468d2528572ae20741 : SUCCESS in 1h 05m 01s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/57c97b97ea36401f85e4003b6a0dfa8a : SUCCESS in 55m 15s (non-voting)\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/30682e4e8fc648ce98bd384cf4115881 : SUCCESS in 51m 04s\n- kolla-ansible-centos8s-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/ab0e73eb1c4b4786bc41c6cf8ca55f19 : SUCCESS in 1h 27m 42s (non-voting)\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/00c9f74df147499e8782660d9e8018c1 : SUCCESS in 1h 11m 32s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/bc0345a9de2342558331e330ce4b7feb : SUCCESS in 1h 22m 41s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/253c0b1821d349ebac28fc69c725cfce : SUCCESS in 1h 12m 10s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/e064b64ab42e43e997e58fcf7d9e1f1b : SUCCESS in 1h 18m 34s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/ba7e619e09c8478892814ffd404368b8 : SUCCESS in 1h 05m 59s (non-voting)\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/7adc35c498424d128ceddc34ac089997 : SUCCESS in 55m 31s (non-voting)\n- kolla-ansible-centos8s-source-cephadm https://zuul.opendev.org/t/openstack/build/3acd6e0e9fec4543843527276cb25c6b : SUCCESS in 1h 40m 04s (non-voting)\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/5998e7722eb841bd91f94f6abd0f845c : SUCCESS in 1h 13m 17s (non-voting)\n- kolla-ansible-centos8s-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/09949fd3150d42198ed5274ee10475fe : FAILURE in 44m 25s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/a940aa68f2dc434d932e1a82354e4899 : FAILURE in 25m 01s (non-voting)\n- kolla-ansible-centos8s-hashi-vault https://zuul.opendev.org/t/openstack/build/06e0e594778749f9aef340bf6c56dbd3 : SUCCESS in 5m 43s (non-voting)","accounts_in_message":[],"_revision_number":59},{"id":"4da54a5bf85614da3e9378b89f97cd1be851c6ec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-10-22 11:38:59.000000000","message":"Uploaded patch set 60: Patch Set 59 was rebased.","accounts_in_message":[],"_revision_number":60},{"id":"b6babbff6ffd00ff9e71b0cdecd520afa18802fc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-10-22 13:15:46.000000000","message":"Uploaded patch set 61: Patch Set 60 was rebased.","accounts_in_message":[],"_revision_number":61},{"id":"eaf41715dfa717f2b1b434776b1a4947539ac3c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-22 14:56:23.000000000","message":"Patch Set 61: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/14d666d931b64193aab337ffb404aa31 : SUCCESS in 6m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3535a05e30cc4ebc921ab77f5069902b : SUCCESS in 7m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4f0447b05b0049949686553026412ff9 : SUCCESS in 7m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5238cd98da904c7890ad448251853058 : SUCCESS in 6m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/475fc2f7ca7245f18dee440e3c6ccd70 : SUCCESS in 9m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c01500e38df94946b1d0cc19e0a108ba : SUCCESS in 6m 21s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/0fcd6757cc34448b89298654f8acccec : SUCCESS in 10m 21s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/22d1dfe8de6a421ba586f8af1fa67547 : SUCCESS in 1h 04m 39s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/dbdf3bf962fa46d58695f3bd857767cd : SUCCESS in 1h 00m 42s (non-voting)\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/3cc6c0f2b414459cb3c88a9f61f8a81b : SUCCESS in 48m 41s\n- kolla-ansible-centos8s-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/48b093a7c97b4209aaea4df00292e496 : FAILURE in 1h 02m 43s (non-voting)\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/2fdf3f7ffe31485a8ebd95d358661ceb : SUCCESS in 52m 58s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/1e991678341b402792a37897703f447a : SUCCESS in 1h 17m 52s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/6d55357920534296b40ddf344c4e398e : SUCCESS in 1h 16m 55s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/bc2f874d685a4af18e341e278a1d18fe : SUCCESS in 1h 04m 15s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/1e4db8db10a44f04a52130b0c1ba1309 : SUCCESS in 1h 11m 33s (non-voting)\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/cf12d2854b0c44fd86611e87eb154be5 : SUCCESS in 52m 40s (non-voting)\n- kolla-ansible-centos8s-source-cephadm https://zuul.opendev.org/t/openstack/build/afefcf5b1b4349b5a8cafe66be6b4b87 : SUCCESS in 1h 32m 29s (non-voting)\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/620570388c2a4eb9bcf555ff41daaca5 : SUCCESS in 1h 16m 28s (non-voting)\n- kolla-ansible-centos8s-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/ed18619bf3b34a6095aafbfa9eecd4a7 : SUCCESS in 1h 34m 57s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/6dbfa8442f8c49ed83090566dd360b50 : SUCCESS in 1h 31m 40s (non-voting)\n- kolla-ansible-centos8s-hashi-vault https://zuul.opendev.org/t/openstack/build/e0ebaf606aea4d0eb06d288f3bbeea84 : SUCCESS in 6m 10s (non-voting)","accounts_in_message":[],"_revision_number":61},{"id":"eb99bcd1473b2c62950bebd8eefa853185a7124e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-22 15:33:49.000000000","message":"Patch Set 61:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f0182642373a4feb8724304704c6e36e : SUCCESS in 1h 48m 14s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/4d3d89232d904ab89f1b8e285ece9eaa : SUCCESS in 2h 15m 36s (non-voting)","accounts_in_message":[],"_revision_number":61},{"id":"1d1a6268210cf8e8e2ad3752ae6e7f8262b7247c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-01-11 15:10:33.000000000","message":"Uploaded patch set 62: Patch Set 61 was rebased.","accounts_in_message":[],"_revision_number":62},{"id":"6fa0fd8f0744011863232cb632c3ad92cee2afa7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-01-11 15:26:28.000000000","message":"Uploaded patch set 63: Patch Set 62 was rebased.","accounts_in_message":[],"_revision_number":63},{"id":"0d7cf1383242cd95192f449a43a16503f1e9c147","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-11 16:11:18.000000000","message":"Patch Set 63:\n\nBuild failed (ARM64 pipeline).\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/c5386d73e87a45848930f55ac16ccdf5 : FAILURE in 29m 42s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/c71a3353bd024d88b564f685aa62dc75 : FAILURE in 30m 44s","accounts_in_message":[],"_revision_number":63},{"id":"e8dc5e25b7047778677610ce7cca0e691542ccb8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-11 18:20:19.000000000","message":"Patch Set 63: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/07d87201be00461781ff9086e140de7f : SUCCESS in 8m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/72098b051d43461bb52226681dbd4e0a : SUCCESS in 7m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bcac77bda9694ec9bafc6dafbb9b98e7 : SUCCESS in 9m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b63f4d9bb4b547f9a6ad876f06105b09 : SUCCESS in 8m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/336abd7267db4701a6fe39657f3f1352 : SUCCESS in 8m 06s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/31aaa5c59a8b4456b64e4621338384c9 : SUCCESS in 9m 48s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/72dc10ea2b87471096852a56a5fb6404 : FAILURE in 29m 49s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/417d8944ee77419d9eb7633e838b58e3 : FAILURE in 26m 56s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c7488b9506cc412ebf084a6236d840b9 : FAILURE in 20m 50s\n- kolla-ansible-centos8s-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/7d56e1394eeb432f84f309dda35c9a02 : FAILURE in 29m 49s (non-voting)\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/c975c99a837e4bd3a02cda19c7aeecc3 : FAILURE in 24m 07s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/ebb6c0f58be444e78fac1923c092dcc6 : FAILURE in 1h 09m 17s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/8e2eeacedbd0480d92dfc2c6fb369590 : FAILURE in 57m 44s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/df2bf5ed002c4926a9f1431216e94a99 : FAILURE in 52m 48s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/a4b8c790655b47e2ac597b371c5d993c : FAILURE in 33m 51s (non-voting)\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/482ea5a3ed7f40c9aba44ca2a1385bdb : FAILURE in 24m 35s (non-voting)\n- kolla-ansible-centos8s-source-cephadm https://zuul.opendev.org/t/openstack/build/5a7600d8ea934b908391711bde040187 : FAILURE in 37m 57s (non-voting)\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/338f3e7029c942b7bc07140f0bd4a8f6 : FAILURE in 26m 15s (non-voting)\n- kolla-ansible-centos8s-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/1bcdf4f468db422ebb315b0915db63c7 : FAILURE in 1h 17m 51s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/38d5a5197fc949078cd6b0fc78c18fe6 : FAILURE in 1h 05m 08s (non-voting)\n- kolla-ansible-centos8s-hashi-vault https://zuul.opendev.org/t/openstack/build/85e7eb5023c14ab18cb336abf5a54291 : SUCCESS in 5m 28s (non-voting)","accounts_in_message":[],"_revision_number":63},{"id":"1626cacc728d6973df3697853899115a4dcb63f0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-01-13 14:06:20.000000000","message":"Uploaded patch set 64: Patch Set 63 was rebased.","accounts_in_message":[],"_revision_number":64},{"id":"7b505eddf60ef2e0f2471fee16d6d8d2204723f0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-13 15:13:12.000000000","message":"Patch Set 64:\n\nBuild failed (ARM64 pipeline).\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/854436ae1827427aa653634699854f95 : FAILURE in 38m 27s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/b15ca308b49642af9cc4d9f323e4133a : FAILURE in 34m 34s","accounts_in_message":[],"_revision_number":64},{"id":"78a00a4a859ec0bbaabbc2da5c4516325b597507","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-13 17:34:27.000000000","message":"Patch Set 64: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f5aee112621242e2b9a1345631ed0e3f : SUCCESS in 6m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3bfd63f4c9544d5d865413e8f7ab02c7 : SUCCESS in 7m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/df43dbc39dde4049ae676ce476f796c3 : SUCCESS in 9m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bd5500fe951f43f28318c2006583e5d6 : SUCCESS in 11m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89b745a01ffb49ae9811f313b0189ea6 : SUCCESS in 8m 01s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/5c96d2dbc64441ab82a48d1576c5b024 : SUCCESS in 10m 11s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/58f4920a8c49485a8c99f514b08f8ec9 : SUCCESS in 1h 11m 25s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/b5bde719031344118f93e252832e0ce0 : SUCCESS in 55m 00s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d36bbf70c45a446aae0e53055edad995 : SUCCESS in 52m 59s\n- kolla-ansible-centos8s-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/d684d1aeb5374220a44598da8ebdd5ad : SUCCESS in 1h 18m 26s (non-voting)\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/81b90c5b579b48d389072db6b4ea28e1 : SUCCESS in 1h 03m 16s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/4a8ec936ecfa49aa9f626996c7a00015 : SUCCESS in 1h 46m 26s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/a6157e2e5d9144a6ac7cb30cc0bd1e46 : SUCCESS in 1h 02m 05s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/c62de1c985194510be23f3f7a464df84 : SUCCESS in 1h 17m 10s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/ba0f14d23a82481a8a639db17a56b358 : SUCCESS in 1h 07m 59s (non-voting)\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9b2eeaf9875242c68781209064cc877d : SUCCESS in 1h 02m 24s (non-voting)\n- kolla-ansible-centos8s-source-cephadm https://zuul.opendev.org/t/openstack/build/ba93c45ece004b53b269a194446a8d3b : SUCCESS in 1h 33m 37s (non-voting)\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/d3c71f668743441ba7f2a14838140397 : SUCCESS in 1h 18m 03s (non-voting)\n- kolla-ansible-centos8s-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/acd8316fc97a463487cda3e17b0fafae : SUCCESS in 1h 50m 20s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/72b438d797c643a18e2f9b6ac2dc206a : SUCCESS in 1h 42m 59s (non-voting)\n- kolla-ansible-centos8s-hashi-vault https://zuul.opendev.org/t/openstack/build/9e80aa6040954b95983b0528c27ae108 : SUCCESS in 5m 07s (non-voting)","accounts_in_message":[],"_revision_number":64},{"id":"81210cf5042aac17943bffecfb83c061a1cc53b4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-01-20 08:43:26.000000000","message":"Uploaded patch set 65: Patch Set 64 was rebased.","accounts_in_message":[],"_revision_number":65},{"id":"94d34886a6f91968e78d88e1a8260875a3dd774a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-20 09:09:18.000000000","message":"Patch Set 65:\n\nBuild failed (ARM64 pipeline).\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/dfbdfea66faf4d5b9f3115d916da05d0 : FAILURE in 23m 17s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/061fa74bfdf24120a3fd57c1e2a2e63d : FAILURE in 22m 56s","accounts_in_message":[],"_revision_number":65},{"id":"e680ac6b8e5c9dc38c2bb10b74d905f1bcb126bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-20 10:10:07.000000000","message":"Patch Set 65: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5fdf72bfafea46d88db7fce123f0a1ac : SUCCESS in 7m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62883ae3249e4f9b982804cb2ae60c9a : SUCCESS in 6m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/33e4f29c602f4232be66774406b6aa25 : SUCCESS in 6m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d7709838254e406a8a402c952e886923 : SUCCESS in 5m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56eac0b628134d4c96927a2f67046181 : SUCCESS in 6m 58s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/8431848cf790416eaa6f1d62573ea7cc : SUCCESS in 7m 59s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/03f2034a444f4a78837a0b8e9539dff4 : RETRY_LIMIT in 2m 19s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/0466c28d73734381a335ede0bf6e6360 : SUCCESS in 54m 12s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/3c42e0492d1d42d8944a516f1594e709 : SUCCESS in 54m 26s\n- kolla-ansible-centos8s-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/f3d6439fae5646faaa84d75137f83f0a : RETRY_LIMIT in 2m 46s (non-voting)\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/0b95c49cd438488a84440f87d3752a6d : SUCCESS in 55m 53s (non-voting)\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/55fde407c3a14ba095a09ca8a47bb848 : SUCCESS in 1h 09m 36s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/c9f313ba68534729a5c757cd1a31c05b : SUCCESS in 59m 18s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/8370922db49846ae85ef9637eadd14cd : RETRY_LIMIT in 2m 34s (non-voting)\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/30e54909bc6b4293944d45cbee6c0fc6 : SUCCESS in 49m 30s (non-voting)\n- kolla-ansible-centos8s-source-cephadm https://zuul.opendev.org/t/openstack/build/1aa1219ef06742d38c718e607561fa5b : RETRY_LIMIT in 2m 47s (non-voting)\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/916cb2f4781a4f63afe9bee5ac14f950 : SUCCESS in 1h 16m 19s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/7223de75de7049ba9b47f5f765118588 : FAILURE in 1h 20m 12s (non-voting)\n- kolla-ansible-centos8s-hashi-vault https://zuul.opendev.org/t/openstack/build/0f6d8f041ac244209ecff890cb995048 : RETRY_LIMIT in 2m 15s (non-voting)","accounts_in_message":[],"_revision_number":65},{"id":"b917422fc6a9c1c1b4da1deda7f2b01edf077032","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-01-20 17:10:00.000000000","message":"Patch Set 65:\n\n(1 comment)","accounts_in_message":[],"_revision_number":65},{"id":"182d6a4d5d44163d711eb4641a416f9ac8c4ff5f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-20 17:47:41.000000000","message":"Patch Set 65:\n\nBuild failed (ARM64 pipeline).\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/36b4e59323574312963733328146e956 : FAILURE in 27m 44s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/98313d43c33746ca8599995a1e039fe4 : FAILURE in 31m 17s","accounts_in_message":[],"_revision_number":65},{"id":"fe9e2a09b8bbadb97e706b9d9e75498ae8424767","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-20 19:00:17.000000000","message":"Patch Set 65: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d88cc8a90a084103af43a3747b36e4de : SUCCESS in 8m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7337a4d904bc4e9399d79622da7ff7a5 : SUCCESS in 7m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b797576df1af40e0bbde2a6269198888 : SUCCESS in 8m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6fdce0d0181549b68ac71d04ee868b9c : SUCCESS in 9m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42fed259fa4f4993902a13c58570c695 : SUCCESS in 9m 29s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/3956791aca734554a8bdeab9b79b4456 : SUCCESS in 9m 52s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/f7383d94d9f54003af7644383c073f1d : SUCCESS in 1h 22m 08s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/2a172108ac54428fb3b8631767bd2abe : SUCCESS in 1h 02m 06s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b003caca7b834787b0718d779c04752e : SUCCESS in 54m 56s\n- kolla-ansible-centos8s-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/26781d7b59284443bd42efe75ef203b9 : SUCCESS in 1h 30m 31s (non-voting)\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/1dc0dfeabf304877beade444941078d4 : SUCCESS in 1h 08m 22s (non-voting)\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/4cdf28b699b84fcab2e2a7acfbaf1ad8 : SUCCESS in 56m 42s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/9fcee1767ba44a13a90e8ab0d272f39d : SUCCESS in 1h 04m 17s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/f895324c2e664880990003e81dbe819b : SUCCESS in 1h 34m 10s (non-voting)\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9849a54e5c6441c4ac3568424ac2e1b2 : SUCCESS in 46m 23s (non-voting)\n- kolla-ansible-centos8s-source-cephadm https://zuul.opendev.org/t/openstack/build/e767f6e85ac74def9d5502465b7f9a00 : SUCCESS in 1h 33m 31s (non-voting)\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/667cd576bd7d4310b13f339aae21a893 : SUCCESS in 1h 26m 02s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/e5e3db109e244d2a9515cb6d49a5640f : SUCCESS in 1h 27m 27s (non-voting)\n- kolla-ansible-centos8s-hashi-vault https://zuul.opendev.org/t/openstack/build/dd390e0f6e1941e7b8d822d6216aa3cc : SUCCESS in 5m 43s (non-voting)","accounts_in_message":[],"_revision_number":65},{"id":"2e3bb60c35c38298454b0b8c109d750383321f4e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-01-28 09:13:48.000000000","message":"Uploaded patch set 66: Patch Set 65 was rebased.","accounts_in_message":[],"_revision_number":66},{"id":"b79100d353f23eb2dca8dd79de21e3e69bbeacc4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-28 09:55:54.000000000","message":"Patch Set 66:\n\nBuild failed (ARM64 pipeline).\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/c4b1bd20675d4f778ce99277026559b0 : FAILURE in 35m 40s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/fac9d02a10d54266a43af8124c5833c0 : FAILURE in 31m 05s","accounts_in_message":[],"_revision_number":66},{"id":"6e9947bce88185f69fe1a1da1a5f5e9d29ac8137","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-28 10:49:14.000000000","message":"Patch Set 66: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/64a18ffd2a17441189c3784cff26e4a0 : SUCCESS in 7m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b376721f09734c9f90a2e7e2596383d2 : SUCCESS in 6m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e90f031100b649a191c9e4433b390411 : SUCCESS in 6m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f3e8cee0f4ae48c29631dc86f937227c : SUCCESS in 6m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a86538e70808404a995cd4065e206628 : SUCCESS in 6m 32s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/9888d7bfa0c04b25b185ef8557d5ca3e : SUCCESS in 8m 07s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/8d7b4aba5f6248d0bf182e7484bc9fd4 : SUCCESS in 58m 54s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/a550ec6d220a4de78220f553e352079a : SUCCESS in 47m 33s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/19570b44d7a04692aa6f4a45a9c7732a : SUCCESS in 45m 00s\n- kolla-ansible-centos8s-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/ebbcb709c93d4d49a3b7de8341603f53 : SUCCESS in 1h 16m 07s (non-voting)\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/93c0a8febb16490894d11e98ba2a8cd4 : SUCCESS in 51m 22s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/963a339dabcf46f8874e4cbe4f0ba67a : SUCCESS in 1h 20m 37s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/297117f6c0004a1789bd802e0834e528 : SUCCESS in 1h 19m 38s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ba6aa4f5c358469083b110a16a16bfcd : SUCCESS in 1h 00m 44s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/0a8a617ac4f348cd83461024b7978302 : SUCCESS in 1h 09m 56s (non-voting)\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/74bfbb6f2aff441aa5394cdf68dc1e89 : SUCCESS in 51m 34s (non-voting)\n- kolla-ansible-centos8s-source-cephadm https://zuul.opendev.org/t/openstack/build/331511cfaddf433dbc96b64f8abbacf7 : SUCCESS in 1h 27m 10s (non-voting)\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/ebb2c3cf97de41d4a267a99ffec2c9b1 : FAILURE in 23m 44s (non-voting)\n- kolla-ansible-centos8s-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/6a08eed4a68745408928e2841f496cd6 : FAILURE in 38m 19s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/c61181b5eb04471ba0202f6525094a60 : SUCCESS in 1h 19m 36s (non-voting)\n- kolla-ansible-centos8s-hashi-vault https://zuul.opendev.org/t/openstack/build/5ae8cd55cec3457e91ab4f42217d0ac0 : SUCCESS in 3m 37s (non-voting)","accounts_in_message":[],"_revision_number":66},{"id":"7afa544eaf72abcd1764d6766b38b2250748a81f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-02-01 08:38:31.000000000","message":"Uploaded patch set 67: Patch Set 66 was rebased.","accounts_in_message":[],"_revision_number":67},{"id":"c74d7848f742bbc6cc010399e138eae22a499c40","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-01 09:12:43.000000000","message":"Patch Set 67:\n\nBuild failed (ARM64 pipeline).\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/47d7f61bdb7b494691fb566679aae258 : FAILURE in 26m 47s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/0fcce4eb087649a180fcd22fc1f017d6 : FAILURE in 30m 09s","accounts_in_message":[],"_revision_number":67},{"id":"b8ea5295f337777b3649621be072e3455fae7576","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-01 10:40:55.000000000","message":"Patch Set 67: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fed75c9ca19443879e9528aafb350de6 : SUCCESS in 6m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28013e8bd84d40e29b78c2bcba43b945 : SUCCESS in 5m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4bcc6d16dca645bda0ef5e8e6a5e67c4 : SUCCESS in 5m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/375ce946325b40a5a64c12956fba0e6d : SUCCESS in 7m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/acd38bae593d44c39024d627e632dc09 : SUCCESS in 6m 56s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/74490bbf1c6741bbb161c7b0f967bd8e : SUCCESS in 7m 41s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/be5427a416d846b28afbc61663621a4a : SUCCESS in 1h 07m 06s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/ab3584b722be4d37b80aca20ee997968 : SUCCESS in 57m 35s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ad44e25ab0fc4142b3e11223269326dc : SUCCESS in 52m 34s\n- kolla-ansible-centos8s-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/d99fc6509f0948ca8a04ea00f2f2b2e7 : SUCCESS in 1h 18m 08s (non-voting)\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/f14e32cdcd964b27b70c6a14ccd77d04 : SUCCESS in 1h 10m 40s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/fe14fe778a6546dcb70db53d136b2708 : SUCCESS in 1h 26m 58s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/b792431db77747f38fde7f784c46fd73 : SUCCESS in 1h 11m 15s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/62afb01899d54273922a051ff5acea3d : SUCCESS in 1h 01m 39s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/d67a040d4f1049778f3a9db85f63c0aa : SUCCESS in 1h 16m 32s (non-voting)\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6e91ae1d931149bd87de94fdc94784ed : SUCCESS in 54m 45s (non-voting)\n- kolla-ansible-centos8s-source-cephadm https://zuul.opendev.org/t/openstack/build/5360c63e56434fceb8ff2528c4fdf896 : SUCCESS in 1h 28m 43s (non-voting)\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/6990fde0251f48fca096d3dbd420911b : SUCCESS in 1h 22m 27s (non-voting)\n- kolla-ansible-centos8s-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/05780786d12f4a2b845ffc23222cf415 : SUCCESS in 1h 55m 16s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/e35a740b4bd74e77826c64cef4449ffd : SUCCESS in 1h 34m 31s (non-voting)\n- kolla-ansible-centos8s-hashi-vault https://zuul.opendev.org/t/openstack/build/ae54bede6adf45eca32055228cb65e03 : SUCCESS in 3m 36s (non-voting)","accounts_in_message":[],"_revision_number":67},{"id":"ae577f73efc66262a899b7ad9e3904ea6daad295","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-03-11 09:55:06.000000000","message":"Uploaded patch set 68: Patch Set 67 was rebased.","accounts_in_message":[],"_revision_number":68},{"id":"1dd3b099500171726578d58a83acd963bd0418fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-11 12:41:58.000000000","message":"Patch Set 68: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5794a814dd9f4402a53046e4497b193e : SUCCESS in 8m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d920e392a294b35be7e22a77933b9e3 : SUCCESS in 8m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/97a7c73d63d34566a8b13e098740acdc : SUCCESS in 8m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3e4cbac570944fd78bf2ad8545bb18ed : SUCCESS in 8m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6431139d52fe475c8da3bd4d755508e9 : SUCCESS in 8m 00s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/111401f1bcd34d28a6d3a17a6d63720d : SUCCESS in 10m 05s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/c2fa9631964c48ba9c2702a9c2dc3952 : SUCCESS in 1h 11m 48s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/bf50fff62d17424286098eba5d3dbbf8 : SUCCESS in 55m 06s\n- kolla-ansible-rocky8-source https://zuul.opendev.org/t/openstack/build/1de2b7bdd353462c9e7b7e4d2ca6451d : SUCCESS in 1h 14m 03s (non-voting)\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a7673a5ec0504ca9a19174fa3bd122e0 : TIMED_OUT in 2h 03m 50s\n- kolla-ansible-centos8s-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/5faf06dc399745268a2f7ed6198ed8b0 : SUCCESS in 1h 20m 29s (non-voting)\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/e85191a9679b4cbc90747851b026d3ad : TIMED_OUT in 2h 04m 56s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/5fa43ab14ced4ad898e14ac1d5a0e585 : SUCCESS in 1h 23m 30s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/002bea850c414a7788308d9806faf7ee : TIMED_OUT in 2h 33m 33s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/3019b81d768948beaa583ca22a202d0f : SUCCESS in 1h 03m 17s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/ccdb581c9bfd45c5a8bae972b61da7fb : TIMED_OUT in 2h 04m 47s (non-voting)\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6ac6cb8f74d44ad388f14ea9c23505ba : FAILURE in 46m 22s (non-voting)\n- kolla-ansible-centos8s-source-cephadm https://zuul.opendev.org/t/openstack/build/2a1e1cb671a341feb873771d5455dc01 : FAILURE in 33m 45s (non-voting)\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/4085444986534ce784f4f2f3508169f5 : SUCCESS in 1h 58m 39s (non-voting)\n- kolla-ansible-centos8s-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/849f2a10bae24b31abc46765a3834c84 : SUCCESS in 1h 58m 15s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/d92c6e7eac764dc6af0702cfb340a20c : SUCCESS in 1h 25m 47s (non-voting)\n- kolla-ansible-centos8s-hashi-vault https://zuul.opendev.org/t/openstack/build/ddc58aef4c10415bb6375c2df5e03091 : SUCCESS in 6m 09s (non-voting)","accounts_in_message":[],"_revision_number":68},{"id":"14504e742b9258a602eef57bb9434e835935b181","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-03-11 12:59:34.000000000","message":"Patch Set 68:\n\n(1 comment)","accounts_in_message":[],"_revision_number":68},{"id":"141d1e191ebd5affa0d5a3420c2454bfbe8adfe7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-11 13:58:10.000000000","message":"Patch Set 68:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/c478c40ce8ed4d8c9e1a679250281bd4 : FAILURE in 2h 06m 39s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/4922b209dda14a169758629c1cda652b : SUCCESS in 2h 11m 54s","accounts_in_message":[],"_revision_number":68},{"id":"e1f54b67bcf931fbe633928e39145733d7585848","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-11 15:46:05.000000000","message":"Patch Set 68:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/41b09c1ad8704741aeb295743c888c2a : SUCCESS in 9m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76aaaa9bbb074886b3d74603517cf021 : SUCCESS in 10m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1f3412a3f05144039ee2045eacd38aa6 : SUCCESS in 8m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f92e535b71e14f81814e18bcefce38b6 : SUCCESS in 8m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9e1a73455b34e96b78228ba35d89b41 : SUCCESS in 8m 40s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/7674398f2a1b4071ae06bd209f1918d0 : SUCCESS in 10m 56s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/2b11594fb874471fbf4f9d7495fed5ab : SUCCESS in 1h 36m 59s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/9515894b7e984b22b0ebdfd5f993aa2f : SUCCESS in 58m 58s\n- kolla-ansible-rocky8-source https://zuul.opendev.org/t/openstack/build/510fcd01a5294dff96e28c8f7c586ecc : SUCCESS in 1h 11m 29s (non-voting)\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7873f2d03b804403a2ccc2617f7dbab1 : TIMED_OUT in 2h 04m 45s\n- kolla-ansible-centos8s-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/5168582698004d8f93d6b46d085fc7da : SUCCESS in 1h 23m 42s (non-voting)\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/51363009b0354edaad5ed7940af28ab6 : SUCCESS in 59m 11s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/6f6f12e315b84693bdddbcded67da46f : SUCCESS in 1h 26m 26s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/d25355e4cc7243bab17b38cbefdeaebe : SUCCESS in 1h 12m 02s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/bc68f8a83d3345aaae1fed9af38a3067 : TIMED_OUT in 2h 33m 28s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/93352957ff054ab5b0a0f91d8907c9fa : TIMED_OUT in 2h 03m 57s (non-voting)\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/c2d7502b114641efb30baf824e1adf3d : FAILURE in 51m 06s (non-voting)\n- kolla-ansible-centos8s-source-cephadm https://zuul.opendev.org/t/openstack/build/658b737cd2304c8390ac728ba14926be : FAILURE in 34m 50s (non-voting)\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/bc0aea08013b412788cd7bd1bd7e1b51 : SUCCESS in 1h 12m 29s (non-voting)\n- kolla-ansible-centos8s-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/bd16ee2e9b7643a39ef63f0ceb09c80c : FAILURE in 44m 54s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/0c054ed5c54d4f989688ae3bf3b70865 : SUCCESS in 1h 35m 16s (non-voting)\n- kolla-ansible-centos8s-hashi-vault https://zuul.opendev.org/t/openstack/build/aa89b9055e2e4e1aa9ef60fa1c644d59 : SUCCESS in 5m 32s (non-voting)","accounts_in_message":[],"_revision_number":68},{"id":"2b968b96eb913813e7b5941ba7183f794eb988ff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-03-15 17:20:41.000000000","message":"Uploaded patch set 69: Patch Set 68 was rebased.","accounts_in_message":[],"_revision_number":69},{"id":"d63e675926d89b1158f3250fd812e60af306c18c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-15 18:21:08.000000000","message":"Patch Set 69:\n\nBuild failed (ARM64 pipeline).\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/5326491e403c4952ace5dccaaef1cf7c : FAILURE in 22m 28s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/98266d6a396746499199d130f0e88192 : FAILURE in 23m 55s","accounts_in_message":[],"_revision_number":69},{"id":"421cefdf05fd209ce6871e3f42fb1485ed0b970b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-15 19:05:30.000000000","message":"Patch Set 69: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/717c7ced6032425bbd410ecdc07fa524 : SUCCESS in 8m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90e1d655fd2b4550b8ba37e6dfb54305 : SUCCESS in 8m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0ac59d619c5948d385e70a7cb959aa13 : SUCCESS in 7m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f063117dee694f59b59575c49d0e425e : SUCCESS in 8m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6ed5ea6dc934febbacd5a8b758d535c : SUCCESS in 7m 34s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/9e50e8bbd0244ccab52d5283adc469f4 : FAILURE in 10m 20s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/30b0824cedd84e41a12cd4188e4c3bfd : SUCCESS in 1h 14m 10s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/c2330a7fc9964b0fa4611aa95df953c4 : SUCCESS in 56m 25s\n- kolla-ansible-rocky8-source https://zuul.opendev.org/t/openstack/build/658bccdfe1a14e999d09605455a74b14 : SUCCESS in 1h 13m 20s (non-voting)\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b8f95512f8c3450b902a072bc609eacf : SUCCESS in 57m 08s\n- kolla-ansible-centos8s-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/aa7f9962fc3b45e7a7f76c8dd735e55b : SUCCESS in 1h 14m 43s (non-voting)\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/33ce6054c4694cfca968ca01992aacd0 : SUCCESS in 1h 04m 48s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/3bba3f13f791409a81046879257eaa44 : SUCCESS in 1h 20m 37s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/0616ee4ca9754417bce446b23c1176a3 : SUCCESS in 1h 07m 15s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/2619a905331a4eaca2e5f93c292101c6 : SUCCESS in 59m 09s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/fb24eb5e19bb440a9c76a41a4bb4eaa5 : SUCCESS in 1h 08m 27s (non-voting)\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6d40c51b025948e0835b2f4f5b754cd1 : FAILURE in 43m 48s (non-voting)\n- kolla-ansible-centos8s-source-cephadm https://zuul.opendev.org/t/openstack/build/5176627dd857486c88196c0f21f87584 : SUCCESS in 1h 42m 05s (non-voting)\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/213a14cf504b401ab5f10fd1cae3396d : SUCCESS in 1h 19m 59s (non-voting)\n- kolla-ansible-centos8s-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/af5b82f191544c5999a556ff4aa8f472 : FAILURE in 51m 15s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/0e1a6aa3334f47dba52e8afaef170ef0 : SUCCESS in 1h 30m 19s (non-voting)\n- kolla-ansible-centos8s-hashi-vault https://zuul.opendev.org/t/openstack/build/fe8596835f774fd4adb4441cb87c486a : FAILURE in 4m 54s (non-voting)","accounts_in_message":[],"_revision_number":69},{"id":"78e05b00e43bb20de8e6f76cbd293063f42afae4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-03-18 08:23:32.000000000","message":"Uploaded patch set 70: Patch Set 69 was rebased.","accounts_in_message":[],"_revision_number":70},{"id":"679e5ea599042744ac06d5262faa979a826762a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-03-18 10:11:32.000000000","message":"Uploaded patch set 71: Patch Set 70 was rebased.","accounts_in_message":[],"_revision_number":71},{"id":"8ef789f814fb70e6f4c41d9bc8535e55bb5a79d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-18 12:05:05.000000000","message":"Patch Set 71: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/304fe459d30941d0897674eba43dfa17 : SUCCESS in 8m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b0ec74f60724cf19a1ec539e3e1fefd : SUCCESS in 8m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3ba66f6bf83846b69b659d22453ed04e : SUCCESS in 7m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/88d89709cc55420688bc0dc9381cafc0 : SUCCESS in 8m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a43961c661104d9f9cb0a84a5da5270b : SUCCESS in 7m 15s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/8ae45eb94dbe4b6ea0afedd508486c2b : SUCCESS in 9m 17s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/2e24b851499d4c349bc2975d6e440cfc : SUCCESS in 1h 11m 15s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/84d68dff7bee4a558f529d9c2fce5029 : SUCCESS in 1h 13m 04s\n- kolla-ansible-rocky8-source https://zuul.opendev.org/t/openstack/build/29490816475847af9919d1b5b4446061 : SUCCESS in 1h 16m 57s (non-voting)\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/05f4ad5efaf940daae53215910d66cee : SUCCESS in 54m 36s\n- kolla-ansible-centos8s-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/2961ff39498a483eae095540f1e12655 : SUCCESS in 1h 24m 03s (non-voting)\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/4bbeacefe32244668faef9fad751bcd5 : SUCCESS in 1h 04m 30s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/4ab9ac1552ff450b928bafd83549424f : SUCCESS in 1h 20m 45s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/3c46eef1e96f4bb58a4466498bc5491d : SUCCESS in 1h 16m 04s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/db92548310de4564b26bbfc2944211c4 : SUCCESS in 1h 07m 44s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/adda4e8c55694497a88d51ce85da1d81 : SUCCESS in 1h 06m 14s (non-voting)\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/27c2649cfe0c41f78eb08faff92f2734 : FAILURE in 45m 18s (non-voting)\n- kolla-ansible-centos8s-source-cephadm https://zuul.opendev.org/t/openstack/build/a7d34fc6ce2442939ad52d30d5ded997 : FAILURE in 33m 24s (non-voting)\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/63f76388e7f84aa2af8bf49df5b0e578 : SUCCESS in 1h 19m 58s (non-voting)\n- kolla-ansible-centos8s-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/13a6f6e0456d4627874114cef54bb446 : FAILURE in 49m 00s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/5f158344c54d42be86480a24d61e7535 : SUCCESS in 1h 42m 16s (non-voting)\n- kolla-ansible-centos8s-hashi-vault https://zuul.opendev.org/t/openstack/build/236e908bf5c34c26864ed537d22c40c8 : SUCCESS in 5m 27s (non-voting)","accounts_in_message":[],"_revision_number":71},{"id":"5902c8ccb7838ae7d45e0cf411c174e12edbee72","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-18 12:35:50.000000000","message":"Patch Set 71:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/3a7f99f2dfde4394bdfafae3c19db9e8 : SUCCESS in 2h 00m 51s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/f0cb7247603846ccb6c70f3c8a7b028e : SUCCESS in 2h 17m 35s","accounts_in_message":[],"_revision_number":71},{"id":"2d335fa7413cd7ba2f2074dac7d383597f644885","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-04-21 12:31:10.000000000","message":"Uploaded patch set 72: Patch Set 71 was rebased.","accounts_in_message":[],"_revision_number":72},{"id":"f331a3f08dbeae69b3870c0f623d527f70298412","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-21 14:04:40.000000000","message":"Patch Set 72: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/68c46fb54d6a4305ad8cf81c563d44a6 : SUCCESS in 5m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31cd7d14fbfd4b8fa5cb4a72455e41c5 : SUCCESS in 6m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/19bcd94e473741f090a4ab880479f2c8 : SUCCESS in 5m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/97f3821c7bd3425f897ae51b8eae1fd2 : SUCCESS in 5m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/75c5529b746f4690becc36623d9de01b : SUCCESS in 7m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c90d3233e734575b4644d2e5bfda2a6 : SUCCESS in 6m 14s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/cdb3c13f82014581a1d7349c35b9d910 : SUCCESS in 8m 02s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/322c9288da9d4b8cbedb219bbe9dd9b3 : SUCCESS in 1h 14m 17s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/b304c7d2a0eb49158bddd86c94531041 : SUCCESS in 47m 41s\n- kolla-ansible-openeuler-source https://zuul.opendev.org/t/openstack/build/73ccbd65911047249ad12a051cc07b9b : SUCCESS in 1h 16m 21s (non-voting)\n- kolla-ansible-rocky8-source https://zuul.opendev.org/t/openstack/build/f71adce1b82d4e539917d61ea3d42a99 : SUCCESS in 1h 20m 03s (non-voting)\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/43268aae255f43278c2e970f7a45e63e : SUCCESS in 49m 11s\n- kolla-ansible-centos8s-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/f15dac99414c461e99af5fd9090169d0 : SUCCESS in 1h 12m 10s (non-voting)\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/07f3670fb24c493fa9a3681f9a56785a : SUCCESS in 55m 05s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/3d8c174c0c6743d0959e079d0dd758c7 : SUCCESS in 1h 18m 53s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/ca7f1f6909f64abf9c847e939ef0c0b1 : SUCCESS in 1h 16m 34s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/7b06d9e5449e4183925ff4871c6e2ed6 : SUCCESS in 1h 10m 32s\n- kolla-ansible-centos8s-source-cephadm https://zuul.opendev.org/t/openstack/build/6d95aa547ce64c9ab0af0c7fa46be8c1 : SUCCESS in 1h 20m 40s (non-voting)\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/bcd6125b8a184d7abdc0e0a81cab6a80 : SUCCESS in 1h 11m 58s (non-voting)\n- kolla-ansible-centos8s-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/493a83aabf094cdbaa497a1ae7251912 : FAILURE in 40m 21s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/c11102a2f7614995b42b514b4a61af57 : SUCCESS in 1h 25m 17s (non-voting)\n- kolla-ansible-centos8s-hashi-vault https://zuul.opendev.org/t/openstack/build/3eebec5be5a849429b7cd148bf1299e9 : SUCCESS in 3m 41s (non-voting)","accounts_in_message":[],"_revision_number":72},{"id":"ac28c017e85f708b70a493b358203fe7a29113f7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-21 14:41:21.000000000","message":"Patch Set 72:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/8803c874903d4a3db3ed83ca69790c85 : SUCCESS in 1h 52m 52s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/62a7d995faf043deaa74a1e30c8aa9fe : SUCCESS in 2h 07m 09s","accounts_in_message":[],"_revision_number":72},{"id":"1957ed507bc3e53bcc2a2bfacf2cdd4b1d2db6c9","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-21 15:10:50.000000000","message":"Patch Set 72:\n\n(1 comment)","accounts_in_message":[],"_revision_number":72},{"id":"2500b56decb3eb0c58a28acd3746fad4af7ed8f0","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-21 15:12:40.000000000","message":"Patch Set 72:\n\n(1 comment)","accounts_in_message":[],"_revision_number":72},{"id":"98188bacd95de6751fee81a25b6c703afbaff3e8","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-04-21 15:12:52.000000000","message":"Patch Set 72:\n\n(1 comment)","accounts_in_message":[],"_revision_number":72},{"id":"891ae89fb70cb610559910c48be2a9feb201ae48","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-04-21 15:13:33.000000000","message":"Patch Set 72:\n\n(1 comment)","accounts_in_message":[],"_revision_number":72},{"id":"c9dbe1aad1eab81c8c2d4e593f6de2428644341a","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-21 15:13:45.000000000","message":"Patch Set 72:\n\n(1 comment)","accounts_in_message":[],"_revision_number":72},{"id":"17edf6debfb1b4ad8101ccbf33935143daf0d20a","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-04-21 15:14:01.000000000","message":"Patch Set 72:\n\n(1 comment)","accounts_in_message":[],"_revision_number":72},{"id":"ff61efa6e17da516b664cc4f7506cd653a5197cd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-04-21 16:14:43.000000000","message":"Uploaded patch set 73: Patch Set 72 was rebased.","accounts_in_message":[],"_revision_number":73},{"id":"fef7fd8014853c38e121d88f13c24405d6df8890","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-04-21 16:59:42.000000000","message":"Uploaded patch set 74: New patch set was added with same tree, parent, and commit message as Patch Set 73.","accounts_in_message":[],"_revision_number":74},{"id":"bd3af3f51e0e0aba81613c18f136a21e72fb6544","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-21 16:59:58.000000000","message":"Patch Set 73: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":73},{"id":"e72a7015a921ca5047949b2c44f3a4c728391554","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-21 17:00:03.000000000","message":"Patch Set 73: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":73},{"id":"358719b7a7a399c1d3305f983e3bf052d2c67c52","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-04-21 17:15:28.000000000","message":"Patch Set 74:\n\n(1 comment)","accounts_in_message":[],"_revision_number":74},{"id":"db6a8af77f46927bf0f1f63f7cb46517c8c60884","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-04-21 17:42:26.000000000","message":"Uploaded patch set 75: Patch Set 74 was rebased.","accounts_in_message":[],"_revision_number":75},{"id":"05049bb0cad907302ae673007b27f00558dcf90d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-21 19:10:47.000000000","message":"Patch Set 75: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3fca3b61844e470293f860cc50ab6e76 : SUCCESS in 6m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cbf6fef02147487e8cc9273e996568eb : SUCCESS in 6m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a6b80d047ad547d2aba8d5ee5f26cbc7 : SUCCESS in 6m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c3b266035bf1412e82ecb67fd9f72fe7 : SUCCESS in 6m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bf7893c103d347a8819492c90d9d2fa2 : SUCCESS in 8m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b83ec00c04b9489aa750f8b99b268c28 : SUCCESS in 5m 55s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/1c81058736b24e11bbf0638ba7dfdc82 : SUCCESS in 7m 51s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/612dcd88431b4a49bf2e7b01b1d6ce88 : FAILURE in 35m 09s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/e508241c0b4745c795599b60d8763b31 : FAILURE in 29m 56s\n- kolla-ansible-openeuler-source https://zuul.opendev.org/t/openstack/build/b4263c78f4f04dba8344da44931f53d0 : FAILURE in 28m 32s (non-voting)\n- kolla-ansible-rocky8-source https://zuul.opendev.org/t/openstack/build/4ce8ecb501624eeda1cc7a7c369028d7 : FAILURE in 33m 32s (non-voting)\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/33843bb9bd7a470d925c5857106e2466 : FAILURE in 21m 56s\n- kolla-ansible-centos8s-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/fc55a14b0c6f4be2b52f0aedb0b06e06 : FAILURE in 30m 40s (non-voting)\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/c244bf6cee7d487995f4473f0ef896d2 : FAILURE in 22m 52s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/b1d5e26324114ac99267829e7634e536 : FAILURE in 1h 05m 13s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/44a32c4525e24d818053c3112773d2fe : FAILURE in 53m 22s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/33edd1fd311043f49b109fd30a51bfd2 : FAILURE in 50m 06s\n- kolla-ansible-centos8s-source-cephadm https://zuul.opendev.org/t/openstack/build/248069b2932e4293acea7e942ab8b7df : FAILURE in 38m 17s (non-voting)\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/9a852f9f648b4bdda6a456aa46df9aca : FAILURE in 30m 07s (non-voting)\n- kolla-ansible-centos8s-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/1db3b6a077b2494a9d41695902ce34d3 : FAILURE in 1h 22m 47s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/575fc48cb41243639d327ed156b79300 : FAILURE in 1h 05m 06s (non-voting)\n- kolla-ansible-centos8s-hashi-vault https://zuul.opendev.org/t/openstack/build/81ab452f797a446b8a9b1539af7ad8c8 : SUCCESS in 4m 08s (non-voting)","accounts_in_message":[],"_revision_number":75},{"id":"f92c7988a141acc47e76696cf51bf035076fa278","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-04-21 19:55:37.000000000","message":"Patch Set 75:\n\n(1 comment)","accounts_in_message":[],"_revision_number":75},{"id":"f083487cf0e756f1acd15843a08580c96624b577","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-21 23:52:11.000000000","message":"Patch Set 75:\n\nBuild failed (ARM64 pipeline).\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/4c4306ed7e224544a74730455616704d : FAILURE in 1h 16m 29s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/f7c1bd6ceaff425db34be57626ad8925 : FAILURE in 1h 48m 32s","accounts_in_message":[],"_revision_number":75},{"id":"221feca80204fe1306368b0bcbd18c0d7960d49a","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-04-22 07:32:43.000000000","message":"Patch Set 75:\n\n(1 comment)","accounts_in_message":[],"_revision_number":75},{"id":"2a690531088e121bc89f7003129a88fbb0c7b393","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-22 09:12:03.000000000","message":"Patch Set 75: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a1db83ca1092457eb8283824a72738f0 : SUCCESS in 6m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6f9d1cf529a45abaaa39f40a71d7f38 : SUCCESS in 6m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/27b90a77e2d14ed3819c7368d833dee6 : SUCCESS in 6m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eb3438e2bcab41eba6f37cc4fe1c6b6b : SUCCESS in 6m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a24e479eab414bcca148bb448635e353 : SUCCESS in 7m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/422450dd93d140bcbe4a8651319b3314 : SUCCESS in 5m 31s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/d62d887f975c486191a577205da4125e : SUCCESS in 7m 40s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/11f9af86d901404fbbb1786bb2b4aab5 : SUCCESS in 1h 06m 45s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/b452c7dd3da842cc9b869e5cbdd5b6e9 : SUCCESS in 42m 21s\n- kolla-ansible-openeuler-source https://zuul.opendev.org/t/openstack/build/760fa25565b74b56971592a6c490259c : SUCCESS in 1h 07m 49s (non-voting)\n- kolla-ansible-rocky8-source https://zuul.opendev.org/t/openstack/build/4485f7f1666f48ec9c0e765ca83dd3bd : SUCCESS in 1h 05m 17s (non-voting)\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/1dd927b23b5a440c85d5e55dc9568a02 : SUCCESS in 38m 11s\n- kolla-ansible-centos8s-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/2017a7ea704640cba917e23e486cb327 : FAILURE in 44m 31s (non-voting)\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/1f649a4d895640c0be22ffe522504fac : SUCCESS in 45m 36s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/78fd7162b632402ebd5c3e3c24c29ff9 : SUCCESS in 1h 16m 22s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/ef750ff08277448e9786b5f9be903713 : SUCCESS in 48m 58s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f2fd3aa9f5ce43c08d1bd46f2d0c6d1f : SUCCESS in 56m 54s\n- kolla-ansible-centos8s-source-cephadm https://zuul.opendev.org/t/openstack/build/beae1fab15dd41649d7b660d471edc91 : SUCCESS in 1h 16m 21s (non-voting)\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/a42f85b2f28246c9882722c2b70e3f13 : SUCCESS in 58m 59s (non-voting)\n- kolla-ansible-centos8s-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/c87147654ec947c28c3ca9a446a63469 : SUCCESS in 1h 28m 46s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/835cd19c87464dcf8b8be3f11557a93b : SUCCESS in 1h 22m 19s (non-voting)\n- kolla-ansible-centos8s-hashi-vault https://zuul.opendev.org/t/openstack/build/8d15a6109a6b4bc1b5f192a5979e2009 : SUCCESS in 3m 55s (non-voting)","accounts_in_message":[],"_revision_number":75},{"id":"518465919fc420a1cc2c983070194b2626c04aa0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-22 09:15:31.000000000","message":"Patch Set 75:\n\nBuild failed (ARM64 pipeline).\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/ecbc269a73734ff6b0212785618d12d7 : SUCCESS in 1h 13m 43s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/48e144c5519d46fa825347aa4b891518 : FAILURE in 1h 40m 23s","accounts_in_message":[],"_revision_number":75},{"id":"d330b44802ddd5c82cebbdd3e79c06ca51115089","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-05-02 11:08:00.000000000","message":"Patch Set 76: Patch Set 75 was rebased","accounts_in_message":[],"_revision_number":76},{"id":"6e2011282061d9fe23e880a2c4f857a5207bcd2f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-02 13:03:18.000000000","message":"Patch Set 76:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/15a7c1269c824786b3a99b9835fd7aa5 : SUCCESS in 1h 50m 27s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/9af55dc8f30a42fabd6cf6d2f8b4949e : SUCCESS in 1h 44m 49s","accounts_in_message":[],"_revision_number":76},{"id":"cf0abe037bf0023852f7d9f0b7247890899f1306","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-02 13:23:23.000000000","message":"Patch Set 76: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7defea19bc25426fa94fde68209f3075 : SUCCESS in 5m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ec9af92806e47e98b9a6a96e5e5a869 : SUCCESS in 6m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/94485611fc874fb2b2553db569b66bb7 : SUCCESS in 6m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/13455592b79a42368e6c62feba90d7bc : SUCCESS in 7m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c980d400fb6c45bcb579dd7c43946008 : SUCCESS in 7m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f4db133c60d4fcd98d5c726cc851cdb : SUCCESS in 5m 30s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/315e9fd692284267a5798adad4ad0bf7 : SUCCESS in 7m 44s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/855f3fdc4fde4f738bb0e55a0b628ebe : SUCCESS in 1h 02m 24s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/ee845f1983bc4feea334a9c6d1fcc5d4 : SUCCESS in 48m 59s\n- kolla-ansible-openeuler-source https://zuul.opendev.org/t/openstack/build/4034ad585ae34c19a144a370f4b80a25 : SUCCESS in 58m 27s (non-voting)\n- kolla-ansible-rocky8-source https://zuul.opendev.org/t/openstack/build/0024109d88bb48d5a2a621e03946159b : SUCCESS in 53m 56s (non-voting)\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/afed5d011bb14ea285303709e96dd6dd : SUCCESS in 41m 38s\n- kolla-ansible-centos8s-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/ce3035b4987f4ea6b763e52c78e525c6 : SUCCESS in 54m 12s (non-voting)\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/362e90daa4fa4c92bca18705323b6ea4 : SUCCESS in 50m 02s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/be62aad309484fb48bb38198f181b6ad : SUCCESS in 1h 29m 59s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/f936f06201994e49a4e618ff7a55b563 : SUCCESS in 1h 55m 48s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/7a838e13b9f74b97b949416dbeb80e94 : SUCCESS in 1h 54m 22s\n- kolla-ansible-ubuntu-binary-upgrade https://zuul.opendev.org/t/openstack/build/fb89d2b01b81440facc38ee36347c740 : SUCCESS in 1h 49m 30s\n- kolla-ansible-centos8s-source-cephadm https://zuul.opendev.org/t/openstack/build/6d98aaf4c2b34082b9c8108abfe6bdaa : SUCCESS in 2h 08m 21s (non-voting)\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/a9d6a64574034ac4b6a5124a189ac3b2 : SUCCESS in 1h 10m 57s (non-voting)\n- kolla-ansible-centos8s-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/d0d4dedfc06a455084529c8f6c598f35 : FAILURE in 1h 07m 07s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/6f1d1ed35afd4669baf46422d707253f : SUCCESS in 1h 56m 14s (non-voting)\n- kolla-ansible-centos8s-hashi-vault https://zuul.opendev.org/t/openstack/build/a1f4c54bdb9641cf8a4cf623b89f1168 : SUCCESS in 3m 13s (non-voting)","accounts_in_message":[],"_revision_number":76},{"id":"782a837110e731b23f01f9c4019fe043c6e807fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-05-04 09:52:35.000000000","message":"Uploaded patch set 77.","accounts_in_message":[],"_revision_number":77},{"id":"a9919d8b15bfe5b6dd36f4693f1c9550973effe8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-05-04 09:54:13.000000000","message":"Uploaded patch set 78: Patch Set 77 was rebased.","accounts_in_message":[],"_revision_number":78},{"id":"bc6fa24fe98958252f58e38ddcb2dffe26db08b7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-05-04 09:58:59.000000000","message":"Uploaded patch set 79: Patch Set 78 was rebased.","accounts_in_message":[],"_revision_number":79},{"id":"662565f530ab0bed20414d651fc2aead2adb5576","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-05-04 10:26:55.000000000","message":"Uploaded patch set 80.","accounts_in_message":[],"_revision_number":80},{"id":"32634c0f72a3c95bd8dab5e9b5c2fe537cd5ea9a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-05-04 10:44:35.000000000","message":"Uploaded patch set 81.","accounts_in_message":[],"_revision_number":81},{"id":"06e9dbefd8071ddbce26bfdc8ef49c3daa25f42a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-04 12:12:45.000000000","message":"Patch Set 81: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f9c79e875b364ae98ca89e4591dbea1f : SUCCESS in 7m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9589ba12fb3a4334900e99209bf44884 : SUCCESS in 5m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/31ecde31c8ca4e81aa44f2c6f62c6213 : SUCCESS in 5m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b88331a7af7e4123b18a6d195ad0578a : SUCCESS in 5m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3315ed31b2e74c0abf1f026c4880306a : SUCCESS in 7m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8212d2e6e67e4a56ae5a2648b79bdb93 : SUCCESS in 5m 28s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/ea2028036f2c435086944b5f8c126c32 : SUCCESS in 8m 10s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/7f9218767992447ea84d4a57e9ac57b6 : SUCCESS in 1h 05m 42s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/ee1e6bc6a61e4ad58da027a02419f67e : SUCCESS in 43m 31s\n- kolla-ansible-openeuler-source https://zuul.opendev.org/t/openstack/build/e5f4da68d73a49e2baf38676101d4e87 : SUCCESS in 1h 05m 32s (non-voting)\n- kolla-ansible-rocky8-source https://zuul.opendev.org/t/openstack/build/7fb2c9dc217a49239facf899eb93b74e : SUCCESS in 57m 43s (non-voting)\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/9f35e5f895d64cc8845a526f359b1571 : SUCCESS in 49m 07s\n- kolla-ansible-centos8s-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/6c42771f5db94c7c9fee6892d3b042fe : SUCCESS in 59m 32s (non-voting)\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/43b9e1a399e3439d8c2be4a76d8180f1 : SUCCESS in 52m 45s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/a9a55e17909d40b7ad4dbeab07c2cc37 : SUCCESS in 1h 00m 17s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/211c106865fb429a8b667c86cc87a4df : SUCCESS in 49m 10s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/e0a4e6b06e2a4b21ac49f7d2d760f0ca : SUCCESS in 44m 14s\n- kolla-ansible-ubuntu-binary-upgrade https://zuul.opendev.org/t/openstack/build/64d02d29af0247599e9830cd68b4eb19 : SUCCESS in 42m 24s\n- kolla-ansible-centos8s-source-cephadm https://zuul.opendev.org/t/openstack/build/b8839149b514409b83a0cbab3f7131ac : SUCCESS in 1h 25m 27s (non-voting)\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/0611bf5d72724c87b39365d8b53d84a5 : SUCCESS in 55m 22s (non-voting)\n- kolla-ansible-centos8s-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/9892495e24e546b099e116e57f60ee3d : FAILURE in 17m 58s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/8f232381d4984776b571b68dc2b8ed3e : SUCCESS in 1h 19m 00s (non-voting)\n- kolla-ansible-centos8s-hashi-vault https://zuul.opendev.org/t/openstack/build/3036f072c97d49fba2abef8151d66063 : SUCCESS in 3m 15s (non-voting)","accounts_in_message":[],"_revision_number":81},{"id":"3b3e3201e17e328d3f3299f00d376203381c5279","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-04 13:09:54.000000000","message":"Patch Set 81:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a6e11f22d9a44bc2aa4cf096fe9d7c36 : SUCCESS in 1h 40m 11s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/62c21b5d8a134cc8994abdb3a2bec4dc : SUCCESS in 1h 47m 39s","accounts_in_message":[],"_revision_number":81},{"id":"52e616537416b8446894a9340fbe98bee9219e89","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-05-05 10:36:18.000000000","message":"Uploaded patch set 82.","accounts_in_message":[],"_revision_number":82},{"id":"c16870ac4e471ca6218dec1108c3464259d52855","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-05 12:21:37.000000000","message":"Patch Set 82:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/7c936394afd145e98983db923ac1e904 : SUCCESS in 1h 32m 34s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/eeae06ee8b774dc6a6ec8b72d1315689 : SUCCESS in 1h 41m 56s","accounts_in_message":[],"_revision_number":82},{"id":"f0cd80789ced58eb39e0c2224190fe7eae7582ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-05 13:02:39.000000000","message":"Patch Set 82: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3e59a270a2d843df98651d9b5a095b15 : SUCCESS in 6m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0b310e0922640aab6f2a965ecf63fed : SUCCESS in 5m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a1fbd367be21402499f640141a7cbb0a : SUCCESS in 6m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0edc99d9909542419de190f019bded3e : SUCCESS in 5m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0c188285a91441a499c7bd292ca96739 : SUCCESS in 9m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4f60b76ea85840b991a46d6968986213 : SUCCESS in 6m 32s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/fccdff6fc29e484ba8d57486093510ce : SUCCESS in 8m 46s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/58f3cb22d6284ccfbe5dab10bf9b2112 : SUCCESS in 56m 44s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/cc48b4f8d9174952adec937e5639215d : SUCCESS in 1h 54m 47s\n- kolla-ansible-openeuler-source https://zuul.opendev.org/t/openstack/build/5cb5d230e2a841b192af777cde1e69c6 : SUCCESS in 1h 16m 47s (non-voting)\n- kolla-ansible-rocky8-source https://zuul.opendev.org/t/openstack/build/3963b5df5a934699b7a9ab4ad3878fe0 : SUCCESS in 1h 11m 33s (non-voting)\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/0c039a5fe2a64349b8c1ceb7e0541341 : SUCCESS in 1h 51m 50s\n- kolla-ansible-centos8s-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/3cc895c3e0034c86813159f0725e73ab : SUCCESS in 58m 19s (non-voting)\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/f40c336159b04ea1a3519d9145668ff8 : SUCCESS in 1h 59m 33s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/81a3fe584e7c4487afa43d005e167dbc : SUCCESS in 59m 47s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/73c2a8ba09954a59b1ad93fdb1f80744 : SUCCESS in 53m 49s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/30798e0d624d44e48c74b6fb96f6f025 : SUCCESS in 48m 41s\n- kolla-ansible-ubuntu-binary-upgrade https://zuul.opendev.org/t/openstack/build/476dfb1d4b244b91bebbd407f8f6782a : SUCCESS in 1h 58m 23s\n- kolla-ansible-centos8s-source-cephadm https://zuul.opendev.org/t/openstack/build/6219099fba07415ea2dc3d9046ce71bc : SUCCESS in 1h 06m 56s (non-voting)\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/8244fed97d1646fba07291e4840b51b0 : SUCCESS in 1h 59m 39s (non-voting)\n- kolla-ansible-centos8s-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/a9afb560002f4cac9d02466309ef2639 : SUCCESS in 1h 17m 04s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/445f907dcf0b471890acc93ceffa6db5 : SUCCESS in 2h 18m 15s (non-voting)\n- kolla-ansible-centos8s-hashi-vault https://zuul.opendev.org/t/openstack/build/fa786c796a674bcfa7b8c7fb4510ca59 : SUCCESS in 3m 13s (non-voting)","accounts_in_message":[],"_revision_number":82},{"id":"2443bd56626e53c3f23df00314d1164f7ae3ea2e","author":{"_account_id":167,"name":"Christian Berendt","email":"berendt@osism.tech","username":"berendt"},"date":"2022-05-13 17:49:22.000000000","message":"Patch Set 82: Code-Review+1","accounts_in_message":[],"_revision_number":82},{"id":"1ce525b29d7b275cdb4df2ae870144c09e5ac19b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-05-31 09:26:22.000000000","message":"Uploaded patch set 83: Patch Set 82 was rebased.","accounts_in_message":[],"_revision_number":83},{"id":"0f9d5ced58592a124ec3ec18f0defa905122b662","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-05-31 10:59:38.000000000","message":"Uploaded patch set 84: Patch Set 83 was rebased.","accounts_in_message":[],"_revision_number":84},{"id":"d211ee4186ad6dfd4506257fc783ed83d390ed50","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-31 12:39:29.000000000","message":"Patch Set 84: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dd9d20ced81b480989a1d6df9890a7cb : SUCCESS in 6m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d152fd4657e4cab90b6b3068ba8ed2a : SUCCESS in 5m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c7c48ea8bd264affa34fc103f1f1c91e : SUCCESS in 5m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c7078fcea9d14e1fa0e04841f6ef40b6 : SUCCESS in 6m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3454bbd6cba34bbba5d55847bfaebb0f : SUCCESS in 5m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20f55a288ee94a4e9a481fa3d29aa444 : SUCCESS in 6m 31s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/924dbd15184240ce94447563f802bdf0 : SUCCESS in 6m 08s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/b82d97916e094e4e931a41508e5b10d2 : SUCCESS in 59m 38s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/c57ec952e1a44fa2a5e31ae9a54efb11 : SUCCESS in 43m 43s\n- kolla-ansible-openeuler-source https://zuul.opendev.org/t/openstack/build/a30d773769fc49df864307f1419fcd11 : SUCCESS in 1h 07m 36s (non-voting)\n- kolla-ansible-rocky8-source https://zuul.opendev.org/t/openstack/build/3a6a0af399d94a899261909213c61392 : SUCCESS in 1h 05m 02s (non-voting)\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/18324deb174d46618eb4526a41d43019 : SUCCESS in 38m 47s\n- kolla-ansible-centos8s-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/cddc2bd199414f748bf5a6a3479a9f0d : SUCCESS in 1h 13m 15s (non-voting)\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/1a2536db752d46839a05beac73ecc7d4 : SUCCESS in 47m 39s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/501ae132d8884b74a9d30268df5cfff1 : SUCCESS in 1h 15m 16s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/5ff21c4c8af548f9a9ce512c6bdc0955 : SUCCESS in 59m 46s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/a121fd817b054c5188c07a96beccd013 : SUCCESS in 47m 44s\n- kolla-ansible-ubuntu-binary-upgrade https://zuul.opendev.org/t/openstack/build/6d09b891e14c46cf9c0e6cc26d39e160 : SUCCESS in 52m 44s\n- kolla-ansible-centos8s-source-cephadm https://zuul.opendev.org/t/openstack/build/49d5e6043087485db07226e44c5bf999 : SUCCESS in 1h 14m 49s (non-voting)\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/ff419a0f665146c595d68128f7d2b7b3 : SUCCESS in 1h 09m 42s (non-voting)\n- kolla-ansible-centos8s-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/1ecd58114a984a4f8256bc4acb0c968f : FAILURE in 53m 04s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/c04d495b1bf045fdb5415db1ab5c6000 : SUCCESS in 1h 25m 49s (non-voting)\n- kolla-ansible-centos8s-hashi-vault https://zuul.opendev.org/t/openstack/build/dd53bad72e694861946712dc69c3ed73 : SUCCESS in 3m 14s (non-voting)","accounts_in_message":[],"_revision_number":84},{"id":"77c9c4bf2ec39f8bae18a4956067c3f5b0216c61","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-05-31 12:50:49.000000000","message":"Uploaded patch set 85: Patch Set 84 was rebased.","accounts_in_message":[],"_revision_number":85},{"id":"b1209735b5e68e5d5bf7bf1c0c6b632ce6d55dea","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-31 14:52:28.000000000","message":"Patch Set 85:\n\nBuild failed (ARM64 pipeline).\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/06922cb2eb2542a29469638c4557f035 : FAILURE in 1h 16m 57s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/6549d68d09d64727a8ad779e07518342 : FAILURE in 1h 53m 55s","accounts_in_message":[],"_revision_number":85},{"id":"a1b5a9a0326918d8d066fbd62f8eb71b248de986","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-31 16:19:09.000000000","message":"Patch Set 85: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f4f8e6ae0cef47dbbcd12c88415e7065 : SUCCESS in 6m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d73440362734406ad0d86f48a674b65 : SUCCESS in 6m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b21ff428ba9a417fab90e1a9c114c3a0 : SUCCESS in 5m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4e7b9f6ac8944fc6985e104622a28122 : SUCCESS in 6m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/01791d3ee3ee465a9aecba9e980a05e7 : SUCCESS in 6m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85a5385884774573b24e2835bf569fcf : SUCCESS in 6m 17s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/7738ea91cc4c492e86e1448ff76e4832 : SUCCESS in 7m 58s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/428710d3c81e4f539e0a73e0c5cf99e4 : SUCCESS in 1h 05m 02s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/75d482824587462f979895dca058479f : SUCCESS in 1h 01m 03s\n- kolla-ansible-openeuler-source https://zuul.opendev.org/t/openstack/build/d193c6b6e5864160b97312fa9a2f493e : SUCCESS in 1h 10m 14s (non-voting)\n- kolla-ansible-rocky8-source https://zuul.opendev.org/t/openstack/build/ecd9afb2edcf40d1862a0d0be3a13e77 : SUCCESS in 1h 03m 03s (non-voting)\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/106c19f4a6264f7eb70dda70dcbc8780 : SUCCESS in 59m 26s\n- kolla-ansible-centos8s-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/90d7e00dd2954d2dbf1c2dacfb4749d4 : SUCCESS in 1h 14m 15s (non-voting)\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/a28b5c05dde6418e9e45ae865da324c3 : SUCCESS in 56m 07s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/5a38bcb6b9734cb39335c01d6e5410ed : SUCCESS in 1h 15m 08s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/bf82e59ad4944700a3aa3c5061a478b3 : SUCCESS in 1h 09m 05s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d3f7d6ee52414bdc9b12a5d4496718b9 : SUCCESS in 1h 00m 32s\n- kolla-ansible-ubuntu-binary-upgrade https://zuul.opendev.org/t/openstack/build/81c46661df42451680bf04c710fec7eb : SUCCESS in 1h 02m 16s\n- kolla-ansible-centos8s-source-cephadm https://zuul.opendev.org/t/openstack/build/2ddcd21e96e14ab6a9d3d7379d6af336 : SUCCESS in 1h 30m 46s (non-voting)\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/af8bb2c442b14baf991c5e360f10cab2 : SUCCESS in 1h 09m 25s (non-voting)\n- kolla-ansible-centos8s-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/0f08b430e6cb4aeebe6a10a3a478a317 : SUCCESS in 1h 51m 02s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/1a7ac0d023d94f3d9627c1e6fc72a1b6 : SUCCESS in 1h 35m 28s (non-voting)\n- kolla-ansible-centos8s-hashi-vault https://zuul.opendev.org/t/openstack/build/be91d42af0044bcb97c268adb219af5d : SUCCESS in 3m 40s (non-voting)","accounts_in_message":[],"_revision_number":85},{"id":"ce9075440579137b114079949071655780e5f7a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-06-01 11:16:57.000000000","message":"Uploaded patch set 86: Patch Set 85 was rebased.","accounts_in_message":[],"_revision_number":86},{"id":"4daa9a945f95e3fbbd928a97c49920539bee2674","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-01 12:41:57.000000000","message":"Patch Set 86:\n\nBuild failed (ARM64 pipeline).\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/6252f6717cdb4c248fc84d11cdf80599 : FAILURE in 58m 45s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/13816d08611f47969b85efc4c458a45a : FAILURE in 1h 17m 09s","accounts_in_message":[],"_revision_number":86},{"id":"071c5341d884e4b1c96362369e75e2ca71718746","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-01 12:56:53.000000000","message":"Patch Set 86: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d7b4b6dfe8704317a4a50b3a48ab872a : SUCCESS in 6m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d8ca895f16941b0b16183a77b98ea46 : SUCCESS in 6m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eaaa12266fdb4ea0854550499b0384a7 : SUCCESS in 6m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d459f9e151f94447969dc33277261102 : SUCCESS in 6m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cb8645789971442baa49839c96b6019f : SUCCESS in 6m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39f07278fb9c49cbbeb6e2d9db6e75b9 : SUCCESS in 5m 41s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/da94a29d7ce947028461690392a27948 : SUCCESS in 8m 00s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/3c140123ce724bc4a5d0ea32b3ebb7b3 : SUCCESS in 1h 23m 29s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/903d59f830934da2ad05236a9fdc4a68 : SUCCESS in 38m 01s\n- kolla-ansible-openeuler-source https://zuul.opendev.org/t/openstack/build/94688dd8ca77451bba9984c9db63bbd0 : SUCCESS in 1h 31m 18s (non-voting)\n- kolla-ansible-rocky8-source https://zuul.opendev.org/t/openstack/build/99381a2e0a6f41c0b5a770c621836419 : SUCCESS in 1h 31m 38s (non-voting)\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/09075dcf24a94a4a975c16512eb162ee : SUCCESS in 40m 55s\n- kolla-ansible-centos8s-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/3a253cdee42c48d1a85bb647283deb3e : SUCCESS in 54m 49s (non-voting)\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/6bd83e3e77c148baa6feca0d5d01a903 : SUCCESS in 45m 40s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/48cd9121358f487d9ddcfe1302a9314a : SUCCESS in 1h 20m 21s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/fbe5bf573b6047f0961add152125494a : SUCCESS in 47m 34s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/9088a152094a4dcc9f2408260e0bc84a : SUCCESS in 57m 53s\n- kolla-ansible-ubuntu-binary-upgrade https://zuul.opendev.org/t/openstack/build/38a9b76f83404050a77ca6c74f108656 : SUCCESS in 1h 00m 43s\n- kolla-ansible-centos8s-source-cephadm https://zuul.opendev.org/t/openstack/build/db7b8b993d1948e4bc0283ee08867000 : SUCCESS in 1h 07m 46s (non-voting)\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/5bd4dd7839964307873c63c84f41851c : SUCCESS in 1h 05m 03s (non-voting)\n- kolla-ansible-centos8s-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/860e3a643224405c99a6ca7d8a210bba : SUCCESS in 1h 23m 14s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/fa3d07b468c64ce892960b46bad3e72e : FAILURE in 1h 06m 55s (non-voting)\n- kolla-ansible-centos8s-hashi-vault https://zuul.opendev.org/t/openstack/build/91093c6bf6ed410aac9c784d96f7a2f8 : SUCCESS in 3m 28s (non-voting)","accounts_in_message":[],"_revision_number":86},{"id":"9fdd16c32498b2db51a141bbcdef0520bee51432","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-06-02 12:52:42.000000000","message":"Uploaded patch set 87: Patch Set 86 was rebased.","accounts_in_message":[],"_revision_number":87},{"id":"98378c391064936e05d21d3630c279f057944c9c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-02 14:51:20.000000000","message":"Patch Set 87:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/8c98126a09224b349f1f117c19c18e75 : SUCCESS in 1h 36m 22s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/65ff745ddade46368b272341c2e74cb3 : SUCCESS in 1h 54m 43s","accounts_in_message":[],"_revision_number":87},{"id":"d7c316f8a5981ddc325aa3eabb6df5ff7dcbd0eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-02 16:51:57.000000000","message":"Patch Set 87: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a9bc5a86795e47c4a16321c801116c43 : SUCCESS in 5m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c6d698895f14101a7aa6c9de18bfa99 : SUCCESS in 5m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/324cc99eb573480dadbebaf04b85399e : SUCCESS in 6m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/071b670364704e5c91e6280bee9b7ab0 : SUCCESS in 6m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/15b220acbdb144eea33386eb8be223fd : SUCCESS in 6m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de12b10638be4192bd15721e0c9339e2 : SUCCESS in 6m 41s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/17de5c99245649b6bcfa88453d6b0523 : SUCCESS in 7m 49s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/51ca9029c2e44256a1a0e8421007092f : SUCCESS in 1h 01m 41s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/c672da207adf4db99bdf945f034c1bbd : SUCCESS in 38m 52s\n- kolla-ansible-openeuler-source https://zuul.opendev.org/t/openstack/build/ca75bb13c07944bbb63f4b1b431f7204 : SUCCESS in 1h 08m 59s (non-voting)\n- kolla-ansible-rocky8-source https://zuul.opendev.org/t/openstack/build/2b285163560b4761991f94630539fa32 : SUCCESS in 59m 26s (non-voting)\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/138482a5a92e451990b7655e565f04f6 : SUCCESS in 58m 18s\n- kolla-ansible-centos8s-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/79b250d9330b46388dd21225b862b887 : SUCCESS in 1h 06m 31s (non-voting)\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/23c7fa30fe2345e6bae49cd5b89b1758 : SUCCESS in 56m 01s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/fe6f5e7169304596b9f8bf4f605d5d14 : SUCCESS in 1h 18m 19s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/2917a355de0e46c79a3dd658e27d6380 : SUCCESS in 1h 00m 06s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/7ca6227c3d994f049b6128f438a3e961 : SUCCESS in 59m 46s\n- kolla-ansible-ubuntu-binary-upgrade https://zuul.opendev.org/t/openstack/build/f609783512934b8998416b6acc431590 : SUCCESS in 53m 14s\n- kolla-ansible-centos8s-source-cephadm https://zuul.opendev.org/t/openstack/build/af6d16f6eacf44e79b068a8071d33d01 : SUCCESS in 1h 34m 58s (non-voting)\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/424a24a8186a4a60937dac27fa58cea8 : SUCCESS in 1h 08m 23s (non-voting)\n- kolla-ansible-centos8s-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/a88708317f0b4d5c814503a8736e5314 : FAILURE in 54m 15s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/82697c14858545d0bc9137d4f8ac7faf : SUCCESS in 1h 33m 42s (non-voting)\n- kolla-ansible-centos8s-hashi-vault https://zuul.opendev.org/t/openstack/build/149d9f9d91ac4bae95c59555cd5c6cf1 : SUCCESS in 3m 29s (non-voting)","accounts_in_message":[],"_revision_number":87},{"id":"e972803f891147b5d65e4eaa9481783f958206bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-06-03 09:08:12.000000000","message":"Uploaded patch set 88: Patch Set 87 was rebased.","accounts_in_message":[],"_revision_number":88},{"id":"d8de1e292f51428604fb95a8ffe4f1aab84a48c0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-03 11:17:32.000000000","message":"Patch Set 88:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/cd829397ddc847c6bacc69446d9b6a8c : SUCCESS in 1h 26m 15s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/95f88f10938b4a9ab1b93d13d1e4e477 : SUCCESS in 2h 06m 14s","accounts_in_message":[],"_revision_number":88},{"id":"50776f7730dcab8d8e63fdc00d13e841754c8b57","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-03 11:17:40.000000000","message":"Patch Set 88: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/55caf20461ec486b917159f2d49e75ca : SUCCESS in 5m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9dff44bd17f42ad9a897d8a9cf5e9bf : SUCCESS in 5m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/695749fb65db42a7bbbaaa49014ac30c : SUCCESS in 6m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c653ec4e3d394026a6efd0b7236ab6cc : SUCCESS in 6m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0f9c437bf9404bcda14d79694e9115fe : SUCCESS in 6m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f7ec9a6c5ab4f7687ba9a26e60022ee : SUCCESS in 6m 07s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/1d05ad2566694509906753ef2092ecd2 : SUCCESS in 7m 03s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/3ee8de0e51194321b0e2ac105950afdb : SUCCESS in 58m 25s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/b474b297c306461e85d37913c50bc226 : SUCCESS in 39m 54s\n- kolla-ansible-openeuler-source https://zuul.opendev.org/t/openstack/build/cd7122faa7164c3b89bd86716c6aacce : SUCCESS in 2h 01m 17s (non-voting)\n- kolla-ansible-rocky8-source https://zuul.opendev.org/t/openstack/build/e021caf524e94fdb8429be2de91e7e83 : SUCCESS in 1h 45m 01s (non-voting)\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/5b8df047f2bf4c0c83abdb0342f1e73b : SUCCESS in 51m 35s\n- kolla-ansible-centos8s-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/b290d3d3410e4181b05e1a9a47e786db : SUCCESS in 58m 41s (non-voting)\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/339d2e1356c041878324496fcb78bfb4 : SUCCESS in 44m 13s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/f22e57cf88214ef484612afae34247d5 : SUCCESS in 1h 10m 33s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/4ce76e653ed34ec3a4e4063179d2ee3f : SUCCESS in 1h 02m 26s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/eeaccfd58b454119b6e00b403593e596 : SUCCESS in 56m 56s\n- kolla-ansible-ubuntu-binary-upgrade https://zuul.opendev.org/t/openstack/build/e90b4393df19447390cf02ed3553a668 : SUCCESS in 57m 51s\n- kolla-ansible-centos8s-source-cephadm https://zuul.opendev.org/t/openstack/build/6a0250c1d557451bb504ab784bdc893c : SUCCESS in 2h 00m 40s (non-voting)\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/7bb1cfe86981499b868acea491e9d2c2 : SUCCESS in 1h 07m 56s (non-voting)\n- kolla-ansible-centos8s-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/9e66f8f0b31c4c02b13e17f744ae98bc : SUCCESS in 1h 42m 44s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/938240bd31c14438b745297570657d89 : SUCCESS in 1h 26m 29s (non-voting)\n- kolla-ansible-centos8s-hashi-vault https://zuul.opendev.org/t/openstack/build/34a7894614aa47e38230a6e7dbf11570 : SUCCESS in 3m 30s (non-voting)","accounts_in_message":[],"_revision_number":88},{"id":"d8edb9af32da7f37f54b5c946f38c879713c3681","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-06-13 17:22:32.000000000","message":"Uploaded patch set 89: Patch Set 88 was rebased.","accounts_in_message":[],"_revision_number":89},{"id":"2dc1eed9fcfa624dc826121f82a3d74b7b703dcb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-13 19:03:17.000000000","message":"Patch Set 89:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/bc0022de46fe401f9f8cc659e21a84ea : SUCCESS in 1h 01m 50s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/605fc8a667b442c48e9707c7dccade4c : SUCCESS in 1h 37m 33s","accounts_in_message":[],"_revision_number":89},{"id":"f9e3c60a5d3b370bad7bdd3e5f711a8df85ac157","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-13 19:26:24.000000000","message":"Patch Set 89: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/28e62f5b883241c29ced3e8a32427baf : SUCCESS in 6m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac5c1c9ac67e4808a66da0dac493265d : SUCCESS in 5m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aba8dd2e3f5d4593845a65a107e749d0 : SUCCESS in 6m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c3be4c2cce824e809d7fd19e17b03d78 : SUCCESS in 6m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8b65ea6f5f1b4647b09a69eb31ffa445 : SUCCESS in 6m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/beb79c053f9c4da9840e42702496a8e7 : SUCCESS in 7m 02s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/9fe534f58e1247c493d137caaeb2c84f : SUCCESS in 8m 29s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/45b1e192fd5346b28b6062343372ff5d : SUCCESS in 53m 06s\n- kolla-ansible-openeuler-source https://zuul.opendev.org/t/openstack/build/10a74678d8664a26ab23c995be0ff174 : SUCCESS in 1h 01m 16s (non-voting)\n- kolla-ansible-rocky8-source https://zuul.opendev.org/t/openstack/build/7876cc3d44a0431985441bb651ab8efe : SUCCESS in 55m 11s (non-voting)\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/beed264dfa574421a00f1ab4e801aa87 : SUCCESS in 43m 51s\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/b9e774eaf20d445aa3d967f0a3656f14 : SUCCESS in 49m 03s (non-voting)\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/61206557605646919273f94bbd401f61 : SUCCESS in 57m 40s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f820d74a723c44868c803c03302bbaa2 : SUCCESS in 56m 07s\n- kolla-ansible-ubuntu-binary-upgrade https://zuul.opendev.org/t/openstack/build/4b2728485e4b40f7a7d65f469ecd07f7 : SUCCESS in 55m 37s\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/7ef02a6d44804b3888260ad42841b3fd : SUCCESS in 1h 02m 56s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/bda38351f40849abaa1127895d86468a : SUCCESS in 1h 27m 55s (non-voting)","accounts_in_message":[],"_revision_number":89},{"id":"e74f23b418b89b50f62df60475a27921bf0b8b60","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-06-17 17:07:34.000000000","message":"Uploaded patch set 90: Patch Set 89 was rebased.","accounts_in_message":[],"_revision_number":90},{"id":"c4bc7569e624a917aa87b167faed1c1f3f875f44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-17 18:39:36.000000000","message":"Patch Set 90: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/498611b3341c45538bc85da6453a3942 : SUCCESS in 6m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1fb9ebc773843248a78eaa2f10b430a : SUCCESS in 6m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4ea14d20d3fb45ad948250aef672e578 : SUCCESS in 6m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/03b6e54c9796486b8bf83050c29e421f : SUCCESS in 7m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/17ce3c7e9a0547f19cf1a4d81f15fbc4 : SUCCESS in 6m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/585c922a6d1948b0a6925afa78df0c02 : SUCCESS in 6m 18s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/e2469d797a954f5387f018a9ff88a6a5 : SUCCESS in 8m 37s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/c4ff78b4433d4442bb12f3306e41246c : SUCCESS in 1h 01m 20s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/95ecd90dc4d24f9391cb99bdc30233d8 : SUCCESS in 40m 13s\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/445882631f7041f999449b8e5e151bea : SUCCESS in 1h 01m 23s (non-voting)\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/e24eebfdb14546cc963bb9676372de6b : SUCCESS in 59m 21s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b5d8c5a83e7e428fad8951dff549a8d3 : SUCCESS in 49m 47s\n- kolla-ansible-ubuntu-binary-upgrade https://zuul.opendev.org/t/openstack/build/38ea825f997642a5944a3c34348abc39 : SUCCESS in 1h 00m 52s\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/8897bfc958fb4d749532f7e9248fd039 : SUCCESS in 1h 10m 29s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/8a9bc25643304048bccc8b0a23216122 : SUCCESS in 1h 29m 23s (non-voting)","accounts_in_message":[],"_revision_number":90},{"id":"f9bdd59033f09382a928a58ca63bdd5023c9d9d9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-06-17 18:59:52.000000000","message":"Patch Set 90:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/767b73efb50e4ef4805caf2d43fb6cd8 : SUCCESS in 1h 25m 31s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/d8c864ee90b3458aac53149a39059815 : SUCCESS in 1h 48m 55s","accounts_in_message":[],"_revision_number":90},{"id":"f6257f5449e727142287c39a4d586b4a04437e89","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-07-02 13:18:10.000000000","message":"Patch Set 90:\n\n(1 comment)","accounts_in_message":[],"_revision_number":90},{"id":"066b2cbb38d78a1d7b51a09828b1ca59a953d6e4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-07-04 16:42:41.000000000","message":"Uploaded patch set 91: Patch Set 90 was rebased.","accounts_in_message":[],"_revision_number":91},{"id":"2be4d3901011a4af07d8eba0ffc65efaee54867e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-04 18:09:11.000000000","message":"Patch Set 91: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b4129e342c447939d761910dec73d57\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e8a2395d12854e71ab38af399d50af16 : SUCCESS in 7m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8efe2713617549d19a466561c86db530 : SUCCESS in 6m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7fec220a48db401998945d00e1b1504c : SUCCESS in 6m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c6512159c8ff4ef89de893d8fdee8221 : SUCCESS in 6m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4d3e668566e340ebae007480e2639d78 : SUCCESS in 7m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/026b801d20134582995be9929354ab3c : SUCCESS in 7m 02s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/775de88de0ec42659bfe11263166425b : SUCCESS in 7m 46s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/ba1b6b90a95347548afe3a986a68bf5c : SUCCESS in 51m 39s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/3f53b60f280946098ce5472478ffee11 : SUCCESS in 47m 03s\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/f66f9b673a934a4f9b9bc6770ef06f0d : SUCCESS in 53m 40s (non-voting)\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/e5a4084d51aa47769315fcc73813d4a0 : SUCCESS in 54m 33s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f37105e8554b48c0865f825ce57b7397 : SUCCESS in 59m 51s\n- kolla-ansible-ubuntu-binary-upgrade https://zuul.opendev.org/t/openstack/build/9219f158ead04331ac1b3b4251bcd9ef : SUCCESS in 53m 29s\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/c2b1340be44d44f48b882ba372310b78 : SUCCESS in 1h 10m 03s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/cf9f06e775a24caa98e87ac01f840dc2 : SUCCESS in 1h 22m 06s (non-voting)","accounts_in_message":[],"_revision_number":91},{"id":"a7a834cb3d6b9b19b4cd0d543dcb7a9d71e55447","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-04 18:13:10.000000000","message":"Patch Set 91:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8413ce69d2a94d7da731704b0fe74307\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/035081712e864c04b6dab99aae374bd9 : SUCCESS in 1h 09m 45s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/f812e56e1aa6423481b9e0bbb155115d : SUCCESS in 1h 21m 30s","accounts_in_message":[],"_revision_number":91},{"id":"81e4f983e28e28e3a84d761a6d7c86316f97faeb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-07-04 18:26:36.000000000","message":"Uploaded patch set 92: Patch Set 91 was rebased.","accounts_in_message":[],"_revision_number":92},{"id":"15d8f379fd78a5affdc8e2483703a6c53815f2d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-04 19:55:11.000000000","message":"Patch Set 92: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5eeab4403855472484d20f41b6dfdcde\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/02a05287e30a450d9062733223048f03 : SUCCESS in 6m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2abc9b2454c242fda773d154f84ae324 : SUCCESS in 6m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f63fdfb7b0584f1b91e16e0fa4c21c38 : SUCCESS in 5m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b2b97aeb74cb4c09969a6e41fe636fda : SUCCESS in 5m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7c1f30f3f1254423aa283dcd40c90a36 : SUCCESS in 6m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05339f3b843b40a18b4104b68be368ba : SUCCESS in 5m 23s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/ebb80498ca6a4f37a58644d01770585f : SUCCESS in 8m 12s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/697aa3dbe42e453d8a58abdae9d6df71 : SUCCESS in 57m 09s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/0bbe0dd9ce534b1bbb2d76d9b4e231d4 : SUCCESS in 45m 08s\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/45176a44d85f49148e340f07b01778e7 : SUCCESS in 49m 59s (non-voting)\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/77383fe9779f43a8b97336fa11fdfc5d : SUCCESS in 1h 03m 29s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/89f1546f10c142979e6aa9f317c37b87 : SUCCESS in 50m 57s\n- kolla-ansible-ubuntu-binary-upgrade https://zuul.opendev.org/t/openstack/build/1f207ac536924b05b7cdaae049c35e7d : SUCCESS in 51m 39s\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/d86e4b1dd5fc45c38ff6e55f210a4540 : SUCCESS in 1h 06m 12s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/1c386dfb97424a438b1019b5f1985377 : FAILURE in 1h 23m 10s (non-voting)","accounts_in_message":[],"_revision_number":92},{"id":"31cce9eea99b3374787126a109a1ccd40fe5e312","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-04 20:02:57.000000000","message":"Patch Set 92:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5daa1e105007400ea4f59f0b1fb61bf7\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/29b420edfddc48969f82e401d5fc06f0 : SUCCESS in 1h 06m 32s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/47afc688eeac469a918966b3909ea3e3 : SUCCESS in 1h 33m 45s","accounts_in_message":[],"_revision_number":92},{"id":"3a9f11e3a77c907006134d6e18ec327f7ad67ade","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-07-05 21:02:41.000000000","message":"Uploaded patch set 93: Patch Set 92 was rebased.","accounts_in_message":[],"_revision_number":93},{"id":"c256181a3ec9b47ca5cf7735ef6336709b5e6d49","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-05 22:12:26.000000000","message":"Patch Set 93: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c85077d860f44ea3881e108fb7e27d3a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/63ca34f654e64f2284cb27757a9ee580 : SUCCESS in 6m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c78b7bb6daf14a5a8bbe960de2125d32 : SUCCESS in 5m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/55fe39aabdb845b5b28d91031082cc3d : SUCCESS in 6m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e5ce302583e54d8f982f3e3edfd0ddec : SUCCESS in 7m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/21813587224e4c65b20d64fb35153b1e : SUCCESS in 6m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6e93f003ae341e9890b394cc94a32e2 : SUCCESS in 6m 46s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/9803bdf616d84ea79eb562c3dd69379e : SUCCESS in 6m 59s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/74f5093a11b6448a89342f250b576793 : SUCCESS in 43m 40s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/eec7f2d1d7d54e929c36d6dbf3cc31ad : SUCCESS in 46m 43s\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/b6574013a446457abac4d79018f1bb1b : SUCCESS in 47m 09s (non-voting)\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/5f7685aca9b648ec91ef3f741ad6d0c5 : SUCCESS in 58m 04s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d273abee2f044983a067718c613cb5e9 : SUCCESS in 53m 27s\n- kolla-ansible-ubuntu-binary-upgrade https://zuul.opendev.org/t/openstack/build/e3a33eb8f33d435d863c1b2e0003168c : SUCCESS in 50m 33s\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/e2d5850d53c3498eb0c2bf4a7c6bafa6 : SUCCESS in 1h 07m 09s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/0d7072248cb943038c46a951af9492a3 : SUCCESS in 1h 07m 17s (non-voting)","accounts_in_message":[],"_revision_number":93},{"id":"b88282eb77997bef97c58705e10abb237ded30a0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-05 22:34:49.000000000","message":"Patch Set 93:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6113c081d6004ad8abad4029f1cf5e8a\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/63a8c03600b4454c8701f77e9589502d : SUCCESS in 1h 14m 08s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/70c337fdad52412c83bc67a381806d5e : SUCCESS in 1h 29m 20s","accounts_in_message":[],"_revision_number":93},{"id":"b61cac2f4b533b119f47df59edf5b4f168115f5f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-07-06 14:58:32.000000000","message":"Uploaded patch set 94: Patch Set 93 was rebased.","accounts_in_message":[],"_revision_number":94},{"id":"7ec5dcee459e791be2ad04251f20f87c5d5ad04b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-07-06 15:04:38.000000000","message":"Uploaded patch set 95: Patch Set 94 was rebased.","accounts_in_message":[],"_revision_number":95},{"id":"ff96e4cecd56c7dcf75f9475ab8be21dd5603cec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-07-06 16:00:06.000000000","message":"Uploaded patch set 96: Patch Set 95 was rebased.","accounts_in_message":[],"_revision_number":96},{"id":"412d696db7485efa1948b0e7b306a68a8ecf520b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-06 17:31:18.000000000","message":"Patch Set 96: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa16ae56924e4895975886cd055e5488\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e036839f1353459dbe4c5cc400127337 : SUCCESS in 5m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aeacf8beacf14f5190f0e8a377bcd6c1 : SUCCESS in 6m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6ba4e9c5e1794373adb1aaed2dfe2f4c : SUCCESS in 5m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/22e0cb57149c40cf9d1357438beb7d19 : SUCCESS in 6m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/55c62ef4c2c2454f921c9b664410b51c : SUCCESS in 6m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91dd74a78e2e4ab5b9d7951ff7f0bf68 : SUCCESS in 6m 32s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/3cb4a0ac221e4768861c4cfab1c7d91f : SUCCESS in 8m 09s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/a0e4b3fb2dec42bdad404b3ca5a662e2 : SUCCESS in 51m 10s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d6f361e04ea64ff78ef2fdc9a7bdda20 : SUCCESS in 42m 15s\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/75919493de0e4c9680b4bd5ddf5fe32a : SUCCESS in 55m 31s (non-voting)\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/ea422c54b31349f28e31e2f757d006a3 : SUCCESS in 51m 05s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b323219a8a8a420d8f85d75e23883cc8 : SUCCESS in 52m 21s\n- kolla-ansible-ubuntu-binary-upgrade https://zuul.opendev.org/t/openstack/build/751835ee4a8c4a55b4bc6319c0af20db : SUCCESS in 59m 03s\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/5d55da2500bb4f43b6bce4b37a747edc : SUCCESS in 1h 02m 40s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/32c0a9ffd871467e99b862209d25a5f9 : SUCCESS in 1h 27m 40s (non-voting)","accounts_in_message":[],"_revision_number":96},{"id":"c32d1f465c51b106541c00ee65bf69d5e1d5e4f8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-06 17:42:12.000000000","message":"Patch Set 96:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3fd9e792c618440098e191052a9274f9\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/caee257c8fe546a180ccef04c972c51b : SUCCESS in 59m 55s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/8d831e774fbf4ea9af0172902e891919 : SUCCESS in 1h 39m 26s","accounts_in_message":[],"_revision_number":96},{"id":"c0006aa510a1290dc9cd0313123d5c82befeb452","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-07-07 08:22:32.000000000","message":"Patch Set 96: Code-Review+2","accounts_in_message":[],"_revision_number":96},{"id":"cdb048d6e239e64192bbec47e462b655b133e460","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-07-12 16:07:38.000000000","message":"Uploaded patch set 97: Patch Set 96 was rebased.","accounts_in_message":[],"_revision_number":97},{"id":"bffbc7deff383e1442080bfaf3dd503cd753e14a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-07-12 16:25:31.000000000","message":"Uploaded patch set 98: Patch Set 97 was rebased.","accounts_in_message":[],"_revision_number":98},{"id":"9bf8d395af61807668721b42834126aea5f744ee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-07-12 17:53:24.000000000","message":"Uploaded patch set 99: Patch Set 98 was rebased.","accounts_in_message":[],"_revision_number":99},{"id":"f2461cacf414f3920f014ec445de134646603bd3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 19:12:44.000000000","message":"Patch Set 99: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f1eaded2bfa749b6bbc54e140a331584\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9c468b5ba73c4574a0dac0bb8dbde24c : SUCCESS in 7m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/40c409b77922405ba84dc9325c55553f : SUCCESS in 7m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ec2c5ef25f454df4a6cd18ffee2f70cd : SUCCESS in 7m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2cd7996307894d99b10ed4d32ee32717 : SUCCESS in 7m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9d6738449d65492fa1e40f8c286c448e : SUCCESS in 6m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c80371ed5fe4b52a1b76f9b69e91519 : SUCCESS in 8m 52s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/30a1f56efc2b44a6b4f1f57e4fbb0a14 : SUCCESS in 7m 20s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/943cf027618640c0ac0611788fbde32b : SUCCESS in 50m 29s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/20ea553c89a34931ad268349d80f856a : SUCCESS in 52m 20s\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/1335c1716ff84c6ba2273fbe5f93706d : SUCCESS in 40m 44s (non-voting)\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/a80e495d420047adb3f9a6f3c0e75408 : SUCCESS in 1h 03m 09s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/3e0dfce2d7e540e0ba23e9da4978f952 : SUCCESS in 59m 03s\n- kolla-ansible-ubuntu-binary-upgrade https://zuul.opendev.org/t/openstack/build/c6e1f70d672b49f3a9b9d10581c91dec : SUCCESS in 57m 42s\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/581ba07bb29a4b3a9516e5f0094ac0ad : SUCCESS in 1h 14m 19s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/07d39ac3e8044f3c83c028d5476e79c5 : FAILURE in 28m 28s (non-voting)","accounts_in_message":[],"_revision_number":99},{"id":"8a7af37fc5e417b8aff3f0d90c73945585eada94","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-12 19:47:22.000000000","message":"Patch Set 99:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f8a02b4ab2924683b31510db5c8635d6\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/1b8d7bf89bf2425a9d3c12da69d7de63 : SUCCESS in 1h 08m 22s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/406b9dba8c1840c0b70d101c98abe610 : SUCCESS in 1h 50m 33s","accounts_in_message":[],"_revision_number":99},{"id":"6ea930bd1f492080333a537889795ce82e4a5c3f","author":{"_account_id":29759,"name":"zhangmeng","email":"zhangmeng9502@gmail.com","username":"zhangmeng9502"},"date":"2022-07-13 06:45:42.000000000","message":"Patch Set 99: Code-Review+1","accounts_in_message":[],"_revision_number":99},{"id":"366316f8daa12736f25aa29b9e0fd00e6ede9bf6","author":{"_account_id":33493,"name":"chenwei","display_name":"chenwei","email":"jzpcoming@163.com","username":"chenwei"},"date":"2022-07-13 09:34:33.000000000","message":"Patch Set 99: Code-Review+1","accounts_in_message":[],"_revision_number":99},{"id":"0ccc36221e7dfaeae3af7b48c616a07a2e5fc195","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-07-13 14:23:36.000000000","message":"Uploaded patch set 100: Patch Set 99 was rebased.","accounts_in_message":[],"_revision_number":100},{"id":"a6ac35036a76349da7dbe2fbf9fb32856e597217","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-07-13 14:58:36.000000000","message":"Uploaded patch set 101: Patch Set 100 was rebased.","accounts_in_message":[],"_revision_number":101},{"id":"09a6b05b87da4f77d3af4d5a259609b7e36fea59","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-07-13 16:17:41.000000000","message":"Uploaded patch set 102: Patch Set 101 was rebased.","accounts_in_message":[],"_revision_number":102},{"id":"3934fe314a71f8ed81093fb4f10bdb5752b46f86","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-07-13 16:35:30.000000000","message":"Uploaded patch set 103: Patch Set 102 was rebased.","accounts_in_message":[],"_revision_number":103},{"id":"f5280bf624000427cbf2874413f5bef282d8a429","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-07-13 16:52:09.000000000","message":"Uploaded patch set 104: Patch Set 103 was rebased.","accounts_in_message":[],"_revision_number":104},{"id":"70ff2b1b6825dc9481190656beec3cdc6b0e5eef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-13 18:30:10.000000000","message":"Patch Set 104: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/798b75f4ee9041a998b22c69c80d1434\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bc15932bfbd942d99b09fcb783ca010a : SUCCESS in 6m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bd309236dd8048f282838bee88269893 : SUCCESS in 7m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c39ad0e324c74ebdbce431857a3a2de6 : SUCCESS in 7m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a9b56e7065ed46da8ae32b131a0aeb8a : SUCCESS in 7m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f943a4d3cbed42c8924de75c17195d74 : SUCCESS in 7m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/017d988e37cc402aad1c97c6dc25a818 : SUCCESS in 7m 18s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/33465af7fe3b496ab740696aaec4569d : SUCCESS in 8m 14s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/0c99c67c23bd4b4b96c697aa10b690b2 : SUCCESS in 49m 54s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a83fc0b9b5d54c328960a022e322668b : SUCCESS in 47m 50s\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/0b05293f6b1241b694c46a870189ad70 : SUCCESS in 53m 07s (non-voting)\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/395aec876f0e4be48368f8af32620e83 : SUCCESS in 1h 02m 16s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/666e1a2105dc40ae8fe1335b3a42b106 : SUCCESS in 1h 00m 48s\n- kolla-ansible-ubuntu-binary-upgrade https://zuul.opendev.org/t/openstack/build/d435b4cd3319463fa45a9c488e3348ba : SUCCESS in 1h 01m 12s\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/92d7a545efd7458295bc83fe25ca34bb : SUCCESS in 1h 09m 09s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/ea4a5697fae84ea5b56596dceb41467c : SUCCESS in 1h 31m 32s (non-voting)","accounts_in_message":[],"_revision_number":104},{"id":"1788d370d56545ca7bce6a2495a6ce965fc00339","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-13 18:44:28.000000000","message":"Patch Set 104:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b3fe0f7b46e74a3389d8093545593c54\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/e92ad18b01604c2d86646174d9e202a0 : FAILURE in 1h 13m 28s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/df6d42cb2d95472195e0b5dc4eefb189 : SUCCESS in 1h 49m 14s","accounts_in_message":[],"_revision_number":104},{"id":"f1028a69ccc43dd97001ebee453d2d5244e82bd9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-07-13 19:29:45.000000000","message":"Uploaded patch set 105: Patch Set 104 was rebased.","accounts_in_message":[],"_revision_number":105},{"id":"12875a9be09974bb0d564aa12c23ba024a875d92","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-13 21:07:50.000000000","message":"Patch Set 105: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6e3af66ea848476e944d87cd085ced22\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/888f8e24e77e4c9b8ca40047ba548f6d : SUCCESS in 6m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b0d6c8b730149b0b8567df11c2f7a79 : SUCCESS in 6m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/719a48479993446bbf418ba0997bae2c : SUCCESS in 7m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a601370bfd4c45e8836cac5ad4ecb633 : SUCCESS in 6m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/71e114c9cf2b4371936969bf755508ec : SUCCESS in 5m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18c434be09674c6ca6fff51671c6e375 : SUCCESS in 5m 16s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/2ac23a6c23c54055b2c1e2360ca28b79 : SUCCESS in 8m 31s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/c978d9190ed341bcb34a15f582515108 : SUCCESS in 57m 45s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c632b4fd46a14c61b9e80a6fb838daae : SUCCESS in 56m 01s\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/1b83110803ba45a192c33cfe7930291f : SUCCESS in 56m 34s (non-voting)\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/2810a8bf00844354a681d3877e55c574 : POST_FAILURE in 1h 03m 48s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/74389573257c46b58404ccce73ac45dc : POST_FAILURE in 1h 30m 11s\n- kolla-ansible-ubuntu-binary-upgrade https://zuul.opendev.org/t/openstack/build/ea79a4b8d2c94000b7adc9673cdf23ce : POST_FAILURE in 1h 03m 45s\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/04518b7f78a94f229001eb4ba1251330 : SUCCESS in 1h 13m 58s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/04711e9430a049b3a241841e9bb34fbf : SUCCESS in 1h 29m 01s (non-voting)","accounts_in_message":[],"_revision_number":105},{"id":"f74b2ce8685db951cd95f03bcec7ac80373ce5e0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-13 21:31:46.000000000","message":"Patch Set 105:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e6e4df3e51f64508ba3b78cc8612a8b9\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/44e3631468684365960eb7e64b24c159 : SUCCESS in 1h 00m 46s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/54f07612a27b4d2a8b184d9b12d726b9 : SUCCESS in 1h 59m 15s","accounts_in_message":[],"_revision_number":105},{"id":"c16c3a4ad216ed683ef81b789a2bca5e96e65727","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-07-21 14:21:34.000000000","message":"Uploaded patch set 106: Patch Set 105 was rebased.","accounts_in_message":[],"_revision_number":106},{"id":"fd8d04d52d462e8abbec6f716fdd5721b66bfa12","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-07-21 14:29:12.000000000","message":"Uploaded patch set 107: Patch Set 106 was rebased.","accounts_in_message":[],"_revision_number":107},{"id":"0f88f4afad3724b91586c09e667ab4ed03cf6a2a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-21 14:58:35.000000000","message":"Patch Set 107: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d5aa617bfae943148b4874a41081bab9\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/175aaa67a32e4129b2830909af7567fd : SUCCESS in 9m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ded890e37a6d48859203ea755f92fdaf : SUCCESS in 9m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/210d03a758a84baeab919188b58e85e3 : SUCCESS in 9m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ac04e664b8f84df39ec8c702e2a6cb46 : SUCCESS in 9m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8c12ec733f9545d7a1cc74a28d8b2156 : SUCCESS in 8m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/435969c1e8714a4592f07b51779a6af9 : SUCCESS in 9m 47s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f82a730b864845c2bdf4362e702cf6e6 : SUCCESS in 10m 04s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/ce9203c7d3a74f418f69566088299c0b : FAILURE in 22m 35s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/263f9f6b08e646729dbd85f4420c3904 : FAILURE in 20m 19s\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/a05c80f732f14fd4abd7a7b40e76953f : FAILURE in 15m 38s (non-voting)\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/517dc0ec7b2f488699d594198bd62c41 : FAILURE in 19m 19s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/24541374dd5e449993aaec61d4a905f4 : FAILURE in 22m 02s\n- kolla-ansible-ubuntu-binary-upgrade https://zuul.opendev.org/t/openstack/build/e390a237bb07498a89ee297a55902092 : FAILURE in 21m 50s\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/95198317478847e5b6bc9b8bf3a987d6 : FAILURE in 19m 33s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/309a1048108a4858b66778b66a2fbfde : FAILURE in 15m 22s (non-voting)","accounts_in_message":[],"_revision_number":107},{"id":"36efa82df1c459cd9dd1fb8cda7b518d1f7cb88f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-21 15:07:40.000000000","message":"Patch Set 107:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e4b6c3887bae42a5b1aad33ed52f3a76\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/ef9f723799b04b2f81809e74f6e6e81c : FAILURE in 26m 40s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/f7b71ecdd3764776be07f6599e8f6e8f : FAILURE in 28m 05s","accounts_in_message":[],"_revision_number":107},{"id":"0916728e24024616a33b8fa5ac92e089a632bd12","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-07-21 17:01:33.000000000","message":"Uploaded patch set 108: Patch Set 107 was rebased.","accounts_in_message":[],"_revision_number":108},{"id":"3eb91b1f382e336f15a265ece51d7990f4b3b1aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-21 17:35:48.000000000","message":"Patch Set 108: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/34cbbfaf6de64a4f8973eca0008f4023\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f69f00eb09424ea3aed95784fcb051dd : SUCCESS in 6m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4a0e59791bc47c691ac990ecb5b66e3 : SUCCESS in 6m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9bad57c712fe442dacd92eecf3b32e55 : SUCCESS in 7m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1e6c8f81d3304f46970b613b383dc4db : SUCCESS in 7m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c245a7269d334c249ca58659a8a13cc9 : SUCCESS in 6m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/605bb3a8a72e4d38bd924d11b50cae73 : SUCCESS in 7m 34s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f5eaa82a468846b39f6e3833b9e573d7 : SUCCESS in 7m 40s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/9c642bb6af09443cbc879d67b870d0c4 : FAILURE in 19m 16s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/d81df51b090b4662b85ecd3a05f3cc56 : FAILURE in 16m 28s\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/dfffcebc3f5d4696a27d773a7873a4fb : FAILURE in 17m 45s (non-voting)\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/fb96abd532ca4250ab2284430c177e10 : FAILURE in 31m 45s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/348a40f96ffd4a3da5699a824615fe8b : FAILURE in 16m 20s\n- kolla-ansible-ubuntu-binary-upgrade https://zuul.opendev.org/t/openstack/build/0628b13f5f1f45c590ab4dbb4820e219 : FAILURE in 17m 39s\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/2f7dbc1ef48947db852d59ac236d347a : FAILURE in 17m 13s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/e2023b489444409295424d385679ea24 : FAILURE in 18m 39s (non-voting)","accounts_in_message":[],"_revision_number":108},{"id":"5bb66784399410f366391bf38145a74a545aa84c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-21 17:38:10.000000000","message":"Patch Set 108:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/619c6b1fa74b466195ac96e5c5ae456d\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/9b80b20b3fa042adb25c4a030f268972 : FAILURE in 26m 00s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/ed1d8b335bf24c629b14dc8982c6db98 : FAILURE in 28m 43s","accounts_in_message":[],"_revision_number":108},{"id":"123eeebf399c48b546cc9473614953a9212a6dca","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-07-22 10:32:24.000000000","message":"Patch Set 108:\n\n(1 comment)","accounts_in_message":[],"_revision_number":108},{"id":"3109bdd1be33b696e94cfde22b4e947ad614ecaa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-22 11:53:17.000000000","message":"Patch Set 108:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":108},{"id":"66cc4cb0fe68baf758a8b3542ccc3e966d1c0689","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-22 11:53:18.000000000","message":"Patch Set 108:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":108},{"id":"ad6e92b59424b1cd82d369f1ea22398b79f3f4f5","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-07-22 12:18:04.000000000","message":"Patch Set 108:\n\n(1 comment)","accounts_in_message":[],"_revision_number":108},{"id":"c427f88d3d3f4b206a855c701d635a5e8e0d01f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-22 13:30:00.000000000","message":"Patch Set 108: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/50652bdd88d64441ad4993ea96a17a75\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cb607139ca77463eaf9dd5eced5e9117 : SUCCESS in 7m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ccc25954f61d4dbeaa9d740395bb14d4 : SUCCESS in 6m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f7ed92ed401b44bda72f8c007e862899 : SUCCESS in 5m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/de16efd274bf435e8c68a655ee0e458d : SUCCESS in 5m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/94ab79b74bf0456f8bbb6f934f0c7873 : SUCCESS in 5m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e892e433e8c9416f91f7546856053d47 : SUCCESS in 4m 52s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/83aa9694090d4580986cc19c705dcc46 : SUCCESS in 7m 39s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/1eb0a195c83a4e54ad3b24a40da71d1f : SUCCESS in 57m 10s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/95e695e3f1c74fd1b0bf23c58a286d44 : SUCCESS in 49m 43s\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/b422de1ef9c440769a3dc99b8b4b71ea : SUCCESS in 59m 51s (non-voting)\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/b1f5755f3e704c5aaf36b0fd91c27e95 : SUCCESS in 1h 07m 37s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ac15ede9262847d69fc9e8903d532814 : SUCCESS in 1h 06m 56s\n- kolla-ansible-ubuntu-binary-upgrade https://zuul.opendev.org/t/openstack/build/cce41018b2d445c5966ff4da545f49db : SUCCESS in 50m 31s\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/a19ea7d4f4d445489de53060e2ca3dc5 : FAILURE in 20m 00s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/6c002ab5e5ad431ba0979b1ec7ef69d1 : FAILURE in 18m 57s (non-voting)","accounts_in_message":[],"_revision_number":108},{"id":"6eda0d699ce136e01f15d1361463d8e783a5f769","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-22 14:04:33.000000000","message":"Patch Set 108:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b858792d2bc5433183d3a0a51172e742\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/16c0a96559d24886a2cdfa69f46a08a7 : SUCCESS in 1h 44m 30s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/4e6a411f17fb4c41abc3a81284ff2e5b : SUCCESS in 1h 32m 35s","accounts_in_message":[],"_revision_number":108},{"id":"7674e31a4bcb10e40633abe8b71ef9801c4fe9af","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-07-23 18:01:00.000000000","message":"Uploaded patch set 109: Patch Set 108 was rebased.","accounts_in_message":[],"_revision_number":109},{"id":"a1a29eae76603f490490275c69bcf65ae1e66a2d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-23 19:24:30.000000000","message":"Patch Set 109: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a66fe25ddd944ad5930cb375b0d0434c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b2c78403213c492e903e8e9c35586608 : SUCCESS in 7m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/09e71a219e9341809c47f8626440d392 : SUCCESS in 7m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6db90406457340549f212505926c9680 : SUCCESS in 7m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dff121f6c42c4cb882131f6a3542c69c : SUCCESS in 7m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c2b1b3f01001421a9aa1f4fb8e12bb44 : SUCCESS in 6m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ddc168505a045e7a78c08645d5605e7 : SUCCESS in 6m 26s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/f9bc5103a71b45a2b08ebc2e4ed7f5b2 : SUCCESS in 9m 09s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/0e89b42cb2f9444f8d76c4a19c676539 : SUCCESS in 47m 12s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/5d8ffdaf9d6d4f36ad326a3e0a8a6cd0 : SUCCESS in 55m 55s\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/15f50355401141ffad80a09aa05403f2 : SUCCESS in 1h 03m 51s (non-voting)\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/40cfa76dcbd64db6b84e27a9fd9a4cd5 : SUCCESS in 1h 16m 27s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/69b31f7529d34edabfcd72cd4338ec8c : SUCCESS in 1h 03m 38s\n- kolla-ansible-ubuntu-binary-upgrade https://zuul.opendev.org/t/openstack/build/9759f04875914ca08bd3fa19468c07ff : SUCCESS in 1h 08m 13s\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/85cf73f10cfa456f864a957e57eb13f8 : FAILURE in 23m 04s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/ce13483010ed40f5b8be6f0b1d3a7a6d : FAILURE in 21m 55s (non-voting)","accounts_in_message":[],"_revision_number":109},{"id":"0ccb9a126243ac81c9d45560b8d92fd5ae871a52","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-23 20:12:44.000000000","message":"Patch Set 109:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/40645e1abead4f3e82e715942bccae6c\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/df41412ce94947fab761667889ca3f71 : SUCCESS in 1h 46m 14s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/f343874a43864ef9b24a468f8c412c36 : SUCCESS in 2h 08m 54s","accounts_in_message":[],"_revision_number":109},{"id":"b5cc286ee7dbb73afae167e34143f066a765ff11","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2022-07-24 14:55:37.000000000","message":"Patch Set 109: Code-Review+1","accounts_in_message":[],"_revision_number":109},{"id":"bcbdc4dce5e93d04f1d4cdf280de6a6b213399e3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-07-26 09:12:43.000000000","message":"Uploaded patch set 110: Patch Set 109 was rebased.","accounts_in_message":[],"_revision_number":110},{"id":"569fced48677b4a39dcda7f08953d01047390cfb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-07-26 09:40:11.000000000","message":"Uploaded patch set 111: Patch Set 110 was rebased.","accounts_in_message":[],"_revision_number":111},{"id":"78a6d0b0e177080f5d710c9518a907c765fb7b38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-26 10:52:37.000000000","message":"Patch Set 111: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2271c9c165a24e7cb70e1aa08749f57a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2c623f47235d49d6aa05b987de7d1c85 : SUCCESS in 6m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2534e565f95645a4963d42d9cc6d632d : SUCCESS in 7m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d622996a0707496e8ae2c57047f10f70 : SUCCESS in 8m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d6c2497d8ee34bfe9fe0f17479dfce9d : SUCCESS in 7m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e408ec1dc6a94b87a4094fd677e548ee : SUCCESS in 6m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9ea99408f9d4f45890bac2784069012 : SUCCESS in 8m 23s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/8af9bec99b494ea5a36eb27375f094ef : SUCCESS in 9m 00s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/2c23db13893a4eb1a5930e9e46192b13 : SUCCESS in 46m 19s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/05ce32d9851c480d87bc703dc6ef2b14 : SUCCESS in 55m 43s\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/41b03f5ad19d4e398da338fb711f0b13 : SUCCESS in 1h 04m 52s (non-voting)\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/0b378a7f0f8344f6b983379950140649 : SUCCESS in 1h 03m 15s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/389c76e480e74dbfbf82169fa52602c1 : SUCCESS in 1h 09m 28s\n- kolla-ansible-ubuntu-binary-upgrade https://zuul.opendev.org/t/openstack/build/65e3cd4f4b5943ab81c080d594045d80 : SUCCESS in 54m 59s\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/8770230c648b46cb8be046d8c748388c : FAILURE in 22m 00s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/3d5935ca826b47039ea3f9b1ccd3d8c3 : FAILURE in 18m 40s (non-voting)","accounts_in_message":[],"_revision_number":111},{"id":"9088c6fd17385e85f03ee119ae3cfa7f6fd3f573","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-26 10:59:00.000000000","message":"Patch Set 111:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/419c008fed2542398b318ba066870f86\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/bc2f76163f2c47babf9a4a015f9d05ba : FAILURE in 1h 09m 09s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/0c3cd96177be4b76be632d34dba6e7d4 : FAILURE in 28m 54s","accounts_in_message":[],"_revision_number":111},{"id":"7510258c4e4f1661ddf3cb37d8f6eca53218a86d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-07-26 12:26:43.000000000","message":"Uploaded patch set 112: Patch Set 111 was rebased.","accounts_in_message":[],"_revision_number":112},{"id":"635710bf82253468928da7c336d655f35447fae9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-07-26 12:28:18.000000000","message":"Uploaded patch set 113: New patch set was added with same tree, parent, and commit message as Patch Set 112.","accounts_in_message":[],"_revision_number":113},{"id":"42c44804cb2d2aac6b053b348a06d8d92fd9bf36","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-07-26 12:30:21.000000000","message":"Uploaded patch set 114: Patch Set 113 was rebased.","accounts_in_message":[],"_revision_number":114},{"id":"5fef0f7f5e0c95dbc77cac003a776abeb057d4f8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-07-26 13:44:52.000000000","message":"Uploaded patch set 115: Patch Set 114 was rebased.","accounts_in_message":[],"_revision_number":115},{"id":"38de296399a4e40d9bbd11516db3bc69249f41ef","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-07-26 13:49:44.000000000","message":"Uploaded patch set 116: Patch Set 115 was rebased.","accounts_in_message":[],"_revision_number":116},{"id":"fda98f10657ee783adaeba8bb36a5353deeb5c22","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-26 15:13:27.000000000","message":"Patch Set 116: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/85773c89f83343f898ed84f526356337\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7625fb0b91444bae83a26583afd6de83 : SUCCESS in 6m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e031e22dd1ad41f08ce200da789b14b5 : SUCCESS in 5m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/685694b694da4a6da12a499bca72c138 : SUCCESS in 6m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bad5394d038544bcac548baa277a305a : SUCCESS in 7m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4ede60a0132245fab9c744db3ce82471 : SUCCESS in 6m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0fecf72eb11b46f38aa92ff8a236b5f6 : SUCCESS in 7m 20s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/95a355d0878e4f1e8270f6fc5f639452 : SUCCESS in 8m 19s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/2c3b31e560884ea3a480fd8467536736 : SUCCESS in 1h 15m 14s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/45c1ccd2ba6c4a779f1a78deaaba154d : SUCCESS in 41m 24s\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/bd8ddaa78b354b378302040c9ac60428 : SUCCESS in 1h 00m 19s (non-voting)\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/bb9834f9a54c4e4b903546c83c8c2f0e : SUCCESS in 1h 07m 44s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/e9c1b6f010ae46018cd96604385c1bd3 : SUCCESS in 55m 20s\n- kolla-ansible-ubuntu-binary-upgrade https://zuul.opendev.org/t/openstack/build/36e3e7aea02e41668216096572f26e64 : SUCCESS in 1h 08m 23s\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/5d844d264d674f6faaf8718085825a19 : FAILURE in 26m 01s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/29e800dce03542f89fc33fb4279a4e38 : FAILURE in 1h 17m 40s (non-voting)","accounts_in_message":[],"_revision_number":116},{"id":"6433e16f15e135243de981552503a8305f5dfd89","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-07-26 15:55:28.000000000","message":"Uploaded patch set 117: New patch set was added with same tree, parent, and commit message as Patch Set 116.","accounts_in_message":[],"_revision_number":117},{"id":"62d76d07795b338faf2682c57f4ead9bf016775a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-26 17:16:51.000000000","message":"Patch Set 117: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b46b0a89784d4ce297fe667cef4f0f9b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/be28fbd17b8e434f8d13b4f03b982270 : SUCCESS in 6m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec2421d1c48447959a057d7ed39ad599 : SUCCESS in 5m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9b77f2f3cf194e6798960f5e6ff7209b : SUCCESS in 6m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e0e2aa2fc0c2475e9ad0bf129e6123f0 : SUCCESS in 6m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dead531bae38444da787ab18e1658925 : SUCCESS in 7m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/661326edac864b789da858573bea1e7a : SUCCESS in 6m 37s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/ad39d81410024a74ac172ecfdc0f92e1 : SUCCESS in 9m 14s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/db6d551ad56e429185f410e9a75ee658 : SUCCESS in 1h 03m 48s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/13f756e0583a4d97b032de370350456c : SUCCESS in 59m 08s\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/2da85bcb33b840969243896abad20b13 : SUCCESS in 1h 00m 58s (non-voting)\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/40746a545a4b46bf894d9e290a2f312c : SUCCESS in 1h 14m 06s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/6a08f71ebe1f43dca7961f57b965af5d : SUCCESS in 1h 01m 14s\n- kolla-ansible-ubuntu-binary-upgrade https://zuul.opendev.org/t/openstack/build/c5012b9400c84a70ba67e2511ba9e2b2 : SUCCESS in 1h 06m 44s\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/45b80e55d7294148b2d5ca634384c031 : FAILURE in 21m 19s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/5b4dac962c27455d9cf27449f2aa447b : FAILURE in 1h 19m 21s (non-voting)","accounts_in_message":[],"_revision_number":117},{"id":"fbb3c39f4f10066a40aa755849cf82e81b12203a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-26 20:04:19.000000000","message":"Patch Set 117:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/92b3c8d322294bfb88662307e3982c20\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f2012c87c4474cbe911ecb5cb5c964fc : TIMED_OUT in 3h 04m 25s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/8b2529db4e9d40f9971f7ca68ca87088 : RETRY_LIMIT in 4m 23s","accounts_in_message":[],"_revision_number":117},{"id":"07eeae167864ef9d767cd75482f3d1fefa90979a","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-07-27 12:45:02.000000000","message":"Patch Set 117: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":117},{"id":"603dc321d17a5a404d5063ef71f550af5a6ba3bc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-07-28 14:01:11.000000000","message":"Uploaded patch set 118: Patch Set 117 was rebased.","accounts_in_message":[],"_revision_number":118},{"id":"7d6c92583212236b8899c3e9e5c5ecd07a96d083","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-07-28 15:09:15.000000000","message":"Uploaded patch set 119: Patch Set 118 was rebased.","accounts_in_message":[],"_revision_number":119},{"id":"e1ed58e6f9f4c336c82204440b41eef5ccb25306","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-28 16:53:50.000000000","message":"Patch Set 119: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/eb1b53ca9f1f45cbb5a68ab25a046a54\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f8518aa15f7f4b6b95280e8b8ecedcd9 : SUCCESS in 7m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f907fbc01de45069309803f3b647a88 : SUCCESS in 5m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2d53e54d12d94966a2bf847403555859 : SUCCESS in 5m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/29b0bb3d56364cceb9574f8a4d55ad12 : SUCCESS in 6m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f8e9f462923b43738d8f2780c0b80ad2 : SUCCESS in 6m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd98bc440323428d9fb0ef823ec27a55 : SUCCESS in 7m 30s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/c6cee3ff17044eacb728f11247e8bc3e : SUCCESS in 7m 58s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/d1f0d8267d4949808b33293773333f84 : SUCCESS in 1h 13m 42s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/b6c5906c2430497699c9d5b8b101367f : FAILURE in 41m 58s\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/0faad09f772e41f5b0026aede1b7c90d : FAILURE in 42m 07s (non-voting)\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/ca5508f3c02d42ba9fa6cf8ad5480946 : SUCCESS in 1h 12m 53s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/c02846b5add04e23b9e023ccc5ffdcf1 : FAILURE in 1h 12m 18s\n- kolla-ansible-ubuntu-binary-upgrade https://zuul.opendev.org/t/openstack/build/4d8ba08dfb184fa6b6e7c2519a2829b2 : FAILURE in 1h 14m 23s\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/cd6f3bc56feb4388bce11f157b060851 : FAILURE in 20m 55s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/eba98831b0124608bf2d7bd6b0e8ab06 : FAILURE in 1h 41m 39s (non-voting)","accounts_in_message":[],"_revision_number":119},{"id":"4a4c79d1ce8d22e4a06205e474507f6fc0800990","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-07-28 17:27:38.000000000","message":"Uploaded patch set 120: Patch Set 119 was rebased.","accounts_in_message":[],"_revision_number":120},{"id":"b45f7cf570550ec79b0007935d5f3585f70e6505","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-07-28 17:43:28.000000000","message":"Uploaded patch set 121: New patch set was added with same tree, parent, and commit message as Patch Set 120.","accounts_in_message":[],"_revision_number":121},{"id":"e464ab55db2d2de12b6a44d58320fe9bef5f493e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-07-28 18:04:23.000000000","message":"Uploaded patch set 122: Patch Set 121 was rebased.","accounts_in_message":[],"_revision_number":122},{"id":"c4ebd4fe2d4060eb53fb0788d1aa151140cc9884","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-07-28 18:08:57.000000000","message":"Uploaded patch set 123: Patch Set 122 was rebased.","accounts_in_message":[],"_revision_number":123},{"id":"2b3d0144979071f6d579bedf98bb3de76f05ecf1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-07-28 18:14:30.000000000","message":"Uploaded patch set 124: New patch set was added with same tree, parent, and commit message as Patch Set 123.","accounts_in_message":[],"_revision_number":124},{"id":"dba14ca41795b877d7965411919fc60ae1a46795","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-28 19:36:39.000000000","message":"Patch Set 124: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5c79dc369f2842ea94410af6d7dc9a1a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bf8c29ad74104612bbdd1ba8e5ccf4cc : SUCCESS in 6m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d9d5c626c1a450d92e6750a04ad5461 : SUCCESS in 6m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/994b39471ab14d7c891bd48ef765b218 : SUCCESS in 6m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2680ea9a08ca45ffbc3025a3b108176d : SUCCESS in 6m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4d5ade1f1eda422f844143206477a55f : SUCCESS in 5m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a5ce94871fa49d68cb514f3d65939e8 : SUCCESS in 7m 11s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/e737a8eea3d94ec99597a6031b2334f9 : SUCCESS in 8m 29s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/7496c99adfcb4d0e8f923982e05a6fc8 : SUCCESS in 54m 11s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7d5fbdd40a524de9a87aa0fd3b0cefb7 : SUCCESS in 49m 10s\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/bc61a7e342ef49b9bd0e96e6b700d878 : SUCCESS in 46m 59s (non-voting)\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/fcb0614fd2c64ea786c76b4ae8b182a3 : SUCCESS in 55m 18s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/7852c8294aa14067b36a6578861670ca : SUCCESS in 50m 59s\n- kolla-ansible-ubuntu-binary-upgrade https://zuul.opendev.org/t/openstack/build/b33aba4bc13f421280ed1c7cce6bb3f1 : SUCCESS in 54m 48s\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/b3888e28f0bf47f6bde9d9115fc3649a : FAILURE in 13m 40s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/e72814fc0bdd49d0889fe901a44fa38b : SUCCESS in 1h 19m 55s (non-voting)","accounts_in_message":[],"_revision_number":124},{"id":"f65690416ee90020d38410357feb2ddde7a43f64","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-28 19:48:13.000000000","message":"Patch Set 124:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7b80ebc312724d419ed39584038d8f42\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/eee54696cdf248db828844358adb8ed4 : SUCCESS in 1h 03m 46s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/33d633640f97482fa01c514e6928b6a7 : SUCCESS in 1h 31m 14s","accounts_in_message":[],"_revision_number":124},{"id":"b8083793dbadfb5adbda7952f475db1472b62513","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-07-28 20:15:13.000000000","message":"Uploaded patch set 125: New patch set was added with same tree, parent, and commit message as Patch Set 124.","accounts_in_message":[],"_revision_number":125},{"id":"b3c0dd72d04dcc7363c1b9c5383afcc29a757eb1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-28 21:49:22.000000000","message":"Patch Set 125: Verified+1\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/67347ce78c7247aea77c9c6efec109c5\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f69f1ee0859648a3adb125b2a04425f1 : SUCCESS in 1h 27m 41s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/bb47171a8efd49c88adbb082ce2a4143 : SUCCESS in 1h 31m 40s","accounts_in_message":[],"_revision_number":125},{"id":"193e9d5f11d83c636daa7829c6678d22a6860cb4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-28 21:49:39.000000000","message":"Patch Set 125:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/07f4fb57755942238f5cb69b1b701f34\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/742996ec4c214da9bf98ff9b0429a150 : SUCCESS in 8m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6888c656242247f7994d5a902e62c56f : SUCCESS in 6m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/96b18e06bf464715885400c484867686 : SUCCESS in 8m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/274a125d542e40549110889382a8d145 : SUCCESS in 7m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c1f4c492ca524db7bcde3ce067802450 : SUCCESS in 6m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82f1b5a7db3b48a1b81ffc7926edec3e : SUCCESS in 10m 13s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/5145a28467ba4f3088b7e950933d8cfe : SUCCESS in 6m 55s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/b6eefbbc68e9453290dea3ebce1cc889 : SUCCESS in 55m 32s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/ca580fc1e5c14e9998d24d8e992a5c0f : SUCCESS in 48m 00s\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/339dab3b258644e590ce7f2991615d9a : SUCCESS in 46m 06s (non-voting)\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/6232d1830352419a818bcbdf11ba95f6 : SUCCESS in 1h 04m 39s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d275e722b7914953aff28fd09eac9a3f : SUCCESS in 49m 25s\n- kolla-ansible-ubuntu-binary-upgrade https://zuul.opendev.org/t/openstack/build/9cf4718c8c5943a9b0545f9b480b954f : SUCCESS in 50m 35s\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/d739c052510347afa0c3aaeec8547a9e : FAILURE in 15m 29s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/9ae7ab4071da4d929563c577e96e7c06 : SUCCESS in 1h 24m 45s (non-voting)","accounts_in_message":[],"_revision_number":125},{"id":"dfa46df138f7c05f6b99d024e000861338e3a2dc","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2022-07-29 07:09:37.000000000","message":"Patch Set 125: Code-Review+1","accounts_in_message":[],"_revision_number":125},{"id":"61e600442003c6b602265121e0eda7180b6c06c2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-07-29 13:07:34.000000000","message":"Uploaded patch set 126: Patch Set 125 was rebased.","accounts_in_message":[],"_revision_number":126},{"id":"5f0227a6d137d3e2584c9d6aa35b344debc7c03e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-29 14:32:05.000000000","message":"Patch Set 126:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b9bea084ea83459ba447808f38291ca7\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f64621813f2d4e898f718e72a9ab2bfd : SUCCESS in 1h 21m 52s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/1a92d67c40494402982debecbb783d29 : FAILURE in 1h 03m 18s","accounts_in_message":[],"_revision_number":126},{"id":"839bf75dd58b0d005fc35a1732ec32dfb964c50b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-29 14:45:51.000000000","message":"Patch Set 126: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7aae6aeff1a941c5b9e82871f7682345\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6f863c58ab574982bb029085f483a98f : SUCCESS in 6m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36e33edab12c467d8fc47cc40bc68b2d : SUCCESS in 6m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/46b5ccad9bd644979bda5e7d86fc4c4c : SUCCESS in 7m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aaa2dbdd0b1a494fbfda29792012a98d : SUCCESS in 7m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2aaf8a96e8aa4a5b88233dfc1bcc244f : SUCCESS in 6m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b66da67da604259a331e62447d509de : SUCCESS in 7m 33s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/5eab760777c047fdb97424a83195f50d : SUCCESS in 8m 26s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/c6308d9583ab439b8b4fb32f9751a5e9 : SUCCESS in 1h 04m 19s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e6b553be05df48d981df1436d69ecd86 : SUCCESS in 46m 17s\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/ff7c5d0fa0f0490f9e6803f8677b7e5c : SUCCESS in 49m 32s (non-voting)\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/d3ea48eb1dc94686836c7e57ceb14be5 : SUCCESS in 1h 02m 39s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/3953e8882b024b0eaceff1446ea466f0 : SUCCESS in 1h 00m 45s\n- kolla-ansible-ubuntu-binary-upgrade https://zuul.opendev.org/t/openstack/build/a1858cb3faa84cc0ab013bc95520e87b : SUCCESS in 58m 54s\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/0f92b466c6314720bf5f7a52e0be9c52 : FAILURE in 14m 05s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/4aee4324f5c0441496f550c8c6463c8d : SUCCESS in 1h 26m 24s (non-voting)","accounts_in_message":[],"_revision_number":126},{"id":"326d4b2d8dec85fa7c72ac1900ea87106c89fad7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-07-30 14:24:37.000000000","message":"Uploaded patch set 127: Patch Set 126 was rebased.","accounts_in_message":[],"_revision_number":127},{"id":"e5de618cb4ccf9238c4d161defe2f1492a7bdd11","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-30 15:51:37.000000000","message":"Patch Set 127:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/43e05450e1f34452b417ecbbf075ab9b\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f636108dd3e94b62945947d6b130207a : SUCCESS in 1h 06m 56s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/c7e8212809724ac1b39a52def977b2ff : FAILURE in 1h 16m 36s","accounts_in_message":[],"_revision_number":127},{"id":"f6105d27cde6e5c49c8064d4549a5819661e7857","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-30 15:54:33.000000000","message":"Patch Set 127: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/09eafefab5444696912124acc7b3af90\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/77246f73326f4d45973dcf7a066cd470 : SUCCESS in 5m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ed07a6354814deaa3cd2ac5e58e35ea : SUCCESS in 5m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/35c66d0157804039a814d1c71a0f7ede : SUCCESS in 5m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d91cbbd522b54eab97bebe7680da2aac : SUCCESS in 5m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8bc3fd72e1c24b34b4ba5ffd987769a2 : SUCCESS in 6m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2fa31bdd209e477f9170809ddb7fc124 : SUCCESS in 4m 49s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/bbb1cd7803f64a008c6f9588eff6f565 : SUCCESS in 7m 19s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/af5e78503b5e42cda1b84d79f5309fb8 : SUCCESS in 54m 03s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e77674f6da5a456b829674424b947e9c : SUCCESS in 44m 02s\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/49463ad86c01411b8ced913b4a6f75e6 : SUCCESS in 49m 15s (non-voting)\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/b4155f8260a142e298dac216dad98a6e : SUCCESS in 54m 25s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/4e646c2394b7409c8d11f78908c2583f : SUCCESS in 44m 34s\n- kolla-ansible-ubuntu-binary-upgrade https://zuul.opendev.org/t/openstack/build/ec34844922614bac9aeb99cc66e9e54c : SUCCESS in 51m 36s\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/0f31958dd0754269af72eef6dfebffd1 : FAILURE in 13m 28s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/040e206989ee4879a06a22436a5ca325 : SUCCESS in 1h 28m 15s (non-voting)","accounts_in_message":[],"_revision_number":127},{"id":"d05dcda208ef9aacd1d2df2f74a492aaa591cff0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-08-02 11:34:47.000000000","message":"Uploaded patch set 128: Patch Set 127 was rebased.","accounts_in_message":[],"_revision_number":128},{"id":"55a72f83d724c2f5780d206b1518e8b63f73b60c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-02 13:03:29.000000000","message":"Patch Set 128: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/80b2c6f7ef5b488e9d4f17cda8a9931c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ca9c3bea0b86488aa6cc56c1d4f8ce45 : SUCCESS in 6m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4663372d487245668f3b39e7a97f534a : SUCCESS in 7m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d5332450de69495299e6dcb30fe8ae7d : SUCCESS in 5m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/efd2ad18c64a44e0aeff771d656d88ce : SUCCESS in 6m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/39ac89cc4a2b4283a82d78cecbe8bf71 : SUCCESS in 6m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e59cf4a027f84ea1af55348dd936dbfb : SUCCESS in 8m 03s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/75adf1d39fd5438caafe14fc20cfdeea : SUCCESS in 7m 41s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/e9730f2e07f84c99baefc4eb8b866075 : SUCCESS in 43m 54s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a9da11bf03d840ea88626d27f5cae020 : SUCCESS in 37m 34s\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/af3d967a6d5a4bb0bc5cf72d852750e6 : SUCCESS in 48m 14s (non-voting)\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/9e9861ad70f848da98959cf914811966 : SUCCESS in 55m 55s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/d0e0f7c8def54767867189abd88c0d31 : SUCCESS in 1h 06m 11s\n- kolla-ansible-ubuntu-binary-upgrade https://zuul.opendev.org/t/openstack/build/b226579e44f44ae58a2f76468c131769 : SUCCESS in 1h 01m 35s\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/d766051c672948818fa8f757248a86a2 : FAILURE in 12m 35s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/de33f9e1056e46789701c5435ba6b881 : SUCCESS in 1h 08m 28s (non-voting)","accounts_in_message":[],"_revision_number":128},{"id":"75f061d62d4e034647ce2d25548dafe31e8126b6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-02 13:09:16.000000000","message":"Patch Set 128:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/21de7259d16f4ddfb03c9f01b76c8a21\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/8f96ddbf52f14132bfbbc7c3ee8e0cf1 : SUCCESS in 1h 31m 24s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/d1366a029cb14492af4947eff82e08d5 : FAILURE in 1h 19m 18s","accounts_in_message":[],"_revision_number":128},{"id":"18d421b8f47cac60093c614d769aeaae52df85b3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-08-02 15:49:06.000000000","message":"Uploaded patch set 129: Patch Set 128 was rebased.","accounts_in_message":[],"_revision_number":129},{"id":"2612b0410c5d9e77d67ff5aa1677a4f0fb13c208","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-02 17:20:04.000000000","message":"Patch Set 129: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5c4d3c102da24352b74dfaf0c74f6514\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9929d19de0bf494e918cb577fc33e447 : SUCCESS in 7m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/65c6574ef2144e0a9796846cb650f63f : SUCCESS in 8m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/91212eee8a66442ab5c84a69bae80fe4 : SUCCESS in 8m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/77a43592ec324176b9b2d3aa853de92b : SUCCESS in 7m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7062c0629e174caa9330cd8ec7227d75 : SUCCESS in 8m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d8731268c764d0c8b236d1f3a5c74fb : SUCCESS in 5m 01s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/74e1250047024cf781900e0a51a949ad : SUCCESS in 10m 12s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/43395d6155ab4b9ba3bcc4a2e77f63b1 : SUCCESS in 47m 02s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/4853db678b124789a9dc9bce39cc6a2f : SUCCESS in 52m 14s\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/df3083b737434ea69d8fa08f99512011 : SUCCESS in 55m 28s (non-voting)\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/b4810247e109419c83f3232f4c0a2a51 : SUCCESS in 1h 08m 47s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/3107fe70f5104bc5a61c5258609c216d : SUCCESS in 1h 03m 34s\n- kolla-ansible-ubuntu-binary-upgrade https://zuul.opendev.org/t/openstack/build/3440e896f8ce45a3a6ecd1f6285c21c2 : SUCCESS in 57m 17s\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/1678fc5a64f9412d8a71a425800aa2a6 : FAILURE in 15m 18s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/6775a549e7fc4b06b80d218b156ce6a4 : SUCCESS in 1h 24m 50s (non-voting)","accounts_in_message":[],"_revision_number":129},{"id":"65851ec8fff0fd129697983dc6a3226f2527c27d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-02 17:27:29.000000000","message":"Patch Set 129:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9fa4f99d184544a78eb08887091b8245\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/a3ae3c5c85ac4de0b6875030c5c7a27b : SUCCESS in 1h 12m 16s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/c43f6f4aff344dbf971ce60ba800e9e5 : FAILURE in 1h 33m 39s","accounts_in_message":[],"_revision_number":129},{"id":"beebbe96330fb1111fd8759e55642bf15f5e8b35","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-08-09 10:17:34.000000000","message":"Uploaded patch set 130: Patch Set 129 was rebased.","accounts_in_message":[],"_revision_number":130},{"id":"5420c97dbe2bdc4159adffb0ca0965267a02734c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-09 12:00:37.000000000","message":"Patch Set 130: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f5200c223bbd49419cd646f9becae82c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/54a0559cb40742b0bea3fb5abf6df13e : SUCCESS in 6m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f45baaa15656420aaf2353692192e461 : SUCCESS in 5m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ee8712e3269348c8b8094653f5586465 : SUCCESS in 6m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ce39a90bfd1c4e2db40f716d53316e61 : SUCCESS in 7m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7252582406174258b6fb146156b4ef2a : SUCCESS in 7m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/929b08d924fe4484b9efae46850cc7c8 : SUCCESS in 5m 09s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/c4ea6be53c414017bb8802105b617cdb : SUCCESS in 8m 55s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/624dd0d0f79947fc9ed8450c1793a245 : SUCCESS in 53m 01s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/e5e2b135c9be424ebfb250fc95bc6f4d : SUCCESS in 43m 36s\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/18d3f9df0ba14b60818098f9b0d6a81d : SUCCESS in 58m 32s (non-voting)\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/e7a6e8061cc74fe89b6f7d6a74af0896 : SUCCESS in 1h 05m 42s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/a43c2c8f4fac4eea9ab9548a3198dc13 : SUCCESS in 58m 16s\n- kolla-ansible-ubuntu-binary-upgrade https://zuul.opendev.org/t/openstack/build/bb58a1b07d6f4caba92a5691aed02456 : SUCCESS in 1h 02m 12s\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/d58ce5dec644497e959d8cf8c89551ec : SUCCESS in 58m 49s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/6d55d2b7cb3645928178484c8894693d : SUCCESS in 1h 41m 08s (non-voting)","accounts_in_message":[],"_revision_number":130},{"id":"8f9092112ce0007f38c8e2a4d3ed9a93be3387e3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-09 12:13:38.000000000","message":"Patch Set 130:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c7f17cecb74e4d4aa5cc022bf349f967\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/1f7ba28cec14426fa9e3388c8d3b2798 : SUCCESS in 1h 13m 24s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/650c9f6a24fc48668363f15ae96ea73b : FAILURE in 1h 47m 40s","accounts_in_message":[],"_revision_number":130},{"id":"a4ba6bd8fb0754a7eb8f652e6272fdd72c573ee2","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-08-09 13:12:07.000000000","message":"Patch Set 130: Review-Priority+1","accounts_in_message":[],"_revision_number":130},{"id":"86df9b4fe369cbd83354d69701192d66fc93dd00","author":{"_account_id":30434,"name":"Jingwei.Zhang","email":"zhang.jingwei@algoblu.com","username":"jwzhang"},"date":"2022-08-09 13:14:38.000000000","message":"Patch Set 130: Code-Review+1","accounts_in_message":[],"_revision_number":130},{"id":"e047375e58c9b74ec1844651aa9d4c8d82f3789e","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-08-09 13:17:05.000000000","message":"Patch Set 130: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":130},{"id":"9d85f7d5a4f2b6cb9cfcde71443960bb5693daae","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2022-08-09 15:42:46.000000000","message":"Patch Set 130:\n\n(1 comment)","accounts_in_message":[],"_revision_number":130},{"id":"2f6e84577c64b6b1f6bdab5d35704012e48adbbb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-09 17:00:12.000000000","message":"Patch Set 130:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a40df5c595343178bde3d9c79b51bc6\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d14dc5201a7e4f519af61e571ac281b5 : SUCCESS in 5m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0c3178057944d08baecbfb8cb1d6349 : SUCCESS in 5m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/32d66589891a4af5beb0b078890f0345 : SUCCESS in 6m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/72f1ee5818ae4fcb9b811ac35c2e39c2 : SUCCESS in 5m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a69173f5dee74c90a19469da9704b656 : SUCCESS in 6m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ab8ee494aa7460dbc2198c82a7dc5f3 : SUCCESS in 5m 07s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/4382c16adb964ee3be8ebd052e51381b : SUCCESS in 8m 03s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/c907903b665842c4affbde9d445a7add : SUCCESS in 48m 00s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/3710c0d4b1de4f459bbd4eb0b5ad62ac : SUCCESS in 40m 16s\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/7f41f145c89d4a209431cf02ed19010b : SUCCESS in 46m 18s (non-voting)\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/782247a491b044eea681f8bccb39d811 : SUCCESS in 56m 06s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/3ebd18872b944108be785c9d01d35b75 : SUCCESS in 52m 34s\n- kolla-ansible-ubuntu-binary-upgrade https://zuul.opendev.org/t/openstack/build/adf8e6a508404b479a0e2068705d2bb2 : SUCCESS in 52m 07s\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/9defc567f4184a5f97c035283f79ef7e : SUCCESS in 59m 49s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/f0368564d04c4cedbbced57ad90f139f : SUCCESS in 1h 15m 29s (non-voting)","accounts_in_message":[],"_revision_number":130},{"id":"18ee6504c7356ca0c8529f277c3060aedbd677d9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-09 17:07:49.000000000","message":"Patch Set 130:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a417012e4624f6b8dccc1707d79b383\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/ed4c3bbcd88248d6bec2a26e719ce22a : SUCCESS in 1h 22m 47s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/a48bdbc5ca584b798247a76171b519f5 : FAILURE in 1h 15m 36s","accounts_in_message":[],"_revision_number":130},{"id":"76068b31240d1e33de0517b02c6c59bca4a8da62","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-10 11:09:38.000000000","message":"Patch Set 130: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":130},{"id":"da29afcc619b4c0f4622117fa8ebf98d2a4bba96","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-10 12:22:51.000000000","message":"Patch Set 130: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/63ac93225eec411d9b7ce29705ed69f8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d358d43a90d441c98c5739a11c6604d : SUCCESS in 5m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d184457604614a28958986c0c57ed8bd : SUCCESS in 6m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9ff811580e8c4cd99bbafd224bc7c6f6 : SUCCESS in 6m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b73dd5573ff4b2c8fd538c5bb301aac : SUCCESS in 6m 47s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/3c5ba9536c1b47649ad03332cd6b74f2 : SUCCESS in 7m 52s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/1f4f2172747a40ce9f726aaff7e6c947 : SUCCESS in 52m 28s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/67bddc51e36446d0bcdf0ebfcf318f61 : SUCCESS in 52m 23s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/0f49e2ac4952474bb0e186c8ac9cc710 : SUCCESS in 40m 57s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/bd0f3eb8f6f44a608e09ac8697d4437c : SUCCESS in 49m 53s","accounts_in_message":[],"_revision_number":130},{"id":"7f39765d7acbb1d40496f2da8bf68b6b78123937","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-10 12:22:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":130},{"id":"ff403f4ad9b9028fe32a612ec8e6f11a312ca541","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-10 12:24:55.000000000","message":"Patch Set 130:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/74502b05f7824ec2aa61ae542cf4d7af\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e5eca4006bc4cc79eda8086b9ebbe21 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":130}],"current_revision_number":130,"current_revision":"1b3fcfed64c9dffb5693d2930ac1f0a696b4aa34","revisions":{"4e13886149803da3e1612abb71059650b16680f4":{"kind":"REWORK","_number":1,"created":"2021-01-13 13:56:00.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/1"}}},"commit":{"parents":[{"commit":"9e32aad621df27440650ba224791d7ce90a12a35","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/9e32aad621df27440650ba224791d7ce90a12a35"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:44.000000000","tz":0},"subject":"[CI] Enable proxysql test","message":"[CI] Enable proxysql test\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/4e13886149803da3e1612abb71059650b16680f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/4e13886149803da3e1612abb71059650b16680f4"}]},"branch":"refs/heads/master"},"2de8ffecafcba24c2af7b0ae221e16eab4830de1":{"kind":"NO_CHANGE","_number":2,"created":"2021-01-13 15:01:34.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/2"}}},"commit":{"parents":[{"commit":"a683f38a5d8e57e9777146c71731368f69498069","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/a683f38a5d8e57e9777146c71731368f69498069"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 14:51:11.000000000","tz":60},"subject":"[CI] Enable proxysql test","message":"[CI] Enable proxysql test\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/2de8ffecafcba24c2af7b0ae221e16eab4830de1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/2de8ffecafcba24c2af7b0ae221e16eab4830de1"}]},"branch":"refs/heads/master"},"d93864f3a37cbeaec503699a5b91b17d4c98c4c5":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-01-13 15:12:01.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/3"}}},"commit":{"parents":[{"commit":"a90c340031c9810cc1b42a00702f0b5fe61bb20e","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/a90c340031c9810cc1b42a00702f0b5fe61bb20e"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 15:11:50.000000000","tz":0},"subject":"[CI] Enable proxysql test","message":"[CI] Enable proxysql test\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/d93864f3a37cbeaec503699a5b91b17d4c98c4c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/d93864f3a37cbeaec503699a5b91b17d4c98c4c5"}]},"branch":"refs/heads/master"},"099b42da5ba2776a17726e3fe27fb93a04a336ad":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2021-01-13 18:17:22.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/4"}}},"commit":{"parents":[{"commit":"a90c340031c9810cc1b42a00702f0b5fe61bb20e","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/a90c340031c9810cc1b42a00702f0b5fe61bb20e"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 18:15:12.000000000","tz":0},"subject":"[CI] Enable proxysql test","message":"[CI] Enable proxysql test\n\nDepends-On: https://review.opendev.org/c/openstack/kolla/+/769385\nDepends-On: https://review.opendev.org/c/openstack/kolla/+/765781\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/099b42da5ba2776a17726e3fe27fb93a04a336ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/099b42da5ba2776a17726e3fe27fb93a04a336ad"}]},"branch":"refs/heads/master"},"b29cb762017438d87c4313031266258e4eddba29":{"kind":"REWORK","_number":5,"created":"2021-01-13 22:55:08.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/5"}}},"commit":{"parents":[{"commit":"541e55cddb1158c4fc1dd72a350bde3e9a7d161e","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/541e55cddb1158c4fc1dd72a350bde3e9a7d161e"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 22:30:35.000000000","tz":0},"subject":"[CI] Enable proxysql test","message":"[CI] Enable proxysql test\n\nDepends-On: https://review.opendev.org/c/openstack/kolla/+/769385\nDepends-On: https://review.opendev.org/c/openstack/kolla/+/765781\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/b29cb762017438d87c4313031266258e4eddba29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/b29cb762017438d87c4313031266258e4eddba29"}]},"branch":"refs/heads/master"},"39397a2f143b0edd9b41a8808faccf37c28360b6":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2021-01-14 00:30:33.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/6"}}},"commit":{"parents":[{"commit":"2882d28dcda8cc0b16d059dfa6d7288f45911a79","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/2882d28dcda8cc0b16d059dfa6d7288f45911a79"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-14 00:30:19.000000000","tz":0},"subject":"[CI] Enable proxysql test","message":"[CI] Enable proxysql test\n\nDepends-On: https://review.opendev.org/c/openstack/kolla/+/769385\nDepends-On: https://review.opendev.org/c/openstack/kolla/+/765781\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/39397a2f143b0edd9b41a8808faccf37c28360b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/39397a2f143b0edd9b41a8808faccf37c28360b6"}]},"branch":"refs/heads/master"},"5b9a412563c94742b34aa5327b5020ad798f564d":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2021-01-14 10:43:39.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/7"}}},"commit":{"parents":[{"commit":"86f2384c009fb5886a03f71c891f59f4f2e86e96","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/86f2384c009fb5886a03f71c891f59f4f2e86e96"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-14 09:11:37.000000000","tz":0},"subject":"[CI] Enable proxysql test","message":"[CI] Enable proxysql test\n\nDepends-On: https://review.opendev.org/c/openstack/kolla/+/769385\nDepends-On: https://review.opendev.org/c/openstack/kolla/+/765781\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/5b9a412563c94742b34aa5327b5020ad798f564d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/5b9a412563c94742b34aa5327b5020ad798f564d"}]},"branch":"refs/heads/master"},"dde83562424d7179efabc7a7ade6176d1c32b083":{"kind":"NO_CHANGE","_number":8,"created":"2021-01-14 10:46:10.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/8"}}},"commit":{"parents":[{"commit":"a769b7844d2fc3e62393fa6eff56777a9e17b66a","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/a769b7844d2fc3e62393fa6eff56777a9e17b66a"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-14 10:45:35.000000000","tz":0},"subject":"[CI] Enable proxysql test","message":"[CI] Enable proxysql test\n\nDepends-On: https://review.opendev.org/c/openstack/kolla/+/769385\nDepends-On: https://review.opendev.org/c/openstack/kolla/+/765781\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/dde83562424d7179efabc7a7ade6176d1c32b083"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/dde83562424d7179efabc7a7ade6176d1c32b083"}]},"branch":"refs/heads/master"},"a22c7c7dfc7a36ea7578105627ae15ab1c9645d9":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2021-01-14 23:20:11.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/9"}}},"commit":{"parents":[{"commit":"6a85ef2f6fd616cc3942a11fde6ca4c7641a7789","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/6a85ef2f6fd616cc3942a11fde6ca4c7641a7789"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-14 20:22:50.000000000","tz":0},"subject":"[CI] Enable proxysql test","message":"[CI] Enable proxysql test\n\nDepends-On: https://review.opendev.org/c/openstack/kolla/+/769385\nDepends-On: https://review.opendev.org/c/openstack/kolla/+/765781\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/a22c7c7dfc7a36ea7578105627ae15ab1c9645d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/a22c7c7dfc7a36ea7578105627ae15ab1c9645d9"}]},"branch":"refs/heads/master"},"b47935d7d6df323b74c2a16b5dc75561adabeede":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2021-01-18 11:53:40.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/10"}}},"commit":{"parents":[{"commit":"5c768f75050ff9d89ebba0f2774a1d06ffc435c2","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/5c768f75050ff9d89ebba0f2774a1d06ffc435c2"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-18 11:14:54.000000000","tz":0},"subject":"[CI] Enable proxysql test","message":"[CI] Enable proxysql test\n\nDepends-On: https://review.opendev.org/c/openstack/kolla/+/769385\nDepends-On: https://review.opendev.org/c/openstack/kolla/+/765781\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/b47935d7d6df323b74c2a16b5dc75561adabeede"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/b47935d7d6df323b74c2a16b5dc75561adabeede"}]},"branch":"refs/heads/master"},"01f329887b6d742822c7df42feb49df3abffc270":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2021-01-20 00:33:51.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/11"}}},"commit":{"parents":[{"commit":"04168e54750eef3e4b519f188bdd04f551c5ba37","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/04168e54750eef3e4b519f188bdd04f551c5ba37"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-20 00:32:52.000000000","tz":0},"subject":"[CI] Enable proxysql test","message":"[CI] Enable proxysql test\n\nDepends-On: https://review.opendev.org/c/openstack/kolla/+/769385\nDepends-On: https://review.opendev.org/c/openstack/kolla/+/765781\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/01f329887b6d742822c7df42feb49df3abffc270"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/01f329887b6d742822c7df42feb49df3abffc270"}]},"branch":"refs/heads/master"},"ab6762e1c025286ad8840c40fd08dbddd8d906d1":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2021-01-20 09:09:43.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/12"}}},"commit":{"parents":[{"commit":"f0d9c373a5b12587356f37681c212ccb62e47884","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/f0d9c373a5b12587356f37681c212ccb62e47884"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-20 09:08:42.000000000","tz":0},"subject":"[CI] Enable proxysql test","message":"[CI] Enable proxysql test\n\nDepends-On: https://review.opendev.org/c/openstack/kolla/+/769385\nDepends-On: https://review.opendev.org/c/openstack/kolla/+/765781\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/ab6762e1c025286ad8840c40fd08dbddd8d906d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/ab6762e1c025286ad8840c40fd08dbddd8d906d1"}]},"branch":"refs/heads/master"},"3484672f1d4b9b9ea9542853d0a57d454d6322c7":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2021-01-26 20:07:45.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/21/770621/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/13"}}},"commit":{"parents":[{"commit":"066faf2bc43e7cf8fba834b6228c19eebcdfb677","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/066faf2bc43e7cf8fba834b6228c19eebcdfb677"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-01-26 20:07:45.000000000","tz":0},"subject":"[CI] Enable proxysql test","message":"[CI] Enable proxysql test\n\nDepends-On: https://review.opendev.org/c/openstack/kolla/+/769385\nDepends-On: https://review.opendev.org/c/openstack/kolla/+/765781\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/3484672f1d4b9b9ea9542853d0a57d454d6322c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/3484672f1d4b9b9ea9542853d0a57d454d6322c7"}]},"branch":"refs/heads/master","description":"Rebase"},"7b059d18a7fd19cd5a89a9e8e4d5e539ebe9b263":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2021-01-27 07:25:35.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/21/770621/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/14"}}},"commit":{"parents":[{"commit":"936ca048ed71c112a34d84c6dcaa56a54e4462d8","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/936ca048ed71c112a34d84c6dcaa56a54e4462d8"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-01-27 07:25:35.000000000","tz":0},"subject":"[CI] Enable proxysql test","message":"[CI] Enable proxysql test\n\nDepends-On: https://review.opendev.org/c/openstack/kolla/+/769385\nDepends-On: https://review.opendev.org/c/openstack/kolla/+/765781\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/7b059d18a7fd19cd5a89a9e8e4d5e539ebe9b263"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/7b059d18a7fd19cd5a89a9e8e4d5e539ebe9b263"}]},"branch":"refs/heads/master","description":"Rebase"},"50b6eb277224408810b9c0d8e37370a385436091":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2021-01-27 11:36:48.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/15"}}},"commit":{"parents":[{"commit":"3f1e3046352c2f51f4a1418d0395cb16c00a2d21","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/3f1e3046352c2f51f4a1418d0395cb16c00a2d21"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-27 11:36:16.000000000","tz":0},"subject":"[CI] Enable proxysql test","message":"[CI] Enable proxysql test\n\nDepends-On: https://review.opendev.org/c/openstack/kolla/+/769385\nDepends-On: https://review.opendev.org/c/openstack/kolla/+/765781\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/50b6eb277224408810b9c0d8e37370a385436091"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/50b6eb277224408810b9c0d8e37370a385436091"}]},"branch":"refs/heads/master"},"17bc646840ed61dfb8d48dc809bcedac28c1b465":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2021-02-01 12:13:25.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/16"}}},"commit":{"parents":[{"commit":"3061045516de0c0243f0fc03413d2e74a679ac58","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/3061045516de0c0243f0fc03413d2e74a679ac58"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-02-01 12:11:58.000000000","tz":0},"subject":"[CI] Enable proxysql test","message":"[CI] Enable proxysql test\n\nDepends-On: https://review.opendev.org/c/openstack/kolla/+/769385\nDepends-On: https://review.opendev.org/c/openstack/kolla/+/765781\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/17bc646840ed61dfb8d48dc809bcedac28c1b465"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/17bc646840ed61dfb8d48dc809bcedac28c1b465"}]},"branch":"refs/heads/master"},"22010aaecefa170b179dd624948a57f6872492c2":{"kind":"TRIVIAL_REBASE","_number":17,"created":"2021-02-01 17:19:28.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/17"}}},"commit":{"parents":[{"commit":"5f85cab156442becd89d0c02ea442fd3c7a76e09","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/5f85cab156442becd89d0c02ea442fd3c7a76e09"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-02-01 17:19:03.000000000","tz":0},"subject":"[CI] Enable proxysql test","message":"[CI] Enable proxysql test\n\nDepends-On: https://review.opendev.org/c/openstack/kolla/+/769385\nDepends-On: https://review.opendev.org/c/openstack/kolla/+/765781\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/22010aaecefa170b179dd624948a57f6872492c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/22010aaecefa170b179dd624948a57f6872492c2"}]},"branch":"refs/heads/master"},"b460827fb5c7d07dea4d033fc858d8891b7af888":{"kind":"TRIVIAL_REBASE","_number":18,"created":"2021-02-01 17:34:43.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/18"}}},"commit":{"parents":[{"commit":"544d6956fd254644c35f852d35cdad4829dbeaf3","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/544d6956fd254644c35f852d35cdad4829dbeaf3"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-02-01 17:34:24.000000000","tz":0},"subject":"[CI] Enable proxysql test","message":"[CI] Enable proxysql test\n\nDepends-On: https://review.opendev.org/c/openstack/kolla/+/769385\nDepends-On: https://review.opendev.org/c/openstack/kolla/+/765781\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/b460827fb5c7d07dea4d033fc858d8891b7af888"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/b460827fb5c7d07dea4d033fc858d8891b7af888"}]},"branch":"refs/heads/master"},"c3db2c0feaf12c94e93f140e59bd7ea4b0c9de6b":{"kind":"REWORK","_number":19,"created":"2021-02-02 19:00:45.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/21/770621/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/19"}}},"commit":{"parents":[{"commit":"7282c483963e1c3344f876674e589f29b2972ce9","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/7282c483963e1c3344f876674e589f29b2972ce9"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2021-02-02 19:00:29.000000000","tz":60},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/c3db2c0feaf12c94e93f140e59bd7ea4b0c9de6b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/c3db2c0feaf12c94e93f140e59bd7ea4b0c9de6b"}]},"branch":"refs/heads/master"},"f772b8437e2a7714e2508636cf5d1378736d8af1":{"kind":"TRIVIAL_REBASE","_number":20,"created":"2021-02-09 11:19:31.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/20"}}},"commit":{"parents":[{"commit":"292f9c7a82e241a1d9bec56ed2533fb5ed745c76","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/292f9c7a82e241a1d9bec56ed2533fb5ed745c76"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-02-09 11:18:58.000000000","tz":60},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/f772b8437e2a7714e2508636cf5d1378736d8af1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/f772b8437e2a7714e2508636cf5d1378736d8af1"}]},"branch":"refs/heads/master"},"d43ea6a9f56a3cec773a2d5d2201c596909d626d":{"kind":"TRIVIAL_REBASE","_number":21,"created":"2021-02-09 12:00:28.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/21"}}},"commit":{"parents":[{"commit":"671369f5730d1bb4bd3105407383380a97e7cf73","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/671369f5730d1bb4bd3105407383380a97e7cf73"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-02-09 12:00:15.000000000","tz":60},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/d43ea6a9f56a3cec773a2d5d2201c596909d626d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/d43ea6a9f56a3cec773a2d5d2201c596909d626d"}]},"branch":"refs/heads/master"},"fea0571815fae726873bfc620989fe439aeb1c86":{"kind":"TRIVIAL_REBASE","_number":22,"created":"2021-02-10 11:49:01.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/22"}}},"commit":{"parents":[{"commit":"ad8c4de45936d25617fb326aa3c868a1b0b702bf","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/ad8c4de45936d25617fb326aa3c868a1b0b702bf"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-02-10 11:48:43.000000000","tz":60},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/fea0571815fae726873bfc620989fe439aeb1c86"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/fea0571815fae726873bfc620989fe439aeb1c86"}]},"branch":"refs/heads/master"},"7d0561c49f9ad0308f1983a55b104c63c82c9ae0":{"kind":"TRIVIAL_REBASE","_number":23,"created":"2021-02-10 12:32:00.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/23"}}},"commit":{"parents":[{"commit":"062938d00b2a4e9505a1a84dcde49fbc852f4528","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/062938d00b2a4e9505a1a84dcde49fbc852f4528"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-02-10 12:31:07.000000000","tz":60},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/7d0561c49f9ad0308f1983a55b104c63c82c9ae0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/7d0561c49f9ad0308f1983a55b104c63c82c9ae0"}]},"branch":"refs/heads/master"},"c1f00f7c8d12d7987d9216e012192e68db4f46f7":{"kind":"TRIVIAL_REBASE","_number":24,"created":"2021-02-11 13:41:59.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/24"}}},"commit":{"parents":[{"commit":"7a8f4db000b8ed9fafb97d39cc5063c521d0d8ee","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/7a8f4db000b8ed9fafb97d39cc5063c521d0d8ee"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-02-11 13:41:36.000000000","tz":60},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/c1f00f7c8d12d7987d9216e012192e68db4f46f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/c1f00f7c8d12d7987d9216e012192e68db4f46f7"}]},"branch":"refs/heads/master"},"e27c639e2637d1931bd14380569f3facf827e15b":{"kind":"TRIVIAL_REBASE","_number":25,"created":"2021-02-15 18:16:50.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/25"}}},"commit":{"parents":[{"commit":"9f265c7b1e301f18ef3b910e302fc90750dd3f07","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/9f265c7b1e301f18ef3b910e302fc90750dd3f07"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-02-15 18:15:57.000000000","tz":60},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/e27c639e2637d1931bd14380569f3facf827e15b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/e27c639e2637d1931bd14380569f3facf827e15b"}]},"branch":"refs/heads/master"},"00dad479fee359b2ac724624a455d5dbb251073d":{"kind":"TRIVIAL_REBASE","_number":26,"created":"2021-02-26 11:25:29.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/26"}}},"commit":{"parents":[{"commit":"fd7a67f331ca490129c91f34f5a0bfbae0fba3d4","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/fd7a67f331ca490129c91f34f5a0bfbae0fba3d4"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-02-26 11:16:04.000000000","tz":60},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/00dad479fee359b2ac724624a455d5dbb251073d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/00dad479fee359b2ac724624a455d5dbb251073d"}]},"branch":"refs/heads/master"},"246499ea1c3e531c31e7afac507cda7e41bb7ece":{"kind":"TRIVIAL_REBASE","_number":27,"created":"2021-03-16 12:39:58.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/27"}}},"commit":{"parents":[{"commit":"3062f78e78fe2e06c884cd5091d7700df02f64fc","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/3062f78e78fe2e06c884cd5091d7700df02f64fc"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-03-13 16:59:20.000000000","tz":0},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/246499ea1c3e531c31e7afac507cda7e41bb7ece"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/246499ea1c3e531c31e7afac507cda7e41bb7ece"}]},"branch":"refs/heads/master"},"97c15291a65ca5d2f300b911f56e575f5e4bc593":{"kind":"TRIVIAL_REBASE","_number":28,"created":"2021-03-17 23:57:21.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/28"}}},"commit":{"parents":[{"commit":"72a19cd7828247140087d983099ac6a8b14ffdcc","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/72a19cd7828247140087d983099ac6a8b14ffdcc"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-03-17 23:56:22.000000000","tz":0},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/97c15291a65ca5d2f300b911f56e575f5e4bc593"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/97c15291a65ca5d2f300b911f56e575f5e4bc593"}]},"branch":"refs/heads/master"},"bddef043c71219cc0368c66e342f4934fa573982":{"kind":"TRIVIAL_REBASE","_number":29,"created":"2021-03-18 14:58:34.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/29"}}},"commit":{"parents":[{"commit":"b30d1468c564cdd9516525bf93d12c24ae795483","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/b30d1468c564cdd9516525bf93d12c24ae795483"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-03-18 12:49:30.000000000","tz":0},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/bddef043c71219cc0368c66e342f4934fa573982"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/bddef043c71219cc0368c66e342f4934fa573982"}]},"branch":"refs/heads/master"},"05d48e9bf117c4b82cc686fc07f9fd6968c528e1":{"kind":"TRIVIAL_REBASE","_number":30,"created":"2021-03-18 19:33:54.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/30"}}},"commit":{"parents":[{"commit":"165218e91815c1113e35c92316efa4815dcce99f","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/165218e91815c1113e35c92316efa4815dcce99f"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-03-18 17:52:03.000000000","tz":0},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/05d48e9bf117c4b82cc686fc07f9fd6968c528e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/05d48e9bf117c4b82cc686fc07f9fd6968c528e1"}]},"branch":"refs/heads/master"},"f48a89820391dd317ae91c5a4a6c654f1a9d4914":{"kind":"TRIVIAL_REBASE","_number":31,"created":"2021-03-18 19:46:00.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/31"}}},"commit":{"parents":[{"commit":"1d270d1e383da0ae57b48d56748347dfee4bfdc4","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/1d270d1e383da0ae57b48d56748347dfee4bfdc4"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-03-18 19:44:39.000000000","tz":0},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/f48a89820391dd317ae91c5a4a6c654f1a9d4914"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/f48a89820391dd317ae91c5a4a6c654f1a9d4914"}]},"branch":"refs/heads/master"},"c33923a131196b15e47b1d7cef381f608f2201f2":{"kind":"TRIVIAL_REBASE","_number":32,"created":"2021-03-19 14:37:45.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/32"}}},"commit":{"parents":[{"commit":"fe46a1810da796a51e1b2d9f4420e6c5baba2356","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/fe46a1810da796a51e1b2d9f4420e6c5baba2356"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-03-19 14:37:13.000000000","tz":0},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/c33923a131196b15e47b1d7cef381f608f2201f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/c33923a131196b15e47b1d7cef381f608f2201f2"}]},"branch":"refs/heads/master"},"5ac170e978f9ace53005264eff9e96e063808770":{"kind":"NO_CHANGE","_number":33,"created":"2021-03-19 16:50:59.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/33","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/33","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/33 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/33 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/33 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/33"}}},"commit":{"parents":[{"commit":"f687789121d25ccf8c98394cb164c01f2400abba","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/f687789121d25ccf8c98394cb164c01f2400abba"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-03-19 16:49:57.000000000","tz":0},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/5ac170e978f9ace53005264eff9e96e063808770"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/5ac170e978f9ace53005264eff9e96e063808770"}]},"branch":"refs/heads/master"},"e67b988ce0976fd903400e4c5ef367cfb11dac79":{"kind":"NO_CHANGE","_number":34,"created":"2021-03-20 07:44:31.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/34","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/34","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/34 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/34 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/34 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/34"}}},"commit":{"parents":[{"commit":"ceecc1ebaa3aab08b262f58e36502e6a0a626804","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/ceecc1ebaa3aab08b262f58e36502e6a0a626804"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-03-20 06:46:27.000000000","tz":0},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/e67b988ce0976fd903400e4c5ef367cfb11dac79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/e67b988ce0976fd903400e4c5ef367cfb11dac79"}]},"branch":"refs/heads/master"},"ea53364c75a5b88ebd9477fbfcf1686d22df8fa5":{"kind":"TRIVIAL_REBASE","_number":35,"created":"2021-03-23 15:23:00.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/35","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/35","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/35 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/35 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/35 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/35"}}},"commit":{"parents":[{"commit":"4e4cd58fef98bda4febc29b1621a7c1c4deb8dd7","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/4e4cd58fef98bda4febc29b1621a7c1c4deb8dd7"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-03-23 13:59:57.000000000","tz":0},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/ea53364c75a5b88ebd9477fbfcf1686d22df8fa5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/ea53364c75a5b88ebd9477fbfcf1686d22df8fa5"}]},"branch":"refs/heads/master"},"c9e2f1923264839020969be7263f5802e8edf391":{"kind":"TRIVIAL_REBASE","_number":36,"created":"2021-03-24 19:43:44.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/36","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/36","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/36 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/36 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/36 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/36"}}},"commit":{"parents":[{"commit":"dcc3fc93ca3b309600c03fef5d6aa0f3dc609ee5","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/dcc3fc93ca3b309600c03fef5d6aa0f3dc609ee5"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-03-24 19:34:53.000000000","tz":0},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/c9e2f1923264839020969be7263f5802e8edf391"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/c9e2f1923264839020969be7263f5802e8edf391"}]},"branch":"refs/heads/master"},"dae9901d9b380ba872858d17938976e5e91b481c":{"kind":"TRIVIAL_REBASE","_number":37,"created":"2021-03-27 18:19:07.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/37","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/37","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/37 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/37 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/37 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/37"}}},"commit":{"parents":[{"commit":"8b4b554354336da644620b92eaa0e97cba559e6e","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/8b4b554354336da644620b92eaa0e97cba559e6e"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-03-27 18:17:56.000000000","tz":60},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/dae9901d9b380ba872858d17938976e5e91b481c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/dae9901d9b380ba872858d17938976e5e91b481c"}]},"branch":"refs/heads/master"},"26ac17e0873fa84b7b96d20602d50d69ef91922b":{"kind":"TRIVIAL_REBASE","_number":38,"created":"2021-03-28 22:22:44.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/38","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/38","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/38 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/38 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/38 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/38"}}},"commit":{"parents":[{"commit":"42315caf6d4040ff6338071898611b0266e77cb9","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/42315caf6d4040ff6338071898611b0266e77cb9"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-03-28 22:21:51.000000000","tz":0},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/26ac17e0873fa84b7b96d20602d50d69ef91922b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/26ac17e0873fa84b7b96d20602d50d69ef91922b"}]},"branch":"refs/heads/master"},"560791290313b52458e96e4294cd1a96eeb88710":{"kind":"TRIVIAL_REBASE","_number":39,"created":"2021-03-29 10:23:41.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/39","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/39","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/39 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/39 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/39 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/39"}}},"commit":{"parents":[{"commit":"8b93897613f8d9e1e8224579e90e3527d8f786ab","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/8b93897613f8d9e1e8224579e90e3527d8f786ab"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-03-29 10:23:10.000000000","tz":0},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/560791290313b52458e96e4294cd1a96eeb88710"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/560791290313b52458e96e4294cd1a96eeb88710"}]},"branch":"refs/heads/master"},"06a6b5ef224d5b221d1423733fc3cd43b6ebc898":{"kind":"TRIVIAL_REBASE","_number":40,"created":"2021-03-29 11:00:25.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/40","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/40","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/40 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/40 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/40 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/40"}}},"commit":{"parents":[{"commit":"6e2a4a3731355e5fa246d47cd82f6a98adc7200d","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/6e2a4a3731355e5fa246d47cd82f6a98adc7200d"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-03-29 11:00:04.000000000","tz":0},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/06a6b5ef224d5b221d1423733fc3cd43b6ebc898"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/06a6b5ef224d5b221d1423733fc3cd43b6ebc898"}]},"branch":"refs/heads/master"},"3e2e38fbea34df885c4182723893616c07418622":{"kind":"TRIVIAL_REBASE","_number":41,"created":"2021-04-07 12:22:10.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/41","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/41","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/41 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/41 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/41 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/41"}}},"commit":{"parents":[{"commit":"6df1040952c5833150830f7df991b1f031dba579","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/6df1040952c5833150830f7df991b1f031dba579"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-04-07 12:15:10.000000000","tz":120},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/3e2e38fbea34df885c4182723893616c07418622"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/3e2e38fbea34df885c4182723893616c07418622"}]},"branch":"refs/heads/master"},"343749801b99afa168b8acbdf4ed41891fb6c1b7":{"kind":"TRIVIAL_REBASE","_number":42,"created":"2021-04-07 12:42:33.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/42","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/42","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/42 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/42 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/42 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/42"}}},"commit":{"parents":[{"commit":"0a1a523ffccf5cbe1b76e73881573b4a884b646e","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/0a1a523ffccf5cbe1b76e73881573b4a884b646e"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-04-07 12:42:12.000000000","tz":120},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/343749801b99afa168b8acbdf4ed41891fb6c1b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/343749801b99afa168b8acbdf4ed41891fb6c1b7"}]},"branch":"refs/heads/master"},"36e9c3669dc0218d9c25c3b917f20623829c651d":{"kind":"TRIVIAL_REBASE","_number":43,"created":"2021-04-07 21:27:30.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/43","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/43","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/43 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/43 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/43 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/43"}}},"commit":{"parents":[{"commit":"7c036b4a95e9d5871f819fbcc1253032f9fabdf0","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/7c036b4a95e9d5871f819fbcc1253032f9fabdf0"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-04-07 21:19:42.000000000","tz":120},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/36e9c3669dc0218d9c25c3b917f20623829c651d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/36e9c3669dc0218d9c25c3b917f20623829c651d"}]},"branch":"refs/heads/master"},"e0427506683950551958ded265b0d76296c8be69":{"kind":"TRIVIAL_REBASE","_number":44,"created":"2021-04-10 07:27:27.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/44","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/44","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/44 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/44 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/44 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/44"}}},"commit":{"parents":[{"commit":"907df5f5b19f2e487d360f9d5e0f060dc8415888","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/907df5f5b19f2e487d360f9d5e0f060dc8415888"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-04-10 07:25:40.000000000","tz":120},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/e0427506683950551958ded265b0d76296c8be69"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/e0427506683950551958ded265b0d76296c8be69"}]},"branch":"refs/heads/master"},"f31f7059799ee6c87cc1880ad89f71ef2c2508d1":{"kind":"TRIVIAL_REBASE","_number":45,"created":"2021-04-10 07:30:38.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/45","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/45","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/45 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/45 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/45 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/45"}}},"commit":{"parents":[{"commit":"d81543eb38fefcfc386f728a50f71ecd81a2bd26","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/d81543eb38fefcfc386f728a50f71ecd81a2bd26"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-04-10 07:30:07.000000000","tz":120},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/f31f7059799ee6c87cc1880ad89f71ef2c2508d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/f31f7059799ee6c87cc1880ad89f71ef2c2508d1"}]},"branch":"refs/heads/master"},"ffb7bf177bd0bb6b42eb83af48b79e8aaf398fe4":{"kind":"TRIVIAL_REBASE","_number":46,"created":"2021-04-14 14:54:25.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/46","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/46","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/46 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/46 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/46 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/46"}}},"commit":{"parents":[{"commit":"8c7a5f0118c281f2482f1396b1555563ae90d304","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/8c7a5f0118c281f2482f1396b1555563ae90d304"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-04-14 14:10:30.000000000","tz":120},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/ffb7bf177bd0bb6b42eb83af48b79e8aaf398fe4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/ffb7bf177bd0bb6b42eb83af48b79e8aaf398fe4"}]},"branch":"refs/heads/master"},"d690e2abfcbf7e493270fd9457f80f63954a07a4":{"kind":"TRIVIAL_REBASE","_number":47,"created":"2021-04-19 16:34:03.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/47","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/47","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/47 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/47 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/47 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/47"}}},"commit":{"parents":[{"commit":"68b59c07d2a03365023fae87d8d08ba87698734f","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/68b59c07d2a03365023fae87d8d08ba87698734f"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-04-19 16:33:27.000000000","tz":120},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/d690e2abfcbf7e493270fd9457f80f63954a07a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/d690e2abfcbf7e493270fd9457f80f63954a07a4"}]},"branch":"refs/heads/master"},"bca4194a04c7d26c06385f1f5f594cec555b4e4f":{"kind":"TRIVIAL_REBASE","_number":48,"created":"2021-06-24 10:48:52.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/48","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/48","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/48 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/48 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/48 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/48"}}},"commit":{"parents":[{"commit":"3072bb8f5fc68114d7ea56d1939c7017ed10d0e8","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/3072bb8f5fc68114d7ea56d1939c7017ed10d0e8"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-06-24 10:41:57.000000000","tz":120},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/bca4194a04c7d26c06385f1f5f594cec555b4e4f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/bca4194a04c7d26c06385f1f5f594cec555b4e4f"}]},"branch":"refs/heads/master"},"13025740896ac4eb053401968489b28ff377c30c":{"kind":"TRIVIAL_REBASE","_number":49,"created":"2021-06-24 13:30:51.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/49","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/49","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/49 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/49 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/49 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/49"}}},"commit":{"parents":[{"commit":"4c655a054ea7aa75e3c69e33a0bbef838f527c15","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/4c655a054ea7aa75e3c69e33a0bbef838f527c15"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-06-24 13:17:32.000000000","tz":120},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/13025740896ac4eb053401968489b28ff377c30c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/13025740896ac4eb053401968489b28ff377c30c"}]},"branch":"refs/heads/master"},"3b12ff4450c40f13395354b7dd613abc3cb4accd":{"kind":"TRIVIAL_REBASE","_number":50,"created":"2021-07-09 10:38:43.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/50","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/50","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/50 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/50 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/50 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/50"}}},"commit":{"parents":[{"commit":"81a9f6b7c5abd1938c98b9f5914df0a5906fc5f6","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/81a9f6b7c5abd1938c98b9f5914df0a5906fc5f6"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-07-09 10:37:54.000000000","tz":120},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/3b12ff4450c40f13395354b7dd613abc3cb4accd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/3b12ff4450c40f13395354b7dd613abc3cb4accd"}]},"branch":"refs/heads/master"},"eaffc7fe4d403b27c4864f6d870a60ea07783e63":{"kind":"TRIVIAL_REBASE","_number":51,"created":"2021-08-02 09:04:24.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/51","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/51","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/51 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/51 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/51 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/51"}}},"commit":{"parents":[{"commit":"00a56547038ab1105166be70962c53d3230c9165","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/00a56547038ab1105166be70962c53d3230c9165"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-08-02 09:03:55.000000000","tz":120},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/eaffc7fe4d403b27c4864f6d870a60ea07783e63"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/eaffc7fe4d403b27c4864f6d870a60ea07783e63"}]},"branch":"refs/heads/master"},"9b8f753675e84a99203d1758f3c23c239cbe8f77":{"kind":"TRIVIAL_REBASE","_number":52,"created":"2021-08-02 10:41:32.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/52","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/52","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/52 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/52 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/52 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/52"}}},"commit":{"parents":[{"commit":"5b301ab3c3f4f0605641c8c76718c814f48870c7","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/5b301ab3c3f4f0605641c8c76718c814f48870c7"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-08-02 10:39:40.000000000","tz":120},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/9b8f753675e84a99203d1758f3c23c239cbe8f77"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/9b8f753675e84a99203d1758f3c23c239cbe8f77"}]},"branch":"refs/heads/master"},"918035eac61037f5d6ad79917b496df9587b2a77":{"kind":"TRIVIAL_REBASE","_number":53,"created":"2021-08-17 15:20:48.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/53","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/53","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/53 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/53 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/53 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/53"}}},"commit":{"parents":[{"commit":"6b11fc0207deea5ec79535df36b738133267b539","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/6b11fc0207deea5ec79535df36b738133267b539"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-08-17 15:20:31.000000000","tz":120},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/918035eac61037f5d6ad79917b496df9587b2a77"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/918035eac61037f5d6ad79917b496df9587b2a77"}]},"branch":"refs/heads/master"},"6cf12a81febec1eea5bdefd52bd4e1b57e50abca":{"kind":"TRIVIAL_REBASE","_number":54,"created":"2021-08-17 21:22:57.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/54","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/54","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/54 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/54 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/54 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/54"}}},"commit":{"parents":[{"commit":"738763c9362c7aabb9a48c75e076173004c01d59","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/738763c9362c7aabb9a48c75e076173004c01d59"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-08-17 21:21:40.000000000","tz":120},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/6cf12a81febec1eea5bdefd52bd4e1b57e50abca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/6cf12a81febec1eea5bdefd52bd4e1b57e50abca"}]},"branch":"refs/heads/master"},"ffed9d95447b82d4dc573b2fd0947121f40235c8":{"kind":"TRIVIAL_REBASE","_number":55,"created":"2021-08-19 10:07:22.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/55","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/55","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/55 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/55 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/55 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/55"}}},"commit":{"parents":[{"commit":"5001d845855626dd9eb2eb39fdb6b1c063743d81","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/5001d845855626dd9eb2eb39fdb6b1c063743d81"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-08-19 10:06:40.000000000","tz":120},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/ffed9d95447b82d4dc573b2fd0947121f40235c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/ffed9d95447b82d4dc573b2fd0947121f40235c8"}]},"branch":"refs/heads/master"},"36856faecc39edffbc427a590cca6c7ace05a753":{"kind":"TRIVIAL_REBASE","_number":56,"created":"2021-08-19 19:28:46.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/56","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/56","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/56 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/56 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/56 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/56"}}},"commit":{"parents":[{"commit":"2f6943dc0f544534c35e91672a82e96a3f9199ad","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/2f6943dc0f544534c35e91672a82e96a3f9199ad"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-08-19 19:22:35.000000000","tz":120},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/36856faecc39edffbc427a590cca6c7ace05a753"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/36856faecc39edffbc427a590cca6c7ace05a753"}]},"branch":"refs/heads/master"},"4775f172d8165798b3dcc81abeb0f0986071a568":{"kind":"TRIVIAL_REBASE","_number":57,"created":"2021-09-15 19:57:14.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/57","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/57","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/57 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/57 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/57 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/57"}}},"commit":{"parents":[{"commit":"febd1dec36df0be6d56a7639b2f797f200c30e1b","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/febd1dec36df0be6d56a7639b2f797f200c30e1b"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-09-15 19:56:22.000000000","tz":120},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/4775f172d8165798b3dcc81abeb0f0986071a568"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/4775f172d8165798b3dcc81abeb0f0986071a568"}]},"branch":"refs/heads/master"},"60e2e727b6e40487bdbf91fd43f978f29266f101":{"kind":"TRIVIAL_REBASE","_number":58,"created":"2021-09-21 15:30:40.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/58","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/58","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/58 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/58 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/58 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/58"}}},"commit":{"parents":[{"commit":"f6bafec3cfa4b5a8f8bd2cac43059d65214c5c2e","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/f6bafec3cfa4b5a8f8bd2cac43059d65214c5c2e"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-09-21 15:29:44.000000000","tz":120},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/60e2e727b6e40487bdbf91fd43f978f29266f101"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/60e2e727b6e40487bdbf91fd43f978f29266f101"}]},"branch":"refs/heads/master"},"b8a95e4b7a1b0306cffc63470d8aee9540dcdb0c":{"kind":"TRIVIAL_REBASE","_number":59,"created":"2021-10-01 14:36:36.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/59","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/59","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/59 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/59 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/59 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/59"}}},"commit":{"parents":[{"commit":"6a9646f4c0abfe372fad12a13a996a6e11a97238","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/6a9646f4c0abfe372fad12a13a996a6e11a97238"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-10-01 14:21:37.000000000","tz":120},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/b8a95e4b7a1b0306cffc63470d8aee9540dcdb0c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/b8a95e4b7a1b0306cffc63470d8aee9540dcdb0c"}]},"branch":"refs/heads/master"},"f692575981764fb555082a2549f28a5d2505f97d":{"kind":"TRIVIAL_REBASE","_number":60,"created":"2021-10-22 11:38:59.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/60","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/60","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/60 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/60 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/60 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/60"}}},"commit":{"parents":[{"commit":"6c762315de3986d2e7bedba398053f576157fba0","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/6c762315de3986d2e7bedba398053f576157fba0"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-10-22 11:38:14.000000000","tz":120},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/f692575981764fb555082a2549f28a5d2505f97d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/f692575981764fb555082a2549f28a5d2505f97d"}]},"branch":"refs/heads/master"},"caa8e3e777f5adea1072e7f96f8b526fd6d58cff":{"kind":"TRIVIAL_REBASE","_number":61,"created":"2021-10-22 13:15:46.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/61","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/61","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/61 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/61 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/61 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/61"}}},"commit":{"parents":[{"commit":"7870254dd7f0d25dbb487ed27acb3f212c889ac6","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/7870254dd7f0d25dbb487ed27acb3f212c889ac6"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-10-22 13:14:53.000000000","tz":120},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/caa8e3e777f5adea1072e7f96f8b526fd6d58cff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/caa8e3e777f5adea1072e7f96f8b526fd6d58cff"}]},"branch":"refs/heads/master"},"325f0fa37f051708b694a5a24dede778f7e24547":{"kind":"TRIVIAL_REBASE","_number":62,"created":"2022-01-11 15:10:33.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/62","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/62","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/62 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/62 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/62 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/62"}}},"commit":{"parents":[{"commit":"84b88bd2b202ae158cbd126c609c9b9f0cd3706d","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/84b88bd2b202ae158cbd126c609c9b9f0cd3706d"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2022-01-11 15:10:21.000000000","tz":60},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/325f0fa37f051708b694a5a24dede778f7e24547"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/325f0fa37f051708b694a5a24dede778f7e24547"}]},"branch":"refs/heads/master"},"a72e392b85cd168e9de20d702978bb837520ac70":{"kind":"TRIVIAL_REBASE","_number":63,"created":"2022-01-11 15:26:28.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/63","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/63","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/63 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/63 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/63 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/63"}}},"commit":{"parents":[{"commit":"a443e96c6ed7ae2f3143919d0511d275679ef58d","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/a443e96c6ed7ae2f3143919d0511d275679ef58d"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2022-01-11 15:16:15.000000000","tz":60},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/a72e392b85cd168e9de20d702978bb837520ac70"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/a72e392b85cd168e9de20d702978bb837520ac70"}]},"branch":"refs/heads/master"},"663e1a3a35755d3e866e481ef9fdf259c595b0d7":{"kind":"TRIVIAL_REBASE","_number":64,"created":"2022-01-13 14:06:20.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/64","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/64","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/64 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/64 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/64 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/64"}}},"commit":{"parents":[{"commit":"75a5a15eea5870c0ffe6ee0b2d6d9f2d4c9efe4f","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/75a5a15eea5870c0ffe6ee0b2d6d9f2d4c9efe4f"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2022-01-13 13:37:44.000000000","tz":60},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/663e1a3a35755d3e866e481ef9fdf259c595b0d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/663e1a3a35755d3e866e481ef9fdf259c595b0d7"}]},"branch":"refs/heads/master"},"12363a7b8b5d4e02e402b8a6139d683f7dc97837":{"kind":"TRIVIAL_REBASE","_number":65,"created":"2022-01-20 08:43:26.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/65","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/65","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/65 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/65 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/65 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/65"}}},"commit":{"parents":[{"commit":"7e11fad74f79eaf7b34502bea3e9bf71aa8580d2","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/7e11fad74f79eaf7b34502bea3e9bf71aa8580d2"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2022-01-20 08:40:39.000000000","tz":60},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/12363a7b8b5d4e02e402b8a6139d683f7dc97837"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/12363a7b8b5d4e02e402b8a6139d683f7dc97837"}]},"branch":"refs/heads/master"},"a5da4deaa281e0380581c3848b187c5d33b673b3":{"kind":"TRIVIAL_REBASE","_number":66,"created":"2022-01-28 09:13:48.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/66","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/66","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/66 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/66 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/66 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/66"}}},"commit":{"parents":[{"commit":"62fe0594a5920f4381eccfc3617a0f5c6a717315","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/62fe0594a5920f4381eccfc3617a0f5c6a717315"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2022-01-28 08:58:49.000000000","tz":60},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/a5da4deaa281e0380581c3848b187c5d33b673b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/a5da4deaa281e0380581c3848b187c5d33b673b3"}]},"branch":"refs/heads/master"},"ec882f0411eb57472b2c761844789790468daf70":{"kind":"TRIVIAL_REBASE","_number":67,"created":"2022-02-01 08:38:31.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/67","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/67","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/67 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/67 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/67 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/67"}}},"commit":{"parents":[{"commit":"8c93abbe12530d88098dc4a88093ed6bad96bb58","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/8c93abbe12530d88098dc4a88093ed6bad96bb58"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2022-02-01 08:38:21.000000000","tz":60},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/ec882f0411eb57472b2c761844789790468daf70"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/ec882f0411eb57472b2c761844789790468daf70"}]},"branch":"refs/heads/master"},"be64fdaf21b38e3bb26e9b5f08300a6f663568db":{"kind":"TRIVIAL_REBASE","_number":68,"created":"2022-03-11 09:55:06.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/68","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/68","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/68 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/68 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/68 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/68"}}},"commit":{"parents":[{"commit":"9b0143b498496a68221c2844c2424c9bfdea9a54","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/9b0143b498496a68221c2844c2424c9bfdea9a54"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2022-03-11 09:54:50.000000000","tz":60},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/be64fdaf21b38e3bb26e9b5f08300a6f663568db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/be64fdaf21b38e3bb26e9b5f08300a6f663568db"}]},"branch":"refs/heads/master"},"e62421d40adf67b5afc8b4363849805b5425c7c6":{"kind":"TRIVIAL_REBASE","_number":69,"created":"2022-03-15 17:20:41.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/69","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/69","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/69 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/69 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/69 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/69"}}},"commit":{"parents":[{"commit":"91cbd2dfc72c31a83acfe7596a4e54140390f0bc","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/91cbd2dfc72c31a83acfe7596a4e54140390f0bc"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2022-03-15 17:20:14.000000000","tz":60},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/e62421d40adf67b5afc8b4363849805b5425c7c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/e62421d40adf67b5afc8b4363849805b5425c7c6"}]},"branch":"refs/heads/master"},"752fea7e8d45d25c686b71c15305f139fa5e569c":{"kind":"TRIVIAL_REBASE","_number":70,"created":"2022-03-18 08:23:32.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/70","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/70","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/70 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/70 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/70 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/70"}}},"commit":{"parents":[{"commit":"a33eaa743ab893bbbccf6b29df70ae52739c6470","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/a33eaa743ab893bbbccf6b29df70ae52739c6470"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2022-03-18 08:23:16.000000000","tz":60},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/752fea7e8d45d25c686b71c15305f139fa5e569c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/752fea7e8d45d25c686b71c15305f139fa5e569c"}]},"branch":"refs/heads/master"},"9e9774db5cfdd073bada2ea87948f99e32ec6fc8":{"kind":"TRIVIAL_REBASE","_number":71,"created":"2022-03-18 10:11:32.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/71","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/71","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/71 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/71 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/71 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/71"}}},"commit":{"parents":[{"commit":"14eea2f407608dc89c08b8f0129c8208832e84c3","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/14eea2f407608dc89c08b8f0129c8208832e84c3"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2022-03-18 10:11:21.000000000","tz":60},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/9e9774db5cfdd073bada2ea87948f99e32ec6fc8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/9e9774db5cfdd073bada2ea87948f99e32ec6fc8"}]},"branch":"refs/heads/master"},"a88f11a285c1cc62c7d6524d9608df7c9e35e460":{"kind":"TRIVIAL_REBASE","_number":72,"created":"2022-04-21 12:31:10.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/72","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/72","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/72 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/72 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/72 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/72"}}},"commit":{"parents":[{"commit":"78e51165d760502bcb90def129b28a25d73962b4","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/78e51165d760502bcb90def129b28a25d73962b4"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2022-04-21 12:30:13.000000000","tz":120},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/a88f11a285c1cc62c7d6524d9608df7c9e35e460"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/a88f11a285c1cc62c7d6524d9608df7c9e35e460"}]},"branch":"refs/heads/master"},"b9a95ce10f8b82ca32b8aedfd7198b6700bcdcbc":{"kind":"TRIVIAL_REBASE","_number":73,"created":"2022-04-21 16:14:43.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/73","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/73","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/73 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/73 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/73 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/73"}}},"commit":{"parents":[{"commit":"00bc23874aa187d46457b502ed7e8e95bf5434d8","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/00bc23874aa187d46457b502ed7e8e95bf5434d8"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2022-04-21 16:10:45.000000000","tz":120},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/b9a95ce10f8b82ca32b8aedfd7198b6700bcdcbc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/b9a95ce10f8b82ca32b8aedfd7198b6700bcdcbc"}]},"branch":"refs/heads/master"},"a32345293f5c15191736245559212e4f058263ea":{"kind":"NO_CHANGE","_number":74,"created":"2022-04-21 16:59:42.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/74","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/74","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/74 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/74 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/74 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/74"}}},"commit":{"parents":[{"commit":"e9512206eeb6d8931e59b7a29f2ffbcc4b6cc61f","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/e9512206eeb6d8931e59b7a29f2ffbcc4b6cc61f"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2022-04-21 16:59:15.000000000","tz":120},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/a32345293f5c15191736245559212e4f058263ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/a32345293f5c15191736245559212e4f058263ea"}]},"branch":"refs/heads/master"},"48044242454b8b81f7a017c908e79339eef0fbf9":{"kind":"TRIVIAL_REBASE","_number":75,"created":"2022-04-21 17:42:26.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/75","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/75","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/75 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/75 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/75 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/75"}}},"commit":{"parents":[{"commit":"2a1222c549eaabd7e0908ff520316170d2330510","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/2a1222c549eaabd7e0908ff520316170d2330510"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2022-04-21 17:42:13.000000000","tz":120},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/48044242454b8b81f7a017c908e79339eef0fbf9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/48044242454b8b81f7a017c908e79339eef0fbf9"}]},"branch":"refs/heads/master"},"a21a63299347646a74c454e4552699e51da252a0":{"kind":"TRIVIAL_REBASE","_number":76,"created":"2022-05-02 11:08:00.000000000","uploader":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"ref":"refs/changes/21/770621/76","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/76","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/76 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/76 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/76 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/76"}}},"commit":{"parents":[{"commit":"61f70e4b4e2ca16415dc4b49710cf6b7332be2be","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/61f70e4b4e2ca16415dc4b49710cf6b7332be2be"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Radosław Piliszek","email":"radoslaw.piliszek@gmail.com","date":"2022-05-02 11:08:00.000000000","tz":0},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/a21a63299347646a74c454e4552699e51da252a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/a21a63299347646a74c454e4552699e51da252a0"}]},"branch":"refs/heads/master","description":"Rebase"},"d61bf7513d64fd21332cfe33ef90f0faa97ab129":{"kind":"REWORK","_number":77,"created":"2022-05-04 09:52:35.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/77","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/77","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/77 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/77 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/77 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/77"}}},"commit":{"parents":[{"commit":"43bd98b0f984456e9a8d21cd51c8f01ca485684e","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/43bd98b0f984456e9a8d21cd51c8f01ca485684e"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2022-05-04 09:17:50.000000000","tz":0},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/d61bf7513d64fd21332cfe33ef90f0faa97ab129"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/d61bf7513d64fd21332cfe33ef90f0faa97ab129"}]},"branch":"refs/heads/master"},"bb7e66c9143e3a4eb18c7e29be56e286d60c4770":{"kind":"TRIVIAL_REBASE","_number":78,"created":"2022-05-04 09:54:13.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/78","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/78","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/78 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/78 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/78 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/78"}}},"commit":{"parents":[{"commit":"59dbf88b9e6df08a00d7db858eeee33ae71c74ac","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/59dbf88b9e6df08a00d7db858eeee33ae71c74ac"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2022-05-04 09:54:04.000000000","tz":120},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/bb7e66c9143e3a4eb18c7e29be56e286d60c4770"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/bb7e66c9143e3a4eb18c7e29be56e286d60c4770"}]},"branch":"refs/heads/master"},"508a0102e1d39ef96f54dea3c5ccecdba35c370e":{"kind":"TRIVIAL_REBASE","_number":79,"created":"2022-05-04 09:58:59.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/79","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/79","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/79 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/79 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/79 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/79"}}},"commit":{"parents":[{"commit":"c3e9133ae3db18a41b2751eda44de3a876a030c2","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/c3e9133ae3db18a41b2751eda44de3a876a030c2"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2022-05-04 09:58:50.000000000","tz":120},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/508a0102e1d39ef96f54dea3c5ccecdba35c370e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/508a0102e1d39ef96f54dea3c5ccecdba35c370e"}]},"branch":"refs/heads/master"},"c74bf2c1b0d1c4bd7eafa69cb0a5609308b34693":{"kind":"REWORK","_number":80,"created":"2022-05-04 10:26:55.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/80","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/80","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/80 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/80 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/80 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/80"}}},"commit":{"parents":[{"commit":"c3e9133ae3db18a41b2751eda44de3a876a030c2","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/c3e9133ae3db18a41b2751eda44de3a876a030c2"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2022-05-04 10:26:34.000000000","tz":120},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/c74bf2c1b0d1c4bd7eafa69cb0a5609308b34693"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/c74bf2c1b0d1c4bd7eafa69cb0a5609308b34693"}]},"branch":"refs/heads/master"},"6c5632b1128d11c6a9a5fef37dcd17226ee3aec9":{"kind":"REWORK","_number":81,"created":"2022-05-04 10:44:35.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/81","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/81","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/81 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/81 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/81 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/81"}}},"commit":{"parents":[{"commit":"c3e9133ae3db18a41b2751eda44de3a876a030c2","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/c3e9133ae3db18a41b2751eda44de3a876a030c2"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2022-05-04 10:44:21.000000000","tz":120},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/6c5632b1128d11c6a9a5fef37dcd17226ee3aec9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/6c5632b1128d11c6a9a5fef37dcd17226ee3aec9"}]},"branch":"refs/heads/master"},"30a55b1c495ec24e27fa32379d5435ed1be790a2":{"kind":"REWORK","_number":82,"created":"2022-05-05 10:36:18.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/82","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/82","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/82 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/82 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/82 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/82"}}},"commit":{"parents":[{"commit":"32cee6064b0bbef76d76cd94f7b37c62b4e50b3f","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/32cee6064b0bbef76d76cd94f7b37c62b4e50b3f"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2022-05-05 10:35:59.000000000","tz":120},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/30a55b1c495ec24e27fa32379d5435ed1be790a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/30a55b1c495ec24e27fa32379d5435ed1be790a2"}]},"branch":"refs/heads/master"},"d61ed163e93d16e6004f811fbdd16406f16f64d2":{"kind":"TRIVIAL_REBASE","_number":83,"created":"2022-05-31 09:26:22.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/83","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/83","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/83 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/83 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/83 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/83"}}},"commit":{"parents":[{"commit":"90423a98172fa22c52801a356e2fb2c3a3b47b8d","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/90423a98172fa22c52801a356e2fb2c3a3b47b8d"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2022-05-31 09:25:58.000000000","tz":120},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/d61ed163e93d16e6004f811fbdd16406f16f64d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/d61ed163e93d16e6004f811fbdd16406f16f64d2"}]},"branch":"refs/heads/master"},"f212193cb3e799fc5e3aea97d2760d7a9db46d26":{"kind":"TRIVIAL_REBASE","_number":84,"created":"2022-05-31 10:59:38.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/84","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/84","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/84 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/84 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/84 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/84"}}},"commit":{"parents":[{"commit":"25079a6f251e04689eb4967575ff73c40c966fb1","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/25079a6f251e04689eb4967575ff73c40c966fb1"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2022-05-31 10:59:15.000000000","tz":120},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/f212193cb3e799fc5e3aea97d2760d7a9db46d26"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/f212193cb3e799fc5e3aea97d2760d7a9db46d26"}]},"branch":"refs/heads/master"},"4770bc018c3fa550f6e8df431689d0c11d0c94c4":{"kind":"TRIVIAL_REBASE","_number":85,"created":"2022-05-31 12:50:49.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/85","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/85","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/85 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/85 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/85 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/85"}}},"commit":{"parents":[{"commit":"b17cfc2a2ed94a014abfce0d27ab2cc5b6851176","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/b17cfc2a2ed94a014abfce0d27ab2cc5b6851176"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2022-05-31 12:50:26.000000000","tz":120},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/4770bc018c3fa550f6e8df431689d0c11d0c94c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/4770bc018c3fa550f6e8df431689d0c11d0c94c4"}]},"branch":"refs/heads/master"},"cc1594d689fb106faf41d4f9c57e470c46bfba66":{"kind":"TRIVIAL_REBASE","_number":86,"created":"2022-06-01 11:16:57.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/86","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/86","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/86 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/86 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/86 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/86"}}},"commit":{"parents":[{"commit":"232ec1e636f32ea23c315a49cee4929ab6de9e44","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/232ec1e636f32ea23c315a49cee4929ab6de9e44"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2022-06-01 11:16:20.000000000","tz":120},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/cc1594d689fb106faf41d4f9c57e470c46bfba66"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/cc1594d689fb106faf41d4f9c57e470c46bfba66"}]},"branch":"refs/heads/master"},"ac7c7dd0e9645008d937a6bfc185cb373856ba14":{"kind":"TRIVIAL_REBASE","_number":87,"created":"2022-06-02 12:52:42.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/87","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/87","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/87 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/87 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/87 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/87"}}},"commit":{"parents":[{"commit":"8f02f0cad5fe4ca3a1bd1679531dcc17491d3851","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/8f02f0cad5fe4ca3a1bd1679531dcc17491d3851"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2022-06-02 12:52:30.000000000","tz":120},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/ac7c7dd0e9645008d937a6bfc185cb373856ba14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/ac7c7dd0e9645008d937a6bfc185cb373856ba14"}]},"branch":"refs/heads/master"},"65c4a3be1d9bfa4176efa8bcf832897b081ab3cb":{"kind":"TRIVIAL_REBASE","_number":88,"created":"2022-06-03 09:08:12.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/88","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/88","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/88 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/88 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/88 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/88"}}},"commit":{"parents":[{"commit":"1f12a0c824c31945966a374d12b74757a62311c1","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/1f12a0c824c31945966a374d12b74757a62311c1"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2022-06-03 08:59:45.000000000","tz":120},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/65c4a3be1d9bfa4176efa8bcf832897b081ab3cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/65c4a3be1d9bfa4176efa8bcf832897b081ab3cb"}]},"branch":"refs/heads/master"},"e603926d44076b4f5f5e8be3de780e0628ba9014":{"kind":"TRIVIAL_REBASE","_number":89,"created":"2022-06-13 17:22:32.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/89","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/89","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/89 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/89 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/89 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/89"}}},"commit":{"parents":[{"commit":"ee17eb66acd404c39bae4157d5fa3a746aafab56","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/ee17eb66acd404c39bae4157d5fa3a746aafab56"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2022-06-13 17:21:08.000000000","tz":120},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/e603926d44076b4f5f5e8be3de780e0628ba9014"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/e603926d44076b4f5f5e8be3de780e0628ba9014"}]},"branch":"refs/heads/master"},"9e9b9f82779c64d857b4464f99f473c15c5ef9e9":{"kind":"TRIVIAL_REBASE","_number":90,"created":"2022-06-17 17:07:34.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/90","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/90","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/90 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/90 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/90 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/90"}}},"commit":{"parents":[{"commit":"a9d12c0723a13d25fcad172f6335120b54be80c7","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/a9d12c0723a13d25fcad172f6335120b54be80c7"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2022-06-17 17:07:22.000000000","tz":120},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/9e9b9f82779c64d857b4464f99f473c15c5ef9e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/9e9b9f82779c64d857b4464f99f473c15c5ef9e9"}]},"branch":"refs/heads/master"},"0329c7131532e1c15d7474f250414699a13a0f7e":{"kind":"TRIVIAL_REBASE","_number":91,"created":"2022-07-04 16:42:41.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/91","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/91","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/91 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/91 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/91 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/91"}}},"commit":{"parents":[{"commit":"2b4bcea43da4a3e5be5f3a4a530c92eaf0dca777","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/2b4bcea43da4a3e5be5f3a4a530c92eaf0dca777"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2022-07-04 16:42:23.000000000","tz":120},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/0329c7131532e1c15d7474f250414699a13a0f7e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/0329c7131532e1c15d7474f250414699a13a0f7e"}]},"branch":"refs/heads/master"},"0b13184f5f8698e24a015cede2bb2fbba77e40a4":{"kind":"TRIVIAL_REBASE","_number":92,"created":"2022-07-04 18:26:36.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/92","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/92","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/92 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/92 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/92 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/92"}}},"commit":{"parents":[{"commit":"30e113edbeb15e83624b915c7b3768d0992150ef","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/30e113edbeb15e83624b915c7b3768d0992150ef"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2022-07-04 18:25:30.000000000","tz":120},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/0b13184f5f8698e24a015cede2bb2fbba77e40a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/0b13184f5f8698e24a015cede2bb2fbba77e40a4"}]},"branch":"refs/heads/master"},"7ff20c0e685811491cbac980f9c6b7795951ba71":{"kind":"TRIVIAL_REBASE","_number":93,"created":"2022-07-05 21:02:41.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/93","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/93","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/93 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/93 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/93 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/93"}}},"commit":{"parents":[{"commit":"494b5b4ee726f70bb2de32ab6e6e7d7ecca1078b","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/494b5b4ee726f70bb2de32ab6e6e7d7ecca1078b"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2022-07-05 21:01:23.000000000","tz":0},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/7ff20c0e685811491cbac980f9c6b7795951ba71"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/7ff20c0e685811491cbac980f9c6b7795951ba71"}]},"branch":"refs/heads/master"},"fed6afe8d88b3a9330d1ca4b41a6e20eb22ed742":{"kind":"TRIVIAL_REBASE","_number":94,"created":"2022-07-06 14:58:32.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/94","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/94","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/94 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/94 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/94 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/94"}}},"commit":{"parents":[{"commit":"9a43aa0cedb2f6a585b0aac5db9333cdbc2ecd87","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/9a43aa0cedb2f6a585b0aac5db9333cdbc2ecd87"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2022-07-06 14:55:52.000000000","tz":120},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/fed6afe8d88b3a9330d1ca4b41a6e20eb22ed742"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/fed6afe8d88b3a9330d1ca4b41a6e20eb22ed742"}]},"branch":"refs/heads/master"},"041addbeb0eb26e164d22567e9547a29265c483a":{"kind":"TRIVIAL_REBASE","_number":95,"created":"2022-07-06 15:04:38.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/95","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/95","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/95 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/95 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/95 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/95"}}},"commit":{"parents":[{"commit":"031d337955f08b88fdcd541e86a8bd135aa43630","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/031d337955f08b88fdcd541e86a8bd135aa43630"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2022-07-06 15:04:28.000000000","tz":120},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/041addbeb0eb26e164d22567e9547a29265c483a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/041addbeb0eb26e164d22567e9547a29265c483a"}]},"branch":"refs/heads/master"},"f2e241c126e740e9843e0c55b12ab737be462f7e":{"kind":"TRIVIAL_REBASE","_number":96,"created":"2022-07-06 16:00:06.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/96","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/96","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/96 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/96 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/96 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/96"}}},"commit":{"parents":[{"commit":"160be50293cf89b844cae41a51c74586e369046b","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/160be50293cf89b844cae41a51c74586e369046b"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2022-07-06 15:59:42.000000000","tz":120},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/f2e241c126e740e9843e0c55b12ab737be462f7e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/f2e241c126e740e9843e0c55b12ab737be462f7e"}]},"branch":"refs/heads/master"},"1f764af0b7e4d4f13db9a07d852bf44fc5eae2e2":{"kind":"TRIVIAL_REBASE","_number":97,"created":"2022-07-12 16:07:38.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/97","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/97","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/97 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/97 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/97 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/97"}}},"commit":{"parents":[{"commit":"432f92d516f5c56c5957475e055d8f36c499ca99","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/432f92d516f5c56c5957475e055d8f36c499ca99"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2022-07-12 16:07:25.000000000","tz":120},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/1f764af0b7e4d4f13db9a07d852bf44fc5eae2e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/1f764af0b7e4d4f13db9a07d852bf44fc5eae2e2"}]},"branch":"refs/heads/master"},"c51e08abc1d93145c22c2327987be95323c36b0d":{"kind":"TRIVIAL_REBASE","_number":98,"created":"2022-07-12 16:25:31.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/98","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/98","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/98 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/98 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/98 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/98"}}},"commit":{"parents":[{"commit":"4c11c8c82023abdb457deaa96db7334ed4e510a0","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/4c11c8c82023abdb457deaa96db7334ed4e510a0"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2022-07-12 16:23:20.000000000","tz":120},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/c51e08abc1d93145c22c2327987be95323c36b0d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/c51e08abc1d93145c22c2327987be95323c36b0d"}]},"branch":"refs/heads/master"},"1c67715c07a730d7836e7a43f120edc4fcf0526d":{"kind":"TRIVIAL_REBASE","_number":99,"created":"2022-07-12 17:53:24.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/99","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/99","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/99 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/99 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/99 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/99"}}},"commit":{"parents":[{"commit":"5a156b0c2830a6300f844cf3af88788d5e9dd9a3","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/5a156b0c2830a6300f844cf3af88788d5e9dd9a3"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2022-07-12 17:53:06.000000000","tz":120},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/1c67715c07a730d7836e7a43f120edc4fcf0526d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/1c67715c07a730d7836e7a43f120edc4fcf0526d"}]},"branch":"refs/heads/master"},"6fe4a2cb2b33e32bfe949c641bedc427318e0c70":{"kind":"TRIVIAL_REBASE","_number":100,"created":"2022-07-13 14:23:36.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/100","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/100","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/100 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/100 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/100 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/100"}}},"commit":{"parents":[{"commit":"a4893a6b89e70a74d7ae3c7aa1c00125f82f2ce6","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/a4893a6b89e70a74d7ae3c7aa1c00125f82f2ce6"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2022-07-13 14:19:36.000000000","tz":120},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/6fe4a2cb2b33e32bfe949c641bedc427318e0c70"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/6fe4a2cb2b33e32bfe949c641bedc427318e0c70"}]},"branch":"refs/heads/master"},"2cfe03669fa0c7123a02e4cca00f72196162fb91":{"kind":"TRIVIAL_REBASE","_number":101,"created":"2022-07-13 14:58:36.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/101","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/101","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/101 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/101 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/101 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/101"}}},"commit":{"parents":[{"commit":"90e2c7f70ed80bb7fe898a481cff3a095acb55f8","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/90e2c7f70ed80bb7fe898a481cff3a095acb55f8"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2022-07-13 14:58:26.000000000","tz":120},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/2cfe03669fa0c7123a02e4cca00f72196162fb91"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/2cfe03669fa0c7123a02e4cca00f72196162fb91"}]},"branch":"refs/heads/master"},"fd2aa2989b4812fe0c2af6bf9d6a1eae827f5399":{"kind":"TRIVIAL_REBASE","_number":102,"created":"2022-07-13 16:17:41.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/102","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/102","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/102 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/102 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/102 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/102"}}},"commit":{"parents":[{"commit":"e177fda0b74d0eb7790a92223624d3d08f0b59ef","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/e177fda0b74d0eb7790a92223624d3d08f0b59ef"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2022-07-13 16:17:34.000000000","tz":120},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/fd2aa2989b4812fe0c2af6bf9d6a1eae827f5399"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/fd2aa2989b4812fe0c2af6bf9d6a1eae827f5399"}]},"branch":"refs/heads/master"},"c0afb8ef1e42f7a344ee262525892360ea2ca434":{"kind":"TRIVIAL_REBASE","_number":103,"created":"2022-07-13 16:35:30.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/103","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/103","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/103 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/103 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/103 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/103"}}},"commit":{"parents":[{"commit":"5c8d9985561327527f531cab4a1f6286586da12d","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/5c8d9985561327527f531cab4a1f6286586da12d"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2022-07-13 16:35:21.000000000","tz":120},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/c0afb8ef1e42f7a344ee262525892360ea2ca434"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/c0afb8ef1e42f7a344ee262525892360ea2ca434"}]},"branch":"refs/heads/master"},"df56a012a39b0746c85b9269b46b28163fcc495e":{"kind":"TRIVIAL_REBASE","_number":104,"created":"2022-07-13 16:52:09.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/104","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/104","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/104 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/104 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/104 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/104"}}},"commit":{"parents":[{"commit":"fed6b10cf9545443be2b67061e5238d8423928be","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/fed6b10cf9545443be2b67061e5238d8423928be"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2022-07-13 16:51:58.000000000","tz":120},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/df56a012a39b0746c85b9269b46b28163fcc495e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/df56a012a39b0746c85b9269b46b28163fcc495e"}]},"branch":"refs/heads/master"},"b2445fb06ea6b572da360c565498df2d7fc86c05":{"kind":"TRIVIAL_REBASE","_number":105,"created":"2022-07-13 19:29:45.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/105","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/105","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/105 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/105 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/105 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/105"}}},"commit":{"parents":[{"commit":"3a3270ecd8725d0c1f51e36bb19cd81d57986163","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/3a3270ecd8725d0c1f51e36bb19cd81d57986163"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2022-07-13 19:21:16.000000000","tz":120},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/b2445fb06ea6b572da360c565498df2d7fc86c05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/b2445fb06ea6b572da360c565498df2d7fc86c05"}]},"branch":"refs/heads/master"},"a51755627bd203f0858e8ea3c66ee4d7cf624797":{"kind":"TRIVIAL_REBASE","_number":106,"created":"2022-07-21 14:21:34.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/106","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/106","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/106 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/106 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/106 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/106"}}},"commit":{"parents":[{"commit":"3e58f48b26c491f9f3d835c51eadcf34ae0bee3e","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/3e58f48b26c491f9f3d835c51eadcf34ae0bee3e"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2022-07-21 14:21:24.000000000","tz":120},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/a51755627bd203f0858e8ea3c66ee4d7cf624797"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/a51755627bd203f0858e8ea3c66ee4d7cf624797"}]},"branch":"refs/heads/master"},"ea6907e7d87232a41102beba73feeaccfa59943b":{"kind":"TRIVIAL_REBASE","_number":107,"created":"2022-07-21 14:29:12.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/107","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/107","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/107 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/107 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/107 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/107"}}},"commit":{"parents":[{"commit":"0cdfe31a309c792717b679b9a111ef7bb550c130","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/0cdfe31a309c792717b679b9a111ef7bb550c130"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2022-07-21 14:28:58.000000000","tz":120},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/ea6907e7d87232a41102beba73feeaccfa59943b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/ea6907e7d87232a41102beba73feeaccfa59943b"}]},"branch":"refs/heads/master"},"5d512fbb7aaf72e305b4d4d522391dd6c6da7fba":{"kind":"TRIVIAL_REBASE","_number":108,"created":"2022-07-21 17:01:33.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/108","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/108","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/108 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/108 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/108 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/108"}}},"commit":{"parents":[{"commit":"3de76d9a5adbac12cedde88b039ec3cbceaa94b4","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/3de76d9a5adbac12cedde88b039ec3cbceaa94b4"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2022-07-21 16:58:42.000000000","tz":120},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/5d512fbb7aaf72e305b4d4d522391dd6c6da7fba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/5d512fbb7aaf72e305b4d4d522391dd6c6da7fba"}]},"branch":"refs/heads/master"},"be25e63b7936b5fc640807950ea766446405cd21":{"kind":"TRIVIAL_REBASE","_number":109,"created":"2022-07-23 18:01:00.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/109","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/109","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/109 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/109 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/109 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/109"}}},"commit":{"parents":[{"commit":"28c50ee1788f32762c2e0f847fbcf5ab9c350337","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/28c50ee1788f32762c2e0f847fbcf5ab9c350337"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2022-07-23 18:00:46.000000000","tz":120},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/be25e63b7936b5fc640807950ea766446405cd21"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/be25e63b7936b5fc640807950ea766446405cd21"}]},"branch":"refs/heads/master"},"aa8be0654240b66797f6fdc3444d610db15b904d":{"kind":"TRIVIAL_REBASE","_number":110,"created":"2022-07-26 09:12:43.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/110","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/110","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/110 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/110 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/110 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/110"}}},"commit":{"parents":[{"commit":"a5b19d6e9d81885f3a80b52b0ec63f9cd41f049e","subject":"Add proxysql configuration","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/a5b19d6e9d81885f3a80b52b0ec63f9cd41f049e"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2022-07-26 09:12:00.000000000","tz":120},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/aa8be0654240b66797f6fdc3444d610db15b904d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/aa8be0654240b66797f6fdc3444d610db15b904d"}]},"branch":"refs/heads/master"},"a1fa8943b6205233a822c89ccdc6d3e2c79e10ba":{"kind":"TRIVIAL_REBASE","_number":111,"created":"2022-07-26 09:40:11.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/111","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/111","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/111 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/111 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/111 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/111"}}},"commit":{"parents":[{"commit":"197067d7b574f264772aa73948ee549faeac55b4","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/197067d7b574f264772aa73948ee549faeac55b4"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2022-07-26 09:38:56.000000000","tz":120},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/a1fa8943b6205233a822c89ccdc6d3e2c79e10ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/a1fa8943b6205233a822c89ccdc6d3e2c79e10ba"}]},"branch":"refs/heads/master"},"e264659a5703eb24a89bb65d6b0525c426ae4811":{"kind":"TRIVIAL_REBASE","_number":112,"created":"2022-07-26 12:26:43.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/112","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/112","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/112 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/112 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/112 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/112"}}},"commit":{"parents":[{"commit":"55cc3684538a2d16829d38707d46b3a871d894f1","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/55cc3684538a2d16829d38707d46b3a871d894f1"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2022-07-26 12:25:46.000000000","tz":120},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/e264659a5703eb24a89bb65d6b0525c426ae4811"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/e264659a5703eb24a89bb65d6b0525c426ae4811"}]},"branch":"refs/heads/master"},"87fed6a32df55d4cdde70c5228fb17d0f6e76d3e":{"kind":"NO_CHANGE","_number":113,"created":"2022-07-26 12:28:18.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/113","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/113","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/113 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/113 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/113 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/113"}}},"commit":{"parents":[{"commit":"d03277d230c5484a6cc7e04ffc7f26f78f6b5691","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/d03277d230c5484a6cc7e04ffc7f26f78f6b5691"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2022-07-26 12:28:07.000000000","tz":120},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/87fed6a32df55d4cdde70c5228fb17d0f6e76d3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/87fed6a32df55d4cdde70c5228fb17d0f6e76d3e"}]},"branch":"refs/heads/master"},"c14a779156eb957045f829f0294a3f263ed2cfdf":{"kind":"TRIVIAL_REBASE","_number":114,"created":"2022-07-26 12:30:21.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/114","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/114","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/114 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/114 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/114 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/114"}}},"commit":{"parents":[{"commit":"79f0a1f4778f1953152aba45a6969303d78a7151","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/79f0a1f4778f1953152aba45a6969303d78a7151"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2022-07-26 12:30:07.000000000","tz":120},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/c14a779156eb957045f829f0294a3f263ed2cfdf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/c14a779156eb957045f829f0294a3f263ed2cfdf"}]},"branch":"refs/heads/master"},"97315c02d247e18077b5353ec4673cd2f5bc795f":{"kind":"TRIVIAL_REBASE","_number":115,"created":"2022-07-26 13:44:52.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/115","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/115","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/115 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/115 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/115 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/115"}}},"commit":{"parents":[{"commit":"84de04f46b1e4edb7cb728cadf8eec07d5a5dc5b","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/84de04f46b1e4edb7cb728cadf8eec07d5a5dc5b"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2022-07-26 13:43:12.000000000","tz":120},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/97315c02d247e18077b5353ec4673cd2f5bc795f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/97315c02d247e18077b5353ec4673cd2f5bc795f"}]},"branch":"refs/heads/master"},"8884ec654b523761d579c46b761637079c88fe1e":{"kind":"TRIVIAL_REBASE","_number":116,"created":"2022-07-26 13:49:44.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/116","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/116","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/116 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/116 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/116 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/116"}}},"commit":{"parents":[{"commit":"b1cb95d52f274b34c0d356be5aa11215403badc2","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/b1cb95d52f274b34c0d356be5aa11215403badc2"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2022-07-26 13:49:35.000000000","tz":120},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/8884ec654b523761d579c46b761637079c88fe1e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/8884ec654b523761d579c46b761637079c88fe1e"}]},"branch":"refs/heads/master"},"36138c6160af1e4f6144646bf50e363a5c4bf7c3":{"kind":"NO_CHANGE","_number":117,"created":"2022-07-26 15:55:28.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/117","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/117","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/117 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/117 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/117 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/117"}}},"commit":{"parents":[{"commit":"7b6441eabcce15c4831440bb27a868a2ce2d9143","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/7b6441eabcce15c4831440bb27a868a2ce2d9143"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2022-07-26 15:55:09.000000000","tz":120},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/36138c6160af1e4f6144646bf50e363a5c4bf7c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/36138c6160af1e4f6144646bf50e363a5c4bf7c3"}]},"branch":"refs/heads/master"},"36abe5a4886bc6f5eee3c459904c5e2f405b5562":{"kind":"TRIVIAL_REBASE","_number":118,"created":"2022-07-28 14:01:11.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/118","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/118","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/118 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/118 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/118 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/118"}}},"commit":{"parents":[{"commit":"8218a645bd62e8e04dfd1a0723837f8f993db8db","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/8218a645bd62e8e04dfd1a0723837f8f993db8db"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2022-07-28 14:01:00.000000000","tz":120},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/36abe5a4886bc6f5eee3c459904c5e2f405b5562"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/36abe5a4886bc6f5eee3c459904c5e2f405b5562"}]},"branch":"refs/heads/master"},"258de6a9eefa207f5c32521b6f7b96e715f7e840":{"kind":"TRIVIAL_REBASE","_number":119,"created":"2022-07-28 15:09:15.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/119","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/119","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/119 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/119 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/119 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/119"}}},"commit":{"parents":[{"commit":"8d21e1d336cd9a2832ca7df5cedcf34dfa9a096b","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/8d21e1d336cd9a2832ca7df5cedcf34dfa9a096b"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2022-07-28 15:09:03.000000000","tz":120},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/258de6a9eefa207f5c32521b6f7b96e715f7e840"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/258de6a9eefa207f5c32521b6f7b96e715f7e840"}]},"branch":"refs/heads/master"},"e2277abcc4dfa40f3e12f5328bad77b329fff4d0":{"kind":"TRIVIAL_REBASE","_number":120,"created":"2022-07-28 17:27:38.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/120","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/120","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/120 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/120 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/120 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/120"}}},"commit":{"parents":[{"commit":"bd5c93a78207832d535184f9d10c0d07a92d64d2","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/bd5c93a78207832d535184f9d10c0d07a92d64d2"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2022-07-28 17:23:02.000000000","tz":120},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/e2277abcc4dfa40f3e12f5328bad77b329fff4d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/e2277abcc4dfa40f3e12f5328bad77b329fff4d0"}]},"branch":"refs/heads/master"},"8b2100801566ceb116822bf3692198b543f1eb3d":{"kind":"NO_CHANGE","_number":121,"created":"2022-07-28 17:43:28.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/121","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/121","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/121 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/121 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/121 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/121"}}},"commit":{"parents":[{"commit":"d060fc1142afb23c71fcd5eaffb28b6f783bf994","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/d060fc1142afb23c71fcd5eaffb28b6f783bf994"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2022-07-28 17:43:05.000000000","tz":120},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/8b2100801566ceb116822bf3692198b543f1eb3d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/8b2100801566ceb116822bf3692198b543f1eb3d"}]},"branch":"refs/heads/master"},"d409527b9f63f3aee283017ac0f35492696c9879":{"kind":"TRIVIAL_REBASE","_number":122,"created":"2022-07-28 18:04:23.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/122","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/122","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/122 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/122 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/122 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/122"}}},"commit":{"parents":[{"commit":"7a4442a8e2f43e8c7f323d00df47c63d16de90ca","subject":"Edit services roles to support database sharding","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/7a4442a8e2f43e8c7f323d00df47c63d16de90ca"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2022-07-28 18:04:10.000000000","tz":120},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/d409527b9f63f3aee283017ac0f35492696c9879"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/d409527b9f63f3aee283017ac0f35492696c9879"}]},"branch":"refs/heads/master"},"b17b5e2716d5d8c9519a0822ebfaf8c2993154d0":{"kind":"TRIVIAL_REBASE","_number":123,"created":"2022-07-28 18:08:57.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/123","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/123","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/123 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/123 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/123 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/123"}}},"commit":{"parents":[{"commit":"89c6a04b9d89abefa2f197172b93585ecb5fb789","subject":"Add proxysql-config role","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/89c6a04b9d89abefa2f197172b93585ecb5fb789"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2022-07-28 18:07:59.000000000","tz":120},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/b17b5e2716d5d8c9519a0822ebfaf8c2993154d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/b17b5e2716d5d8c9519a0822ebfaf8c2993154d0"}]},"branch":"refs/heads/master"},"01333cf980905192e9ac9aa6864bd65315337012":{"kind":"NO_CHANGE","_number":124,"created":"2022-07-28 18:14:30.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/124","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/124","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/124 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/124 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/124 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/124"}}},"commit":{"parents":[{"commit":"ffa296acb84a984f7ddc9eed40e31d0ce2687ff8","subject":"Add proxysql-config role and wrap it in loadbalancer-config role","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/ffa296acb84a984f7ddc9eed40e31d0ce2687ff8"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2022-07-28 18:14:21.000000000","tz":120},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/01333cf980905192e9ac9aa6864bd65315337012"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/01333cf980905192e9ac9aa6864bd65315337012"}]},"branch":"refs/heads/master"},"122f3f1e865269f7a90e00e71c484c8487c293c3":{"kind":"NO_CHANGE","_number":125,"created":"2022-07-28 20:15:13.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/125","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/125","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/125 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/125 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/125 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/125"}}},"commit":{"parents":[{"commit":"8741f6729ccb4412d578d47b9cc2a2b5ed4e02f7","subject":"Add proxysql-config role and wrap it in loadbalancer-config role","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/8741f6729ccb4412d578d47b9cc2a2b5ed4e02f7"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2022-07-28 20:14:40.000000000","tz":120},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/122f3f1e865269f7a90e00e71c484c8487c293c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/122f3f1e865269f7a90e00e71c484c8487c293c3"}]},"branch":"refs/heads/master"},"ccc5b604be5388d6a5d76bd3ebfa2c051ea0b593":{"kind":"TRIVIAL_REBASE","_number":126,"created":"2022-07-29 13:07:34.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/126","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/126","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/126 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/126 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/126 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/126"}}},"commit":{"parents":[{"commit":"27a56a711a9702f543f4e81479caec5120cec4c8","subject":"Add proxysql-config role and wrap it in loadbalancer-config role","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/27a56a711a9702f543f4e81479caec5120cec4c8"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2022-07-29 13:06:46.000000000","tz":120},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/ccc5b604be5388d6a5d76bd3ebfa2c051ea0b593"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/ccc5b604be5388d6a5d76bd3ebfa2c051ea0b593"}]},"branch":"refs/heads/master"},"4084463ccebccb270d3186e2a22d69a1c2200fb2":{"kind":"TRIVIAL_REBASE","_number":127,"created":"2022-07-30 14:24:37.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/127","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/127","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/127 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/127 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/127 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/127"}}},"commit":{"parents":[{"commit":"f5783be125e6468f1f8d8a184af698f543e7045a","subject":"Add proxysql-config role and wrap it in loadbalancer-config role","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/f5783be125e6468f1f8d8a184af698f543e7045a"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2022-07-30 14:22:05.000000000","tz":120},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/4084463ccebccb270d3186e2a22d69a1c2200fb2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/4084463ccebccb270d3186e2a22d69a1c2200fb2"}]},"branch":"refs/heads/master"},"77dc3b43f6b9a2489133ad76451167a1a6c6e767":{"kind":"TRIVIAL_REBASE","_number":128,"created":"2022-08-02 11:34:47.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/128","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/128","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/128 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/128 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/128 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/128"}}},"commit":{"parents":[{"commit":"fa231bc166f2ca693d1fcb7f8862208a609ea811","subject":"Add proxysql-config role and wrap it in loadbalancer-config role","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/fa231bc166f2ca693d1fcb7f8862208a609ea811"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2022-08-01 16:30:19.000000000","tz":120},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/77dc3b43f6b9a2489133ad76451167a1a6c6e767"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/77dc3b43f6b9a2489133ad76451167a1a6c6e767"}]},"branch":"refs/heads/master"},"c4aafa4acc655b1bfc0d4adc9ebe83341dba1d13":{"kind":"TRIVIAL_REBASE","_number":129,"created":"2022-08-02 15:49:06.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/129","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/129","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/129 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/129 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/129 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/129"}}},"commit":{"parents":[{"commit":"a325453decb9f8f201d5c9bd6b95cf47cf7041e1","subject":"Add proxysql-config role and wrap it in loadbalancer-config role","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/a325453decb9f8f201d5c9bd6b95cf47cf7041e1"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2022-08-02 15:48:56.000000000","tz":120},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/c4aafa4acc655b1bfc0d4adc9ebe83341dba1d13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/c4aafa4acc655b1bfc0d4adc9ebe83341dba1d13"}]},"branch":"refs/heads/master"},"1b3fcfed64c9dffb5693d2930ac1f0a696b4aa34":{"kind":"TRIVIAL_REBASE","_number":130,"created":"2022-08-09 10:17:34.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/21/770621/130","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/770621/130","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/130 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/130 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/130 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/770621/130"}}},"commit":{"parents":[{"commit":"ae7f4ddc3ee2fa8a697fc1c50f1721058f30687a","subject":"Add proxysql-config role and wrap it in loadbalancer-config role","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/ae7f4ddc3ee2fa8a697fc1c50f1721058f30687a"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-01-13 13:52:32.000000000","tz":0},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2022-08-09 10:16:43.000000000","tz":120},"subject":"[CI] Test ProxySQL with shards in the nova cells scenario","message":"[CI] Test ProxySQL with shards in the nova cells scenario\n\nThere are two shards:\nOne 2-node (to test the clustering), one 1-node.\n\nChange-Id: If3a60ad4cc39d6ad0cd72a934f5f7497cd44021b\nCo-Authored-By: Radosław Piliszek \u003cradoslaw.piliszek@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/1b3fcfed64c9dffb5693d2930ac1f0a696b4aa34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/1b3fcfed64c9dffb5693d2930ac1f0a696b4aa34"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"}}]}],"submit_requirements":[]}
