)]}'
{"id":"openstack%2Fkolla-ansible~791237","triplet_id":"openstack%2Fkolla-ansible~master~I30988176c8d5a59cf61527af1a43c5d33ba6268d","project":"openstack/kolla-ansible","branch":"master","hashtags":[],"change_id":"I30988176c8d5a59cf61527af1a43c5d33ba6268d","subject":"Trivial if conditional fix in keystone.json","status":"MERGED","created":"2021-05-13 13:48:13.000000000","updated":"2021-05-31 17:12:23.000000000","submitted":"2021-05-31 14:30:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"791237","meta_rev_id":"e9f3cb82d6f103ccdfb2668541964f5a8d3c150f","_number":791237,"virtual_id_number":791237,"owner":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-31 14:30:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2021-05-13 14:05:07.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2021-05-31 13:38:26.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2021-05-31 13:38:26.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-13 14:05:07.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-05-13 14:07:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2021-05-13 14:56:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-31 13:38:26.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"}],"messages":[{"id":"7ca07ba18148ba84df4b2cae98c21ea172777bc4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"date":"2021-05-13 13:48:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"232b8c0a569337ea9bab3bb1a8e0af1a936159fc","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-05-13 14:05:07.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4f9616b7edbc93a4ab27a0af6679b860fff79924","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-13 14:07:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/8c3091f3eaf44d6ca9e8258842acd8a2 : FAILURE in 13m 18s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/16a3d8d22ae94e10a0513b09d85412ae : FAILURE in 11m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f8e1fd3979b024f17b3ded23ae97a457f69c4720","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-13 14:56:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/358dd071bc034c818d7792ae219e5754 : SUCCESS in 6m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06a573ee09e0470ab27e533dcb0d0d1d : SUCCESS in 5m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/93e55c2f51ce4c9aa84b24eefa2fd353 : SUCCESS in 6m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/93f5f059ac4c47958ae376b20d37117c : SUCCESS in 6m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d3a4a1fd37df41db981418912a7576c9 : SUCCESS in 5m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3669acc80baf480797ebecbfe0e49bfd : SUCCESS in 8m 17s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/a82d1ddb4c2a4de3842442f1ef47c554 : SUCCESS in 8m 40s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/38df44aae26e4ec0af29004e8c1ac95f : SUCCESS in 56m 00s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/5d0b140a9be94a4c9bcd2753a598fd5a : FAILURE in 10m 56s (non-voting)\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7f5b037d8fc0470091b8761854d2d984 : SUCCESS in 47m 52s\n- kolla-ansible-centos8s-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/4bc7903249b844af9e2d1935cee41e05 : SUCCESS in 48m 20s (non-voting)\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/9339004c56074d4fb47a2d4ce8253147 : SUCCESS in 49m 57s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/94223f1acf0b4ab4afa837e421ac6c52 : SUCCESS in 55m 12s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/12845e35dcdf4f3da6a4bc034112753c : FAILURE in 53m 51s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/356c2869fa5f4d92b67203ed7d0a1f62 : SUCCESS in 46m 11s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/29a0aaf690714ee18b03e175b1b79b0b : SUCCESS in 52m 23s (non-voting)\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/d4d3ff7d201e463ca0d099aaccf91156 : SUCCESS in 33m 06s (non-voting)\n- kolla-ansible-centos8s-source-cells https://zuul.opendev.org/t/openstack/build/c8ae90361c384c628a05917b427b54dc : SUCCESS in 55m 56s (non-voting)\n- kolla-ansible-centos8s-source-cephadm https://zuul.opendev.org/t/openstack/build/9f2faa3318c242299a9b9d35f7434be9 : SUCCESS in 49m 50s (non-voting)\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/a457b38b433048a58d45f51d07e4334d : SUCCESS in 53m 11s (non-voting)\n- kolla-ansible-centos8s-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/045cd66573824e36b459c031f725177b : SUCCESS in 57m 46s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/bc275178b20146239833107701526827 : SUCCESS in 1h 01m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"907e59d2d00b1db839da8f3533f1a859c5f6c0ee","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-05-31 13:38:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ae483ad84add1c4af93a681c1ae638652f072bbc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-31 13:38:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a7ed29ca94f63ec647c3a17096d6b5028db90f56","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-05-31 13:53:29.000000000","message":"Patch Set 1:\n\nI wonder though how it used to work since the sources are obviously missing and it\u0027s not optional","accounts_in_message":[],"_revision_number":1},{"id":"5dee7d22b6526a2a4fe9bc59e118be124ac9fc14","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-31 14:30:37.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70c6f2b23647410db58da91dd63bfd29 : SUCCESS in 5m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bca2631241d0416caafd20e2a0debc05 : SUCCESS in 5m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cd4cfde95bde49d0a7cba3d5a9838590 : SUCCESS in 5m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d21f0510f1a43318afcd64c99ce5cbb : SUCCESS in 5m 48s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/1951263a01984160ad97dc6c470830a0 : SUCCESS in 6m 38s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/851d104efd244780a71296d9dc9a66cd : SUCCESS in 50m 57s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/834a4139b94d47f2abe44d95bae0d5f4 : SUCCESS in 48m 55s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/c383c8b57dfc49309cab0977476fca78 : SUCCESS in 37m 30s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/018ae6d3653b40a98d03cf06cb8942dd : SUCCESS in 49m 29s","accounts_in_message":[],"_revision_number":1},{"id":"d12f69cf427bb050f458d19d0e429533498f67d1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-31 14:30:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"0fd0f02e7e4a216cc00e0ac14f090c88ccc1465f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-31 14:32:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b23aefe8b0e4468a8b6467064f9fa78 : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":1},{"id":"e9f3cb82d6f103ccdfb2668541964f5a8d3c150f","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-05-31 17:12:23.000000000","message":"Patch Set 1:\n\n\u003e Patch Set 1:\n\u003e \n\u003e I wonder though how it used to work since the sources are obviously missing and it\u0027s not optional\n\nah, we have it \"booleised\" by default","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d43642ac370cdf58af977790a3faa7c174e17849","revisions":{"d43642ac370cdf58af977790a3faa7c174e17849":{"kind":"REWORK","_number":1,"created":"2021-05-13 13:48:13.000000000","uploader":{"_account_id":27339,"name":"Michal Arbet","email":"michal.arbet@ultimum.io","username":"michalarbet"},"ref":"refs/changes/37/791237/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/37/791237/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/37/791237/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/37/791237/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/37/791237/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/37/791237/1"}}},"commit":{"parents":[{"commit":"9df0098669af18e34bed9a55e7adab72ef2f23c5","subject":"Merge \"Disable Alertmanager\u0027s peer gossip in non-HA deployments\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/9df0098669af18e34bed9a55e7adab72ef2f23c5"}]}],"author":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-05-13 13:47:00.000000000","tz":120},"committer":{"name":"Michal Arbet","email":"michal.arbet@ultimum.io","date":"2021-05-13 13:47:00.000000000","tz":120},"subject":"Trivial if conditional fix in keystone.json","message":"Trivial if conditional fix in keystone.json\n\nChange-Id: I30988176c8d5a59cf61527af1a43c5d33ba6268d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/d43642ac370cdf58af977790a3faa7c174e17849"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/d43642ac370cdf58af977790a3faa7c174e17849"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
