)]}'
{"id":"openstack%2Fkolla-ansible~791321","triplet_id":"openstack%2Fkolla-ansible~master~If23a6b1273c2639d1296becc9d222546d52f63ac","project":"openstack/kolla-ansible","branch":"master","topic":"bug/1927880","hashtags":[],"change_id":"If23a6b1273c2639d1296becc9d222546d52f63ac","subject":"fluentd: Fix check for external elasticsearch","status":"MERGED","created":"2021-05-13 21:30:57.000000000","updated":"2021-08-11 14:35:18.000000000","submitted":"2021-08-11 14:32:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"791321-bug/1927880","meta_rev_id":"c1b3318dafe47dc8f2a2afd4b534426a27b262c6","_number":791321,"virtual_id_number":791321,"owner":{"_account_id":33469,"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","username":"sshambar"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-08-11 14:32:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":33469,"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","username":"sshambar"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2021-08-11 14:35:18.000000000","post_submit":true,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2021-08-11 10:25:27.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":33469,"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","username":"sshambar"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":1,"date":"2021-08-11 10:25:27.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":33469,"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","username":"sshambar"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":33469,"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","username":"sshambar"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-08-11 14:35:18.000000000","post_submit":true,"_account_id":33469,"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","username":"sshambar"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":33469,"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","username":"sshambar"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"_account_id":33469,"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","username":"sshambar"}],"CC":[{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-13 21:35:26.000000000","updated_by":{"_account_id":33469,"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","username":"sshambar"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2021-05-24 07:26:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-22 11:28:57.000000000","updated_by":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"reviewer":{"_account_id":32657,"name":"Piotr Parczewski","email":"piotr@stackhpc.com","username":"piotrp"},"state":"CC"},{"updated":"2021-08-11 10:25:27.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"}],"messages":[{"id":"debedae7fadf1b726c9fc7ef433a30c3be1ee381","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":33469,"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","username":"sshambar"},"date":"2021-05-13 21:30:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b9a87a255a584050653761859dc708bb616b022c","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":33469,"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","username":"sshambar"},"date":"2021-05-13 21:32:50.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":1},{"id":"ba2597b764eeff49c548545c83c8c79c07aaeabc","author":{"_account_id":33469,"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","username":"sshambar"},"date":"2021-05-13 21:35:26.000000000","message":"Patch Set 1: Backport-Candidate+1\n\nHere\u0027s the proposed fix for bug #1927880","accounts_in_message":[],"_revision_number":1},{"id":"a9536c9ad31f7db59a9325f022af016c0f2ce529","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-05-17 09:19:21.000000000","message":"Patch Set 1:\n\nThanks. Could you add a release note?\n\nhttps://docs.openstack.org/kolla-ansible/latest/contributor/release-notes.html","accounts_in_message":[],"_revision_number":1},{"id":"b9bf5e9ae437f108e8eb3c3acf1db662acbfb50a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33469,"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","username":"sshambar"},"date":"2021-05-24 06:00:24.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"21a50547538f3e86c58b61e4d1f3b4006dff48e7","author":{"_account_id":33469,"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","username":"sshambar"},"date":"2021-05-24 06:01:45.000000000","message":"Patch Set 2: Backport-Candidate+1\n\nAdded release note","accounts_in_message":[],"_revision_number":2},{"id":"c381baf15f23899a14e8553ff147f840b3e70b85","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-24 07:26:59.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7540376b67264caeada8d4d289a37839 : SUCCESS in 7m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b756651b98a845c998a1743683af67d8 : SUCCESS in 8m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6732c1b32f5d435b8858119de078446d : SUCCESS in 11m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5713b65a2c9748eba4bce7ef6b0674c8 : SUCCESS in 7m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bbb2d9e6db5b457faab89ad4fe6c988d : SUCCESS in 7m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cdd9802772f84c08a510316754491005 : SUCCESS in 11m 55s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/0f8e8e055af949828c7a9caadc2cf8af : FAILURE in 7m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/667e3b29adae442aa01bcf353fdc0b92 : SUCCESS in 6m 02s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/cb4e8058c0d1492cbd8f6b7cc96237e2 : SUCCESS in 1h 00m 50s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/b546a1f3396f497baee4bfbbd0cc6a84 : SUCCESS in 45m 34s (non-voting)\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/a660f5653c054a67ac6e84a067652bf9 : SUCCESS in 48m 20s\n- kolla-ansible-centos8s-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/237b55648d914ea6a5228dab216e9583 : SUCCESS in 55m 48s (non-voting)\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/6d6ef31bdbdb448d994ca5e916785dca : SUCCESS in 55m 22s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/0504774cb2984ad99a79d44acefd747f : SUCCESS in 1h 09m 50s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/a88280752e8c4e34800f5419233ee770 : FAILURE in 53m 48s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/556cf3046698468086218d28a91fc3ae : SUCCESS in 46m 33s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/fba0bfd92dd14b648444dd892ae0343d : SUCCESS in 39m 39s (non-voting)\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/f77285eea21f4c4f9942ce19ff27ac66 : SUCCESS in 39m 32s (non-voting)\n- kolla-ansible-centos8s-source-cells https://zuul.opendev.org/t/openstack/build/900fb93eba67434e8b6c3337a40aef3a : SUCCESS in 56m 10s (non-voting)\n- kolla-ansible-centos8s-source-prometheus-efk https://zuul.opendev.org/t/openstack/build/0758278745ac4daea326cdec94cb2b06 : SUCCESS in 25m 48s (non-voting)\n- kolla-ansible-ubuntu-source-prometheus-efk https://zuul.opendev.org/t/openstack/build/df7af70c3aea4190991f411b3177a5aa : SUCCESS in 22m 00s (non-voting)\n- kolla-ansible-centos8s-source-cephadm https://zuul.opendev.org/t/openstack/build/57b022bd70da438a833d59769eda8870 : SUCCESS in 1h 02m 30s (non-voting)\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/4b2d5fc2ab4a4d23b1d57b1664bae641 : SUCCESS in 59m 29s (non-voting)\n- kolla-ansible-centos8s-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/8c9ee50c61ac4388bdf336b436ffef87 : SUCCESS in 1h 10m 06s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/f797ad911e994915bdc404cb572a524c : FAILURE in 55m 06s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0a55d7b40c353c3d329e235ce6d72dd476cbace4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-24 07:41:54.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/0d0888fea400462eb25070adb817bcf6 : SUCCESS in 1h 15m 28s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/610f09e6c86d43f6983c101c9ee842a0 : FAILURE in 1h 35m 43s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8e9f485ef66c091df0e99c6e2092838fb82fdd97","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-05-24 09:09:27.000000000","message":"Patch Set 2: Code-Review+2\n\nThanks!\n\n(CI broken, we\u0027ll recheck when fixed)","accounts_in_message":[],"_revision_number":2},{"id":"ca02a99caa3c1fdcd56426db1e734a056f990dc1","tag":"autogenerated:gerrit:setAssignee","author":{"_account_id":33469,"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","username":"sshambar"},"date":"2021-05-24 18:08:35.000000000","message":"Assignee added: Scott Shambarger \u003cscott-openstack@shambarger.net\u003e","accounts_in_message":[],"_revision_number":2},{"id":"c186d3c7ad57ec552ba826c418a5b68d6da58e69","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33469,"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","username":"sshambar"},"date":"2021-05-24 19:28:06.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"15e514a2145eb104ec68ea339b9869ab613a1c5a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-24 20:45:07.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/957e87aa87174d2d84ba3b5308cf665c : SUCCESS in 1h 02m 27s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/f0a03b867fc444ce98f086f5d0ddbbb6 : SUCCESS in 1h 12m 48s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c3fc9224910a9a2e40cdb4365d92172baedb2885","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-24 21:12:24.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/07c19eb53d344de59a61a506859a939f : SUCCESS in 7m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da92c04c573b4df1bb36b271bf441910 : SUCCESS in 6m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/15c4577d79d24cb78efc591228c690e5 : SUCCESS in 5m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9db8bc3036c54bf384b44a728428c26e : SUCCESS in 6m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2db9b198ebba4d62a1ca126a9b1f974d : SUCCESS in 5m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d43782010b64065abf1c4374607cd58 : SUCCESS in 7m 06s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/19fa2111a0414e3f96c5291586164698 : SUCCESS in 8m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3d65e5dc7fb64a4d91574802a4b8f7df : SUCCESS in 5m 47s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/8b037a5867b849cea3d0a65c9d54a395 : RETRY_LIMIT in 2m 35s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/76da45b68902490280db8b8988614be0 : SUCCESS in 41m 04s (non-voting)\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/9c23fa299f5e4de8b31b270dc725d121 : SUCCESS in 44m 52s\n- kolla-ansible-centos8s-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/963107596cb14ee1bfab2e8d730b9ba9 : RETRY_LIMIT in 3m 56s (non-voting)\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/d41b20bc2aa249b183e57e635d33337a : SUCCESS in 1h 14m 28s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/4c94da22a26d46b0b7d09d6fb21a56d3 : RETRY_LIMIT in 2m 25s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/1cb87d149bdf4b96ac7418076720924b : SUCCESS in 54m 17s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/01e3ebcd00334f2fbab96e7dd017efe1 : SUCCESS in 53m 31s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/0d3db698d1af4b4d8f49dafa352fa2a9 : RETRY_LIMIT in 2m 19s (non-voting)\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/8346952fecf64c1c9a0ebc2d7f8abff1 : SUCCESS in 42m 42s (non-voting)\n- kolla-ansible-centos8s-source-cells https://zuul.opendev.org/t/openstack/build/ddeb56b92da144d98249c5922fd2b67a : RETRY_LIMIT in 4m 38s (non-voting)\n- kolla-ansible-centos8s-source-prometheus-efk https://zuul.opendev.org/t/openstack/build/98a56a034b6e471c84b58ffaba49b487 : RETRY_LIMIT in 2m 33s (non-voting)\n- kolla-ansible-ubuntu-source-prometheus-efk https://zuul.opendev.org/t/openstack/build/6e99341c914c49e587b167e9cf70d9c2 : SUCCESS in 28m 02s (non-voting)\n- kolla-ansible-centos8s-source-cephadm https://zuul.opendev.org/t/openstack/build/f2bd9e235eb5462cad2959157c0859db : RETRY_LIMIT in 4m 31s (non-voting)\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/b7861b0d373a4affac754d9e7e093c91 : FAILURE in 1h 01m 42s (non-voting)\n- kolla-ansible-centos8s-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/38c742094c114c50add094d23569949f : RETRY_LIMIT in 4m 04s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/12b312d8da4d416188a1dbef4099f904 : SUCCESS in 1h 37m 19s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"082cf7288a16e50bb136e40c64aca4fa9b507b40","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-05-25 08:37:39.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"3a2adf9c5c5266b22841cb03703a042a247bd543","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-25 10:22:27.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/d94255d9bbdc42d58312d51b7e5c316a : SUCCESS in 1h 26m 23s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/dd26a0fbeb0d4e719bad54d9566a53e6 : SUCCESS in 1h 42m 24s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"4f6276606d7c9a005c87814fd2e07dc401eb933f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-25 10:24:19.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a410e187e71e4571a21b20d190838948 : SUCCESS in 5m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e34236dd66640bdadc85f4cc86b2a7f : SUCCESS in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/25980bb3d56c4d9ba16ced071c141bfc : SUCCESS in 5m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d6d8a0e2ba19443f934a13bcf40f6e94 : SUCCESS in 5m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0594cebdcdf043c3b2d0fa1526563ab0 : SUCCESS in 6m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff94584dd13f429e8a4568015a242410 : SUCCESS in 7m 42s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/69fa5d3bfe0245feb62dcf8aa02aa8f8 : SUCCESS in 7m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/47470e86433c450884f3177161bf1142 : SUCCESS in 5m 03s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/1ec1201ec3ac43c0afd576b7f902b295 : FAILURE in 39m 15s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/cbf203aa9e774aef81fc786f94c2350e : SUCCESS in 40m 57s (non-voting)\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/727a3324c50641da9b9ea9bc964f4fff : SUCCESS in 42m 24s\n- kolla-ansible-centos8s-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/671c5250e1e747aa9627d014dd24d198 : FAILURE in 46m 09s (non-voting)\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/9a4a1f2739bc4dfeb0d60c853699b6bf : SUCCESS in 55m 43s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/2f4bbe13f47d403fab4eeef31b4c2954 : FAILURE in 49m 42s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/29d9ce03246642e5bcd4782b22c40bd7 : SUCCESS in 59m 50s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/7accd281af1d4dfeaa40f0a3424bc897 : SUCCESS in 55m 06s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/42518fc13ee54f468a74557c055f8435 : FAILURE in 32m 10s (non-voting)\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/9bc184b6fd7e4abab68b956549ee60f7 : SUCCESS in 41m 00s (non-voting)\n- kolla-ansible-centos8s-source-cells https://zuul.opendev.org/t/openstack/build/2f05408b8e3f4884b9f4a7ea151453af : FAILURE in 47m 04s (non-voting)\n- kolla-ansible-centos8s-source-prometheus-efk https://zuul.opendev.org/t/openstack/build/516de71f576041e29ae0b235e636c14d : SUCCESS in 1h 45m 41s (non-voting)\n- kolla-ansible-ubuntu-source-prometheus-efk https://zuul.opendev.org/t/openstack/build/e652563e5d3e48dba4e828deb10f0431 : SUCCESS in 19m 43s (non-voting)\n- kolla-ansible-centos8s-source-cephadm https://zuul.opendev.org/t/openstack/build/4a86d04d1e284435b0e2b6110067ffa6 : FAILURE in 1h 00m 57s (non-voting)\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/d67f1fe8791648b98d565daca2ec569a : SUCCESS in 54m 13s (non-voting)\n- kolla-ansible-centos8s-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/d58aad3ed533461cb9e49723f0c0dfb1 : FAILURE in 1h 24m 03s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/2cf408660f8a41859e20e0e60d4fbe85 : SUCCESS in 1h 14m 00s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"b95cb503c0b4822ab0432a66a5e4932fd2fd70af","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33469,"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","username":"sshambar"},"date":"2021-05-27 17:26:04.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"207d0ad5366e617c76efebc7087ef0268e876855","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 18:57:52.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1b99d660ea13414689151e7c29a71a7d : SUCCESS in 6m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42fdebfef51245b6a936796aaa1a1a85 : SUCCESS in 6m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b4d5fa0970114e149ee0259ce1c5b182 : SUCCESS in 6m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1dd22ab8b8254fb9b902f3c0c54f9ed3 : SUCCESS in 6m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/20ae54a78ccb4b22af686547f4cfc63a : SUCCESS in 6m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c1fcad3612442268afc23cc37f0c393 : SUCCESS in 8m 51s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/3efb202c389d4d39b463ed99719be09c : SUCCESS in 8m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/da3fd856b8b64debadb58712a9400140 : SUCCESS in 4m 57s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/3c57a28cae9e4e42b6c08493a04d2a93 : SUCCESS in 51m 53s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/dfe1874f4dcf47c29bb339c564206c29 : SUCCESS in 45m 42s (non-voting)\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/894716e07a43492a95b2546c9fa31cdc : SUCCESS in 53m 57s\n- kolla-ansible-centos8s-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/885b40dc55c7430e956431f3ac5c6cae : SUCCESS in 58m 40s (non-voting)\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/d33a9061780e42e0b3147158c852a18c : SUCCESS in 55m 57s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/8e26d1ec2d544f75b09f719936d42863 : SUCCESS in 1h 01m 01s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/16e53a9feba2495fb91edfb79e32a477 : SUCCESS in 47m 54s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/6ab6870fc21447cfa97fd9f17ed84339 : SUCCESS in 54m 09s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/94f912532e38485fbac74fc1dcfb8298 : SUCCESS in 57m 59s (non-voting)\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/6e6f6074f85c4c03a5fdff706690dd95 : SUCCESS in 39m 34s (non-voting)\n- kolla-ansible-centos8s-source-prometheus-efk https://zuul.opendev.org/t/openstack/build/4d9f9336bc0e434188ffe80d9961a136 : SUCCESS in 25m 13s (non-voting)\n- kolla-ansible-ubuntu-source-prometheus-efk https://zuul.opendev.org/t/openstack/build/01f6dc3600544ff18dc5bd1c04d03d87 : SUCCESS in 22m 29s (non-voting)\n- kolla-ansible-centos8s-source-cephadm https://zuul.opendev.org/t/openstack/build/fded0e8cb8f1477ab54a65446b790be8 : SUCCESS in 1h 08m 09s (non-voting)\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/e0c0edb22a564913b88719b3868fda01 : SUCCESS in 1h 18m 09s (non-voting)\n- kolla-ansible-centos8s-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/7b99ab47a06149e187a58f2f726adfac : SUCCESS in 1h 25m 35s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/b053b09dbad4422bb6c043a88e061dab : SUCCESS in 1h 11m 48s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"3a0d2c2089c3a1613228d1d94431b7ba5ddd2fc6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 19:00:53.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/f73266d099834fa091dc1da07fe7633e : SUCCESS in 1h 13m 58s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/754eacc3041b4ce28ba5f8b512f22321 : SUCCESS in 1h 33m 14s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"d5b2e17ab03ca61b14c578fda20bda29c7d4a240","author":{"_account_id":33469,"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","username":"sshambar"},"date":"2021-07-11 19:59:50.000000000","message":"Patch Set 4:\n\nAre you waiting on me for anything on this review?  Not sure reviews have to wait for merging, but this seems like a long time...","accounts_in_message":[],"_revision_number":4},{"id":"7d658256a792aa981f94e6f9e1c38b663c0c1b2e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2021-07-22 09:03:57.000000000","message":"Patch Set 4:\n\n\u003e Patch Set 4:\n\u003e \n\u003e Are you waiting on me for anything on this review?  Not sure reviews have to wait for merging, but this seems like a long time...\n\nWe require 2 +2s to merge. Feel free to ping someone on IRC, #openstack-kolla on OFTC.","accounts_in_message":[],"_revision_number":4},{"id":"10149adb6bcd2e1b0f7d5e33957fd06783b25f0b","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-08-11 10:25:27.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n\u003e Patch Set 4:\n\u003e \n\u003e Are you waiting on me for anything on this review?  Not sure reviews have to wait for merging, but this seems like a long time...\n\nApologies, Scott. High volume of changes in certain time periods makes it hard not to lose the track of what is mergable.","accounts_in_message":[],"_revision_number":4},{"id":"1a497abc329eae8cadf1fbbd95c4c6c954d40038","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-11 10:25:52.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"405d9b1f2a283a331245f454e40235a89c781945","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-11 11:43:23.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2bdf4cf65be40f385e1fb098695c049 : FAILURE in 4m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/01610624c027413bb2feb5774b6dc317 : SUCCESS in 6m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/038e475801d54ca7b28d388175f4d34f : FAILURE in 4m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2df2f11ad56f4591af7277af766b58e7 : SUCCESS in 7m 01s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/a813db59e44640de8283960d941d6e49 : SUCCESS in 7m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/abb1b375d0cf44b8a86c5431a55fa014 : SUCCESS in 5m 09s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/ea963f88834a409eb664f11c470f45c3 : SUCCESS in 55m 01s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/58a017cff9ce4d289c23a83c4a8926b0 : SUCCESS in 55m 51s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/7372fc0c40c4424ca212f65f95d55676 : SUCCESS in 1h 11m 35s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/b66bb7b42af948a2a8d256bd5cb8e0cd : SUCCESS in 52m 16s","accounts_in_message":[],"_revision_number":4},{"id":"3b149691b44cd2c8d39b6bb154d806b41fbfda6c","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2021-08-11 11:45:48.000000000","message":"Patch Set 4:\n\nrecheck\n\npypi mirror failure","accounts_in_message":[],"_revision_number":4},{"id":"4171408c6dbfd016862836ac820cc2246a2e2712","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-11 13:11:33.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/fb4a35802f204e588fcd3206ee0ba9d1 : SUCCESS in 1h 18m 57s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/2dd67c9fa6fe44cd8723567b03fd4445 : FAILURE in 55m 43s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"612eb9022ce5b6992bdeb6d23fd6fad2cf1529c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-11 13:13:35.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c6f47779d7774bb89936a27cd5a611c1 : SUCCESS in 7m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60e071e2da1249bd8d3810486cff44fc : SUCCESS in 5m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/49302ba2648b41e687c11ee2d8734bf8 : SUCCESS in 5m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ae220d8df5094c3f889b36e87f58500b : SUCCESS in 6m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7872607d212d40d1983dfe7f2c19b86a : SUCCESS in 6m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9179007abeba4bc89f590507d8984207 : SUCCESS in 7m 14s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/2821c7910fe248419981085c3ad1b0cc : SUCCESS in 8m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/045e79334bd54a8bb2dc2760587c0318 : SUCCESS in 5m 02s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/24987ee2068f49048611aa6a176b2268 : SUCCESS in 55m 30s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/1d10a4307876416398c8173046a30762 : SUCCESS in 54m 52s (non-voting)\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/06efe375b7ab49539baf4e5a97fd6e10 : SUCCESS in 45m 22s\n- kolla-ansible-centos8s-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/0091de9ad0454cd3b5e4ab4de4a5a3cd : SUCCESS in 1h 17m 36s (non-voting)\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/280a915c1a3546eab3c928fb84b2a5bd : SUCCESS in 56m 01s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/982dfb21a49d4d0b8fe8f27445cf3188 : SUCCESS in 1h 03m 04s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/ac456b48d4374f3aa7d0a73dc13e7263 : FAILURE in 36m 38s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/ef493cb725e449629d729a69fb50b4ce : SUCCESS in 45m 48s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/734dc14adde443ca84753d309f897a2e : SUCCESS in 54m 10s (non-voting)\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/402d7cc68a974b32a15a93de9d4eae0f : SUCCESS in 38m 32s (non-voting)\n- kolla-ansible-centos8s-source-prometheus-efk https://zuul.opendev.org/t/openstack/build/3f77382e077e4596a2630c3032738410 : SUCCESS in 27m 27s (non-voting)\n- kolla-ansible-ubuntu-source-prometheus-efk https://zuul.opendev.org/t/openstack/build/ad115d634ca444f995cb54aebdb1e993 : SUCCESS in 23m 48s (non-voting)\n- kolla-ansible-centos8s-source-cephadm https://zuul.opendev.org/t/openstack/build/9b22ab1e39354a7a901f078aaa51e2ea : FAILURE in 46m 31s (non-voting)\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/1d1ab4e8a75946c7916e59a84b69fbde : SUCCESS in 1h 06m 10s (non-voting)\n- kolla-ansible-centos8s-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/86ee745fddf0435ab6a9567f61dd62fe : SUCCESS in 1h 24m 41s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/90294b0ea79a476c9b27c4c1afcc09f1 : SUCCESS in 1h 05m 35s (non-voting)\n- kolla-ansible-centos8s-hashi-vault https://zuul.opendev.org/t/openstack/build/8e3b6cb2d02a40e5bb0d2d6b623de39d : SUCCESS in 3m 50s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"4741abdff5f8e0c31d70e7f5096e2cb9a42f14c8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-11 13:14:09.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"0be6bbc56145ac3cd8ae73c54aaae2fca020b536","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-11 14:32:59.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba640765148c4a12bd7799baf855b68e : SUCCESS in 7m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/79bdd2c9d7a1427da02ab0f58dcae73a : SUCCESS in 5m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ea353a7981f5475c8da6b9ecf24bb30a : SUCCESS in 6m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6685f8a477da4c55a78d0d2abdd0c970 : SUCCESS in 6m 24s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/6efdee21507d4f41ac6c31ebeb77ad35 : SUCCESS in 7m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f324314fe68241138cc48149175f7784 : SUCCESS in 4m 29s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/dc9c050520c147c29fe6451707936a6d : SUCCESS in 53m 34s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/78831b1f3fbc4e589979b337b53d9452 : SUCCESS in 56m 43s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/46f20c5a7dcf4b7a8a239c1b60f76f2d : SUCCESS in 1h 08m 35s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/7a7f32d4bad34b7c9be90a0232920676 : SUCCESS in 54m 27s","accounts_in_message":[],"_revision_number":4},{"id":"4383dcb6a89ad4a575b530ae07c615b87db337bf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-11 14:32:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"fb2a780dfc02f5fc68d8ee7067f0adb3243f40e0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-11 14:35:18.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b79d08c2d083452d87fac848341fb9db : SUCCESS in 1m 18s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/497da4dcc980449a972231dc7f70dc39 : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"283b0dec639c032945f5d1d87a66f0c7541c0562","revisions":{"c1163efac44d2d9dde9affc39b88402b543adcad":{"kind":"REWORK","_number":1,"created":"2021-05-13 21:30:57.000000000","uploader":{"_account_id":33469,"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","username":"sshambar"},"ref":"refs/changes/21/791321/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/791321/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/791321/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/791321/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/791321/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/791321/1"}}},"commit":{"parents":[{"commit":"9df0098669af18e34bed9a55e7adab72ef2f23c5","subject":"Merge \"Disable Alertmanager\u0027s peer gossip in non-HA deployments\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/9df0098669af18e34bed9a55e7adab72ef2f23c5"}]}],"author":{"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","date":"2021-05-13 21:08:32.000000000","tz":-420},"committer":{"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","date":"2021-05-13 21:08:32.000000000","tz":-420},"subject":"fluentd: Fix check for external elasticsearch","message":"fluentd: Fix check for external elasticsearch\n\nWhen enable_elasticsearch is False, and kolla_internal_fqdn is set\nand elasticsearch_address is the default, fluentd is still enabling\noutput to elasticsearch.\n\nelasticsearch_address defaults to kolla_internal_fqdn, but the\ncheck to enable logging to an \"external\" elasticsearch host\ncompares with kolla_internal_vip_address (fqdn !\u003d vip address).\n\nThis patch fixes the comparison to use the kolla_internal_fqdn.\n\nCloses-Bug: #1927880\nChange-Id: If23a6b1273c2639d1296becc9d222546d52f63ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/c1163efac44d2d9dde9affc39b88402b543adcad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/c1163efac44d2d9dde9affc39b88402b543adcad"}]},"branch":"refs/heads/master"},"5552ba70a7a5d06cb390b304729c43fe1a6b26d7":{"kind":"REWORK","_number":2,"created":"2021-05-24 06:00:24.000000000","uploader":{"_account_id":33469,"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","username":"sshambar"},"ref":"refs/changes/21/791321/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/791321/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/791321/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/791321/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/791321/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/791321/2"}}},"commit":{"parents":[{"commit":"1f5a3a288b2ba51ee465bb7c78851c22e24e031e","subject":"Merge \"CI: add grafana to monasca image list\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/1f5a3a288b2ba51ee465bb7c78851c22e24e031e"}]}],"author":{"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","date":"2021-05-13 21:08:32.000000000","tz":-420},"committer":{"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","date":"2021-05-24 05:59:08.000000000","tz":-420},"subject":"fluentd: Fix check for external elasticsearch","message":"fluentd: Fix check for external elasticsearch\n\nWhen enable_elasticsearch is False, and kolla_internal_fqdn is set\nand elasticsearch_address is the default, fluentd is still enabling\noutput to elasticsearch.\n\nelasticsearch_address defaults to kolla_internal_fqdn, but the\ncheck to enable logging to an \"external\" elasticsearch host\ncompares with kolla_internal_vip_address (fqdn !\u003d vip address).\n\nThis patch fixes the comparison to use the kolla_internal_fqdn.\n\nCloses-Bug: #1927880\nChange-Id: If23a6b1273c2639d1296becc9d222546d52f63ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/5552ba70a7a5d06cb390b304729c43fe1a6b26d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/5552ba70a7a5d06cb390b304729c43fe1a6b26d7"}]},"branch":"refs/heads/master"},"1ac30febd334ab54504e3992f0ed5521917aaedd":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-05-24 19:28:06.000000000","uploader":{"_account_id":33469,"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","username":"sshambar"},"ref":"refs/changes/21/791321/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/791321/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/791321/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/791321/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/791321/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/791321/3"}}},"commit":{"parents":[{"commit":"fa4f99950a4e4b0037996b564b87f1293047e1a2","subject":"Merge \"CI: Configure IP on a linux bridge instead of OVS br-ex\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/fa4f99950a4e4b0037996b564b87f1293047e1a2"}]}],"author":{"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","date":"2021-05-13 21:08:32.000000000","tz":-420},"committer":{"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","date":"2021-05-24 19:28:06.000000000","tz":0},"subject":"fluentd: Fix check for external elasticsearch","message":"fluentd: Fix check for external elasticsearch\n\nWhen enable_elasticsearch is False, and kolla_internal_fqdn is set\nand elasticsearch_address is the default, fluentd is still enabling\noutput to elasticsearch.\n\nelasticsearch_address defaults to kolla_internal_fqdn, but the\ncheck to enable logging to an \"external\" elasticsearch host\ncompares with kolla_internal_vip_address (fqdn !\u003d vip address).\n\nThis patch fixes the comparison to use the kolla_internal_fqdn.\n\nCloses-Bug: #1927880\nChange-Id: If23a6b1273c2639d1296becc9d222546d52f63ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/1ac30febd334ab54504e3992f0ed5521917aaedd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/1ac30febd334ab54504e3992f0ed5521917aaedd"}]},"branch":"refs/heads/master","description":"Rebase"},"283b0dec639c032945f5d1d87a66f0c7541c0562":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-05-27 17:26:04.000000000","uploader":{"_account_id":33469,"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","username":"sshambar"},"ref":"refs/changes/21/791321/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/791321/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/791321/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/791321/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/791321/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/791321/4"}}},"commit":{"parents":[{"commit":"3b9dafdf29eb911d2506dcbabe842412e20b1998","subject":"Merge \"Indented two spaces to match the other things in this block\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/3b9dafdf29eb911d2506dcbabe842412e20b1998"}]}],"author":{"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","date":"2021-05-13 21:08:32.000000000","tz":-420},"committer":{"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","date":"2021-05-27 17:26:04.000000000","tz":0},"subject":"fluentd: Fix check for external elasticsearch","message":"fluentd: Fix check for external elasticsearch\n\nWhen enable_elasticsearch is False, and kolla_internal_fqdn is set\nand elasticsearch_address is the default, fluentd is still enabling\noutput to elasticsearch.\n\nelasticsearch_address defaults to kolla_internal_fqdn, but the\ncheck to enable logging to an \"external\" elasticsearch host\ncompares with kolla_internal_vip_address (fqdn !\u003d vip address).\n\nThis patch fixes the comparison to use the kolla_internal_fqdn.\n\nCloses-Bug: #1927880\nChange-Id: If23a6b1273c2639d1296becc9d222546d52f63ac\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/283b0dec639c032945f5d1d87a66f0c7541c0562"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/283b0dec639c032945f5d1d87a66f0c7541c0562"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":33469,"name":"Scott Shambarger","email":"scott-openstack@shambarger.net","username":"sshambar"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
