)]}'
{"id":"openstack%2Fkolla-ansible~823921","triplet_id":"openstack%2Fkolla-ansible~master~I2c34a6b5894402bbebeb3fb96768789bc3c7fe84","project":"openstack/kolla-ansible","branch":"master","topic":"OIDC-cert-optional","attention_set":{},"removed_from_attention_set":{"29100":{"account":{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},"last_update":"2022-01-10 16:33:12.000000000","reason":"Jason Anderson replied on the change"},"10910":{"account":{"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"},"last_update":"2022-01-11 10:38:16.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2c34a6b5894402bbebeb3fb96768789bc3c7fe84","subject":"OpenID Connect certifiate file is optional","status":"MERGED","created":"2022-01-09 21:57:33.000000000","updated":"2022-01-11 10:40:50.000000000","submitted":"2022-01-11 10:38:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"823921-OIDC-cert-optional","meta_rev_id":"9419d3b2707f00ba301a6f8e99eea0756b7283b4","_number":823921,"virtual_id_number":823921,"owner":{"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-01-11 10:38:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"recommended":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2022-01-10 09:21:46.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2022-01-11 09:23:15.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":2,"date":"2022-01-10 16:15:33.000000000","_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":1,"date":"2022-01-10 16:33:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2022-01-11 09:23:15.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},{"value":0,"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}],"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-01-09 22:08:25.000000000","updated_by":{"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"},"reviewer":{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},"state":"REVIEWER"},{"updated":"2022-01-09 23:17:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-01-10 09:14:46.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2022-01-10 09:21:46.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2022-01-10 15:07:27.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"CC"},{"updated":"2022-01-10 16:15:33.000000000","updated_by":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"reviewer":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"state":"REVIEWER"}],"messages":[{"id":"cd2bf46a6378dab8b4d29e615b32b3ed3c83ae73","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"},"date":"2022-01-09 21:57:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a49bccd120af58a7f2e8ab91aae5394673ebaeee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-09 23:17:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-deploy-guide https://zuul.opendev.org/t/openstack/build/1a09b23c33724d80bf1f6d1d95e024a9 : SUCCESS in 3m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/903839bfcb9c46bc824a91bfab023a2f : SUCCESS in 6m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/559fa13de4294c90a86f9aefee37cded : SUCCESS in 5m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c2e39767145c4928a519bb09d9ad2574 : SUCCESS in 5m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/82696a62ba5a4682bb926b07634a08f8 : SUCCESS in 6m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf3eaef126384aaf92f8d8162c6951a8 : SUCCESS in 6m 11s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/2f6072775d384cc2a32650109c9bbe9f : FAILURE in 6m 39s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/54b4c6d7a95342569700d2f023e6617c : SUCCESS in 59m 52s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/e9a30bd3fea746bc8d78cbe7c18028c8 : SUCCESS in 1h 03m 29s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/28b4096852924262ac770a47929b42f1 : SUCCESS in 59m 57s\n- kolla-ansible-centos8s-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/980c335088a14d148b706d0114fdce1b : SUCCESS in 45m 37s (non-voting)\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/0f366547b79e4302b33ce8b16e7532fc : SUCCESS in 49m 04s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/2632fb98f1cc44ceac3cdb37af99f252 : SUCCESS in 53m 07s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/36300b0714d84ab7937479df3a40c362 : SUCCESS in 47m 52s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/7b0eb467e9e6475cb889446ea5fdd9b3 : SUCCESS in 45m 31s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/54f6f9b76a024b43af51f6d07f9b4dfe : SUCCESS in 42m 49s (non-voting)\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/72afe8a755c94cde818ad069a2a11c76 : SUCCESS in 31m 55s (non-voting)\n- kolla-ansible-centos8s-source-cephadm https://zuul.opendev.org/t/openstack/build/8833280db48549a4bcb7895df3c5147f : SUCCESS in 1h 15m 09s (non-voting)\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/bcef7595307b411c89196959f45bed93 : SUCCESS in 45m 33s (non-voting)\n- kolla-ansible-centos8s-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/fa5dab93450d4131bf0e2b94a57c9adc : SUCCESS in 1h 07m 38s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/30ae1369d4a747c293fffe2759c31b96 : SUCCESS in 57m 21s (non-voting)\n- kolla-ansible-centos8s-hashi-vault https://zuul.opendev.org/t/openstack/build/a47833a6f5be47b9a006b2d50892499a : SUCCESS in 2m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2fae215552bdf201092261c0ca18eb8d58200e9e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-09 23:53:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/9546730e3fc64f1895d3112ad74ade48 : SUCCESS in 1h 27m 14s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/453d03eca0e74745b1adb3de36a9d46e : SUCCESS in 1h 54m 16s","accounts_in_message":[],"_revision_number":1},{"id":"b9dd4334cb32b35c42527ab504a7973d62923748","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-01-10 09:14:46.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"7be18680e87d7dfc19485a3a9f5482f65c92ee6d","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2022-01-10 09:21:46.000000000","message":"Patch Set 1: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"76558660838bd5f68b46252d498d17b097158c32","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2022-01-10 15:07:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"205d7928c7194b624810652d2cb1bc2ac31f4f01","author":{"_account_id":30491,"name":"Radosław Piliszek","display_name":"Radek","email":"radek@piliszek.it","username":"yoctozepto","status":"self-employed techologist, collaborating mostly with 7bulls.com"},"date":"2022-01-10 16:15:33.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"502bfe640fbe5daddbd9658940eecd9cca617615","author":{"_account_id":29100,"name":"Jason Anderson","email":"jasonanderson@uchicago.edu","username":"jasonanderson"},"date":"2022-01-10 16:33:12.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"bba9ffdc122a6b9e6133a93146478cd209d91ab9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-10 16:53:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-deploy-guide https://zuul.opendev.org/t/openstack/build/3cb870e384af4fdba2fc3689ae051527 : SUCCESS in 5m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/949b15fa553d47b0b896caef5a797a65 : SUCCESS in 6m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf6201615458483aa2208907868c24fa : SUCCESS in 6m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ed722bec835d4976960dfa263c9beaaa : SUCCESS in 6m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e506d48a6c6c48ba9c99fd0f9e14d067 : SUCCESS in 7m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dfaf631f52c641a7a4d415985b26eaf5 : SUCCESS in 7m 16s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/32bc37c9703c481982604c6e00e4b491 : SUCCESS in 8m 26s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/acbf027d6598484cbca77a5a38fd662f : SUCCESS in 56m 12s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/a7778137d935470391fe3594f84d5225 : SUCCESS in 48m 14s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/159a7a2cf636413aa6d6bda7af2f14ae : SUCCESS in 51m 00s\n- kolla-ansible-centos8s-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/382179b956394a959193f1717839581f : SUCCESS in 1h 10m 54s (non-voting)\n- kolla-ansible-ubuntu-source-multinode-ipv6 https://zuul.opendev.org/t/openstack/build/48f28d9bc17343e98e7753c57caaf985 : SUCCESS in 50m 50s (non-voting)\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/508aec0390a74a07b1629c0bc849bb8b : SUCCESS in 1h 33m 42s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/8d2b0c0b5ab14a4dbb82b248ddd23b5c : SUCCESS in 59m 45s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/49a4855e173b4e5880b7c9aa2b804431 : SUCCESS in 58m 31s\n- kolla-ansible-centos8s-binary https://zuul.opendev.org/t/openstack/build/a99fda22081c44f8ba070069ba341bf8 : SUCCESS in 58m 24s (non-voting)\n- kolla-ansible-ubuntu-binary https://zuul.opendev.org/t/openstack/build/0f23c9f2c78041ed897a2b2b11a8e654 : SUCCESS in 53m 56s (non-voting)\n- kolla-ansible-centos8s-source-cephadm https://zuul.opendev.org/t/openstack/build/46e01294971a4b819dfa9c0099771d8f : FAILURE in 18m 14s (non-voting)\n- kolla-ansible-ubuntu-source-cephadm https://zuul.opendev.org/t/openstack/build/74c357df2d714884ac8eeca234737aae : SUCCESS in 55m 10s (non-voting)\n- kolla-ansible-centos8s-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/1f05abebe72c48d8a7ccc14ce39f25e1 : SUCCESS in 1h 29m 38s (non-voting)\n- kolla-ansible-ubuntu-source-upgrade-cephadm https://zuul.opendev.org/t/openstack/build/65c318740f1f4d4fa78ab7d5f40f497a : SUCCESS in 1h 20m 49s (non-voting)\n- kolla-ansible-centos8s-hashi-vault https://zuul.opendev.org/t/openstack/build/3f9c48ba523c48e0bcdf8dadf15c6537 : SUCCESS in 5m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f74bab816304059254f89271e752e4086e8a2a28","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-10 17:16:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\n\n- kolla-ansible-debian-source-aarch64 https://zuul.opendev.org/t/openstack/build/0efcdc23f18d4fbebc74ab808be50e82 : SUCCESS in 1h 33m 53s (non-voting)\n- kolla-ansible-debian-source-upgrade-aarch64 https://zuul.opendev.org/t/openstack/build/ab2adb5d02fd481594870adb5ed9f6e9 : SUCCESS in 1h 47m 55s","accounts_in_message":[],"_revision_number":1},{"id":"177c8d43f1ef7e28d626d88f9e5500b8ba43685f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2022-01-11 09:23:15.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2a002c1b4cc555481285bdf0835287622b9d2126","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-11 09:23:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ae1b386028be7ba6c0ec11830389fc8a46e367c2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-11 10:38:16.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-deploy-guide https://zuul.opendev.org/t/openstack/build/0fc99af23eb44ef0b8944afd0be8b5ca : SUCCESS in 6m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a2f9dc79c714a9a993a12e831b54486 : SUCCESS in 6m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5ac2803567eb428eb7017d9f959950cc : SUCCESS in 6m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a41b13fb7856425db5bd477c04e759e6 : SUCCESS in 6m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5782702510354a62bdb398bf000f51ef : SUCCESS in 10m 02s\n- openstack-tox-linters https://zuul.opendev.org/t/openstack/build/42b0daaec4534262a93bc4821dc8bcfe : SUCCESS in 7m 47s\n- kolla-ansible-centos8s-source https://zuul.opendev.org/t/openstack/build/1e402ca134a84a6ab7f382ac339cbf47 : SUCCESS in 1h 06m 08s\n- kolla-ansible-centos8s-source-upgrade https://zuul.opendev.org/t/openstack/build/0ae0eb53c65342dfba6d9cb6bcfdd8b0 : SUCCESS in 1h 11m 12s\n- kolla-ansible-debian-source https://zuul.opendev.org/t/openstack/build/e920d93424af4c0bbbd8ffa6e65215e1 : SUCCESS in 48m 22s\n- kolla-ansible-debian-source-upgrade https://zuul.opendev.org/t/openstack/build/d363a7348e354a5ba38d10d9abe44756 : SUCCESS in 1h 02m 14s\n- kolla-ansible-ubuntu-source https://zuul.opendev.org/t/openstack/build/52773a8218be4991a08db7633d18b13b : SUCCESS in 48m 30s\n- kolla-ansible-ubuntu-source-upgrade https://zuul.opendev.org/t/openstack/build/f8c8b93373f04f5a9e076c448a573d4e : SUCCESS in 52m 00s","accounts_in_message":[],"_revision_number":1},{"id":"6956e08844060f6cb6fc2c2dc1e4758dd2097caa","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-11 10:38:16.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"9419d3b2707f00ba301a6f8e99eea0756b7283b4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-01-11 10:40:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-deploy-guide https://zuul.opendev.org/t/openstack/build/8641c0c312974380920a1d7aaa4f016f : SUCCESS in 1m 07s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3d7fa3b25b64c01a9ec33cd2c874cec : SUCCESS in 1m 34s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"78f29fdc5deecdf00e86930ca7737b3e67fc9eed","revisions":{"78f29fdc5deecdf00e86930ca7737b3e67fc9eed":{"kind":"REWORK","_number":1,"created":"2022-01-09 21:57:33.000000000","uploader":{"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"},"ref":"refs/changes/21/823921/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kolla-ansible","ref":"refs/changes/21/823921/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/823921/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/823921/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kolla-ansible refs/changes/21/823921/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kolla-ansible refs/changes/21/823921/1"}}},"commit":{"parents":[{"commit":"dc5eaa4ec738d966c6d0dfb9180f9c45505c4693","subject":"Merge \"Use Volume V3 API in OpenStack exporter\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/dc5eaa4ec738d966c6d0dfb9180f9c45505c4693"}]}],"author":{"name":"Stig Telfer","email":"stig@stackhpc.com","date":"2022-01-09 21:55:34.000000000","tz":0},"committer":{"name":"Stig Telfer","email":"stig@stackhpc.com","date":"2022-01-09 21:55:34.000000000","tz":0},"subject":"OpenID Connect certifiate file is optional","message":"OpenID Connect certifiate file is optional\n\nSome ID provider configurations do not require a certificate file.\nChange the logic to allow this, and update documentation accordingly.\n\nChange-Id: I2c34a6b5894402bbebeb3fb96768789bc3c7fe84\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/78f29fdc5deecdf00e86930ca7737b3e67fc9eed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kolla-ansible/commit/78f29fdc5deecdf00e86930ca7737b3e67fc9eed"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"}},{"label":"Backport-Candidate","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
